)]}'
{"id":"openstack%2Fnetworking-bgpvpn~809502","triplet_id":"openstack%2Fnetworking-bgpvpn~stable%2Fvictoria~Id5bc15a22921a34399e926736d0a9a478598aff7","project":"openstack/networking-bgpvpn","branch":"stable/victoria","topic":"c7-to-c8-stable/victoria","attention_set":{},"removed_from_attention_set":{"20182":{"account":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"last_update":"2022-02-24 16:24:04.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-02-24 16:24:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5bc15a22921a34399e926736d0a9a478598aff7","subject":"Run TripleO jobs on CentOS8 instead of CentOS7","status":"MERGED","created":"2021-09-17 06:16:39.000000000","updated":"2022-02-24 16:30:21.000000000","submitted":"2022-02-24 16:24:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809502-c7-to-c8-stable/victoria","cherry_pick_of_change":809525,"cherry_pick_of_patch_set":1,"meta_rev_id":"010453223e6ce3905d5c77d280f5f1bd8ae3ee78","_number":809502,"virtual_id_number":809502,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 16:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-02-24 16:30:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-02-24 15:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-02-24 15:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 09:17:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-06 07:21:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-10-06 07:22:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-10-11 12:09:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2021-10-11 16:00:38.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-10-25 12:04:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"39092644669fc85400b41ebdc033763597cf3cb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-17 06:16:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fcde9011fa088c4b2960a6acb884d4cd3d3d6f00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 09:17:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969b7f1307b7465ba478cca060a8a8f0 : SUCCESS in 10m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ec4498f9ed8439d8d5f086e6bf180bd : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccb38bbea2c540df9dd6b903bbd421fe : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43addff5cbc421d84e0596942c1df50 : SUCCESS in 10m 47s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/0272a0fd03cb415bb86f54da2b91327d : SUCCESS in 2h 03m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/221b86112c744510a09108e7693c5010 : FAILURE in 4m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/402b114de1c74921b073b552a298ebf5 : SUCCESS in 1h 24m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ce0c0af897e4eec916b1ca82cf6089c : SUCCESS in 9m 17s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/0021a1b824194b13b92738b263e469f5 : SUCCESS in 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"81ea3b032da1d60190c5a418afb72e05716c8a63","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-06 07:21:34.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI\u0027m ok with that change but please update commit message to include \"cherry-pick from ...\" info there.","accounts_in_message":[],"_revision_number":1},{"id":"d2c8ec81c29aacb8ce41b7e31e330a5dd913d4c1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-10-06 07:22:09.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"c50d6d9ee97616e87511547e471cef3cc59a67cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-10-11 07:09:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"12b2ee31bbb129536f1aa9e9adee4bc4863772e6","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-10-11 07:11:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5e52c3588440519cdce0ba86c616b65fb4acd957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 09:18:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c1a91df16d40399327027463fedaf7 : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c64236204d6247b192d6e9d857f0e8e0 : SUCCESS in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73099c05b126484b92543b1b77456eeb : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d86786256b446dbbd8fb74e43026e33 : SUCCESS in 8m 52s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/95f2dc20492f4e3dabae29a9376e9cbf : SUCCESS in 2h 03m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e7b3679228d410cbd10eb08e9c7d697 : FAILURE in 7m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/d90e90b6b4ee4f83b9ac98469830f3f9 : SUCCESS in 1h 13m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e3aa9b6a57b4c8a921aa71746f36049 : SUCCESS in 11m 20s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/657684d6a9764d8b80d65b71bf8ca1b6 : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5cc10b2fef58a3999e1ca730de235cad699ed3ca","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-10-11 12:09:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5fa6ca5a3f85ec8811e5d725fa0eebffbef636a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-10-11 16:00:38.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13f74c12e3920a0bdeb03190260969b0eb77abcb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-10-25 12:04:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68094dbb0b6fc2296f77069741256d2e261c98c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 14:19:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c97781db641843deab3e158e22b6cdfd : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89199aa57e254b4aa654034922522dd3 : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6a6acf4a83c490cbae1af45d08db029 : SUCCESS in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d640052a7641afa78f82c9db307d40 : SUCCESS in 9m 45s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1a30c0de831041e9b875566f696717d1 : SUCCESS in 2h 04m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca74bd88760648a3ad2f34283ba0e8b7 : FAILURE in 6m 48s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/9bf5ae342fd74156acdce89287f1aaa3 : SUCCESS in 1h 01m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efbd106fe6f64ae9bc6ed0ab0f14453f : SUCCESS in 9m 15s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/c792f879ac0d47b189fb79fa48a78e30 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":2},{"id":"92aaf1bf435f5cb3dac6031843123e21de382c8e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-03 11:06:40.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"32cfbfa797e036985168b95b7922835ce2e98c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 13:07:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7758425db6a48c6aad4b5dc86a6642f : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad6b0cbac90e441fbeac9442ed0d51a0 : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b8887e91d2347dab04a1bd7211f4f2d : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f55ccc0d3aa44e39fdd9ab3763a6d6e : SUCCESS in 7m 38s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b232c7fcc667497da026059a267656f3 : SUCCESS in 1h 58m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46ce253b88874e5faf557511e84a4804 : FAILURE in 5m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/8b46be72044744ac8866664a66f59ab6 : SUCCESS in 1h 14m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bb923f8de5349b785dd5d1073129975 : SUCCESS in 9m 02s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/afdad83ff1da40fdb4d54c0b15d91558 : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c024bb37647467aa8cb734d3432995cdb60272e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-03 14:13:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9700a8d1327c4df89308c86dff329edb1411a185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 16:19:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6ad6a471c943ca97a5178deb588af6 : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4e7bccdf1df40d9bbb3dc537cfe70ce : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2573e14d2c94463681c457f31f08e962 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b99ff7bffb425da26c1698e0569e80 : SUCCESS in 9m 25s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ed613e3453a248abaa1f47b347bc5326 : SUCCESS in 2h 00m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61b398f594ef449fa6aac3eecdc64824 : FAILURE in 5m 57s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/26f26eb3a40741a6830190f23469c3f8 : SUCCESS in 1h 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1137b7d154384975a2f3c6525ce2a1b8 : SUCCESS in 10m 12s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/fb2a5548ce7440b1929172701065eccf : SUCCESS in 26m 51s","accounts_in_message":[],"_revision_number":3},{"id":"27941d10dafaff8e3065ef6782619906549e5804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-23 16:29:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d6d06fe89c127066578a87ffdbfc7ad72a3a9a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 18:52:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/935abf8e7e824d37aac95f13ee64d9c6 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a708f9e6e1224660ad284d7c83b01e73 : SUCCESS in 9m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83ca58918a0a443fa2e7653b69694b6a : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ffe5755de440039cce29913143c23b : SUCCESS in 8m 12s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/540098481729439dbd9fcc63f8f1f15d : SUCCESS in 2h 14m 26s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/587d6e6bbddb47dda9fd37907540f423 : SUCCESS in 1h 14m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78225ede81004f8fa92482e4bea3bc5f : SUCCESS in 8m 07s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/052f5f80c71c46f88b8d123fe96d6380 : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":4},{"id":"03ed576f84e74826c46dab9bbab2dc3cec672ff0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-02-24 15:00:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d7d3e0d8c8639e8a0d2a06cdfca9360498aa776d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 15:02:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"794eb320dbdbe1ff8d053dbc3d4c60b1a514e932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:24:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba88eea1faf413892a0dc011041add7 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/456305aad6024728978651c216418144 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d1ebf18b7694a36b0057e206478f5df : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a65cc7c11b94d9f9b1d56522775037a : SUCCESS in 7m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/a5f6829a904448c382ecd1afb15fe49d : SUCCESS in 1h 17m 27s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/3bc358663b0246a2b6c0a88d693b7ee3 : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":4},{"id":"dec7dbc63c1c07c631f592ff9bb18d7c87edd3e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:24:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2881cb48e3e08befa1ca760ce54f3c0e91730908","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:30:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b3b2488cb54e2da0ccd88a463142be : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3f5f7051a58930905147b5471845e76fc8345b32","revisions":{"1417c0097c5555fb2f0c47db238b840624feb717":{"kind":"REWORK","_number":1,"created":"2021-09-17 06:16:39.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/02/809502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/02/809502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/1"}}},"commit":{"parents":[{"commit":"b849ea5c74e3729df25e868a266d8076ad85f4d7","subject":"Switch tempest jobs for stable/victoria branch to victoria jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/b849ea5c74e3729df25e868a266d8076ad85f4d7"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:14:37.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:16:39.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id5bc15a22921a34399e926736d0a9a478598aff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1417c0097c5555fb2f0c47db238b840624feb717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1417c0097c5555fb2f0c47db238b840624feb717"}]},"branch":"refs/heads/stable/victoria"},"522ac104e98d026305203a948de98c4333c525be":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-11 07:09:40.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/02/809502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/02/809502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/2"}}},"commit":{"parents":[{"commit":"b849ea5c74e3729df25e868a266d8076ad85f4d7","subject":"Switch tempest jobs for stable/victoria branch to victoria jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/b849ea5c74e3729df25e868a266d8076ad85f4d7"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:14:37.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-10-11 07:09:40.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id5bc15a22921a34399e926736d0a9a478598aff7\n(cherry picked from commit e403f297b02906641a040969e1d5b62a08f62cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/522ac104e98d026305203a948de98c4333c525be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/522ac104e98d026305203a948de98c4333c525be"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"},"fd97210fd5b86d5f33b9df7cd1680cc15ac34e5c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-03 14:13:06.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/809502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/02/809502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/3"}}},"commit":{"parents":[{"commit":"094a4955e5da0111c85922ab30def2b639fa4fa1","subject":"Make lower-constraints job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/094a4955e5da0111c85922ab30def2b639fa4fa1"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:14:37.000000000","tz":330},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-11-03 14:13:06.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id5bc15a22921a34399e926736d0a9a478598aff7\n(cherry picked from commit e403f297b02906641a040969e1d5b62a08f62cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/fd97210fd5b86d5f33b9df7cd1680cc15ac34e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/fd97210fd5b86d5f33b9df7cd1680cc15ac34e5c"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"3f5f7051a58930905147b5471845e76fc8345b32":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-23 16:29:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/809502/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/02/809502/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/02/809502/4"}}},"commit":{"parents":[{"commit":"3bf3c084fa8b57e4010c2dfe7d78fe121f3cebce","subject":"[stable/victoria] Dropping lower-constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/3bf3c084fa8b57e4010c2dfe7d78fe121f3cebce"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-17 06:14:37.000000000","tz":330},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-02-23 16:29:12.000000000","tz":0},"subject":"Run TripleO jobs on CentOS8 instead of CentOS7","message":"Run TripleO jobs on CentOS8 instead of CentOS7\n\nAs we are cleaning up the c7 jobs and obselete featuresets [1].\nThis change replaces usage of CentOS7 in TripleO jobs by CentOS8.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Id5bc15a22921a34399e926736d0a9a478598aff7\n(cherry picked from commit e403f297b02906641a040969e1d5b62a08f62cf2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/3f5f7051a58930905147b5471845e76fc8345b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/3f5f7051a58930905147b5471845e76fc8345b32"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
