)]}'
{"id":"openstack%2Fnetworking-bgpvpn~839980","triplet_id":"openstack%2Fnetworking-bgpvpn~master~Id90e65ce8ad132b4518e3f979b4d7ad8917ce267","project":"openstack/networking-bgpvpn","branch":"master","topic":"839980","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2022-08-09 12:09:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id90e65ce8ad132b4518e3f979b4d7ad8917ce267","subject":"Address RemovedInDjango40Warning","status":"MERGED","created":"2022-04-29 17:09:50.000000000","updated":"2022-08-09 12:11:09.000000000","submitted":"2022-08-09 12:09:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839980-839980","meta_rev_id":"4fbe9dad16047a90f0940800c7dc1820fd5c8796","_number":839980,"virtual_id_number":839980,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-09 12:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2022-05-02 14:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-08-09 10:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-08-09 10:45:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 19:01:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-02 10:56:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-05-02 14:45:29.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-08-09 10:45:56.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"75043ad7356e416e7969b12e90a98d6125d54545","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-04-29 17:09:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fce75058f657b92f8965d2b70a99c45d42a5b67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:01:58.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d04ded7ecae433fa8832ff46dc54faa : FAILURE in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87413d36ff974781a95e19115d71a781 : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/799788dfd0f04df0871c8c1e30da12e7 : SUCCESS in 8m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a79fd54d0c954a33b71136da3cbf2d93 : SUCCESS in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d9334483e9f48a1b7f045d8a86138d4 : SUCCESS in 7m 33s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/e088eb6f228a463586494f74f7a6f668 : SUCCESS in 1h 47m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fbebababa8b436f9620c9ae4e990e06 : FAILURE in 4m 12s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dcb4d7c606d4222a1b5d86f19e149f2 : SUCCESS in 7m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4588be3288954830a236e41882713a2b : SUCCESS in 1h 18m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/970c6f4c5117481c800354af8c377f79 : SUCCESS in 10m 00s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/8cd96bb5fe4a4213b3fe1d5a325c6855 : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a3e0a6e8b0b73a8fd7b5b7797732b246f2a7c083","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 10:56:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba5261aa8c4aa87997925a8e4d53ff762d1685e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-05-02 12:10:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f4bda64940c5a1d89b1d110f42dc50e3f5b14a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 14:24:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ecd4c1e6eb34e0ea23508bb252975b9 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47a288bb446f4802ae531ad09ed959bf : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/933d669e8f5d48b59b2c56ccf9e9d12d : SUCCESS in 7m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60b80291bc1f4490bd8dab210f7c685b : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f92fe6b6714530913bd132ee472de2 : SUCCESS in 12m 17s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/507c2f390e2641e0beed810093e7aeb5 : SUCCESS in 2h 11m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45eeaed621d442479aa1d2d807b68398 : FAILURE in 4m 08s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a96dec68649f4130ba8e8a45e453d3d5 : SUCCESS in 6m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a156d3c0212246e6accfe451d754e8cc : SUCCESS in 56m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8f685645d70414ca67195ad10254c10 : SUCCESS in 8m 40s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/0afa5adcea9f4311b2252f9c6b42ce0e : SUCCESS in 15m 34s","accounts_in_message":[],"_revision_number":2},{"id":"b9a8855d13ac3b0581238aa1fb8d8b4dcc8cccda","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 14:45:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c5e2da93acaccdd0795284806c514454d8647595","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-08-09 10:45:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e82a572eb23131ddf767404e3945cef881e996ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 10:46:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe39f9de6b7dbb40b8a37d843f42e826ac44dc45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:09:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/550238b001a44161b842825e5f3c865e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c08b440a7cf64667b4a6dc1faaa45ab4 : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5867a73de00142e1921e79beb3e0c73b : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16e6b479e0f8458f8d3e0ada4d4bc8af : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8890d65bd149238640786eb8397d35 : SUCCESS in 8m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/722f462ab1054bd3b53a7dffe1549e18 : SUCCESS in 1h 12m 48s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/eab134adb1e64174975cbd98ae2110e1 : SUCCESS in 22m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ea2604657fe9a4051f9d7d8a519c99ef701489de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:09:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4fbe9dad16047a90f0940800c7dc1820fd5c8796","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:11:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308f5dbfb0be4df2a00d0c48ea3b59e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd33e514a4b478b9d3742999e1f7ed5 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d44be7d590b2078c77d627a203470f6042ae6484","revisions":{"3d8bef3dabba0dbcc6f511215c85c6b1dfa21cab":{"kind":"REWORK","_number":1,"created":"2022-04-29 17:09:50.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/80/839980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/80/839980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/80/839980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/80/839980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/80/839980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/80/839980/1"}}},"commit":{"parents":[{"commit":"9f78f6e393bef578651be239a7e851c122976c38","subject":"CI: fix weekly job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/9f78f6e393bef578651be239a7e851c122976c38"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-04-29 16:54:11.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-04-29 17:09:46.000000000","tz":330},"subject":"Address RemovedInDjango40Warning","message":"Address RemovedInDjango40Warning\n\nIn Django 3.1, django.conf.urls.url() is deprecated\nin favor of django.urls.re_path(). For more info see [1]\nThese were already replaced in Horizon repo by [2].\n\n[1] https://docs.djangoproject.com/en/4.0/releases/3.1/#id2\n[2] https://review.opendev.org/c/openstack/horizon/+/827093\n\nChange-Id: Id90e65ce8ad132b4518e3f979b4d7ad8917ce267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/3d8bef3dabba0dbcc6f511215c85c6b1dfa21cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/3d8bef3dabba0dbcc6f511215c85c6b1dfa21cab"}]},"branch":"refs/heads/master"},"d44be7d590b2078c77d627a203470f6042ae6484":{"kind":"REWORK","_number":2,"created":"2022-05-02 12:10:28.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/80/839980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/80/839980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/80/839980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/80/839980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/80/839980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/80/839980/2"}}},"commit":{"parents":[{"commit":"9f78f6e393bef578651be239a7e851c122976c38","subject":"CI: fix weekly job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/9f78f6e393bef578651be239a7e851c122976c38"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-04-29 16:54:11.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-05-02 12:10:23.000000000","tz":330},"subject":"Address RemovedInDjango40Warning","message":"Address RemovedInDjango40Warning\n\nIn Django 3.1, django.conf.urls.url() is deprecated\nin favor of django.urls.re_path(). For more info see [1]\nThese were already replaced in Horizon repo by [2].\n\n[1] https://docs.djangoproject.com/en/4.0/releases/3.1/#id2\n[2] https://review.opendev.org/c/openstack/horizon/+/827093\n\nChange-Id: Id90e65ce8ad132b4518e3f979b4d7ad8917ce267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/d44be7d590b2078c77d627a203470f6042ae6484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/d44be7d590b2078c77d627a203470f6042ae6484"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
