)]}'
{"id":"openstack%2Fnetworking-bgpvpn~875913","triplet_id":"openstack%2Fnetworking-bgpvpn~master~Icb61aa418476cd4e2d66e918e37fc08abe5191a9","project":"openstack/networking-bgpvpn","branch":"master","attention_set":{},"removed_from_attention_set":{"11805":{"account":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"last_update":"2023-03-01 15:15:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb61aa418476cd4e2d66e918e37fc08abe5191a9","subject":"Drop rc1 and b1 from min versions in requirements.txt","status":"MERGED","created":"2023-03-01 13:20:40.000000000","updated":"2023-03-01 15:16:39.000000000","submitted":"2023-03-01 15:15:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875913","meta_rev_id":"32c78c317704aeb0493b267afc925dbafb9a8844","_number":875913,"virtual_id_number":875913,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-01 15:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-03-01 14:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-03-01 13:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-03-01 14:43:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 13:32:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-03-01 13:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-01 14:43:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"7e0ad89bd50800084928e962a9ce70dc00bb2c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-03-01 13:20:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0a5ad552634c8dfc1faa165072ee5c6e0b14925","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-01 13:32:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a4eb8906e0132dd4cc813d3673e336252558401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:58:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaa975c8bce5473b8c5305f9427d6e45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22b70d316bce49ceb5fcaed7decad097 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63f83d00074d4cdfadf9ed760aa58e2b : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c7783db3a8d4ab7a711aef77dae4c7c : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef63b28484a44dfbe1d6d4707dc392a : SUCCESS in 4m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c69e3bf006c9485ab9076edb458d275c : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da3b4cd848244d1f9d4e3cc73d864498 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a17edaa27f8406f95571dcbe13082d0 : SUCCESS in 7m 33s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/3decaeb5652e4cdda2c2c74bc40b5484 : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"32be2d8013f0b3d6c0b22f57caa86d52847638de","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-01 14:43:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eeacac2a3972c493ae6fa4f77782f73d6721e999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 14:43:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ccd116f0238700d2d90cfbb20df726fde5cdad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:15:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b9038ea4ab4e00a4430d5ba7c52978\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc1df3cb42d46f0b1ae989e311779dd : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ab8040b7012485e9716520b829f830a : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fed6c07a20bc430db5de1f3448130dd3 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2afeeeac92241a4b574478042f0c070 : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b069ae4f4e24fa4a7c37b4c006a0f9e : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12cf5d20613a461d99bd2df9b97c633b : SUCCESS in 8m 00s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/1695d7283d754c4899771ef36a0e830f : SUCCESS in 25m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0054bdef6050fd11af5b0bfbdfda25dcd336c7d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:15:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"32c78c317704aeb0493b267afc925dbafb9a8844","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:16:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d500b6c0d0334778abc79f85752df216\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d83ef22d3d4e8b8083be54ba255330 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"902b82169db078bcd3765ce1115c11ab872c72a6","revisions":{"902b82169db078bcd3765ce1115c11ab872c72a6":{"kind":"REWORK","_number":1,"created":"2023-03-01 13:20:40.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/13/875913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/13/875913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/13/875913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/13/875913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/13/875913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/13/875913/1"}}},"commit":{"parents":[{"commit":"fc202ff7bbf6093dd3fc731e5572abd73c9031cf","subject":"Merge \"Remove TripleO job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/fc202ff7bbf6093dd3fc731e5572abd73c9031cf"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-03-01 13:18:38.000000000","tz":-300},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-03-01 13:20:40.000000000","tz":-300},"subject":"Drop rc1 and b1 from min versions in requirements.txt","message":"Drop rc1 and b1 from min versions in requirements.txt\n\nThe min versions for python3-networking-bagpipe and\npython3-neutron have been set to a release candidate\nand beta version for a while now. Let\u0027s set those to\nthe official corresponding release versions.\n\nChange-Id: Icb61aa418476cd4e2d66e918e37fc08abe5191a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/902b82169db078bcd3765ce1115c11ab872c72a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/902b82169db078bcd3765ce1115c11ab872c72a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
