)]}'
{"id":"openstack%2Fnetworking-bgpvpn~900287","triplet_id":"openstack%2Fnetworking-bgpvpn~master~I4f96a06da3bd322ea1bf43c8d89d5377b3e1129d","project":"openstack/networking-bgpvpn","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-04-04 09:02:23.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-04-04 09:02:23.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-04-04 09:02:23.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-05-07 19:43:24.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-04-04 09:02:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4f96a06da3bd322ea1bf43c8d89d5377b3e1129d","subject":"Add pyproject.toml to support pip 23.1","status":"MERGED","created":"2023-11-07 13:21:53.000000000","updated":"2025-05-07 19:44:15.000000000","submitted":"2025-05-07 19:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"900287-pip-23.1-support","meta_rev_id":"049e1707fb030390b5ca90e4db10b9e27b9fdf02","_number":900287,"virtual_id_number":900287,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 19:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-05-07 11:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-05-07 13:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-05-07 16:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-05-07 11:30:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 14:42:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-26 09:00:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-01-26 09:00:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-01-26 09:00:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-01-26 09:00:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-05-17 20:20:12.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2024-12-20 16:54:56.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REMOVED"}],"messages":[{"id":"9b1b13a5e88c096a442af3c4b993eab1f67763aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-07 13:21:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"287d2f4c99bb76f850fe465db427815bf71d666d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 14:42:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe93f73f0eab409f84c82ca23d460e02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc573bea67d4dc69effae8ebb148b45 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79530ff0d33d4fa8abc2a896701909ba : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd6ff6049da64002b3c69c81ec5b6cef : FAILURE in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea5002c438e1464ea85a4fa8ac015e14 : SUCCESS in 8m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f4ec560d78643d7a39b20e994b0605e : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9bf6536dcfc4b12b37e2bbf4cafb129 : SUCCESS in 9m 48s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ef7822aa176a44bf8b8aa96a3f5be194 : RETRY_LIMIT in 5m 30s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/346e0f8b95c44c8dbe39a9a10618135d : SUCCESS in 1h 19m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffe9bd6e934f43a098611a5fbe187e02 : SUCCESS in 8m 33s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/9fe499a068bd46198001fb3fa3efe7ae : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e587b579b23fb7141fff6d6e4aa726fcaaea84ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-09 15:44:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"49b1958f6f6f6a51b0d3314c779f6e30bffd5b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:06:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ceb9a6d1fc44e979a77ccd9edd602c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9559e4a5d3d34dfba93634de1b59e159 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66ae6cfb3f5646268466a7a8375fddb0 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69a9db619d1445598adfc9903f607fa7 : SUCCESS in 8m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17d9f9ce29eb4d568c3066b01eabf340 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae6673f1158482d9f14828d7f3d187b : SUCCESS in 8m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6f8996ec7d09457583fd71d24e18b27b : FAILURE in 7m 34s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ea5511d43cc2414db8e28827cfe7e193 : SUCCESS in 1h 19m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abec8742dfe44972b665ee66f6e8a9fe : SUCCESS in 8m 29s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/068dee78627546b6bcd51062e0509721 : SUCCESS in 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"55d3b2cd036eee365152760399a1b5c7c76430bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-26 09:00:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cdcf919e46c3ec4fa8099518668cf7d2e5e4a453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 11:02:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/928513f978354cadbac3b8eddc7bb514\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6b4c7925374511b508ff430cec4bea : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f55bc51aef554faeb2cc24ed7940228d : FAILURE in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e5ea567cd0e47ae9044c3c543ad6dfa : FAILURE in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d08dc2043604adabb42cd9f3ac41c45 : FAILURE in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6683505bcc8346a38e330552815cd3b2 : SUCCESS in 8m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/dd9385298d4046849e3ed718609806a6 : FAILURE in 7m 29s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/eade9b52910447c0a9611496912b77fc : SUCCESS in 1h 49m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc791f1d9a364d3e9c3c2747e77e2c75 : SUCCESS in 8m 35s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/abbef09bdd6b4c0d901bacafcc02482a : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":3},{"id":"1962ef39a0e04e4a580a0a2261c5a79f9db1b6bd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-04 09:02:23.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3},{"id":"9b9e650ffd964d90f72655b67b280f84ba328f3b","tag":"autogenerated:gerrit:restore","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-19 11:12:54.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"38dde6f83f1497bcd4146dde3d4dc75146922cb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-19 11:13:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c396c38994603b9acda0f4e6821d535324ecbcf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:24:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0306768bad141f9aea2abc2020632a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/984ab0e109cb4a859079ebbc1ca625d4 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bec16819ff604573ac85e71a0f0ed62e : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7956d33af2f84185b444a848c3f80ab6 : SUCCESS in 7m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e766b71264504739a9936375473c03b9 : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab902034c2eb4eb2b6373db37959b9ac : SUCCESS in 7m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0129fabfb2bf4c5da18859910788beab : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff19a3da3ad48fbb2c338dda97575b2 : SUCCESS in 7m 29s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5b9b7c75ebbc42358279c37eb6b30eab : SUCCESS in 7m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/26ad7a77bc42420d9e991d902e677f93 : TIMED_OUT in 2h 02m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/967926e24c914ff5b2af9c8e069c95a4 : SUCCESS in 8m 25s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/9660343acc364a50af4667194a77cf80 : SUCCESS in 26m 13s","accounts_in_message":[],"_revision_number":4},{"id":"faec5592dcd98fecdddc1dc6cd781686d9663219","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-22 11:48:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c43f59c0c4f9a54b225322941448776a11911d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 13:33:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a47411bd6bc44a38a8729f3cc15111dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a382e82df6457dae87cb04a83228d0 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82b21791e42f43e29a262dad54509fca : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/665eed3ea65247e7af1583549ac312b0 : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b201ed0742a43c6bc121f65112ad896 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6cac9ec575541f3b8eb1a0d321ca7cc : SUCCESS in 8m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2dfd371ed0d44898bb1d01fc867a8fcc : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb07ad5cfba4f2d90ad519dd5c4336b : SUCCESS in 7m 54s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/51cfc633a3364687a0b9b1d224074228 : SUCCESS in 6m 11s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2b8973744879455f9629e8a163baea68 : SUCCESS in 1h 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97790bf59d9647cc8a8c090ee34f3fc2 : SUCCESS in 6m 55s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/d506b7acdd3f4b10b470c5f21a00cc38 : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e74e5cf1269a14846f319e96bee94da9e2260453","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-17 20:20:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6be8a3ebc4d8a67074c1d53f154511f5218a8ca","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-12-20 16:54:56.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_9656\u003e.","accounts_in_message":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"_revision_number":4},{"id":"184aef14fbd7af337e16d7b310e168ba3bb257dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-02 09:22:10.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4078a7c847e5131d841b1d1cf979f2592f0e6eaa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-02 09:22:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a47e4c74996c72632f609bb2b6a7ab0aeb2ea5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 11:26:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6697280c78543518fd71acc2e48ae1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec9f2c0ac154445a67d909531fbb197 : SUCCESS in 10m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02fb6c9e4e294b108c58b783ef8a941d : SUCCESS in 9m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64218a7cb1234dfb85d02b42c74f9c71 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69b885f1154e4505853b943b0fda6922 : SUCCESS in 7m 51s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a1afdb9dd9b14fe79139573f4aa5d47f : SUCCESS in 8m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/92af07f5147049ebadc3dc32df1cb993 : SUCCESS in 1h 58m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3403a5efea74314a2bb29434d5fec10 : SUCCESS in 8m 56s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/19f8abd53f624442bafdcc3734d3c654 : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":5},{"id":"fee77ed1e45ff9da73c82f016c8c071dc9226c08","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-03-11 18:34:44.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b44adfe75e457a73af009cdd470629ae9e66f9e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"real_author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-07 11:29:42.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_8313\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"_revision_number":6},{"id":"d2da19bda95be05f99907161627b17b0087b933e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-07 11:30:26.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"2873a887d4367031a716906c3a48ffb6757e74ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 12:14:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5b47396d71f4af9aef94e4ffaad76c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c1142a6e51c4879b09ea26d09d29834 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/760c126814aa4a58b6b10161caa82b42 : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5478659d9b304d08ae672b3a179eec6d : SUCCESS in 6m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/676a4a0c04ed4fcb846d8e1bdb1d91ca : SUCCESS in 10m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ef39e78f82b4a138812875a07f542c3 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d63e6b7bacf4bd5ad71dac1da4d405e : SUCCESS in 9m 30s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/764b711bdf2e4b5dabe7b602bce17a14 : SUCCESS in 6m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/adbb044ce73e4f41977c6c65586c0190 : FAILURE in 38m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/784ff313a963415eb66a4e4ff2f47194 : SUCCESS in 7m 33s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/91dcd7ec807c43cc815b48eae40b043f : SUCCESS in 18m 36s","accounts_in_message":[],"_revision_number":6},{"id":"32c8f643cd21de6d0bf8734077e2fe42a32d0235","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-07 13:15:55.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"312c013571647736aceadb7f0fc6a9a1db302e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:18:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bdaf43ff07e4e47894af75dd88702d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/267b4918afd541aeab13b9d541820dbc : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/784a80bcc0bf410fa6f84bd6fa9558d7 : SUCCESS in 8m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e78811209e349958f0e0515dffa470b : SUCCESS in 6m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e3b33ea7b5042a291654c37e743ad50 : SUCCESS in 10m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce43d238176a4d61835e1ad82c3d08dd : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372d79d5d049481e9a8fb590873b5fd1 : SUCCESS in 8m 25s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/704d293c5ada4489b41b6f3d3ceed78b : SUCCESS in 6m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8a0885182d8c45a09804385216baec6d : SUCCESS in 1h 01m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fc259f1f71e45a0b7f937a55890c94e : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":6},{"id":"042d78e0126d2c068b1b78ba1d74c00455522b5f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-07 16:07:22.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ce331eece08e86f6e51ec0da5698b97cb09f1170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 16:07:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f90d5ebeecf6371796fd67870fb03450cde3b6ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 16:25:06.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b083be2451db48358b33e30fa500bdd6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1431c910e6ff4072bf30de78b3280287 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aa511fc9fa64e439ce06df308ad7d91 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0820d141edc40dc920c9b6abd8118a1 : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/658f3a1bf39a4fabb8cabc9418a0d7a9 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df82207803104be7916f72f0fb3aa14e : SUCCESS in 8m 31s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/8dec620e2f364d45bae5758892e11596 : SUCCESS in 8m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6aa27b9d986046f19b826737be549922 : FAILURE in 14m 10s","accounts_in_message":[],"_revision_number":6},{"id":"9131bc936c1e9c29977df9b1f31ff4686bb020c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-07 16:27:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9383e33dfdad2976609b289e45548a5093ecf028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 18:01:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5aada2cf09a44feb13ad754a99a9954\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58dff4761cb94533b2e30cc9fad6e5c9 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d56cbc3d5264a0f8ab1060231bcc0bc : SUCCESS in 7m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a932628d2b949beb4cab27cb926b596 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6917f7cc2f704d5ba0270ad83e87078c : SUCCESS in 6m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b5a528f08e24beab2a0f83d3b62c5e2 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83dc3eb63e0c46b19b897542af2929db : SUCCESS in 10m 06s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7efe2d6e11994cb2849e18556ac30ae4 : SUCCESS in 6m 53s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6a3839b46e954b518a47e401af365543 : SUCCESS in 1h 33m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95fff46b17b04a18bf4ac4bea0917501 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":6},{"id":"936fab54b95cc368b27aeccf1743f0ff107c166c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 18:02:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8360ef271abad2fac027de3bad052e060926bc40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 19:43:24.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8741c88c29ad463b9b2d867f3754a9ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824c3aba4d8b4d85877ac6bc87ae73a9 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af0385ed17b4456b9721edb0ae108d6d : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a517e8e035d2431f8760d8bd455ee9e6 : SUCCESS in 5m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/592c4d547d9447bf97c473c7da4887ac : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fda4eca00cd400582f75c33cfa2e017 : SUCCESS in 9m 49s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/397a3ea260234c1dbaac85ed7878e1d4 : SUCCESS in 7m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2b1228c55e974ee8823835bfece80300 : SUCCESS in 1h 40m 59s","accounts_in_message":[],"_revision_number":6},{"id":"4083048bf44c568779aa08b36aa57ed8faa82780","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 19:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"049e1707fb030390b5ca90e4db10b9e27b9fdf02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 19:44:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/392c219f72584e338496142590a96def\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c750a6fc4a4683b1bfb4933eb83411 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1044a82287af01f3f627a233370aaf03fd85830d","revisions":{"7f04459bd56450ca0b2ccd48d9046e4dad7c2512":{"kind":"REWORK","_number":1,"created":"2023-11-07 13:21:53.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/900287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/87/900287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/1"}}},"commit":{"parents":[{"commit":"867e32b9a7df41e5003bda5cc0144f2fb87c5abf","subject":"py311: add required projects to py311 job and add it to weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/867e32b9a7df41e5003bda5cc0144f2fb87c5abf"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-07 13:21:20.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-07 13:21:20.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I4f96a06da3bd322ea1bf43c8d89d5377b3e1129d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/7f04459bd56450ca0b2ccd48d9046e4dad7c2512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/7f04459bd56450ca0b2ccd48d9046e4dad7c2512"}]},"branch":"refs/heads/master"},"1f317b82d115d500b2b263bbf6b383a106ee96d5":{"kind":"REWORK","_number":2,"created":"2023-11-09 15:44:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/900287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/87/900287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/2"}}},"commit":{"parents":[{"commit":"867e32b9a7df41e5003bda5cc0144f2fb87c5abf","subject":"py311: add required projects to py311 job and add it to weekly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/867e32b9a7df41e5003bda5cc0144f2fb87c5abf"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-07 13:21:20.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-09 15:44:22.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I4f96a06da3bd322ea1bf43c8d89d5377b3e1129d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1f317b82d115d500b2b263bbf6b383a106ee96d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1f317b82d115d500b2b263bbf6b383a106ee96d5"}]},"branch":"refs/heads/master"},"2568b8ae0ac97941302e45fbc7c2c7c901c92ba7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-01-26 09:00:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/900287/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/87/900287/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/3"}}},"commit":{"parents":[{"commit":"deafdf6b3af77dd796ffb624cf160488568901e8","subject":"Update python classifier with py3.11 in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/deafdf6b3af77dd796ffb624cf160488568901e8"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-07 13:21:20.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2024-01-26 09:00:12.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I4f96a06da3bd322ea1bf43c8d89d5377b3e1129d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2568b8ae0ac97941302e45fbc7c2c7c901c92ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/2568b8ae0ac97941302e45fbc7c2c7c901c92ba7"}]},"branch":"refs/heads/master","description":"Rebase"},"0aab4614c35ea6f6fd7bc246dc6340a9f65915d0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-19 11:13:04.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/900287/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/87/900287/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/4"}}},"commit":{"parents":[{"commit":"4b2dee22057d75d491ba9d637975985a113adc38","subject":"Merge \"coveragerc: Remove reference to non-existant path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/4b2dee22057d75d491ba9d637975985a113adc38"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-07 13:21:20.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2024-04-19 11:13:04.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I4f96a06da3bd322ea1bf43c8d89d5377b3e1129d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/0aab4614c35ea6f6fd7bc246dc6340a9f65915d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/0aab4614c35ea6f6fd7bc246dc6340a9f65915d0"}]},"branch":"refs/heads/master","description":"Rebase"},"07177343f2809a3fe9dd1df130331fec0c266c6f":{"kind":"REWORK","_number":5,"created":"2025-01-02 09:22:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/87/900287/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/87/900287/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/5"}}},"commit":{"parents":[{"commit":"496907993338a55ea4b613f2f4e9521dbd7456ba","subject":"pep8: Issue on Noble (U. 24.04) with pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/496907993338a55ea4b613f2f4e9521dbd7456ba"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-07 13:21:20.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2025-01-02 09:21:52.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I4f96a06da3bd322ea1bf43c8d89d5377b3e1129d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/07177343f2809a3fe9dd1df130331fec0c266c6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/07177343f2809a3fe9dd1df130331fec0c266c6f"}]},"branch":"refs/heads/master"},"1044a82287af01f3f627a233370aaf03fd85830d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-05-07 11:29:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"real_uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/900287/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/87/900287/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/87/900287/6"}}},"commit":{"parents":[{"commit":"26a124b8cd80299f4fd5db11cab4d7cb3fe0c6b9","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/26a124b8cd80299f4fd5db11cab4d7cb3fe0c6b9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-11-07 13:21:20.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2025-05-07 11:29:42.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\nFor details see the below mail thread:\nhttps://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/HVFN5RBSHRTM3B2UUKPAWKH6H6AT6CYR/\n\nand the commit message of the keystone patch:\nhttps://review.opendev.org/c/899519\n\nChange-Id: I4f96a06da3bd322ea1bf43c8d89d5377b3e1129d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1044a82287af01f3f627a233370aaf03fd85830d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1044a82287af01f3f627a233370aaf03fd85830d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
