)]}'
{"id":"openstack%2Fnetworking-bgpvpn~928639","triplet_id":"openstack%2Fnetworking-bgpvpn~master~I1c3fb99bbe6e8eb020768b2a29bb0adfedadecb6","project":"openstack/networking-bgpvpn","branch":"master","topic":"release-health-check-cwrc-dalmatian","hashtags":[],"change_id":"I1c3fb99bbe6e8eb020768b2a29bb0adfedadecb6","subject":"DNM: gate health test","status":"ABANDONED","created":"2024-09-09 15:26:48.000000000","updated":"2024-09-10 08:03:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a678fabc782da0b6858e619bce1eb829745621e3","_number":928639,"virtual_id_number":928639,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-09 17:01:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 17:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"403a7ad23b4abe6bee5b26b48c28e0351b163ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-09 15:26:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e778441ed27bb3df5a26ec20b0f523f0650c7902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 17:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26781b35e4e9457d9cffe2f1fab93825\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0915e6f24b274098abea2f6f5ddbc5d7 : SUCCESS in 9m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc641ed0b8344b099b07be90000c795 : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f5ab12b84024ca5b3d38bdba9411180 : SUCCESS in 8m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23e8d8f711b94693a404a71277279aae : SUCCESS in 8m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64167f0d08ca4060b891d748bd39d26d : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d9d72defa940e0b612b73547147e3e : SUCCESS in 7m 20s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/9f50f625aa4f42e28ab7611240e7faaa : SUCCESS in 6m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/dbed284a2958485bbc042112e03f3ee6 : SUCCESS in 1h 29m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e323ad48e464e6983e7b31a017522f0 : SUCCESS in 8m 20s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/7009cb9bdb764970a5ef5267aa852875 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a678fabc782da0b6858e619bce1eb829745621e3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-10 08:03:18.000000000","message":"Abandoned\n\ngate is not blocked, abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b1bd6b10f1222bb9c3859c71cebf3c5e722a3dd","revisions":{"5b1bd6b10f1222bb9c3859c71cebf3c5e722a3dd":{"kind":"REWORK","_number":1,"created":"2024-09-09 15:26:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/39/928639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/39/928639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/39/928639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/39/928639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/39/928639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/39/928639/1"}}},"commit":{"parents":[{"commit":"e62d94fc981f2a54381c6f1d3978e1e89c58d560","subject":"Add pre-commit configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/e62d94fc981f2a54381c6f1d3978e1e89c58d560"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 15:26:39.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 15:26:47.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I1c3fb99bbe6e8eb020768b2a29bb0adfedadecb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/5b1bd6b10f1222bb9c3859c71cebf3c5e722a3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/5b1bd6b10f1222bb9c3859c71cebf3c5e722a3dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
