)]}'
{"id":"openstack%2Fnetworking-bgpvpn~936732","triplet_id":"openstack%2Fnetworking-bgpvpn~stable%2F2023.2~Ia6c14ea8843ebccffecada2808038b23859ef7af","project":"openstack/networking-bgpvpn","branch":"stable/2023.2","topic":"stadium-stable-jobs","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-11-29 16:58:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia6c14ea8843ebccffecada2808038b23859ef7af","subject":"Stable Only: Use branched tempest jobs","status":"MERGED","created":"2024-11-29 13:05:30.000000000","updated":"2024-11-29 17:01:58.000000000","submitted":"2024-11-29 16:58:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936732-stadium-stable-jobs","meta_rev_id":"de9676fbed0ff864b0875bd1cf246ec16bf1ba8c","_number":936732,"virtual_id_number":936732,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-29 16:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-11-29 15:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-11-29 15:27:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-11-29 15:27:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-29 14:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-29 15:07:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-11-29 15:27:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"210c4b65a617e7aa86d9ba05d138006721a7cf89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-29 13:05:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5632abef9967ffd609fd3958ae152d649742210c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 14:46:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f7ae38ea7c45ee8c64d0beedb9b79a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1553350c66b4011ab90effb20cf13dd : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/caf16539b1154ac4a1e921be1a534676 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60b215ef4fff414c9f31a2784b972997 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83171943db7c4f98973d6f844631b926 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c1fce7e08d8436abd1ad84f19133190 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd784459673c48529c502cf660174ee0 : SUCCESS in 8m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/e4ff2994d5dc4c0e96edd3a0c7356607 : SUCCESS in 1h 29m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/841708ea105b47eb96b79c5ad09e8dd0 : SUCCESS in 9m 13s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/cb8a934201264eaea97e53afce6c3671 : SUCCESS in 22m 40s","accounts_in_message":[],"_revision_number":1},{"id":"66ce16b58e06ca555e67522b5b4ccae25c5671ca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-29 15:07:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"795f51e6881baa7ffb605a61ba8d75541e0fd76e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-29 15:27:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7c3c6115f84e0c3f21ab3ef13a6bc9992e788d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 15:28:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa1644c59e6e094548daa6190076173ffcf791bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 16:58:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e390e533514a4c2295e85aeac8a6f9f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3599cfc6f064b7281924b5b5bda7f61 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f93235e4737246a4ba76c847537845a9 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebaed18613ba47eda1ebe223cd2f5c98 : SUCCESS in 7m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd45a5f524454ba781e89ded5621ccce : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8862980c217437f80bca16251030d61 : SUCCESS in 7m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/d44a28a74f9142a5b1eb0c2d65be1b2a : SUCCESS in 1h 25m 24s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/c289097dfa1647d8906b2c944ae64c34 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"83d6ecd5cb6a2eba4e4f699009c831a8d83207e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 16:58:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"de9676fbed0ff864b0875bd1cf246ec16bf1ba8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 17:01:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6bd78915cd2471faf4a9499714e9df8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c8e277b5d14c5f80d63020e880f299 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d190d85fe8ceca495a3b1989d506c000504a7f7","revisions":{"4d190d85fe8ceca495a3b1989d506c000504a7f7":{"kind":"REWORK","_number":1,"created":"2024-11-29 13:05:30.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/32/936732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/32/936732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/32/936732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/32/936732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/32/936732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/32/936732/1"}}},"commit":{"parents":[{"commit":"a955fb7d17e34d4303ace8d5ee7441f6c5b7e310","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/a955fb7d17e34d4303ace8d5ee7441f6c5b7e310"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-29 13:05:16.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-29 13:05:18.000000000","tz":330},"subject":"Stable Only: Use branched tempest jobs","message":"Stable Only: Use branched tempest jobs\n\nChange-Id: Ia6c14ea8843ebccffecada2808038b23859ef7af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/4d190d85fe8ceca495a3b1989d506c000504a7f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/4d190d85fe8ceca495a3b1989d506c000504a7f7"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
