)]}'
{"id":"openstack%2Fnetworking-bgpvpn~950212","triplet_id":"openstack%2Fnetworking-bgpvpn~master~I90deb35bb6ed843aa132171907c9efe2a09fee99","project":"openstack/networking-bgpvpn","branch":"master","topic":"set-service-type-on-provider-config-register","attention_set":{},"removed_from_attention_set":{"30314":{"account":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"last_update":"2025-06-12 14:11:13.000000000","reason":"\u003cGERRIT_ACCOUNT_30314\u003e replied on the change","reason_account":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"}}},"hashtags":[],"change_id":"I90deb35bb6ed843aa132171907c9efe2a09fee99","subject":"Set service type when registering provider config","status":"MERGED","created":"2025-05-16 17:19:53.000000000","updated":"2025-06-12 16:56:06.000000000","submitted":"2025-06-12 16:54:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950212-set-service-type-on-provider-config-register","meta_rev_id":"eb02d9f0bf463f0ede9c203a075e796e83ae2ca5","_number":950212,"virtual_id_number":950212,"owner":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-12 16:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-06-10 14:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-05-21 07:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-06-10 14:17:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 18:58:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-19 13:37:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-05-21 07:27:59.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-06-10 14:17:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"ccc91140e3eaa51ae5419e19ef4bb8a6fe886484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2025-05-16 17:19:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a4b89d59bf0918bc5dd26c88cbb7df5e8fb6326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:58:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd86907640d4cce940965b7756067d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c06baff80ac4bca90cb516e3e6cbb15 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7213385b83846609cb94b1751c113fb : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb4be0f8b9b74d11a2921d1134d93de6 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78818efa5cbc4be6b5103fbf373ac579 : SUCCESS in 7m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abe44008aec44f8cb2845c2cb0c96d29 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe05c6feb224efa83971d293960b40c : SUCCESS in 9m 06s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c73f447597f6482999c992302ef921dd : SUCCESS in 6m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a48082899f5f4d94a7e3024a69c9c247 : SUCCESS in 1h 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c6e94642c1c4ff6baf67504c3c5cf40 : SUCCESS in 4m 39s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/430513283655413ca5b31a8bc8d05dd5 : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ae5a5f2f0069fc480886ec86fa042be14f208f45","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-05-21 07:27:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b562fd1d98e542d19c0cbfdc7e20aa4b79b88e7c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-06-10 14:17:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d4b586727c94884897fd34d0ae6d8601f7d26e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 14:18:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48f98ee66ee53739ba1434e5ad45558ad8005d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 16:14:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d10b001ead9041d281cfac8647a058ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2c178a2e86d47de8bd18bc7c5eba64e : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89b1a25d2f1e47e387fdea1017fe2861 : FAILURE in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26d240119c3740c8b9d4fc044c7224fd : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23fcbf50391c40a2997710244ce0eef0 : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/396253c535a440f6a7d2935c20c490a7 : SUCCESS in 5m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/78f46d847b8d4f60b934f8dda40d312d : SUCCESS in 6m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f33066449e7d43938b8fcdcdc8ebcb65 : SUCCESS in 1h 45m 14s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/be83f752568e45449c0f97f9d18453c5 : SUCCESS in 18m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7629c64d3dd7cf165ada908282ed379ddaccedf5","author":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"date":"2025-06-12 14:11:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b4097ad4013b91a24a3cc719aed5d515cd3382e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 15:15:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4241a8b85dbd48f4b04f4226b10c874a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb216b503a443ea8d02c02c188d7c8e : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01b2a1ae89e445e198488d9c23d1a6e4 : SUCCESS in 8m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3fcd38ff50544349b5d8a4f65a7819e9 : SUCCESS in 9m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba6835fad0234eb9886541c437c23a68 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489c45f22a564590b61e79601af9bc1a : SUCCESS in 10m 08s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0889c2fd721f453ba27717ec8e878e1c : SUCCESS in 9m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/bda6b3ff41354eeb9fb01d3bd91fc2b2 : SUCCESS in 1h 02m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6537741495e84c36bbce11f5fb9e27ae : SUCCESS in 10m 36s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/5a384273603743959d89a32a3f9e2709 : SUCCESS in 21m 33s","accounts_in_message":[],"_revision_number":1},{"id":"16b7a7aa46436935899afb967be9e35e01878da8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 15:15:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9c04c4d7bfe77bdb36e1831ea3d56e02f5db69e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 16:54:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a6c7dd67d044aefa77cd1d038db6817\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9f61d64ea645d8a289c1b5597a57f6 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/983027c821784e0085d784111f681e8b : SUCCESS in 7m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d22f190f60d402faa9394399402f47b : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7870e48c962b44598e3261047ee1f8e7 : SUCCESS in 9m 18s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e176837573904f02942515f051dd4af2 : SUCCESS in 4m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/fcaaa5b38e6245a1929f8ebeea9493ff : SUCCESS in 1h 37m 30s\n- networking-bgpvpn-functional-full https://zuul.opendev.org/t/openstack/build/8ba87693ca3146698e560ad1baa2c1ea : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"eb8b45acd8ec1ab10dd1a25d628d2fbfa25326ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 16:54:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb02d9f0bf463f0ede9c203a075e796e83ae2ca5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 16:56:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d552b1f71544ee18ea29dc442734369\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f97925b9a2648d8a2bbdadaab6930be : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"000191eba886c7c77be013c9a800634691053324","revisions":{"000191eba886c7c77be013c9a800634691053324":{"kind":"REWORK","_number":1,"created":"2025-05-16 17:19:53.000000000","uploader":{"_account_id":30314,"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","username":"seba"},"ref":"refs/changes/12/950212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-bgpvpn","ref":"refs/changes/12/950212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/12/950212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/12/950212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-bgpvpn refs/changes/12/950212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-bgpvpn refs/changes/12/950212/1"}}},"commit":{"parents":[{"commit":"1044a82287af01f3f627a233370aaf03fd85830d","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/1044a82287af01f3f627a233370aaf03fd85830d"}]}],"author":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2025-05-16 16:59:17.000000000","tz":120},"committer":{"name":"Sebastian Lohff","email":"sebastian.lohff@sap.com","date":"2025-05-16 17:05:27.000000000","tz":120},"subject":"Set service type when registering provider config","message":"Set service type when registering provider config\n\nWhen registering the provider configuration with the service type\nmanager we need to specify a service type for ProviderConfiguration(),\nas described in the Neutron Contributors documentation[0]. If this is\nnot done it might lead to duplicate service providers configuration, as\nthe ProviderConfiguration no longer filters out other services.\n\n[0] https://docs.openstack.org/neutron/latest/contributor/contribute.html#service-providers\n\nChange-Id: I90deb35bb6ed843aa132171907c9efe2a09fee99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/000191eba886c7c77be013c9a800634691053324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-bgpvpn/commit/000191eba886c7c77be013c9a800634691053324"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
