)]}'
{"id":"openstack%2Fnetworking-generic-switch~1004210","triplet_id":"openstack%2Fnetworking-generic-switch~master~Ide1ce2cb446b7204bf1b689241de6af92400a0c5","project":"openstack/networking-generic-switch","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-07 10:15:19.000000000","reason":"\u003cGERRIT_ACCOUNT_24245\u003e replied on the change","reason_account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ide1ce2cb446b7204bf1b689241de6af92400a0c5","subject":"tox: Drop redundant passenv","status":"NEW","created":"2026-09-05 18:03:06.000000000","updated":"2026-09-07 10:15:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d7ad2338ec5566fb8620cb40cab30173f517203","_number":1004210,"virtual_id_number":1004210,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-05 18:36:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2026-09-07 10:15:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 18:36:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 10:15:19.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"}],"messages":[{"id":"97352ecf51b71715a37bafd4d2b5bf5ee08dc59c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:03:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12445588119966c06d45666173e5de8e53b6d7be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:12:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8963b8dfbe4975b62fa591ee9a6dacde1dc2d108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:15:42.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fb3b5de519e52dcef640c43021dd9f41b577e587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 18:36:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecc5f60ca6b34810916760741f1702a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5923588247714787806d5aefb35314c2 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff5b8472eeb5424a9e4738133bb45a0d : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2432e53ce528449c9a4b56b0cb7fdaab : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54ddab883dbd4f398f699f77175d7df0 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4085a7d4525c4af7b9e132377b2210f2 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/931d1274421d4d3a81e256c3c883ff46 : SUCCESS in 3m 34s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/a50f85cb69d54f05a4cde614b4e438b9 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":3},{"id":"9d7ad2338ec5566fb8620cb40cab30173f517203","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-09-07 10:15:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8fb5544c3293b3237043264198126ad7e24a6322","revisions":{"8acd6823991d2a2e047ed75a5004d6bc1cdd036e":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:03:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/1004210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/10/1004210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/1"}}},"commit":{"parents":[{"commit":"42c59790d62a3ece7c9674a678a20e3ddf80b5b1","subject":"Merge \"batching: Fix missed batch result watches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/42c59790d62a3ece7c9674a678a20e3ddf80b5b1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:02:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:02:55.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https: //tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Ide1ce2cb446b7204bf1b689241de6af92400a0c5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/8acd6823991d2a2e047ed75a5004d6bc1cdd036e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/8acd6823991d2a2e047ed75a5004d6bc1cdd036e"}]},"branch":"refs/heads/master"},"8c83b94af2eac5557acdfa782b3fff4c72da0898":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-05 18:12:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/1004210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/10/1004210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/2"}}},"commit":{"parents":[{"commit":"42c59790d62a3ece7c9674a678a20e3ddf80b5b1","subject":"Merge \"batching: Fix missed batch result watches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/42c59790d62a3ece7c9674a678a20e3ddf80b5b1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:02:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:12:14.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Ide1ce2cb446b7204bf1b689241de6af92400a0c5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/8c83b94af2eac5557acdfa782b3fff4c72da0898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/8c83b94af2eac5557acdfa782b3fff4c72da0898"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8fb5544c3293b3237043264198126ad7e24a6322":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-05 18:15:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/1004210/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/10/1004210/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/10/1004210/3"}}},"commit":{"parents":[{"commit":"42c59790d62a3ece7c9674a678a20e3ddf80b5b1","subject":"Merge \"batching: Fix missed batch result watches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/42c59790d62a3ece7c9674a678a20e3ddf80b5b1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:02:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:15:42.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Ide1ce2cb446b7204bf1b689241de6af92400a0c5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/8fb5544c3293b3237043264198126ad7e24a6322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/8fb5544c3293b3237043264198126ad7e24a6322"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
