)]}'
{"id":"openstack%2Fnetworking-generic-switch~524138","triplet_id":"openstack%2Fnetworking-generic-switch~master~I77f57ac1730eac2869fbdfeb112d1a9ea2b55fb1","project":"openstack/networking-generic-switch","branch":"master","topic":"bug/1690115","hashtags":[],"change_id":"I77f57ac1730eac2869fbdfeb112d1a9ea2b55fb1","subject":"Add support for trunk ports on Dell switches","status":"MERGED","created":"2017-11-30 11:07:27.000000000","updated":"2019-03-19 15:05:50.000000000","submitted":"2017-12-15 10:42:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"524138-1513334520214-7f84d19e","meta_rev_id":"744ef46bd825b782e1ad06727feae989422348ca","_number":524138,"virtual_id_number":524138,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26576,"name":"yanpeifei","email":"yanpeifei@gohighsec.com","username":"yanpeifei","inactive":true},{"value":0,"_account_id":21697,"name":"Sumedh Sathaye","email":"Sumedh_Sathaye@Dell.com","username":"swsyes"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"date":"2017-12-15 08:49:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2017-12-15 10:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"recommended":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"all":[{"value":1,"date":"2017-12-15 08:45:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26576,"name":"yanpeifei","email":"yanpeifei@gohighsec.com","username":"yanpeifei","inactive":true},{"value":0,"_account_id":21697,"name":"Sumedh Sathaye","email":"Sumedh_Sathaye@Dell.com","username":"swsyes"},{"value":1,"date":"2017-12-14 12:56:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2017-12-14 23:11:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2017-12-14 13:51:52.000000000","_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":1,"date":"2017-12-14 13:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"all":[{"value":0,"_account_id":26576,"name":"yanpeifei","email":"yanpeifei@gohighsec.com","username":"yanpeifei","inactive":true},{"value":0,"date":"2017-12-18 20:14:02.000000000","post_submit":true,"_account_id":21697,"name":"Sumedh Sathaye","email":"Sumedh_Sathaye@Dell.com","username":"swsyes"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2017-12-15 08:43:24.000000000","_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":26576,"name":"yanpeifei","email":"yanpeifei@gohighsec.com","username":"yanpeifei","inactive":true},{"value":0,"_account_id":21697,"name":"Sumedh Sathaye","email":"Sumedh_Sathaye@Dell.com","username":"swsyes"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":21697,"name":"Sumedh Sathaye","email":"Sumedh_Sathaye@Dell.com","username":"swsyes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":26576,"name":"yanpeifei","email":"yanpeifei@gohighsec.com","username":"yanpeifei","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-14 12:56:00.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2017-12-14 13:55:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2017-12-14 23:11:53.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2017-12-15 08:43:24.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2017-12-15 08:45:52.000000000","updated_by":{"_account_id":26576,"name":"yanpeifei","email":"yanpeifei@gohighsec.com","username":"yanpeifei","inactive":true},"reviewer":{"_account_id":26576,"name":"yanpeifei","email":"yanpeifei@gohighsec.com","username":"yanpeifei","inactive":true},"state":"REVIEWER"},{"updated":"2017-12-15 10:42:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-12-18 20:14:02.000000000","updated_by":{"_account_id":21697,"name":"Sumedh Sathaye","email":"Sumedh_Sathaye@Dell.com","username":"swsyes"},"reviewer":{"_account_id":21697,"name":"Sumedh Sathaye","email":"Sumedh_Sathaye@Dell.com","username":"swsyes"},"state":"REVIEWER"}],"messages":[{"id":"b9c55686b78e3078906e023f307efe4891477cc3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2017-11-30 11:07:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77054b14f2a634ba0c0f486053dd5a0f7f2d8cb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-30 13:01:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/38/524138/1/check/openstack-tox-pep8/45fd54b/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/38/524138/1/check/openstack-tox-py27/d459543/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/38/524138/1/check/openstack-tox-py35/f31e0a5/ : SUCCESS in 5m 09s\n- networking-generic-switch-tempest-dsvm-dlm http://logs.openstack.org/38/524138/1/check/networking-generic-switch-tempest-dsvm-dlm/4ea4be1/ : SUCCESS in 19m 32s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode http://logs.openstack.org/38/524138/1/check/ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode/58fe354/ : SUCCESS in 1h 25m 52s\n- ironic-grenade-dsvm-multinode-multitenant http://logs.openstack.org/38/524138/1/check/ironic-grenade-dsvm-multinode-multitenant/a9edade/ : SUCCESS in 1h 52m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a91eb8c816b9551a34dcad7df619a1e2c95716ad","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2017-12-04 14:14:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3a75dbdcf0e7157e774dd0dca9d33ca39500a7a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2017-12-14 09:17:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e346038d2de0b803bbf7a2e2e33afb06f922240","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 11:16:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/38/524138/2/check/openstack-tox-pep8/9b15131/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/38/524138/2/check/openstack-tox-py27/4d837f1/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/38/524138/2/check/openstack-tox-py35/3f1876a/ : SUCCESS in 4m 34s\n- networking-generic-switch-tempest-dsvm-dlm http://logs.openstack.org/38/524138/2/check/networking-generic-switch-tempest-dsvm-dlm/15c0dde/ : SUCCESS in 26m 13s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode http://logs.openstack.org/38/524138/2/check/ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode/f69a644/ : SUCCESS in 1h 19m 16s\n- ironic-grenade-dsvm-multinode-multitenant http://logs.openstack.org/38/524138/2/check/ironic-grenade-dsvm-multinode-multitenant/dae713b/ : SUCCESS in 1h 55m 41s","accounts_in_message":[],"_revision_number":2},{"id":"130e320781907fdf99a7b47362b2803276b2b312","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2017-12-14 12:56:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6dcc9fd485b77197f4f20016001dbdf6f2b5e87b","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2017-12-14 13:51:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11398e89d4d169981e176c6afb6d55b7ed805488","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2017-12-14 13:55:06.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good to me, it matches the static configuration performed on our Dell Force10 S6000 switches.","accounts_in_message":[],"_revision_number":2},{"id":"569153834d916613a06ba1fc4b89bd24a6c27d58","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2017-12-14 23:11:53.000000000","message":"Patch Set 2: Code-Review+1\n\nsuggest also updating device list at https://github.com/openstack/networking-generic-switch\nto include Dell Force10","accounts_in_message":[],"_revision_number":2},{"id":"4174f413a49b0d75340089655951e6170de4d12f","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2017-12-15 08:43:24.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"17e2782054d4ca3bc945e0bbb32a567601a8449b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 08:43:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4fb9825041075f16628e7cd91116b09aa5f4cb77","author":{"_account_id":26576,"name":"yanpeifei","email":"yanpeifei@gohighsec.com","username":"yanpeifei","inactive":true},"date":"2017-12-15 08:45:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5cfbbba5471aec21bd77e72b11ecb4ec17a89cff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2017-12-15 08:49:27.000000000","message":"Patch Set 2:\n\n\u003e suggest also updating device list at https://github.com/openstack/networking-generic-switch\n \u003e to include Dell Force10\n\nI\u0027ll submit a follow up patch for this.","accounts_in_message":[],"_revision_number":2},{"id":"ce406746997f8e90cac755c1860ab8be0bf6a161","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2017-12-15 08:49:55.000000000","message":"Patch Set 2:\n\n\u003e \u003e suggest also updating device list at https://github.com/openstack/networking-generic-switch\n \u003e \u003e to include Dell Force10\n \u003e \n \u003e I\u0027ll submit a follow up patch for this.\n\nthanks Mark, just wanted to ask about it :)","accounts_in_message":[],"_revision_number":2},{"id":"8ab2d23c292cb13a8298321dc01d581ae7f3f133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 10:42:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6c6ee1c0175d653a89ab2c1fd2db2a07ea56f7bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 10:42:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/38/524138/2/gate/openstack-tox-pep8/fe508f4/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/38/524138/2/gate/openstack-tox-py27/9ed9014/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/38/524138/2/gate/openstack-tox-py35/72d4335/ : SUCCESS in 6m 53s\n- networking-generic-switch-tempest-dsvm-dlm http://logs.openstack.org/38/524138/2/gate/networking-generic-switch-tempest-dsvm-dlm/4f894d7/ : SUCCESS in 22m 00s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode http://logs.openstack.org/38/524138/2/gate/ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode/62ce840/ : SUCCESS in 1h 22m 45s\n- ironic-grenade-dsvm-multinode-multitenant http://logs.openstack.org/38/524138/2/gate/ironic-grenade-dsvm-multinode-multitenant/d2f7e5f/ : SUCCESS in 1h 56m 15s","accounts_in_message":[],"_revision_number":2},{"id":"c7898deff048a011b17e038af7128f7135e385fc","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2019-03-19 05:14:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"559c202c6889ddcc266c3a515cc0df4d933458e3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-03-19 08:33:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d9b957b6d00837d980755a7c1650c29635fd136","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2019-03-19 15:05:50.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nThanks for the info! That makes sense. I suppose something else is going on with how we\u0027re doing things; it seems like it\u0027s not causing it to jump back to the top, rather it\u0027s causing a log out.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d8d58a81565116a8fae796d55ab5ee6fe90d83c7","revisions":{"b2f197bc24ee12c31dc53b9beb78f0a1a05f018a":{"kind":"REWORK","_number":1,"created":"2017-11-30 11:07:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/38/524138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/38/524138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/38/524138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/38/524138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/38/524138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/38/524138/1"}}},"commit":{"parents":[{"commit":"6f035ddf808f1cfd3fa876668ddbc0bea82fdbc6","subject":"Support configuration of trunk ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/6f035ddf808f1cfd3fa876668ddbc0bea82fdbc6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2017-05-11 09:39:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2017-11-30 11:07:02.000000000","tz":0},"subject":"Add support for trunk ports on Dell switches","message":"Add support for trunk ports on Dell switches\n\nOn Dell Force10 switches, interfaces are added as tagged members of a\nVLAN using the following commands:\n\ninterface vlan \u003cid\u003e\ntagged \u003cif\u003e\n\nThey can be removed similarly:\n\ninterface vlan \u003cid\u003e\nno tagged \u003cif\u003e\n\nThis change uses these commands to support configuration of trunk ports\nfor all networks.\n\nChange-Id: I77f57ac1730eac2869fbdfeb112d1a9ea2b55fb1\nPartial-Bug: #1690115\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/b2f197bc24ee12c31dc53b9beb78f0a1a05f018a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/b2f197bc24ee12c31dc53b9beb78f0a1a05f018a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6f035ddf808f1cfd3fa876668ddbc0bea82fdbc6","is_merged_in_target_branch":false,"change_id":"I6c40f7f31058bc6e5e893743b6cdcd896d4850d1","change_number":524135,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d8d58a81565116a8fae796d55ab5ee6fe90d83c7":{"kind":"REWORK","_number":2,"created":"2017-12-14 09:17:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/38/524138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/38/524138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/38/524138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/38/524138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/38/524138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/38/524138/2"}}},"commit":{"parents":[{"commit":"16c50b691bbe67c267f0ebfde8a79a54f19ea0cd","subject":"Support configuration of trunk ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/16c50b691bbe67c267f0ebfde8a79a54f19ea0cd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2017-05-11 09:39:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2017-12-14 09:11:57.000000000","tz":0},"subject":"Add support for trunk ports on Dell switches","message":"Add support for trunk ports on Dell switches\n\nOn Dell Force10 switches, interfaces are added as tagged members of a\nVLAN using the following commands:\n\ninterface vlan \u003cid\u003e\ntagged \u003cif\u003e\n\nThey can be removed similarly:\n\ninterface vlan \u003cid\u003e\nno tagged \u003cif\u003e\n\nThis change uses these commands to support configuration of trunk ports\nfor all networks.\n\nChange-Id: I77f57ac1730eac2869fbdfeb112d1a9ea2b55fb1\nPartial-Bug: #1690115\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d8d58a81565116a8fae796d55ab5ee6fe90d83c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d8d58a81565116a8fae796d55ab5ee6fe90d83c7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"16c50b691bbe67c267f0ebfde8a79a54f19ea0cd","is_merged_in_target_branch":true,"change_id":"I6c40f7f31058bc6e5e893743b6cdcd896d4850d1","change_number":524135,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
