)]}'
{"id":"openstack%2Fnetworking-generic-switch~603920","triplet_id":"openstack%2Fnetworking-generic-switch~master~Ib7d68d1360e3cdbfdc2ff185a9760461e75ac8c4","project":"openstack/networking-generic-switch","branch":"master","hashtags":[],"change_id":"Ib7d68d1360e3cdbfdc2ff185a9760461e75ac8c4","subject":"Set config filename for neutron-metering","status":"MERGED","created":"2018-09-19 21:31:05.000000000","updated":"2018-09-21 13:34:49.000000000","submitted":"2018-09-21 13:34:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"603920-1537536889881-72bc25b4","meta_rev_id":"35f893bc0bd0e6fb5122c5587631e29f6195ec2c","_number":603920,"virtual_id_number":603920,"owner":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2018-09-21 13:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-09-21 11:52:47.000000000","_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},{"value":2,"date":"2018-09-21 11:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-09-21 11:52:47.000000000","_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-21 11:48:34.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2018-09-21 11:52:47.000000000","updated_by":{"_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},"reviewer":{"_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},"state":"REVIEWER"},{"updated":"2018-09-21 13:34:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17b00bb0c3f979d28a6f909e3b62ba9343cdaafc","author":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"date":"2018-09-19 21:31:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fae17a643d781c931a18669b56256f80ee82a7d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-20 04:09:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/603920/1/check/openstack-tox-pep8/9133e98/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/20/603920/1/check/openstack-tox-py27/314630d/ : SUCCESS in 5m 46s\n- openstack-tox-py35 http://logs.openstack.org/20/603920/1/check/openstack-tox-py35/8cc29b5/ : SUCCESS in 5m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/603920/1/check/build-openstack-sphinx-docs/88f2714/html/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/20/603920/1/check/openstack-tox-docs/f8a3b8a/html/ : SUCCESS in 5m 16s\n- networking-generic-switch-tempest-dsvm-dlm http://logs.openstack.org/20/603920/1/check/networking-generic-switch-tempest-dsvm-dlm/6c33b06/ : SUCCESS in 20m 31s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode http://logs.openstack.org/20/603920/1/check/ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode/aafdc58/ : SUCCESS in 59m 15s\n- ironic-grenade-dsvm-multinode-multitenant http://logs.openstack.org/20/603920/1/check/ironic-grenade-dsvm-multinode-multitenant/f38d59a/ : FAILURE in 1h 19m 56s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/603920/1/check/openstack-tox-lower-constraints/acfa309/ : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5158d51ec77d888c109260c511954c42ec0d856e","author":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"date":"2018-09-21 09:13:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8f7d6571eb14e1eb79268edb9b4095f9ae5c80b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 11:11:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/603920/2/check/openstack-tox-pep8/7ce2f78/ : SUCCESS in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/20/603920/2/check/openstack-tox-py27/45cd37b/ : SUCCESS in 6m 48s\n- openstack-tox-py35 http://logs.openstack.org/20/603920/2/check/openstack-tox-py35/afa294b/ : SUCCESS in 5m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/603920/2/check/build-openstack-sphinx-docs/751876f/html/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/20/603920/2/check/openstack-tox-docs/b9d0819/html/ : SUCCESS in 5m 40s\n- networking-generic-switch-tempest-dsvm-dlm http://logs.openstack.org/20/603920/2/check/networking-generic-switch-tempest-dsvm-dlm/b4db16b/ : SUCCESS in 23m 08s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode http://logs.openstack.org/20/603920/2/check/ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode/7465e8b/ : SUCCESS in 1h 17m 14s\n- ironic-grenade-dsvm-multinode-multitenant http://logs.openstack.org/20/603920/2/check/ironic-grenade-dsvm-multinode-multitenant/0975900/ : FAILURE in 1h 43m 13s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/603920/2/check/openstack-tox-lower-constraints/4d66973/ : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"9ae2186623c693758d0aceee35b543169e053d31","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-09-21 11:48:34.000000000","message":"Patch Set 2: Code-Review+2\n\nWe need to figure out why grenade does not set these, but it can wait","accounts_in_message":[],"_revision_number":2},{"id":"7aaff83fee624367c21d051eb248d525104edea9","author":{"_account_id":6637,"name":"Sam Betts","email":"sam@code-smash.net","username":"sambetts"},"date":"2018-09-21 11:52:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8a17dff81bc004fc9e572e39b594aef7ad8c6dc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 11:52:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c468b7c91c42ca04818246fe82a2c64a4e91c97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 13:34:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/603920/2/gate/openstack-tox-pep8/f4835a7/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/20/603920/2/gate/openstack-tox-py27/9ee9ec2/ : SUCCESS in 5m 49s\n- openstack-tox-py35 http://logs.openstack.org/20/603920/2/gate/openstack-tox-py35/764be96/ : SUCCESS in 6m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/603920/2/gate/build-openstack-sphinx-docs/c3b5090/html/ : SUCCESS in 5m 05s\n- openstack-tox-docs http://logs.openstack.org/20/603920/2/gate/openstack-tox-docs/c4774a6/html/ : SUCCESS in 5m 39s\n- networking-generic-switch-tempest-dsvm-dlm http://logs.openstack.org/20/603920/2/gate/networking-generic-switch-tempest-dsvm-dlm/37670cf/ : SUCCESS in 20m 32s\n- ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode http://logs.openstack.org/20/603920/2/gate/ironic-tempest-dsvm-ipa-wholedisk-agent_ipmitool-tinyipa-multinode/f48c5ab/ : SUCCESS in 1h 11m 11s\n- ironic-grenade-dsvm-multinode-multitenant http://logs.openstack.org/20/603920/2/gate/ironic-grenade-dsvm-multinode-multitenant/5631863/ : FAILURE in 1h 32m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/603920/2/gate/openstack-tox-lower-constraints/5278d0d/ : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":2},{"id":"35f893bc0bd0e6fb5122c5587631e29f6195ec2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-21 13:34:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2a653da30d09cf92c2ab9f9d2752d919176f6c1","revisions":{"29c17be4bba4723075337b888d3caba20a0ba686":{"kind":"REWORK","_number":1,"created":"2018-09-19 21:31:05.000000000","uploader":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"ref":"refs/changes/20/603920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/20/603920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/20/603920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/20/603920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/20/603920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/20/603920/1"}}},"commit":{"parents":[{"commit":"9fa0acce4a8816fe87611da1075d2da3ddbd637f","subject":"Install doc requirements in venv tox environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/9fa0acce4a8816fe87611da1075d2da3ddbd637f"}]}],"author":{"name":"Derek Higgins","email":"derekh@redhat.com","date":"2018-09-19 21:23:22.000000000","tz":60},"committer":{"name":"Derek Higgins","email":"derekh@redhat.com","date":"2018-09-19 21:24:14.000000000","tz":60},"subject":"Testing","message":"Testing\n\nChange-Id: Ib7d68d1360e3cdbfdc2ff185a9760461e75ac8c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/29c17be4bba4723075337b888d3caba20a0ba686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/29c17be4bba4723075337b888d3caba20a0ba686"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9fa0acce4a8816fe87611da1075d2da3ddbd637f","is_merged_in_target_branch":true,"change_id":"I6fd91e444fca4d4239aa67fae08cbd7a25a6ccf6","change_number":598995,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d2a653da30d09cf92c2ab9f9d2752d919176f6c1":{"kind":"REWORK","_number":2,"created":"2018-09-21 09:13:30.000000000","uploader":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"ref":"refs/changes/20/603920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/20/603920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/20/603920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/20/603920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/20/603920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/20/603920/2"}}},"commit":{"parents":[{"commit":"9fa0acce4a8816fe87611da1075d2da3ddbd637f","subject":"Install doc requirements in venv tox environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/9fa0acce4a8816fe87611da1075d2da3ddbd637f"}]}],"author":{"name":"Derek Higgins","email":"derekh@redhat.com","date":"2018-09-19 21:23:22.000000000","tz":60},"committer":{"name":"Derek Higgins","email":"derekh@redhat.com","date":"2018-09-21 09:12:13.000000000","tz":60},"subject":"Set config filename for neutron-metering","message":"Set config filename for neutron-metering\n\nRequired now for the grenade jobs as they have switched\nover to using this script since rocky has been released.\nTask: #26593\n\nStory: #2003808\nChange-Id: Ib7d68d1360e3cdbfdc2ff185a9760461e75ac8c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d2a653da30d09cf92c2ab9f9d2752d919176f6c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d2a653da30d09cf92c2ab9f9d2752d919176f6c1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9fa0acce4a8816fe87611da1075d2da3ddbd637f","is_merged_in_target_branch":true,"change_id":"I6fd91e444fca4d4239aa67fae08cbd7a25a6ccf6","change_number":598995,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
