)]}'
{"id":"openstack%2Fnetworking-generic-switch~692404","triplet_id":"openstack%2Fnetworking-generic-switch~master~Ib257f4e7710ed6615e68c7034ff4e223bbf677b2","project":"openstack/networking-generic-switch","branch":"master","topic":"story/2006789","hashtags":[],"change_id":"Ib257f4e7710ed6615e68c7034ff4e223bbf677b2","subject":"Create and delete networks in parallel","status":"ABANDONED","created":"2019-10-31 16:17:18.000000000","updated":"2026-03-02 14:07:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e00e094537e59c9d42cb634ddbc70d0487a83916","_number":692404,"virtual_id_number":692404,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-03 12:55:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 16:28:32.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-03-27 18:20:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8086b20b486f1699e0178d208617abb1a4884666","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 16:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9234c1b14171117f729c26f5fcecdc4eab195049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 18:02:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdda0d504a94598bce9d8fac7b38f2b : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18509794d2ac468c83b213468527ed4e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4550b71310a74c30a4c6b56a0f1550f7 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ef7a20549bb4c928c2ee9b76cc1e526 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a3bf0919104f5a85aa3bf06f134d3c : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c00719c921e7498699a20c813eeca3ae : SUCCESS in 3m 53s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/c799ade6a6cd4f13bc0df37f21c1ab2d : SUCCESS in 21m 01s\n- networking-generic-switch-tempest-dlm-python2 https://zuul.opendev.org/t/openstack/build/b763043120a544408932f09e90337259 : SUCCESS in 20m 58s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/1b717050d37b47618e6a87577ffd59f2 : SUCCESS in 1h 38m 58s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/046706a723c94ea5935e590072e5b61c : FAILURE in 1h 22m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eebbe2087b44d78a191a8bb74db6dda : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"773fabccf226f855e2cfb006d0dd344c313184e5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 16:24:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81e1a64cc5976d16fc8b6887631081bb4f625855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:11:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11fe4be1558c4adb9ff70fa79eeb2542 : FAILURE in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17d8c5853a30426286ed71ea06b61db4 : FAILURE in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2495400f9d0c4304b2049f7cf447aff4 : FAILURE in 9m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a73af24cd9b4847a3e2305bc594b897 : FAILURE in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e80d0fb4bfe4f3cb0d65ab4880df524 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7f7b1fd1d5949da8aea13bbd436e73b : SUCCESS in 4m 23s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/784fdab28b314f61a9c2551f095b339f : FAILURE in 46m 22s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/cd8a77d3753a417eb1b8b7a61abb0df4 : FAILURE in 2h 32m 15s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/759abe3b50f1440fa1318da37fbb148b : FAILURE in 1h 37m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c5db10b4201420cab003432b92a12db : FAILURE in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"09ada015305954dd7a5bb4ea485ae98ba53d1259","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 11:25:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c20f720955672a29812465108eca607624fb7891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 14:26:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361febd3a85247c1965b2394e1a19122 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d92dccd1d4140ffb78d8303898e4262 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc2da7b8daf14d3793b4519b122c7325 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67a527e6ab1f46958d6dc439e0c8956a : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d3141d1abd49ca9520f73cfac21dcb : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c2ebcd08f924d10b93b72c4a90f6fe4 : SUCCESS in 3m 11s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/9c5f45cff0104e049d162a07425dc9f7 : FAILURE in 42m 53s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d54c1a4b36e447b19a2181b824943d6c : FAILURE in 2h 55m 50s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/0239b400a6e74c4f84f1468f9c9b0abe : FAILURE in 1h 49m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a16455e2d9e40eb98f33f696055ade3 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":3},{"id":"772e0f58f11736f4dfb862afccf54be9e1bb5410","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-03-27 16:28:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e3cc93f0a4652bb3c83c6a9e96722c45ca25cc5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 18:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/950f2fb89d88498999203379527c3286 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac504f2083f949cba5d6ccccb5067183 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/345f3ffa03e74acdb00ad28fb9532a71 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/637fbdfb5b2d4dd1b3786c6ef8f4667d : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6773c970414b54a3fe0cdc7822211e : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71b2fa48db68403696ce7bfb8e87cfb1 : SUCCESS in 4m 32s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/31cb15c9d3554a0c9e63ed7cc4d54f4b : SUCCESS in 21m 07s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/2f0548262546483380d614c299802899 : FAILURE in 40m 17s (non-voting)\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/ae45493059e848c28460349b6ba511e9 : FAILURE in 1h 45m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d051684b67454f13bc6bddec619801cd : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":3},{"id":"bce1574a7ff341833cc9e0dff96bc547dc45581b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 10:44:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f1b9cbf1ee0770f3b8f3d91d1a1787abf3719a62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 12:55:02.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e6f9a1249140e5a6d5457db7d5c28d : FAILURE in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f209a35e6d534049b995d798042d8b04 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aacd7820edc342bb964c5148a9b1ab8c : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abf6597a2fc44b228b04ed0dda69c26a : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f21398ac84e4bbb9cd5f95a0c4a5cc0 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d5ad73ddf41495cb4219f03e5003ec8 : SUCCESS in 3m 17s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/d57057e52522404aa9373ff3025f8fb9 : SUCCESS in 24m 18s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d70e4971094c4bd586e56d163b39ecc8 : FAILURE in 2h 03m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e00e094537e59c9d42cb634ddbc70d0487a83916","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-03-02 14:07:02.000000000","message":"Abandoned\n\nobsolete","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"03671046105ca89c70b6d981aa0e88507492039f","revisions":{"9431f315bc679ad59e813d4312265b015fc3f0ca":{"kind":"REWORK","_number":1,"created":"2019-10-31 16:17:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/692404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/04/692404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/1"}}},"commit":{"parents":[{"commit":"dd7e84204b941a9f90666be029b44f0a4ef8330b","subject":"Merge \"Update the constraints url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/dd7e84204b941a9f90666be029b44f0a4ef8330b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-31 15:04:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-31 16:16:58.000000000","tz":0},"subject":"Create and delete networks in parallel","message":"Create and delete networks in parallel\n\nCurrently, networks are created and deleted on devices serially. In\nsystems with many devices, this can lead to these requests taking a long\ntime to complete.\n\nThis change executes device configuration in parallel for network\ncreation and deletion, which should improve this from O(N) to O(1). It\nis not necessary to execute other operations in parallel, since they\nonly affect a single device. If we ever support multi-chassis LAG\nconfiguration, we could configure it in parallel.\n\nFailure handling should be as before this change, with the exception\nthat network creation will no longer stop after the first failed device,\nsince all are configured in parallel.\n\nChange-Id: Ib257f4e7710ed6615e68c7034ff4e223bbf677b2\nStory: 2006789\nTask: 37328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/9431f315bc679ad59e813d4312265b015fc3f0ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/9431f315bc679ad59e813d4312265b015fc3f0ca"}]},"branch":"refs/heads/master"},"ee845fd05d727d714ea2ea2257afb2bec07a5d46":{"kind":"REWORK","_number":2,"created":"2020-03-12 16:24:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/692404/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/04/692404/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/2"}}},"commit":{"parents":[{"commit":"7907f2cb88d0ae26c76d473d6a5e683fcd8e0698","subject":"Merge \"Allow network name format to be configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7907f2cb88d0ae26c76d473d6a5e683fcd8e0698"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-31 15:04:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 16:24:46.000000000","tz":0},"subject":"Create and delete networks in parallel","message":"Create and delete networks in parallel\n\nCurrently, networks are created and deleted on devices serially. In\nsystems with many devices, this can lead to these requests taking a long\ntime to complete.\n\nThis change executes device configuration in parallel for network\ncreation and deletion, which should improve this from O(N) to O(1). It\nis not necessary to execute other operations in parallel, since they\nonly affect a single device. If we ever support multi-chassis LAG\nconfiguration, we could configure it in parallel.\n\nFailure handling should be as before this change, with the exception\nthat network creation will no longer stop after the first failed device,\nsince all are configured in parallel.\n\nChange-Id: Ib257f4e7710ed6615e68c7034ff4e223bbf677b2\nStory: 2006789\nTask: 37328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ee845fd05d727d714ea2ea2257afb2bec07a5d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ee845fd05d727d714ea2ea2257afb2bec07a5d46"}]},"branch":"refs/heads/master"},"e6cf1591b91e93d0ed354123d6d38f5614eec911":{"kind":"REWORK","_number":3,"created":"2020-03-13 11:25:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/692404/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/04/692404/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/3"}}},"commit":{"parents":[{"commit":"7907f2cb88d0ae26c76d473d6a5e683fcd8e0698","subject":"Merge \"Allow network name format to be configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7907f2cb88d0ae26c76d473d6a5e683fcd8e0698"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-31 15:04:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-13 11:24:59.000000000","tz":0},"subject":"Create and delete networks in parallel","message":"Create and delete networks in parallel\n\nCurrently, networks are created and deleted on devices serially. In\nsystems with many devices, this can lead to these requests taking a long\ntime to complete.\n\nThis change executes device configuration in parallel for network\ncreation and deletion, which should improve this from O(N) to O(1). It\nis not necessary to execute other operations in parallel, since they\nonly affect a single device. If we ever support multi-chassis LAG\nconfiguration, we could configure it in parallel.\n\nFailure handling should be as before this change, with the exception\nthat network creation will no longer stop after the first failed device,\nsince all are configured in parallel.\n\nChange-Id: Ib257f4e7710ed6615e68c7034ff4e223bbf677b2\nStory: 2006789\nTask: 37328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/e6cf1591b91e93d0ed354123d6d38f5614eec911"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/e6cf1591b91e93d0ed354123d6d38f5614eec911"}]},"branch":"refs/heads/master"},"03671046105ca89c70b6d981aa0e88507492039f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-03 10:44:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/692404/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/04/692404/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/04/692404/4"}}},"commit":{"parents":[{"commit":"6b1fca9439b8df5ad26b6ac9374733b62c5ddee9","subject":"Update minversion of tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/6b1fca9439b8df5ad26b6ac9374733b62c5ddee9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-31 15:04:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 10:44:39.000000000","tz":0},"subject":"Create and delete networks in parallel","message":"Create and delete networks in parallel\n\nCurrently, networks are created and deleted on devices serially. In\nsystems with many devices, this can lead to these requests taking a long\ntime to complete.\n\nThis change executes device configuration in parallel for network\ncreation and deletion, which should improve this from O(N) to O(1). It\nis not necessary to execute other operations in parallel, since they\nonly affect a single device. If we ever support multi-chassis LAG\nconfiguration, we could configure it in parallel.\n\nFailure handling should be as before this change, with the exception\nthat network creation will no longer stop after the first failed device,\nsince all are configured in parallel.\n\nChange-Id: Ib257f4e7710ed6615e68c7034ff4e223bbf677b2\nStory: 2006789\nTask: 37328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/03671046105ca89c70b6d981aa0e88507492039f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/03671046105ca89c70b6d981aa0e88507492039f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
