)]}'
{"id":"openstack%2Fnetworking-generic-switch~743269","triplet_id":"openstack%2Fnetworking-generic-switch~master~I4d96ea3f3c3d55ff7d742b9e08aa75ce990eee9a","project":"openstack/networking-generic-switch","branch":"master","topic":"jg-cumulus","hashtags":[],"change_id":"I4d96ea3f3c3d55ff7d742b9e08aa75ce990eee9a","subject":"Add support for Cumulus 4.x NCLU driver","status":"MERGED","created":"2020-07-27 15:56:17.000000000","updated":"2020-08-03 12:09:52.000000000","submitted":"2020-08-03 12:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743269-1596456520797-c0251aed","meta_rev_id":"a2b00e4e8d53d81243aaf7ec7ec7e593417ab3fa","_number":743269,"virtual_id_number":743269,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 12:08:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-08-03 09:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-30 10:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-08-03 09:47:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 10:06:12.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-03 09:47:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-03 12:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec28d20869b9c6fce870ba4e3baf7e669c3d8ce9","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-27 15:56:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93e274280418055eb0d6355483aab84f14daead5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 17:54:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd7ce6f4789a400f954884622f98f828 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f34775c65b4747b596e8a06b85bf08 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9009a734094d43e9ad0c8517cf9fae28 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d848345c4fd84324963315309d69935a : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2716ef53b3534edea28879055fe5b2fb : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bfd086749c945d68482de52e8aacdcb : SUCCESS in 3m 57s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/834f4af92a8648aeb1568e86ba6ad751 : SUCCESS in 25m 36s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/49e297ca5fe84db987e0be0e3ed9d6f0 : SUCCESS in 1h 43m 19s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/a367c67446a04630ba01e6080920fa4e : FAILURE in 15m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e60a9247d6249d191d501c7995c0cbc : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0a7e8c3445958d562639f841e92c719846f6907a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-30 10:06:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c54a248af053df6a5542be5675243016b963485","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-03 09:47:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nCould you follow up with docs?","accounts_in_message":[],"_revision_number":1},{"id":"c035e6d83bcccf0b5d3c0cb9a47473a104b27c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 09:47:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9bbdc5836ec99fbc979ce0346835dd0088cf8b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 12:08:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"440a0f08eeaa33cb9baf0d738d76d741f894aacc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 12:08:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e6ebb5727774afb9025729d50305e36 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27beea345e574887a1005b042e25ef4a : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dd30270d31a497881580a3cbf20fef8 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52c53ddba18d403cafdc57410a3639fd : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5b7c95c0d8a42c097ac23dc5cec4d20 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb68276988b04bde914a3f5109386217 : SUCCESS in 3m 17s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/9c6f42542bb5469a8b4592402e57f6f1 : SUCCESS in 24m 37s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/2a9cc9f3dc2245b3a5da0badd5c8fffa : SUCCESS in 2h 16m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33af4eae981c4a64b12f61fc5e3ddb12 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b9fab79d061a9f68363569c0d59a581150bc3e9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 12:09:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f769f3f49049cbaa555fb7ccd1107e : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4c36821122549c29fe78787edb5d10e : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7971c9c6b80f650b22c7b1284188dd92f4b1b0f7","revisions":{"7971c9c6b80f650b22c7b1284188dd92f4b1b0f7":{"kind":"REWORK","_number":1,"created":"2020-07-27 15:56:17.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/69/743269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/69/743269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/69/743269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/69/743269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/69/743269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/69/743269/1"}}},"commit":{"parents":[{"commit":"efeba57e7b5a53dcf4b60d66f3c0c7748fb2dfd5","subject":"Set min version of tox to 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/efeba57e7b5a53dcf4b60d66f3c0c7748fb2dfd5"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 15:42:07.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 15:56:07.000000000","tz":60},"subject":"Add support for Cumulus 4.x NCLU driver","message":"Add support for Cumulus 4.x NCLU driver\n\nThis adds a new driver for Cumulus switches that support using NCLU.\nIt actually uses the netmiko linux driver, given we are really just\naccessing a debian user space, and there is currently no cumulus\nspecific driver in netmiko.\n\nIt borrows quite heavily from the support in networking-ansible that\nuses these roles to configure cumulus switches:\nhttps://github.com/ansible-network/network-runner\n\nHowever, this driver adopts the more typical networking-generic-switch\nconfiguration operations, where existing port configuration is retained\nas the port cycles through various VLANs. In a similar way, we make\ndisabling a port optional, and support users specifying a default VLAN,\nsuch as the ironic inspection network.\n\nChange-Id: I4d96ea3f3c3d55ff7d742b9e08aa75ce990eee9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7971c9c6b80f650b22c7b1284188dd92f4b1b0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7971c9c6b80f650b22c7b1284188dd92f4b1b0f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
