)]}'
{"id":"openstack%2Fnetworking-generic-switch~743283","triplet_id":"openstack%2Fnetworking-generic-switch~master~I8c458bbc94df5630cfede5434bcdbe527988059c","project":"openstack/networking-generic-switch","branch":"master","topic":"batching","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-03-02 15:56:21.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-02 15:56:21.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2023-03-02 15:56:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c458bbc94df5630cfede5434bcdbe527988059c","subject":"Support batching up commands","status":"MERGED","created":"2020-07-27 17:03:07.000000000","updated":"2023-09-20 18:45:46.000000000","submitted":"2023-03-02 15:56:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743283-batching","meta_rev_id":"ad7783ebc3054afc5842091577828066d994d6c4","_number":743283,"virtual_id_number":743283,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 15:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2023-03-01 09:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2023-03-01 20:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2023-03-01 20:31:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"}],"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 11:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 19:09:12.000000000","updated_by":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"CC"},{"updated":"2023-02-08 09:57:28.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"CC"},{"updated":"2023-02-21 14:51:27.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-02-23 12:37:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-03-01 09:42:09.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2023-03-01 20:31:41.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"8f326c9e1ad0aa9ccef720f3bfb916e7e6321b99","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-27 17:03:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14bb7669a429ea2f1e7ad7898940bd6afd6cbbdd","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-27 17:03:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7ed6194189241444f5d59b1b2f1be38e6721c768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 19:03:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6104a9810f94aebb54cd2be44f61a5a : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96cf243eda32414ba55b464dcd40535d : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca0f5a2c929d45c5a2d89f3071050c96 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc53deb5e503474e9a96c705727121dd : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01207724156d4e3f81aed5f08643e5b6 : SUCCESS in 8m 46s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/678202ca432a42b8bd4c5ff9229544de : SUCCESS in 30m 06s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/6001e6e5c1954906883887aa57e9bac1 : SUCCESS in 1h 50m 25s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/14eaf500acee43f9b2fa0a20c942cbde : FAILURE in 13m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1f66c0ae0d64822893fc004de46a98d : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"274bb099ea1eb134725181b2df6b36b2260e37cc","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-28 10:33:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1982fb6228828d0bff4d2911ceca2e65c78e6eec","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-28 10:44:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"196ea57f7bb172911c29734a389bbd6705f75352","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-28 11:29:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e70e89ace13bf894144b5bd14a06af7825502ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 12:55:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a518beb219b42dca4a008ab7267111d : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b32bdafc79f49689d2e521594f10e5a : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a383e09a81c49dab0e976bad2059f38 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/836e1a55324e4323bcd6fadde3b44f6d : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4cda5d17644a5e8a2d3fe4092a112f : SUCCESS in 6m 19s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/1ecc5677ef4441519218339a7a81c0c1 : SUCCESS in 25m 15s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/f655bec8fda04fe798a089c8859f71f9 : SUCCESS in 1h 22m 56s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/0f93bfea30b947a4be509daf70c127f6 : FAILURE in 8m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9551042ff26d420d9f2daf1078643399 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":4},{"id":"901fd8589363aa8d428440903eb8d7b2c0f9ba88","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-28 13:37:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e747d6bf0f2cde714174fd2b9ece6102a36466dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 15:07:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9aa52981ef53465e970d7cbfd268c631 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8aba1b28b744f879ef9b8025834de42 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88e551a826564af7b37faa1219786398 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9b1bfa408f7401980ad2450d70bfade : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5a4ea0ea704138b500fa0cfd1feb3d : SUCCESS in 7m 51s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/a50be03f529f4ca5ac83a98cc6e46540 : SUCCESS in 20m 43s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/0cb25ee775d144f79d24c2fde2f55266 : SUCCESS in 1h 25m 39s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/47666ba545454b3f8b3e7cff5f4a1ef7 : FAILURE in 11m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/727083361fd941a28b2275b12f6e75c1 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":5},{"id":"598c765295a5df0ba01d5c42e9499fdaa35de8e6","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-29 10:04:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"eaaa52abcc16b8b5428f9f80e7ea7bbd8a341d99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:39:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3bee4595b0714abd8a7bbd143c51505d : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f9c6d6044c84071bbbfbf91d920ad6f : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e4800a24f7c478ebc75e3cc48a14d02 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a6b5adee2a94663922a827a01bf9d1c : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d04fcecac80437ea703b874b9fcf5ec : SUCCESS in 6m 38s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/039b65757c0b4bfa8b561d17e9047d15 : SUCCESS in 18m 33s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/8b868fff897e4bd0b14f6a18554de963 : SUCCESS in 1h 25m 15s\n- ironic-grenade-dsvm-multinode-multitenant https://zuul.opendev.org/t/openstack/build/2b2c7da2eb8e4f149c84d813341bee5a : FAILURE in 8m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10a702d8ebe34613999f3ce916a8bec3 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":6},{"id":"bd63b83449722fd0e2be9b0977c2260622ca6807","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-07-30 08:59:26.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"24f228ca9c9031dbed3e53d56a3666c3454b3932","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2021-01-04 19:09:12.000000000","message":"Patch Set 6:\n\nThis looks very promising, and I suspect can improve performance quite a bit. Do you have any numbers to share so far?\n\nI was wondering, is this something that tooz could help with? https://docs.openstack.org/tooz/latest/user/tutorial/lock.html -- right now this requires the user deploy etcd; with tooz I think it opens up several other options for how to do the locking. It may also reduce the amount of business logic needed around the locking mechanism.\n\nCool!","accounts_in_message":[],"_revision_number":6},{"id":"b12dda2026552a4ae7dc95374c2df59a8866a20a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-08 09:51:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"567059d33a7e00b663cd2ed1fd6ed5a2db7d8d54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-08 10:30:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"abe269858c88096dd197c3f1da9bd57f5eae41be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 12:01:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29717ed199cd4cd593a016b782c85e08\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e07cf8557c69490f8269020b7c91c21f : SUCCESS in 1m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe365b5725e4446abcba55941727cf8e : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb04edc60a614661ae8c1f3916a087c9 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b17d95671ce4347bf4c91dadc40cae5 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfcd8cc4666141189199b38b0dca9661 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4e039468011465287e943cca35f6a52 : SUCCESS in 2m 49s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/4b82710dc70b4614b2f30dc22bcb84dc : SUCCESS in 15m 49s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/ff6d2b82cd4241858c6f0c635ec8eb17 : SUCCESS in 1h 26m 31s","accounts_in_message":[],"_revision_number":8},{"id":"767b87721b86cc1455411da12aaa7712d7b8143e","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-02-21 14:51:27.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b640d84c01fdb1906582f757184602a1e30be00f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-23 12:36:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"81191cd0bb0c0615ad753631588baf2f9575093c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-23 12:37:00.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b55a2fd69bf9f87e746e115932e54e24375d8c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-23 12:44:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d4388e40da95b44009a51c5e2f7885bf9ab51da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 13:37:19.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f83ffa7869d04ef4a027382e3643d4b3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b6451b5bc9544472ab22c6842a2ef65b : SUCCESS in 1m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe48448a9bbf4a609b2e80c0e07f0ccb : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6776f4cbe6014b7a88fe6a5139d41192 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f13faff2d2684d568847ba26dfa7e112 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a775654bfb25442c9ad5e8e5624790aa : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4097cb6aba243dca854678c7371fb0e : SUCCESS in 2m 53s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/3fe7a59bb196410784a207f9f03775de : SUCCESS in 25m 13s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/69d7ea4c6c7f46bbb993618292159883 : FAILURE in 48m 46s","accounts_in_message":[],"_revision_number":10},{"id":"8f54d96b7f3a930615aa50b26eb344eb3d9744b7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-23 20:03:58.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e1c5be18181dc99a16c9dd10e79ed96a4f3bc0c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:43:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01dbac07003f4d8eb11a1d8b02996f71\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4642f356ad514baa99a8f4cde2ab4c9a : SUCCESS in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3db90704848482a995ec2da06908b25 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/127e1932690d49849881d93e5ad3371b : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85a1c4fa910b45b3b99107986dd33005 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c30c3e7f91542d594d25b4cff233510 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/371a611ba2ab4794be7cc3e7e5a66b9e : SUCCESS in 3m 33s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/8894c8b4e36d471782a0c51ed59aff70 : SUCCESS in 14m 16s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d9133910675b4ab28b8ac86bc9d413b1 : SUCCESS in 1h 34m 09s","accounts_in_message":[],"_revision_number":10},{"id":"244fa9d044666adc863ba1c6fc51bf3bc7d662cf","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-02-27 14:12:53.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0253f1a4b00f4704ac9c6dab8b368e0832d2d37f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-27 15:51:43.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"f7ff98cced2a6f4ae6e343d97961f8b6f720152b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-27 15:52:11.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"28600e0ab277912034fca5d5b07c3c67b668818c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 17:50:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/962ebad979774e178702c587f33683fd\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3f0a230a80d0472282cf0a4d77d38922 : SUCCESS in 1m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48c318ca10ca45c4853ff7c44817f841 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f83380b0ef644c868e46bb72f09e2a21 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b6ccac0ee774f7da790b3222f2ebf59 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bcc2563d9fa4d288c4096e81b4d24bd : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a81af6082e54144b024b2b3dbc515d4 : SUCCESS in 3m 21s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/14e6404280684b5fa3ccc436116e6676 : SUCCESS in 25m 40s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/cf28f957133d41969d7f088631d3f277 : SUCCESS in 1h 53m 54s","accounts_in_message":[],"_revision_number":11},{"id":"fa4a3792f8e14fe3ab58505a7b9917f3e4a05781","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-01 09:42:09.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b3c395f5b666ee49786042876f192b77e095c5a0","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-03-01 20:31:41.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"b92c4395da45e5cae48764c69e9659d1aca70afa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 20:32:15.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"f6115062390fb1f93a71f779abd2fe40f46fe018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 23:41:15.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6505f49646a941458ea68f8d582b9943\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/926b2d0f51e44095ad3a4cc15fb15db8 : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d47bf7d81ae4863a4a21f68c05b6a33 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77a3fc4855054681a370c4500ce5ca45 : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6132f77d6d44f0cb462e3e51b7ecfdd : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18f2beaca8044f7388349cf63a5e8f1e : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a68c2c707714ce29c7d479e19a39208 : SUCCESS in 2m 47s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/c58d5d49d0414521ad066b518873683a : SUCCESS in 38m 17s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/d1c0d94a52664dd4ae6272196f6d76f5 : FAILURE in 2h 57m 04s","accounts_in_message":[],"_revision_number":11},{"id":"af847be8fb9b2e7f3b0e3758f9d793f0dc3c60ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-02 09:07:09.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5a1d1a23611a296e88744ccee6960d7f85e8ab2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:35:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6f4caded6f045c99337bba4855c2386\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e65fc87dc6f4e178644858eea7e1bbb : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98163f9f900f484990c8672ed473939b : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f0fb04ba9646e8ae0ccfec6cf960ee : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e0574e5af74484d9c84c0c63cbf53af : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edcd2a13d0c843aa82796455f371cd2b : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f48ccd89e1449338068359d20e6200b : SUCCESS in 2m 57s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/6f37c9b9624244e592fa748c84878940 : SUCCESS in 26m 24s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/cbd21e1fd0ec4b548198fec57d5c4468 : SUCCESS in 1h 22m 33s","accounts_in_message":[],"_revision_number":11},{"id":"3d2c35fbd1363cb95be2d24743877d3fbb41e5a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:35:39.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"2cc5b4cfc734cc48838f13bb01ab6f6730f7dd36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 12:14:03.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/688780f7bf274f9aaeec0e495616eeff\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a419ecfb63f46bfbb52165a26625b7c : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb22e21b2f2f44778d413f60037fbae3 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88f1f59bce2f4637aa4c9e34633c571f : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70f7784571b94d7d90c3bf964470aedd : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39fc4c9238824684bb2c2b19f41f97ef : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/815316ad85f74004bcef1c21af7236c3 : SUCCESS in 2m 51s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/b06a18bb0a4e470b9473bfb5317ae171 : SUCCESS in 13m 14s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/59c1fd27c96e4557b62805caf0194100 : FAILURE in 1h 31m 14s","accounts_in_message":[],"_revision_number":11},{"id":"b29ab2620f0a6ccb0115a966a374be4ad9c680fc","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-03-02 12:23:07.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d095ecbf94c38d27548b2862b76fb595d1acbcef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:12:21.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ade180070a4eb388554ed3e088c922\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fbc6629d11d7456c8f6ab82ff9c71367 : SUCCESS in 1m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502d696fa84c42fa9ac8b1d2da79c066 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d7f024104074c5c9bb6a47864308dc9 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e17042b9c7204d058c708a8d609bf14d : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2424edb7ad354e419327b0ae36d39b64 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e26bf510741844f999162a16a86ec7f1 : SUCCESS in 2m 42s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/61e08f66e4524470b61d6fb386e86bbe : SUCCESS in 27m 51s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/84a5c19e78064772b1880b83049f1814 : SUCCESS in 1h 42m 05s","accounts_in_message":[],"_revision_number":11},{"id":"a6eecefc6d078f5f29fe0e3fe2a5eabb679df649","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:17:17.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"3a31acb156131d6c779b00638399b67e047b750e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:56:20.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/967044a9815f4878b0c76d04ed2b81cd\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/70614ccf631c4c7a9f7bb5b5173cfce0 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9f5c59a573420c8ff88b20cf3db0f2 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/063687839da143c781bdf2be1fec89ff : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53fa39e48fd24353b0885ce0618dfeef : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de15956c98b54df2977b84350bdd5c8c : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93495fa1ffb34b5ab19ebfa855627028 : SUCCESS in 3m 08s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/ce0c34f9fe60496aadb440eab9bf5597 : SUCCESS in 23m 47s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/4251821e69f94e65b3aef2f247612b7a : SUCCESS in 1h 33m 06s","accounts_in_message":[],"_revision_number":11},{"id":"a267f1b0c472887f6d3b836701b2fc969d85ccf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:56:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"9b54da59f3443b11059bbaf0440cec8912cd6acf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:57:25.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d651b3f7e0b8487baf8cfd62fec862c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9927a74a2dd44819be41665e82ff9f5f : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40731a787f1542ad9f1b6de195b51721 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":11},{"id":"b5a377c8d8d2f39685cac2a998794e67da83c080","tag":"autogenerated:gerrit:revert","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-20 18:42:02.000000000","message":"Created a revert of this change as I4d36d2fc9e1077b0265f5ca0dc0fea922e4371e3","accounts_in_message":[],"_revision_number":11},{"id":"ad7783ebc3054afc5842091577828066d994d6c4","tag":"autogenerated:gerrit:revert","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-20 18:45:46.000000000","message":"Created a revert of this change as I45cb52edc7d6852d5263d4b5c3c468597087572e","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"45b237b754cd2d9c6f9c588e759fa74da5b9eacc","revisions":{"d3f70b0078b524b1cfefed6897cc9a798f0e8672":{"kind":"REWORK","_number":1,"created":"2020-07-27 17:03:07.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/83/743283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/1"}}},"commit":{"parents":[{"commit":"7971c9c6b80f650b22c7b1284188dd92f4b1b0f7","subject":"Add support for Cumulus 4.x NCLU driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7971c9c6b80f650b22c7b1284188dd92f4b1b0f7"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"subject":"WIP: Attempt to batch up cmds","message":"WIP: Attempt to batch up cmds\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look at batch up commands to send to the switch together, then\nwaits for them all to complete before returning the result. It does this\nby using etcd keys as a queueing system. We pull off the queue using the\nversion at which the keys were added, giving a FIFO style queue.\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d3f70b0078b524b1cfefed6897cc9a798f0e8672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d3f70b0078b524b1cfefed6897cc9a798f0e8672"}]},"branch":"refs/heads/master"},"08750e6f460691a4d23a8d931cfb5a26ce2ebad9":{"kind":"REWORK","_number":2,"created":"2020-07-28 10:33:24.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/83/743283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/2"}}},"commit":{"parents":[{"commit":"7971c9c6b80f650b22c7b1284188dd92f4b1b0f7","subject":"Add support for Cumulus 4.x NCLU driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7971c9c6b80f650b22c7b1284188dd92f4b1b0f7"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-28 10:33:12.000000000","tz":60},"subject":"WIP: Attempt to batch up cmds","message":"WIP: Attempt to batch up cmds\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look at batch up commands to send to the switch together, then\nwaits for them all to complete before returning the result. It does this\nby using etcd keys as a queueing system. We pull off the queue using the\nversion at which the keys were added, giving a FIFO style queue.\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/08750e6f460691a4d23a8d931cfb5a26ce2ebad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/08750e6f460691a4d23a8d931cfb5a26ce2ebad9"}]},"branch":"refs/heads/master"},"111aa18ce13d6eddaad852cd57ecf3c06ec7430d":{"kind":"REWORK","_number":3,"created":"2020-07-28 10:44:41.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/83/743283/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/3"}}},"commit":{"parents":[{"commit":"7971c9c6b80f650b22c7b1284188dd92f4b1b0f7","subject":"Add support for Cumulus 4.x NCLU driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7971c9c6b80f650b22c7b1284188dd92f4b1b0f7"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-28 10:44:30.000000000","tz":60},"subject":"WIP: Attempt to batch up cmds","message":"WIP: Attempt to batch up cmds\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look at batch up commands to send to the switch together, then\nwaits for them all to complete before returning the result. It does this\nby using etcd keys as a queueing system. We pull off the queue using the\nversion at which the keys were added, giving a FIFO style queue.\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/111aa18ce13d6eddaad852cd57ecf3c06ec7430d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/111aa18ce13d6eddaad852cd57ecf3c06ec7430d"}]},"branch":"refs/heads/master"},"5ef9f272b495a40b54ca66418e396a310dc1e133":{"kind":"REWORK","_number":4,"created":"2020-07-28 11:29:00.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/83/743283/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/4"}}},"commit":{"parents":[{"commit":"7971c9c6b80f650b22c7b1284188dd92f4b1b0f7","subject":"Add support for Cumulus 4.x NCLU driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7971c9c6b80f650b22c7b1284188dd92f4b1b0f7"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-28 11:28:49.000000000","tz":60},"subject":"WIP: Attempt to batch up cmds","message":"WIP: Attempt to batch up cmds\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look at batch up commands to send to the switch together, then\nwaits for them all to complete before returning the result. It does this\nby using etcd keys as a queueing system. We pull off the queue using the\nversion at which the keys were added, giving a FIFO style queue.\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/5ef9f272b495a40b54ca66418e396a310dc1e133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/5ef9f272b495a40b54ca66418e396a310dc1e133"}]},"branch":"refs/heads/master"},"ff5a84bb8d6fb950c08ec25995c2543d4539a9f9":{"kind":"REWORK","_number":5,"created":"2020-07-28 13:37:13.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/83/743283/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/5"}}},"commit":{"parents":[{"commit":"7971c9c6b80f650b22c7b1284188dd92f4b1b0f7","subject":"Add support for Cumulus 4.x NCLU driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7971c9c6b80f650b22c7b1284188dd92f4b1b0f7"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-28 13:32:18.000000000","tz":60},"subject":"WIP: Attempt to batch up cmds","message":"WIP: Attempt to batch up cmds\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look to batch up commands to send to the switch together, then\nwait for them all to complete before returning the result. It does this\nby using etcd keys as a queueing system. We pull off the queue using the\nversion at which the keys were added, giving a FIFO style queue.\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ff5a84bb8d6fb950c08ec25995c2543d4539a9f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ff5a84bb8d6fb950c08ec25995c2543d4539a9f9"}]},"branch":"refs/heads/master"},"95ad991ca824fd3a9f08ab4f266e65d33542f9a8":{"kind":"REWORK","_number":6,"created":"2020-07-29 10:04:07.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/83/743283/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/6"}}},"commit":{"parents":[{"commit":"7971c9c6b80f650b22c7b1284188dd92f4b1b0f7","subject":"Add support for Cumulus 4.x NCLU driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7971c9c6b80f650b22c7b1284188dd92f4b1b0f7"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-29 10:03:52.000000000","tz":60},"subject":"WIP: Attempt to batch up cmds","message":"WIP: Attempt to batch up cmds\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look to batch up commands to send to the switch together, then\nwait for them all to complete before returning the result. It does this\nby using etcd keys as a queueing system. We pull off the queue using the\nversion at which the keys were added, giving a FIFO style queue.\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/95ad991ca824fd3a9f08ab4f266e65d33542f9a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/95ad991ca824fd3a9f08ab4f266e65d33542f9a8"}]},"branch":"refs/heads/master"},"4ac05195e5a22f1172eedf878372e9432352dae3":{"kind":"REWORK","_number":7,"created":"2023-02-08 09:51:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/743283/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/7"}}},"commit":{"parents":[{"commit":"39cb7f0455badb63fb35441e31973cb6d05173e2","subject":"Document ngs_trunk_ports and ngs_physical_networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/39cb7f0455badb63fb35441e31973cb6d05173e2"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-02-08 09:51:23.000000000","tz":0},"subject":"WIP: Attempt to batch up cmds","message":"WIP: Attempt to batch up cmds\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look to batch up commands to send to the switch together, then\nwait for them all to complete before returning the result. It does this\nby using etcd keys as a queueing system. We pull off the queue using the\nversion at which the keys were added, giving a FIFO style queue.\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/4ac05195e5a22f1172eedf878372e9432352dae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/4ac05195e5a22f1172eedf878372e9432352dae3"}]},"branch":"refs/heads/master"},"a915d2b39273c8c65eee4346d6ac4552ef170fd6":{"kind":"REWORK","_number":8,"created":"2023-02-08 10:30:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/743283/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/8"}}},"commit":{"parents":[{"commit":"39cb7f0455badb63fb35441e31973cb6d05173e2","subject":"Document ngs_trunk_ports and ngs_physical_networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/39cb7f0455badb63fb35441e31973cb6d05173e2"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-02-08 10:30:30.000000000","tz":0},"subject":"Support batching up commands","message":"Support batching up commands\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look to batch up commands to send to the switch together using a\nsingle connection. The results of each port\u0027s commands are returned when\navailable.\n\nThis is implemented using etcd as a queueing system. Commands are added\nto an input key, then a worker thread processes the available commands\nfor a particular switch device. We pull off the queue using the version\nat which the keys were added, giving a FIFO style queue. The result of\neach command set are added to an output key, which the original request\nthread is watching. Distributed locks are used to serialise the\nprocessing of commands for each switch device.\n\nVarious neat etcd features are used here to alleviate some of the issues\nof distributed task coordination, including transactions, leases,\nwatches, historical key/value tracking, etc.\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/a915d2b39273c8c65eee4346d6ac4552ef170fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/a915d2b39273c8c65eee4346d6ac4552ef170fd6"}]},"branch":"refs/heads/master"},"ab4d37c2fa6ccd04beee276a02731d35046bfdca":{"kind":"REWORK","_number":9,"created":"2023-02-23 12:36:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/743283/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/9"}}},"commit":{"parents":[{"commit":"0c7f61bb39ded112d62f51894876398f86a5425a","subject":"Add ngs_ssh_disabled_algorithms setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/0c7f61bb39ded112d62f51894876398f86a5425a"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-02-23 12:35:52.000000000","tz":0},"subject":"Support batching up commands","message":"Support batching up commands\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look to batch up commands to send to the switch together using a\nsingle connection. The results of each port\u0027s commands are returned when\navailable.\n\nThis is implemented using etcd as a queueing system. Commands are added\nto an input key, then a worker thread processes the available commands\nfor a particular switch device. We pull off the queue using the version\nat which the keys were added, giving a FIFO style queue. The result of\neach command set are added to an output key, which the original request\nthread is watching. Distributed locks are used to serialise the\nprocessing of commands for each switch device.\n\nVarious neat etcd features are used here to alleviate some of the issues\nof distributed task coordination, including transactions, leases,\nwatches, historical key/value tracking, etc.\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ab4d37c2fa6ccd04beee276a02731d35046bfdca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ab4d37c2fa6ccd04beee276a02731d35046bfdca"}]},"branch":"refs/heads/master"},"494f69811e406c0247d22b169733e1b67cbafb2b":{"kind":"REWORK","_number":10,"created":"2023-02-23 12:44:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/743283/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/10"}}},"commit":{"parents":[{"commit":"0c7f61bb39ded112d62f51894876398f86a5425a","subject":"Add ngs_ssh_disabled_algorithms setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/0c7f61bb39ded112d62f51894876398f86a5425a"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-02-23 12:44:38.000000000","tz":0},"subject":"Support batching up commands","message":"Support batching up commands\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look to batch up commands to send to the switch together using a\nsingle connection. The results of each port\u0027s commands are returned when\navailable.\n\nThis is implemented using etcd as a queueing system. Commands are added\nto an input key, then a worker thread processes the available commands\nfor a particular switch device. We pull off the queue using the version\nat which the keys were added, giving a FIFO style queue. The result of\neach command set are added to an output key, which the original request\nthread is watching. Distributed locks are used to serialise the\nprocessing of commands for each switch device.\n\nVarious neat etcd features are used here to alleviate some of the issues\nof distributed task coordination, including transactions, leases,\nwatches, historical key/value tracking, etc.\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/494f69811e406c0247d22b169733e1b67cbafb2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/494f69811e406c0247d22b169733e1b67cbafb2b"}]},"branch":"refs/heads/master"},"45b237b754cd2d9c6f9c588e759fa74da5b9eacc":{"kind":"REWORK","_number":11,"created":"2023-02-27 15:51:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/743283/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/83/743283/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/83/743283/11"}}},"commit":{"parents":[{"commit":"0c7f61bb39ded112d62f51894876398f86a5425a","subject":"Add ngs_ssh_disabled_algorithms setting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/0c7f61bb39ded112d62f51894876398f86a5425a"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2020-07-27 16:59:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-02-27 15:51:43.000000000","tz":0},"subject":"Support batching up commands","message":"Support batching up commands\n\nWhen you have around 60 baremetal nodes attached to a single switch, it\ntakes a long time to execute all those commands. This gets worse when\nyou limit the number of concurrent ssh connections.\n\nHere we look to batch up commands to send to the switch together using a\nsingle connection. The results of each port\u0027s commands are returned when\navailable.\n\nThis is implemented using etcd as a queueing system. Commands are added\nto an input key, then a worker thread processes the available commands\nfor a particular switch device. We pull off the queue using the version\nat which the keys were added, giving a FIFO style queue. The result of\neach command set are added to an output key, which the original request\nthread is watching. Distributed locks are used to serialise the\nprocessing of commands for each switch device.\n\nVarious neat etcd features are used here to alleviate some of the issues\nof distributed task coordination, including transactions, leases,\nwatches, historical key/value tracking, etc.\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I8c458bbc94df5630cfede5434bcdbe527988059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/45b237b754cd2d9c6f9c588e759fa74da5b9eacc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/45b237b754cd2d9c6f9c588e759fa74da5b9eacc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[]}
