)]}'
{"id":"openstack%2Fnetworking-generic-switch~868318","triplet_id":"openstack%2Fnetworking-generic-switch~master~I5b5c09524ec3db27e694d9a0f1a3bbd777853017","project":"openstack/networking-generic-switch","branch":"master","topic":"g5k-upstreaming/dell-os10-force10-errors","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-02-27 19:16:04.000000000","reason":"removed on reply"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2026-02-28 03:55:54.000000000","reason":"Change was submitted"},"34936":{"account":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"last_update":"2026-02-28 03:55:54.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2026-02-28 03:55:54.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I5b5c09524ec3db27e694d9a0f1a3bbd777853017","subject":"Detect device configuration errors on Dell Force10 and OS10","status":"MERGED","created":"2022-12-21 13:55:20.000000000","updated":"2026-02-28 03:57:44.000000000","submitted":"2026-02-28 03:55:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868318-g5k-upstreaming/dell-os10-force10-errors","meta_rev_id":"98a33d32afff02b533040099ec29d0b1f2a9c7a4","_number":868318,"virtual_id_number":868318,"owner":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-28 03:55:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-02-28 02:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-02-27 19:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-02-27 14:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-02-28 02:24:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 15:32:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 09:46:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-01-04 16:56:34.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-09-26 20:02:53.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-02-28 02:24:56.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"7f584d0d4375a4c791f6fe7e89b05dee680ffaad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"date":"2022-12-21 13:55:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"428984276a938a366a0a9182c9eec8ebd7534a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 15:32:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b954c0b62648f2b8d1c77a04afa8a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b955174a598d4c7680a72ad6adc98eee : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab83aa96ad6b41fdbaa2d449c3d1fd45 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0473bbae58d4f16b19792879ac59698 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9629913d6314474ba78b914e229b52b2 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/772d1fb3e772444fa1675cb2b19df2d0 : SUCCESS in 2m 46s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/3155a4bd6990460393c256c7554e91a9 : SUCCESS in 23m 04s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/f9947155555a4ac98dfc8391e331e561 : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9f6889861fda0699a1ec6caf1089d86e26ebab6a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-04 09:46:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aaef3462d10e8d7991c96f96d9458684f14eddee","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-01-04 16:56:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"242cff77a01d2fe7165204bff0e68107aeb63905","author":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"date":"2023-01-04 18:26:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff8ec0c2519aaf151da2bb7e99aa9211bc12087a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"date":"2025-08-21 14:33:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"14be39e7aae487eca3cfba4f879dace0d9550e5a","author":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"date":"2025-08-21 14:36:51.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7f543ebab4478f90795ea3cbb4feba38057a7206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 16:37:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed667fb31c804b6bb15985dff0331785\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1b57b79d07047cb97fdacda2805299b : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793f20d06c0f4aac958d9b0bfeba6657 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf15b8148786430191c522d9269e0a5b : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b076d035e044fbd81baff6e02afe43b : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52e6f13864364a9d8ca0c2a10ee0aa84 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448b55ffc29a443aa1f92daec500eec8 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51e3a96755a14000aac56fc7611e58f5 : SUCCESS in 2m 31s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/75a88230896f4d41ab9a912cf1bb488d : SUCCESS in 20m 43s\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/3c9fc76228624605980e1d85591a0ced : FAILURE in 2h 00m 03s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/3ab7419b126148e3ad61ddef7674cd4e : FAILURE in 3m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"414d3960cfc686b1c5ab314e7a5f3f4e918987cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"date":"2025-09-08 07:39:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"803f68b246aa42956db45f2320f66b51df7a29ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 10:16:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432cedae0c9b4b1e8047e6b0322c731e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/67eb9f98617d4037baa9e1fd0f810f6a : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d4133329e884292b72bbaa0bc8017c4 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bd23eba7ff9407fa85176d196cd9c46 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de414295e1384782b35674615ef133a0 : SUCCESS in 6m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/795e2f5cda954248938b8c97133ec72f : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89629f3035e14d2e9eb498c4bc2401c2 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8c459591f8e412c8427820fc8cc4bac : SUCCESS in 2m 02s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/291a5fb3b4ff49739e4b885d57e94f1a : SUCCESS in 13m 46s\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/fd1e54155b694bd19a4772c9b23138f3 : SUCCESS in 2h 30m 36s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/93b2f9c88df046558542657f90d52d91 : FAILURE in 2m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29a35c44c8646d037d4011b01841b873e633be5a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-09-24 17:45:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ed295e7d57ec0f9e2fbae91b7f57f9599c80f926","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-09-26 20:02:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a86a0974f1b34df48c57f20149adcd55ece0c2ed","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-09-26 20:02:58.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4683991b36d928e9a8faf9855bc798a0c72b8517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 20:44:56.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/networking-generic-switch for 868318,3","accounts_in_message":[],"_revision_number":3},{"id":"a5b447c0ce562933f3807dab7209f5f0e071bc56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"date":"2025-09-29 13:05:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1ab5eb4eb35af3b3c6dfe06c35f10f319ae0fc05","author":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"date":"2025-09-29 13:09:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"228ebaf84090d1de82bda4c67a303ea4e207fb9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:33:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7166afa2b68a48c5a67011ba99ffb6dc\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e8bd717a84554b0d8a659c90fdfd2313 : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f001dbedb224c7f91dc8d8cecf70774 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca7caf3bbefa4b9c8037b4b180cc6c29 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32eba2bfd9da40378a22030aa7cf0eff : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e976b42c24d47cca15ce3db93a5f3f9 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d9b5c5de784539a18f407e77f64d22 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5222b9d52d2741df92a2cf47f2cd6914 : SUCCESS in 2m 31s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/729f6196caae4d619a2a13f11b5700c8 : SUCCESS in 21m 05s\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/2b29e2777aa6429a92aecd95a9bfa573 : SUCCESS in 2h 22m 57s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/31659f4314d04cacbc3d39a7898bdd11 : FAILURE in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4685336431d89a35857d93496fd4e7d5192f0973","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-09-29 15:34:30.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"043437faf58a019bd1d3cea877686cb0ab830a0c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-10-03 15:44:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7ebdf33fd430a15ee694e23bde71692a2b31e4fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 15:45:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ae4a8cc662cac3049f4f49889cd4c743a47ab710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 16:22:34.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/934670c5e8114605a574487a4c4f71e8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8bbb7a4ba502488bbccc68a0ae503a36 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1277cdd8678e4097b1fad55e12dced53 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a012ebc9a4f34fcaa62b2d4f9fb25fdb : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/769adce04ee14aee84a052ee30429209 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288ffe8aef1f44ab841bd9fd173e200b : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73509e8930404e87bdbd2aab230cd09c : SUCCESS in 3m 48s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/0ba743539f1e4f35a43898cf577d615d : SUCCESS in 17m 10s\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/193ae4491723416a8069328ec667c273 : FAILURE in 36m 34s","accounts_in_message":[],"_revision_number":4},{"id":"53039e3f6a6bb8d8568a19c4e96b6144abfe1593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-27 14:38:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9d8611a35acd78ba4fe5f34469a28cfc743d1440","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-27 14:39:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5abe1a3f57ef14eed037e2e0acac40723d0b4878","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-27 14:39:38.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":5},{"id":"760b87320d8260b789b00b44a97d066cd07352ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 15:39:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02866abe8020410b9bac6f28b7681213\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d941e817344b06b729f31899159164 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/650a3e17a2b5434d865703498f30cf79 : SUCCESS in 6m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de79fc1d50ae4f75a1eb1ae914e4cf17 : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dec922d0f0324a41804eff9101bc9f6f : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233b48fcacba4c63a5b3707f8a812e69 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e30c871df4a409c8c9b0d4a3c47727c : SUCCESS in 2m 39s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/dcdcb37be07946eb9c3940f88a0228d9 : SUCCESS in 9m 52s\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/cc14aa58170247d3a109dda3834f5db1 : FAILURE in 51m 05s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/fc04430b992e4cccab0552ecd637e344 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":5},{"id":"ed5fb442a50a3c919e56215f7e2e5f0a65ac6979","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-02-27 19:16:04.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"72fb81ec82451a2e19bd8302f13941fc7416fbbf","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-27 19:19:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5e3f8208ff8ec591d3eb015f0ae76b0e13c8536d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-02-27 23:37:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"34c81c448631d49ff1d59e597bd9b9a70a91878b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 00:18:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/664d62955bd949229b965d75f7e05dfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81c4a79d6434e83a1d518a3345d2e3f : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19838db461de4506b234e619dfad65a9 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/836562a03f144261833f721f8e7e8d57 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52485ee899c144fda5e434314142ed1b : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08ba208f1ca4192b1e3575dcf840424 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14c52060b5b942cca79921d9cb63effc : SUCCESS in 2m 22s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/5b8c47cc59324b56b3b3cbf1b17b96e6 : SUCCESS in 14m 27s\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/f620d551601e432fabfc69a5e73d4b4e : SUCCESS in 40m 16s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/9902ffdf2eba4b5fafbf8647c747b16f : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":5},{"id":"79f5565781bcf15d1d81858ee5696c922c462a7c","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-02-28 02:24:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8655a724042e7668d4f0802813e34ef3ad36cc05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 02:25:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d4eca9adab076b56f8a0ac80960b4175af627690","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 03:55:54.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5078134786b94eccab5dcb7c9c8c1d10\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7db07bb9fd04db89fbdd6317f47b8e6 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c68669a4b6074c60a719bbe1ba041956 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab7cb924a2ea4da5b22f91c6705754b7 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a67f9479224fe7904fe79e97f26c4c : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95579b78f4ce4429b23e04776d2c6088 : SUCCESS in 2m 33s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/040401bdf994406f9a52c2e0966ac964 : SUCCESS in 23m 55s\n- ironic-tempest-ipa-wholedisk-direct-multinode https://zuul.opendev.org/t/openstack/build/377da07b603f4c3dbd12a426edabd429 : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":5},{"id":"c95d7ecd2246bd7daec072275c6c100b046ce502","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 03:55:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"98a33d32afff02b533040099ec29d0b1f2a9c7a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 03:57:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32723f85683c42d897dcf52c511d891f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8ca73b01da4f5fbf1a0e878a1d8ba2 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26a8ea103d6b4d668e26c63c8b428ac9 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"119f960f00c7866fafb938742eb9d914cf726972","revisions":{"4e3ba88b1347a8ecade6d650acc70ad16a49af4a":{"kind":"REWORK","_number":1,"created":"2022-12-21 13:55:20.000000000","uploader":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"ref":"refs/changes/18/868318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/18/868318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/1"}}},"commit":{"parents":[{"commit":"c83ad2fd061830bafa619cc1f34943fb9c9b8279","subject":"Add SONiC devices to docs page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/c83ad2fd061830bafa619cc1f34943fb9c9b8279"}]}],"author":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2022-02-16 17:16:25.000000000","tz":60},"committer":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2022-12-21 13:54:35.000000000","tz":60},"subject":"Add error message pattern for Dell Force10 and OS10 devices","message":"Add error message pattern for Dell Force10 and OS10 devices\n\nThis allows to detect errors when configuring the device and to make sure\nthat port creation fails in Neutron, instead of silently ignoring the\nerror.\n\nAlso fix small doc issues with Dell OS10 support.\n\nChange-Id: I5b5c09524ec3db27e694d9a0f1a3bbd777853017\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/4e3ba88b1347a8ecade6d650acc70ad16a49af4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/4e3ba88b1347a8ecade6d650acc70ad16a49af4a"}]},"branch":"refs/heads/master"},"4833f560254b4856ce0c97ae0ad10cba356907a4":{"kind":"REWORK","_number":2,"created":"2025-08-21 14:33:35.000000000","uploader":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"ref":"refs/changes/18/868318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/18/868318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/2"}}},"commit":{"parents":[{"commit":"6c1abe4b002530aa6cde832777d1760017c92090","subject":"Enhance doc generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/6c1abe4b002530aa6cde832777d1760017c92090"}]}],"author":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2025-08-21 14:33:07.000000000","tz":120},"committer":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2025-08-21 14:33:07.000000000","tz":120},"subject":"Detect device configuration errors on Dell Force10 and OS10","message":"Detect device configuration errors on Dell Force10 and OS10\n\nAdding an error message pattern allows to detect errors when configuring\nthe device.  It makes sure that port creation fails in Neutron, instead of\nsilently ignoring the error.\n\nBump netmiko requirement because we need this fix for OS9:\n\n  https://github.com/ktbyers/netmiko/pull/3070\n\nChange-Id: I5b5c09524ec3db27e694d9a0f1a3bbd777853017\nSigned-off-by: Baptiste Jonglez \u003cbaptiste.jonglez@inria.fr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/4833f560254b4856ce0c97ae0ad10cba356907a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/4833f560254b4856ce0c97ae0ad10cba356907a4"}]},"branch":"refs/heads/master"},"a9e71f23fcb4e1612b4185fe18c8a6fb6755d31c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-08 07:39:59.000000000","uploader":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"ref":"refs/changes/18/868318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/18/868318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/3"}}},"commit":{"parents":[{"commit":"2f7f3e12369e46294a2be5c780db373ba831124c","subject":"Support security groups in netmiko base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/2f7f3e12369e46294a2be5c780db373ba831124c"}]}],"author":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2025-08-21 14:33:07.000000000","tz":120},"committer":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2025-09-08 07:39:50.000000000","tz":120},"subject":"Detect device configuration errors on Dell Force10 and OS10","message":"Detect device configuration errors on Dell Force10 and OS10\n\nAdding an error message pattern allows to detect errors when configuring\nthe device.  It makes sure that port creation fails in Neutron, instead of\nsilently ignoring the error.\n\nBump netmiko requirement because we need this fix for OS9:\n\n  https://github.com/ktbyers/netmiko/pull/3070\n\nChange-Id: I5b5c09524ec3db27e694d9a0f1a3bbd777853017\nSigned-off-by: Baptiste Jonglez \u003cbaptiste.jonglez@inria.fr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/a9e71f23fcb4e1612b4185fe18c8a6fb6755d31c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/a9e71f23fcb4e1612b4185fe18c8a6fb6755d31c"}]},"branch":"refs/heads/master"},"0f25773ed4cb1d638840a208bd18b76e16c354f9":{"kind":"REWORK","_number":4,"created":"2025-09-29 13:05:20.000000000","uploader":{"_account_id":34936,"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","username":"bjonglez"},"ref":"refs/changes/18/868318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/18/868318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/4"}}},"commit":{"parents":[{"commit":"b413567e6671a7011192f953bf60787fa0aeffc9","subject":"Merge \"test_devices: Fix test to avoid scary backtrace\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/b413567e6671a7011192f953bf60787fa0aeffc9"}]}],"author":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2025-08-21 14:33:07.000000000","tz":120},"committer":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2025-09-29 13:04:03.000000000","tz":120},"subject":"Detect device configuration errors on Dell Force10 and OS10","message":"Detect device configuration errors on Dell Force10 and OS10\n\nAdding an error message pattern allows to detect errors when configuring\nthe device.  It makes sure that port creation fails in Neutron, instead of\nsilently ignoring the error.\n\nBump netmiko requirement because we need this fix for OS9:\n\n  https://github.com/ktbyers/netmiko/pull/3070\n\nChange-Id: I5b5c09524ec3db27e694d9a0f1a3bbd777853017\nSigned-off-by: Baptiste Jonglez \u003cbaptiste.jonglez@inria.fr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/0f25773ed4cb1d638840a208bd18b76e16c354f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/0f25773ed4cb1d638840a208bd18b76e16c354f9"}]},"branch":"refs/heads/master"},"119f960f00c7866fafb938742eb9d914cf726972":{"kind":"REWORK","_number":5,"created":"2026-02-27 14:38:58.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/18/868318/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/18/868318/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/18/868318/5"}}},"commit":{"parents":[{"commit":"0de61285bf4f1906fc8dfd54400be3a54afb055b","subject":"vxlan: OVS testing patch for \u0027vxlan\u0027 binding model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/0de61285bf4f1906fc8dfd54400be3a54afb055b"}]}],"author":{"name":"Baptiste Jonglez","email":"baptiste.jonglez@inria.fr","date":"2025-08-21 14:33:07.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2026-02-27 14:38:36.000000000","tz":-480},"subject":"Detect device configuration errors on Dell Force10 and OS10","message":"Detect device configuration errors on Dell Force10 and OS10\n\nAdding an error message pattern allows to detect errors when configuring\nthe device.  It makes sure that port creation fails in Neutron, instead of\nsilently ignoring the error.\n\nBump netmiko requirement because we need this fix for OS9:\n\n  https://github.com/ktbyers/netmiko/pull/3070\n\nChange-Id: I5b5c09524ec3db27e694d9a0f1a3bbd777853017\nSigned-off-by: Baptiste Jonglez \u003cbaptiste.jonglez@inria.fr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/119f960f00c7866fafb938742eb9d914cf726972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/119f960f00c7866fafb938742eb9d914cf726972"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
