)]}'
{"id":"openstack%2Fnetworking-generic-switch~881959","triplet_id":"openstack%2Fnetworking-generic-switch~master~I8da7802019f2c1bd162a98a8a592104a18b55c7d","project":"openstack/networking-generic-switch","branch":"master","topic":"881959","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2023-06-07 19:39:06.000000000","reason":"\u003cGERRIT_ACCOUNT_9542\u003e replied on the change","reason_account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}},"35976":{"account":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"last_update":"2023-11-15 13:22:10.000000000","reason":"\u003cGERRIT_ACCOUNT_35976\u003e replied on the change","reason_account":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"}},"26922":{"account":{"_account_id":26922,"name":"Steven Relf","email":"s.relf@eschercloud.ai","username":"srelf"},"last_update":"2023-11-15 14:15:05.000000000","reason":"Change was abandoned"},"8493":{"account":{"_account_id":8493,"name":"Serge Kovaleff"},"last_update":"2023-11-15 14:15:05.000000000","reason":"Change was abandoned"},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2023-11-15 14:15:05.000000000","reason":"Change was abandoned"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2023-11-15 14:15:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8da7802019f2c1bd162a98a8a592104a18b55c7d","subject":"Fix fallback to plug_port_to_network on 802_3ad bonds","status":"ABANDONED","created":"2023-05-02 11:17:46.000000000","updated":"2023-11-15 14:15:05.000000000","total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"6bc92527caf55a4310ab45737e16edff8ccc262f","_number":881959,"virtual_id_number":881959,"owner":{"_account_id":26922,"name":"Steven Relf","email":"s.relf@eschercloud.ai","username":"srelf"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-25 11:47:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8493,"name":"Serge Kovaleff"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"all":[{"value":-1,"date":"2023-06-07 19:39:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8493,"name":"Serge Kovaleff"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8493,"name":"Serge Kovaleff"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8493,"name":"Serge Kovaleff"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":8493,"name":"Serge Kovaleff"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 13:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-11 11:18:49.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2023-05-23 09:15:18.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"state":"CC"},{"updated":"2023-05-23 15:03:29.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"state":"REVIEWER"},{"updated":"2023-05-25 09:52:37.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2023-05-25 09:52:37.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2023-05-25 09:52:37.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":8493,"name":"Serge Kovaleff"},"state":"REVIEWER"}],"messages":[{"id":"f34cdcc5b75754c00e60cdfb4c4702aed90221dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26922,"name":"Steven Relf","email":"s.relf@eschercloud.ai","username":"srelf"},"date":"2023-05-02 11:17:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcbe55a9a7b84694b058fd9621b14c3465363629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 13:18:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c4cb65ce2c42e2bff631eddfd5625f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06c2694a1f84af9aa6825f106eb806e : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/308e9b7f351248dfb8de5fd493fe2055 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65a206b4d55440298a92e844f6f9707d : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b212efec3f9a4945bfccc66c25763a2c : SUCCESS in 7m 08s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/e70552ec566e4fb5a9d5114e5397e5c5 : SUCCESS in 24m 03s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/99646a1d53914df28d9f0e29a09f673d : SUCCESS in 1h 50m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0c663b664bf3678cd1f91ee1d411489af10df3da","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-05-11 11:18:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bba2f50fe00314a6b5bcb756f04a2d8cd7a67765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"date":"2023-05-23 07:51:57.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e9a44210d7139630e93bcd3fff4a2885514803f6","author":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"date":"2023-05-23 09:15:18.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"654c094321035ab546ff67d89806274fb5ce7efd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"date":"2023-05-23 09:29:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd2c1c9aae89afcf7b511f2a1344b69de5fbd1b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 10:46:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fdc46fd673a4d6fbf732222e3a387fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8291d9d21354a9d83472d06040a2d38 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/255a9e1ac09c4da7ba2643fd77cd1b89 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83f639275cbc43abb132e995d2e7428d : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53fb32090c4a450c85e7ac6ce32f0942 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898aee85091d4500924e8d0b3f500bf4 : SUCCESS in 6m 42s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/05367994dc554d18b3d188498ee8fe0d : SUCCESS in 17m 58s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/ab19c7cd0a0b411abb895594e8d8b169 : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":3},{"id":"1388caf629a3ab8491070cdd91f36fcaf6f6fc15","author":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"date":"2023-05-23 15:03:29.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"943b11cd933277497fa1aac0ce12465cc5fa6b8a","author":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"date":"2023-05-25 09:52:37.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"347044bb95aa747c2c2dd2d7227f164503bd3d4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"date":"2023-05-25 10:01:13.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"91a9344b88a5ea460fa7860cf9767e1752d46978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 11:47:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83c382ab9001410b919c26c3dc9d313a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f8ba9925626455c9f60c6327d278f29 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/457222339bd54304a7d1bd1f098d1d6a : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84c8d10cba344115abcdb49477f84ee3 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/655cff5468e94dc4a40f2825fde4fba9 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf10871cd894876aa793497116f91d2 : SUCCESS in 6m 46s\n- networking-generic-switch-tempest-dlm https://zuul.opendev.org/t/openstack/build/033f3c18f45544e59fa521debaa2d636 : SUCCESS in 13m 29s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/6547f125995a41ab9d3c8bb3c8490d68 : SUCCESS in 1h 41m 20s","accounts_in_message":[],"_revision_number":4},{"id":"2bc93863028d066e018e11c266de17d82a520dda","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-06-07 19:39:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"89ea4268bd5d79fc9dcb783b662a3b9c2a182667","author":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"date":"2023-11-15 13:22:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6bc92527caf55a4310ab45737e16edff8ccc262f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26922,"name":"Steven Relf","email":"s.relf@eschercloud.ai","username":"srelf"},"date":"2023-11-15 14:15:05.000000000","message":"Abandoned\n\nfixed in master","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7f77d1e4232b617e109d0b4c4fb1c2ab9b337cc3","revisions":{"ad583e7170b8ca06c838486d4989095e3e0cf45f":{"kind":"REWORK","_number":1,"created":"2023-05-02 11:17:46.000000000","uploader":{"_account_id":26922,"name":"Steven Relf","email":"s.relf@eschercloud.ai","username":"srelf"},"ref":"refs/changes/59/881959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/59/881959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/1"}}},"commit":{"parents":[{"commit":"ec91cbb93ecc1eca52431f31ef0ae1317d98e87b","subject":"Merge \"Fixes, updates for CI configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ec91cbb93ecc1eca52431f31ef0ae1317d98e87b"}]}],"author":{"name":"Steve Relf","email":"s.relf@eschercloud.ai","date":"2023-05-02 11:15:38.000000000","tz":60},"committer":{"name":"Steve Relf","email":"s.relf@eschercloud.ai","date":"2023-05-02 11:15:38.000000000","tz":60},"subject":"Ensures that if a switch plugin has an empty plug/unplug bond to network function that the old code path of plug/unplug port to network is followed, otherwise this can result in a no-op happening","message":"Ensures that if a switch plugin has an empty plug/unplug bond to network\nfunction that the old code path of plug/unplug port to network is\nfollowed, otherwise this can result in a no-op happening\n\nCo-authored-by: Steven Relf s.relf@eschercloud.ai\nCo-authored-by: Graeme Moss g.moss@eschercloud.ai\nChange-Id: I8da7802019f2c1bd162a98a8a592104a18b55c7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ad583e7170b8ca06c838486d4989095e3e0cf45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ad583e7170b8ca06c838486d4989095e3e0cf45f"}]},"branch":"refs/heads/master"},"aac3d4313a7f22a4482b7e6c21855cc3c1b361e7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-23 07:51:57.000000000","uploader":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"ref":"refs/changes/59/881959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/59/881959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/2"}}},"commit":{"parents":[{"commit":"ec91cbb93ecc1eca52431f31ef0ae1317d98e87b","subject":"Merge \"Fixes, updates for CI configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ec91cbb93ecc1eca52431f31ef0ae1317d98e87b"}]}],"author":{"name":"Steve Relf","email":"s.relf@eschercloud.ai","date":"2023-05-02 11:15:38.000000000","tz":60},"committer":{"name":"Graeme Moss","email":"grami.moss@gmail.com","date":"2023-05-23 07:51:57.000000000","tz":0},"subject":"Fix fallback to plug_port_to_network on 802_3ad bonds","message":"Fix fallback to plug_port_to_network on 802_3ad bonds\n\nEnsures that if a switch plugin has an empty plug/unplug bond to network\nfunction that the old code path of plug/unplug port to network is\nfollowed, otherwise this can result in a no-op happening\n\nCo-authored-by: Steven Relf s.relf@eschercloud.ai\nCo-authored-by: Graeme Moss g.moss@eschercloud.ai\nChange-Id: I8da7802019f2c1bd162a98a8a592104a18b55c7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/aac3d4313a7f22a4482b7e6c21855cc3c1b361e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/aac3d4313a7f22a4482b7e6c21855cc3c1b361e7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bbc95942634836a1076597edaf83d5373d1850cf":{"kind":"REWORK","_number":3,"created":"2023-05-23 09:29:13.000000000","uploader":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"ref":"refs/changes/59/881959/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/59/881959/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/3"}}},"commit":{"parents":[{"commit":"ec91cbb93ecc1eca52431f31ef0ae1317d98e87b","subject":"Merge \"Fixes, updates for CI configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ec91cbb93ecc1eca52431f31ef0ae1317d98e87b"}]}],"author":{"name":"Steve Relf","email":"s.relf@eschercloud.ai","date":"2023-05-02 11:15:38.000000000","tz":60},"committer":{"name":"Graeme Moss","email":"g.moss@eschercloud.ai","date":"2023-05-23 09:24:50.000000000","tz":60},"subject":"Fix fallback to plug_port_to_network on 802_3ad bonds","message":"Fix fallback to plug_port_to_network on 802_3ad bonds\n\nEnsures that if a switch plugin has an empty plug/unplug bond to network\nfunction that the old code path of plug/unplug port to network is\nfollowed, otherwise this can result in a no-op happening\n\nCo-authored-by: Steven Relf s.relf@eschercloud.ai\nCo-authored-by: Graeme Moss g.moss@eschercloud.ai\nChange-Id: I8da7802019f2c1bd162a98a8a592104a18b55c7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/bbc95942634836a1076597edaf83d5373d1850cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/bbc95942634836a1076597edaf83d5373d1850cf"}]},"branch":"refs/heads/master"},"7f77d1e4232b617e109d0b4c4fb1c2ab9b337cc3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-05-25 10:01:13.000000000","uploader":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"ref":"refs/changes/59/881959/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/59/881959/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/59/881959/4"}}},"commit":{"parents":[{"commit":"ec91cbb93ecc1eca52431f31ef0ae1317d98e87b","subject":"Merge \"Fixes, updates for CI configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/ec91cbb93ecc1eca52431f31ef0ae1317d98e87b"}]}],"author":{"name":"Steve Relf","email":"s.relf@eschercloud.ai","date":"2023-05-02 11:15:38.000000000","tz":60},"committer":{"name":"Graeme Moss","email":"g.moss@eschercloud.ai","date":"2023-05-25 10:01:13.000000000","tz":0},"subject":"Fix fallback to plug_port_to_network on 802_3ad bonds","message":"Fix fallback to plug_port_to_network on 802_3ad bonds\n\nEnsures that if a switch plugin has an empty plug/unplug bond to network\nfunction that the old code path of plug/unplug port to network is\nfollowed, otherwise this can result in a no-op happening\n\nCo-authored-by: Steven Relf s.relf@eschercloud.ai\nCo-authored-by: Graeme Moss g.moss@eschercloud.ai\nCloses-Bug: #2020764\nChange-Id: I8da7802019f2c1bd162a98a8a592104a18b55c7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7f77d1e4232b617e109d0b4c4fb1c2ab9b337cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/7f77d1e4232b617e109d0b4c4fb1c2ab9b337cc3"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
