)]}'
{"id":"openstack%2Fnetworking-generic-switch~992889","triplet_id":"openstack%2Fnetworking-generic-switch~master~I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe","project":"openstack/networking-generic-switch","branch":"master","topic":"restconf","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2026-09-21 13:20:41.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe","subject":"Add RESTCONF documentation","status":"MERGED","created":"2026-06-11 11:36:09.000000000","updated":"2026-09-21 13:20:41.000000000","submitted":"2026-09-21 13:20:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992889-restconf","meta_rev_id":"3cb5c4eef6e3bb4d56e4e1160d5726ed270966f2","_number":992889,"virtual_id_number":992889,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 13:20:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-21 11:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-09-18 15:50:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-21 11:32:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-11 11:47:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 15:50:27.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2026-09-21 11:32:29.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"ac58ce131d653403aad737d81b24669b33c858b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-06-11 11:36:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cf38936567ac1bc118c4cecd682adace7cc8c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 11:47:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81db55358fe6421e9d5e91ad9907b1a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2a3b8d7a7b45e0beaf83687b82809e : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eccb3a1005bd45b6a5e4caee00234ee8 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d54a2a98f7844842b95a2b1d19cff67b : SUCCESS in 4m 43s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/c2189a7273c243c88ca187c08e204203 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9083717ee6c97a9742991c326ef617e361974741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-06-12 07:24:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6fdcc054e71a275df5b0e95e31933dc92346d9e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-06-12 14:22:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c79e511d8a72096e8b2a742a5efde35de44a30a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 14:38:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31a83f507885413c92483bedec235544\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8d4896fbf54742beec55e7ff8081c8 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c607e72dd8431ab384346b8b8c22a5 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc869dbc0beb4bf88702200015c854b5 : SUCCESS in 4m 34s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/4da198fdcdda41399ad16bf628afd25e : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"beef7cc3631883503da59fa2c02cfdb53027d4ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-07-22 10:36:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d9ac0bf33f87abdd9dca334ede6cfa0f3f5c86ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 10:51:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d38e5656a0ab4deeb27775f4bb05890b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4acc8eca5c148b0bb513c4d415b42ff : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e84c9c909b943e2a21639491929ec7c : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25178d656e77405899ed5f16cbda061d : SUCCESS in 4m 07s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/28ca95dff2a345708b0a7925dd306166 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":4},{"id":"5f7871f332b73b763d608eca264bd8ba2e3d9fe4","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-09-07 08:50:05.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":4},{"id":"ac77fa395a8da8bba7dfd1f1408c8a0e53815a36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-09-09 12:25:15.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"40e847f4e382254db3b346ce536d5347d84e7aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 12:40:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c64d0f7dd6054c24a2df4da8a7952cc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0733b5b88104c5182941a593c2d6e38 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adbeff31483c4eb39276012823f77f18 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68483b1df7fc48dc8a19c8157a1af2f1 : SUCCESS in 3m 46s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/073a767ccc56450f9d28d2b873d6341b : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":5},{"id":"c03be263d2c954ef2d13de73f60c6e7c79a6c07e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-09-09 15:11:35.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"ec9d064fba9a7aaf56e7e2907e9508c4fc53e39b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:40:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8696e39d2ab4e17baffd9a2da14ac97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b827ca528b499fb343024cdb920630 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad586c07756c4b39a0264afdb39cf58b : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/237c0e5aa90247d3a64bb46163e8c596 : SUCCESS in 5m 04s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/6e824ba25c934b83b345c62aaf4fece8 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":6},{"id":"8ab0274a71e5f1dd99075810b6053e2dfc9474ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-09-11 09:27:38.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"40feab4e1a8d37f205f999b08a8fa895ec58ae87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 09:40:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9007af95ea3441c996eb7a63408f01b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d05cd86ad048ae824a340fc4ded029 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef7d92b3a30432ea2336e53771e05fa : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb950f1514404a72a88b70357ce1ce84 : SUCCESS in 4m 15s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/6667c0d906b547cdae7dfa9246c6fdca : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":7},{"id":"a26a4977ac2a131f5b198e00414588816f8e7a9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2026-09-18 08:17:30.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"81ee1f7ccdeffd61457d7e2412ad551d32de051a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 08:38:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0696f8d4181e46dd9e907ec8b76b5c7a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91dacdeea5324318b7f3206456428470 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a0dca6d697f4678b1631a6c35f16e5c : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29a7acc9e4344c3cbffa10d9e49fa969 : SUCCESS in 4m 43s\n- networking-generic-switch-tox-codespell https://zuul.opendev.org/t/openstack/build/2c9c9dfff2d5467998a453d07e66c309 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":8},{"id":"31a24066844e664e789165d2fafd5081ad1fc67a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2026-09-18 15:50:27.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"dcff747e58691933b84f16214fc9cb1c91ce879f","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 11:32:29.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ed129e54b93a450290dff308f30159a02ff71a19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 11:32:55.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c27e72c58af63bf74d5d9ef0669ae802ccaedbc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:20:41.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0eaa46b24254985ba825098ab5d8080\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62d6a8f9e67c44229f30f89e2796c777 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98b323d4b5c452bbcb198ff161f2e65 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d19e1fcf75c24cb182ba2480e2eb1b72 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":8},{"id":"3cb5c4eef6e3bb4d56e4e1160d5726ed270966f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:20:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"71f636dcb8e30378e2d9c1407f748ff67b1f4ec1","revisions":{"d3a78c3cfb5cc6e6b6ee92723f5437739d214788":{"kind":"REWORK","_number":1,"created":"2026-06-11 11:36:09.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/89/992889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/89/992889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/1"}}},"commit":{"parents":[{"commit":"251c23035f61e1045ea6559fed97da5610623855","subject":"Add RestconfOpenConfigSwitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/251c23035f61e1045ea6559fed97da5610623855"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-09 06:34:47.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-11 11:29:26.000000000","tz":120},"subject":"Add RESTCONF documentation","message":"Add RESTCONF documentation\n\nDocument the RESTCONF management interface transport.\n\nChange-Id: I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d3a78c3cfb5cc6e6b6ee92723f5437739d214788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d3a78c3cfb5cc6e6b6ee92723f5437739d214788"}]},"branch":"refs/heads/master"},"88e71fa8bbd3d863e453d7e534c66489321f23bb":{"kind":"REWORK","_number":2,"created":"2026-06-12 07:24:45.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/89/992889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/89/992889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/2"}}},"commit":{"parents":[{"commit":"c8cf86ce430004f843f3f35cc71fa78f2ff63bfc","subject":"Add RestconfOpenConfigSwitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/c8cf86ce430004f843f3f35cc71fa78f2ff63bfc"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-09 06:34:47.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-11 22:09:16.000000000","tz":120},"subject":"Add RESTCONF documentation","message":"Add RESTCONF documentation\n\nDocument the RESTCONF management interface transport.\n\nChange-Id: I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/88e71fa8bbd3d863e453d7e534c66489321f23bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/88e71fa8bbd3d863e453d7e534c66489321f23bb"}]},"branch":"refs/heads/master"},"3e42086987cb35bc5460adbfb52c01aa6ec4fd86":{"kind":"REWORK","_number":3,"created":"2026-06-12 14:22:10.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/89/992889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/89/992889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/3"}}},"commit":{"parents":[{"commit":"88bd06a5dc7e015e399ff3916ada619de3b58a5a","subject":"Add RestconfOpenConfigSwitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/88bd06a5dc7e015e399ff3916ada619de3b58a5a"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-09 06:34:47.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-12 14:20:41.000000000","tz":120},"subject":"Add RESTCONF documentation","message":"Add RESTCONF documentation\n\nDocument the RESTCONF management interface transport.\n\nChange-Id: I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/3e42086987cb35bc5460adbfb52c01aa6ec4fd86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/3e42086987cb35bc5460adbfb52c01aa6ec4fd86"}]},"branch":"refs/heads/master"},"d169b615e3a1a6953bebd15c1b705e41f0cb67a5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-22 10:36:45.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/89/992889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/89/992889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/4"}}},"commit":{"parents":[{"commit":"08e4675c20700f647f29189578e7a39289987ed7","subject":"Add RestconfOpenConfigSwitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/08e4675c20700f647f29189578e7a39289987ed7"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-09 06:34:47.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-07-22 10:17:58.000000000","tz":120},"subject":"Add RESTCONF documentation","message":"Add RESTCONF documentation\n\nDocument the RESTCONF management interface transport.\n\nChange-Id: I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d169b615e3a1a6953bebd15c1b705e41f0cb67a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/d169b615e3a1a6953bebd15c1b705e41f0cb67a5"}]},"branch":"refs/heads/master"},"0258256927a94313e0b045bfacc631db54026004":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-09 12:25:15.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/89/992889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/89/992889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/5"}}},"commit":{"parents":[{"commit":"f3a32b618c40f0454d3e7fc55ec6980f3ec50965","subject":"Add RestconfOpenConfigSwitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/f3a32b618c40f0454d3e7fc55ec6980f3ec50965"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-09 06:34:47.000000000","tz":120},"committer":{"name":"Harald Jensas","email":"hjensas@redhat.com","date":"2026-09-09 12:23:26.000000000","tz":120},"subject":"Add RESTCONF documentation","message":"Add RESTCONF documentation\n\nDocument the RESTCONF management interface transport.\n\nChange-Id: I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/0258256927a94313e0b045bfacc631db54026004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/0258256927a94313e0b045bfacc631db54026004"}]},"branch":"refs/heads/master"},"f7805470a443084f39bd344264ff10bddc3db59f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-09-09 15:11:35.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/89/992889/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/89/992889/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/6"}}},"commit":{"parents":[{"commit":"4d5b9fdd3b42953b89b1dc876be4f274d2d84557","subject":"Add RestconfOpenConfigSwitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/4d5b9fdd3b42953b89b1dc876be4f274d2d84557"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-09 06:34:47.000000000","tz":120},"committer":{"name":"Harald Jensas","email":"hjensas@redhat.com","date":"2026-09-09 15:06:07.000000000","tz":120},"subject":"Add RESTCONF documentation","message":"Add RESTCONF documentation\n\nDocument the RESTCONF management interface transport.\n\nChange-Id: I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/f7805470a443084f39bd344264ff10bddc3db59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/f7805470a443084f39bd344264ff10bddc3db59f"}]},"branch":"refs/heads/master"},"c02089f5d7dc879b53712da5040c0551e62c490c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-09-11 09:27:38.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/89/992889/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/89/992889/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/7"}}},"commit":{"parents":[{"commit":"62af7b3de32d48bea7b7beb724453cdbbb78052a","subject":"Add RestconfOpenConfigSwitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/62af7b3de32d48bea7b7beb724453cdbbb78052a"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-09 06:34:47.000000000","tz":120},"committer":{"name":"Harald Jensas","email":"hjensas@redhat.com","date":"2026-09-11 09:27:10.000000000","tz":120},"subject":"Add RESTCONF documentation","message":"Add RESTCONF documentation\n\nDocument the RESTCONF management interface transport.\n\nChange-Id: I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/c02089f5d7dc879b53712da5040c0551e62c490c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/c02089f5d7dc879b53712da5040c0551e62c490c"}]},"branch":"refs/heads/master"},"71f636dcb8e30378e2d9c1407f748ff67b1f4ec1":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-09-18 08:17:30.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/89/992889/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-generic-switch","ref":"refs/changes/89/992889/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-generic-switch refs/changes/89/992889/8"}}},"commit":{"parents":[{"commit":"a4f3799be9a6869e326aadadb177c76074b9530c","subject":"Add RestconfOpenConfigSwitch driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/a4f3799be9a6869e326aadadb177c76074b9530c"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2026-06-09 06:34:47.000000000","tz":120},"committer":{"name":"Harald Jensas","email":"hjensas@redhat.com","date":"2026-09-18 08:16:58.000000000","tz":120},"subject":"Add RESTCONF documentation","message":"Add RESTCONF documentation\n\nDocument the RESTCONF management interface transport.\n\nChange-Id: I45ea964b871f6a30e271db2e8aad3b5b34b9c8fe\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/71f636dcb8e30378e2d9c1407f748ff67b1f4ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-generic-switch/commit/71f636dcb8e30378e2d9c1407f748ff67b1f4ec1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
