)]}'
{"id":"openstack%2Fnetworking-midonet~199387","triplet_id":"openstack%2Fnetworking-midonet~master~Ifd462e0fac8f6b3237766a29f928152c9669133b","project":"openstack/networking-midonet","branch":"master","topic":"ci-test","hashtags":[],"change_id":"Ifd462e0fac8f6b3237766a29f928152c9669133b","subject":"TEST FOR CI DO NOT REVIEW","status":"ABANDONED","created":"2015-07-08 03:21:22.000000000","updated":"2020-09-25 15:33:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"38cd3d4a1a9e2eba5681e2b898622ea691973ab3","_number":199387,"virtual_id_number":199387,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"date":"2020-09-25 15:33:31.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-08 07:20:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-2,"date":"2019-10-08 05:13:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-05-12 10:04:48.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2017-09-22 07:40:21.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-10-13 02:10:09.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2019-04-18 15:59:38.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-10-01 23:41:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-10-08 07:20:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 15:33:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4d7ef05a595fa4087cc31eb153954f81ab727070","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-07-08 03:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3807d33c2c2178113186dc977256d9a013854758","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-08 04:14:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/1/check/gate-networking-midonet-pep8/cd6c14e/ : SUCCESS in 2m 42s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/1/check/gate-networking-midonet-docs/12f94ae//doc/build/html/ : SUCCESS in 2m 57s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/1/check/gate-networking-midonet-python27/220c0bc/ : SUCCESS in 5m 11s\n- check-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/1/check/check-tempest-dsvm-networking-midonet/7b24f24/ : SUCCESS in 48m 09s\n- check-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/1/check/check-tempest-dsvm-networking-midonet-v2/a99c8f5/ : SUCCESS in 53m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9bdc90f37fd4386bb5c76cab08da98c6d3150a2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-07-09 02:40:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"89e2c0e090b3723842873c7b201e8d43a1b0c911","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-07-09 02:40:41.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"1eb7e01e5ca42f63b153a31b464a83975ce888c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-09 03:19:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/1/check/gate-networking-midonet-pep8/6ac2dc4/ : SUCCESS in 2m 41s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/1/check/gate-networking-midonet-docs/3305d75//doc/build/html/ : SUCCESS in 2m 35s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/1/check/gate-networking-midonet-python27/680eaa8/ : SUCCESS in 6m 30s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/1/check/gate-networking-midonet-python34/2c380cb/ : FAILURE in 13m 52s\n- check-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/1/check/check-tempest-dsvm-networking-midonet/0719a33/ : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b41f74ce0ef60569dab23e7e74b49b325525e072","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-09 03:22:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- check-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/1/experimental/check-tempest-dsvm-networking-midonet-v2/6ce0fcc/ : SUCCESS in 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86ef3404ea5a7ab33c45ef46ed410cfdadba394d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-07-09 08:20:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fca8790660d3cedecd71e9c9a58a85a75d5e3973","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-07-09 08:22:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2ca539c01d8380e7d338995f458038d2da3476b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-07-09 09:13:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/3/check/gate-networking-midonet-pep8/d4cf042/ : SUCCESS in 3m 04s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/3/check/gate-networking-midonet-docs/5cfbf96//doc/build/html/ : SUCCESS in 3m 00s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/3/check/gate-networking-midonet-python27/0f733e7/ : SUCCESS in 5m 39s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/3/check/gate-networking-midonet-python34/c02a111/ : SUCCESS in 4m 37s\n- check-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/3/check/check-tempest-dsvm-networking-midonet/084dc91/ : SUCCESS in 47m 38s","accounts_in_message":[],"_revision_number":3},{"id":"6798da26868d0af92a6e7061b19ef1f49410af65","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-08-27 05:37:02.000000000","message":"Patch Set 3: Code-Review-2","accounts_in_message":[],"_revision_number":3},{"id":"02714cb22965e7e875b79677d604f16745e9618e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-08-27 06:05:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e778897d7278bc6c65e3db43a49d4f51017518f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-08-27 06:07:22.000000000","message":"Topic changed from py3 to ci-test","accounts_in_message":[],"_revision_number":4},{"id":"20426a3097a78235d3a03383ff330a2370d3dca8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-08-27 07:01:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/67120b5/ : SUCCESS in 2m 44s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/0d7656e//doc/build/html/ : SUCCESS in 3m 19s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/204cc4f/ : SUCCESS in 8m 52s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/11dab23/ : SUCCESS in 4m 43s\n- gate-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/4/check/gate-tempest-dsvm-networking-midonet/a89ebaa/ : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":4},{"id":"fc5c2222386e67dc3e9178bf8b5f4acd635daf20","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-01 09:03:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"03f4600fa115716a181b58c2c72e24129b8cac50","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-01 10:08:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/4b5dff7/ : SUCCESS in 2m 52s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/8033593//doc/build/html/ : SUCCESS in 3m 11s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/0c4e4df/ : FAILURE in 3m 16s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/c7e7b35/ : FAILURE in 3m 54s\n- gate-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/4/check/gate-tempest-dsvm-networking-midonet/e409a11/ : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":4},{"id":"02ea1882114ec9c3996a056106eeb5fee4edf8aa","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-04 08:21:47.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5fa9ae44b6ba254a929b58c4d9e058f23c3f90ac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-04 08:44:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/0797068/ : SUCCESS in 3m 17s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/87fe2f7//doc/build/html/ : SUCCESS in 3m 06s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/c0867f6/ : FAILURE in 3m 18s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/c31cf1b/ : FAILURE in 4m 00s\n- gate-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/4/check/gate-tempest-dsvm-networking-midonet/0ab701f/ : FAILURE in 22m 57s","accounts_in_message":[],"_revision_number":4},{"id":"52160d5e08ecf413b192c120fc63fd91d64db7da","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-07 13:24:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"673128e2d84726bab618d59f1913a7acc930ee26","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-07 13:47:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/87a12b9/ : SUCCESS in 3m 02s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/e9c0d04//doc/build/html/ : SUCCESS in 3m 13s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/78349a5/ : FAILURE in 3m 15s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/d5317e0/ : FAILURE in 4m 02s\n- gate-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/4/check/gate-tempest-dsvm-networking-midonet/02a7309/ : FAILURE in 23m 15s","accounts_in_message":[],"_revision_number":4},{"id":"6ab001ee1ec000388f1633eefb8a305eaa8e71e2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-07 14:31:33.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"64ec1e2a892d032c74d63a46136e47f0b1b3b605","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-07 15:12:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/4/experimental/gate-tempest-dsvm-networking-midonet-v2/d4045b7/ : FAILURE in 38m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/4/experimental/gate-tempest-dsvm-networking-midonet-ml2/471a3b2/ : FAILURE in 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bfac09d88cec7dd913dd36fbfe56ddf539e8ba5e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-10 03:35:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d6e41ec60b8c14439569d4588ee1187719c0cfc4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-10 04:09:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/0c99d62/ : SUCCESS in 3m 21s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/c7632c7//doc/build/html/ : SUCCESS in 2m 07s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/8093ae4/ : FAILURE in 3m 54s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/2c09a2e/ : FAILURE in 4m 06s\n- gate-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/4/check/gate-tempest-dsvm-networking-midonet/d5ffaa4/ : FAILURE in 33m 55s","accounts_in_message":[],"_revision_number":4},{"id":"ba2faaca42cf16a2c24ebdb3106552d569640670","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-10 04:20:45.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"72e91149fef2e1982be319c54e561c4e8acdd9f0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-10 05:01:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/4/experimental/gate-tempest-dsvm-networking-midonet-v2/5c29449/ : FAILURE in 31m 35s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/4/experimental/gate-tempest-dsvm-networking-midonet-ml2/bc9b061/ : FAILURE in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"467cd4aa4718220750a2d7ca948745ecf3b50ddf","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-15 03:46:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d16ea60268184b4e7e6cca13847c8db31b9a86c5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-15 03:47:39.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"4e3b4861f585e4238b330cd527e77313b56f0ebb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-15 04:22:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/3429d71/ : SUCCESS in 2m 22s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/8c1bb2c//doc/build/html/ : SUCCESS in 2m 30s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/9b2fe31/ : FAILURE in 4m 26s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/8973411/ : FAILURE in 3m 16s\n- gate-tempest-dsvm-networking-midonet http://logs.openstack.org/87/199387/4/check/gate-tempest-dsvm-networking-midonet/33e6b4a/ : FAILURE in 34m 50s","accounts_in_message":[],"_revision_number":4},{"id":"32ab17209e269783e86065f4bc3fab541bdbad94","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-15 04:34:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/4/experimental/gate-tempest-dsvm-networking-midonet-v2/254bb27/ : FAILURE in 27m 09s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/4/experimental/gate-tempest-dsvm-networking-midonet-ml2/8576591/ : FAILURE in 46m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7696d6ae9a7a3af0d866528e3cc86d65537adcb7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-28 02:43:19.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e9f60775a9d1be949be5718082e37cd5c84b1cca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-28 02:52:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/085909d/ : SUCCESS in 2m 34s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/76b1c5b//doc/build/html/ : SUCCESS in 2m 11s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/70373d5/ : FAILURE in 4m 59s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/685ba4a/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"26bb356e0fa3d66711f3c4be9c91e8ccd31830f7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-09-30 06:37:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5da9e9aaa31bc570cae82ec46fa826c7cdaba4b8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-09-30 06:44:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/a6e7840/ : SUCCESS in 2m 24s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/33e441f//doc/build/html/ : SUCCESS in 2m 28s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/bdfb59a/ : SUCCESS in 4m 58s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/cbd4d13/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":4},{"id":"cf5587a5d202ac16cf6d8d91ddd4f5b839fc8d77","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-10-13 06:29:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"dfa61860ca3a52fc09b789c963b6ddd8ffa30425","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-13 06:49:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/d8e5c0c/ : SUCCESS in 3m 09s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/81b8278//doc/build/html/ : SUCCESS in 5m 12s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/a6032f3/ : FAILURE in 3m 36s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/cb895af/ : FAILURE in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"c0597ac9c5ba6df618bd80bff84dc2cf327f210b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-10-15 08:08:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fa324a8efe2101a3f7f43f273ad0b1706a9c29dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-15 08:17:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/7ecf044/ : SUCCESS in 2m 40s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/e102d9e//doc/build/html/ : SUCCESS in 2m 56s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/c1337eb/ : SUCCESS in 6m 48s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/2fa43a0/ : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":4},{"id":"a7d4ac81381c4be921695d3c9e093adb425d1f4e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-10-23 12:06:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9d986103caa789c3c2de3f67f42bf65b29376895","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-10-23 12:06:17.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"bca790b0d4334758c698a90fe01de295516977dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-23 12:32:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-pep8/4890cc7/ : SUCCESS in 2m 27s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/4/check/gate-networking-midonet-docs/3fb3e23//doc/build/html/ : SUCCESS in 2m 54s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python27/2c60620/ : SUCCESS in 5m 54s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/4/check/gate-networking-midonet-python34/f07e389/ : SUCCESS in 5m 43s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/4/check/gate-tempest-dsvm-networking-midonet-v1/9af6451/ : FAILURE in 23m 14s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/4/check/gate-tempest-dsvm-networking-midonet-v2/55523a4/ : FAILURE in 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4a3e4bcbf3130f67c63fc182d59ba140852df102","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-23 13:01:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/4/experimental/gate-tempest-dsvm-networking-midonet-ml2/2de2d54/ : FAILURE in 54m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c5830e28946eedcb714b6ac67dff550ca51c3365","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-10-28 06:30:11.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b9025172d9c62a8fe0b125a15c0136796a2a8c3a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-10-28 06:30:22.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"0fd0b5baa7cf4c9b7759e9e2d907dffe089d0d06","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-28 06:55:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/5/experimental/gate-tempest-dsvm-networking-midonet-ml2/6387ca5/ : FAILURE in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c6c419b3592cd67d0ea2caab05bcc2735ffe2133","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-28 08:10:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-pep8/10e35b6/ : SUCCESS in 3m 24s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/5/check/gate-networking-midonet-docs/0ea55c5//doc/build/html/ : SUCCESS in 2m 35s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-python27/644863b/ : SUCCESS in 7m 12s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-python34/bda4f5f/ : SUCCESS in 5m 29s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/5/check/gate-tempest-dsvm-networking-midonet-v1/f817718/ : FAILURE in 1h 36m 18s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/5/check/gate-tempest-dsvm-networking-midonet-v2/465248b/ : FAILURE in 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c4e339c1c96183c8c38d8197b9d6902c111d48b2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-11-05 10:21:03.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a369b81105a8fba1536dbc6ee407a1f88b08c037","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-11-05 10:21:12.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"5edfd9ac462a46e5b4f6c660c0137ac5afc2aa08","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-05 11:31:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-pep8/32eab07/ : SUCCESS in 3m 23s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/5/check/gate-networking-midonet-docs/7d93962//doc/build/html/ : SUCCESS in 3m 38s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-python27/61ceaac/ : SUCCESS in 5m 36s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-python34/cd6b55d/ : SUCCESS in 6m 07s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/5/check/gate-tempest-dsvm-networking-midonet-v1/2af2395/ : FAILURE in 52m 58s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/5/check/gate-tempest-dsvm-networking-midonet-v2/26d452e/ : FAILURE in 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a0cce36d8a0a16a0d332f3c2f7326576495238f3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-05 11:36:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/5/experimental/gate-tempest-dsvm-networking-midonet-ml2/dfd9ede/ : SUCCESS in 56m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c1e3d834665b069030cb5c0db43746bbee236b0a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-11-05 11:39:23.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2fa1c837b5bafbc7f92ace63f2815d159c53a0b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-05 13:05:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-pep8/0243514/ : SUCCESS in 3m 17s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/5/check/gate-networking-midonet-docs/c3fb1a6//doc/build/html/ : SUCCESS in 3m 45s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-python27/c11a4c9/ : SUCCESS in 5m 59s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/5/check/gate-networking-midonet-python34/953b55d/ : SUCCESS in 6m 00s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/5/check/gate-tempest-dsvm-networking-midonet-v1/39130c9/ : FAILURE in 57m 17s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/5/check/gate-tempest-dsvm-networking-midonet-v2/4a57f46/ : SUCCESS in 1h 08m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2b01b8fb592c0914f21b0527bf17de1a5684e211","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-11-06 08:59:10.000000000","message":"Patch Set 6: Commit message was updated","accounts_in_message":[],"_revision_number":6},{"id":"bd5dcfa8247372159131b1196e1e324fc20be5a9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-11-06 08:59:14.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"2543f56eb06da21b0d42b1d60c3a73ce29403024","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-06 10:01:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/7/check/gate-networking-midonet-pep8/9bb0054/ : SUCCESS in 4m 18s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/7/check/gate-networking-midonet-docs/b85800d//doc/build/html/ : SUCCESS in 3m 41s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/7/check/gate-networking-midonet-python27/1dc1bdf/ : SUCCESS in 7m 13s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/7/check/gate-networking-midonet-python34/a50a49c/ : SUCCESS in 8m 42s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/7/check/gate-tempest-dsvm-networking-midonet-v1/360243e/ : FAILURE in 59m 42s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/7/check/gate-tempest-dsvm-networking-midonet-v2/24d603e/ : SUCCESS in 52m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"efd9a0f797823ef6a4254c899cca4453dfc1d785","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-11-17 07:27:34.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"85008de1daa67d3acc3a01aca5dc32200ff1832c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-17 08:42:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/7/check/gate-networking-midonet-pep8/8f9af17/ : SUCCESS in 4m 35s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/7/check/gate-networking-midonet-docs/d7137e4//doc/build/html/ : SUCCESS in 4m 15s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/7/check/gate-networking-midonet-python27/8ee7412/ : SUCCESS in 7m 36s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/7/check/gate-networking-midonet-python34/7c1439a/ : SUCCESS in 7m 53s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/7/check/gate-tempest-dsvm-networking-midonet-v1/f7a5159/ : SUCCESS in 58m 52s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/7/check/gate-tempest-dsvm-networking-midonet-v2/3f4bd21/ : SUCCESS in 1h 03m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"42a575dcb5686205383b84d022fdd3d5bcbac423","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-11-30 09:15:53.000000000","message":"Patch Set 8: Commit message was updated","accounts_in_message":[],"_revision_number":8},{"id":"3d75a9cdf77bee86039277c560a76742f138d56a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-30 10:40:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-pep8/a7ed45d/ : SUCCESS in 3m 28s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/8/check/gate-networking-midonet-docs/619f588//doc/build/html/ : SUCCESS in 4m 30s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python27/d61a4a1/ : SUCCESS in 8m 11s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python34/376c3f1/ : SUCCESS in 7m 04s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v1/f933f14/ : SUCCESS in 1h 20m 53s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v2/3bfc160/ : FAILURE in 56m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"daa7d56429accbe5d782b1bf38536a807f719459","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-01 02:05:09.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f5fe484901a57a606f43955b226c9f4b222c9f0c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-01 03:41:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-pep8/cb6150c/ : SUCCESS in 4m 15s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/8/check/gate-networking-midonet-docs/f9283c0//doc/build/html/ : SUCCESS in 4m 32s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python27/d170cff/ : SUCCESS in 6m 04s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python34/a76afc1/ : SUCCESS in 8m 38s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v1/e89d4d0/ : SUCCESS in 1h 08m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v2/81087fc/ : FAILURE in 1h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4ba23d227d15a88288194a6b0bd13225f681affe","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-04 02:12:56.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"84a31204cd6ab6bebda966ce815e8a74102ef7a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-04 03:33:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-pep8/5d810e8/ : SUCCESS in 4m 17s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/8/check/gate-networking-midonet-docs/7c6b2e5//doc/build/html/ : SUCCESS in 4m 43s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python27/e128862/ : SUCCESS in 8m 23s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python34/c32837a/ : SUCCESS in 9m 00s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v1/085d58c/ : FAILURE in 1h 10m 47s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v2/30deae6/ : FAILURE in 1h 00m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7451326e2d62d523b4bccdc03b5946270e5c23f6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-07 03:38:26.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"14af5a7385c214ba3da931c212cdb09fd49beaa6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-07 05:20:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-pep8/d5ff4c0/ : SUCCESS in 4m 09s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/8/check/gate-networking-midonet-docs/b2226c2//doc/build/html/ : SUCCESS in 4m 21s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python27/b2a6654/ : SUCCESS in 8m 11s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python34/25215cb/ : SUCCESS in 6m 30s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v1/3ddf458/ : FAILURE in 1h 35m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v2/3fb3f3b/ : FAILURE in 1h 07m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"515b8a0765714da2a68575224ddb08d9e97c0e5e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-08 05:58:33.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f49553a6ebb382312301de4979ac1e7a033a96f6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-08 05:58:43.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"f8a7a0ddab335d879abbf155b0b9194fc823da20","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-08 07:29:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/8/experimental/gate-tempest-dsvm-networking-midonet-ml2/94df01f/ : FAILURE in 1h 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0023aafdbff04d83a324abb53231061b9a3f8001","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-08 07:38:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-pep8/0fb4f63/ : SUCCESS in 5m 08s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/8/check/gate-networking-midonet-docs/9f3d7b6//doc/build/html/ : SUCCESS in 5m 40s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python27/1942403/ : SUCCESS in 8m 34s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/8/check/gate-networking-midonet-python34/eee7fd2/ : SUCCESS in 7m 25s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v1/1382d07/ : FAILURE in 39m 41s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/8/check/gate-tempest-dsvm-networking-midonet-v2/b59966c/ : FAILURE in 1h 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"551ae75a86013e29c59f0b4679cdd990a70d34bc","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-10 08:52:40.000000000","message":"Patch Set 8:\n\ncheck experimental","accounts_in_message":[],"_revision_number":8},{"id":"765561a701ad9dc1bfeedd25a2c6d71942766d3b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-10 09:19:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/8/experimental/gate-tempest-dsvm-networking-midonet-ml2/b0402f1/ : FAILURE in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"87c62008dfa41ddad5e0353b672d445e94073257","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-15 10:41:18.000000000","message":"Patch Set 9: Commit message was updated","accounts_in_message":[],"_revision_number":9},{"id":"a3e4143e0d789b1877d59b6f532a1350280251b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-15 11:33:05.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"a9a42a4f01ac9609a6576f26b87d3fc53ab8bc8a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-15 11:35:48.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"86b90de173ff9f05b49fc73b3effec2c236af8f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-15 13:19:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/9/check/gate-networking-midonet-pep8/ff5e77c/ : SUCCESS in 4m 50s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/9/check/gate-networking-midonet-docs/0d51ee2//doc/build/html/ : SUCCESS in 4m 48s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/9/check/gate-networking-midonet-python27/2b7f01a/ : SUCCESS in 8m 37s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/9/check/gate-networking-midonet-python34/9bcb992/ : SUCCESS in 8m 24s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/9/check/gate-tempest-dsvm-networking-midonet-v1/983bc1e/ : SUCCESS in 1h 18m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/9/check/gate-tempest-dsvm-networking-midonet-v2/34f8711/ : FAILURE in 1h 26m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/9/check/gate-tempest-dsvm-networking-midonet-ml2/59ed972/ : FAILURE in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"35c701de2adc5462e45f12d1e9f293044d2ecd7b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2015-12-15 14:10:09.000000000","message":"Patch Set 10: Commit message was updated","accounts_in_message":[],"_revision_number":10},{"id":"8ad15383eeaabe347f2250de66f11a876a97ccfd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-15 16:19:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-pep8/1570041/ : SUCCESS in 4m 54s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/10/check/gate-networking-midonet-docs/8d534a2//doc/build/html/ : SUCCESS in 5m 04s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python27/52388e4/ : SUCCESS in 8m 42s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python34/582cf1a/ : SUCCESS in 10m 42s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v1/410f8e3/ : FAILURE in 1h 04m 49s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v2/2fa4d37/ : FAILURE in 1h 13m 05s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-ml2/a2a46d6/ : FAILURE in 1h 00m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ee136b513886d20d7e6450e0cc32f5eb0935c6b4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-12 04:52:24.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"beb3797f8d6e8389f3f10a8f912c0076d2a41b21","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-12 06:39:19.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-pep8/3b4138e/ : FAILURE in 5m 38s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/10/check/gate-networking-midonet-docs/2e60daa//doc/build/html/ : SUCCESS in 5m 00s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python27/8b8bd8e/ : SUCCESS in 7m 58s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python34/032fe2b/ : SUCCESS in 23m 11s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v1/639e3f0/ : SUCCESS in 56m 14s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v2/8f35b59/ : SUCCESS in 1h 04m 14s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-ml2/2f5eb95/ : SUCCESS in 1h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a2e6fad7e3cb715798471615eb24020f7dab7da8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-13 03:30:22.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"775b9a75911d101dc65444a89a67b45079aeb0a3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-13 05:47:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-pep8/9c2a8ec/ : SUCCESS in 5m 30s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/10/check/gate-networking-midonet-docs/eda806b//doc/build/html/ : SUCCESS in 5m 41s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python27/67cf8b6/ : SUCCESS in 11m 04s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python34/df18961/ : SUCCESS in 22m 37s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v1/383ecf3/ : SUCCESS in 1h 14m 46s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v2/d5599ac/ : SUCCESS in 1h 14m 03s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-ml2/f2be44a/ : SUCCESS in 1h 33m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"304f47a0b1e1c6f1d5f29e87a99a3ad3d388b9c0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-15 04:01:01.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"d79c0e711704e759cf54b789c2c16305178094dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-15 05:41:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-pep8/eaf1639/ : SUCCESS in 6m 10s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/10/check/gate-networking-midonet-docs/3e47232//doc/build/html/ : SUCCESS in 5m 42s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python27/dbcae24/ : SUCCESS in 11m 25s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python34/d796a9e/ : SUCCESS in 22m 42s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v1/ceda93f/ : FAILURE in 1h 33m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v2/dde22ab/ : SUCCESS in 1h 15m 06s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-ml2/f1032b5/ : SUCCESS in 1h 39m 59s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"901976c43b8a85dbc64f0608c2121d7cb8be8e65","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-18 04:54:10.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"33f5d55c1207dcf9213bfd382d443f1eaecc4e42","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-18 06:52:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-pep8/b077e09/ : SUCCESS in 5m 56s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/10/check/gate-networking-midonet-docs/5cbeea4//doc/build/html/ : SUCCESS in 5m 35s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python27/8ed36e3/ : SUCCESS in 10m 21s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/10/check/gate-networking-midonet-python34/1bf63ea/ : SUCCESS in 32m 12s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v1/338dc73/ : FAILURE in 1h 07m 28s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-v2/62611ac/ : FAILURE in 1h 57m 03s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/10/check/gate-tempest-dsvm-networking-midonet-ml2/70ed5cd/ : SUCCESS in 1h 49m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c7ebf9e860f48f1951bfb3c37602258b1bfee5b1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-18 14:02:31.000000000","message":"Patch Set 11: Published edit on patch set 10","accounts_in_message":[],"_revision_number":11},{"id":"72d2997723a6123ff758cb50651c7ca05e5dec84","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-18 20:32:24.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-pep8/c931021/ : SUCCESS in 5m 46s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/11/check/gate-networking-midonet-docs/b34e38d//doc/build/html/ : SUCCESS in 5m 44s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-python27/de790ab/ : SUCCESS in 7m 39s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-python34/2cd17db/ : SUCCESS in 29m 28s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-v1/178c429/ : FAILURE in 1h 09m 54s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-v2/fe58693/ : FAILURE in 1h 55m 34s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-ml2/9d82657/ : FAILURE in 1h 57m 07s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7d85c103a2e03c3f8d3b3c233827fdaeba1e3cc1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-19 02:08:05.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"afa81eadc147af7fb30825f3eedd9428cd57a40e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-19 05:03:11.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-pep8/4381d82/ : SUCCESS in 6m 03s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/11/check/gate-networking-midonet-docs/da6ab95//doc/build/html/ : SUCCESS in 4m 22s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-python27/f456d0c/ : SUCCESS in 7m 52s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-python34/3dfacfa/ : SUCCESS in 29m 34s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-v1/f734095/ : SUCCESS in 1h 09m 13s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-v2/3610b49/ : FAILURE in 1h 56m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-ml2/60cff92/ : SUCCESS in 1h 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d574e8d8d785aedee2fe560a1d7689002c729b2d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-19 06:58:55.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"e8b4fa10bbe5cee84a25e4b39a29d5fcae40d5ba","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-19 08:54:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-pep8/0fd8e12/ : SUCCESS in 6m 12s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/11/check/gate-networking-midonet-docs/458bece//doc/build/html/ : SUCCESS in 4m 10s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-python27/22cebec/ : SUCCESS in 10m 07s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-python34/926aa8c/ : SUCCESS in 24m 30s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-v1/823aab1/ : SUCCESS in 1h 25m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-v2/c41bf53/ : SUCCESS in 1h 44m 53s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-ml2/b21d1ec/ : SUCCESS in 1h 49m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7abe3d9c55dd061c24ed7158c65ecfd549236a95","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-19 12:01:31.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"8df145bf0cbeed77e733dcd3e669980ff2cdf0d3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-19 15:18:08.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-pep8/b203253/ : SUCCESS in 4m 17s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/11/check/gate-networking-midonet-docs/69b7544//doc/build/html/ : SUCCESS in 4m 57s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-python27/375d5a2/ : SUCCESS in 7m 32s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/11/check/gate-networking-midonet-python34/943757e/ : SUCCESS in 22m 34s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-v1/a7bad30/ : FAILURE in 1h 22m 59s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-v2/41c685b/ : FAILURE in 1h 59m 43s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/11/check/gate-tempest-dsvm-networking-midonet-ml2/b10bd74/ : SUCCESS in 1h 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"cd5f3fccce8d78cb2109b57cc26d6ddec8af62f3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-20 04:03:54.000000000","message":"Patch Set 12: Published edit on patch set 11","accounts_in_message":[],"_revision_number":12},{"id":"696b414cd3e3aa6d628ed3d78d8cd205d031d517","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-20 04:42:35.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-pep8/7d9ad33/ : SUCCESS in 4m 22s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/12/check/gate-networking-midonet-docs/ece71c6//doc/build/html/ : SUCCESS in 5m 26s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-python27/83b74ae/ : SUCCESS in 10m 02s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-python34/8f76a38/ : SUCCESS in 25m 27s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-v1/d157820/ : FAILURE in 6m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-v2/e3df040/ : FAILURE in 9m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-ml2/26c3477/ : FAILURE in 7m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"290371707bd3ac578f05144d5278325d6226a98b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-21 09:47:15.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"5b69b1209348c231a6135086e33316f486982c4a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-21 11:57:32.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-pep8/fd9c5af/ : SUCCESS in 5m 05s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/12/check/gate-networking-midonet-docs/da38364//doc/build/html/ : SUCCESS in 6m 09s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-python27/0f15579/ : SUCCESS in 8m 40s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-python34/8992d21/ : SUCCESS in 24m 56s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-v1/82fabbc/ : FAILURE in 1h 21m 05s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-v2/be9191c/ : FAILURE in 1h 55m 40s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-ml2/e132a38/ : FAILURE in 1h 57m 22s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e0cd94f497812f2c6d56f7a02cf4b0adcfca62de","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-25 07:44:15.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"20ba7b15e9ca10178dc5e6b96c12d2aa3065e22f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-25 08:39:46.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-pep8/e207d20/ : SUCCESS in 5m 41s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/12/check/gate-networking-midonet-docs/270429c//doc/build/html/ : SUCCESS in 5m 44s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-python27/cac9b3e/ : FAILURE in 10m 38s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/12/check/gate-networking-midonet-python34/fc82227/ : FAILURE in 35m 15s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-v1/787cab5/ : FAILURE in 48m 24s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-v2/4216fa6/ : FAILURE in 51m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/12/check/gate-tempest-dsvm-networking-midonet-ml2/5cfc51a/ : FAILURE in 48m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"71edf226c7efa9e305dbee12ac2f5707a3aca071","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-25 12:58:18.000000000","message":"Patch Set 13: Published edit on patch set 12","accounts_in_message":[],"_revision_number":13},{"id":"3b15f13fd34752e1d245d9a87a7daad047a6954a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-25 14:51:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/13/check/gate-networking-midonet-pep8/da91fe1/ : SUCCESS in 4m 33s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/13/check/gate-networking-midonet-docs/66495ee//doc/build/html/ : SUCCESS in 4m 41s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/13/check/gate-networking-midonet-python27/1b69a3d/ : SUCCESS in 10m 28s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/13/check/gate-networking-midonet-python34/58da54d/ : SUCCESS in 29m 41s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/13/check/gate-tempest-dsvm-networking-midonet-v1/5500ea9/ : FAILURE in 1h 00m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/13/check/gate-tempest-dsvm-networking-midonet-v2/f1bac7a/ : FAILURE in 1h 46m 20s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/13/check/gate-tempest-dsvm-networking-midonet-ml2/825fe97/ : FAILURE in 1h 42m 21s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4aa4df528e66bed556def380a21e12c3ee908969","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-25 21:44:59.000000000","message":"Patch Set 14: Published edit on patch set 13","accounts_in_message":[],"_revision_number":14},{"id":"ab05adf6bb54ac73ada9e45275ffec1d8a4306ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-25 22:57:14.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/14/check/gate-networking-midonet-pep8/52be561/ : SUCCESS in 6m 17s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/14/check/gate-networking-midonet-docs/28df5dd//doc/build/html/ : SUCCESS in 6m 19s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/14/check/gate-networking-midonet-python27/be091ab/ : FAILURE in 8m 24s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/14/check/gate-networking-midonet-python34/d78cbc2/ : FAILURE in 27m 38s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/14/check/gate-tempest-dsvm-networking-midonet-v1/068264d/ : FAILURE in 1h 03m 50s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/14/check/gate-tempest-dsvm-networking-midonet-v2/c69a0d8/ : FAILURE in 1h 02m 23s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/14/check/gate-tempest-dsvm-networking-midonet-ml2/7b77adf/ : FAILURE in 1h 05m 26s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6fee0a10029b3b82df8a87a2a50c059f11154e29","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-25 23:58:39.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"071ced26fc4dc92b3550fcb96f01c87c0283ec2a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-26 01:22:17.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/14/check/gate-networking-midonet-pep8/211a375/ : SUCCESS in 5m 46s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/14/check/gate-networking-midonet-docs/c5a96d9//doc/build/html/ : SUCCESS in 5m 28s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/14/check/gate-networking-midonet-python27/7f05c03/ : SUCCESS in 8m 44s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/14/check/gate-networking-midonet-python34/5321409/ : SUCCESS in 26m 28s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/14/check/gate-tempest-dsvm-networking-midonet-v1/9aeb490/ : SUCCESS in 58m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/14/check/gate-tempest-dsvm-networking-midonet-v2/6487074/ : FAILURE in 1h 14m 10s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/14/check/gate-tempest-dsvm-networking-midonet-ml2/846fa3e/ : FAILURE in 1h 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2b9683557c00ba15b97de27e5366e75550b85175","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-26 09:06:49.000000000","message":"Patch Set 15: Published edit on patch set 14","accounts_in_message":[],"_revision_number":15},{"id":"741a09678ecd2ba17f35be6e2fc03c483f0a2418","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-26 10:41:20.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/15/check/gate-networking-midonet-pep8/442b9e6/ : SUCCESS in 4m 57s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/15/check/gate-networking-midonet-docs/9884006//doc/build/html/ : SUCCESS in 4m 23s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/15/check/gate-networking-midonet-python27/ecad645/ : SUCCESS in 7m 44s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/15/check/gate-networking-midonet-python34/f426102/ : SUCCESS in 30m 27s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/15/check/gate-tempest-dsvm-networking-midonet-v1/0cfb577/ : SUCCESS in 58m 48s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/15/check/gate-tempest-dsvm-networking-midonet-v2/7376a21/ : FAILURE in 1h 29m 16s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/15/check/gate-tempest-dsvm-networking-midonet-ml2/842567f/ : FAILURE in 1h 22m 38s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"59ef4b3c95c38a2c3692418a2473fc56eb4e2837","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-27 12:07:29.000000000","message":"Patch Set 16: Published edit on patch set 15","accounts_in_message":[],"_revision_number":16},{"id":"b5c63b2e6028d75f2f4763cb75809c74c8b760f0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-27 12:37:54.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"55f06949549a577484732f4ad62ac6a9637421cb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-27 13:52:43.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/16/check/gate-networking-midonet-pep8/2be7298/ : SUCCESS in 5m 02s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/16/check/gate-networking-midonet-docs/53227d1//doc/build/html/ : SUCCESS in 6m 00s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/16/check/gate-networking-midonet-python27/e63a39b/ : SUCCESS in 8m 35s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/16/check/gate-networking-midonet-python34/eefc53b/ : SUCCESS in 24m 25s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/16/check/gate-tempest-dsvm-networking-midonet-v1/d4c826a/ : SUCCESS in 57m 41s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/16/check/gate-tempest-dsvm-networking-midonet-v2/32ea87b/ : FAILURE in 1h 38m 03s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/16/check/gate-tempest-dsvm-networking-midonet-ml2/9fb5055/ : FAILURE in 1h 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d7fa2ac4c2b0837f9391cf8da15216d2a89c72a9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-27 14:17:19.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"9d94354b84330bcb26b2cfe72b9eb216d4d4b459","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-27 20:13:16.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/16/check/gate-networking-midonet-pep8/a0d50fc/ : SUCCESS in 4m 57s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/16/check/gate-networking-midonet-docs/cb3039b//doc/build/html/ : SUCCESS in 4m 49s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/16/check/gate-networking-midonet-python27/de5f91d/ : SUCCESS in 8m 58s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/16/check/gate-networking-midonet-python34/d70ccc1/ : SUCCESS in 28m 03s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/16/check/gate-tempest-dsvm-networking-midonet-v1/3c8e979/ : SUCCESS in 58m 43s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/16/check/gate-tempest-dsvm-networking-midonet-v2/37160c0/ : FAILURE in 1h 31m 27s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/16/check/gate-tempest-dsvm-networking-midonet-ml2/eaaeca7/ : FAILURE in 1h 30m 28s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"3893fc6b6fd0d27abc1a21b2754837fdba36eb64","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-28 09:55:20.000000000","message":"Patch Set 17: Published edit on patch set 16","accounts_in_message":[],"_revision_number":17},{"id":"06674b4dd36a51884be9e3d6923c0f29831a9103","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-28 12:00:42.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-pep8/c58e2d9/ : SUCCESS in 5m 02s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/17/check/gate-networking-midonet-docs/0faa8ab//doc/build/html/ : SUCCESS in 5m 08s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-python27/c6210e8/ : SUCCESS in 8m 55s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-python34/3a88225/ : SUCCESS in 24m 39s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-v1/1cda39f/ : SUCCESS in 56m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-v2/c1a1e04/ : FAILURE in 25m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-ml2/4a2afcf/ : FAILURE in 1h 56m 55s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"24833d50b6cbc7c77efee4d003355ffde2efa9f6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-01-28 13:51:57.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"2945977cd3320e45797f4f6830675fe5bceb9489","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-28 22:19:42.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-pep8/1ceac55/ : SUCCESS in 4m 58s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/17/check/gate-networking-midonet-docs/6721aa3//doc/build/html/ : SUCCESS in 4m 56s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-python27/79a217b/ : SUCCESS in 8m 29s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-python34/4162000/ : SUCCESS in 26m 32s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-v1/01d44cc/ : SUCCESS in 1h 11m 55s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-v2/d41cc19/ : FAILURE in 1h 33m 40s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-ml2/f287662/ : FAILURE in 1h 36m 33s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d0c41aee4b8f59984877b11c711ceca7d9818c92","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-01 09:34:22.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"f9f08c03b76137547a2b498a93ca5c9266191044","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-01 11:36:59.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-pep8/549c46e/ : SUCCESS in 5m 25s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/17/check/gate-networking-midonet-docs/e995acc//doc/build/html/ : SUCCESS in 5m 50s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-python27/bf1813d/ : SUCCESS in 9m 09s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-python34/bd2a30b/ : SUCCESS in 25m 14s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-v1/cb73df8/ : SUCCESS in 1h 00m 11s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-v2/905430d/ : FAILURE in 1h 37m 28s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-ml2/bcaa360/ : FAILURE in 1h 57m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"0466a15523164e2514c49a0f6f32ac71c096a0d7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-01 14:43:13.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"8c688fa6fea42e867711eeae400774d4384aefdc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-01 16:39:50.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-pep8/2348d1e/ : SUCCESS in 4m 41s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/17/check/gate-networking-midonet-docs/9e90cfd//doc/build/html/ : SUCCESS in 5m 09s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-python27/266c6ff/ : SUCCESS in 8m 33s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/17/check/gate-networking-midonet-python34/3da6563/ : SUCCESS in 24m 33s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-v1/3f64160/ : FAILURE in 56m 40s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-v2/a8e4eb8/ : FAILURE in 1h 53m 22s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/17/check/gate-tempest-dsvm-networking-midonet-ml2/078dd50/ : FAILURE in 1h 45m 45s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7af7afc3e1b613923ad3500cfbaa12c7f27c43d9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-02 03:11:51.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"5d67789a4e268adb4f4f71bf1e0512fbf7e9cc15","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-02 05:26:41.000000000","message":"Patch Set 18: Published edit on patch set 17","accounts_in_message":[],"_revision_number":18},{"id":"d72962c79024edf3d6699a246c4f342fbafd57d2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-02 09:43:48.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/457e169/ : SUCCESS in 5m 43s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/1bd80b5//doc/build/html/ : SUCCESS in 5m 11s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/5f2a3b3/ : SUCCESS in 8m 17s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/ba19286/ : SUCCESS in 25m 01s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/1f92033/ : FAILURE in 53m 32s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/e1d0c24/ : SUCCESS in 1h 37m 10s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/da39ac4/ : SUCCESS in 1h 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"a1472b7f9ff31e43ac72505a304f99290208fb57","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-04 03:13:13.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"d493c836e175c5021fad9ca9ceda1324ee91ba47","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-04 06:38:52.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/cf8d103/ : SUCCESS in 10m 17s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/7d1f8ad//doc/build/html/ : SUCCESS in 5m 31s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/6a4d34f/ : SUCCESS in 9m 31s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/89721ce/ : SUCCESS in 26m 22s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/ff311ed/ : FAILURE in 59m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/fb73b50/ : SUCCESS in 1h 14m 13s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/8b998a7/ : SUCCESS in 1h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"660163b01598a22bb37b6ccc37a6ce1cc087cd00","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-04 09:16:30.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"13e039ea0b473ee7f1f9b52c6b7e7c7bc6759e61","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-04 15:16:01.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/0967bf3/ : SUCCESS in 5m 05s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/c3dfc67//doc/build/html/ : SUCCESS in 5m 32s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/6aa56f8/ : SUCCESS in 9m 02s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/8e6fa2d/ : SUCCESS in 25m 41s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/a29f909/ : FAILURE in 59m 18s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/a23ed93/ : FAILURE in 1h 56m 42s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/3da3bbd/ : SUCCESS in 1h 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e024fa2372f3b01861236a5287d17d38565a1b20","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-05 02:42:01.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"77e883ef8b9a93ac884671c1d89d663aad83e1e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-05 04:27:29.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/experimental/gate-networking-midonet-v2-dsvm-rally-nv/59a15f0/ : FAILURE in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"fd3ea676b8e728015cc6f6234ff863ef1933d43c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-08 13:21:21.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"96e433dc0b363b372a2c99c99b653e06f63960c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-08 14:57:06.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/8c12505/ : SUCCESS in 4m 09s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/93e77df//doc/build/html/ : SUCCESS in 4m 30s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/2505be7/ : SUCCESS in 7m 47s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/6cb9077/ : SUCCESS in 23m 32s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/b19d29f/ : SUCCESS in 49m 05s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/4344770/ : FAILURE in 1h 23m 44s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/132739b/ : FAILURE in 1h 07m 53s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0f647255ce27254f74edebc1bf9bcfed7cc43ef8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-10 09:53:25.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"80e5515e64b09b1931f64880b28770514266f035","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-10 10:41:09.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/experimental/gate-networking-midonet-v2-dsvm-rally-nv/4594ff0/ : SUCCESS in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"92f15b5718d309bff7bd94719059a115e85f6da2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-12 01:48:16.000000000","message":"Patch Set 18:\n\ncheck experimental","accounts_in_message":[],"_revision_number":18},{"id":"af880762c5cfead693da3954d927c49899939a59","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-12 02:37:06.000000000","message":"Patch Set 18:\n\nBuild succeeded (experimental pipeline).\n\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/experimental/gate-networking-midonet-v2-dsvm-rally-nv/3dc0d15/ : SUCCESS in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"f047aa9c838c96be35c04f767410984f48a51821","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-15 03:42:25.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"6e7d01a259514b344533c322a65b9d732ffe6ded","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-15 04:38:19.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/fd5b7c8/ : SUCCESS in 3m 59s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/bb94739//doc/build/html/ : SUCCESS in 4m 12s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/64a5418/ : SUCCESS in 7m 53s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/98cf4bf/ : SUCCESS in 24m 20s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/2493f29/ : SUCCESS in 33m 42s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/e302cbb/ : SUCCESS in 55m 47s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/824415f/ : SUCCESS in 49m 28s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"286c963edb9507749439505b97ffe4ba2253d28d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-16 01:52:31.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"11a49f621c015f4ac97891648767aa04e7f79998","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-16 03:09:16.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/c409731/ : SUCCESS in 4m 37s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/c252baf//doc/build/html/ : SUCCESS in 4m 48s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/12e00a0/ : SUCCESS in 8m 30s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/cdc07df/ : SUCCESS in 26m 02s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/6a5faa1/ : SUCCESS in 37m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/e2872a2/ : SUCCESS in 1h 16m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/f5e3a4d/ : SUCCESS in 58m 35s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-v2-dsvm-rally-nv/68ead3e/ : SUCCESS in 48m 21s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"90fee415b3342e4bc6cf4fdcc69666f67b244c42","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-16 10:14:46.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"8af2a54be08c31b2dfd3887388619186321db2bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-16 11:20:05.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/09eff78/ : SUCCESS in 3m 58s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/5fd8be1//doc/build/html/ : SUCCESS in 4m 36s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/0b073a7/ : SUCCESS in 7m 47s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/936846c/ : SUCCESS in 25m 43s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/8f8ec7a/ : SUCCESS in 36m 17s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/1c4a694/ : SUCCESS in 57m 16s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/de58ad7/ : SUCCESS in 1h 03m 31s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-v2-dsvm-rally-nv/5ac2154/ : SUCCESS in 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3161ed015fa81b9feea7174d8b38847355f27654","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-18 05:32:41.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"1bd712f3a97b339acb66acd67219eb7a53b81b18","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-18 06:33:42.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/f852af2/ : SUCCESS in 3m 59s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/929eb24//doc/build/html/ : SUCCESS in 4m 02s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/74f8200/ : SUCCESS in 7m 42s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/91283de/ : SUCCESS in 26m 33s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/e767060/ : SUCCESS in 34m 49s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/61fe39d/ : SUCCESS in 1h 00m 03s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/d787711/ : SUCCESS in 1h 00m 54s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-v2-dsvm-rally-nv/6754f1e/ : SUCCESS in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"cb3cc0aaad1b9b3b3a27518329622adecafc89ab","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-24 01:30:11.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"8fbe8f593968ff5484cdaa0cc66abcea5848a5ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-24 02:20:00.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/7833d31/ : SUCCESS in 4m 13s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/92cb75e//doc/build/html/ : SUCCESS in 4m 05s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/18120cf/ : SUCCESS in 7m 55s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/a0702d4/ : SUCCESS in 28m 13s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/d4590c9/ : SUCCESS in 49m 40s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/72b26f5/ : FAILURE in 9m 34s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/66929a1/ : FAILURE in 9m 34s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-v2-dsvm-rally-nv/e0d9ef4/ : FAILURE in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"d6ad32f769c51372abee08347ba4b1eb1408b769","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-24 06:43:56.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"b58c8db7e4550c0acc2f9ae104314ff503be98bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-24 07:49:17.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/4d6951f/ : SUCCESS in 4m 06s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/65ce7a7//doc/build/html/ : SUCCESS in 4m 05s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/f51aac5/ : SUCCESS in 7m 30s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/6946d32/ : SUCCESS in 24m 44s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/ddf62c9/ : SUCCESS in 36m 26s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/7f488f6/ : SUCCESS in 1h 04m 19s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/1c749b0/ : SUCCESS in 1h 03m 56s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-v2-dsvm-rally-nv/c7703be/ : SUCCESS in 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"549ae92ccf394e365e23ebb96683916e382f6f24","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-02-29 05:02:49.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"ac67b49c3bfe67f8b8075eb69e27d9d4e7c403a0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-29 11:26:56.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-pep8/c0f312d/ : SUCCESS in 4m 29s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/18/check/gate-networking-midonet-docs/a84757f//doc/build/html/ : SUCCESS in 4m 04s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python27/4e1f2ae/ : SUCCESS in 8m 10s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-python34/2ce11c8/ : SUCCESS in 26m 45s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v1/1f2c822/ : SUCCESS in 42m 05s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-v2/e3b8cdb/ : FAILURE in 1h 10m 18s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/18/check/gate-tempest-dsvm-networking-midonet-ml2/1c00603/ : FAILURE in 1h 09m 09s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/18/check/gate-networking-midonet-v2-dsvm-rally-nv/51b5877/ : SUCCESS in 44m 47s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6710ded56c6a3a1df187288b68b29ec6a7bf9d24","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-01 13:21:05.000000000","message":"Patch Set 19: Published edit on patch set 18","accounts_in_message":[],"_revision_number":19},{"id":"febc7a783423baf428e40efd09ffcf1d52c812bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-01 14:58:32.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-pep8/f246667/ : SUCCESS in 4m 00s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/19/check/gate-networking-midonet-docs/7b02a9e//doc/build/html/ : SUCCESS in 4m 00s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-python27/ca59d1c/ : FAILURE in 7m 45s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-python34/0ad907f/ : SUCCESS in 30m 14s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-v1/6f9e0e9/ : SUCCESS in 40m 06s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-v2/7ef2fc5/ : SUCCESS in 1h 15m 16s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-ml2/a4a3d5d/ : SUCCESS in 1h 08m 33s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-v2-dsvm-rally-nv/bf2c828/ : SUCCESS in 39m 55s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"4df76721b79518304a44141b37f6c22ca1e3fa94","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-02 03:28:18.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"811ace7a1c66c6fa685d7ee93b84f7db9da23c0e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-02 04:42:12.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-pep8/925b886/ : SUCCESS in 4m 25s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/19/check/gate-networking-midonet-docs/bdd32cd//doc/build/html/ : SUCCESS in 4m 58s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-python27/eb9aaa7/ : FAILURE in 10m 13s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-python34/6d18aff/ : SUCCESS in 26m 24s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-v1/891c827/ : SUCCESS in 46m 04s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-v2/7b5f41f/ : SUCCESS in 1h 13m 46s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-ml2/55ba68e/ : SUCCESS in 1h 04m 28s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-v2-dsvm-rally-nv/e916353/ : SUCCESS in 41m 33s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"fa113ec77b6b9d1e18c42a33154caa676dec9555","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-02 07:16:00.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"edd9c47ec864847135ce37e6c3bde9e653086794","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-02 08:13:50.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-pep8/8216723/ : SUCCESS in 4m 23s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/19/check/gate-networking-midonet-docs/51808d9//doc/build/html/ : SUCCESS in 4m 29s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-python27/963668c/ : FAILURE in 9m 39s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-python34/4408dc8/ : SUCCESS in 26m 23s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-v1/73f0fd2/ : FAILURE in 34m 16s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-v2/b56de2f/ : FAILURE in 27m 34s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/19/check/gate-tempest-dsvm-networking-midonet-ml2/eb07279/ : FAILURE in 27m 21s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/19/check/gate-networking-midonet-v2-dsvm-rally-nv/ca4acc3/ : FAILURE in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ed8e2df1e2e58e75e33b965d316e22fd79d55317","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-03 03:32:08.000000000","message":"Patch Set 20: Published edit on patch set 19","accounts_in_message":[],"_revision_number":20},{"id":"dd388e6ebdafad43f33cb0a2aef4eecd5d4139ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-03 04:12:47.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/20/check/gate-networking-midonet-pep8/5eef3b7/ : SUCCESS in 3m 52s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/20/check/gate-networking-midonet-docs/6502ed2//doc/build/html/ : SUCCESS in 4m 10s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/20/check/gate-networking-midonet-python27/f5cf3e6/ : FAILURE in 8m 54s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/20/check/gate-networking-midonet-python34/279dd42/ : SUCCESS in 29m 57s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/20/check/gate-tempest-dsvm-networking-midonet-v1/370c72d/ : FAILURE in 24m 35s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/20/check/gate-tempest-dsvm-networking-midonet-v2/f4a9de4/ : FAILURE in 26m 08s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/20/check/gate-tempest-dsvm-networking-midonet-ml2/2d6f461/ : FAILURE in 25m 09s (non-voting)\n- gate-networking-midonet-v2-dsvm-rally-nv http://logs.openstack.org/87/199387/20/check/gate-networking-midonet-v2-dsvm-rally-nv/bb70e03/ : FAILURE in 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ccf158e09fd2c0504eb0d539ff8bd5b7ee085b3a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-07 03:49:20.000000000","message":"Patch Set 21: Published edit on patch set 20","accounts_in_message":[],"_revision_number":21},{"id":"79bfd2de5e92609da9bc23a71c514d6c10a6b8fa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 04:56:37.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/21/check/gate-networking-midonet-pep8/ad171e8/ : SUCCESS in 4m 10s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/21/check/gate-networking-midonet-docs/d644e00//doc/build/html/ : SUCCESS in 2m 55s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/21/check/gate-networking-midonet-python27/a7f16d3/ : SUCCESS in 7m 34s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/21/check/gate-networking-midonet-python34/0e3da04/ : SUCCESS in 23m 49s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/21/check/gate-tempest-dsvm-networking-midonet-v1/28ef4eb/ : FAILURE in 45m 21s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/21/check/gate-tempest-dsvm-networking-midonet-v2/dad6efe/ : FAILURE in 1h 07m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/21/check/gate-tempest-dsvm-networking-midonet-ml2/3d63e9e/ : SUCCESS in 1h 02m 25s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/21/check/gate-networking-midonet-v2-rally-dsvm-nv/bd0a65d/ : FAILURE in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f77f154b523254b67e60c456a2559149360c8039","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-07 07:57:47.000000000","message":"Patch Set 21:\n\nrally job failure is due to https://review.openstack.org/#/c/284149/","accounts_in_message":[],"_revision_number":21},{"id":"2fef9b4ead538c53a62e332ae5f29e089daf5bcc","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-07 07:58:46.000000000","message":"Patch Set 22: Published edit on patch set 21","accounts_in_message":[],"_revision_number":22},{"id":"a0a0649908609acb783c0c1b0443d193c2931ac9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-07 09:18:16.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/22/check/gate-networking-midonet-pep8/129a416/ : SUCCESS in 5m 32s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/22/check/gate-networking-midonet-docs/210b8f1//doc/build/html/ : SUCCESS in 4m 15s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/22/check/gate-networking-midonet-python27/2cb70fe/ : SUCCESS in 7m 56s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/22/check/gate-networking-midonet-python34/f24786c/ : SUCCESS in 28m 20s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/22/check/gate-tempest-dsvm-networking-midonet-v1/b31cde2/ : SUCCESS in 40m 13s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/22/check/gate-tempest-dsvm-networking-midonet-v2/d5c6dc7/ : SUCCESS in 1h 06m 31s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/22/check/gate-tempest-dsvm-networking-midonet-ml2/e5d37e3/ : SUCCESS in 1h 05m 43s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/22/check/gate-networking-midonet-v2-rally-dsvm-nv/d0754bb/ : FAILURE in 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"58552d6a7dbb164ff6ae49e2e35e1b7753714c22","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-13 23:56:58.000000000","message":"Patch Set 23: Published edit on patch set 22","accounts_in_message":[],"_revision_number":23},{"id":"97805de850d5de807372dae4c240c5e0d000cf6d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-14 00:57:28.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/7611338/ : SUCCESS in 6m 21s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/23/check/gate-networking-midonet-docs/5ce5ef5//doc/build/html/ : SUCCESS in 5m 55s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/2c760aa/ : SUCCESS in 10m 02s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/3025d02/ : SUCCESS in 20m 36s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/df90727/ : SUCCESS in 40m 34s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/15ce531/ : SUCCESS in 1h 00m 22s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/8639889/ : SUCCESS in 59m 36s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/91be18a/ : SUCCESS in 52m 36s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"4b1e7a3670806ea5f46cc9f5cf45c180ea7f1340","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-18 05:55:13.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"f7b8218557575280b33ef768f6f906898485ae84","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-18 06:49:30.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/cba1e3f/ : SUCCESS in 2m 56s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/23/check/gate-networking-midonet-docs/ad8cd3e//doc/build/html/ : SUCCESS in 2m 55s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/c4697a0/ : SUCCESS in 41m 41s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/568d37d/ : SUCCESS in 20m 13s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/587d62c/ : FAILURE in 17m 29s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/ea7c345/ : FAILURE in 46m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/075f53c/ : FAILURE in 22m 14s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/ff0a0f4/ : FAILURE in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"146aa92a5705c6bb574171cfb4cfef2a31fd0b4e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-18 09:36:55.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"7cf4cc88977c34c0c7a9c18812b7557e79b0ea08","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-18 10:09:04.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/70d4e6c/ : SUCCESS in 5m 51s\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/23/check/gate-networking-midonet-docs/2853e98//doc/build/html/ : SUCCESS in 6m 49s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/5bf827e/ : SUCCESS in 11m 27s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/19dfa84/ : SUCCESS in 25m 42s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/d01bca8/ : FAILURE in 18m 03s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/70508ad/ : FAILURE in 10m 20s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/bc4d474/ : FAILURE in 16m 00s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/06fc7a4/ : FAILURE in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"dec61f49744a18130b075d2e198a8f8b174fe419","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-24 07:22:10.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"4c2b2eec3f2572283ebad48f07453ed728850a72","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-24 08:30:26.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/23/check/gate-networking-midonet-docs/102994b//doc/build/html/ : SUCCESS in 4m 51s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/41a1ac6/ : SUCCESS in 4m 46s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/c28b792/ : SUCCESS in 9m 04s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/578b3f2/ : SUCCESS in 32m 02s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/81fdb76/ : FAILURE in 31m 46s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/bae5a67/ : SUCCESS in 1h 07m 25s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/5e81829/ : SUCCESS in 52m 05s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/e70a371/ : SUCCESS in 56m 38s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"94dd8326fc6943e8733491c827194a3f78dc38d6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-28 06:29:37.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"94008b34558b1a46edd74b3daeb3091d331b2388","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-28 07:30:58.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/23/check/gate-networking-midonet-docs/ad7891e//doc/build/html/ : SUCCESS in 6m 04s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/450b731/ : SUCCESS in 3m 14s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/1a5b8e8/ : SUCCESS in 9m 22s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/c173de6/ : SUCCESS in 31m 35s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/4804f6a/ : FAILURE in 30m 57s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/c2f602a/ : SUCCESS in 55m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/6ffa176/ : SUCCESS in 59m 35s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/47a5eae/ : SUCCESS in 52m 51s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"7a4316e976fde89c5d0fd7b45fb776e49420e897","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-03-31 05:08:15.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"4930cc628e194e93ab8efe89150e029d4a564214","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-31 06:18:10.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/23/check/gate-networking-midonet-docs/3a7a8ac//doc/build/html/ : SUCCESS in 5m 18s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/9470d08/ : SUCCESS in 3m 13s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/eebc4a6/ : SUCCESS in 7m 48s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/6ece432/ : SUCCESS in 24m 25s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/3f2ad44/ : FAILURE in 32m 06s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/a2fc935/ : SUCCESS in 1h 08m 51s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/19fa80f/ : SUCCESS in 1h 03m 36s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/e17be2e/ : SUCCESS in 53m 11s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"e54e924b1a16647e0a7ada957a9382b499ed3ffc","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-05 02:43:24.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"9190370df99b7ee711e51d3e8b10535f4f05570b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-05 03:51:56.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/23/check/gate-networking-midonet-docs/4dac71d//doc/build/html/ : SUCCESS in 2m 56s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/439a18c/ : SUCCESS in 4m 35s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/6fdca9d/ : FAILURE in 10m 32s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/6df59a4/ : SUCCESS in 33m 43s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/72ab035/ : SUCCESS in 59m 47s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/71bf21a/ : SUCCESS in 56m 01s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/1481676/ : SUCCESS in 1h 08m 24s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/755f3d9/ : SUCCESS in 58m 05s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"1caa51fb54824c3d9471d6fff1b7be38eb6e173c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-11 14:49:08.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"bb1e2c723c39af42c00114794b7e89fecfab337d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-11 16:06:22.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/23/check/gate-networking-midonet-docs/790fbf5//doc/build/html/ : SUCCESS in 3m 56s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/8479000/ : SUCCESS in 3m 44s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/fb69665/ : SUCCESS in 7m 26s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/7a41474/ : SUCCESS in 28m 58s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/aa6b968/ : SUCCESS in 29m 58s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/0ff0db9/ : SUCCESS in 1h 07m 07s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/ec5f34c/ : SUCCESS in 1h 03m 34s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/a07399a/ : FAILURE in 1h 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"eefcf0094f48c18bec5943cf40bb548221d7221a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-13 02:24:26.000000000","message":"Patch Set 23:\n\nrecheck","accounts_in_message":[],"_revision_number":23},{"id":"d58e32876aaad4ec93711609a380d91d6e750476","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-13 03:32:16.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-docs/dfacc50/ : FAILURE in 3m 56s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-pep8/c85592d/ : SUCCESS in 4m 12s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python27/4a9fccd/ : SUCCESS in 8m 07s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-python34/ad9afb7/ : SUCCESS in 34m 19s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v1/6507a4e/ : SUCCESS in 38m 22s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-v2/8b06ca6/ : FAILURE in 1h 07m 44s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/23/check/gate-tempest-dsvm-networking-midonet-ml2/25f9f5b/ : SUCCESS in 1h 05m 17s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/23/check/gate-networking-midonet-v2-rally-dsvm-nv/6240c37/ : FAILURE in 56m 42s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"2826adfea61a212f00a41dc1403b4fdf64932c08","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-14 09:52:32.000000000","message":"Patch Set 24: Published edit on patch set 23","accounts_in_message":[],"_revision_number":24},{"id":"7d40b005aafc823562ee27063a42440a2fc011bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-14 10:05:16.000000000","message":"Patch Set 24: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":24},{"id":"27ea694b43efd42a4f91d9780bbeac28e69d75af","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-14 16:55:34.000000000","message":"Patch Set 25: Published edit on patch set 24","accounts_in_message":[],"_revision_number":25},{"id":"a4f748e0b79bd073bd0622d080a911e0f7ab439a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-14 20:13:53.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/25/check/gate-networking-midonet-docs/3fb63c1//doc/build/html/ : SUCCESS in 2m 47s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-pep8/5e1ab58/ : SUCCESS in 4m 45s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python27/38395b6/ : SUCCESS in 10m 39s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python34/5cebf21/ : SUCCESS in 26m 46s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v1/8a9c7f7/ : FAILURE in 1h 57m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v2/7787b8c/ : FAILURE in 1h 56m 01s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-ml2/d781671/ : FAILURE in 1h 57m 22s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-v2-rally-dsvm-nv/dc5a1d4/ : FAILURE in 1h 04m 38s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"4ac588721cfca935d75a63d672929993510fadf2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-15 03:18:34.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"de17473cf6cdab92c38df9066cc11e2a26267e82","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-15 05:14:38.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/25/check/gate-networking-midonet-docs/2affb66//doc/build/html/ : SUCCESS in 3m 52s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-pep8/813d3c1/ : SUCCESS in 4m 01s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python27/cfefa9d/ : SUCCESS in 7m 36s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python34/cd00350/ : SUCCESS in 31m 09s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v1/f2f1467/ : FAILURE in 1h 55m 48s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v2/59ddfc9/ : FAILURE in 1h 55m 53s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-ml2/6b5f8f4/ : FAILURE in 1h 55m 48s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-v2-rally-dsvm-nv/4333fe0/ : FAILURE in 56m 06s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"e0c6ce6b8e741436b4c680cbb2a28d98de51326c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-17 10:46:04.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"a4a7ec8b443ce738ce887cf907710c6eb14d0307","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-17 12:43:50.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/25/check/gate-networking-midonet-docs/6f775ec//doc/build/html/ : SUCCESS in 4m 52s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-pep8/da7902c/ : SUCCESS in 6m 58s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python27/5760666/ : SUCCESS in 11m 53s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python34/1009854/ : SUCCESS in 32m 51s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v1/28549d7/ : FAILURE in 1h 57m 39s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v2/d72e49c/ : FAILURE in 1h 56m 25s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-ml2/4d9e8ff/ : FAILURE in 1h 56m 33s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-v2-rally-dsvm-nv/c5eac14/ : SUCCESS in 52m 10s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/25/check/networking-midonet-coverage/e82b7f6/ : FAILURE in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"2a7dbe7f6e6b0e0116a21f69f4d92b33887bfd06","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-18 11:08:29.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"303e269fe150bc9820b4c9ad3ed71fedbea1eb82","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-18 13:23:27.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-docs/397ce54/ : FAILURE in 8m 37s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-pep8/e0794a9/ : SUCCESS in 3m 39s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python27/aac482b/ : FAILURE in 6m 29s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python34/f8f6985/ : FAILURE in 6m 44s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v1/1ead1d5/ : FAILURE in 1h 56m 19s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v2/18035d4/ : FAILURE in 1h 55m 36s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-ml2/3f9e52a/ : FAILURE in 1h 55m 54s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-v2-rally-dsvm-nv/d0e7151/ : FAILURE in 49m 27s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/25/check/networking-midonet-coverage/b799d72/ : FAILURE in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"b701f969d46d4f7d8090b27060564512b35a3e0d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-18 19:30:20.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"6eba71f66b83b182ee50831a7ac7993135b667e5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-19 09:28:26.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/25/check/gate-networking-midonet-docs/8ddc212//doc/build/html/ : SUCCESS in 3m 48s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-pep8/a401c1f/ : SUCCESS in 4m 01s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python27/ac3cfa4/ : SUCCESS in 9m 03s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-python34/de45afb/ : SUCCESS in 28m 24s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v1/26677f4/ : FAILURE in 1h 56m 08s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-v2/db94e7e/ : FAILURE in 1h 56m 16s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/25/check/gate-tempest-dsvm-networking-midonet-ml2/867fe9d/ : FAILURE in 1h 55m 47s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/25/check/gate-networking-midonet-v2-rally-dsvm-nv/d03029f/ : SUCCESS in 51m 48s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/25/check/networking-midonet-coverage/ef5cf11/ : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"31e70080193ba6261a8e6ad3b7544f3c2be856d7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-20 05:24:55.000000000","message":"Patch Set 26: Published edit on patch set 25","accounts_in_message":[],"_revision_number":26},{"id":"7a5c205385978e7057f63e6f92228d71f3984fc2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-20 05:27:43.000000000","message":"Patch Set 26: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"6aedacfca414b0a468418ac3be0f7bd3b4802bf0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-20 05:49:57.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"9e86bb85920439c245b99f8372325fc8a88b92e9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-20 06:43:55.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/26/check/gate-networking-midonet-docs/dee6b1d//doc/build/html/ : SUCCESS in 6m 00s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/26/check/gate-networking-midonet-pep8/9877af6/ : SUCCESS in 5m 51s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/26/check/gate-networking-midonet-python27/f3c9179/ : SUCCESS in 12m 40s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/26/check/gate-networking-midonet-python34/7479732/ : SUCCESS in 25m 33s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/26/check/gate-tempest-dsvm-networking-midonet-v1/85183ad/ : FAILURE in 29m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/26/check/gate-tempest-dsvm-networking-midonet-v2/d3ef2d4/ : FAILURE in 37m 11s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/26/check/gate-tempest-dsvm-networking-midonet-ml2/b93609b/ : FAILURE in 37m 33s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/26/check/gate-networking-midonet-v2-rally-dsvm-nv/17681f8/ : FAILURE in 50m 14s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/26/check/networking-midonet-coverage/4fed0b8/ : SUCCESS in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"5c76c024137f450da74b2a210e93d526d2b390ec","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-20 07:00:51.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"45c6db7ebf385345b2822b1dd077c10a98081633","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-20 08:26:28.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/26/check/gate-networking-midonet-docs/f17c5b0//doc/build/html/ : SUCCESS in 3m 42s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/26/check/gate-networking-midonet-pep8/0cca590/ : SUCCESS in 3m 27s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/26/check/gate-networking-midonet-python27/de4aa66/ : SUCCESS in 8m 42s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/26/check/gate-networking-midonet-python34/d3a5d16/ : SUCCESS in 26m 26s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/26/check/gate-tempest-dsvm-networking-midonet-v1/3e7a0f7/ : SUCCESS in 43m 43s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/26/check/gate-tempest-dsvm-networking-midonet-v2/d7260d3/ : SUCCESS in 1h 04m 21s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/26/check/gate-tempest-dsvm-networking-midonet-ml2/757e228/ : SUCCESS in 1h 23m 59s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/26/check/gate-networking-midonet-v2-rally-dsvm-nv/06cfb5a/ : FAILURE in 50m 11s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/26/check/networking-midonet-coverage/cf3d101/ : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"4a2c330a2e78159b209a244c2c996863c61d7320","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-22 07:41:03.000000000","message":"Patch Set 27: Published edit on patch set 26","accounts_in_message":[],"_revision_number":27},{"id":"2adf5f7dd79d5d14f26bf36e979544f7f1f0d235","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-22 08:49:37.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/27/check/gate-networking-midonet-docs/960fec8//doc/build/html/ : SUCCESS in 4m 21s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-pep8/1b710b7/ : SUCCESS in 5m 35s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-python27/5e97201/ : SUCCESS in 10m 41s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-python34/9cc988f/ : SUCCESS in 29m 23s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-v1/ef814e2/ : SUCCESS in 39m 40s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-v2/dfabb08/ : SUCCESS in 1h 05m 06s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-ml2/84bc5b8/ : SUCCESS in 1h 02m 18s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-v2-rally-dsvm-nv/b4df2aa/ : SUCCESS in 52m 23s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/27/check/networking-midonet-coverage/8a76d98/ : FAILURE in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"3db02c28dc22a1189ea91401b9791be9bc479288","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-23 08:31:39.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"cd6c4b54faa35a1d01747f4bab13098069c2e913","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-23 09:38:57.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/27/check/gate-networking-midonet-docs/3e832c9//doc/build/html/ : SUCCESS in 3m 57s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-pep8/9fbd4e2/ : SUCCESS in 4m 41s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-python27/571e6c2/ : SUCCESS in 8m 16s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-python34/7b162f8/ : SUCCESS in 30m 56s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-v1/4301346/ : FAILURE in 47m 29s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-v2/98e855c/ : FAILURE in 1h 07m 11s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-ml2/4c3531a/ : FAILURE in 1h 04m 06s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-v2-rally-dsvm-nv/09d6b4a/ : SUCCESS in 38m 09s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/27/check/networking-midonet-coverage/af4653c/ : FAILURE in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"f0022cb3afb5988ee8b1d32a509f9e700181db8c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-04-27 22:01:54.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"26154bc7652dd8d124ac868def263d22d789d456","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-04-27 23:21:14.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/27/check/gate-networking-midonet-docs/8bf244c//doc/build/html/ : SUCCESS in 2m 46s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-pep8/4cae443/ : SUCCESS in 2m 49s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-python27/3605753/ : SUCCESS in 9m 33s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-python34/e0e6458/ : SUCCESS in 35m 58s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-v1/8ad4354/ : SUCCESS in 37m 16s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-v2/2518c87/ : SUCCESS in 1h 13m 59s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-ml2/948d5a1/ : SUCCESS in 1h 17m 23s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-v2-rally-dsvm-nv/afd1680/ : SUCCESS in 47m 02s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/27/check/networking-midonet-coverage/c3b71fd/ : SUCCESS in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"051a7946e242dfa7f442f1db3b2d8567f6bd7519","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2016-05-12 10:04:48.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"dc70845eb61e8f1fdd61c63427179373eaa364b4","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2016-05-12 10:09:05.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"84bc4641a24d81762a1538ec6077e4a8d815fd6d","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2016-05-12 10:15:31.000000000","message":"Patch Set 27:\n\nmido-recheck","accounts_in_message":[],"_revision_number":27},{"id":"8f1dfb1ce3eb95c6b8153683d2a6e90ca3a54863","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2016-05-12 10:18:06.000000000","message":"Patch Set 27:\n\nmido-recheck","accounts_in_message":[],"_revision_number":27},{"id":"9ec39228e9cee06303098a086e99f4c1f23bc881","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-05-12 11:12:06.000000000","message":"Patch Set 27:\n\nwhat\u0027s mido-recheck?","accounts_in_message":[],"_revision_number":27},{"id":"fb343fe60911c999cc589d967da103a23601d306","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-12 11:14:34.000000000","message":"Patch Set 27:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/27/check/gate-networking-midonet-docs/a4b437e//doc/build/html/ : SUCCESS in 3m 36s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-pep8/f6aeb56/ : SUCCESS in 2m 16s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-python27/d1c3085/ : SUCCESS in 10m 01s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-python34/e34c84a/ : SUCCESS in 32m 02s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-v1/7fe83fd/ : SUCCESS in 34m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-v2/db0e8a9/ : SUCCESS in 59m 35s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/27/check/gate-tempest-dsvm-networking-midonet-ml2/b121dc7/ : SUCCESS in 42m 37s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/27/check/gate-networking-midonet-v2-rally-dsvm-nv/1976989/ : SUCCESS in 56m 12s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/27/check/networking-midonet-coverage/199ccfb/ : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"84ab4a2c1efc6dc804388d604d374bce14a3ce91","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2016-05-12 23:10:48.000000000","message":"Patch Set 27:\n\nJust a temporary comment to test with so not to trigger the openstack infra jobs.  I\u0027m setting up private jenkins to listen to the events from openstack gerrit.","accounts_in_message":[],"_revision_number":27},{"id":"9579dc05adbfa8b34fc8bb63449968da532e60ca","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2016-05-13 03:25:21.000000000","message":"Patch Set 27:\n\nmido-recheck","accounts_in_message":[],"_revision_number":27},{"id":"b7236abdd90ce4c3ddfa6831ef214009bdc87d07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-13 09:09:49.000000000","message":"Patch Set 27: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":27},{"id":"2d02ffd7f0f6bfff2f1dc22fccdf37eeec242522","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-05-16 05:15:55.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"502a0d4f451a35cdcbe93532d85046fb93e41453","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-16 06:05:53.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/2a73ebf//doc/build/html/ : SUCCESS in 5m 05s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/9e4d4e1/ : SUCCESS in 3m 07s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/741ff94/ : SUCCESS in 9m 20s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/59dfc9b/ : SUCCESS in 30m 33s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/39f2360/ : FAILURE in 49m 36s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/bb6e24e/ : FAILURE in 30m 01s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/58d8c72/ : FAILURE in 32m 02s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/504d197/ : FAILURE in 29m 36s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/f10a8f3/ : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"bf76f4e2221288e01bee64fac129564052301881","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-05-23 03:28:55.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"f4c674b0aeee4c1175aca00bd2047e35edf584b9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-23 04:47:19.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/0b81aca//doc/build/html/ : SUCCESS in 6m 01s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/2f285f2/ : SUCCESS in 6m 52s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/d277299/ : SUCCESS in 9m 33s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/840e8cd/ : SUCCESS in 34m 05s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/8904b06/ : SUCCESS in 42m 34s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/e01c02f/ : SUCCESS in 1h 05m 38s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/a44b7c0/ : SUCCESS in 1h 14m 36s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/849b280/ : FAILURE in 1h 04m 59s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/810b510/ : SUCCESS in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"af0963a46baccf60917007e74b208cef8fe6ff8d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-05-24 03:22:12.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"d6a1786d26aa3e679dddc6fa70adc5b22b1cee66","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-24 04:40:38.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/800f2c7//doc/build/html/ : SUCCESS in 4m 11s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/7dd07a0/ : SUCCESS in 2m 50s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/34b9f3e/ : FAILURE in 8m 53s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/3249a1e/ : FAILURE in 33m 55s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/401c956/ : SUCCESS in 55m 07s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/15afadb/ : SUCCESS in 1h 12m 05s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/5c2bc08/ : SUCCESS in 1h 11m 54s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/99a6297/ : FAILURE in 55m 32s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/f2f74d0/ : FAILURE in 9m 38s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"19ac9a6029b8aaf37a93c53b65b83a6ae07ceb6d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-05-24 14:20:32.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"2d2823eda4e044c1437fe42bcf46ab4cbb59903f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-24 16:28:42.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/1e3ef2c//doc/build/html/ : SUCCESS in 5m 50s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/87ddeca/ : SUCCESS in 5m 26s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/27e719c/ : SUCCESS in 8m 59s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/5fa407c/ : FAILURE in 40m 15s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/b226f75/ : SUCCESS in 56m 41s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/3cbdd13/ : SUCCESS in 1h 09m 31s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/913573c/ : SUCCESS in 53m 07s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/53df860/ : FAILURE in 1h 02m 23s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/6b35bab/ : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"9eac785ffa2dd4b6d86a7e0d9c73a27bbbc07e77","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-08 06:30:53.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"b910f96e27ffd446834997ca2b3121274d79271b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-08 07:47:31.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/29f1561//doc/build/html/ : SUCCESS in 4m 58s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/eeba409/ : SUCCESS in 6m 03s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/af187d4/ : SUCCESS in 38m 32s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/0ee32ee/ : SUCCESS in 31m 52s\n- gate-networking-midonet-releasenotes http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-releasenotes/af089e3/ : FAILURE in 6m 09s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/9830069/ : SUCCESS in 49m 36s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/0522ccb/ : SUCCESS in 1h 12m 57s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/24b8ac2/ : SUCCESS in 1h 12m 49s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/992ea59/ : FAILURE in 25m 13s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/f6dd931/ : FAILURE in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"1d0c11926131cdbb00bc8c976e2a72c98b6350f8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-13 07:24:11.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"b175fc63f6dac13280ab5918e92d2a00c01d8d8b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-13 08:32:41.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/e8025ef//doc/build/html/ : SUCCESS in 4m 31s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/d49a1bd/ : SUCCESS in 4m 37s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/5ed551d/ : FAILURE in 32m 12s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/64c243a/ : FAILURE in 35m 57s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-releasenotes/804d8ae//releasenotes/build/html/ : SUCCESS in 5m 24s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/d46800e/ : FAILURE in 42m 09s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/b9cbb65/ : SUCCESS in 48m 59s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/d12126e/ : SUCCESS in 1h 05m 01s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/7bad31a/ : FAILURE in 22m 59s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/a84624a/ : FAILURE in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"afcaf5d48586ca85383d1486cbe0ff8fef497021","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-16 15:35:59.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"863622fe9d0219fd1f7c3d8e8f3d74df9e4aaac4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-16 18:18:28.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/ff0ac3c//doc/build/html/ : SUCCESS in 5m 33s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/3919730/ : SUCCESS in 6m 06s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/2a8e402/ : SUCCESS in 25m 43s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/be49b71/ : SUCCESS in 26m 11s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-releasenotes/8420ffd//releasenotes/build/html/ : SUCCESS in 4m 49s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/cb93ad1/ : SUCCESS in 54m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/f61452d/ : FAILURE in 48m 10s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/bd6c929/ : FAILURE in 32m 22s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/f702a59/ : FAILURE in 14m 23s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/ddad542/ : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"1b3e9f6bbcf2724430abc1cc8d0fc33523632c5e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-20 03:47:38.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"612318b27d9336dccb07067c35ca2b46885030c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-20 04:48:55.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-docs/30ce721/ : POST_FAILURE in 13m 47s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/b8b06c3/ : POST_FAILURE in 10m 00s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/c979403/ : POST_FAILURE in 43m 41s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/f3f65ea/ : POST_FAILURE in 48m 17s\n- gate-networking-midonet-releasenotes http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-releasenotes/1f40560/ : POST_FAILURE in 14m 17s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/c53150e/ : POST_FAILURE in 52m 20s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/e1fe38d/ : POST_FAILURE in 1h 01m 05s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/2adc547/ : POST_FAILURE in 51m 13s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/1163c6c/ : POST_FAILURE in 56m 50s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/6cddc7d/ : POST_FAILURE in 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"a2f62fde32ab6b10c375d4a4f80ad910f0df81a6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-21 06:08:19.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"29772d59ea2528e568b064c7ae69f521b395afb0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-21 07:05:01.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/035fc47//doc/build/html/ : SUCCESS in 6m 24s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/3773d70/ : SUCCESS in 6m 41s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/c281c6b/ : SUCCESS in 28m 52s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/e21931e/ : SUCCESS in 33m 40s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-releasenotes/6402c85//releasenotes/build/html/ : SUCCESS in 6m 24s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/c7e79bf/ : SUCCESS in 49m 13s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/64452e4/ : FAILURE in 50m 05s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/33b1ead/ : FAILURE in 40m 07s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/68f7195/ : FAILURE in 54m 14s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/a8c908f/ : FAILURE in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"302faa9ba99d86b4a70cae9abd7a1ac0650140b3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-21 08:34:41.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"8040d5a22828dc700a804b0951b50b61e4d0c548","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-21 09:33:40.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-docs/25492a8//doc/build/html/ : SUCCESS in 5m 16s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-pep8/14842fc/ : SUCCESS in 5m 18s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python27/be1ba1c/ : SUCCESS in 42m 22s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-python34/e26cf09/ : SUCCESS in 39m 07s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/28/check/gate-networking-midonet-releasenotes/98c2290//releasenotes/build/html/ : SUCCESS in 7m 50s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v1/5385550/ : SUCCESS in 50m 48s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-v2/958a166/ : FAILURE in 46m 55s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/28/check/gate-tempest-dsvm-networking-midonet-ml2/1895549/ : FAILURE in 45m 26s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/28/check/gate-networking-midonet-v2-rally-dsvm-nv/82abe13/ : FAILURE in 51m 03s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/28/check/networking-midonet-coverage/095031c/ : FAILURE in 7m 46s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"1985cdcb83f321f364e9f23c701dbccd48f605c5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-22 14:23:35.000000000","message":"Patch Set 29: Published edit on patch set 28","accounts_in_message":[],"_revision_number":29},{"id":"bd30b355bcf71ecd98391e31e9eb083a341814d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-22 15:24:02.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/29/check/gate-networking-midonet-docs/cbd238a//doc/build/html/ : SUCCESS in 6m 17s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/29/check/gate-networking-midonet-pep8/8217518/ : SUCCESS in 6m 55s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/29/check/gate-networking-midonet-python27/f396467/ : SUCCESS in 41m 52s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/29/check/gate-networking-midonet-python34/02d0756/ : SUCCESS in 25m 28s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/29/check/gate-networking-midonet-releasenotes/5e12a63//releasenotes/build/html/ : SUCCESS in 6m 01s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/29/check/gate-tempest-dsvm-networking-midonet-v1/2b1c219/ : FAILURE in 56m 49s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/29/check/gate-tempest-dsvm-networking-midonet-v2/2397a22/ : SUCCESS in 53m 06s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/29/check/gate-tempest-dsvm-networking-midonet-ml2/36f62e2/ : SUCCESS in 42m 26s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/29/check/gate-networking-midonet-v2-rally-dsvm-nv/652758f/ : FAILURE in 52m 02s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/29/check/networking-midonet-coverage/eaac559/ : FAILURE in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"634c3b972afad0eafb3909c8326bd8053ab09247","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-23 06:37:57.000000000","message":"Patch Set 30: Published edit on patch set 29","accounts_in_message":[],"_revision_number":30},{"id":"2c86718e853b950415a91f82dd9ad8f62a60f896","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-23 07:33:29.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/30/check/gate-networking-midonet-docs/46bc186//doc/build/html/ : SUCCESS in 5m 11s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-pep8/4024a65/ : SUCCESS in 5m 53s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-python27/84c9afc/ : SUCCESS in 42m 42s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-python34/7e93cb7/ : SUCCESS in 37m 23s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/30/check/gate-networking-midonet-releasenotes/daf0744//releasenotes/build/html/ : SUCCESS in 6m 26s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-v1/9adf877/ : SUCCESS in 45m 26s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-v2/ae22ce9/ : SUCCESS in 45m 44s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-ml2/a0de490/ : SUCCESS in 40m 50s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-v2-rally-dsvm-nv/46f773d/ : FAILURE in 49m 11s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/30/check/networking-midonet-coverage/ad77a04/ : FAILURE in 8m 09s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"75f7824aad5cb99b6110bb9bafdde7284d26a908","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-24 09:57:40.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"8342bf1b7fc8aab985b7d1b5b6efed005019f284","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-24 11:22:28.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/30/check/gate-networking-midonet-docs/ac29a5c//doc/build/html/ : SUCCESS in 6m 18s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-pep8/d0dc386/ : SUCCESS in 4m 38s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-python27/904a99c/ : SUCCESS in 26m 01s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-python34/31552f3/ : SUCCESS in 25m 47s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/30/check/gate-networking-midonet-releasenotes/6127765//releasenotes/build/html/ : SUCCESS in 6m 42s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-v1/b7cbb6c/ : FAILURE in 51m 35s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-v2/950841b/ : FAILURE in 1h 08m 53s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-ml2/8290f04/ : FAILURE in 42m 55s\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-v2-rally-dsvm-nv/500308c/ : FAILURE in 53m 08s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/30/check/networking-midonet-coverage/e785653/ : FAILURE in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"06f27ca944842e5fbe8df27613c36f228cd7537b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-24 19:12:48.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"218d291a92d1a6c37a07caecd9302048e8ed8823","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-24 20:11:19.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/30/check/gate-networking-midonet-docs/3a89f03//doc/build/html/ : SUCCESS in 3m 56s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-pep8/4b865d2/ : SUCCESS in 3m 58s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-python27/5dc7d7b/ : SUCCESS in 42m 03s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-python34/f919125/ : SUCCESS in 30m 50s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/30/check/gate-networking-midonet-releasenotes/ea32676//releasenotes/build/html/ : SUCCESS in 10m 23s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-v1/8b418cb/ : SUCCESS in 48m 56s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-v2/71f1d7c/ : SUCCESS in 43m 35s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-ml2/0a788c7/ : SUCCESS in 40m 41s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-v2-rally-dsvm-nv/0fc826c/ : FAILURE in 56m 44s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/30/check/networking-midonet-coverage/e6b8b7c/ : FAILURE in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"2034a1b3fdacf8886b168cbb8a4f2e96921a1ea6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-26 16:10:37.000000000","message":"Patch Set 30:\n\nrecheck","accounts_in_message":[],"_revision_number":30},{"id":"4efe2bfa79511e4d7a59f43f9cd760e0b08e5f48","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-26 17:04:03.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/30/check/gate-networking-midonet-docs/58adc14//doc/build/html/ : SUCCESS in 7m 25s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-pep8/aeea3cf/ : SUCCESS in 5m 24s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-python27/a5b9474/ : SUCCESS in 36m 49s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-python34/0439f87/ : SUCCESS in 34m 19s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/30/check/gate-networking-midonet-releasenotes/e08b510//releasenotes/build/html/ : SUCCESS in 6m 43s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-v1/fdd5c01/ : FAILURE in 53m 10s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-v2/80f1a87/ : SUCCESS in 48m 09s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/30/check/gate-tempest-dsvm-networking-midonet-ml2/93f4276/ : SUCCESS in 35m 30s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/30/check/gate-networking-midonet-v2-rally-dsvm-nv/7619c9e/ : SUCCESS in 48m 05s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/30/check/networking-midonet-coverage/b5493b3/ : FAILURE in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"555b313bc0304f5117bb935f81430d81966184f1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-27 04:10:51.000000000","message":"Patch Set 31: Published edit on patch set 30","accounts_in_message":[],"_revision_number":31},{"id":"ba1a59e9e6a92809e6b4cc56f5a6a58e0a3c4edf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-27 05:01:15.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/31/check/gate-networking-midonet-docs/cf4a574//doc/build/html/ : SUCCESS in 7m 02s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/31/check/gate-networking-midonet-pep8/e0b789e/ : SUCCESS in 7m 28s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/31/check/gate-networking-midonet-python27/d681b7d/ : FAILURE in 35m 45s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/31/check/gate-networking-midonet-python34/b4fdd60/ : FAILURE in 31m 55s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/31/check/gate-networking-midonet-releasenotes/0081b53//releasenotes/build/html/ : SUCCESS in 5m 37s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/31/check/gate-tempest-dsvm-networking-midonet-v1/3825695/ : POST_FAILURE in 47m 23s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/31/check/gate-tempest-dsvm-networking-midonet-v2/96e2fbe/ : FAILURE in 35m 59s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/31/check/gate-tempest-dsvm-networking-midonet-ml2/b90a87b/ : FAILURE in 37m 55s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/31/check/gate-networking-midonet-v2-rally-dsvm-nv/315a81f/ : SUCCESS in 48m 08s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/31/check/networking-midonet-coverage/a762d83/ : FAILURE in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"1a49e88f0d1662ad7a3a7bc188e84f484e062612","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-27 05:13:55.000000000","message":"Patch Set 32: Published edit on patch set 31","accounts_in_message":[],"_revision_number":32},{"id":"4121194abb446b9d2251b5836884bfdd0601006c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-27 06:11:38.000000000","message":"Patch Set 32:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/32/check/gate-networking-midonet-docs/327b4d2//doc/build/html/ : SUCCESS in 5m 20s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/32/check/gate-networking-midonet-pep8/5699ffd/ : SUCCESS in 5m 17s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/32/check/gate-networking-midonet-python27/de390de/ : SUCCESS in 38m 29s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/32/check/gate-networking-midonet-python34/74fb84b/ : SUCCESS in 33m 54s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/32/check/gate-networking-midonet-releasenotes/d6e63df//releasenotes/build/html/ : SUCCESS in 5m 44s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/32/check/gate-tempest-dsvm-networking-midonet-v1/1bf7e1d/ : FAILURE in 54m 14s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/32/check/gate-tempest-dsvm-networking-midonet-v2/63d3526/ : SUCCESS in 40m 03s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/32/check/gate-tempest-dsvm-networking-midonet-ml2/a7463fb/ : SUCCESS in 33m 00s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/32/check/gate-networking-midonet-v2-rally-dsvm-nv/0c0e7c2/ : FAILURE in 49m 49s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/32/check/networking-midonet-coverage/3a46bbe/ : FAILURE in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"e615778855da17abc55f27cdda9c1cc47e263c8a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-27 11:53:54.000000000","message":"Patch Set 33: Published edit on patch set 32","accounts_in_message":[],"_revision_number":33},{"id":"4ff7e6efd88072cec293a9b9aee7f48d6d06618f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-27 11:54:18.000000000","message":"Patch Set 33:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":33},{"id":"b52a3bd70740527ebfe14a77e539d7b26fb2109b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-27 13:28:06.000000000","message":"Patch Set 34: Published edit on patch set 33","accounts_in_message":[],"_revision_number":34},{"id":"0e86da437e66cdced81f59443923807d4c7757d7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-27 14:38:18.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/2e644a3//doc/build/html/ : SUCCESS in 7m 41s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/ff902be/ : SUCCESS in 7m 15s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/8093925/ : SUCCESS in 33m 56s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/b116b09/ : SUCCESS in 35m 41s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/07f0be4//releasenotes/build/html/ : SUCCESS in 11m 35s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/0849f72/ : FAILURE in 46m 05s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/4085430/ : FAILURE in 45m 49s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/cde8584/ : SUCCESS in 45m 40s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/751c279/ : FAILURE in 45m 57s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/c710afc/ : FAILURE in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"33e62d9d7d991c50f462ea1fd01021e1ee9c38a6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-28 04:32:04.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"2c1218143667bccaad95b9458b63fb2589ff5c67","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-28 05:22:10.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/5cdf78e//doc/build/html/ : SUCCESS in 7m 47s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/6e86c4c/ : SUCCESS in 7m 30s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/2f97670/ : SUCCESS in 34m 33s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/a49ba68/ : SUCCESS in 33m 08s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/d314caf//releasenotes/build/html/ : SUCCESS in 6m 49s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/493df26/ : FAILURE in 42m 35s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/7e61eef/ : SUCCESS in 37m 44s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/cf3ac0e/ : SUCCESS in 39m 21s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/dd04500/ : SUCCESS in 47m 13s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/0ab8280/ : FAILURE in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"9c91281559d4f973f302b8554f76d1eb9a660012","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-06-29 04:44:42.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"520ea4b30ba2d4d25918ef942cbb55836528d622","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-29 05:34:58.000000000","message":"Patch Set 34:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/132640b//doc/build/html/ : SUCCESS in 2m 44s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/5211dc1/ : SUCCESS in 5m 41s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/a882157/ : SUCCESS in 32m 23s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/623781d/ : SUCCESS in 36m 37s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/7291e80//releasenotes/build/html/ : SUCCESS in 7m 31s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/ca387b9/ : FAILURE in 20m 47s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/b801ebd/ : SUCCESS in 45m 26s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/aec79dc/ : SUCCESS in 40m 16s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/1c58ec8/ : SUCCESS in 48m 01s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/8c08a30/ : FAILURE in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"cad3771569baf4cfbe8e3c4d881d0e2850b7efa3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-02 14:03:44.000000000","message":"Patch Set 34: Cherry Picked\n\nThis patchset was cherry picked to branch stable/mitaka as commit 24e96859503d770662d1abf2303b0d89f0c5351f","accounts_in_message":[],"_revision_number":34},{"id":"7706865aa56cb1c0736a10cca0a7a7eb75e7d745","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-07 16:51:54.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"3488fc1cef4bc281878eb6c8869c98d0a8d51b87","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-07 18:39:01.000000000","message":"Patch Set 34:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/4d5e00a//doc/build/html/ : SUCCESS in 3m 57s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/3a28c42/ : SUCCESS in 4m 19s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/80a0eb5/ : SUCCESS in 30m 15s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/92991de/ : SUCCESS in 34m 25s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35-nv/8beaa05/ : FAILURE in 8m 27s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/54f7f7c//releasenotes/build/html/ : SUCCESS in 4m 31s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/7e7d565/ : FAILURE in 36m 07s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/6fe7fe3/ : SUCCESS in 1h 00m 54s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/c0fc8da/ : SUCCESS in 1h 41m 35s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/eec7092/ : FAILURE in 55m 35s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/b4f475f/ : FAILURE in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"1f734ef3e1374aa4685d1b61113e6b3a3ec433e3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-11 06:41:25.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"26518fc1be0774903c692ef864b5bb9a8123255d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-11 08:13:33.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/3421c08//doc/build/html/ : SUCCESS in 2m 39s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/829c079/ : SUCCESS in 6m 24s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/4cf9069/ : SUCCESS in 36m 09s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/a9b3718/ : SUCCESS in 21m 07s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35-nv/f958638/ : FAILURE in 7m 10s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/5300d63//releasenotes/build/html/ : SUCCESS in 5m 42s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/aff34a6/ : FAILURE in 34m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/2d7f4ac/ : FAILURE in 1h 30m 55s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/c6d4ad3/ : FAILURE in 1h 31m 37s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/6f819ae/ : SUCCESS in 58m 13s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/5d354fa/ : FAILURE in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"f5761a080680ef48394937af33b1ad305006f6ff","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-12 04:30:56.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"06de37decc3e07aa8f2e7d3c633110aa197f524e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-12 05:55:44.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/07aea8a//doc/build/html/ : SUCCESS in 5m 19s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/7a6f14c/ : SUCCESS in 7m 58s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/8dae7b8/ : SUCCESS in 33m 09s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/438970d/ : SUCCESS in 32m 21s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35-nv/741a10f/ : FAILURE in 8m 30s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/1e16b62//releasenotes/build/html/ : SUCCESS in 6m 03s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/a7429f3/ : FAILURE in 1h 02m 25s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/f987240/ : FAILURE in 1h 04m 59s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/d4cac72/ : FAILURE in 1h 21m 29s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/3e19195/ : FAILURE in 1h 00m 15s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/c387de3/ : FAILURE in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"be1045c55af04cf44d74eaf4996e9df9ced018a3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-13 01:00:00.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"e7a999daf8c4d6bc3d08a364bd319d878d4246f7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-13 03:24:35.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/945bf77//doc/build/html/ : SUCCESS in 2m 47s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/448f195/ : SUCCESS in 5m 52s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/1bc9f20/ : SUCCESS in 32m 33s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/40ab011/ : SUCCESS in 32m 07s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35-nv/d2e4f64/ : FAILURE in 7m 11s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/fdc9878//releasenotes/build/html/ : SUCCESS in 6m 31s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/e4d6f11/ : FAILURE in 53m 10s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/8bea9c4/ : FAILURE in 1h 53m 01s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/a52bc74/ : FAILURE in 1h 35m 51s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/61017cd/ : SUCCESS in 57m 00s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/d6b3dfe/ : FAILURE in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"d27c4af7cb0a8eaa06a9f168bf0fbd239143dce4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-14 07:36:56.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"37f1d5279463faaba6dc28011e44b91781d99d24","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-14 08:49:01.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/2157fdd//doc/build/html/ : SUCCESS in 2m 00s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/b197f5e/ : SUCCESS in 4m 08s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/954a7f3/ : SUCCESS in 20m 26s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/3e28bce/ : SUCCESS in 32m 29s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35-nv/fd85fb1/ : FAILURE in 7m 20s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/11ae146//releasenotes/build/html/ : SUCCESS in 5m 25s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/992575d/ : FAILURE in 37m 43s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/d5fdcdb/ : FAILURE in 1h 10m 05s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/726ae54/ : FAILURE in 57m 22s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/49dc93b/ : SUCCESS in 46m 30s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/8980a7d/ : FAILURE in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"369f00ee97589b02c4ffd09fa770a5e0d0325ae8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-18 11:43:54.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"378398d97b32bd39205f0e0ec679c2f3c9b6304c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-18 12:58:14.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/f10b436//doc/build/html/ : SUCCESS in 2m 40s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/3d28e49/ : SUCCESS in 3m 50s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/d971110/ : SUCCESS in 37m 12s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/2b715e0/ : SUCCESS in 32m 29s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35/e27452d/ : SUCCESS in 38m 11s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/239ed7c//releasenotes/build/html/ : SUCCESS in 5m 57s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/7560b18/ : FAILURE in 43m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/2116066/ : SUCCESS in 1h 10m 50s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/d885562/ : SUCCESS in 1h 05m 53s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/165d94d/ : SUCCESS in 1h 01m 29s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/6d86c6a/ : FAILURE in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"1c57d22cd55ab86c120f57403a003c31a7c552ac","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-20 12:15:52.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"c849a000ac6b153d0159a6b7385fa99241acd9e3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-20 13:24:32.000000000","message":"Patch Set 34:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/0f358aa//doc/build/html/ : SUCCESS in 2m 18s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/564a416/ : SUCCESS in 4m 31s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/1d93ef9/ : SUCCESS in 24m 30s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/2ac72e8/ : SUCCESS in 28m 54s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35/a7b92b9/ : SUCCESS in 37m 54s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/d3a05af//releasenotes/build/html/ : SUCCESS in 7m 00s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/fa5c8ce/ : FAILURE in 37m 29s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/7243199/ : SUCCESS in 37m 57s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/7da501a/ : SUCCESS in 36m 02s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/afc5e41/ : SUCCESS in 1h 04m 01s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/73f8dae/ : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"79a4a896dc17c5a0ed8d9ae169db996d01754e27","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-21 13:12:19.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"07e000fc8a0300e534b1c3502d81da0abc16f0d1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-21 14:27:06.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs/17c6c92//doc/build/html/ : SUCCESS in 5m 14s\n- gate-networking-midonet-pep8 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8/e5d56a8/ : SUCCESS in 8m 50s\n- gate-networking-midonet-python27 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27/3dde6b4/ : SUCCESS in 37m 14s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/187f85d/ : SUCCESS in 33m 22s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35/4e5de9c/ : FAILURE in 45m 12s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/3839690//releasenotes/build/html/ : SUCCESS in 4m 14s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/28a49bc/ : FAILURE in 38m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/5ea8b86/ : SUCCESS in 54m 18s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/b3abf43/ : FAILURE in 10m 37s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/e4dae75/ : SUCCESS in 1h 11m 36s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/45771ae/ : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"0753bbf60baa30d7b9de8348dac9a8519d848851","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-26 06:36:38.000000000","message":"Patch Set 34:\n\nrecheck","accounts_in_message":[],"_revision_number":34},{"id":"72420da26eb88fce9bd10a0994493ea66cf684aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-26 08:26:43.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-docs-ubuntu-xenial/41f8f4b//doc/build/html/ : SUCCESS in 7m 33s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-pep8-ubuntu-xenial/4960e6e/ : SUCCESS in 5m 49s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python27-ubuntu-xenial/eb371fb/ : FAILURE in 55m 33s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python34/22244aa/ : SUCCESS in 34m 27s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-python35-nv/b0b82b9/ : FAILURE in 45m 18s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/34/check/gate-networking-midonet-releasenotes/3508e89//releasenotes/build/html/ : SUCCESS in 7m 37s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v1/531680c/ : FAILURE in 51m 13s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-v2/4e3c896/ : SUCCESS in 36m 03s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/34/check/gate-tempest-dsvm-networking-midonet-ml2/bbf09b7/ : SUCCESS in 47m 44s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/34/check/gate-networking-midonet-v2-rally-dsvm-nv/3a421b9/ : SUCCESS in 1h 18m 10s (non-voting)\n- networking-midonet-coverage http://logs.openstack.org/87/199387/34/check/networking-midonet-coverage/27cf247/ : FAILURE in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"b87bdc177b5d0a9c21a5ccd859fc8835de3ee492","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-27 06:17:06.000000000","message":"Patch Set 35: Published edit on patch set 34","accounts_in_message":[],"_revision_number":35},{"id":"28b9fefbc933cd353ed51e4dd862335c9ca96177","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-27 07:43:02.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/bfb2b25//doc/build/html/ : SUCCESS in 5m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/a7120e7/ : SUCCESS in 19m 10s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/162f113/ : SUCCESS in 14m 49s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/fadf469/ : FAILURE in 45m 28s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/7b37e14/ : FAILURE in 46m 02s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/b8b7b1e//releasenotes/build/html/ : SUCCESS in 8m 32s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/0545708/ : SUCCESS in 37m 10s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/cf9bcd7/ : SUCCESS in 54m 04s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/b8543cb/ : SUCCESS in 50m 15s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/4a13af0/ : FAILURE in 1h 25m 09s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/9538292/ : SUCCESS in 16m 43s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"a4a8660792223491b839eb3e5a55b386b5847ba9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-29 06:48:51.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"f7fb1663802eed77f8f4b786cc8c45bb777bd093","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-29 09:00:05.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/c43a416//doc/build/html/ : SUCCESS in 3m 17s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/2fffe40/ : SUCCESS in 4m 13s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/63d0ffb/ : SUCCESS in 12m 46s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/229d892/ : SUCCESS in 11m 39s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/fc1100a/ : SUCCESS in 12m 24s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/08808a8//releasenotes/build/html/ : SUCCESS in 6m 05s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/78dd2a8/ : SUCCESS in 49m 27s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/ca1be89/ : FAILURE in 2h 10m 51s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/69c81fc/ : SUCCESS in 1h 53m 59s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/49cfe50/ : FAILURE in 1h 25m 04s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/7c4d416/ : SUCCESS in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"c42459043577396b1c38619ec6f26f8b539abe20","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-30 14:12:01.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"b34ad25a715a065dae6bbe60ce4d54f501377ed0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-30 15:48:47.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/b189095/ : POST_FAILURE in 2m 56s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/4fda924/ : POST_FAILURE in 6m 48s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/f500f44/ : POST_FAILURE in 12m 38s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/e4ff115/ : POST_FAILURE in 11m 28s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/e3eac69/ : POST_FAILURE in 13m 51s (non-voting)\n- gate-networking-midonet-releasenotes http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/953a789/ : POST_FAILURE in 6m 53s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/dafae07/ : SUCCESS in 58m 10s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/f94cf8a/ : SUCCESS in 1h 36m 35s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/c3e1d02/ : SUCCESS in 1h 23m 53s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/9741480/ : FAILURE in 1h 17m 38s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/13f7ddc/ : POST_FAILURE in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"bdd5e659ce3375ce69ea5350e550943e8da7cad4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-07-30 18:10:33.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"3919743a3d6c97431061c009efbeda9d04c13f75","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-30 20:01:42.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/c56856c//doc/build/html/ : SUCCESS in 2m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/3353779/ : SUCCESS in 7m 54s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/349eda8/ : SUCCESS in 14m 00s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/2e37f8f/ : SUCCESS in 18m 18s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/c5a5260/ : SUCCESS in 14m 43s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/b9abd38//releasenotes/build/html/ : SUCCESS in 5m 41s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/4b9b276/ : SUCCESS in 1h 05m 26s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/e9d012d/ : SUCCESS in 53m 38s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/0a60187/ : SUCCESS in 1h 12m 59s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/0c66aba/ : FAILURE in 1h 46m 46s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/0d55806/ : FAILURE in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"1ca2cd326e4cbe8c7d07e625cd724f707422702a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-01 00:09:20.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"16d2f03add9c379724f6affe231c644d671d3887","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-01 01:38:22.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/5f606ae//doc/build/html/ : SUCCESS in 3m 38s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/f4654a4/ : FAILURE in 7m 02s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/8565414/ : FAILURE in 12m 36s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/f8f0aed/ : FAILURE in 13m 19s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/cb0ac4c/ : FAILURE in 14m 23s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/3830ac6//releasenotes/build/html/ : SUCCESS in 5m 59s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/8ac9e80/ : SUCCESS in 49m 37s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/ff10630/ : SUCCESS in 1h 19m 25s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/4487afc/ : SUCCESS in 1h 21m 52s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/8e25d26/ : FAILURE in 1h 27m 33s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/bf8ee41/ : FAILURE in 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"8d1f8b8d0fb40d3495e5bf4325e6697d0d95e8cb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-02 00:51:42.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"9987306885ed6f61547c4eff5a23a982e48236be","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-02 02:26:50.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/9d945dd//doc/build/html/ : SUCCESS in 3m 38s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/7ed4ace/ : FAILURE in 4m 16s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/e616325/ : FAILURE in 13m 23s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/e0cce4e/ : FAILURE in 12m 29s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/009d81c/ : FAILURE in 14m 59s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/7a5f359//releasenotes/build/html/ : SUCCESS in 6m 01s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/366cd6b/ : SUCCESS in 1h 00m 59s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/f62d211/ : FAILURE in 1h 26m 23s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/74c3239/ : FAILURE in 1h 07m 31s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/cceffa8/ : FAILURE in 1h 12m 07s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/451a436/ : FAILURE in 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"13226c715f30b5bc74acd63c616baeb5ffa4b92a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-03 03:03:44.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"789c428a8c06e1aed6b93b6b0b7c96a1481c8538","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-03 07:41:59.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/1f23576//doc/build/html/ : SUCCESS in 8m 29s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/5d6a785/ : FAILURE in 9m 29s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/f52215f/ : SUCCESS in 16m 40s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/f1c1c6e/ : SUCCESS in 13m 08s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/0fd8832/ : FAILURE in 45m 35s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/834448b//releasenotes/build/html/ : SUCCESS in 11m 02s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/950af0c/ : SUCCESS in 58m 52s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/15e9a6f/ : FAILURE in 1h 27m 52s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/e44d6ee/ : FAILURE in 42m 26s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/21b8082/ : SUCCESS in 59m 29s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/f731d9a/ : FAILURE in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"0e9e158ca3c504b518f601c889c79f6f36b0b7c9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-03 22:44:28.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"c7c2a87ed78a6a4a2d108bedd642b4dabe45039a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-04 00:28:38.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/a137df7//doc/build/html/ : SUCCESS in 2m 22s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/0563160/ : SUCCESS in 6m 02s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/72c3289/ : FAILURE in 14m 43s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/65f771a/ : SUCCESS in 13m 24s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/e4feb63/ : SUCCESS in 19m 21s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/f517df3//releasenotes/build/html/ : SUCCESS in 7m 10s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/4254e93/ : FAILURE in 1h 04m 07s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/e0f67fe/ : SUCCESS in 1h 32m 37s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/8b9dd19/ : FAILURE in 1h 29m 55s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/1012656/ : FAILURE in 1h 22m 20s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/ca7e02b/ : FAILURE in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"faeee82850da0f87ec31ecd34609f9228a85f035","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-05 07:06:23.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"a707d7e96f07f2f6229d617ed8d726738be478c5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-05 08:26:40.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/eb17016//doc/build/html/ : SUCCESS in 3m 44s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/b11d3a7/ : SUCCESS in 8m 16s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/2982cbf/ : SUCCESS in 21m 27s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/3442235/ : SUCCESS in 14m 11s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/03b2110/ : SUCCESS in 13m 31s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/d1960ba//releasenotes/build/html/ : SUCCESS in 7m 27s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/fa42993/ : FAILURE in 47m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/098eb11/ : FAILURE in 1h 00m 13s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/02a9e05/ : FAILURE in 1h 14m 53s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/18ec0aa/ : SUCCESS in 1h 15m 46s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/b85a20a/ : FAILURE in 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"eb634d1e230322c002703f77a1ecffc86b715e28","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-07 23:49:10.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"9a9f87826f9117c4f0d497dce9e93ea2c0c3e75e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-08 01:15:40.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-docs-ubuntu-xenial/aa9bd48//doc/build/html/ : SUCCESS in 3m 48s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-pep8-ubuntu-xenial/c5732e1/ : SUCCESS in 9m 18s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python27-ubuntu-xenial/401a00a/ : FAILURE in 11m 13s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python34/ad1e325/ : FAILURE in 14m 01s\n- gate-networking-midonet-python35-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-python35-nv/48d7e6a/ : FAILURE in 13m 19s (non-voting)\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/35/check/gate-networking-midonet-releasenotes/7281a92//releasenotes/build/html/ : SUCCESS in 7m 04s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v1/183a77f/ : FAILURE in 1h 01m 07s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-v2/88cf271/ : FAILURE in 1h 26m 11s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/35/check/gate-tempest-dsvm-networking-midonet-ml2/5838358/ : FAILURE in 59m 30s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/35/check/gate-networking-midonet-v2-rally-dsvm-nv/fbdcce6/ : FAILURE in 1h 18m 56s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/35/check/networking-midonet-coverage-ubuntu-xenial/2280914/ : FAILURE in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"e4a3ebe425479125bd5babf3a26f05a0c15f1c05","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-09 11:19:52.000000000","message":"Patch Set 36: Published edit on patch set 35","accounts_in_message":[],"_revision_number":36},{"id":"e6b453e8da46b9f5c327e3311fa4440b041e6b13","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-09 13:42:00.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/36/check/gate-networking-midonet-docs-ubuntu-xenial/78b22cc//doc/build/html/ : SUCCESS in 6m 07s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/36/check/gate-networking-midonet-pep8-ubuntu-xenial/ac1e017/ : SUCCESS in 14m 02s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/36/check/gate-networking-midonet-python27-ubuntu-xenial/18bac17/ : SUCCESS in 20m 52s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/36/check/gate-networking-midonet-python34/5ab996f/ : SUCCESS in 11m 36s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/36/check/gate-networking-midonet-python35/2730caa/ : SUCCESS in 24m 22s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/36/check/gate-networking-midonet-releasenotes/6cf80fb//releasenotes/build/html/ : SUCCESS in 4m 21s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/36/check/gate-tempest-dsvm-networking-midonet-v1/86071b7/ : SUCCESS in 47m 17s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/36/check/gate-tempest-dsvm-networking-midonet-v2/faeac29/ : SUCCESS in 1h 20m 50s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/36/check/gate-tempest-dsvm-networking-midonet-ml2/ceb0331/ : FAILURE in 1h 01m 40s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/36/check/gate-networking-midonet-v2-rally-dsvm-nv/d39e60e/ : FAILURE in 1h 15m 34s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/36/check/networking-midonet-coverage-ubuntu-xenial/a374181/ : SUCCESS in 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"c8a83e24f0e670304d8af518885ce7bf30ffb0e3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-09 22:17:02.000000000","message":"Patch Set 37: Published edit on patch set 36","accounts_in_message":[],"_revision_number":37},{"id":"4a6d27cf1f11aaead3e1277bd3ddbc2918268c0d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-10 02:00:28.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/37/check/gate-networking-midonet-docs-ubuntu-xenial/36d29fa//doc/build/html/ : SUCCESS in 2m 24s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/37/check/gate-networking-midonet-pep8-ubuntu-xenial/22bbe7c/ : SUCCESS in 8m 49s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/37/check/gate-networking-midonet-python27-ubuntu-xenial/0d93a15/ : SUCCESS in 15m 50s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/37/check/gate-networking-midonet-python34/289bf80/ : SUCCESS in 12m 50s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/37/check/gate-networking-midonet-python35/7feb13d/ : SUCCESS in 13m 39s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/37/check/gate-networking-midonet-releasenotes/2af6b5e//releasenotes/build/html/ : SUCCESS in 8m 20s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/37/check/gate-tempest-dsvm-networking-midonet-v1/c9719b0/ : SUCCESS in 1h 01m 50s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/37/check/gate-tempest-dsvm-networking-midonet-v2/b29a6bc/ : SUCCESS in 1h 26m 34s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/37/check/gate-tempest-dsvm-networking-midonet-ml2/06207c5/ : FAILURE in 1h 02m 56s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/37/check/gate-networking-midonet-v2-rally-dsvm-nv/5a70024/ : SUCCESS in 57m 54s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/37/check/networking-midonet-coverage-ubuntu-xenial/00b3f40/ : SUCCESS in 17m 25s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"aaa5168a36c382ed5bd1e91d063d5358702f98fd","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-10 04:33:38.000000000","message":"Patch Set 38: Published edit on patch set 37","accounts_in_message":[],"_revision_number":38},{"id":"0574511586669b92d710531583018225e33008bc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-10 07:22:25.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/38/check/gate-networking-midonet-docs-ubuntu-xenial/b9d84b0//doc/build/html/ : SUCCESS in 4m 14s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/38/check/gate-networking-midonet-pep8-ubuntu-xenial/ffcd9d7/ : SUCCESS in 16m 41s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/38/check/gate-networking-midonet-python27-ubuntu-xenial/ba47039/ : SUCCESS in 34m 38s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/38/check/gate-networking-midonet-python34/deb0cae/ : SUCCESS in 14m 37s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/38/check/gate-networking-midonet-python35/7750d9a/ : SUCCESS in 12m 31s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/38/check/gate-networking-midonet-releasenotes/329651f//releasenotes/build/html/ : SUCCESS in 4m 12s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/38/check/gate-tempest-dsvm-networking-midonet-v1/3157e96/ : FAILURE in 56m 36s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/38/check/gate-tempest-dsvm-networking-midonet-v2/203260c/ : SUCCESS in 1h 42m 27s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/38/check/gate-tempest-dsvm-networking-midonet-ml2/8f1d7ea/ : FAILURE in 1h 02m 46s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/38/check/gate-networking-midonet-v2-rally-dsvm-nv/7bd66c8/ : FAILURE in 1h 30m 39s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/38/check/networking-midonet-coverage-ubuntu-xenial/cd742ad/ : SUCCESS in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"7a02ba4570f4bcf6f77e81abbfe1bc1fd5ba44b7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-11 12:24:39.000000000","message":"Patch Set 39: Published edit on patch set 38","accounts_in_message":[],"_revision_number":39},{"id":"a2549c4bf2b68a08332231df2147855d1f2428b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-11 14:01:04.000000000","message":"Patch Set 39:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-docs-ubuntu-xenial/f40e1b7//doc/build/html/ : SUCCESS in 10m 41s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-pep8-ubuntu-xenial/5cac7e5/ : SUCCESS in 7m 34s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python27-ubuntu-xenial/cb57400/ : FAILURE in 16m 00s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python34/784ca06/ : FAILURE in 13m 09s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python35/b54b0ae/ : FAILURE in 16m 47s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-releasenotes/dbb7b3e//releasenotes/build/html/ : SUCCESS in 5m 18s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v1/f26b8c5/ : SUCCESS in 59m 39s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v2/98be05c/ : SUCCESS in 1h 28m 54s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-ml2/0e704c4/ : FAILURE in 29m 27s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-v2-rally-dsvm-nv/103de59/ : FAILURE in 1h 16m 40s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/networking-midonet-coverage-ubuntu-xenial/9dda3c0/ : FAILURE in 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"cd6a49ab3149f7fad584f30b558c064d1e6743bb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-12 13:31:49.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"0fae8fc21e4339e3d02d97256827466d52cf6db7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-12 16:07:29.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-docs-ubuntu-xenial/e509b5e//doc/build/html/ : SUCCESS in 4m 05s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-pep8-ubuntu-xenial/d4ec453/ : SUCCESS in 7m 32s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python27-ubuntu-xenial/9998d97/ : SUCCESS in 12m 32s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python34/05dab29/ : SUCCESS in 13m 56s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python35/c9b6b09/ : SUCCESS in 14m 31s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-releasenotes/ab3895d//releasenotes/build/html/ : SUCCESS in 5m 54s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v1/6d798cc/ : SUCCESS in 44m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v2/7e59364/ : SUCCESS in 1h 08m 21s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-ml2/346427d/ : SUCCESS in 1h 25m 05s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-v2-rally-dsvm-nv/30787c0/ : FAILURE in 1h 17m 01s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/networking-midonet-coverage-ubuntu-xenial/254d7ce/ : SUCCESS in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"78b8af1f42872bb718b77b95358736873cb5ca71","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-18 23:58:58.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"c08c84ce3eea195c631bdd0a87629cf31d8bb9aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-19 01:31:38.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-docs-ubuntu-xenial/459df64//doc/build/html/ : SUCCESS in 2m 55s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-pep8-ubuntu-xenial/93f3ae5/ : SUCCESS in 3m 55s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python27-ubuntu-xenial/d9dfb2d/ : SUCCESS in 12m 11s\n- gate-networking-midonet-python34 http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python34/beeea97/ : FAILURE in 12m 11s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python35/780d36e/ : SUCCESS in 12m 51s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-releasenotes/d082dff//releasenotes/build/html/ : SUCCESS in 3m 25s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v1/d946946/ : FAILURE in 31m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v2/fe65114/ : FAILURE in 31m 20s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-ml2/edf15e9/ : FAILURE in 19m 52s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-v2-rally-dsvm-nv/20ed552/ : FAILURE in 51m 10s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/networking-midonet-coverage-ubuntu-xenial/ab03fb2/ : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"bbe6e0987dd1220d9f9f158ca2d638c0d7b72f96","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-22 06:56:17.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"bc0a88af3b591d8d42a5ca18638fc6b0b783a8ef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-22 08:19:40.000000000","message":"Patch Set 39:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-docs-ubuntu-xenial/1f8110b//doc/build/html/ : SUCCESS in 3m 00s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-pep8-ubuntu-xenial/0cb7e2b/ : SUCCESS in 3m 13s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python27-db-ubuntu-xenial/6b3c32e/ : FAILURE in 12m 39s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python34-db/b33517d/ : FAILURE in 13m 28s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python35-db/737a7b3/ : FAILURE in 13m 14s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-releasenotes/c6e0642//releasenotes/build/html/ : SUCCESS in 3m 04s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v1/f322a85/ : SUCCESS in 1h 22m 39s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v2/f7a2612/ : FAILURE in 32m 36s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-ml2/ae478b2/ : FAILURE in 44m 09s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-v2-rally-dsvm-nv/542e104/ : FAILURE in 1h 18m 06s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/networking-midonet-coverage-db-ubuntu-xenial/fd6e3ce/ : FAILURE in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"3459763f1d588f1ea981fab703091e6975094fca","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-22 22:51:52.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"63e01e77f5da064138e8d61ce3e0b4eb864cc28a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-22 23:57:04.000000000","message":"Patch Set 39:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-docs-ubuntu-xenial/08e47af//doc/build/html/ : SUCCESS in 2m 30s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-pep8-ubuntu-xenial/fa92c10/ : SUCCESS in 3m 22s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python27-db-ubuntu-xenial/1be8075/ : FAILURE in 14m 36s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python34-db/a75f763/ : FAILURE in 12m 24s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python35-db/9e14eff/ : FAILURE in 12m 51s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-releasenotes/7f09cc2//releasenotes/build/html/ : SUCCESS in 4m 36s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v1/1653d57/ : SUCCESS in 58m 24s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v2/5691b46/ : FAILURE in 41m 20s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-ml2/3eb7c59/ : FAILURE in 29m 36s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-v2-rally-dsvm-nv/3ed5194/ : FAILURE in 1h 02m 06s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/networking-midonet-coverage-db-ubuntu-xenial/fc88fe0/ : FAILURE in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"848793f5fb47ed0bf9c19cd0f53f613fbd93bcbb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-23 23:10:58.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"e69f1dd462ea853feeb58843eed983da4feff722","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-24 00:24:41.000000000","message":"Patch Set 39:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-docs-ubuntu-xenial/32aebe4//doc/build/html/ : SUCCESS in 2m 52s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-pep8-ubuntu-xenial/aa5c19d/ : SUCCESS in 4m 37s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python27-db-ubuntu-xenial/5ec0f20/ : FAILURE in 16m 08s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python34-db/c4258af/ : FAILURE in 12m 33s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python35-db/980f13b/ : FAILURE in 13m 36s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-releasenotes/40a22fe//releasenotes/build/html/ : SUCCESS in 3m 10s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v1/6890327/ : SUCCESS in 59m 03s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v2/e68a172/ : SUCCESS in 1h 11m 41s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-ml2/41deff3/ : SUCCESS in 1h 00m 15s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-v2-rally-dsvm-nv/aa82eab/ : FAILURE in 1h 00m 00s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/networking-midonet-coverage-db-ubuntu-xenial/dad488a/ : FAILURE in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"27e8614983d695f830c3a9aed7301a6980adb52a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-26 03:24:46.000000000","message":"Patch Set 39:\n\nrecheck","accounts_in_message":[],"_revision_number":39},{"id":"af2b41ee2f080625dbadc8c02ea8c83668f48f2a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-26 05:34:17.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-docs-ubuntu-xenial/4ccad60//doc/build/html/ : SUCCESS in 4m 40s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-pep8-ubuntu-xenial/db05efc/ : SUCCESS in 3m 05s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python27-db-ubuntu-xenial/34f1c78/ : SUCCESS in 15m 16s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python34-db/a44f692/ : SUCCESS in 12m 10s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-python35-db/c3b9093/ : SUCCESS in 12m 45s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/39/check/gate-networking-midonet-releasenotes/0aa56db//releasenotes/build/html/ : SUCCESS in 4m 19s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v1/0ceb362/ : SUCCESS in 42m 07s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-v2/fe11383/ : SUCCESS in 1h 30m 21s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/39/check/gate-tempest-dsvm-networking-midonet-ml2/5488e91/ : FAILURE in 2h 08m 33s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/39/check/gate-networking-midonet-v2-rally-dsvm-nv/7005375/ : FAILURE in 1h 01m 27s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/39/check/networking-midonet-coverage-db-ubuntu-xenial/922a3f1/ : SUCCESS in 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"11ccfd254458c76feff143ef295cfb78684a9441","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-26 06:02:54.000000000","message":"Patch Set 40: Published edit on patch set 39","accounts_in_message":[],"_revision_number":40},{"id":"846d11a048e8b274b471dab4492d143bb73a13c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-26 06:36:03.000000000","message":"Patch Set 40:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":40},{"id":"53aeaef8e19bcc80fa4d5f0e5023263701cfca3b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-26 06:42:05.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"22c51250bf06c58494eddb804225ab6cbb6c95e7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-26 07:59:33.000000000","message":"Patch Set 40:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-docs-ubuntu-xenial/203384a//doc/build/html/ : SUCCESS in 2m 10s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-pep8-ubuntu-xenial/841de76/ : SUCCESS in 2m 53s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python27-db-ubuntu-xenial/e82161b/ : SUCCESS in 11m 24s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python34-db/ad5cafb/ : SUCCESS in 11m 23s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python35-db/6dec689/ : SUCCESS in 12m 14s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-releasenotes/981c352//releasenotes/build/html/ : SUCCESS in 4m 35s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v1/7e0574e/ : FAILURE in 1h 00m 02s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v2/f6d6ed7/ : FAILURE in 1h 02m 22s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-ml2/8fb1121/ : FAILURE in 1h 10m 17s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-v2-rally-dsvm-nv/61e1978/ : FAILURE in 1h 12m 26s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/networking-midonet-coverage-db-ubuntu-xenial/718a03e/ : SUCCESS in 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"72e9321f3a2acad665d30d8dcb97eac6c758a4b9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-30 01:07:51.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"3beba4172df7458630827a442cbcae27f9f0b7dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-30 02:14:02.000000000","message":"Patch Set 40:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-docs-ubuntu-xenial/39d980f//doc/build/html/ : SUCCESS in 3m 35s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-pep8-ubuntu-xenial/c218a6f/ : SUCCESS in 3m 55s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python27-db-ubuntu-xenial/272f9c6/ : SUCCESS in 12m 16s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python34-db/e0bb20d/ : SUCCESS in 11m 44s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python35-db/11fa1ee/ : SUCCESS in 12m 02s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-releasenotes/d572412//releasenotes/build/html/ : SUCCESS in 2m 57s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v1/b07cced/ : SUCCESS in 34m 52s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v2/a358c5a/ : FAILURE in 1h 05m 46s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-ml2/55a95c4/ : SUCCESS in 1h 02m 57s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-v2-rally-dsvm-nv/68176fc/ : FAILURE in 49m 11s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/networking-midonet-coverage-db-ubuntu-xenial/a300477/ : SUCCESS in 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"c4d7899a67b23b6c4d460468a94a104d5904138d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-08-31 04:40:22.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"1345fbdbbb30b5ca9c025225fbd46e7347bb732e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-08-31 08:04:22.000000000","message":"Patch Set 40:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-docs-ubuntu-xenial/d995bf1//doc/build/html/ : SUCCESS in 5m 09s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-pep8-ubuntu-xenial/4f370d3/ : SUCCESS in 6m 49s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python27-db-ubuntu-xenial/451fbff/ : FAILURE in 14m 01s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python34-db/74b6fed/ : FAILURE in 12m 23s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python35-db/ffbc772/ : FAILURE in 17m 28s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-releasenotes/d86ec07//releasenotes/build/html/ : SUCCESS in 6m 56s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v1/6f91913/ : SUCCESS in 35m 18s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v2/5c2b94f/ : FAILURE in 1h 03m 29s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-ml2/3468c19/ : SUCCESS in 1h 03m 37s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-v2-rally-dsvm-nv/3a86854/ : FAILURE in 52m 13s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/networking-midonet-coverage-db-ubuntu-xenial/1f12eec/ : FAILURE in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"ed7f423085b54705e2fa2479fc981c388a7be3eb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-09-01 02:18:13.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"e51b2aaabae0c3d38268d1487eb7684e75816baf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-01 02:33:33.000000000","message":"Patch Set 40:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-docs-ubuntu-xenial/45bfa9c//doc/build/html/ : SUCCESS in 3m 05s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-pep8-ubuntu-xenial/b77b782/ : SUCCESS in 4m 18s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python27-db-ubuntu-xenial/340a7ed/ : FAILURE in 11m 58s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python34-db/bbf8584/ : FAILURE in 15m 00s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python35-db/dd6a31d/ : FAILURE in 13m 32s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-releasenotes/50bb357//releasenotes/build/html/ : SUCCESS in 2m 45s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v1/04a48d3/ : FAILURE in 3m 18s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v2/1e381a7/ : FAILURE in 12m 50s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-ml2/9768241/ : FAILURE in 12m 37s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-v2-rally-dsvm-nv/8d0514a/ : FAILURE in 12m 22s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/networking-midonet-coverage-db-ubuntu-xenial/85440dc/ : FAILURE in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"a0ba23cf4f5160f421a9be6e1c2527351af235b3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-09-05 03:59:25.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"90a537f2408d385148d08f73f68e6f8cd9bb222a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-05 04:48:04.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-docs-ubuntu-xenial/08f57f4//doc/build/html/ : SUCCESS in 2m 07s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-pep8-ubuntu-xenial/bd01a52/ : SUCCESS in 2m 50s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python27-db-ubuntu-xenial/a99fb7d/ : SUCCESS in 12m 00s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python34-db/62eddc2/ : SUCCESS in 12m 33s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python35-db/a1a8a4c/ : SUCCESS in 12m 05s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-releasenotes/461571b//releasenotes/build/html/ : SUCCESS in 2m 42s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v1/70d7f19/ : FAILURE in 23m 17s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v2/22d241b/ : SUCCESS in 39m 11s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-ml2/827adb0/ : FAILURE in 38m 04s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-v2-rally-dsvm-nv/d0c06f9/ : FAILURE in 48m 20s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/networking-midonet-coverage-db-ubuntu-xenial/48e401c/ : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"5657b8ffed486204b7f492208f05f277f29cb21e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-09-05 08:33:00.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"ed962270642107c0cb6fa720d02d1863daaf3960","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-05 09:38:37.000000000","message":"Patch Set 40:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-docs-ubuntu-xenial/6b8e90b//doc/build/html/ : SUCCESS in 2m 17s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-pep8-ubuntu-xenial/76e7ffd/ : SUCCESS in 3m 13s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python27-db-ubuntu-xenial/5429ca6/ : SUCCESS in 13m 37s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python34-db/24b920f/ : SUCCESS in 14m 17s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python35-db/e6da5dc/ : SUCCESS in 12m 26s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-releasenotes/53dcc1e//releasenotes/build/html/ : SUCCESS in 2m 53s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v1/3e8f1c0/ : SUCCESS in 34m 06s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v2/97741db/ : SUCCESS in 1h 04m 51s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-ml2/2675778/ : SUCCESS in 1h 01m 49s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-v2-rally-dsvm-nv/cebf59e/ : FAILURE in 58m 38s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/networking-midonet-coverage-db-ubuntu-xenial/604fc8b/ : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"c8072f68fd68aaad560cb40d50cf6dc38cb2c9fc","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-09-08 06:47:39.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"3d280a8a3be698b9d6fb710fe52152ad5a7b7b88","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-08 08:15:01.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-docs-ubuntu-xenial/72b8ae8//doc/build/html/ : SUCCESS in 4m 50s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-pep8-ubuntu-xenial/ade5973/ : SUCCESS in 4m 47s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python27-db-ubuntu-xenial/3c57184/ : FAILURE in 13m 16s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python34-db/416c056/ : FAILURE in 13m 21s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-python35-db/617f466/ : FAILURE in 11m 30s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/40/check/gate-networking-midonet-releasenotes/29f2805//releasenotes/build/html/ : SUCCESS in 2m 58s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v1/6744b66/ : FAILURE in 47m 14s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-v2/5f14843/ : SUCCESS in 1h 14m 17s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/40/check/gate-tempest-dsvm-networking-midonet-ml2/0313801/ : SUCCESS in 1h 23m 58s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/40/check/gate-networking-midonet-v2-rally-dsvm-nv/c03f376/ : FAILURE in 1h 16m 23s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/40/check/networking-midonet-coverage-db-ubuntu-xenial/3e4ad5c/ : FAILURE in 13m 12s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"e6e9ad5c95ce82953ceb394ec8976a7762b3fa36","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-09-09 08:13:21.000000000","message":"Patch Set 41: Published edit on patch set 40","accounts_in_message":[],"_revision_number":41},{"id":"026e3ac13eec6e2d0c2182a4136d42654a9cf994","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-09 09:35:23.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/41/check/gate-networking-midonet-docs-ubuntu-xenial/e3e6f8c//doc/build/html/ : SUCCESS in 2m 14s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-pep8-ubuntu-xenial/a1e575c/ : SUCCESS in 2m 59s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python27-db-ubuntu-xenial/6d5d480/ : FAILURE in 13m 04s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python34-db/705fd88/ : FAILURE in 11m 39s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python35-db/080ab0d/ : FAILURE in 16m 10s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/41/check/gate-networking-midonet-releasenotes/82858e5//releasenotes/build/html/ : SUCCESS in 4m 17s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-v1/5d84336/ : SUCCESS in 35m 47s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-v2/c23f9e6/ : SUCCESS in 1h 21m 25s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-ml2/6835c73/ : SUCCESS in 1h 15m 56s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-v2-rally-dsvm-nv/4ac4e3f/ : FAILURE in 1h 09m 56s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/networking-midonet-coverage-db-ubuntu-xenial/ccc1c80/ : FAILURE in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"7c1cbd4106e0fce54c7422833d5c4f1ce3275a0f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-09-20 06:59:50.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"0721dca9f78fa88022185f5c958333e52e2ebe31","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-20 08:17:10.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/41/check/gate-networking-midonet-docs-ubuntu-xenial/7b0a99f//doc/build/html/ : SUCCESS in 2m 50s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-pep8-ubuntu-xenial/bfae19f/ : SUCCESS in 2m 59s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python27-db-ubuntu-xenial/bbfe1d9/ : FAILURE in 12m 59s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python34-db/81eff07/ : FAILURE in 12m 23s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python35-db/b98a4b6/ : FAILURE in 13m 31s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/41/check/gate-networking-midonet-releasenotes/d14babb//releasenotes/build/html/ : SUCCESS in 6m 42s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-v1/91ad83a/ : SUCCESS in 36m 03s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-v2/6d086b5/ : SUCCESS in 1h 10m 14s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-ml2/1b559f5/ : SUCCESS in 1h 06m 09s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-v2-rally-dsvm-nv/35526ad/ : FAILURE in 1h 15m 51s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/networking-midonet-coverage-db-ubuntu-xenial/cb198ae/ : FAILURE in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"33a496c4389f970c423f060adca420579dd2d169","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-09-27 06:10:18.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"038220531e82640e8de9491f94aea14464cbe0e3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-27 07:05:56.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/41/check/gate-networking-midonet-docs-ubuntu-xenial/ee42edc//doc/build/html/ : SUCCESS in 3m 09s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-pep8-ubuntu-xenial/91d45fc/ : SUCCESS in 2m 53s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python27-db-ubuntu-xenial/3c9e08b/ : SUCCESS in 8m 52s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python34-db/dfad7e9/ : SUCCESS in 9m 16s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-python35-db/e784c91/ : SUCCESS in 9m 01s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/41/check/gate-networking-midonet-releasenotes/de7d547//releasenotes/build/html/ : SUCCESS in 3m 48s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-v1/17fd033/ : SUCCESS in 31m 31s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-v2/e4d0839/ : SUCCESS in 53m 33s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/41/check/gate-tempest-dsvm-networking-midonet-ml2/9ae8562/ : SUCCESS in 52m 54s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/41/check/gate-networking-midonet-v2-rally-dsvm-nv/42ae15b/ : FAILURE in 54m 31s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/41/check/networking-midonet-coverage-db-ubuntu-xenial/0bfe500/ : SUCCESS in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"926f6d103ebe11ffdb6f9938b18df67eac5b0a7f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-09-30 09:39:10.000000000","message":"Patch Set 42: Published edit on patch set 41","accounts_in_message":[],"_revision_number":42},{"id":"63cb682cc97974fbd9d6dfd3dd31d9d5c4e1566f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-30 10:38:33.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/42/check/gate-networking-midonet-docs-ubuntu-xenial/c71dab1//doc/build/html/ : SUCCESS in 3m 14s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/42/check/gate-networking-midonet-pep8-ubuntu-xenial/f857525/ : SUCCESS in 2m 53s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/42/check/gate-networking-midonet-python27-db-ubuntu-xenial/723300c/ : SUCCESS in 9m 13s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/42/check/gate-networking-midonet-python34-db/4609b18/ : SUCCESS in 9m 17s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/42/check/gate-networking-midonet-python35-db/a08b4c4/ : SUCCESS in 8m 56s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/42/check/gate-networking-midonet-releasenotes/28d7f0f//releasenotes/build/html/ : SUCCESS in 3m 56s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/42/check/gate-tempest-dsvm-networking-midonet-v1/bd3120f/ : SUCCESS in 32m 51s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/42/check/gate-tempest-dsvm-networking-midonet-v2/3454b47/ : SUCCESS in 58m 05s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/42/check/gate-tempest-dsvm-networking-midonet-ml2/046a5a0/ : SUCCESS in 57m 51s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/42/check/gate-networking-midonet-v2-rally-dsvm-nv/4c53f87/ : FAILURE in 54m 45s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/42/check/networking-midonet-coverage-db-ubuntu-xenial/d39fa1c/ : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"c4bce56b73d5cd2db9fd6db1362e9b1f6102747a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-10-03 06:11:32.000000000","message":"Patch Set 43: Published edit on patch set 42","accounts_in_message":[],"_revision_number":43},{"id":"d243447a94acb2a407b25de7da3dc4fb8aee92ad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-03 06:45:48.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/43/check/gate-networking-midonet-docs-ubuntu-xenial/1ccc02b//doc/build/html/ : SUCCESS in 2m 02s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-pep8-ubuntu-xenial/56d2c3c/ : SUCCESS in 4m 32s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python27-db-ubuntu-xenial/50003c4/ : SUCCESS in 12m 01s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python34-db/2c4b08a/ : SUCCESS in 12m 44s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python35-db/766b4b8/ : SUCCESS in 9m 13s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/43/check/gate-networking-midonet-releasenotes/6a25c88//releasenotes/build/html/ : SUCCESS in 2m 39s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-v1/7d4468e/ : FAILURE in 24m 50s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-v2/0a2d8d1/ : FAILURE in 32m 42s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-ml2/40de1b5/ : FAILURE in 30m 15s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-v2-rally-dsvm-nv/7b1bf1b/ : FAILURE in 34m 04s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/networking-midonet-coverage-db-ubuntu-xenial/752d936/ : SUCCESS in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"2b6528a8abc01a8d3b5f423cae01c6a4e05517f5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-10-04 08:44:31.000000000","message":"Patch Set 43:\n\nrecheck","accounts_in_message":[],"_revision_number":43},{"id":"1d9c392c6cde95c94de1897dd57bd4b473a30dce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-04 09:44:43.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/43/check/gate-networking-midonet-docs-ubuntu-xenial/b802fa8//doc/build/html/ : SUCCESS in 2m 57s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-pep8-ubuntu-xenial/557cd36/ : SUCCESS in 3m 33s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python27-db-ubuntu-xenial/fd8d68d/ : SUCCESS in 9m 08s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python34-db/2b680f9/ : SUCCESS in 12m 16s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python35-db/f14dd66/ : SUCCESS in 9m 00s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/43/check/gate-networking-midonet-releasenotes/6f90ed3//releasenotes/build/html/ : SUCCESS in 2m 53s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-v1/7889b41/ : SUCCESS in 31m 32s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-v2/c965cb2/ : SUCCESS in 58m 27s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-ml2/fc3da69/ : SUCCESS in 56m 00s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-v2-rally-dsvm-nv/aa25844/ : FAILURE in 58m 26s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/networking-midonet-coverage-db-ubuntu-xenial/3e56299/ : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"48f571127712c144be8292a45aa602c0e4c47e15","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-10-13 10:21:39.000000000","message":"Patch Set 43:\n\nrecheck","accounts_in_message":[],"_revision_number":43},{"id":"16fdd16e2aa76374576f073db7bdd0481b3847ea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-13 11:20:15.000000000","message":"Patch Set 43:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/43/check/gate-networking-midonet-docs-ubuntu-xenial/1466afb//doc/build/html/ : SUCCESS in 3m 27s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-pep8-ubuntu-xenial/83d53bb/ : SUCCESS in 4m 38s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python27-db-ubuntu-xenial/97adfc8/ : SUCCESS in 14m 12s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python34-db/e497800/ : SUCCESS in 8m 37s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-python35-db/b618985/ : SUCCESS in 9m 36s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/43/check/gate-networking-midonet-releasenotes/d0349a6//releasenotes/build/html/ : SUCCESS in 2m 57s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-v1/5f1f0b0/ : SUCCESS in 31m 02s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-v2/df9b5d3/ : SUCCESS in 58m 24s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/43/check/gate-tempest-dsvm-networking-midonet-ml2/79f8a7a/ : SUCCESS in 57m 57s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/43/check/gate-networking-midonet-v2-rally-dsvm-nv/cedf148/ : FAILURE in 57m 42s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/43/check/networking-midonet-coverage-db-ubuntu-xenial/44cce97/ : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"fed6f9671c74ac3e43558845ecd3615e2ea7db3c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-10-27 08:06:07.000000000","message":"Patch Set 44: Published edit on patch set 43","accounts_in_message":[],"_revision_number":44},{"id":"39bd93419c773a3c40ee6b4e3b91da424f2fe2b6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-27 09:10:18.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/44/check/gate-networking-midonet-docs-ubuntu-xenial/157941e//doc/build/html/ : SUCCESS in 2m 12s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-pep8-ubuntu-xenial/7fb55c6/ : SUCCESS in 6m 09s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-python27-db-ubuntu-xenial/f128254/ : SUCCESS in 8m 18s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-python34-db/c2da20b/ : SUCCESS in 10m 24s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-python35-db/a588611/ : SUCCESS in 8m 54s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/44/check/gate-networking-midonet-releasenotes/6f85ada//releasenotes/build/html/ : SUCCESS in 2m 21s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/44/check/gate-tempest-dsvm-networking-midonet-v1/13211dc/ : FAILURE in 1h 03m 51s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/44/check/gate-tempest-dsvm-networking-midonet-v2/d140f85/ : SUCCESS in 51m 18s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/44/check/gate-tempest-dsvm-networking-midonet-ml2/aea0593/ : SUCCESS in 51m 08s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-v2-rally-dsvm-nv/25b3bd7/ : FAILURE in 51m 19s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/44/check/networking-midonet-coverage-db-ubuntu-xenial/359def2/ : SUCCESS in 9m 51s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"2602143a4d1b468edb02d5216e278647a2ae51b7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-10-27 15:51:58.000000000","message":"Patch Set 44:\n\nrecheck","accounts_in_message":[],"_revision_number":44},{"id":"4dc17ad4384fad57b6fd2eeb08112247116aaa66","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-27 17:04:20.000000000","message":"Patch Set 44:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/44/check/gate-networking-midonet-docs-ubuntu-xenial/af12b8b//doc/build/html/ : SUCCESS in 1m 46s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-pep8-ubuntu-xenial/5f63d24/ : SUCCESS in 4m 47s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-python27-db-ubuntu-xenial/4015280/ : SUCCESS in 7m 58s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-python34-db/aa287ea/ : SUCCESS in 11m 38s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-python35-db/132c6f4/ : SUCCESS in 9m 47s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/44/check/gate-networking-midonet-releasenotes/bcb5494//releasenotes/build/html/ : SUCCESS in 2m 22s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/44/check/gate-tempest-dsvm-networking-midonet-v1/b8befd9/ : SUCCESS in 30m 06s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/44/check/gate-tempest-dsvm-networking-midonet-v2/0f676a2/ : SUCCESS in 1h 11m 58s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/44/check/gate-tempest-dsvm-networking-midonet-ml2/4a759c3/ : SUCCESS in 51m 15s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/44/check/gate-networking-midonet-v2-rally-dsvm-nv/30eda11/ : FAILURE in 50m 36s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/44/check/networking-midonet-coverage-db-ubuntu-xenial/eab8dcb/ : SUCCESS in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"8caedb38807a4fe68a0cc33deebfaf6366d4ac4e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-01 12:16:34.000000000","message":"Patch Set 45: Published edit on patch set 44","accounts_in_message":[],"_revision_number":45},{"id":"7e05705d6f56a916507ad23f14f3a669faed04f8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-01 13:13:06.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/c1dbe10//doc/build/html/ : SUCCESS in 2m 27s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/e3fbc7f/ : SUCCESS in 3m 52s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/28ce51b/ : SUCCESS in 9m 09s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/210d506/ : SUCCESS in 12m 01s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/c6c54f8/ : SUCCESS in 10m 11s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/711b0ea//releasenotes/build/html/ : SUCCESS in 2m 25s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v1/4ae6f3e/ : SUCCESS in 40m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2/ade71bb/ : SUCCESS in 53m 36s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2/76e4ca1/ : SUCCESS in 56m 05s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-v2-rally-dsvm-nv/71dfea1/ : FAILURE in 51m 29s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/ee59a0c/ : FAILURE in 3m 10s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"3b48f230c949741def4e6bf0cfe6720e1ea8f374","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-03 00:23:59.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"a2f47499b22f7969fff6375ec632c386034f9359","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-03 01:15:46.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/f17ff0c//doc/build/html/ : SUCCESS in 1m 41s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/b9a97eb/ : SUCCESS in 3m 01s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/557a41e/ : SUCCESS in 8m 03s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/66cdefa/ : SUCCESS in 9m 02s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/41f8098/ : SUCCESS in 9m 24s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/fccc441//releasenotes/build/html/ : SUCCESS in 2m 52s\n- gate-tempest-dsvm-networking-midonet-v1 http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v1/43b4c2b/ : FAILURE in 41m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-v2 http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2/765cc8c/ : FAILURE in 48m 41s\n- gate-tempest-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2/b0b8e42/ : SUCCESS in 51m 35s (non-voting)\n- gate-networking-midonet-v2-rally-dsvm-nv http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-v2-rally-dsvm-nv/d5f0261/ : FAILURE in 51m 28s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/f4360db/ : FAILURE in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"68e6f5435c20ada22a8c653f42ae9b3263ccd8f6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-08 05:20:40.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"c1bb103df669a154733344123cbcd3eca7a70265","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-08 06:19:12.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/e7d76ac//doc/build/html/ : SUCCESS in 1m 47s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/f9103f4/ : SUCCESS in 5m 03s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/4d7d54d/ : SUCCESS in 9m 16s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/903b9ec/ : SUCCESS in 10m 17s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/2bbb66d/ : SUCCESS in 8m 44s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/ca7656e//releasenotes/build/html/ : SUCCESS in 2m 36s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/bc70327/ : SUCCESS in 51m 45s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/4256ec2/ : SUCCESS in 58m 20s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/2d2934b/ : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"5ecf66d89562a8b486bb6adb9de1a14c900e699f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-13 14:58:06.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"c92ecae93fb859b71241e7543191f41fbd1a8517","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-13 15:28:19.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/363340e//doc/build/html/ : SUCCESS in 2m 47s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/aa8596d/ : SUCCESS in 3m 13s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/deb6fa2/ : FAILURE in 2m 42s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/7c67313/ : FAILURE in 3m 02s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/60e991e/ : FAILURE in 2m 56s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/50278ab//releasenotes/build/html/ : SUCCESS in 2m 33s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/2170a5f/ : FAILURE in 30m 01s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/9ea31f7/ : FAILURE in 24m 29s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/006f0aa/ : FAILURE in 2m 47s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"67f398a01180e0de6c2391aa68c797176a04c8d8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-14 13:13:54.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"dafb68903b05a18a132eee92f492e71bb6cb734b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-14 13:56:47.000000000","message":"Patch Set 45:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/5c4bb9a//doc/build/html/ : SUCCESS in 1m 51s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/dd0d1a3/ : SUCCESS in 4m 55s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/2ac9499/ : FAILURE in 4m 44s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/a3451b6/ : FAILURE in 3m 15s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/306826e/ : FAILURE in 4m 57s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/3a37681//releasenotes/build/html/ : SUCCESS in 3m 59s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/357e7f7/ : FAILURE in 36m 18s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/8dd03b2/ : FAILURE in 39m 45s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/4214d94/ : FAILURE in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"8d9d2cfaf55ee0bd57400b7e53c9ddb5e0ff862a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-15 03:29:22.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"02f046af791213106776db961db56f40368c2b8a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-15 04:08:44.000000000","message":"Patch Set 45:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/96ecf0d//doc/build/html/ : SUCCESS in 3m 02s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/a7087e6/ : SUCCESS in 3m 07s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/843d9e0/ : FAILURE in 7m 25s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/15d935a/ : FAILURE in 3m 30s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/cbcea46/ : FAILURE in 6m 18s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/aa24d65//releasenotes/build/html/ : SUCCESS in 5m 43s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/ee05f44/ : FAILURE in 31m 16s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/23b2e29/ : FAILURE in 38m 16s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/78f9ea7/ : FAILURE in 2m 56s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"5e63c4f99a64d875711cf84fac22da69d7d36fdb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-17 05:46:50.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"01960e90e9a0d0a40b18ca3891408f9f8543f5d3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-17 06:26:03.000000000","message":"Patch Set 45:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/10e89a9//doc/build/html/ : SUCCESS in 1m 48s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/4875240/ : SUCCESS in 3m 09s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/f5f1e1b/ : FAILURE in 2m 49s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/cc3dde8/ : FAILURE in 5m 30s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/10f1c4c/ : FAILURE in 2m 59s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/9ef3a17//releasenotes/build/html/ : SUCCESS in 4m 41s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/33b47ea/ : FAILURE in 39m 01s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/c2dec22/ : FAILURE in 24m 36s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/d25e00e/ : FAILURE in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"568e4abaf56b380ce738c5aa24d37c912faebb08","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-18 04:55:20.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"00b46bf9d5ffa496d2f7106dcf7171b417c1e05c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-18 05:24:24.000000000","message":"Patch Set 45:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/bd42221//doc/build/html/ : SUCCESS in 2m 48s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/0ff29f0/ : SUCCESS in 3m 08s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/0ade2d8/ : FAILURE in 2m 54s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/d823845/ : FAILURE in 3m 07s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/04ab747/ : FAILURE in 4m 24s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/4c9a894//releasenotes/build/html/ : SUCCESS in 2m 41s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/282662e/ : FAILURE in 28m 35s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/ff50f8b/ : FAILURE in 25m 27s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/8befd77/ : FAILURE in 2m 49s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"18880192636ef57fbb94acf9a50421f36f32800c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-22 08:42:12.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"ea66dd835784ca7d34df3a264dd4d271a8f6eeb7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-22 08:42:21.000000000","message":"Patch Set 45:\n\ncheck experimental","accounts_in_message":[],"_revision_number":45},{"id":"dcbf79905164203fd6d0a683895f59279b6d14c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-22 09:40:58.000000000","message":"Patch Set 45:\n\nBuild succeeded (experimental pipeline).\n\n- gate-networking-midonet-v2-rally-dsvm-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/experimental/gate-networking-midonet-v2-rally-dsvm-ubuntu-xenial-nv/e99a58d/ : FAILURE in 56m 15s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/experimental/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/a779aa4/ : FAILURE in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"c37e69bda96e7c0c51bab0b538b1e47491447517","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-22 09:49:50.000000000","message":"Patch Set 45:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/7340d3b//doc/build/html/ : SUCCESS in 2m 06s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/ea27f48/ : SUCCESS in 3m 27s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/2b2cfde/ : SUCCESS in 8m 49s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/37c004f/ : SUCCESS in 9m 38s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/98f72de/ : SUCCESS in 9m 52s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/b015f6f//releasenotes/build/html/ : SUCCESS in 2m 47s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/cb4f86d/ : FAILURE in 1h 05m 09s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/2a33633/ : SUCCESS in 51m 09s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/44b4eed/ : SUCCESS in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"e5da40d6c8f41c917443a9f73938a0a7268a81a1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-23 13:50:21.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"8d4f77c3a9145c9fdda230db42eb9d828cf258f6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-23 14:45:15.000000000","message":"Patch Set 45:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/2afba3f//doc/build/html/ : SUCCESS in 2m 42s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/2feba2f/ : SUCCESS in 4m 11s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/6d7984d/ : FAILURE in 9m 13s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/4a42cf5/ : FAILURE in 12m 37s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/ba15fa3/ : FAILURE in 9m 59s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/46e10d1//releasenotes/build/html/ : SUCCESS in 2m 47s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/4a971d5/ : SUCCESS in 51m 33s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/310a011/ : FAILURE in 36m 43s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/a1f5419/ : FAILURE in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"d92644bf66aaeec5ffef749eb430c13f371ac67f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-11-29 04:51:31.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"1b7038e0f663d10a3914fafca796864fe1fef297","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-29 05:48:03.000000000","message":"Patch Set 45:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/18e1f7d//doc/build/html/ : SUCCESS in 1m 48s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/a690b76/ : SUCCESS in 4m 47s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/9957121/ : FAILURE in 11m 35s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/ce998af/ : FAILURE in 7m 36s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/5c3f376/ : FAILURE in 13m 29s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/7604c04//releasenotes/build/html/ : SUCCESS in 4m 36s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/6c2f841/ : FAILURE in 56m 20s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/62e13bf/ : FAILURE in 55m 59s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/01db834/ : FAILURE in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"fba41757a92128b63132326ece37b274032f2cde","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-02 04:57:02.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"6d3a12743b2cc36e39c2b791976dcaae5897f3bb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-02 06:21:37.000000000","message":"Patch Set 45:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/0a17d6c//doc/build/html/ : SUCCESS in 4m 34s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/d090d00/ : SUCCESS in 5m 21s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/48706b3/ : FAILURE in 8m 35s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/67bf38a/ : FAILURE in 11m 11s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/471071d/ : FAILURE in 7m 40s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/5709795//releasenotes/build/html/ : SUCCESS in 2m 43s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/bf6c645/ : SUCCESS in 1h 22m 21s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/afd6a67/ : SUCCESS in 1h 10m 17s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/b2e4450/ : FAILURE in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"f36ce2816994d861b3b75b43ac4493d2c380f12c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-04 11:01:33.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"87c92c811097b5917b7345648f1b130c12700ed6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-04 11:58:14.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/2c95e09//doc/build/html/ : SUCCESS in 1m 50s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/063a151/ : SUCCESS in 3m 38s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/af55a05/ : SUCCESS in 10m 20s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/c004a07/ : SUCCESS in 11m 21s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/476adb2/ : SUCCESS in 10m 40s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/b1bae47//releasenotes/build/html/ : SUCCESS in 2m 37s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/3cfa618/ : SUCCESS in 51m 12s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/8076c10/ : SUCCESS in 56m 29s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/be7880a/ : SUCCESS in 7m 30s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"695653e225a54186a4ed682c9c9d9fe23abd4ad7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-05 03:55:13.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"1421a914b6ba029f4e50f75e6e44ec316b22b794","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-05 05:16:23.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/df8947a//doc/build/html/ : SUCCESS in 1m 52s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/53470eb/ : SUCCESS in 3m 45s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/d6ab11c/ : SUCCESS in 7m 46s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/fffffae/ : SUCCESS in 7m 49s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/3b9f80d/ : SUCCESS in 7m 31s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/3a87be0//releasenotes/build/html/ : SUCCESS in 4m 46s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/7618421/ : SUCCESS in 1h 12m 28s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/382db19/ : SUCCESS in 1h 20m 09s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/fe509df/ : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"5a99d0940ca7242abff0e9beca908cce70d5a32f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-08 03:59:12.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"22dd6c3e21e79c3f2be8aa039ac48c052b400341","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-08 05:18:14.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/c0cadbe//doc/build/html/ : SUCCESS in 3m 31s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/dad0694/ : SUCCESS in 4m 22s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/98ba6a2/ : SUCCESS in 8m 18s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/6bd2e06/ : SUCCESS in 7m 56s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/bc156d6/ : SUCCESS in 9m 23s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/dead041//releasenotes/build/html/ : SUCCESS in 3m 12s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/8b85ba8/ : SUCCESS in 57m 21s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/7ac07c2/ : SUCCESS in 1h 18m 20s\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/1dfe7b4/ : FAILURE in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"d6fbc97e7585470a75227382b7eed65b6e30a937","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-09 16:39:15.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"cb31e51313345b3ad7f188385617ea586cac64dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-09 18:42:06.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/6b4404e//doc/build/html/ : SUCCESS in 5m 39s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/aa350ab/ : SUCCESS in 6m 04s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/0c95ba6/ : SUCCESS in 11m 29s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/acc9a61/ : SUCCESS in 8m 44s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/09f7000/ : SUCCESS in 11m 11s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/740f938//releasenotes/build/html/ : SUCCESS in 3m 54s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/bcbf130/ : FAILURE in 58m 56s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/b157951/ : FAILURE in 1h 57m 51s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/dfaa323/ : SUCCESS in 1h 02m 17s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/3b17ba5/ : SUCCESS in 57m 19s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/693e2fd/ : SUCCESS in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"1dc1d51afcdba4ae1a95b8a3d9cbcbb46dd21127","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-12 00:30:47.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"52f57cdfc153f0b55a5653c1620e579b906ba2d6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-12 02:03:38.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/cfe9178//doc/build/html/ : SUCCESS in 3m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/3278bb5/ : SUCCESS in 3m 42s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/8cc153a/ : SUCCESS in 8m 20s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/95c0b2a/ : SUCCESS in 8m 52s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/c814633/ : SUCCESS in 9m 15s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/d27b0ff//releasenotes/build/html/ : SUCCESS in 3m 43s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/3e92128/ : SUCCESS in 1h 13m 31s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/9ee6d80/ : SUCCESS in 1h 09m 22s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/7d1cc7f/ : SUCCESS in 1h 32m 38s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/4a5b0a6/ : SUCCESS in 1h 22m 08s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/bc4947a/ : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"80276eb0b0389663fecf48f3b3e4ac92062ee9fb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-12 02:39:19.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"b0c6e912597b0606b86dc09f2af679ec3961f276","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-12 03:57:18.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-docs-ubuntu-xenial/6aaf969//doc/build/html/ : SUCCESS in 3m 22s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-pep8-ubuntu-xenial/86fdd72/ : SUCCESS in 5m 51s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python27-db-ubuntu-xenial/807de82/ : SUCCESS in 7m 45s\n- gate-networking-midonet-python34-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python34-db/118fbad/ : SUCCESS in 8m 56s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/45/check/gate-networking-midonet-python35-db/f747ddb/ : SUCCESS in 11m 50s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/45/check/gate-networking-midonet-releasenotes/d8a8cde//releasenotes/build/html/ : SUCCESS in 5m 27s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/8dd7e06/ : SUCCESS in 51m 25s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/ad1c5bd/ : SUCCESS in 1h 17m 08s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/412c49b/ : SUCCESS in 57m 22s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/df23a51/ : SUCCESS in 57m 56s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/45/check/networking-midonet-coverage-db-ubuntu-xenial/6f8a9d4/ : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"c7a70c2d2e7a35e1f43385fb843f6098d9d88d12","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-12 04:09:56.000000000","message":"Patch Set 45:\n\ncheck experimental","accounts_in_message":[],"_revision_number":45},{"id":"b3cde21b2ddb9d6dff3a7b2fb8d880feb8ca4d30","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-12 04:35:18.000000000","message":"Patch Set 45:\n\nBuild succeeded (experimental pipeline).\n\n- gate-networking-midonet-v2-rally-dsvm-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/experimental/gate-networking-midonet-v2-rally-dsvm-ubuntu-xenial-nv/ced8034/ : FAILURE in 25m 05s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/45/experimental/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/f0ffc5d/ : FAILURE in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"48c52d1861eaa97f9e34685c73083c27c9419f85","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2016-12-14 00:10:06.000000000","message":"Patch Set 46: Published edit on patch set 45","accounts_in_message":[],"_revision_number":46},{"id":"ffda5def399a3aa9a94cb5cdf112cad8dd52033c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-14 01:12:37.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/46/check/gate-networking-midonet-docs-ubuntu-xenial/b7571e1//doc/build/html/ : SUCCESS in 3m 46s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-pep8-ubuntu-xenial/d908399/ : SUCCESS in 5m 32s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-python27-db-ubuntu-xenial/7dc12d5/ : SUCCESS in 8m 55s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-python35-db/a75aca0/ : SUCCESS in 13m 14s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/46/check/gate-networking-midonet-releasenotes/4af1acb//releasenotes/build/html/ : SUCCESS in 6m 53s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/5cc3264/ : SUCCESS in 1h 00m 04s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/332978a/ : SUCCESS in 1h 00m 52s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/d8d797a/ : SUCCESS in 53m 41s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/943eb53/ : SUCCESS in 1h 00m 48s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/networking-midonet-coverage-db-ubuntu-xenial/57b3418/ : SUCCESS in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"135b9cdf8738fb58919a57044976ab987bfe1e22","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-04 02:22:36.000000000","message":"Patch Set 46:\n\nrecheck","accounts_in_message":[],"_revision_number":46},{"id":"121f74e525ba70011f66040534894a738742fe19","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-04 04:12:50.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/46/check/gate-networking-midonet-docs-ubuntu-xenial/d9fe0df//doc/build/html/ : SUCCESS in 3m 47s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-pep8-ubuntu-xenial/7ff3be4/ : SUCCESS in 5m 07s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-python27-db-ubuntu-xenial/abc82d8/ : SUCCESS in 8m 27s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-python35-db/843b170/ : SUCCESS in 8m 56s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/46/check/gate-networking-midonet-releasenotes/452bcc6//releasenotes/build/html/ : SUCCESS in 3m 45s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/7e7ba11/ : FAILURE in 1h 04m 14s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/6311a0e/ : SUCCESS in 1h 01m 51s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/7df7066/ : SUCCESS in 1h 04m 52s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/248c6e2/ : SUCCESS in 1h 06m 55s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/networking-midonet-coverage-db-ubuntu-xenial/14c532d/ : SUCCESS in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"24ed66a1e89fba745c07306eba9603998a740b67","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-01-04 04:13:03.000000000","message":"Patch Set 46:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial: https://bugs.launchpad.net/bugs/1323658\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":46},{"id":"d132c5acff3a710a0e81177507ac04565b434976","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-05 01:10:12.000000000","message":"Patch Set 46:\n\nrecheck","accounts_in_message":[],"_revision_number":46},{"id":"5f5086a61f1776854e0207b99bf847cf0f343b65","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-05 02:35:42.000000000","message":"Patch Set 46:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/46/check/gate-networking-midonet-docs-ubuntu-xenial/28688df//doc/build/html/ : SUCCESS in 3m 28s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-pep8-ubuntu-xenial/e953cf4/ : SUCCESS in 7m 58s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-python27-db-ubuntu-xenial/c8ef25f/ : SUCCESS in 11m 51s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/46/check/gate-networking-midonet-python35-db/96fca9d/ : SUCCESS in 8m 38s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/46/check/gate-networking-midonet-releasenotes/6c4da29//releasenotes/build/html/ : SUCCESS in 3m 29s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/f1cb74e/ : SUCCESS in 1h 10m 35s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/60e6bde/ : SUCCESS in 1h 02m 24s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/39d97e7/ : FAILURE in 1h 24m 32s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/46/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/958b3e7/ : SUCCESS in 1h 20m 46s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/46/check/networking-midonet-coverage-db-ubuntu-xenial/3e974ed/ : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"6f8f72dfa144272a28442c7f6830438c27f19925","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-01-05 02:47:43.000000000","message":"Patch Set 46:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial: https://bugs.launchpad.net/bugs/1532809\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":46},{"id":"4938ac4bf595590f001a0e897f1821bc95ffd08a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-13 08:10:25.000000000","message":"Patch Set 47: Published edit on patch set 46","accounts_in_message":[],"_revision_number":47},{"id":"07e3bdffc95d75166877a63d15c6f259d04f9b03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-13 08:31:49.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/47/check/gate-networking-midonet-docs-ubuntu-xenial/25da220//doc/build/html/ : SUCCESS in 3m 25s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/47/check/gate-networking-midonet-pep8-ubuntu-xenial/85099d1/ : SUCCESS in 4m 49s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/47/check/gate-networking-midonet-python27-db-ubuntu-xenial/e995a5f/ : SUCCESS in 7m 41s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/47/check/gate-networking-midonet-python35-db/ba9dc54/ : SUCCESS in 8m 12s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/47/check/gate-networking-midonet-releasenotes/05b6cd0//releasenotes/build/html/ : SUCCESS in 3m 53s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/47/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/2956631/ : FAILURE in 20m 30s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/47/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/bafd282/ : FAILURE in 20m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/47/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/28d5a34/ : FAILURE in 13m 43s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/47/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/c27d5f9/ : FAILURE in 14m 25s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/47/check/networking-midonet-coverage-db-ubuntu-xenial/6b93171/ : FAILURE in 7m 02s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"ed5b24a4b5bab3dd6894b61a3ba2e6413bd8901b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-13 09:29:39.000000000","message":"Patch Set 48: Published edit on patch set 47","accounts_in_message":[],"_revision_number":48},{"id":"e30e99abf2400ed6a35635c4410e7b8d25d69ec2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-13 09:45:59.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/48/check/gate-networking-midonet-docs-ubuntu-xenial/b2186c8//doc/build/html/ : SUCCESS in 6m 30s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/48/check/gate-networking-midonet-pep8-ubuntu-xenial/251c572/ : SUCCESS in 5m 11s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/48/check/gate-networking-midonet-python27-db-ubuntu-xenial/77ce967/ : FAILURE in 7m 41s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/48/check/gate-networking-midonet-python35-db/e687f60/ : SUCCESS in 12m 42s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/48/check/gate-networking-midonet-releasenotes/da3bd07//releasenotes/build/html/ : SUCCESS in 3m 39s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/48/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/a52be09/ : FAILURE in 14m 51s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/48/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/b2ea356/ : FAILURE in 14m 09s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/48/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/7056d73/ : FAILURE in 13m 21s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/48/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/cabc00c/ : FAILURE in 13m 56s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/48/check/networking-midonet-coverage-db-ubuntu-xenial/2b024d7/ : FAILURE in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"e28d8ba2f915fbe3c3ac66a762f652b47f95f93d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-17 03:14:16.000000000","message":"Patch Set 49: Published edit on patch set 48","accounts_in_message":[],"_revision_number":49},{"id":"24b355651139d6ece66f22c98559f6b96f486a5e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-17 03:30:37.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/49/check/gate-networking-midonet-docs-ubuntu-xenial/7214fdb//doc/build/html/ : SUCCESS in 6m 03s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/49/check/gate-networking-midonet-pep8-ubuntu-xenial/41fae35/ : SUCCESS in 8m 03s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/49/check/gate-networking-midonet-python27-db-ubuntu-xenial/3d8ef7a/ : SUCCESS in 15m 36s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/49/check/gate-networking-midonet-python35-db/7bfc616/ : SUCCESS in 11m 21s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/49/check/gate-networking-midonet-releasenotes/6110095//releasenotes/build/html/ : SUCCESS in 4m 06s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/49/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/9635fc6/ : FAILURE in 13m 30s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/49/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/1b3a1b8/ : FAILURE in 14m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/49/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/3d387ce/ : FAILURE in 13m 56s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/49/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/807c1f2/ : FAILURE in 15m 15s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/49/check/networking-midonet-coverage-db-ubuntu-xenial/c9c5a43/ : SUCCESS in 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"1c3e0d5cf1e999f6a27cd7d12851fb0f7ebe105f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-23 22:30:55.000000000","message":"Patch Set 50: Published edit on patch set 49","accounts_in_message":[],"_revision_number":50},{"id":"5ee296eaee720fcc6e47379be242b6618e64b111","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-23 23:50:02.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/50/check/gate-networking-midonet-docs-ubuntu-xenial/272b454//doc/build/html/ : SUCCESS in 3m 29s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/50/check/gate-networking-midonet-pep8-ubuntu-xenial/cb1f7f8/ : SUCCESS in 9m 28s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/50/check/gate-networking-midonet-python27-db-ubuntu-xenial/f02243e/ : SUCCESS in 11m 16s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/50/check/gate-networking-midonet-python35-db/d904637/ : SUCCESS in 11m 41s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/50/check/gate-networking-midonet-releasenotes/3e00d7c//releasenotes/build/html/ : SUCCESS in 3m 14s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/50/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/d49b1b6/ : FAILURE in 54m 28s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/50/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/75debaf/ : FAILURE in 1h 00m 09s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/50/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/b9c98db/ : FAILURE in 1h 16m 54s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/50/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/5156c50/ : FAILURE in 55m 24s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/50/check/networking-midonet-coverage-db-ubuntu-xenial/369b49a/ : SUCCESS in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"8d24826944ccf3c83a4f30839bd02b109dd4b6f8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-24 06:58:19.000000000","message":"Patch Set 51: Published edit on patch set 50","accounts_in_message":[],"_revision_number":51},{"id":"da2af50d1fd5d7e356be20a7dcdb37c6c8ec6445","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 07:11:33.000000000","message":"Patch Set 51: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":51},{"id":"3690cfaffd12e269e3d0951c543c6a92995cf210","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-24 07:12:05.000000000","message":"Patch Set 51:\n\nrecheck","accounts_in_message":[],"_revision_number":51},{"id":"646cd623bc4b1c89a9f381d5569d6cb6c0611839","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 08:39:23.000000000","message":"Patch Set 51:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":51},{"id":"3af5b7eb2b979621dde3c536a8891dbdb725f4d9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-25 03:36:47.000000000","message":"Patch Set 52: Published edit on patch set 51","accounts_in_message":[],"_revision_number":52},{"id":"e2bd37a8c5b54f1a694fbea21c6fe0a22cd66379","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-25 05:03:34.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-docs-ubuntu-xenial/db76da6//doc/build/html/ : SUCCESS in 3m 48s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-pep8-ubuntu-xenial/388d837/ : SUCCESS in 5m 01s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python27-db-ubuntu-xenial/53c694a/ : SUCCESS in 11m 29s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python35-db/86233d6/ : SUCCESS in 8m 33s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-releasenotes/c42db0a//releasenotes/build/html/ : SUCCESS in 3m 16s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/ea77f17/ : FAILURE in 1h 23m 01s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/7a9827c/ : FAILURE in 1h 15m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/2696e7b/ : FAILURE in 59m 29s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/84a5138/ : FAILURE in 1h 24m 22s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/networking-midonet-coverage-db-ubuntu-xenial/0bb1f88/ : SUCCESS in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"401548357c8ceb9b8039f8249bc3000d3cd33954","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-26 07:13:16.000000000","message":"Patch Set 52:\n\nrecheck","accounts_in_message":[],"_revision_number":52},{"id":"6988ca322a4cc665f68f17cff6eb595ad2df36c0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 08:29:24.000000000","message":"Patch Set 52:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-docs-ubuntu-xenial/1050d09//doc/build/html/ : SUCCESS in 3m 08s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-pep8-ubuntu-xenial/1eca993/ : SUCCESS in 5m 05s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python27-db-ubuntu-xenial/6e7e6df/ : SUCCESS in 9m 14s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python35-db/a8cf7f4/ : SUCCESS in 8m 12s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-releasenotes/0b9c033//releasenotes/build/html/ : SUCCESS in 3m 13s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/3540bb0/ : FAILURE in 1h 12m 18s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/e459461/ : FAILURE in 1h 02m 10s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/1d1e15d/ : FAILURE in 1h 14m 39s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/c1070d2/ : FAILURE in 1h 04m 52s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/networking-midonet-coverage-db-ubuntu-xenial/01dea72/ : SUCCESS in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"6b06dc0557d4716151e7481253724bd154b6366b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-29 22:40:04.000000000","message":"Patch Set 52:\n\nrecheck","accounts_in_message":[],"_revision_number":52},{"id":"d672d51d97e9099ee25c19a2fcac529ef6411c75","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-30 00:23:18.000000000","message":"Patch Set 52:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-docs-ubuntu-xenial/525839b//doc/build/html/ : SUCCESS in 3m 05s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-pep8-ubuntu-xenial/e028569/ : SUCCESS in 4m 56s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python27-db-ubuntu-xenial/d11628f/ : SUCCESS in 8m 31s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python35-db/03043a8/ : SUCCESS in 8m 03s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-releasenotes/7095244//releasenotes/build/html/ : SUCCESS in 3m 02s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/c5da3eb/ : FAILURE in 1h 26m 05s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/71c1f82/ : FAILURE in 1h 05m 41s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/260bd13/ : FAILURE in 59m 55s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/5304886/ : FAILURE in 1h 42m 38s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/networking-midonet-coverage-db-ubuntu-xenial/782eb19/ : SUCCESS in 9m 37s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"0207bbfba7b9d0e379e16e05995279cda8877a9e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-01-31 08:35:07.000000000","message":"Patch Set 52:\n\nrecheck","accounts_in_message":[],"_revision_number":52},{"id":"58052d99d86e0873519d3ecb54b566958cc16403","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-31 10:06:41.000000000","message":"Patch Set 52:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-docs-ubuntu-xenial/7d71c23//doc/build/html/ : SUCCESS in 5m 08s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-pep8-ubuntu-xenial/70013f6/ : SUCCESS in 5m 30s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python27-db-ubuntu-xenial/ad1237f/ : SUCCESS in 9m 00s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python35-db/18e25e5/ : SUCCESS in 8m 29s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-releasenotes/30e2301//releasenotes/build/html/ : SUCCESS in 4m 35s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/83f8830/ : FAILURE in 1h 00m 25s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/2ae1b8d/ : FAILURE in 58m 29s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/c4ec5e7/ : FAILURE in 1h 24m 10s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/464ac96/ : FAILURE in 1h 27m 15s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/networking-midonet-coverage-db-ubuntu-xenial/48e5354/ : SUCCESS in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"dec51d73dad30892a2566ceaee86230fe68283d8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-07 02:03:39.000000000","message":"Patch Set 52:\n\nrecheck","accounts_in_message":[],"_revision_number":52},{"id":"ad46b4ff98b4e326e9022880c5c618b783b3d677","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-07 03:12:13.000000000","message":"Patch Set 52:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-docs-ubuntu-xenial/96032c2//doc/build/html/ : SUCCESS in 5m 51s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-pep8-ubuntu-xenial/c94e146/ : SUCCESS in 5m 01s\n- gate-networking-midonet-python27-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python27-db-ubuntu-xenial/04c9322/ : SUCCESS in 8m 11s\n- gate-networking-midonet-python35-db http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python35-db/62a79f1/ : SUCCESS in 8m 29s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-releasenotes/8e1dc48//releasenotes/build/html/ : SUCCESS in 3m 05s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/39ebd24/ : FAILURE in 1h 01m 18s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/ba10685/ : FAILURE in 1h 00m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/0eba6c4/ : FAILURE in 58m 54s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/b79bba9/ : FAILURE in 1h 07m 45s (non-voting)\n- networking-midonet-coverage-db-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/networking-midonet-coverage-db-ubuntu-xenial/a4fcef4/ : SUCCESS in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"81b8484e736398be68d1d711b0c105c1ee4ba60a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-08 22:41:27.000000000","message":"Patch Set 52:\n\nrecheck","accounts_in_message":[],"_revision_number":52},{"id":"726442fe5abd05ae5e8268cc14169d80e782f55b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-09 00:15:14.000000000","message":"Patch Set 52:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-docs-ubuntu-xenial/b83e475//doc/build/html/ : SUCCESS in 3m 10s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-pep8-ubuntu-xenial/70bd217/ : SUCCESS in 5m 18s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python27-ubuntu-xenial/9bb7a06/ : SUCCESS in 13m 06s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python35/4a7e1af/ : SUCCESS in 9m 20s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-releasenotes/e2584e2//releasenotes/build/html/ : SUCCESS in 3m 07s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/ed926e9/ : FAILURE in 1h 33m 08s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/4d94aa4/ : FAILURE in 1h 04m 41s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/9e265c4/ : FAILURE in 1h 11m 22s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/889316b/ : FAILURE in 58m 07s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/networking-midonet-coverage-ubuntu-xenial/9b33291/ : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"b7087d44a0a2c8b831d68f3a66503200800430bb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-10 05:05:52.000000000","message":"Patch Set 52:\n\nrecheck","accounts_in_message":[],"_revision_number":52},{"id":"33a3dee3f17e821f9cbc44a2d29811053c3f0594","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-10 06:14:17.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-docs-ubuntu-xenial/19175dc//doc/build/html/ : SUCCESS in 3m 18s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-pep8-ubuntu-xenial/ecdd2ed/ : SUCCESS in 8m 07s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python27-ubuntu-xenial/3768dc0/ : SUCCESS in 8m 50s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python35/8afccc7/ : SUCCESS in 16m 24s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-releasenotes/c4369b0//releasenotes/build/html/ : SUCCESS in 3m 15s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/3f17ee0/ : SUCCESS in 54m 59s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/b19a3ba/ : SUCCESS in 57m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/6cc78fc/ : SUCCESS in 1h 05m 43s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/b14a384/ : SUCCESS in 1h 06m 55s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/networking-midonet-coverage-ubuntu-xenial/de4c918/ : SUCCESS in 10m 39s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"43ad29dacdfee279a239a18f694f3a68cc81d450","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-12 21:22:29.000000000","message":"Patch Set 52:\n\nrecheck","accounts_in_message":[],"_revision_number":52},{"id":"9783448b551c7317db00eb1c709a0974796d4e6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-12 23:20:31.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-docs-ubuntu-xenial/e405d0b//doc/build/html/ : SUCCESS in 6m 33s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-pep8-ubuntu-xenial/36ab622/ : SUCCESS in 4m 51s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python27-ubuntu-xenial/0af61aa/ : SUCCESS in 8m 26s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/52/check/gate-networking-midonet-python35/4062ac6/ : SUCCESS in 8m 42s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/52/check/gate-networking-midonet-releasenotes/7fe02cc//releasenotes/build/html/ : SUCCESS in 5m 14s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/13c7b1b/ : FAILURE in 1h 01m 28s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/80bc34c/ : FAILURE in 1h 56m 55s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/52845cf/ : FAILURE in 1h 03m 04s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/3e042b4/ : FAILURE in 1h 57m 23s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial http://logs.openstack.org/87/199387/52/check/networking-midonet-coverage-ubuntu-xenial/0397209/ : SUCCESS in 10m 01s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/52/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/3730cd7/ : SUCCESS in 52m 04s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"59fc6dc80a38fa1249bf4d4fe0f3236a46d871f1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-14 09:22:06.000000000","message":"Patch Set 53: Published edit on patch set 52","accounts_in_message":[],"_revision_number":53},{"id":"369669a434e50a1d148f66f484c003d2790790e8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-14 11:00:01.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/53/check/gate-networking-midonet-docs-ubuntu-xenial/80eb6e1//doc/build/html/ : SUCCESS in 3m 23s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/53/check/gate-networking-midonet-pep8-ubuntu-xenial/a1510f4/ : SUCCESS in 6m 07s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/53/check/gate-networking-midonet-python27-ubuntu-xenial/23bfefb/ : SUCCESS in 8m 34s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/53/check/gate-networking-midonet-python35/3af150a/ : SUCCESS in 9m 25s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/53/check/gate-networking-midonet-releasenotes/b055f3b//releasenotes/build/html/ : SUCCESS in 4m 26s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/53/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/7822103/ : SUCCESS in 34m 10s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/53/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/de63cdd/ : SUCCESS in 1h 01m 48s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/53/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/5085b14/ : SUCCESS in 54m 36s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/53/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/80a226f/ : SUCCESS in 1h 31m 22s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/53/check/networking-midonet-coverage-ubuntu-xenial-nv/1376235/ : SUCCESS in 13m 33s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/53/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/759e277/ : SUCCESS in 1h 08m 43s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"88b337f16debd6d5b1c06b5c2c5f35aeabdaf12a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-16 00:44:03.000000000","message":"Patch Set 53: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ocata as commit 0e44cf0ba289567c15bd9538bb484cebf07e908d","accounts_in_message":[],"_revision_number":53},{"id":"9357d9cba25b1b606542aba2904023a27c67ee58","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-16 02:49:07.000000000","message":"Patch Set 54: Published edit on patch set 53","accounts_in_message":[],"_revision_number":54},{"id":"16cce9f3074c34d76ab80b7a98b702562b139103","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-16 03:00:27.000000000","message":"Patch Set 54:\n\ncheck experimental","accounts_in_message":[],"_revision_number":54},{"id":"9b148a7108b0faf4f65a046c3d698750725f6acb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-16 03:19:35.000000000","message":"Patch Set 54:\n\nBuild succeeded (experimental pipeline).\n\n- gate-networking-midonet-v2-rally-dsvm-ubuntu-xenial-nv http://logs.openstack.org/87/199387/54/experimental/gate-networking-midonet-v2-rally-dsvm-ubuntu-xenial-nv/3a1cd32/ : FAILURE in 15m 01s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"e5a06bbc6b175bfab3054e4ae6b2039311c90ede","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-16 04:10:41.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/54/check/gate-networking-midonet-docs-ubuntu-xenial/4bf5c3c//doc/build/html/ : SUCCESS in 5m 50s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/54/check/gate-networking-midonet-pep8-ubuntu-xenial/02d46ce/ : SUCCESS in 5m 38s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/54/check/gate-networking-midonet-python27-ubuntu-xenial/b67e53b/ : SUCCESS in 9m 13s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/54/check/gate-networking-midonet-python35/2fb1650/ : SUCCESS in 11m 41s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/54/check/gate-networking-midonet-releasenotes/8726249//releasenotes/build/html/ : SUCCESS in 3m 19s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/54/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/8f46b55/ : SUCCESS in 29m 16s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/54/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/e978f62/ : SUCCESS in 56m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/54/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/23ac15c/ : SUCCESS in 48m 31s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/54/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/4fb4730/ : SUCCESS in 1h 19m 36s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/54/check/networking-midonet-coverage-ubuntu-xenial-nv/317343a/ : SUCCESS in 10m 56s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/54/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/0afd8f7/ : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"27121b81d4cc39b92a17e02784b90e00459f66d6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-16 05:18:30.000000000","message":"Patch Set 55: Published edit on patch set 54","accounts_in_message":[],"_revision_number":55},{"id":"273a01d056003574a83693710f0d8589613d65c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-16 06:31:17.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/55/check/gate-networking-midonet-docs-ubuntu-xenial/71fcc0a//doc/build/html/ : SUCCESS in 5m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/55/check/gate-networking-midonet-pep8-ubuntu-xenial/cb9d13b/ : SUCCESS in 7m 34s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/55/check/gate-networking-midonet-python27-ubuntu-xenial/e743c81/ : FAILURE in 15m 41s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/55/check/gate-networking-midonet-python35/2517882/ : FAILURE in 13m 26s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/55/check/gate-networking-midonet-releasenotes/d017c09//releasenotes/build/html/ : SUCCESS in 4m 38s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/55/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/831f15e/ : SUCCESS in 31m 02s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/55/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/e45f01b/ : SUCCESS in 1h 07m 28s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/55/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/d1a34b1/ : SUCCESS in 32m 17s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/55/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/435ea8c/ : SUCCESS in 1h 08m 56s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/55/check/networking-midonet-coverage-ubuntu-xenial-nv/792b88c/ : FAILURE in 10m 58s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/55/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/7100bf3/ : SUCCESS in 56m 43s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"ad59f1dbdc94f20f2c00f75f713d707628169206","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-20 19:04:32.000000000","message":"Patch Set 56: Published edit on patch set 55","accounts_in_message":[],"_revision_number":56},{"id":"b953f78b1532c27cc835f6420c7d82807ff285f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-20 20:17:39.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/56/check/gate-networking-midonet-docs-ubuntu-xenial/977409d//doc/build/html/ : SUCCESS in 5m 13s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/56/check/gate-networking-midonet-pep8-ubuntu-xenial/d9393d8/ : SUCCESS in 4m 56s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/56/check/gate-networking-midonet-python27-ubuntu-xenial/a828bd5/ : FAILURE in 12m 18s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/56/check/gate-networking-midonet-python35/8923316/ : FAILURE in 12m 56s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/56/check/gate-networking-midonet-releasenotes/7c4bfe1//releasenotes/build/html/ : SUCCESS in 3m 09s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/56/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/e572588/ : SUCCESS in 29m 24s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/56/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/2289ff8/ : FAILURE in 1h 11m 32s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/56/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/c344a0f/ : SUCCESS in 35m 52s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/56/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/b778e5e/ : SUCCESS in 1h 05m 52s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/56/check/networking-midonet-coverage-ubuntu-xenial-nv/8e308b4/ : FAILURE in 9m 42s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/56/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/3bddcb5/ : SUCCESS in 41m 50s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"9ed18091fbe43da2dcf5dc745b53ce249804a89d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-20 21:31:29.000000000","message":"Patch Set 57: Published edit on patch set 56","accounts_in_message":[],"_revision_number":57},{"id":"25ea3292d1f08af30ff4ee7472c83deda5486222","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-20 22:49:07.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/57/check/gate-networking-midonet-docs-ubuntu-xenial/1567615//doc/build/html/ : SUCCESS in 3m 30s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/57/check/gate-networking-midonet-pep8-ubuntu-xenial/0677162/ : SUCCESS in 7m 59s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/57/check/gate-networking-midonet-python27-ubuntu-xenial/6c8c914/ : SUCCESS in 13m 44s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/57/check/gate-networking-midonet-python35/8666df1/ : SUCCESS in 9m 53s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/57/check/gate-networking-midonet-releasenotes/53d8570//releasenotes/build/html/ : SUCCESS in 3m 04s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/57/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/6aad4f1/ : FAILURE in 41m 23s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/57/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/5dcbf69/ : FAILURE in 1h 16m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/57/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/1d3f76d/ : SUCCESS in 38m 51s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/57/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/8e40619/ : SUCCESS in 1h 01m 59s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/57/check/networking-midonet-coverage-ubuntu-xenial-nv/0ef0559/ : SUCCESS in 10m 05s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/57/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/eff7b09/ : SUCCESS in 42m 33s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"67b6662f92efc6d93e1c5a57961a0075c16b623d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-21 01:37:05.000000000","message":"Patch Set 57:\n\nrecheck","accounts_in_message":[],"_revision_number":57},{"id":"1e5b0a9826580e57613271a37b77d5a491aa1a71","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-21 02:55:48.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/57/check/gate-networking-midonet-docs-ubuntu-xenial/7e3a53c//doc/build/html/ : SUCCESS in 5m 33s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/57/check/gate-networking-midonet-pep8-ubuntu-xenial/2f5f1af/ : SUCCESS in 7m 48s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/57/check/gate-networking-midonet-python27-ubuntu-xenial/31c25be/ : SUCCESS in 8m 44s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/57/check/gate-networking-midonet-python35/e519347/ : SUCCESS in 9m 12s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/57/check/gate-networking-midonet-releasenotes/11521ce//releasenotes/build/html/ : SUCCESS in 2m 59s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/57/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/d0db047/ : SUCCESS in 31m 01s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/57/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/c43ecf3/ : FAILURE in 1h 18m 08s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/57/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/cf94281/ : SUCCESS in 31m 46s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/57/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/6a50388/ : FAILURE in 1h 09m 30s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/57/check/networking-midonet-coverage-ubuntu-xenial-nv/3793a1e/ : SUCCESS in 13m 48s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/57/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/7c7b61b/ : SUCCESS in 1h 07m 21s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"57ee6c7086132b2f0f7ceef76b96d4ef09bae2df","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-22 16:06:53.000000000","message":"Patch Set 58: Published edit on patch set 57","accounts_in_message":[],"_revision_number":58},{"id":"ada65b4c02d3517f72433a43fab42b0c57144d37","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-22 17:30:55.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-docs-ubuntu-xenial/53ab056//doc/build/html/ : SUCCESS in 5m 22s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-pep8-ubuntu-xenial/2dedaab/ : SUCCESS in 5m 01s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python27-ubuntu-xenial/f42c1db/ : SUCCESS in 8m 58s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python35/30eb953/ : SUCCESS in 9m 22s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-releasenotes/53e9cd8//releasenotes/build/html/ : SUCCESS in 4m 59s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/d40892b/ : SUCCESS in 48m 12s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/a83cdaf/ : FAILURE in 1h 00m 32s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/50a35d6/ : SUCCESS in 56m 53s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/2653fc0/ : SUCCESS in 1h 21m 01s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/networking-midonet-coverage-ubuntu-xenial-nv/b2ddc63/ : SUCCESS in 9m 47s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/b6730d4/ : SUCCESS in 1h 01m 31s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"05d8a769ffcd5637ec93db4c661e91213c1d2096","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-24 13:24:24.000000000","message":"Patch Set 58:\n\nrecheck","accounts_in_message":[],"_revision_number":58},{"id":"4014b1a805d90aeb0a0d047cb87ed275831587f2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-24 14:14:20.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-docs-ubuntu-xenial/66e45d4//doc/build/html/ : SUCCESS in 5m 50s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-pep8-ubuntu-xenial/ec2ee92/ : SUCCESS in 5m 20s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python27-ubuntu-xenial/b7fc171/ : SUCCESS in 9m 37s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python35/2a23411/ : SUCCESS in 9m 19s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-releasenotes/7ff3b02//releasenotes/build/html/ : SUCCESS in 3m 02s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/114ab04/ : FAILURE in 43m 41s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/5243000/ : FAILURE in 8m 09s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/bf0ddce/ : FAILURE in 47m 48s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/6c1da2b/ : FAILURE in 7m 58s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/networking-midonet-coverage-ubuntu-xenial-nv/cb8bb78/ : SUCCESS in 13m 26s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/ccafa56/ : FAILURE in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"113ffd155a5cffe56d0c6af3566c20971252eb33","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-02-27 23:24:20.000000000","message":"Patch Set 58:\n\nrecheck","accounts_in_message":[],"_revision_number":58},{"id":"87f6f3468f663512e846f52bf0d12e4a2af3a728","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-28 00:25:32.000000000","message":"Patch Set 58:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-docs-ubuntu-xenial/7d5a84e//doc/build/html/ : SUCCESS in 3m 36s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-pep8-ubuntu-xenial/f1add50/ : SUCCESS in 5m 52s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python27-ubuntu-xenial/0a82d87/ : FAILURE in 3m 18s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python35/5e677d5/ : FAILURE in 3m 13s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-releasenotes/356c092//releasenotes/build/html/ : SUCCESS in 3m 37s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/78a60f0/ : FAILURE in 33m 05s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/7dc2e95/ : FAILURE in 14m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/e94a801/ : FAILURE in 35m 53s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/2eab5de/ : FAILURE in 13m 22s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/networking-midonet-coverage-ubuntu-xenial-nv/7e6ee5b/ : FAILURE in 5m 23s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/f49c7f2/ : SUCCESS in 40m 44s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/9b89b8f/ : FAILURE in 1h 00m 10s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/67f3015/ : FAILURE in 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"b571977af70b87360961fe806ed06d5f4fe03922","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2017-03-01 17:37:15.000000000","message":"Patch Set 58:\n\nrecheck","accounts_in_message":[],"_revision_number":58},{"id":"41b9de40494d34eaafdbbf20b121c98afd1427af","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-01 18:40:02.000000000","message":"Patch Set 58:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-docs-ubuntu-xenial/c775b32//doc/build/html/ : SUCCESS in 3m 44s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-pep8-ubuntu-xenial/8059bd4/ : SUCCESS in 5m 45s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python27-ubuntu-xenial/b29082e/ : SUCCESS in 9m 00s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python35/17dbe75/ : SUCCESS in 9m 15s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-releasenotes/d18c823//releasenotes/build/html/ : SUCCESS in 3m 28s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/327b6f4/ : FAILURE in 27m 38s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/141abb8/ : FAILURE in 58m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/3ca7f25/ : FAILURE in 57m 21s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/412ee60/ : FAILURE in 51m 28s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/networking-midonet-coverage-ubuntu-xenial-nv/3226270/ : SUCCESS in 14m 26s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/778da51/ : SUCCESS in 47m 25s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/a692550/ : FAILURE in 1h 00m 02s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/9eef5e3/ : FAILURE in 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"696370c2875231ba2cb2748fa5f2ff666da2b418","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-03-01 22:18:04.000000000","message":"Patch Set 58:\n\nIhar,\n\nsee https://review.openstack.org/#/c/438092/","accounts_in_message":[],"_revision_number":58},{"id":"e43ff40e7e3f80313c2ba150f444e606af2f4244","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-04 03:49:57.000000000","message":"Patch Set 58:\n\nrecheck","accounts_in_message":[],"_revision_number":58},{"id":"9383bbb9f95535deaf816f363a80a01f63d67504","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-04 05:19:02.000000000","message":"Patch Set 58:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-docs-ubuntu-xenial/95b8036//doc/build/html/ : SUCCESS in 5m 52s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-pep8-ubuntu-xenial/0ece987/ : SUCCESS in 7m 44s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python27-ubuntu-xenial/5b2bab9/ : FAILURE in 7m 31s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-python35/600b9fe/ : FAILURE in 7m 57s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/58/check/gate-networking-midonet-releasenotes/4eb37a7//releasenotes/build/html/ : SUCCESS in 5m 11s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/0e82dba/ : FAILURE in 25m 50s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/c9aa235/ : FAILURE in 13m 13s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/f6d153a/ : FAILURE in 18m 10s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/1640606/ : FAILURE in 13m 10s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/networking-midonet-coverage-ubuntu-xenial-nv/98bca12/ : FAILURE in 8m 09s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/21dc7d0/ : FAILURE in 1h 27m 02s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/f4ce755/ : FAILURE in 13m 10s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/58/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/07b434f/ : FAILURE in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"21ef82d015a14ad89ee6de5c3c64ce8cd11f1016","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-04 05:55:31.000000000","message":"Patch Set 59: Published edit on patch set 58","accounts_in_message":[],"_revision_number":59},{"id":"b3a4720e3af4727f3e426eccc6219154c413c208","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-04 08:00:45.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/59/check/gate-networking-midonet-docs-ubuntu-xenial/6ca6919//doc/build/html/ : SUCCESS in 3m 15s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/59/check/gate-networking-midonet-pep8-ubuntu-xenial/0bfa662/ : SUCCESS in 7m 56s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/59/check/gate-networking-midonet-python27-ubuntu-xenial/25ba71d/ : FAILURE in 12m 19s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/59/check/gate-networking-midonet-python35/e7847d1/ : FAILURE in 11m 39s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/59/check/gate-networking-midonet-releasenotes/731a4e6//releasenotes/build/html/ : SUCCESS in 3m 07s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/59/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/cdf7f51/ : FAILURE in 1h 56m 46s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/59/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/378fe9d/ : FAILURE in 1h 56m 55s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/59/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/440902b/ : FAILURE in 12m 23s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/59/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/d19f7ab/ : FAILURE in 13m 19s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/59/check/networking-midonet-coverage-ubuntu-xenial-nv/da15d23/ : FAILURE in 15m 42s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/59/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/eb4444e/ : FAILURE in 2h 03m 28s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/59/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/bd84237/ : FAILURE in 19m 18s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/59/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/5e361af/ : FAILURE in 39m 53s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"27d637e4c19b0508727f1d76e8c6a2f5f9de9664","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-04 09:36:13.000000000","message":"Patch Set 60: Published edit on patch set 59","accounts_in_message":[],"_revision_number":60},{"id":"dcc864ab22aee3ce9c45db50f2adceabc98e59ca","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-04 11:00:45.000000000","message":"Patch Set 60: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/60/check/gate-networking-midonet-docs-ubuntu-xenial/a118a66//doc/build/html/ : SUCCESS in 5m 35s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/60/check/gate-networking-midonet-pep8-ubuntu-xenial/362125e/ : SUCCESS in 7m 38s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/60/check/gate-networking-midonet-python27-ubuntu-xenial/2222faf/ : FAILURE in 11m 31s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/60/check/gate-networking-midonet-python35/59c292a/ : FAILURE in 12m 48s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/60/check/gate-networking-midonet-releasenotes/b0ca0c8//releasenotes/build/html/ : SUCCESS in 3m 13s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/60/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/0f28969/ : FAILURE in 36m 42s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/60/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/13a0ea4/ : FAILURE in 1h 11m 19s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/60/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/dfbeeab/ : SUCCESS in 39m 21s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/60/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/a387347/ : FAILURE in 1h 21m 59s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/60/check/networking-midonet-coverage-ubuntu-xenial-nv/10258ff/ : FAILURE in 14m 09s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/60/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/379f2b7/ : SUCCESS in 1h 00m 14s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/60/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/744d89a/ : SUCCESS in 1h 00m 23s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/60/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/d9d08d1/ : FAILURE in 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"0746e32164f9f2276f3bb6016c622755531a5f6c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-05 06:33:01.000000000","message":"Patch Set 61: Published edit on patch set 60","accounts_in_message":[],"_revision_number":61},{"id":"e493a96d2c2d552ad695fa6efaea718a8c5073b1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 07:54:11.000000000","message":"Patch Set 61: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/61/check/gate-networking-midonet-docs-ubuntu-xenial/eb53fb9//doc/build/html/ : SUCCESS in 3m 29s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/61/check/gate-networking-midonet-pep8-ubuntu-xenial/c314e97/ : SUCCESS in 8m 59s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/61/check/gate-networking-midonet-python27-ubuntu-xenial/a83329f/ : SUCCESS in 11m 58s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/61/check/gate-networking-midonet-python35/3ca0e93/ : SUCCESS in 12m 11s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/61/check/gate-networking-midonet-releasenotes/4a22647//releasenotes/build/html/ : SUCCESS in 3m 09s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/61/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/8127cd8/ : FAILURE in 34m 43s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/61/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/8f8a4e6/ : FAILURE in 1h 07m 17s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/61/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/ebada7c/ : SUCCESS in 39m 42s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/61/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/775b4d2/ : SUCCESS in 1h 19m 31s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/61/check/networking-midonet-coverage-ubuntu-xenial-nv/ff927eb/ : SUCCESS in 12m 44s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/61/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/9471bfa/ : SUCCESS in 55m 17s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/61/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/0ac6a3a/ : SUCCESS in 37m 30s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/61/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/61d58d3/ : FAILURE in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"7cfc0e5a1f0f57031fbf78cea0ba6758d75c522b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-06 05:36:55.000000000","message":"Patch Set 61: Cherry Picked\n\nThis patchset was cherry picked to branch stable/newton as commit 973f04964fa32a8fa1dd7369503c8a8a8aef226d","accounts_in_message":[],"_revision_number":61},{"id":"4defa2ae755340103cc6bd4caafc7c717d6a3da9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-06 08:49:31.000000000","message":"Patch Set 62: Published edit on patch set 61","accounts_in_message":[],"_revision_number":62},{"id":"cf09412544e5ec59ba3bee03be738c13448e88e3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-06 09:36:23.000000000","message":"Patch Set 62: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":62},{"id":"c4b404433d0718627a9d8a202ef5b06fba395dc8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-06 09:58:35.000000000","message":"Patch Set 62:\n\nrecheck","accounts_in_message":[],"_revision_number":62},{"id":"6fbafc338bd13cccd8512f5bf0c9a80b71641de9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-06 11:18:49.000000000","message":"Patch Set 62:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/62/check/gate-networking-midonet-docs-ubuntu-xenial/263744f//doc/build/html/ : SUCCESS in 3m 34s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/62/check/gate-networking-midonet-pep8-ubuntu-xenial/6720690/ : SUCCESS in 8m 41s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/62/check/gate-networking-midonet-python27-ubuntu-xenial/10eb64f/ : FAILURE in 13m 17s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/62/check/gate-networking-midonet-python35/36f6f78/ : FAILURE in 12m 15s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/62/check/gate-networking-midonet-releasenotes/519682a//releasenotes/build/html/ : SUCCESS in 5m 00s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/62/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/929d0d1/ : SUCCESS in 28m 55s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/62/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/327a53a/ : FAILURE in 1h 18m 53s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/62/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/4ccc9ce/ : SUCCESS in 38m 29s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/62/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/d54c2f1/ : SUCCESS in 1h 11m 17s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/62/check/networking-midonet-coverage-ubuntu-xenial-nv/da80162/ : FAILURE in 12m 04s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/62/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/3637aac/ : SUCCESS in 51m 39s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/62/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/bfb1834/ : SUCCESS in 38m 05s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/62/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/da6ecb4/ : FAILURE in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"ddda029683648dd6c302e5c18c5eb8c6553aa1b2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-07 12:59:23.000000000","message":"Patch Set 63: Published edit on patch set 62","accounts_in_message":[],"_revision_number":63},{"id":"73c77fec183d94944e921ed4496b424c57d74cc4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-07 14:07:02.000000000","message":"Patch Set 63: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/63/check/gate-networking-midonet-docs-ubuntu-xenial/9f3cce8//doc/build/html/ : SUCCESS in 7m 32s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-pep8-ubuntu-xenial/f95e9ee/ : SUCCESS in 5m 30s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-python27-ubuntu-xenial/c17eb57/ : FAILURE in 12m 10s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-python35/0c59ca8/ : FAILURE in 11m 58s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/63/check/gate-networking-midonet-releasenotes/7fa2725//releasenotes/build/html/ : SUCCESS in 4m 52s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/63/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/edc8025/ : SUCCESS in 37m 25s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/acbe8c6/ : FAILURE in 1h 06m 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/63/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/054c56b/ : SUCCESS in 37m 15s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/3f5ca3c/ : SUCCESS in 1h 06m 17s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/networking-midonet-coverage-ubuntu-xenial-nv/8366f7c/ : FAILURE in 14m 13s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/7f9c833/ : SUCCESS in 50m 12s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/1f1688e/ : SUCCESS in 52m 08s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/c5c130d/ : FAILURE in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"11528d46e296cfc4acedd52b26a3b18fb34771b4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-10 02:50:49.000000000","message":"Patch Set 63:\n\nrecheck","accounts_in_message":[],"_revision_number":63},{"id":"99e3176bbc2dd080b73b702ffa76e1aca57383ce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-10 03:52:56.000000000","message":"Patch Set 63:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/63/check/gate-networking-midonet-docs-ubuntu-xenial/a878141//doc/build/html/ : SUCCESS in 5m 46s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-pep8-ubuntu-xenial/9177bc8/ : SUCCESS in 8m 22s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-python27-ubuntu-xenial/571f271/ : FAILURE in 9m 43s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-python35/5beb459/ : FAILURE in 11m 14s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/63/check/gate-networking-midonet-releasenotes/fe23ef7//releasenotes/build/html/ : SUCCESS in 3m 34s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/63/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/0cc62b6/ : SUCCESS in 30m 41s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/8401211/ : FAILURE in 59m 41s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/63/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/0c544e9/ : FAILURE in 31m 54s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/90ce559/ : SUCCESS in 1h 01m 49s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/networking-midonet-coverage-ubuntu-xenial-nv/fd3ca87/ : FAILURE in 12m 02s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/aea9e60/ : SUCCESS in 51m 18s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/fa3812e/ : SUCCESS in 47m 15s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/63/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/59dff57/ : FAILURE in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"90f220945e329ea542b79f908a431d4b7815c7b4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-10 08:42:52.000000000","message":"Patch Set 64: Published edit on patch set 63","accounts_in_message":[],"_revision_number":64},{"id":"76665cf2b03c7ce55636653e3aeed0832976b5d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-10 10:15:12.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/64/check/gate-networking-midonet-docs-ubuntu-xenial/9431b9a//doc/build/html/ : SUCCESS in 4m 08s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/64/check/gate-networking-midonet-pep8-ubuntu-xenial/158c168/ : SUCCESS in 4m 52s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/64/check/gate-networking-midonet-python27-ubuntu-xenial/1dfa25e/ : SUCCESS in 11m 54s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/64/check/gate-networking-midonet-python35/4f4b114/ : SUCCESS in 11m 54s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/64/check/gate-networking-midonet-releasenotes/78e0dc2//releasenotes/build/html/ : SUCCESS in 4m 57s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/64/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/a3eafc4/ : SUCCESS in 27m 56s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/64/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/a417351/ : FAILURE in 1h 00m 53s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/64/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/b956a88/ : SUCCESS in 37m 36s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/64/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/f1e7435/ : SUCCESS in 1h 31m 05s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/64/check/networking-midonet-coverage-ubuntu-xenial-nv/bbff4a2/ : SUCCESS in 12m 28s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/64/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/e66e4a7/ : SUCCESS in 52m 54s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/64/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/31ce38e/ : SUCCESS in 56m 19s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/64/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/af4e062/ : FAILURE in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"bd24ed9f19ae01a244d288bcb97b610eaadf4f82","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-12 03:36:50.000000000","message":"Patch Set 64:\n\nrecheck","accounts_in_message":[],"_revision_number":64},{"id":"40010442c73900d3d87d9f32f46898e1b68a4e71","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-12 03:48:19.000000000","message":"Patch Set 65: Published edit on patch set 64","accounts_in_message":[],"_revision_number":65},{"id":"39471f691b6b5c5bdbb7fef7fd1d9ab6924990a3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-12 04:24:46.000000000","message":"Patch Set 65:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":65},{"id":"fb52f2f1588e2923c108381eb80eb89e6c518293","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-18 10:21:43.000000000","message":"Patch Set 66: Published edit on patch set 65","accounts_in_message":[],"_revision_number":66},{"id":"339da5699ded2b43d838e7e4fdfa16de8ab3fc6f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-18 11:33:42.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/66/check/gate-networking-midonet-docs-ubuntu-xenial/1855275//doc/build/html/ : SUCCESS in 3m 27s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/66/check/gate-networking-midonet-pep8-ubuntu-xenial/eef56d9/ : SUCCESS in 7m 58s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/66/check/gate-networking-midonet-python27-ubuntu-xenial/6db4cd9/ : FAILURE in 11m 26s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/66/check/gate-networking-midonet-python35/1593477/ : FAILURE in 12m 10s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/66/check/gate-networking-midonet-releasenotes/648046c//releasenotes/build/html/ : SUCCESS in 4m 54s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/66/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/29208c6/ : SUCCESS in 33m 12s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/66/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/118f1fe/ : FAILURE in 1h 02m 49s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/66/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/c23f986/ : SUCCESS in 31m 53s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/66/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/5a1b80f/ : FAILURE in 1h 10m 37s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/66/check/networking-midonet-coverage-ubuntu-xenial-nv/93fd42e/ : FAILURE in 12m 42s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/66/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/35fca3e/ : SUCCESS in 42m 45s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/66/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/1881ab1/ : SUCCESS in 48m 29s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/66/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/954f930/ : FAILURE in 30m 27s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"c49113d2855d8e44be37705b37e83df3f400687a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-20 03:45:54.000000000","message":"Patch Set 67: Published edit on patch set 66","accounts_in_message":[],"_revision_number":67},{"id":"189fda58412c313a5619b030738349f21b97ffc0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-20 04:35:09.000000000","message":"Patch Set 68: Published edit on patch set 67","accounts_in_message":[],"_revision_number":68},{"id":"ac2fd52d67fec5cd202c803a0b435783fc0cf217","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-20 04:59:57.000000000","message":"Patch Set 68: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":68},{"id":"bd3864ff714cfd6ebb0e13d745fba565c3262d66","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-25 04:31:50.000000000","message":"Patch Set 69: Published edit on patch set 68","accounts_in_message":[],"_revision_number":69},{"id":"80c05740ad1bb68473b80dd47172a813702a065c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-25 06:09:31.000000000","message":"Patch Set 69: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/69/check/gate-networking-midonet-docs-ubuntu-xenial/b792508//doc/build/html/ : SUCCESS in 5m 18s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/69/check/gate-networking-midonet-pep8-ubuntu-xenial/465460f/ : SUCCESS in 7m 57s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/69/check/gate-networking-midonet-python27-ubuntu-xenial/c0950f5/ : FAILURE in 10m 47s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/69/check/gate-networking-midonet-python35/15f2d39/ : FAILURE in 10m 16s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/69/check/gate-networking-midonet-releasenotes/f9ccde9//releasenotes/build/html/ : SUCCESS in 9m 09s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/69/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/a4fcfdf/ : SUCCESS in 34m 23s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/69/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/9d21913/ : FAILURE in 1h 24m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/69/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/bccf86b/ : SUCCESS in 31m 01s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/69/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/2b3c19c/ : FAILURE in 1h 06m 38s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/69/check/networking-midonet-coverage-ubuntu-xenial-nv/b95c522/ : FAILURE in 12m 40s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/69/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/96bd57f/ : SUCCESS in 1h 00m 19s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/69/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/0cba428/ : SUCCESS in 38m 21s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/69/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/2b0f6d6/ : FAILURE in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"31f9ea1bb7fc45b3565d5fd553de589c0c82efd5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-26 02:30:57.000000000","message":"Patch Set 70: Published edit on patch set 69","accounts_in_message":[],"_revision_number":70},{"id":"838c27a12803e2f698a42590dbb3683bb797ecec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-26 02:45:12.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/70/check/gate-networking-midonet-docs-ubuntu-xenial/17d4d31//doc/build/html/ : SUCCESS in 5m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/70/check/gate-networking-midonet-pep8-ubuntu-xenial/6c92679/ : SUCCESS in 8m 13s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/70/check/gate-networking-midonet-python27-ubuntu-xenial/2f24c2e/ : SUCCESS in 10m 40s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/70/check/gate-networking-midonet-python35/e756320/ : SUCCESS in 10m 37s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/70/check/gate-networking-midonet-releasenotes/a949ba1//releasenotes/build/html/ : SUCCESS in 8m 23s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/70/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/abd4e3b/ : FAILURE in 3m 58s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/70/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/1c43464/ : FAILURE in 3m 25s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/70/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/aa93461/ : FAILURE in 2m 42s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/70/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/c1dddec/ : FAILURE in 2m 45s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/70/check/networking-midonet-coverage-ubuntu-xenial-nv/f35c66f/ : SUCCESS in 13m 11s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/70/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/4a2a103/ : FAILURE in 2m 57s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/70/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/137e5f5/ : FAILURE in 3m 45s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/70/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/96e8c5a/ : FAILURE in 1m 44s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"8997957ad6d741d9bc1123247fa959375711ae4d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-04-27 05:55:01.000000000","message":"Patch Set 71: Published edit on patch set 70","accounts_in_message":[],"_revision_number":71},{"id":"5203c8d9194790bf090ffc45efdc3c8c5a2a47fb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-27 07:09:50.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/71/check/gate-networking-midonet-docs-ubuntu-xenial/553264a//doc/build/html/ : SUCCESS in 3m 52s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/71/check/gate-networking-midonet-pep8-ubuntu-xenial/b123479/ : SUCCESS in 5m 24s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/71/check/gate-networking-midonet-python27-ubuntu-xenial/fe4aca9/ : SUCCESS in 18m 40s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/71/check/gate-networking-midonet-python35/e722fc5/ : SUCCESS in 11m 59s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/71/check/gate-networking-midonet-releasenotes/4724b54//releasenotes/build/html/ : SUCCESS in 8m 30s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/71/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/0827bd5/ : SUCCESS in 26m 39s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/71/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/374b965/ : SUCCESS in 1h 13m 24s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/71/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/13bd1d4/ : SUCCESS in 31m 36s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/71/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/985674e/ : SUCCESS in 1h 07m 49s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/71/check/networking-midonet-coverage-ubuntu-xenial-nv/8ac5225/ : SUCCESS in 13m 03s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/71/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/e10ba43/ : SUCCESS in 45m 52s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/71/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/4659754/ : SUCCESS in 52m 19s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/71/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/d269201/ : FAILURE in 31m 30s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"313db1a7a93899803c52bd10e9a6758202bf8386","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-01 04:15:47.000000000","message":"Patch Set 72: Published edit on patch set 71","accounts_in_message":[],"_revision_number":72},{"id":"82012fd4371a319570d91d0c44711d019172de1e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-01 05:17:39.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/72/check/gate-networking-midonet-docs-ubuntu-xenial/0696cad//doc/build/html/ : SUCCESS in 6m 39s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-pep8-ubuntu-xenial/1b3e8d3/ : SUCCESS in 5m 12s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-python27-ubuntu-xenial/88bf514/ : SUCCESS in 11m 12s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-python35/11ca6b1/ : SUCCESS in 15m 36s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/72/check/gate-networking-midonet-releasenotes/42bab90//releasenotes/build/html/ : SUCCESS in 5m 01s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/72/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/50b8458/ : SUCCESS in 25m 52s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/899a7a1/ : SUCCESS in 1h 00m 02s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/72/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/53b61e8/ : SUCCESS in 41m 14s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/3705301/ : SUCCESS in 1h 01m 38s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/networking-midonet-coverage-ubuntu-xenial-nv/80c317f/ : SUCCESS in 12m 33s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/7f6513a/ : FAILURE in 30m 02s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/a89563b/ : SUCCESS in 50m 29s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/ab7b70f/ : FAILURE in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"2c97a45a582ffec48c6d3e86aa939421b491f564","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-01 05:19:35.000000000","message":"Patch Set 72:\n\nrecheck","accounts_in_message":[],"_revision_number":72},{"id":"cb8a41fb71ae9fef681efb86d4ab59682aa91ae0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-01 06:36:55.000000000","message":"Patch Set 72:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/72/check/gate-networking-midonet-docs-ubuntu-xenial/198040d//doc/build/html/ : SUCCESS in 6m 00s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-pep8-ubuntu-xenial/8405fe2/ : SUCCESS in 8m 32s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-python27-ubuntu-xenial/e081c73/ : SUCCESS in 10m 18s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-python35/3e7b19b/ : SUCCESS in 14m 49s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/72/check/gate-networking-midonet-releasenotes/cd0a8aa//releasenotes/build/html/ : SUCCESS in 5m 42s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/72/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/7da4183/ : SUCCESS in 28m 12s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/d871cf8/ : FAILURE in 1h 05m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/72/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/ae25b5f/ : SUCCESS in 41m 38s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/bae7b0d/ : SUCCESS in 1h 17m 07s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/networking-midonet-coverage-ubuntu-xenial-nv/6b8a3b6/ : SUCCESS in 12m 50s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/7be4a3e/ : SUCCESS in 38m 50s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/cba0daa/ : SUCCESS in 42m 53s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/72/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/055f68e/ : FAILURE in 47m 41s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"9c5aac8f5cd98a02011b9dcb167d2be19cd9cb44","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-08 07:53:16.000000000","message":"Patch Set 73: Published edit on patch set 72","accounts_in_message":[],"_revision_number":73},{"id":"fbcd8fed87ab4f060d3b90c562883cab44a85e4e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-08 09:11:28.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/5203df4//doc/build/html/ : SUCCESS in 3m 25s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/789a030/ : SUCCESS in 7m 29s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/b1a661d/ : SUCCESS in 11m 18s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/165b48d/ : SUCCESS in 11m 52s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/f627c6c//releasenotes/build/html/ : SUCCESS in 14m 30s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/a613d90/ : SUCCESS in 36m 04s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/2340424/ : SUCCESS in 1h 15m 37s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/df30df3/ : SUCCESS in 41m 22s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/856522f/ : SUCCESS in 1h 10m 24s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/4537908/ : SUCCESS in 11m 58s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/54e86d8/ : SUCCESS in 58m 12s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/8609c87/ : SUCCESS in 38m 08s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/258bc6c/ : FAILURE in 41m 27s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"e40edbb03b8307b9a6cdd266152af4fe8c10d211","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-09 04:29:38.000000000","message":"Patch Set 73:\n\nrecheck","accounts_in_message":[],"_revision_number":73},{"id":"da5c97d9ef6565ea01316e7a53bd2c84cf51f2d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-09 05:49:52.000000000","message":"Patch Set 73: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/dfe8627//doc/build/html/ : SUCCESS in 3m 28s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/40fd7d5/ : SUCCESS in 5m 06s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/c39a8b0/ : SUCCESS in 11m 57s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/bdd6898/ : SUCCESS in 10m 52s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/180dc74//releasenotes/build/html/ : SUCCESS in 5m 28s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/7a38056/ : FAILURE in 36m 49s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/e9d477c/ : SUCCESS in 1h 01m 06s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/539f8f1/ : SUCCESS in 41m 24s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/69b60e0/ : FAILURE in 1h 19m 44s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/8fb6237/ : SUCCESS in 36m 39s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/9dcee38/ : SUCCESS in 40m 35s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/f58152f/ : SUCCESS in 40m 58s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/20aa938/ : FAILURE in 49m 26s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"3e343bef30fec0a9fdd8e1300453eb360f0a06f6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-09 06:05:55.000000000","message":"Patch Set 73:\n\nrecheck test_east_west failure","accounts_in_message":[],"_revision_number":73},{"id":"e0ffeb6103da302c766c3342e23ecad9ff4a692f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-09 07:46:35.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/be91512//doc/build/html/ : SUCCESS in 3m 36s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/4fa40a1/ : SUCCESS in 5m 17s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/e68193e/ : SUCCESS in 11m 21s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/a63fc45/ : SUCCESS in 10m 49s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/71e4997//releasenotes/build/html/ : SUCCESS in 5m 16s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/01849f8/ : SUCCESS in 31m 48s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/600130b/ : SUCCESS in 1h 17m 45s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/1f0cb4c/ : SUCCESS in 42m 36s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/bf382d3/ : SUCCESS in 1h 22m 08s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/a797210/ : SUCCESS in 26m 48s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/ef33a88/ : SUCCESS in 49m 38s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/7e83739/ : SUCCESS in 1h 01m 16s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/5ecc4e6/ : FAILURE in 50m 39s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"7e034b1d396b3904d32c2a98ec8cdc6ed7ef088c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-17 07:22:35.000000000","message":"Patch Set 73:\n\nrecheck","accounts_in_message":[],"_revision_number":73},{"id":"61271098cbbc5f76d4cf0750221c409c6ee6cabb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-17 07:39:27.000000000","message":"Patch Set 73: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/a34aabb//doc/build/html/ : SUCCESS in 6m 32s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/73d7a81/ : SUCCESS in 7m 51s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/c79cccc/ : SUCCESS in 11m 12s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/3c47d4f/ : SUCCESS in 10m 42s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/93e947a//releasenotes/build/html/ : SUCCESS in 11m 11s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/02fad36/ : FAILURE in 10m 26s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/0be8f07/ : FAILURE in 10m 47s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/f69e634/ : FAILURE in 8m 14s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/821bab1/ : FAILURE in 11m 05s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/16c15a1/ : SUCCESS in 13m 58s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/cda52ee/ : FAILURE in 12m 03s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/0aba7c3/ : FAILURE in 8m 23s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/2c31fd2/ : FAILURE in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"0b2cf56e91525532c4ba1cd9b7c05d8084b7c0a7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-17 15:52:54.000000000","message":"Patch Set 73:\n\nrecheck","accounts_in_message":[],"_revision_number":73},{"id":"50218abec810d080340ce1f3e74e5ddcbfd97b40","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-17 16:15:13.000000000","message":"Patch Set 73:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/8ea2376//doc/build/html/ : SUCCESS in 6m 06s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/7ad0e03/ : SUCCESS in 9m 58s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/c005f71/ : SUCCESS in 12m 30s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/963fd63/ : SUCCESS in 12m 11s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/c1b520d//releasenotes/build/html/ : SUCCESS in 6m 34s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/0111d66/ : FAILURE in 13m 53s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/c44a0ab/ : FAILURE in 12m 49s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/326c38a/ : FAILURE in 12m 48s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/27fc066/ : FAILURE in 8m 29s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/5c3c269/ : SUCCESS in 13m 27s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/f915641/ : FAILURE in 17m 23s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/c91bd4d/ : FAILURE in 8m 58s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/d0c51cf/ : FAILURE in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"e4fba7343a1e81d6a3bcc204f3d76660880120b5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-17 17:57:49.000000000","message":"Patch Set 73:\n\nrecheck","accounts_in_message":[],"_revision_number":73},{"id":"36c3075b316dc6b7845ddde05534faa25a721d60","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-17 18:18:38.000000000","message":"Patch Set 73:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/be8672e//doc/build/html/ : SUCCESS in 3m 54s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/f3fd2b1/ : SUCCESS in 5m 43s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/455e156/ : SUCCESS in 12m 22s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/8dbc914/ : SUCCESS in 11m 08s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/71c2139//releasenotes/build/html/ : SUCCESS in 5m 44s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/be2e513/ : FAILURE in 10m 17s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/eec6aa7/ : FAILURE in 11m 14s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/cecb540/ : FAILURE in 11m 15s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/7e9fc19/ : FAILURE in 15m 40s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/77591ea/ : SUCCESS in 15m 18s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/afb19a2/ : FAILURE in 8m 53s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/d74177b/ : FAILURE in 16m 58s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/f65735c/ : FAILURE in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"c02a34db99ee91fe47d38c1dc01ed52310fb63a7","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2017-05-19 15:30:28.000000000","message":"Patch Set 73:\n\nrecheck repo bug","accounts_in_message":[],"_revision_number":73},{"id":"534ceadb23dcf2ad85c9f5584f686218a1e32885","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-19 16:53:59.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/b249741//doc/build/html/ : SUCCESS in 5m 55s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/5467955/ : SUCCESS in 10m 10s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/36e6e41/ : SUCCESS in 11m 54s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/41094a0/ : SUCCESS in 11m 29s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/5310f9f//releasenotes/build/html/ : SUCCESS in 5m 24s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/4642531/ : SUCCESS in 32m 15s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/267e67b/ : SUCCESS in 1h 01m 57s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/748ba27/ : SUCCESS in 38m 42s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/872a725/ : SUCCESS in 1h 21m 50s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/b486fd4/ : SUCCESS in 15m 07s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/973e7c6/ : FAILURE in 52m 52s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/5029043/ : SUCCESS in 51m 02s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/9a249b4/ : FAILURE in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"ba8292a97393a6eb00ecc7380d346bb69a9d3135","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-26 03:08:21.000000000","message":"Patch Set 73:\n\nrecheck","accounts_in_message":[],"_revision_number":73},{"id":"26892a1259943ef5d2b53607b1a18c4043ea9450","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-26 04:24:56.000000000","message":"Patch Set 73:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/90da49e//doc/build/html/ : SUCCESS in 4m 01s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/7bb97b9/ : SUCCESS in 5m 15s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/cd4fb5c/ : SUCCESS in 12m 35s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/e54cb3e/ : SUCCESS in 10m 28s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/d54b376//releasenotes/build/html/ : SUCCESS in 8m 19s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/b7f36e1/ : SUCCESS in 27m 15s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/383a49e/ : FAILURE in 1h 10m 43s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/0c908a6/ : SUCCESS in 29m 49s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/b59e8f3/ : SUCCESS in 1h 14m 24s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/60c3a13/ : SUCCESS in 13m 02s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/0a20bf8/ : FAILURE in 42m 37s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/e751109/ : SUCCESS in 39m 30s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/175c0ef/ : SUCCESS in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"132ddc3ffc0f7cebc80260745b36bd0ac7cab179","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-30 02:32:03.000000000","message":"Patch Set 73:\n\nrecheck","accounts_in_message":[],"_revision_number":73},{"id":"638adf53d777f8aca97832f2da9b46b401c84b00","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-30 03:51:06.000000000","message":"Patch Set 73:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-docs-ubuntu-xenial/5f364d8//doc/build/html/ : SUCCESS in 6m 02s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-pep8-ubuntu-xenial/65d3e20/ : SUCCESS in 4m 51s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python27-ubuntu-xenial/fe32455/ : SUCCESS in 11m 46s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-python35/a33ec27/ : SUCCESS in 12m 26s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/73/check/gate-networking-midonet-releasenotes/b9d89fe//releasenotes/build/html/ : SUCCESS in 15m 25s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/8452eea/ : SUCCESS in 27m 30s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/d788aa0/ : SUCCESS in 1h 13m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/20afa25/ : SUCCESS in 30m 34s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/7895b41/ : SUCCESS in 1h 18m 03s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/networking-midonet-coverage-ubuntu-xenial-nv/aa03e21/ : SUCCESS in 12m 57s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/02beeae/ : FAILURE in 39m 16s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/7ee7c83/ : SUCCESS in 40m 58s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/73/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/5a4fda8/ : SUCCESS in 34m 35s (non-voting)","accounts_in_message":[],"_revision_number":73},{"id":"0ca8b2a6534a18083226d8cfed6d3da218c260c2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-30 09:19:10.000000000","message":"Patch Set 74: Published edit on patch set 73","accounts_in_message":[],"_revision_number":74},{"id":"0d925d41242cc0956b5178d952d21049b0b8cb07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-30 10:48:39.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/74/check/gate-networking-midonet-docs-ubuntu-xenial/bc101c8//doc/build/html/ : SUCCESS in 6m 17s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-pep8-ubuntu-xenial/2f5ef31/ : SUCCESS in 7m 55s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python27-ubuntu-xenial/a5c86a7/ : SUCCESS in 11m 35s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python35/9a759ec/ : SUCCESS in 10m 56s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/74/check/gate-networking-midonet-releasenotes/823f75f//releasenotes/build/html/ : SUCCESS in 9m 41s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/d56d79f/ : FAILURE in 43m 57s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/3ed709d/ : FAILURE in 1h 24m 23s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/4a252e8/ : FAILURE in 46m 32s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/d5387c7/ : FAILURE in 1h 25m 53s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/networking-midonet-coverage-ubuntu-xenial-nv/9f3f807/ : SUCCESS in 13m 50s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/4ff93bb/ : FAILURE in 1h 13m 03s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/1964228/ : SUCCESS in 40m 10s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/05f9df0/ : SUCCESS in 34m 08s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"76709e41a22e4d5a19d9888b338f065bc05fdcf0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-30 11:55:40.000000000","message":"Patch Set 74:\n\nrecheck","accounts_in_message":[],"_revision_number":74},{"id":"00380490761e0e897043937bd42726f5b2804922","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-30 13:22:24.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/74/check/gate-networking-midonet-docs-ubuntu-xenial/de8ce1d//doc/build/html/ : SUCCESS in 5m 47s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-pep8-ubuntu-xenial/d86d124/ : SUCCESS in 5m 37s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python27-ubuntu-xenial/124bf78/ : SUCCESS in 12m 03s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python35/2fb7f0a/ : SUCCESS in 10m 14s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/74/check/gate-networking-midonet-releasenotes/0fbf7a1//releasenotes/build/html/ : SUCCESS in 11m 16s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/0bbda3f/ : SUCCESS in 40m 34s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/525f6b8/ : SUCCESS in 59m 29s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/99a1e74/ : SUCCESS in 41m 00s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/147dbb0/ : SUCCESS in 1h 23m 20s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/networking-midonet-coverage-ubuntu-xenial-nv/f41b191/ : SUCCESS in 13m 49s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/1d86853/ : FAILURE in 52m 16s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/6b4c9c2/ : SUCCESS in 38m 48s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/3b85c27/ : SUCCESS in 45m 14s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"288cab582ebedb12f4e5e4d7a18d12e2fee0e70f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-05-30 14:48:31.000000000","message":"Patch Set 74:\n\nrecheck","accounts_in_message":[],"_revision_number":74},{"id":"550792fb1cf8d8526b46a808eaaf17e24446805e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-30 16:21:02.000000000","message":"Patch Set 74:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/74/check/gate-networking-midonet-docs-ubuntu-xenial/53c3439//doc/build/html/ : SUCCESS in 3m 58s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-pep8-ubuntu-xenial/ae1d866/ : SUCCESS in 8m 35s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python27-ubuntu-xenial/83e180e/ : SUCCESS in 11m 04s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python35/92f7200/ : SUCCESS in 10m 40s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/74/check/gate-networking-midonet-releasenotes/7a57942//releasenotes/build/html/ : SUCCESS in 13m 26s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/2397060/ : SUCCESS in 33m 49s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/96cabae/ : FAILURE in 1h 23m 27s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/d4506dc/ : SUCCESS in 37m 16s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/583576e/ : SUCCESS in 1h 30m 18s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/networking-midonet-coverage-ubuntu-xenial-nv/f12745c/ : SUCCESS in 15m 11s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/85843d7/ : FAILURE in 42m 23s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/9f5d4f0/ : SUCCESS in 55m 04s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/e64d7bc/ : SUCCESS in 43m 30s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"09d7cd33025caafef9a1bd34aece8fbd1a0376c7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-01 08:17:42.000000000","message":"Patch Set 74:\n\nrecheck","accounts_in_message":[],"_revision_number":74},{"id":"c3e6c29c8a064a7f7f1a3ebd80727c961513d1ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-01 08:26:39.000000000","message":"Patch Set 74: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-docs-ubuntu-xenial/e1707ec/ : FAILURE in 2m 17s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-pep8-ubuntu-xenial/93acc74/ : FAILURE in 7m 29s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python27-ubuntu-xenial/1f36ba2/ : FAILURE in 2m 50s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python35/ae17f71/ : FAILURE in 1m 36s\n- gate-networking-midonet-releasenotes http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-releasenotes/aa564bd/ : FAILURE in 1m 37s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/f811a5b/ : FAILURE in 16s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/939d464/ : FAILURE in 15s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/545980c/ : FAILURE in 44s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/dfd7090/ : FAILURE in 50s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/networking-midonet-coverage-ubuntu-xenial-nv/93e13e6/ : FAILURE in 1m 39s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/3749f8b/ : FAILURE in 19s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/1c9ee73/ : FAILURE in 24s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/69100fd/ : FAILURE in 19s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"2de9eb75f08da2f4887923e853f937b395267c26","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-06-01 08:27:33.000000000","message":"Patch Set 74:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-networking-midonet-docs-ubuntu-xenial: unrecognized error\n- gate-networking-midonet-pep8-ubuntu-xenial: unrecognized error\n- gate-networking-midonet-python27-ubuntu-xenial: unrecognized error\n- gate-networking-midonet-python35: unrecognized error\n- gate-networking-midonet-releasenotes: unrecognized error\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial: https://bugs.launchpad.net/bugs/1497006\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial: https://bugs.launchpad.net/bugs/1497006\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":74},{"id":"30a1f2e784f4b17f79e1a3c0f7a7fb021eb5f832","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-01 13:29:46.000000000","message":"Patch Set 74:\n\nrecheck","accounts_in_message":[],"_revision_number":74},{"id":"6511f535b82e199b3bc60f5faaae63c2682c1d78","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-01 13:46:31.000000000","message":"Patch Set 74:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-docs-ubuntu-xenial/05a13dc/ : FAILURE in 2m 22s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-pep8-ubuntu-xenial/3fce407/ : FAILURE in 7m 42s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python27-ubuntu-xenial/27ede78/ : FAILURE in 1m 40s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python35/5209cdd/ : FAILURE in 1m 39s\n- gate-networking-midonet-releasenotes http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-releasenotes/cde2df8/ : FAILURE in 3m 39s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/2b1f10f/ : FAILURE in 49s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/c1cb3b4/ : FAILURE in 16s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/115e528/ : FAILURE in 16s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/8dc47f6/ : FAILURE in 1m 08s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/networking-midonet-coverage-ubuntu-xenial-nv/ff6c6d7/ : FAILURE in 1m 38s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/cc3118c/ : FAILURE in 16s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/789b912/ : FAILURE in 26s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/49ac497/ : FAILURE in 21s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"4d7c8a62fee56a139c625eea7d659a91ae24038f","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-06-01 13:47:47.000000000","message":"Patch Set 74:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-networking-midonet-docs-ubuntu-xenial: unrecognized error\n- gate-networking-midonet-pep8-ubuntu-xenial: unrecognized error\n- gate-networking-midonet-python27-ubuntu-xenial: unrecognized error\n- gate-networking-midonet-python35: unrecognized error\n- gate-networking-midonet-releasenotes: unrecognized error\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial: https://bugs.launchpad.net/bugs/1497006\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial: https://bugs.launchpad.net/bugs/1497006\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":74},{"id":"d5424bbe74f4184972ae08e47d9e7f68a4402fe5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-02 02:22:45.000000000","message":"Patch Set 74:\n\nrecheck bug 1694904","accounts_in_message":[],"_revision_number":74},{"id":"dc008760c5938a21f3ca3519fb3c2c3e3094abc9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-02 03:52:34.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/74/check/gate-networking-midonet-docs-ubuntu-xenial/6fafd62//doc/build/html/ : SUCCESS in 4m 16s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-pep8-ubuntu-xenial/bbc6954/ : SUCCESS in 5m 03s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python27-ubuntu-xenial/fc32819/ : SUCCESS in 11m 26s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-python35/e7a5818/ : SUCCESS in 12m 41s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/74/check/gate-networking-midonet-releasenotes/79255d5//releasenotes/build/html/ : SUCCESS in 8m 45s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/e3ab36f/ : SUCCESS in 34m 42s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/cc7963f/ : SUCCESS in 1h 28m 13s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/f61088d/ : SUCCESS in 34m 30s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/c43fe68/ : SUCCESS in 1h 10m 29s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/networking-midonet-coverage-ubuntu-xenial-nv/15bd28c/ : SUCCESS in 15m 37s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/ce188e4/ : FAILURE in 46m 17s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/b8d1e85/ : SUCCESS in 44m 57s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/74/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/62a6540/ : SUCCESS in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":74},{"id":"5a0e0d34104fd63ddb9e9ae7fe9a5dc3b850feea","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-05 10:03:15.000000000","message":"Patch Set 75: Published edit on patch set 74","accounts_in_message":[],"_revision_number":75},{"id":"abe224fb2895c4941580f2d763ce74d53198fca1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-05 10:13:44.000000000","message":"Patch Set 75: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":75},{"id":"2fce3d923beb07298cf596456ce8c6c5219065de","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-05 10:14:16.000000000","message":"Patch Set 75:\n\nrecheck","accounts_in_message":[],"_revision_number":75},{"id":"225514f8470c6f0c2fff1302cde0869f5a911e8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-05 11:54:13.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/75/check/gate-networking-midonet-docs-ubuntu-xenial/a4fe3d3//doc/build/html/ : SUCCESS in 6m 11s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-pep8-ubuntu-xenial/7a4fb33/ : SUCCESS in 5m 12s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-python27-ubuntu-xenial/1fde36d/ : SUCCESS in 11m 40s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-python35/df2f79e/ : SUCCESS in 14m 21s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/75/check/gate-networking-midonet-releasenotes/613266f//releasenotes/build/html/ : SUCCESS in 8m 39s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/75/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/a60ff8f/ : SUCCESS in 50m 07s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/f66baa5/ : SUCCESS in 1h 30m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/75/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/39f4179/ : SUCCESS in 46m 00s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/87410cf/ : SUCCESS in 1h 36m 46s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/networking-midonet-coverage-ubuntu-xenial-nv/e825fe7/ : SUCCESS in 13m 49s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/fbe1712/ : SUCCESS in 51m 41s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/336d23d/ : SUCCESS in 46m 44s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/c5949c8/ : SUCCESS in 51m 40s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"bc6c6af9e7e54511beff8f5000aa84cd7879e0f5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-07 06:26:00.000000000","message":"Patch Set 75:\n\nrecheck","accounts_in_message":[],"_revision_number":75},{"id":"56f5299a7b128681ed4ea9a41d9b9593776a39b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-07 08:02:37.000000000","message":"Patch Set 75: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/75/check/gate-networking-midonet-docs-ubuntu-xenial/19e0c62//doc/build/html/ : SUCCESS in 4m 03s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-pep8-ubuntu-xenial/6b6f897/ : SUCCESS in 8m 14s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-python27-ubuntu-xenial/48614e7/ : FAILURE in 13m 23s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-python35/17b4be0/ : FAILURE in 13m 31s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/75/check/gate-networking-midonet-releasenotes/f854952//releasenotes/build/html/ : SUCCESS in 9m 08s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/75/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/08ea3ab/ : SUCCESS in 36m 31s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/053041f/ : SUCCESS in 1h 16m 14s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/75/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/e8c2266/ : SUCCESS in 52m 49s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/8117f25/ : SUCCESS in 1h 34m 18s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/networking-midonet-coverage-ubuntu-xenial-nv/5972fc3/ : FAILURE in 13m 44s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/8b00fe9/ : SUCCESS in 50m 00s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/b9a2329/ : SUCCESS in 55m 53s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/75/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/1082fa8/ : SUCCESS in 1h 02m 19s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"71272a20a0fc2fb385dc32dab3d032540787348b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-07 10:13:16.000000000","message":"Patch Set 76: Published edit on patch set 75","accounts_in_message":[],"_revision_number":76},{"id":"4453f6257ecea21a74548ba794d8f66d1fe7dd2d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-07 11:45:53.000000000","message":"Patch Set 76: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-docs-ubuntu-xenial/9797a70//doc/build/html/ : SUCCESS in 8m 05s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-pep8-ubuntu-xenial/2388ba2/ : FAILURE in 5m 38s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python27-ubuntu-xenial/b2c41c4/ : SUCCESS in 11m 40s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python35/426d3bc/ : SUCCESS in 14m 11s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-releasenotes/50af528//releasenotes/build/html/ : SUCCESS in 10m 22s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/9406ae7/ : SUCCESS in 45m 51s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/dcf0953/ : SUCCESS in 1h 17m 37s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/da8e788/ : FAILURE in 53m 01s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/952dc52/ : FAILURE in 1h 29m 51s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/networking-midonet-coverage-ubuntu-xenial-nv/042780b/ : SUCCESS in 14m 39s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/dda81c6/ : SUCCESS in 47m 01s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/2de8e4a/ : SUCCESS in 55m 51s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/1d5bd13/ : SUCCESS in 50m 49s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"e485e9f4efab8dc18e0471da707f49f7d3e9d966","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-06-07 11:46:05.000000000","message":"Patch Set 76:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-networking-midonet-pep8-ubuntu-xenial: unrecognized error\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial: https://bugs.launchpad.net/bugs/1696125\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":76},{"id":"e2501eceb978ca01a70bb72dac0408c1fa7757c2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-07 13:17:10.000000000","message":"Patch Set 76:\n\nrecheck","accounts_in_message":[],"_revision_number":76},{"id":"8019fb171dd98537993925f18697bf3800ef1af0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-07 14:51:58.000000000","message":"Patch Set 76:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-docs-ubuntu-xenial/1a0dfb9//doc/build/html/ : SUCCESS in 6m 14s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-pep8-ubuntu-xenial/48fbed4/ : FAILURE in 4m 34s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python27-ubuntu-xenial/e3c25e4/ : SUCCESS in 13m 17s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python35/9f734e7/ : SUCCESS in 13m 11s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-releasenotes/9d93d1a//releasenotes/build/html/ : SUCCESS in 5m 45s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/8cb8389/ : SUCCESS in 37m 12s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/5c4aed7/ : SUCCESS in 1h 21m 31s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/872634a/ : SUCCESS in 47m 39s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/65e17d1/ : SUCCESS in 1h 32m 44s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/networking-midonet-coverage-ubuntu-xenial-nv/f4b5ea0/ : SUCCESS in 14m 42s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/274d9c4/ : SUCCESS in 1h 06m 40s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/57b3dbe/ : SUCCESS in 1h 09m 50s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/3537f68/ : SUCCESS in 55m 17s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"40e0015f812d88d1d3b4a44b3eec3d8ad801a430","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-08 07:35:17.000000000","message":"Patch Set 76:\n\nrecheck","accounts_in_message":[],"_revision_number":76},{"id":"6a685c4293cf3f99bc3e1c43e76dac99aea07add","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-08 09:03:44.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-docs-ubuntu-xenial/a216d31//doc/build/html/ : SUCCESS in 4m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-pep8-ubuntu-xenial/bb38c90/ : SUCCESS in 10m 20s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python27-ubuntu-xenial/18f0512/ : SUCCESS in 12m 45s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python35/ff5467a/ : SUCCESS in 11m 51s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-releasenotes/c533f46//releasenotes/build/html/ : SUCCESS in 9m 25s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/d9d41c9/ : SUCCESS in 34m 37s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/b407276/ : SUCCESS in 1h 23m 30s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/61b2bed/ : SUCCESS in 40m 32s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/964c595/ : SUCCESS in 1h 25m 59s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/networking-midonet-coverage-ubuntu-xenial-nv/8ee3559/ : SUCCESS in 13m 38s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/e0a6c8c/ : SUCCESS in 1h 05m 58s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/d887dba/ : SUCCESS in 1h 06m 23s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/fe8cf1c/ : SUCCESS in 53m 06s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"7bac221c301a9074cc788eb6b5abaeb040b3fd63","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-09 04:21:46.000000000","message":"Patch Set 76:\n\nrecheck","accounts_in_message":[],"_revision_number":76},{"id":"7004f527eb375d6b1aa9eb238ca2ed3d29341d77","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-09 05:42:40.000000000","message":"Patch Set 76:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-docs-ubuntu-xenial/d6e3c1f//doc/build/html/ : SUCCESS in 4m 28s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-pep8-ubuntu-xenial/c08190d/ : SUCCESS in 6m 01s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python27-ubuntu-xenial/4842f48/ : SUCCESS in 10m 58s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python35/6aa067e/ : SUCCESS in 11m 31s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-releasenotes/b71c09d//releasenotes/build/html/ : SUCCESS in 8m 48s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/567be42/ : SUCCESS in 31m 38s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/fc1d4f4/ : SUCCESS in 1h 19m 23s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/0765e4f/ : SUCCESS in 40m 57s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/167bec1/ : SUCCESS in 1h 17m 45s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/networking-midonet-coverage-ubuntu-xenial-nv/05dcbc2/ : SUCCESS in 12m 18s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/bb896bc/ : SUCCESS in 45m 17s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/c73bc5b/ : SUCCESS in 44m 42s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/168376d/ : SUCCESS in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"761d2530462e88433934cd6ac7355e836224af92","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2017-06-09 22:31:13.000000000","message":"Patch Set 76:\n\nrecheck","accounts_in_message":[],"_revision_number":76},{"id":"133674dc4de1cfd56ce55a6f4d71ec339403d079","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-09 23:53:43.000000000","message":"Patch Set 76:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-docs-ubuntu-xenial/0efc68d//doc/build/html/ : SUCCESS in 3m 38s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-pep8-ubuntu-xenial/afbae1d/ : SUCCESS in 5m 31s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python27-ubuntu-xenial/06631c0/ : SUCCESS in 11m 24s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python35/9b6c89a/ : SUCCESS in 10m 52s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-releasenotes/6f9c038//releasenotes/build/html/ : SUCCESS in 8m 49s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/6cc4410/ : SUCCESS in 34m 55s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/006f347/ : SUCCESS in 1h 17m 27s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/b0d81cf/ : SUCCESS in 45m 12s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/020e024/ : SUCCESS in 1h 20m 46s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/networking-midonet-coverage-ubuntu-xenial-nv/eb1209b/ : SUCCESS in 13m 07s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/76b80c2/ : SUCCESS in 46m 53s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/d14f14d/ : SUCCESS in 46m 38s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/4c7e7cd/ : SUCCESS in 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"37e6f7b223a35912bbbb57ddf2eeac0f605d0c8d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-12 03:30:30.000000000","message":"Patch Set 76:\n\nrecheck","accounts_in_message":[],"_revision_number":76},{"id":"18a852c20034d6ab611a7f1d1f3562ec8d03c650","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-12 04:59:32.000000000","message":"Patch Set 76:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-docs-ubuntu-xenial/cb4c42e//doc/build/html/ : SUCCESS in 6m 06s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-pep8-ubuntu-xenial/7ef7af5/ : SUCCESS in 10m 17s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python27-ubuntu-xenial/f8eccf8/ : SUCCESS in 11m 58s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python35/5bb092f/ : SUCCESS in 11m 58s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-releasenotes/4509ab2//releasenotes/build/html/ : SUCCESS in 5m 21s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/51090f9/ : SUCCESS in 37m 41s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/472b86a/ : SUCCESS in 1h 20m 59s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/9fb4535/ : SUCCESS in 35m 59s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/4845075/ : SUCCESS in 1h 27m 42s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/networking-midonet-coverage-ubuntu-xenial-nv/5427208/ : SUCCESS in 15m 03s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/8d68148/ : SUCCESS in 45m 56s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/c52eff6/ : SUCCESS in 1h 01m 33s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/291c787/ : SUCCESS in 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"a6300782469ca0753d336edd4cf32a4c60b59095","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-15 11:59:33.000000000","message":"Patch Set 76:\n\nrecheck","accounts_in_message":[],"_revision_number":76},{"id":"155c799ac6702091db8e7dd4235d56251976ba5f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 12:51:00.000000000","message":"Patch Set 76: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-docs-ubuntu-xenial/33e33ac//doc/build/html/ : SUCCESS in 3m 51s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-pep8-ubuntu-xenial/8f92769/ : SUCCESS in 5m 32s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python27-ubuntu-xenial/234ee31/ : SUCCESS in 13m 03s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-python35/ee109cd/ : SUCCESS in 11m 55s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/76/check/gate-networking-midonet-releasenotes/11fda6b//releasenotes/build/html/ : SUCCESS in 9m 32s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/5625cc1/ : FAILURE in 18m 36s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/ef36917/ : FAILURE in 13m 23s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/abef49b/ : FAILURE in 12m 29s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/5db58ef/ : FAILURE in 13m 03s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/networking-midonet-coverage-ubuntu-xenial-nv/d1d6053/ : SUCCESS in 14m 29s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/a4f6e95/ : SUCCESS in 49m 24s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/a018115/ : FAILURE in 18m 07s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/76/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/5a667c2/ : FAILURE in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"6e035eb061810a131a06bb524695afbd9fc2d97c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-15 13:48:06.000000000","message":"Patch Set 77: Published edit on patch set 76","accounts_in_message":[],"_revision_number":77},{"id":"3bd3f776e78850abecedddaa27128c268f21bfa0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 15:12:46.000000000","message":"Patch Set 77: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/77/check/gate-networking-midonet-docs-ubuntu-xenial/c737edb//doc/build/html/ : SUCCESS in 3m 56s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-pep8-ubuntu-xenial/4dae00f/ : SUCCESS in 8m 29s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-python27-ubuntu-xenial/7d8aec7/ : SUCCESS in 12m 49s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-python35/41a7d80/ : SUCCESS in 11m 25s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/77/check/gate-networking-midonet-releasenotes/8511dec//releasenotes/build/html/ : SUCCESS in 5m 19s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/e834578/ : FAILURE in 41m 25s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/8199007/ : FAILURE in 1h 13m 21s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/81caf57/ : SUCCESS in 44m 48s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/076440f/ : FAILURE in 1h 22m 13s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/networking-midonet-coverage-ubuntu-xenial-nv/aeaea0f/ : SUCCESS in 14m 23s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/c0515e1/ : SUCCESS in 1h 13m 19s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/0f48b82/ : SUCCESS in 55m 22s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/e0de004/ : SUCCESS in 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"bb578c00495734502d208682aa99cfe93bbef9a0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-15 15:31:28.000000000","message":"Patch Set 77:\n\nrecheck","accounts_in_message":[],"_revision_number":77},{"id":"ca2ec9f725d5523bc8bbbcba8b96a27ebae1676e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 16:54:34.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/77/check/gate-networking-midonet-docs-ubuntu-xenial/96b50a2//doc/build/html/ : SUCCESS in 3m 51s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-pep8-ubuntu-xenial/20f4662/ : SUCCESS in 6m 00s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-python27-ubuntu-xenial/d48f16e/ : SUCCESS in 12m 41s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-python35/b06e1d5/ : SUCCESS in 11m 20s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/77/check/gate-networking-midonet-releasenotes/5b7273c//releasenotes/build/html/ : SUCCESS in 7m 34s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/afe9126/ : SUCCESS in 34m 05s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/44f3b1a/ : FAILURE in 1h 11m 07s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/bdfbe1c/ : SUCCESS in 36m 43s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/82de8c1/ : FAILURE in 1h 20m 47s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/networking-midonet-coverage-ubuntu-xenial-nv/9d241a0/ : SUCCESS in 13m 18s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/47c63b5/ : SUCCESS in 1h 09m 24s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/3de18d9/ : SUCCESS in 57m 18s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/58518b1/ : SUCCESS in 44m 13s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"0af8567c56bc186c9bdd4d86dd6d04b8063688ee","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-16 03:29:57.000000000","message":"Patch Set 77:\n\nrecheck","accounts_in_message":[],"_revision_number":77},{"id":"afd64359bdce2083c5ea65c624c5464efdd74415","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-16 04:58:35.000000000","message":"Patch Set 77:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/77/check/gate-networking-midonet-docs-ubuntu-xenial/b0a2c8e//doc/build/html/ : SUCCESS in 4m 06s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-pep8-ubuntu-xenial/bba860f/ : SUCCESS in 5m 26s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-python27-ubuntu-xenial/655cea2/ : SUCCESS in 12m 35s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-python35/eb58fdc/ : SUCCESS in 12m 39s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/77/check/gate-networking-midonet-releasenotes/9693ede//releasenotes/build/html/ : SUCCESS in 8m 06s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/74e5782/ : SUCCESS in 31m 07s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/c5e9672/ : FAILURE in 1h 13m 32s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/7d73ce8/ : SUCCESS in 36m 01s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/a66cfac/ : FAILURE in 1h 26m 20s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/networking-midonet-coverage-ubuntu-xenial-nv/3567ba1/ : SUCCESS in 13m 48s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/531548f/ : SUCCESS in 42m 21s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/5e50f9f/ : SUCCESS in 39m 46s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/7f6c33d/ : SUCCESS in 32m 13s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"fcaddc162e183fca3dae809e67e58da0c46b2716","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-16 05:34:24.000000000","message":"Patch Set 77:\n\nrecheck","accounts_in_message":[],"_revision_number":77},{"id":"718800e89f5f50c7fbabe4ebc84df9f3903ebe9f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-16 07:08:12.000000000","message":"Patch Set 77:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/77/check/gate-networking-midonet-docs-ubuntu-xenial/99a615e//doc/build/html/ : SUCCESS in 6m 41s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-pep8-ubuntu-xenial/78ecb46/ : SUCCESS in 5m 10s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-python27-ubuntu-xenial/8f7c39f/ : SUCCESS in 12m 12s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-python35/00a50ad/ : SUCCESS in 11m 27s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/77/check/gate-networking-midonet-releasenotes/e7d8b0f//releasenotes/build/html/ : SUCCESS in 5m 17s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/48bdde8/ : SUCCESS in 29m 18s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/18e4022/ : FAILURE in 1h 15m 09s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/14bda19/ : SUCCESS in 32m 38s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/5d64c1f/ : FAILURE in 1h 31m 26s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/networking-midonet-coverage-ubuntu-xenial-nv/680b1e2/ : SUCCESS in 14m 58s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/a711d21/ : SUCCESS in 58m 11s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/a470506/ : SUCCESS in 53m 42s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/77/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/6f4bb4c/ : SUCCESS in 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":77},{"id":"bda19c9431137053acb18a9b4489dc282f11d312","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-19 04:25:53.000000000","message":"Patch Set 78: Published edit on patch set 77","accounts_in_message":[],"_revision_number":78},{"id":"3a91797b3e7e8e6ad772d28426e920244c109e0d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-19 06:01:24.000000000","message":"Patch Set 78: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/78/check/gate-networking-midonet-docs-ubuntu-xenial/165791a//doc/build/html/ : SUCCESS in 5m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/78/check/gate-networking-midonet-pep8-ubuntu-xenial/e6e8e3d/ : SUCCESS in 7m 54s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/78/check/gate-networking-midonet-python27-ubuntu-xenial/7dd2834/ : SUCCESS in 11m 50s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/78/check/gate-networking-midonet-python35/acb8b5e/ : SUCCESS in 11m 12s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/78/check/gate-networking-midonet-releasenotes/3a34bb0//releasenotes/build/html/ : SUCCESS in 8m 22s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/78/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/be8c918/ : SUCCESS in 36m 36s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/78/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/912451f/ : FAILURE in 1h 34m 33s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/78/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/2d83d94/ : SUCCESS in 43m 10s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/78/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/17b2844/ : FAILURE in 1h 31m 26s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/78/check/networking-midonet-coverage-ubuntu-xenial-nv/fc6b7bc/ : SUCCESS in 14m 47s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/78/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/2ac434b/ : SUCCESS in 55m 33s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/78/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/7ccceef/ : SUCCESS in 55m 10s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/78/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/eaa5247/ : SUCCESS in 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":78},{"id":"8652e85c940d7ce9464de8e18f0a116bf0cfc827","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-19 06:58:09.000000000","message":"Patch Set 79: Published edit on patch set 78","accounts_in_message":[],"_revision_number":79},{"id":"45190c93415b3f1bb5c46c1f60087a63c0e2d1ea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-19 08:43:11.000000000","message":"Patch Set 79: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/79/check/gate-networking-midonet-docs-ubuntu-xenial/3c31960//doc/build/html/ : SUCCESS in 5m 51s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-pep8-ubuntu-xenial/b53e042/ : SUCCESS in 7m 48s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-python27-ubuntu-xenial/e439b12/ : SUCCESS in 12m 00s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-python35/a0ceb77/ : SUCCESS in 11m 17s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/79/check/gate-networking-midonet-releasenotes/9be2c46//releasenotes/build/html/ : SUCCESS in 7m 40s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/9305936/ : SUCCESS in 35m 08s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/4b2172e/ : FAILURE in 1h 13m 46s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/605b0f2/ : SUCCESS in 37m 10s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/0e26207/ : FAILURE in 1h 43m 03s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/networking-midonet-coverage-ubuntu-xenial-nv/a4bc8dc/ : SUCCESS in 14m 46s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/2495744/ : SUCCESS in 44m 34s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/3df5cb8/ : SUCCESS in 48m 18s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/744e1d6/ : SUCCESS in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"cc7fbf8038c47f7eb495869d0999d9a88a2f8fc1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-19 10:24:44.000000000","message":"Patch Set 79:\n\nrecheck","accounts_in_message":[],"_revision_number":79},{"id":"a294bfc561dbc6c4e8df8bfe56d63b1c998c17ac","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-19 11:57:24.000000000","message":"Patch Set 79: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/79/check/gate-networking-midonet-docs-ubuntu-xenial/d2fe21e//doc/build/html/ : SUCCESS in 5m 50s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-pep8-ubuntu-xenial/c9c14de/ : SUCCESS in 7m 45s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-python27-ubuntu-xenial/fe87c36/ : SUCCESS in 13m 48s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-python35/01ece0d/ : SUCCESS in 12m 30s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/79/check/gate-networking-midonet-releasenotes/77564c5//releasenotes/build/html/ : SUCCESS in 5m 43s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/939cdb4/ : FAILURE in 45m 54s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/51e0956/ : FAILURE in 1h 20m 43s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/331d691/ : FAILURE in 1h 06m 07s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/47e44f1/ : FAILURE in 1h 30m 55s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/networking-midonet-coverage-ubuntu-xenial-nv/118db47/ : SUCCESS in 14m 12s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/98f81da/ : FAILURE in 1h 08m 46s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/2e3fe93/ : SUCCESS in 58m 17s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/4ad0007/ : SUCCESS in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"08a5bc59b0e42e32d644d6d30202038e8c9392a9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-19 12:20:29.000000000","message":"Patch Set 79:\n\nrecheck","accounts_in_message":[],"_revision_number":79},{"id":"e5c9823a24e06d064b349af3160152650f6f01fc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-19 13:58:48.000000000","message":"Patch Set 79:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/79/check/gate-networking-midonet-docs-ubuntu-xenial/57123d7//doc/build/html/ : SUCCESS in 6m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-pep8-ubuntu-xenial/898439b/ : SUCCESS in 5m 19s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-python27-ubuntu-xenial/5553432/ : SUCCESS in 12m 01s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-python35/96e58ad/ : SUCCESS in 11m 08s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/79/check/gate-networking-midonet-releasenotes/7cea429//releasenotes/build/html/ : SUCCESS in 9m 00s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/cbab97c/ : FAILURE in 52m 28s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/62e34f5/ : FAILURE in 1h 36m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/43eea9c/ : FAILURE in 51m 20s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/2fa02a4/ : FAILURE in 1h 30m 27s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/networking-midonet-coverage-ubuntu-xenial-nv/6f1ec42/ : SUCCESS in 14m 54s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/9306340/ : FAILURE in 1h 00m 28s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/51871bc/ : SUCCESS in 54m 22s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/79/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/540f6b4/ : SUCCESS in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":79},{"id":"0b5c7bff840d0ecc020ef9202508b79d319ee1c7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-19 15:44:56.000000000","message":"Patch Set 80: Published edit on patch set 79","accounts_in_message":[],"_revision_number":80},{"id":"fdd4de21fc3c18cb69e16677005e23fc4e52255b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-19 17:26:10.000000000","message":"Patch Set 80: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-docs-ubuntu-xenial/ae43371//doc/build/html/ : SUCCESS in 6m 01s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-pep8-ubuntu-xenial/20f4eee/ : SUCCESS in 9m 37s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python27-ubuntu-xenial/426c339/ : SUCCESS in 13m 28s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python35/3d41288/ : SUCCESS in 11m 37s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-releasenotes/4d7f6b0//releasenotes/build/html/ : SUCCESS in 9m 40s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/a5ed7c2/ : SUCCESS in 36m 36s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/a6119cf/ : FAILURE in 1h 39m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/6051f54/ : SUCCESS in 38m 13s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/5e5819d/ : FAILURE in 1h 20m 23s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/networking-midonet-coverage-ubuntu-xenial-nv/569fc8a/ : SUCCESS in 12m 34s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/97f24c7/ : SUCCESS in 55m 37s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/a948020/ : SUCCESS in 44m 15s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/3470333/ : SUCCESS in 48m 18s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"71084a0193a994fd782c6f52b607f6f4edb4e354","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-20 03:57:04.000000000","message":"Patch Set 80:\n\nrecheck","accounts_in_message":[],"_revision_number":80},{"id":"a90e376cadb98b3f9a18d1754ed16d0c1db555b1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-20 05:15:01.000000000","message":"Patch Set 80:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-docs-ubuntu-xenial/d1495be//doc/build/html/ : SUCCESS in 6m 34s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-pep8-ubuntu-xenial/2cd0f46/ : SUCCESS in 9m 04s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python27-ubuntu-xenial/e167160/ : SUCCESS in 12m 49s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python35/c7121d5/ : SUCCESS in 11m 43s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-releasenotes/efa0cc4//releasenotes/build/html/ : SUCCESS in 8m 09s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/842c1c7/ : SUCCESS in 31m 59s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/9cbcea2/ : FAILURE in 1h 16m 41s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/c491e63/ : SUCCESS in 34m 30s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/f205b36/ : FAILURE in 1h 11m 02s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/networking-midonet-coverage-ubuntu-xenial-nv/72ed6f6/ : SUCCESS in 14m 37s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/1bd23f0/ : FAILURE in 5m 02s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/45e5907/ : SUCCESS in 50m 26s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/ac5bdf3/ : SUCCESS in 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"73dc35cf420883ceebaa99cd80d96aa9f5aefbf2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-20 05:30:53.000000000","message":"Patch Set 80:\n\nrecheck","accounts_in_message":[],"_revision_number":80},{"id":"2183069540082a33ec97fe74a57b7eca2704791a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-20 06:43:22.000000000","message":"Patch Set 80:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-docs-ubuntu-xenial/f2b7f7c//doc/build/html/ : SUCCESS in 5m 17s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-pep8-ubuntu-xenial/01b0cc7/ : SUCCESS in 8m 14s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python27-ubuntu-xenial/ba0f9d2/ : SUCCESS in 14m 25s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python35/ba2b3b8/ : SUCCESS in 11m 04s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-releasenotes/60d64fb//releasenotes/build/html/ : SUCCESS in 5m 11s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/32bff5c/ : SUCCESS in 29m 57s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/f37bdb8/ : FAILURE in 1h 04m 46s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/cae847d/ : SUCCESS in 31m 34s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/9cd0b56/ : FAILURE in 1h 10m 35s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/networking-midonet-coverage-ubuntu-xenial-nv/8c9cfe8/ : SUCCESS in 14m 03s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-grenade-dsvm-networking-midonet-ubuntu-xenial-nv/177e124/ : SUCCESS in 51m 39s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/fe1fd35/ : SUCCESS in 49m 50s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/ce89fc7/ : SUCCESS in 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"627062ef7e8cc3eeee346858aea451c1a1977a06","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-20 07:17:39.000000000","message":"Patch Set 80:\n\nrecheck","accounts_in_message":[],"_revision_number":80},{"id":"3473ac3ca3469ce539f98d23a3bbf7819e75456e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-20 08:21:14.000000000","message":"Patch Set 80:\n\ncheck experimental","accounts_in_message":[],"_revision_number":80},{"id":"6a1ddd8b277fa6ec7ea85b9d1c8070cb67a59b2f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-20 09:02:31.000000000","message":"Patch Set 80:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-v2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/experimental/gate-tempest-dsvm-networking-midonet-v2-full-legacy-ubuntu-xenial-nv/afd96ca/ : FAILURE in 37m 55s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/experimental/gate-tempest-dsvm-networking-midonet-ml2-full-legacy-ubuntu-xenial-nv/7fa7fde/ : FAILURE in 35m 42s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv http://logs.openstack.org/87/199387/80/experimental/gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv/f986a2c/ : FAILURE in 4m 36s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/80/experimental/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/52abe8e/ : FAILURE in 5m 08s (non-voting)\n- gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/experimental/gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial-nv/32aece2/ : FAILURE in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"3b94470be7d59d0bc26baa8f36d75eaa29c08983","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-20 09:28:42.000000000","message":"Patch Set 80:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-docs-ubuntu-xenial/2cc1e04//doc/build/html/ : SUCCESS in 3m 42s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-pep8-ubuntu-xenial/fee2be5/ : SUCCESS in 5m 26s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python27-ubuntu-xenial/2cb52bb/ : SUCCESS in 12m 47s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python35/359e8b4/ : SUCCESS in 11m 56s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-releasenotes/344cb20//releasenotes/build/html/ : SUCCESS in 5m 28s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/a03d533/ : SUCCESS in 35m 24s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/bd76cf5/ : FAILURE in 1h 14m 29s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/7d422bf/ : SUCCESS in 35m 51s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/783f0d6/ : FAILURE in 1h 28m 51s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/networking-midonet-coverage-ubuntu-xenial-nv/3e71662/ : SUCCESS in 14m 41s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/7d07ac0/ : SUCCESS in 1h 01m 12s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/1718d6b/ : FAILURE in 11m 03s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/e91d71d/ : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"6272c188ab2d01c40ee2373060f19f4836f510a2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-20 09:41:11.000000000","message":"Patch Set 80:\n\nrecheck","accounts_in_message":[],"_revision_number":80},{"id":"d6218842cbeab7e691e50d2b51624f5072509680","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-20 11:15:13.000000000","message":"Patch Set 80:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-docs-ubuntu-xenial/e460241//doc/build/html/ : SUCCESS in 6m 39s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-pep8-ubuntu-xenial/592eae0/ : SUCCESS in 8m 38s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python27-ubuntu-xenial/62b48a6/ : SUCCESS in 12m 04s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-python35/64b45d6/ : SUCCESS in 14m 29s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/80/check/gate-networking-midonet-releasenotes/bb3a298//releasenotes/build/html/ : SUCCESS in 11m 19s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/606f432/ : SUCCESS in 34m 31s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/6898710/ : SUCCESS in 1h 30m 04s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/76bb52b/ : SUCCESS in 41m 51s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/ae074d6/ : SUCCESS in 1h 30m 08s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/networking-midonet-coverage-ubuntu-xenial-nv/5c4dd94/ : SUCCESS in 19m 06s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/7c0f922/ : SUCCESS in 50m 27s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/92f158a/ : SUCCESS in 39m 03s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/80/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/db36407/ : SUCCESS in 49m 27s (non-voting)","accounts_in_message":[],"_revision_number":80},{"id":"94c3d184eb2c09eba8dbad5562cf37d63685b93a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-20 11:24:39.000000000","message":"Patch Set 81: Published edit on patch set 80","accounts_in_message":[],"_revision_number":81},{"id":"28e0ed9160ced7698a2a0f38b26db2dc3eb60735","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-20 13:01:09.000000000","message":"Patch Set 81: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/81/check/gate-networking-midonet-docs-ubuntu-xenial/3139767//doc/build/html/ : SUCCESS in 6m 38s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-pep8-ubuntu-xenial/0d25216/ : SUCCESS in 5m 12s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-python27-ubuntu-xenial/9fd9637/ : SUCCESS in 12m 34s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-python35/4cdc23a/ : SUCCESS in 12m 01s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/81/check/gate-networking-midonet-releasenotes/ab04a26//releasenotes/build/html/ : SUCCESS in 5m 13s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/74f4af6/ : SUCCESS in 40m 57s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/a827668/ : SUCCESS in 1h 30m 06s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/99ea95c/ : SUCCESS in 35m 53s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/5ff0306/ : SUCCESS in 1h 33m 34s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/networking-midonet-coverage-ubuntu-xenial-nv/f6413ba/ : SUCCESS in 14m 00s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/0a91aed/ : SUCCESS in 1h 00m 36s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/3d62cc6/ : SUCCESS in 55m 19s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/ab314e7/ : SUCCESS in 47m 14s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"6ddfe023c56f0305eab97017bcd07b5f7e10913d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-22 09:16:32.000000000","message":"Patch Set 81:\n\nrecheck","accounts_in_message":[],"_revision_number":81},{"id":"c4c4006bc37981dda2fd8cca5ee0dc9d9f128fa8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-22 11:15:38.000000000","message":"Patch Set 81:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/81/check/gate-networking-midonet-docs-ubuntu-xenial/c94949a//doc/build/html/ : SUCCESS in 5m 28s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-pep8-ubuntu-xenial/b4fafa2/ : SUCCESS in 7m 49s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-python27-ubuntu-xenial/2cacf8c/ : SUCCESS in 13m 20s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-python35/c86110d/ : SUCCESS in 10m 30s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/81/check/gate-networking-midonet-releasenotes/3e96cb8//releasenotes/build/html/ : SUCCESS in 8m 25s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/5ab2939/ : SUCCESS in 39m 50s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/20a7947/ : SUCCESS in 1h 12m 24s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/2c18494/ : SUCCESS in 34m 42s\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv/b8130f6/ : SUCCESS in 39m 51s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/b99bb1c/ : SUCCESS in 1h 13m 17s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/81/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/f110900/ : FAILURE in 1h 56m 06s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/networking-midonet-coverage-ubuntu-xenial-nv/d5ae824/ : SUCCESS in 13m 06s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/574ddbe/ : SUCCESS in 1h 01m 21s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/3f5c959/ : SUCCESS in 54m 28s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/81/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/82b4580/ : SUCCESS in 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"c6c6d1e84529ab16a00094c3a371651f3988de91","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-25 10:57:41.000000000","message":"Patch Set 82: Published edit on patch set 81","accounts_in_message":[],"_revision_number":82},{"id":"cda5e5cb8f51b22b68dd2912473ffa2972da5620","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-25 11:59:56.000000000","message":"Patch Set 82: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/82/check/gate-networking-midonet-docs-ubuntu-xenial/14e6003//doc/build/html/ : SUCCESS in 3m 58s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/82/check/gate-networking-midonet-pep8-ubuntu-xenial/1051e8b/ : SUCCESS in 5m 01s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/82/check/gate-networking-midonet-python27-ubuntu-xenial/d9fbdf9/ : SUCCESS in 12m 16s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/82/check/gate-networking-midonet-python35/f73b6af/ : SUCCESS in 11m 36s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/82/check/gate-networking-midonet-releasenotes/ad0b23a//releasenotes/build/html/ : SUCCESS in 8m 03s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/82/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/b0c7046/ : SUCCESS in 28m 41s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/82/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/0a3939d/ : FAILURE in 1h 01m 24s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/82/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/d56cb79/ : SUCCESS in 33m 02s\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv http://logs.openstack.org/87/199387/82/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv/7d5aeb1/ : SUCCESS in 34m 14s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/82/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/46d5123/ : FAILURE in 50m 00s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/82/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/7ba7b74/ : FAILURE in 56m 57s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/82/check/networking-midonet-coverage-ubuntu-xenial-nv/e25d398/ : SUCCESS in 12m 41s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/82/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/bd569d1/ : SUCCESS in 41m 57s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/82/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/5cdb708/ : SUCCESS in 51m 12s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/82/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/2be73d3/ : SUCCESS in 48m 27s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"2fb7983d72fa2bc3dbe8fcb4b3b99eba3fabb56e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-26 04:55:36.000000000","message":"Patch Set 83: Published edit on patch set 82","accounts_in_message":[],"_revision_number":83},{"id":"a47ae42632556403bc8bd9cd89da6097c589f3bc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-26 06:02:01.000000000","message":"Patch Set 83: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/83/check/gate-networking-midonet-docs-ubuntu-xenial/d209694//doc/build/html/ : SUCCESS in 3m 26s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-pep8-ubuntu-xenial/8711e9b/ : SUCCESS in 8m 24s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-python27-ubuntu-xenial/19728a6/ : SUCCESS in 12m 47s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-python35/4dacf42/ : SUCCESS in 12m 25s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/83/check/gate-networking-midonet-releasenotes/8f2b173//releasenotes/build/html/ : SUCCESS in 5m 37s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/d613bf8/ : SUCCESS in 31m 12s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/aae0ac2/ : FAILURE in 49m 36s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/56eb873/ : SUCCESS in 33m 26s\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv/65cb824/ : SUCCESS in 34m 22s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/2da9b4f/ : FAILURE in 52m 38s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/b221d8d/ : FAILURE in 1h 00m 45s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/networking-midonet-coverage-ubuntu-xenial-nv/9f8cddf/ : SUCCESS in 12m 23s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/242060a/ : SUCCESS in 1h 05m 05s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/9cf19e8/ : SUCCESS in 38m 11s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/94c588a/ : FAILURE in 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"e8c416a2c424496eb339ff6e4b4dc2373f79daf8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-26 06:56:05.000000000","message":"Patch Set 83:\n\ncheck experimental","accounts_in_message":[],"_revision_number":83},{"id":"913a85183ff44fc74ede2259e8c0901a654ae8a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-26 07:45:10.000000000","message":"Patch Set 83:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-v2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/experimental/gate-tempest-dsvm-networking-midonet-v2-full-legacy-ubuntu-xenial-nv/a478a1b/ : FAILURE in 48m 36s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/experimental/gate-tempest-dsvm-networking-midonet-ml2-full-legacy-ubuntu-xenial-nv/a2ee1ca/ : FAILURE in 30m 29s (non-voting)\n- gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/experimental/gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial-nv/0b14af5/ : FAILURE in 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"3906d2a402f8e53ec6ec1bf5d9fcce006c1268a5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-27 08:29:28.000000000","message":"Patch Set 83:\n\nrecheck","accounts_in_message":[],"_revision_number":83},{"id":"fda3463045d1093391c89550b4d7d9a1aea91df7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-27 09:52:55.000000000","message":"Patch Set 83: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-docs-ubuntu-xenial/3adb73b/ : POST_FAILURE in 5m 28s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-pep8-ubuntu-xenial/a8ce01a/ : SUCCESS in 7m 06s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-python27-ubuntu-xenial/4d262bf/ : SUCCESS in 10m 39s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-python35/5e7e995/ : SUCCESS in 10m 39s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/83/check/gate-networking-midonet-releasenotes/96a5ed3//releasenotes/build/html/ : SUCCESS in 7m 25s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/083e55d/ : SUCCESS in 33m 14s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/347141d/ : SUCCESS in 1h 19m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/072b742/ : SUCCESS in 33m 47s\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7-nv/388f202/ : SUCCESS in 37m 09s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/98d1e78/ : SUCCESS in 1h 16m 36s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/83/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/aa8e972/ : SUCCESS in 1h 20m 41s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/networking-midonet-coverage-ubuntu-xenial-nv/98db2a0/ : SUCCESS in 14m 24s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/137aab3/ : SUCCESS in 51m 05s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/4f2adfd/ : SUCCESS in 39m 42s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/83/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/ade1bbf/ : SUCCESS in 38m 24s (non-voting)","accounts_in_message":[],"_revision_number":83},{"id":"0fc6396f6736b19457d3a013b882ebf8d4a7c7d4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-29 07:37:35.000000000","message":"Patch Set 84: Published edit on patch set 83","accounts_in_message":[],"_revision_number":84},{"id":"920c7401a33a0b201de6aa03de65d2cf9317447b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 09:06:05.000000000","message":"Patch Set 84:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/84/check/gate-networking-midonet-docs-ubuntu-xenial/e26273d//doc/build/html/ : SUCCESS in 12m 04s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-pep8-ubuntu-xenial/5c6be69/ : SUCCESS in 8m 18s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-python27-ubuntu-xenial/0a18e16/ : SUCCESS in 10m 45s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-python35/a0348dc/ : SUCCESS in 11m 52s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/84/check/gate-networking-midonet-releasenotes/54cb377//releasenotes/build/html/ : SUCCESS in 9m 03s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/f951666/ : SUCCESS in 38m 45s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/0a54d76/ : SUCCESS in 1h 07m 16s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/8f98700/ : SUCCESS in 39m 14s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/6d24175/ : SUCCESS in 1h 27m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/54e4fc9/ : SUCCESS in 41m 05s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/0f1723f/ : SUCCESS in 1h 19m 19s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/networking-midonet-coverage-ubuntu-xenial-nv/68ea497/ : SUCCESS in 15m 16s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/310b4c9/ : SUCCESS in 47m 09s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial-nv/5bca7fd/ : SUCCESS in 49m 05s (non-voting)\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial-nv/7e0af99/ : SUCCESS in 56m 17s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"df2ffcb5d6c40893053a0fc7a9df5bc66065d417","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-06-30 08:38:11.000000000","message":"Patch Set 84:\n\nrecheck","accounts_in_message":[],"_revision_number":84},{"id":"53891fc26e8591ed3c3cb536ce0d8831936a553b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-30 10:39:30.000000000","message":"Patch Set 84: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/84/check/gate-networking-midonet-docs-ubuntu-xenial/a7c4aae//doc/build/html/ : SUCCESS in 4m 08s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-pep8-ubuntu-xenial/ef8bb07/ : SUCCESS in 9m 09s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-python27-ubuntu-xenial/dfac5a6/ : SUCCESS in 11m 03s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-python35/362383f/ : SUCCESS in 12m 14s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/84/check/gate-networking-midonet-releasenotes/8e2c79d//releasenotes/build/html/ : SUCCESS in 7m 56s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/0ce8c52/ : FAILURE in 54m 44s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/56dfd8c/ : FAILURE in 1h 59m 23s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/524538a/ : FAILURE in 1h 17m 14s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/16ce6ee/ : FAILURE in 1h 59m 22s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/f4662b4/ : FAILURE in 1h 08m 03s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/84/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/e87ec7e/ : FAILURE in 1h 59m 15s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/networking-midonet-coverage-ubuntu-xenial-nv/60e7c25/ : SUCCESS in 25m 46s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/84/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/abdaeab/ : SUCCESS in 55m 21s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/290eb06/ : SUCCESS in 59m 35s\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial http://logs.openstack.org/87/199387/84/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial/6f6cc12/ : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":84},{"id":"a7f93fef2a37ea5f1b23c6a746763b49efa9db7f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-07-05 12:44:27.000000000","message":"Patch Set 85: Published edit on patch set 84","accounts_in_message":[],"_revision_number":85},{"id":"dc82fe7936eaabd2b2b204bb72686f61df550252","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-05 14:21:10.000000000","message":"Patch Set 85: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/85/check/gate-networking-midonet-docs-ubuntu-xenial/40c2f66//doc/build/html/ : SUCCESS in 6m 47s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/85/check/gate-networking-midonet-pep8-ubuntu-xenial/5ac0cfa/ : SUCCESS in 6m 22s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/85/check/gate-networking-midonet-python27-ubuntu-xenial/7b125f0/ : SUCCESS in 13m 21s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/85/check/gate-networking-midonet-python35/2ec5dfa/ : SUCCESS in 13m 20s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/85/check/gate-networking-midonet-releasenotes/fe92acc//releasenotes/build/html/ : SUCCESS in 12m 08s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/85/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/0197f38/ : SUCCESS in 32m 36s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/85/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/5f22257/ : FAILURE in 20m 07s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/85/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/6d188ab/ : SUCCESS in 36m 44s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/85/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/53dca1e/ : SUCCESS in 1h 34m 58s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/85/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/675e795/ : SUCCESS in 38m 39s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/85/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/1e2236d/ : SUCCESS in 1h 28m 12s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/85/check/networking-midonet-coverage-ubuntu-xenial-nv/bd08f4c/ : SUCCESS in 14m 11s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/85/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/6cdc2e3/ : SUCCESS in 1h 09m 11s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/85/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/06a4716/ : SUCCESS in 58m 23s\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial http://logs.openstack.org/87/199387/85/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial/11abb93/ : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":85},{"id":"8a7c5bd038bdda970240042c5953f784ffe369f6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-07-10 05:16:25.000000000","message":"Patch Set 86: Published edit on patch set 85","accounts_in_message":[],"_revision_number":86},{"id":"22d4b84b1899af1cd15d3bd455ec1b20fa5ebcb7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-10 06:49:27.000000000","message":"Patch Set 86: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/86/check/gate-networking-midonet-docs-ubuntu-xenial/08a2c18//doc/build/html/ : SUCCESS in 5m 58s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-pep8-ubuntu-xenial/532789c/ : SUCCESS in 8m 19s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-python27-ubuntu-xenial/b3102fe/ : SUCCESS in 12m 44s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-python35/5d9ae3b/ : SUCCESS in 11m 38s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/86/check/gate-networking-midonet-releasenotes/2f655c0//releasenotes/build/html/ : SUCCESS in 9m 02s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/3432439/ : SUCCESS in 34m 26s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/f34fcd9/ : SUCCESS in 1h 13m 22s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/3875873/ : SUCCESS in 46m 42s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/1d3a976/ : FAILURE in 1h 18m 37s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/8c532be/ : SUCCESS in 44m 57s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/61a26c3/ : SUCCESS in 1h 31m 29s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/networking-midonet-coverage-ubuntu-xenial-nv/b43e1c9/ : SUCCESS in 13m 58s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/63bc2e8/ : SUCCESS in 50m 07s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/fc5f17c/ : SUCCESS in 37m 26s\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial/dc04e1d/ : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":86},{"id":"ba04b29d2020b884fbead32a2f6278f857407edc","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-07-10 07:26:16.000000000","message":"Patch Set 86:\n\nrecheck","accounts_in_message":[],"_revision_number":86},{"id":"71e2a861733c905ae4fc7c183d264a4c4ea2bfa1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-10 09:24:21.000000000","message":"Patch Set 86:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/86/check/gate-networking-midonet-docs-ubuntu-xenial/c28f514//doc/build/html/ : SUCCESS in 5m 33s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-pep8-ubuntu-xenial/7775b4e/ : SUCCESS in 5m 25s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-python27-ubuntu-xenial/38698c5/ : SUCCESS in 11m 15s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-python35/54835f9/ : SUCCESS in 9m 59s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/86/check/gate-networking-midonet-releasenotes/afbe6fe//releasenotes/build/html/ : SUCCESS in 5m 49s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/7441010/ : SUCCESS in 30m 03s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/ea51b7f/ : SUCCESS in 1h 10m 51s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/d39283b/ : SUCCESS in 44m 27s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/4f10cdf/ : SUCCESS in 1h 35m 24s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/576fab9/ : SUCCESS in 36m 27s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/b817b55/ : FAILURE in 1h 56m 03s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/networking-midonet-coverage-ubuntu-xenial-nv/350bb61/ : SUCCESS in 13m 49s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/8a01c87/ : SUCCESS in 56m 07s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/166702d/ : SUCCESS in 1h 02m 08s\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial/5e96393/ : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":86},{"id":"0bf90526330e4263f252c3f92953d9371cb840d3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-07-13 04:31:33.000000000","message":"Patch Set 86:\n\nrecheck","accounts_in_message":[],"_revision_number":86},{"id":"83615026071397dbfaf251c48d6b792680a9a633","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-13 06:15:38.000000000","message":"Patch Set 86:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/86/check/gate-networking-midonet-docs-ubuntu-xenial/987e75a//doc/build/html/ : SUCCESS in 6m 37s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-pep8-ubuntu-xenial/ad86cfe/ : SUCCESS in 11m 05s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-python27-ubuntu-xenial/6410f2d/ : SUCCESS in 12m 02s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-python35/16a9150/ : SUCCESS in 13m 03s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/86/check/gate-networking-midonet-releasenotes/44403a6//releasenotes/build/html/ : SUCCESS in 7m 59s\n- gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-v2-ubuntu-xenial/e34e5da/ : SUCCESS in 44m 24s\n- gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-v2-full-ubuntu-xenial-nv/d25d1f6/ : SUCCESS in 1h 27m 18s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/35e7133/ : SUCCESS in 35m 02s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/6231ae6/ : SUCCESS in 1h 15m 45s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/d093470/ : SUCCESS in 55m 55s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/f847b5f/ : SUCCESS in 1h 13m 35s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/networking-midonet-coverage-ubuntu-xenial-nv/69ba890/ : SUCCESS in 14m 17s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/5a5e06e/ : SUCCESS in 58m 55s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/4afc494/ : SUCCESS in 1h 02m 17s\n- gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-rally-dsvm-v2-ubuntu-xenial/b095589/ : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":86},{"id":"e4551eb45a7661cfce1c8079f5601d0dafc1d081","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-07-31 01:59:18.000000000","message":"Patch Set 86:\n\nrecheck","accounts_in_message":[],"_revision_number":86},{"id":"4b938e36e674ae6bfbe631bfa2559b1833d8b976","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-31 03:39:11.000000000","message":"Patch Set 86: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/86/check/gate-networking-midonet-docs-ubuntu-xenial/6c5b2f4//doc/build/html/ : SUCCESS in 7m 15s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-pep8-ubuntu-xenial/9c57c7b/ : SUCCESS in 8m 43s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-python27-ubuntu-xenial/f9bfecf/ : SUCCESS in 13m 50s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-python35/5e0d87d/ : SUCCESS in 12m 28s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/86/check/gate-networking-midonet-releasenotes/ef55302//releasenotes/build/html/ : SUCCESS in 5m 14s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/3d55329/ : FAILURE in 36m 07s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/fec6f46/ : SUCCESS in 1h 21m 50s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/5d97589/ : SUCCESS in 48m 58s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/86/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/4ce9f82/ : SUCCESS in 1h 39m 25s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/networking-midonet-coverage-ubuntu-xenial-nv/c15f2aa/ : SUCCESS in 15m 28s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/86/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial-nv/da2ab73/ : SUCCESS in 1h 04m 14s (non-voting)\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/86/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/37ada19/ : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":86},{"id":"e7fb5fdb22d3a551d4a0c836f5cc28f91126c49b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-08-07 00:52:44.000000000","message":"Patch Set 87: Published edit on patch set 86","accounts_in_message":[],"_revision_number":87},{"id":"dc41e949dd7f97e47797fdb6115421c7cb2ba99e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-07 03:14:32.000000000","message":"Patch Set 87: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/87/check/gate-networking-midonet-docs-ubuntu-xenial/1e7c5b6//doc/build/html/ : SUCCESS in 12m 30s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-networking-midonet-pep8-ubuntu-xenial/2914ab6/ : SUCCESS in 7m 43s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-networking-midonet-python27-ubuntu-xenial/c85c806/ : SUCCESS in 9m 03s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/87/check/gate-networking-midonet-python35/c7799bf/ : SUCCESS in 8m 24s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/87/check/gate-networking-midonet-releasenotes/c457063//releasenotes/build/html/ : SUCCESS in 7m 08s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/f06dc37/ : SUCCESS in 55m 43s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/87/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/a8e55b5/ : FAILURE in 2h 17m 46s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/87/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/3e02cf8/ : FAILURE in 1h 39m 08s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/87/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/0cf1e83/ : FAILURE in 2h 20m 07s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/87/check/networking-midonet-coverage-ubuntu-xenial-nv/7e315d2/ : SUCCESS in 14m 00s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/dacbed9/ : SUCCESS in 1h 20m 53s\n- gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial/7ce3dd7/ : SUCCESS in 1h 12m 20s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/ea31e58/ : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":87},{"id":"604e345fccc80bac7bc7ed9b9a4e18b4fb713693","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-08-07 03:18:21.000000000","message":"Patch Set 87:\n\nrecheck","accounts_in_message":[],"_revision_number":87},{"id":"b459558dac334178f726e682c2318a1821fa8866","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-07 05:20:14.000000000","message":"Patch Set 87:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/87/check/gate-networking-midonet-docs-ubuntu-xenial/da1ed46//doc/build/html/ : SUCCESS in 9m 52s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-networking-midonet-pep8-ubuntu-xenial/92f3636/ : SUCCESS in 7m 16s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-networking-midonet-python27-ubuntu-xenial/a720752/ : SUCCESS in 10m 08s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/87/check/gate-networking-midonet-python35/59a37af/ : SUCCESS in 21m 27s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/87/check/gate-networking-midonet-releasenotes/5395d28//releasenotes/build/html/ : SUCCESS in 9m 19s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/5fb46fc/ : FAILURE in 1h 45m 42s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/87/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/033c263/ : SUCCESS in 1h 33m 51s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/87/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/46113cc/ : SUCCESS in 56m 37s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/87/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/ba7e8fe/ : FAILURE in 1h 57m 57s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/87/check/networking-midonet-coverage-ubuntu-xenial-nv/4942e7b/ : SUCCESS in 12m 58s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/f3aa289/ : SUCCESS in 1h 07m 32s\n- gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial/110136c/ : SUCCESS in 1h 10m 30s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/87/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/da4b237/ : FAILURE in 25m 26s","accounts_in_message":[],"_revision_number":87},{"id":"a88fefc64b299a5971a73086e3019f221155e4ce","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-08-07 23:51:56.000000000","message":"Patch Set 88: Published edit on patch set 87","accounts_in_message":[],"_revision_number":88},{"id":"ac7db5d8031fdf703e6bf23bcd624a13ad76c9f7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-08 01:50:07.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/023406c//doc/build/html/ : SUCCESS in 5m 39s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/8cf355b/ : SUCCESS in 7m 30s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/34b3184/ : SUCCESS in 9m 11s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/d3a45bc/ : SUCCESS in 8m 41s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/47200ab//releasenotes/build/html/ : SUCCESS in 8m 59s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/0e3a3ee/ : SUCCESS in 45m 36s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/ed85719/ : SUCCESS in 1h 39m 18s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/50300dd/ : SUCCESS in 57m 51s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/e7aa2c0/ : FAILURE in 1h 57m 55s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/59f45f0/ : SUCCESS in 10m 25s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/4026815/ : SUCCESS in 56m 15s\n- gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial/4a2da70/ : SUCCESS in 53m 45s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/601e36a/ : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":88},{"id":"caf90b8ab00c955448151fa6e22983a71afd2b27","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-08-11 14:49:18.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"de17ac6960c8eb55c0d7f8a837a280fc17e5313e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-11 17:15:36.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/cf2779c//doc/build/html/ : SUCCESS in 7m 13s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/8790cec/ : SUCCESS in 11m 39s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/952aac6/ : SUCCESS in 9m 39s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/4e31cc2/ : SUCCESS in 12m 02s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/2e68c1c//releasenotes/build/html/ : SUCCESS in 9m 25s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/99c02fb/ : FAILURE in 43m 16s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/2813758/ : FAILURE in 2h 17m 53s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/5cbfa6e/ : SUCCESS in 56m 34s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/5eb8a63/ : SUCCESS in 1h 50m 19s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/3680f55/ : SUCCESS in 12m 07s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/cc44909/ : SUCCESS in 1h 12m 58s\n- gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial/e21f864/ : FAILURE in 33m 33s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/39f9800/ : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":88},{"id":"4eaca617fe3a9e66051d20f3f79004d6bd8fd245","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-08-13 22:19:52.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"e8a957839014a783b5acaa4fdb2875a8fe9cf24c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-08-13 22:20:10.000000000","message":"Patch Set 88: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 3203b6425fa36641b2ff43e0f3e3807986150eb6","accounts_in_message":[],"_revision_number":88},{"id":"834aa652a3b1afc6e016ed8c07e639be6cf59f72","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-14 00:19:34.000000000","message":"Patch Set 88:\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/98e3e8c//doc/build/html/ : SUCCESS in 6m 07s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/6b50c4d/ : SUCCESS in 8m 37s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/41c1b74/ : SUCCESS in 12m 37s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/938803e/ : SUCCESS in 9m 52s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/0379ef0//releasenotes/build/html/ : SUCCESS in 8m 39s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/13d9cbe/ : SUCCESS in 46m 39s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/37d2d24/ : SUCCESS in 1h 23m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/144fe77/ : SUCCESS in 1h 02m 19s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/8a8ca34/ : FAILURE in 1h 59m 28s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/8fa9d74/ : SUCCESS in 11m 59s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/39c57ba/ : SUCCESS in 57m 11s\n- gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-v2-ubuntu-xenial/78f10b5/ : FAILURE in 42m 45s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/cc9f2c7/ : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":88},{"id":"b7ea3b5c266e9f73fc0e83ff8f2a073bf42ff161","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-01 02:43:31.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"b1816d395d93678830be50e13dd6a0a11a078ef8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-01 04:27:49.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/bb54e18//doc/build/html/ : SUCCESS in 9m 29s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/e7bea5e/ : SUCCESS in 7m 24s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/4a3293c/ : SUCCESS in 11m 39s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/075984d/ : SUCCESS in 9m 28s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/926bce0//releasenotes/build/html/ : SUCCESS in 8m 50s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/bf0b319/ : SUCCESS in 47m 55s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/31952ae/ : SUCCESS in 1h 28m 23s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/77dfdf9/ : SUCCESS in 49m 01s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/d968450/ : SUCCESS in 1h 41m 03s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/1a8fb16/ : SUCCESS in 10m 29s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/a1272c6/ : SUCCESS in 55m 23s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/7955a65/ : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":88},{"id":"b41ecd7a224bf6dd3c6ad419e831062fec0ed1be","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-12 16:39:34.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"77a763a71849d9c564b34278c613880b96ad1b08","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-12 18:28:21.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/facc5bb//doc/build/html/ : SUCCESS in 7m 01s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/c202a48/ : SUCCESS in 8m 35s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/df262de/ : SUCCESS in 10m 20s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/f98ae69/ : SUCCESS in 11m 13s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/c31b245//releasenotes/build/html/ : SUCCESS in 12m 22s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/352481a/ : SUCCESS in 48m 12s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/1d80d3b/ : SUCCESS in 1h 35m 47s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/f75aa3e/ : SUCCESS in 56m 26s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/2aa63a9/ : SUCCESS in 1h 43m 27s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/2148a92/ : SUCCESS in 11m 21s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/ce9b4e8/ : SUCCESS in 1h 05m 05s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/32a6014/ : FAILURE in 1h 40m 55s","accounts_in_message":[],"_revision_number":88},{"id":"c4d2db98d88e6dfe50420af0da16a2a56273051f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-14 21:09:57.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"6b93553eed7474ea2079afd6b9c5eb139e30cd16","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-14 21:10:07.000000000","message":"Patch Set 88:\n\ncheck experimental","accounts_in_message":[],"_revision_number":88},{"id":"f92fe55100143071ea9b9b114edd01f0fe37ab77","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-14 22:22:23.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"e18ac856e1c07bb3f97d5226c4dea64bf05b1420","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-14 22:49:29.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/442732b//doc/build/html/ : SUCCESS in 7m 42s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/3101215/ : SUCCESS in 9m 07s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/2552f12/ : SUCCESS in 12m 44s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/592b21a/ : SUCCESS in 13m 17s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/2d8bcc4//releasenotes/build/html/ : SUCCESS in 11m 56s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/acd3f44/ : SUCCESS in 42m 15s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/772eca7/ : FAILURE in 1h 35m 01s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/4837f16/ : SUCCESS in 52m 35s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/f3e9b7b/ : FAILURE in 1h 29m 48s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/78f8088/ : SUCCESS in 16m 08s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/6d2dc9d/ : SUCCESS in 55m 41s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/6d3aaa5/ : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":88},{"id":"83e100eab5be2d3411210c89469f9a5a533f2cc1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-14 22:55:09.000000000","message":"Patch Set 88:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/experimental/gate-tempest-dsvm-networking-midonet-ml2-full-legacy-ubuntu-xenial-nv/9d54f1a/ : FAILURE in 1h 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"d65387707dbd497ff3ede06b38f44694842ed5ef","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-18 23:32:04.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"01b97a7b4a3312b11750587e900e788c45480f55","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-19 01:12:36.000000000","message":"Patch Set 88:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/1c896e5//doc/build/html/ : SUCCESS in 6m 21s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/b3e1cab/ : SUCCESS in 6m 41s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/d013aca/ : SUCCESS in 9m 42s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/5b8f7ce/ : SUCCESS in 9m 28s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/e23a4b1//releasenotes/build/html/ : SUCCESS in 8m 54s\n- gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-ubuntu-xenial/5b713ec/ : SUCCESS in 35m 30s\n- gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-ubuntu-xenial-nv/243846b/ : FAILURE in 1h 18m 43s (non-voting)\n- gate-tempest-dsvm-networking-midonet-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-centos-7/809fd0a/ : SUCCESS in 49m 04s\n- gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-ml2-full-centos-7-nv/d185474/ : FAILURE in 1h 37m 16s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/1066584/ : SUCCESS in 11m 14s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/5b5e14a/ : SUCCESS in 50m 34s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/54f3c6c/ : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":88},{"id":"215b36197a2ccaab8eb8896d9ddf09176c2cc914","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-22 00:33:46.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"f4dee2a00e854ca8fff54a62e82a1ea3efea92cf","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-22 00:38:45.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"18551b1b0141fa5eed7118193100ab98142bbfbc","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-22 00:41:55.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"65a46a0949e8df6618ff6dcdca7d1baa0462c698","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 02:28:54.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/4717114//doc/build/html/ : SUCCESS in 6m 17s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/bde4777/ : SUCCESS in 7m 30s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/e270a8c/ : SUCCESS in 9m 11s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/4b19726/ : SUCCESS in 13m 51s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/02f4c7a//releasenotes/build/html/ : SUCCESS in 10m 30s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial/97cb7f4/ : SUCCESS in 47m 59s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv/35e48f4/ : SUCCESS in 1h 16m 26s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv/5247a5b/ : SUCCESS in 1h 30m 42s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7/87b0e11/ : SUCCESS in 50m 11s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv/25806e4/ : FAILURE in 1h 44m 48s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/782df71/ : SUCCESS in 12m 11s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/19ebbf1/ : FAILURE in 1h 08m 41s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/6c178ee/ : FAILURE in 26m 02s","accounts_in_message":[],"_revision_number":88},{"id":"c3d12171c670f94794a116428228e900025a666e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-09-22 07:40:21.000000000","message":"Patch Set 88:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial: https://bugs.launchpad.net/bugs/1696125\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":88},{"id":"04615564e091c78ff3e67e1ebbbf5073c07f0d56","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-25 02:50:25.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"92211edaee38fa0697939dfe2fd3d761b045f62e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-25 02:50:35.000000000","message":"Patch Set 88:\n\ncheck experimental","accounts_in_message":[],"_revision_number":88},{"id":"b7565bea2ef740a0909aa3f9aae54c5b6f42c48d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-25 03:40:21.000000000","message":"Patch Set 88:\n\nBuild succeeded (experimental pipeline).\n\n- gate-tempest-dsvm-networking-midonet-multinode-ml2-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/experimental/gate-tempest-dsvm-networking-midonet-multinode-ml2-ubuntu-xenial-nv/9b287af/ : FAILURE in 44m 50s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"c1126e082a7342efd3d5759ab03c230a0da44bd4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-25 04:25:59.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/60a5f71//doc/build/html/ : SUCCESS in 7m 04s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/009e582/ : SUCCESS in 8m 21s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/b6b122e/ : SUCCESS in 8m 27s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/e34d129/ : SUCCESS in 10m 08s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/f2d46f7//releasenotes/build/html/ : SUCCESS in 8m 25s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial/d69fbe0/ : SUCCESS in 42m 38s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv/d5c7f7d/ : SUCCESS in 1h 28m 10s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv/6bade0c/ : SUCCESS in 1h 34m 04s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7/b90af3c/ : SUCCESS in 53m 13s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv/cb1f36b/ : FAILURE in 1h 29m 34s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/fe29c5d/ : SUCCESS in 9m 52s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/d2a2644/ : SUCCESS in 59m 43s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/e1a50a2/ : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":88},{"id":"34a42e439d3c7b375430cd112652802a1ed0ed08","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-09-29 02:43:06.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"c74372a77bfa200bfd8f8ba4309e1de1c99da2a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-09-29 05:04:04.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed.\n\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/88/check/build-openstack-sphinx-docs/cc9df12/ : RETRY_LIMIT in 12m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/88/check/openstack-tox-pep8/54dced4/ : RETRY_LIMIT in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/88/check/openstack-tox-py27/cdcffb1/ : RETRY_LIMIT in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/88/check/openstack-tox-py35/9d7a506/ : RETRY_LIMIT in 13m 19s\n- legacy-releasenotes http://logs.openstack.org/87/199387/88/check/legacy-releasenotes/cc62e66/ : POST_FAILURE in 4m 17s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/e66a249/ : SUCCESS in 43m 48s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/193b7f7/ : SUCCESS in 1h 52m 01s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/4e9fd57/ : SUCCESS in 1h 40m 26s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/b87c008/ : SUCCESS in 50m 29s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/f9be033/ : FAILURE in 1h 35m 23s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/88/check/openstack-tox-cover/f967a74/ : RETRY_LIMIT in 3m 04s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/88/check/legacy-grenade-dsvm-networking-midonet-ml2/baed3ca/ : SUCCESS in 1h 31m 25s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/88/check/legacy-networking-midonet-rally-dsvm-ml2/c348b2e/ : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":88},{"id":"f3605088b8baee4f1d5c2ee743186cf8ff3ff086","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-03 10:20:35.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"499225584f373f3133490089974655fbde4404fe","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-05 08:55:10.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"4e089a250fa1f714bce90d9ac78d144f53840800","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-05 11:10:30.000000000","message":"Patch Set 88:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/3e0f692//doc/build/html/ : SUCCESS in 6m 06s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/195abea/ : SUCCESS in 7m 02s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/06ed867/ : SUCCESS in 9m 12s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/c6db36c/ : SUCCESS in 12m 59s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/6f584a5//releasenotes/build/html/ : SUCCESS in 9m 14s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial/9eed8a5/ : SUCCESS in 39m 02s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv/c62666e/ : SUCCESS in 1h 21m 51s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv/2d58deb/ : SUCCESS in 1h 37m 54s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7/4ab87a3/ : SUCCESS in 41m 10s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv/c60f35b/ : SUCCESS in 1h 28m 54s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/5c48d6c/ : SUCCESS in 10m 52s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/728d1b6/ : SUCCESS in 1h 11m 15s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/06e6fd3/ : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":88},{"id":"1f4bc51be71ea194f69285c1d4c4ae96116be22e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-10 01:53:14.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"baa998b5cc11c149ce29800b8954219c033528a6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-10 03:43:37.000000000","message":"Patch Set 88:\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/413b73b//doc/build/html/ : SUCCESS in 5m 51s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/960676f/ : SUCCESS in 8m 11s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/a94c0a4/ : SUCCESS in 11m 05s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/b6db790/ : SUCCESS in 11m 30s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/0ea9406//releasenotes/build/html/ : SUCCESS in 10m 21s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial/0f8da6b/ : SUCCESS in 46m 31s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv/01cdd59/ : SUCCESS in 1h 47m 47s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv/8df66ef/ : SUCCESS in 1h 23m 07s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7/fed424a/ : SUCCESS in 46m 18s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv/7f1f76b/ : SUCCESS in 1h 34m 49s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/a785405/ : SUCCESS in 10m 49s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/916905c/ : SUCCESS in 1h 02m 01s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/85b120b/ : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":88},{"id":"0fbf9ed04debc8f87db6f21114ef86679fc08147","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-10 22:39:48.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"2761a85bbc174b25abae53042445819cf98a9c71","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-11 02:40:20.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/66df22c//doc/build/html/ : SUCCESS in 5m 46s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/77cd9d8/ : SUCCESS in 7m 59s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/14d6d83/ : SUCCESS in 10m 26s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/ee37667/ : SUCCESS in 10m 52s\n- gate-networking-midonet-releasenotes http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/da77b24/ : FAILURE in 1m 39s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial/14dec58/ : SUCCESS in 40m 32s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv/f572603/ : SUCCESS in 1h 27m 08s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv/0eaeef0/ : SUCCESS in 1h 23m 16s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7/5596c30/ : SUCCESS in 56m 24s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv/ebf858b/ : SUCCESS in 1h 24m 41s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/993e3b7/ : SUCCESS in 11m 48s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/b516d32/ : SUCCESS in 57m 36s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/5d3b01b/ : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":88},{"id":"f1b1222665ef7e71478e119747cf6dd0b08a84ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-11 06:20:04.000000000","message":"Patch Set 88:\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/88/check/openstack-tox-pep8/9d9eeb9/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/88/check/openstack-tox-py27/7303c68/ : SUCCESS in 11m 32s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/88/check/openstack-tox-py35/41b70ee/ : SUCCESS in 19m 04s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/88/check/build-openstack-sphinx-docs/48e5a4d/ : RETRY_LIMIT in 3m 11s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/88/check/build-openstack-releasenotes/e013319/ : RETRY_LIMIT in 2m 45s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/84d77cb/ : SUCCESS in 50m 07s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/bccd08d/ : SUCCESS in 1h 27m 12s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/eceb732/ : SUCCESS in 1h 32m 05s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/1503f29/ : SUCCESS in 51m 39s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/4a63a0e/ : SUCCESS in 1h 30m 54s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/88/check/openstack-tox-cover/594e4fc/ : SUCCESS in 12m 26s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/88/check/legacy-grenade-dsvm-networking-midonet-ml2/335a460/ : SUCCESS in 1h 02m 47s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/88/check/legacy-networking-midonet-rally-dsvm-ml2/5210d96/ : FAILURE in 5m 11s","accounts_in_message":[],"_revision_number":88},{"id":"dcb131d229b37b3c9e5e838ce193d523a17ff706","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-13 00:25:41.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"ca7403449219c567dfe5da4d8d561a6eee16a59e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-10-13 02:10:09.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-networking-midonet-docs-ubuntu-xenial http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-docs-ubuntu-xenial/776824b//doc/build/html/ : SUCCESS in 6m 25s\n- gate-networking-midonet-pep8-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-pep8-ubuntu-xenial/240565d/ : SUCCESS in 10m 42s\n- gate-networking-midonet-python27-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python27-ubuntu-xenial/0e0cbd4/ : SUCCESS in 9m 11s\n- gate-networking-midonet-python35 http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-python35/b83ee13/ : SUCCESS in 9m 36s\n- gate-networking-midonet-releasenotes http://docs-draft.openstack.org/87/199387/88/check/gate-networking-midonet-releasenotes/9a724a8//releasenotes/build/html/ : SUCCESS in 9m 04s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-ubuntu-xenial/2ddd48d/ : SUCCESS in 44m 38s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-ubuntu-xenial-nv/e250d29/ : SUCCESS in 1h 16m 12s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-legacy-ubuntu-xenial-nv/aade3b3/ : SUCCESS in 1h 25m 09s (non-voting)\n- gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-centos-7/57f2091/ : SUCCESS in 49m 40s\n- gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv http://logs.openstack.org/87/199387/88/check/gate-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7-nv/c600443/ : SUCCESS in 1h 41m 51s (non-voting)\n- networking-midonet-coverage-ubuntu-xenial-nv http://logs.openstack.org/87/199387/88/check/networking-midonet-coverage-ubuntu-xenial-nv/9efe4e5/ : SUCCESS in 10m 03s (non-voting)\n- gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-grenade-dsvm-networking-midonet-ml2-ubuntu-xenial/9b8201c/ : SUCCESS in 1h 01m 23s\n- gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial http://logs.openstack.org/87/199387/88/check/gate-networking-midonet-rally-dsvm-ml2-ubuntu-xenial/e028237/ : SUCCESS in 1h 08m 03s","accounts_in_message":[],"_revision_number":88},{"id":"e43f6a469a96e5860ac91bdf02e56a2f06fd9865","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-16 09:51:28.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"fbc7bf3c706d936f0956c39d6cee41f9916f1ad9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-16 11:43:12.000000000","message":"Patch Set 88:\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/88/check/openstack-tox-pep8/0fa5e20/ : SUCCESS in 9m 51s\n- openstack-tox-py27 openstack-tox-py27 : MERGER_FAILURE\n- openstack-tox-py35 http://logs.openstack.org/87/199387/88/check/openstack-tox-py35/ad848cf/ : SUCCESS in 16m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/88/check/build-openstack-sphinx-docs/7db130a/ : RETRY_LIMIT in 3m 44s\n- build-openstack-releasenotes build-openstack-releasenotes : RETRY_LIMIT\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 legacy-tempest-dsvm-networking-midonet-aio-ml2 : ERROR Project openstack/manila-ui does not have the default branch master\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/6d6229a/ : SUCCESS in 1h 34m 33s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/b26c07a/ : SUCCESS in 1h 34m 27s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/0087ea9/ : SUCCESS in 54m 45s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/1a3514e/ : SUCCESS in 1h 39m 56s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/88/check/openstack-tox-cover/e7ab372/ : SUCCESS in 17m 08s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 legacy-grenade-dsvm-networking-midonet-ml2 : ERROR Project openstack/neutron does not have the default branch master\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/88/check/legacy-networking-midonet-rally-dsvm-ml2/f4bd2f9/ : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":88},{"id":"cae8d5b8bda42fb81203c5f2d09028ff7a7748aa","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-17 01:56:52.000000000","message":"Patch Set 88:\n\nrecheck","accounts_in_message":[],"_revision_number":88},{"id":"3d46aec16c9ae9922a6e3244303de4c9bd31c9f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-17 03:39:31.000000000","message":"Patch Set 88:\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/88/check/openstack-tox-pep8/ed91598/ : SUCCESS in 7m 53s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/88/check/openstack-tox-py27/b1441ea/ : FAILURE in 11m 07s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/88/check/openstack-tox-py35/a2c2ff5/ : FAILURE in 10m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/88/check/build-openstack-sphinx-docs/e28fb25/ : RETRY_LIMIT in 4m 51s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/88/check/build-openstack-releasenotes/332506d/ : RETRY_LIMIT in 2m 53s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/6882eff/ : SUCCESS in 48m 24s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/74f8902/ : SUCCESS in 1h 33m 19s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/a790db6/ : SUCCESS in 1h 31m 43s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/4ee271e/ : SUCCESS in 44m 57s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/88/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/5494b15/ : SUCCESS in 1h 36m 43s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/88/check/openstack-tox-cover/bd3786b/ : FAILURE in 12m 07s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/88/check/legacy-grenade-dsvm-networking-midonet-ml2/ee52599/ : SUCCESS in 1h 04m 42s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/88/check/legacy-networking-midonet-rally-dsvm-ml2/5114951/ : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":88},{"id":"bb1825925fe63eb9f2966e19cc4f84bee490d79b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-17 05:59:33.000000000","message":"Patch Set 89: Commit message was updated.","accounts_in_message":[],"_revision_number":89},{"id":"b7f14fcad2907448da9b695c8cf985127c73a36a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-17 07:45:38.000000000","message":"Patch Set 89: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/89/check/openstack-tox-pep8/a363c6b/ : SUCCESS in 15m 02s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/89/check/openstack-tox-py27/70e7545/ : SUCCESS in 14m 22s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/89/check/openstack-tox-py35/4b724ab/ : SUCCESS in 12m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/89/check/build-openstack-sphinx-docs/2fe70a7/html/ : SUCCESS in 9m 17s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/89/check/build-openstack-releasenotes/0a4c1a5/ : RETRY_LIMIT in 4m 20s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/89/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/b8a081b/ : SUCCESS in 50m 38s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/89/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/1b722f5/ : SUCCESS in 1h 44m 24s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/89/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/fc13ed5/ : SUCCESS in 1h 19m 50s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/89/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/08bda05/ : SUCCESS in 53m 32s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/89/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/b039bbf/ : SUCCESS in 1h 30m 52s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/89/check/openstack-tox-cover/652a2f3/ : SUCCESS in 17m 49s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/89/check/legacy-grenade-dsvm-networking-midonet-ml2/c737ce8/ : SUCCESS in 1h 23m 12s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/89/check/legacy-networking-midonet-rally-dsvm-ml2/0da6393/ : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":89},{"id":"1936a2c4e8a804a39db0f09b853090c8acc6c0e7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-17 09:43:58.000000000","message":"Patch Set 90: Commit message was updated.","accounts_in_message":[],"_revision_number":90},{"id":"ba8c8da6f15e4f5fad1c5aa5ccc62d2c5da27bf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-17 11:26:57.000000000","message":"Patch Set 90: Verified-1\n\nBuild failed.\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/90/check/openstack-tox-pep8/274904a/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/90/check/openstack-tox-py27/145f91b/ : SUCCESS in 13m 39s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/90/check/openstack-tox-py35/877aab3/ : SUCCESS in 12m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/90/check/build-openstack-sphinx-docs/b3ee729/html/ : SUCCESS in 7m 24s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/90/check/build-openstack-releasenotes/7722ea7/ : RETRY_LIMIT in 3m 56s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/90/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/709acc6/ : SUCCESS in 37m 58s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/90/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/b0be8f5/ : SUCCESS in 1h 27m 52s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/90/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/7388f73/ : SUCCESS in 1h 33m 15s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/90/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/fc2194e/ : SUCCESS in 52m 13s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/90/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/0756ec0/ : SUCCESS in 1h 29m 34s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/90/check/openstack-tox-cover/a9188dc/ : SUCCESS in 13m 18s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/90/check/legacy-grenade-dsvm-networking-midonet-ml2/bbe4abf/ : SUCCESS in 51m 37s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/90/check/legacy-networking-midonet-rally-dsvm-ml2/3e26562/ : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":90},{"id":"066a910d94f3e4aa7c2dfb6722165eb4fdee55ff","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-23 06:59:46.000000000","message":"Patch Set 90:\n\nrecheck","accounts_in_message":[],"_revision_number":90},{"id":"3b94230fff4aa06ec80fe6ab28af3fb1c715880a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-23 07:00:45.000000000","message":"Patch Set 91: Commit message was updated.","accounts_in_message":[],"_revision_number":91},{"id":"bfb21e2feefbd1a26d3b554d3754c24cedba1584","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-23 08:39:49.000000000","message":"Patch Set 91: Verified+1\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/91/check/openstack-tox-pep8/5fd0d14/ : SUCCESS in 7m 36s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/91/check/openstack-tox-py27/6f313b1/ : SUCCESS in 11m 09s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/91/check/openstack-tox-py35/01f59f3/ : SUCCESS in 11m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/91/check/build-openstack-sphinx-docs/d2d8a09/html/ : SUCCESS in 9m 48s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/91/check/build-openstack-releasenotes/4a27707/html/ : SUCCESS in 11m 36s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/a159eb5/ : SUCCESS in 41m 06s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/666f6c5/ : SUCCESS in 1h 29m 04s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/c3bd261/ : SUCCESS in 1h 31m 27s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/8c9765e/ : SUCCESS in 51m 29s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/5f4fdd3/ : FAILURE in 10m 20s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/91/check/openstack-tox-cover/7ab0fd6/cover/ : SUCCESS in 15m 21s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/91/check/legacy-grenade-dsvm-networking-midonet-ml2/c5b4769/ : SUCCESS in 59m 05s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/91/check/legacy-networking-midonet-rally-dsvm-ml2/c48a865/ : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":91},{"id":"3d2e72557c64bbc2ff05c6bfaa6c993d3227ea0a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-24 13:13:27.000000000","message":"Patch Set 91:\n\nrecheck","accounts_in_message":[],"_revision_number":91},{"id":"5f4e611e7935e2f851ae0522a19c585980964764","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-24 15:26:49.000000000","message":"Patch Set 91:\n\nBuild succeeded.\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/91/check/openstack-tox-pep8/1972f29/ : SUCCESS in 8m 11s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/91/check/openstack-tox-py27/f6464c2/ : SUCCESS in 13m 19s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/91/check/openstack-tox-py35/b77760a/ : SUCCESS in 12m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/91/check/build-openstack-sphinx-docs/9d6647d/html/ : SUCCESS in 13m 00s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/91/check/build-openstack-releasenotes/15d908d/html/ : SUCCESS in 12m 25s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/c413f3f/ : SUCCESS in 47m 17s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/731f10e/ : SUCCESS in 1h 27m 00s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/2adb007/ : SUCCESS in 1h 25m 37s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/fe28036/ : SUCCESS in 45m 52s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/91/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/a740865/ : FAILURE in 2h 02m 44s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/91/check/openstack-tox-cover/59182a7/cover/ : SUCCESS in 14m 03s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/91/check/legacy-grenade-dsvm-networking-midonet-ml2/d284e9c/ : SUCCESS in 1h 01m 24s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/91/check/legacy-networking-midonet-rally-dsvm-ml2/7710de6/ : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":91},{"id":"bb7062de3280dc799be2308dde9d4c82d23d29eb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-25 05:31:36.000000000","message":"Patch Set 92: Commit message was updated.","accounts_in_message":[],"_revision_number":92},{"id":"5d6486860c1266500addb2a6a26da0a0e00cac9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 06:16:49.000000000","message":"Patch Set 92: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":92},{"id":"8ec2883ecd7ae8b561ccedb227a240f6aad4620b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-25 06:20:01.000000000","message":"Patch Set 92:\n\nrecheck","accounts_in_message":[],"_revision_number":92},{"id":"df0c01e5b74100ce8c510f8256c9f7f3f1a617b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 07:42:34.000000000","message":"Patch Set 92:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":92},{"id":"cd1073bf42f544fb1f6f34bb8dae9d8bff15a79c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-25 07:43:58.000000000","message":"Patch Set 92:\n\nrecheck","accounts_in_message":[],"_revision_number":92},{"id":"8af641ab854bd95e259e82e1caa4e294802ab086","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 09:41:23.000000000","message":"Patch Set 92: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/92/check/openstack-tox-pep8/b642de1/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/92/check/openstack-tox-py27/a9d9f6f/ : SUCCESS in 11m 53s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/92/check/openstack-tox-py35/a7df918/ : SUCCESS in 12m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/92/check/build-openstack-sphinx-docs/7623639/html/ : SUCCESS in 7m 37s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/92/check/build-openstack-releasenotes/d80b58f/html/ : SUCCESS in 10m 37s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/92/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/4bc7360/ : SUCCESS in 36m 31s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/92/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/397e7c0/ : SUCCESS in 1h 30m 33s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/92/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/b4a76fc/ : SUCCESS in 1h 32m 12s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/92/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/246796a/ : SUCCESS in 50m 00s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/92/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/e7fb056/ : SUCCESS in 1h 49m 41s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/92/check/openstack-tox-cover/3d42aac/cover/ : SUCCESS in 16m 52s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/92/check/legacy-grenade-dsvm-networking-midonet-ml2/28f40e0/ : SUCCESS in 1h 06m 29s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/92/check/legacy-networking-midonet-rally-dsvm-ml2/1634469/ : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":92},{"id":"a9c397ed029a7d43f5ec035eda8e415808994fbb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-30 09:14:35.000000000","message":"Patch Set 92:\n\nrecheck","accounts_in_message":[],"_revision_number":92},{"id":"d25a04d9126be569001b64656080fd82fa18c1ae","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-10-31 02:09:38.000000000","message":"Patch Set 93: Commit message was updated.","accounts_in_message":[],"_revision_number":93},{"id":"7aeb4d84351c8aa3a22f40e43bcfbd738f204e78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-31 03:55:41.000000000","message":"Patch Set 93: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/93/check/openstack-tox-pep8/183713b/ : SUCCESS in 7m 53s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/93/check/openstack-tox-py27/5d81965/ : SUCCESS in 12m 17s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/93/check/openstack-tox-py35/a050c9d/ : SUCCESS in 11m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/93/check/build-openstack-sphinx-docs/8a8ae01/html/ : SUCCESS in 7m 53s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/93/check/build-openstack-releasenotes/9e8c72c/html/ : SUCCESS in 9m 59s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/93/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/5c6dfbf/ : SUCCESS in 49m 20s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/93/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/42b768f/ : SUCCESS in 1h 39m 46s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/93/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/af900c6/ : SUCCESS in 1h 26m 33s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/93/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/3681742/ : SUCCESS in 48m 56s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/93/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/640b112/ : SUCCESS in 1h 43m 10s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/93/check/openstack-tox-cover/82bd2d4/cover/ : SUCCESS in 14m 06s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/93/check/legacy-grenade-dsvm-networking-midonet-ml2/38273bf/ : SUCCESS in 1h 05m 01s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/93/check/legacy-networking-midonet-rally-dsvm-ml2/58631d9/ : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":93},{"id":"afbaf392731a48a0b96288f80a72bb7589b44237","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-01 06:22:21.000000000","message":"Patch Set 94: Commit message was updated.","accounts_in_message":[],"_revision_number":94},{"id":"d00485e6beec31ae50c4b3873c9b91ab970c7ab1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-01 08:11:28.000000000","message":"Patch Set 94: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/94/check/openstack-tox-pep8/e1554a3/ : SUCCESS in 7m 20s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/94/check/openstack-tox-py27/dc8f9d4/ : SUCCESS in 12m 11s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/94/check/openstack-tox-py35/d20e784/ : SUCCESS in 11m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/94/check/build-openstack-sphinx-docs/cc10614/html/ : SUCCESS in 7m 08s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/94/check/build-openstack-releasenotes/0455718/html/ : SUCCESS in 10m 39s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/94/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/8bd7bad/ : SUCCESS in 50m 06s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/94/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/a5804b5/ : SUCCESS in 1h 27m 55s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/94/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/15f062f/ : SUCCESS in 1h 37m 23s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/94/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/ab22f5f/ : SUCCESS in 54m 01s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/94/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/55cf2f9/ : SUCCESS in 1h 37m 19s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/94/check/openstack-tox-cover/98b9118/cover/ : SUCCESS in 13m 36s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/94/check/legacy-grenade-dsvm-networking-midonet-ml2/99ab8a5/ : SUCCESS in 1h 07m 47s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/94/check/legacy-networking-midonet-rally-dsvm-ml2/9987baa/ : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":94},{"id":"d4461b8ca985423029db6ebc32a450a55149ef88","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-06 04:06:39.000000000","message":"Patch Set 95: Commit message was updated.","accounts_in_message":[],"_revision_number":95},{"id":"73ed6aa0108f03d407f349e6c7c8e418c81289a6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-06 05:20:48.000000000","message":"Patch Set 96: Commit message was updated.","accounts_in_message":[],"_revision_number":96},{"id":"c491a34eb58641aed03352a8db68a1e957b86b91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-06 07:09:12.000000000","message":"Patch Set 96: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/96/check/openstack-tox-pep8/1d65f52/ : SUCCESS in 7m 52s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/96/check/openstack-tox-py27/7296b41/ : SUCCESS in 11m 23s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/96/check/openstack-tox-py35/c8f88ee/ : SUCCESS in 11m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/96/check/build-openstack-sphinx-docs/5bb37a4/html/ : SUCCESS in 7m 41s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/96/check/build-openstack-releasenotes/8b13dac/html/ : SUCCESS in 10m 42s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/9b132d8/ : SUCCESS in 46m 44s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/42247c4/ : SUCCESS in 1h 29m 48s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/64040d1/ : SUCCESS in 1h 36m 44s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/e8db277/ : SUCCESS in 57m 10s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/47847ce/ : FAILURE in 1h 42m 42s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/96/check/openstack-tox-cover/b629f7d/cover/ : SUCCESS in 12m 32s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/96/check/legacy-grenade-dsvm-networking-midonet-ml2/b9f7ccb/ : SUCCESS in 1h 06m 37s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/96/check/legacy-networking-midonet-rally-dsvm-ml2/8cc0eff/ : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":96},{"id":"7ffcb799a447a30dec137db452033bdf8ac6e4e4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-07 11:56:24.000000000","message":"Patch Set 96:\n\nrecheck","accounts_in_message":[],"_revision_number":96},{"id":"76841f2dfb56084d9dd07519f0510e0501276455","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-07 13:50:18.000000000","message":"Patch Set 96:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/96/check/openstack-tox-pep8/bdcdd32/ : SUCCESS in 7m 23s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/96/check/openstack-tox-py27/e21ad24/ : SUCCESS in 11m 47s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/96/check/openstack-tox-py35/d9a53f7/ : SUCCESS in 14m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/96/check/build-openstack-sphinx-docs/deb59ba/html/ : SUCCESS in 8m 10s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/96/check/build-openstack-releasenotes/72a31fb/html/ : SUCCESS in 10m 29s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/0cf8673/ : SUCCESS in 40m 30s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/8696f9e/ : SUCCESS in 1h 32m 59s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/d431015/ : SUCCESS in 1h 46m 55s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/75cd06d/ : SUCCESS in 44m 45s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/96/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/796a79f/ : SUCCESS in 1h 27m 39s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/96/check/openstack-tox-cover/c92a4f3/cover/ : SUCCESS in 15m 47s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/96/check/legacy-grenade-dsvm-networking-midonet-ml2/1fcf6de/ : SUCCESS in 1h 04m 10s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/96/check/legacy-networking-midonet-rally-dsvm-ml2/a276a97/ : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":96},{"id":"833880727366481aae42b5ade57f71d11ad4f68c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-10 09:59:47.000000000","message":"Patch Set 97: Commit message was updated.","accounts_in_message":[],"_revision_number":97},{"id":"2321d7851d911fdfa223c995345d437a9f2c215d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-10 11:55:30.000000000","message":"Patch Set 97: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/97/check/openstack-tox-pep8/1f60bc7/ : SUCCESS in 7m 35s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/97/check/openstack-tox-py27/070a775/ : SUCCESS in 11m 08s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/97/check/openstack-tox-py35/a88bc9f/ : SUCCESS in 10m 59s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/97/check/build-openstack-sphinx-docs/d74a980/html/ : SUCCESS in 7m 02s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/97/check/build-openstack-releasenotes/68e2715/html/ : SUCCESS in 9m 18s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/97/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/5c22594/ : SUCCESS in 1h 01m 38s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/97/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/472b906/ : SUCCESS in 1h 53m 08s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/97/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/893025a/ : SUCCESS in 1h 49m 05s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/97/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/37a9659/ : SUCCESS in 52m 11s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/97/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/5559da9/ : SUCCESS in 1h 49m 24s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/97/check/openstack-tox-cover/653cf6f/cover/ : SUCCESS in 11m 00s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/97/check/legacy-grenade-dsvm-networking-midonet-ml2/c6ecf4b/ : SUCCESS in 1h 06m 31s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/97/check/legacy-networking-midonet-rally-dsvm-ml2/c465b5e/ : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":97},{"id":"c119794b609a7f665391dd39c0f323c3548dcd65","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-11 13:23:44.000000000","message":"Patch Set 98: Commit message was updated.","accounts_in_message":[],"_revision_number":98},{"id":"e7ec65bcc3b56f1726b83bf5b52cd03fb6cda559","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-11 14:56:16.000000000","message":"Patch Set 98: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/98/check/openstack-tox-pep8/ac998ad/ : SUCCESS in 7m 04s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/98/check/openstack-tox-py27/7ab49e2/ : SUCCESS in 10m 34s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/98/check/openstack-tox-py35/3444798/ : SUCCESS in 10m 36s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/98/check/build-openstack-sphinx-docs/e36b95b/html/ : SUCCESS in 6m 33s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/98/check/build-openstack-releasenotes/ec04d11/html/ : SUCCESS in 9m 11s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/98/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/42be3f2/ : SUCCESS in 37m 49s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/98/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/e24d5da/ : SUCCESS in 1h 17m 46s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/98/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/22a9dbe/ : FAILURE in 1h 19m 48s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/98/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/9e956ce/ : SUCCESS in 41m 13s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/98/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/5515441/ : SUCCESS in 1h 24m 40s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/98/check/openstack-tox-cover/79c6dca/cover/ : SUCCESS in 11m 22s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/98/check/legacy-grenade-dsvm-networking-midonet-ml2/cf84e9e/ : SUCCESS in 53m 03s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/98/check/legacy-networking-midonet-rally-dsvm-ml2/90a66d1/ : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":98},{"id":"e5b99bfe8cee6436259611928d393559cca47976","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-14 10:39:41.000000000","message":"Patch Set 99: Commit message was updated.","accounts_in_message":[],"_revision_number":99},{"id":"6004f3495120d2df9cf7e4c65ba2b717e4ba24da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-14 12:14:48.000000000","message":"Patch Set 99: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/99/check/openstack-tox-pep8/641f2c9/ : SUCCESS in 9m 29s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/99/check/openstack-tox-py27/998c179/ : SUCCESS in 10m 43s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/99/check/openstack-tox-py35/af92e9b/ : SUCCESS in 10m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/99/check/build-openstack-sphinx-docs/f8e416d/html/ : SUCCESS in 8m 36s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/99/check/build-openstack-releasenotes/5369fb9/html/ : SUCCESS in 10m 00s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/1d99df7/ : SUCCESS in 43m 01s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/486d669/ : SUCCESS in 1h 27m 00s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/04a0b85/ : SUCCESS in 1h 28m 23s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/bd786d7/ : SUCCESS in 45m 14s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/67210bd/ : SUCCESS in 1h 31m 14s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/99/check/openstack-tox-cover/2c15faf/cover/ : SUCCESS in 12m 13s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/99/check/legacy-grenade-dsvm-networking-midonet-ml2/128f1f8/ : SUCCESS in 1h 02m 57s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/99/check/legacy-networking-midonet-rally-dsvm-ml2/4cef8de/ : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":99},{"id":"27c4fb27c6f65cf7d2b5139a077ee849eb2d9a88","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-20 04:37:31.000000000","message":"Patch Set 99:\n\nrecheck","accounts_in_message":[],"_revision_number":99},{"id":"a7b7491787a454333ed53dfdbf429d414cb66aa2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-20 06:18:11.000000000","message":"Patch Set 99:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/99/check/openstack-tox-pep8/8532ad4/ : SUCCESS in 7m 44s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/99/check/openstack-tox-py27/e5f5320/ : SUCCESS in 14m 17s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/99/check/openstack-tox-py35/2f52db1/ : SUCCESS in 12m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/99/check/build-openstack-sphinx-docs/f4f4d48/html/ : SUCCESS in 7m 02s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/99/check/build-openstack-releasenotes/25cf1b4/html/ : SUCCESS in 3m 29s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/34fa1c3/ : SUCCESS in 42m 11s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/ec997ac/ : SUCCESS in 1h 39m 09s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/1a5f380/ : SUCCESS in 1h 37m 11s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/43829fd/ : SUCCESS in 45m 16s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/8ddefe2/ : SUCCESS in 1h 28m 22s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/99/check/openstack-tox-cover/ae2bcf6/cover/ : SUCCESS in 19m 27s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/99/check/legacy-grenade-dsvm-networking-midonet-ml2/b028682/ : SUCCESS in 1h 01m 13s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/99/check/legacy-networking-midonet-rally-dsvm-ml2/4124e0b/ : SUCCESS in 1h 27m 19s","accounts_in_message":[],"_revision_number":99},{"id":"9113204a48f3978934a6a7ba5a6c83eee84e426d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-23 00:55:51.000000000","message":"Patch Set 99:\n\nrecheck","accounts_in_message":[],"_revision_number":99},{"id":"524abbb83cc22ea85e388dbe67cb198898f39bb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-23 02:45:27.000000000","message":"Patch Set 99: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/99/check/openstack-tox-pep8/3af443d/ : SUCCESS in 8m 45s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/99/check/openstack-tox-py27/54570b3/ : SUCCESS in 9m 48s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/99/check/openstack-tox-py35/3a41916/ : SUCCESS in 11m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/99/check/build-openstack-sphinx-docs/b2c6445/html/ : SUCCESS in 6m 03s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/988c6a0/ : FAILURE in 53m 16s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/75af0e3/ : FAILURE in 1h 30m 09s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/5149685/ : SUCCESS in 1h 32m 29s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/e55904c/ : FAILURE in 10m 46s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/dace501/ : SUCCESS in 1h 38m 29s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/99/check/openstack-tox-cover/2ff28ce/cover/ : SUCCESS in 11m 40s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/99/check/legacy-grenade-dsvm-networking-midonet-ml2/17d2b77/ : SUCCESS in 1h 20m 02s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/99/check/legacy-networking-midonet-rally-dsvm-ml2/d6e7d66/ : SUCCESS in 57m 59s\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2/c88fd4e/ : SUCCESS in 50m 32s\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2-full/153181c/ : SUCCESS in 1h 38m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2-full-legacy/a912c5e/ : SUCCESS in 1h 37m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2-centos-7/3e1d7da/ : SUCCESS in 1h 02m 37s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2-full-centos-7/a551430/ : SUCCESS in 1h 44m 11s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/99/check/networking-midonet-grenade-ml2/65cc882/ : SUCCESS in 1h 11m 49s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/99/check/networking-midonet-rally-ml2/01758ac/ : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":99},{"id":"8f790b79ee71e32893c9bfefceaf9b5486526457","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-26 12:59:48.000000000","message":"Patch Set 99:\n\nrecheck","accounts_in_message":[],"_revision_number":99},{"id":"9febdf41d8b0dd9c28168683adfbcb4d1549fa04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-26 14:41:43.000000000","message":"Patch Set 99: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/99/check/openstack-tox-pep8/655ce9c/ : SUCCESS in 7m 20s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/99/check/openstack-tox-py27/4a45d4b/ : SUCCESS in 9m 58s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/99/check/openstack-tox-py35/936f857/ : SUCCESS in 9m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/99/check/build-openstack-sphinx-docs/327d887/html/ : SUCCESS in 7m 06s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/99314d8/ : SUCCESS in 55m 07s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/d9fb95f/ : SUCCESS in 1h 30m 26s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/1dd0759/ : SUCCESS in 1h 40m 02s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/b526ac6/ : SUCCESS in 51m 05s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/99/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/d8e7bd9/ : FAILURE in 10m 25s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/99/check/openstack-tox-cover/e9737dc/cover/ : SUCCESS in 12m 34s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/99/check/legacy-grenade-dsvm-networking-midonet-ml2/2571181/ : SUCCESS in 1h 12m 38s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/99/check/legacy-networking-midonet-rally-dsvm-ml2/fde3f63/ : SUCCESS in 1h 14m 53s\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2/7eb7280/ : SUCCESS in 42m 26s\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2-full/b6e7a2d/ : SUCCESS in 1h 29m 41s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2-full-legacy/6e88057/ : SUCCESS in 1h 35m 19s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2-centos-7/66a0833/ : SUCCESS in 51m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/99/check/networking-midonet-tempest-aio-ml2-full-centos-7/eb0ddc5/ : SUCCESS in 1h 38m 53s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/99/check/networking-midonet-grenade-ml2/8700e77/ : SUCCESS in 1h 14m 32s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/99/check/networking-midonet-rally-ml2/1640c6b/ : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":99},{"id":"110aadc08a190af89b464964edf683d1a82a324f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-29 10:06:35.000000000","message":"Uploaded patch set 100.","accounts_in_message":[],"_revision_number":100},{"id":"b461f0cb7523f2d3922c4007f02179ed8bd56ec9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-11-29 10:06:56.000000000","message":"Patch Set 100: Workflow-1","accounts_in_message":[],"_revision_number":100},{"id":"639b816276a218eb4fbf44c269fcda30725c4d60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-29 12:36:23.000000000","message":"Patch Set 100: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/100/check/openstack-tox-pep8/d089a91/ : SUCCESS in 6m 39s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/100/check/openstack-tox-py27/ccf7ab1/ : SUCCESS in 10m 23s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/100/check/openstack-tox-py35/a983f8d/ : SUCCESS in 10m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/100/check/build-openstack-sphinx-docs/ab54614/html/ : SUCCESS in 8m 46s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/100/check/build-openstack-releasenotes/d08ab08/html/ : SUCCESS in 4m 53s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/80823d2/ : SUCCESS in 47m 12s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/583a155/ : SUCCESS in 1h 29m 17s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/7e1ec51/ : SUCCESS in 1h 29m 32s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/d1ddf23/ : SUCCESS in 51m 38s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/9676b71/ : FAILURE in 2h 07m 03s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/100/check/openstack-tox-cover/d9a3a98/cover/ : SUCCESS in 14m 19s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/100/check/legacy-grenade-dsvm-networking-midonet-ml2/6ec7b2c/ : SUCCESS in 1h 23m 09s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/100/check/legacy-networking-midonet-rally-dsvm-ml2/b349eb9/ : SUCCESS in 1h 03m 23s\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2/a24430d/ : SUCCESS in 53m 42s\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full/cfecde2/ : SUCCESS in 1h 28m 11s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-legacy/86bab92/ : SUCCESS in 1h 50m 15s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-centos-7/bc44d56/ : SUCCESS in 49m 38s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-centos-7/d58ee72/ : FAILURE in 1h 21m 22s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-grenade-ml2/3562cc2/ : SUCCESS in 1h 10m 33s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-rally-ml2/db7b850/ : SUCCESS in 1h 15m 25s","accounts_in_message":[],"_revision_number":100},{"id":"6ff28d1ff4dcd7a354fee80e7fb23fcc29e3f237","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-12-11 03:17:23.000000000","message":"Patch Set 100:\n\nrecheck","accounts_in_message":[],"_revision_number":100},{"id":"b5a5726d68a13f49ca82479a10e8785d7e360039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-11 04:49:58.000000000","message":"Patch Set 100: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/100/check/openstack-tox-pep8/1899442/ : SUCCESS in 10m 23s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/100/check/openstack-tox-py27/030f687/ : FAILURE in 14m 25s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/100/check/openstack-tox-py35/c2c8ba7/ : FAILURE in 9m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/100/check/build-openstack-sphinx-docs/5f5d357/html/ : SUCCESS in 5m 56s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/100/check/build-openstack-releasenotes/210d7ca/html/ : SUCCESS in 3m 57s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2 http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2/e1067ba/ : FAILURE in 39m 47s\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full/fbf4053/ : FAILURE in 46m 52s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-legacy/259772a/ : FAILURE in 55m 32s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7 http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-centos-7/71ad2f5/ : FAILURE in 41m 56s (non-voting)\n- legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/100/check/legacy-tempest-dsvm-networking-midonet-aio-ml2-full-centos-7/f88b1bd/ : FAILURE in 47m 07s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/100/check/openstack-tox-cover/a4c9898/ : POST_FAILURE in 12m 12s (non-voting)\n- legacy-grenade-dsvm-networking-midonet-ml2 http://logs.openstack.org/87/199387/100/check/legacy-grenade-dsvm-networking-midonet-ml2/6c1db8f/ : FAILURE in 57m 27s\n- legacy-networking-midonet-rally-dsvm-ml2 http://logs.openstack.org/87/199387/100/check/legacy-networking-midonet-rally-dsvm-ml2/8edc3eb/ : SUCCESS in 1h 11m 04s\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2/0183926/ : FAILURE in 36m 54s\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full/aff1c49/ : FAILURE in 42m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-legacy/fa5f954/ : FAILURE in 49m 24s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-centos-7/2953a05/ : FAILURE in 7m 53s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-centos-7/e0d3462/ : FAILURE in 50m 15s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-grenade-ml2/2188f87/ : FAILURE in 55m 57s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-rally-ml2/9a74e64/ : SUCCESS in 1h 25m 15s","accounts_in_message":[],"_revision_number":100},{"id":"95bc382cf9e98a532634ecff0641565f943bb2d9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-12-31 04:51:39.000000000","message":"Patch Set 100:\n\nrecheck","accounts_in_message":[],"_revision_number":100},{"id":"61a5046c35268e0a1430eadbba11357351e4011d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-31 11:28:49.000000000","message":"Patch Set 100: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/100/check/openstack-tox-pep8/adf3d9d/ : SUCCESS in 11m 48s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/100/check/openstack-tox-py27/1db1af4/ : SUCCESS in 11m 19s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/100/check/openstack-tox-py35/6aec0ab/ : SUCCESS in 10m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/100/check/build-openstack-sphinx-docs/404460c/html/ : SUCCESS in 12m 12s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/100/check/build-openstack-releasenotes/03ba328/html/ : SUCCESS in 4m 18s\n- openstack-tox-cover http://logs.openstack.org/87/199387/100/check/openstack-tox-cover/7ee6d25/cover/ : SUCCESS in 11m 49s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2/96e4a9c/ : SUCCESS in 49m 56s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-multinode-ml2/e54903d/ : SUCCESS in 1h 07m 09s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full/2b0a9ea/ : SUCCESS in 1h 41m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-legacy/278d256/ : SUCCESS in 1h 59m 50s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-centos-7/3a6e319/ : TIMED_OUT in 6h 35m 28s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-centos-7/71381cc/ : FAILURE in 2h 02m 26s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-grenade-ml2/cfc18c2/ : SUCCESS in 1h 34m 55s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-rally-ml2/a3817d0/ : SUCCESS in 1h 17m 31s","accounts_in_message":[],"_revision_number":100},{"id":"ccbc1fc9b0d12bfaa3b9a30a28e899c4ae8b7e50","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2017-12-31 12:24:20.000000000","message":"Patch Set 100:\n\nrecheck","accounts_in_message":[],"_revision_number":100},{"id":"cb327706a98f11cc80736bd3aa0327cfd46215fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-31 14:22:33.000000000","message":"Patch Set 100:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/100/check/openstack-tox-pep8/3c8dfc7/ : SUCCESS in 9m 24s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/100/check/openstack-tox-py27/2131786/ : SUCCESS in 13m 21s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/100/check/openstack-tox-py35/8e73b59/ : SUCCESS in 13m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/100/check/build-openstack-sphinx-docs/7ab26f0/html/ : SUCCESS in 8m 49s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/100/check/build-openstack-releasenotes/90fa4ec/html/ : SUCCESS in 8m 20s\n- openstack-tox-cover http://logs.openstack.org/87/199387/100/check/openstack-tox-cover/6e52a62/cover/ : SUCCESS in 12m 49s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2/1859dd8/ : SUCCESS in 1h 05m 35s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-multinode-ml2/467d464/ : SUCCESS in 1h 29m 37s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full/fe65873/ : SUCCESS in 1h 32m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-legacy/586eff9/ : SUCCESS in 1h 48m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-centos-7/38565c3/ : SUCCESS in 53m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-centos-7/45288a2/ : SUCCESS in 1h 57m 20s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-grenade-ml2/617547a/ : SUCCESS in 1h 22m 21s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-rally-ml2/daaf373/ : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":100},{"id":"b0c592dc2f60e36fb5af721fd4c2fff626ba242c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-08 19:15:04.000000000","message":"Patch Set 100:\n\nrecheck","accounts_in_message":[],"_revision_number":100},{"id":"7d63d2393467bc669c4ba01199e3bcb349fd44a3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-11 02:47:57.000000000","message":"Patch Set 100:\n\nrecheck","accounts_in_message":[],"_revision_number":100},{"id":"191d8cb8a00809356b4f0113edf7665ce15f6e2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-11 11:56:53.000000000","message":"Patch Set 100: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/100/check/openstack-tox-pep8/6baef5c/ : SUCCESS in 8m 24s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/100/check/openstack-tox-py27/9374adc/ : SUCCESS in 10m 38s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/100/check/openstack-tox-py35/064d173/ : SUCCESS in 15m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/100/check/build-openstack-sphinx-docs/1a4f6c2/html/ : SUCCESS in 6m 02s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/100/check/build-openstack-releasenotes/c8eaa5c/html/ : SUCCESS in 5m 00s\n- openstack-tox-cover http://logs.openstack.org/87/199387/100/check/openstack-tox-cover/bf42b1c/cover/ : SUCCESS in 13m 19s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2/e7921b0/ : FAILURE in 1h 00m 42s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-multinode-ml2/1f7d30a/ : FAILURE in 1h 10m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full/b6c1367/ : FAILURE in 2h 02m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-legacy/73a8b8b/ : POST_FAILURE in 3h 41m 13s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-centos-7/7bc68bd/ : FAILURE in 1h 21m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/100/check/networking-midonet-tempest-aio-ml2-full-centos-7/f1942f5/ : FAILURE in 1h 32m 44s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-grenade-ml2/fd8eda7/ : FAILURE in 1h 24m 32s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/100/check/networking-midonet-rally-ml2/c4892a7/ : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":100},{"id":"ee082c273bf61bdbcc942b02432f0a05994f994e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-11 14:18:48.000000000","message":"Patch Set 101: Commit message was updated.","accounts_in_message":[],"_revision_number":101},{"id":"c906eb742ac19cd96efa0b56dd77fdbf2b89ef84","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-12 05:01:47.000000000","message":"Patch Set 101:\n\nrecheck zuul trouble","accounts_in_message":[],"_revision_number":101},{"id":"725b905da49503370d8ed4c1fcb6b5c56eb376b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-12 06:58:58.000000000","message":"Patch Set 101: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/101/check/openstack-tox-pep8/ae16581/ : SUCCESS in 7m 25s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/101/check/openstack-tox-py27/54e5199/ : SUCCESS in 10m 53s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/101/check/openstack-tox-py35/1799a92/ : SUCCESS in 12m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/101/check/build-openstack-sphinx-docs/028296f/html/ : SUCCESS in 6m 19s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/101/check/build-openstack-releasenotes/71047a8/html/ : SUCCESS in 4m 19s\n- openstack-tox-cover http://logs.openstack.org/87/199387/101/check/openstack-tox-cover/02222a8/cover/ : SUCCESS in 12m 19s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2/769d776/ : SUCCESS in 56m 11s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-multinode-ml2/8b0a170/ : SUCCESS in 1h 14m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full/f1e79b1/ : SUCCESS in 1h 53m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full-legacy/9f5624b/ : SUCCESS in 1h 50m 51s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-centos-7/b8dda63/ : FAILURE in 13m 23s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full-centos-7/eeedbcd/ : SUCCESS in 1h 40m 47s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-grenade-ml2/1f71fe3/ : SUCCESS in 1h 19m 04s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-rally-ml2/abf4255/ : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":101},{"id":"deba7c6637ab07617bd0dcb202771f6c74604a31","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-14 13:06:19.000000000","message":"Patch Set 101:\n\nrecheck","accounts_in_message":[],"_revision_number":101},{"id":"1a0c8d123e9f5db390b6d3c7baaca0f15abe86ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-14 15:39:08.000000000","message":"Patch Set 101: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/101/check/openstack-tox-pep8/e8cc88a/ : SUCCESS in 7m 12s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/101/check/openstack-tox-py27/7668034/ : SUCCESS in 13m 22s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/101/check/openstack-tox-py35/2b89867/ : SUCCESS in 23m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/101/check/build-openstack-sphinx-docs/3ebbbec/html/ : SUCCESS in 6m 40s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/101/check/build-openstack-releasenotes/0d252ba/html/ : SUCCESS in 4m 24s\n- openstack-tox-cover http://logs.openstack.org/87/199387/101/check/openstack-tox-cover/85bde7a/cover/ : SUCCESS in 13m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2/dddbae9/ : SUCCESS in 46m 45s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-multinode-ml2/9d76a53/ : SUCCESS in 1h 06m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full/8e4a15a/ : SUCCESS in 1h 39m 48s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full-legacy/71753c4/ : SUCCESS in 1h 48m 07s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-centos-7/2216d3e/ : SUCCESS in 1h 15m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full-centos-7/6496281/ : FAILURE in 2h 03m 45s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-grenade-ml2/fe86b96/ : FAILURE in 2h 29m 11s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-rally-ml2/beec15b/ : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":101},{"id":"76b88bb633ecb8efb7948281f0f15ca2789a5d93","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-15 15:01:13.000000000","message":"Patch Set 101:\n\nrecheck","accounts_in_message":[],"_revision_number":101},{"id":"de544bf306dbff855121c643a469784f0dcdc0a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-15 17:08:23.000000000","message":"Patch Set 101: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/101/check/openstack-tox-pep8/88a9f39/ : SUCCESS in 7m 38s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/101/check/openstack-tox-py27/f1e5bb9/ : SUCCESS in 12m 58s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/101/check/openstack-tox-py35/f5122a5/ : SUCCESS in 11m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/101/check/build-openstack-sphinx-docs/720d9af/html/ : SUCCESS in 6m 42s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/101/check/build-openstack-releasenotes/bca11a3/html/ : SUCCESS in 4m 50s\n- openstack-tox-cover http://logs.openstack.org/87/199387/101/check/openstack-tox-cover/01cc3be/cover/ : SUCCESS in 13m 15s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2/aef43ce/ : SUCCESS in 1h 06m 37s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-multinode-ml2/2ce0020/ : FAILURE in 1h 01m 09s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full/853bff2/ : SUCCESS in 1h 42m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full-legacy/b260102/ : SUCCESS in 1h 29m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-centos-7/5077a36/ : FAILURE in 11m 19s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full-centos-7/e7fc0ae/ : SUCCESS in 1h 35m 11s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-grenade-ml2/bc0a44d/ : SUCCESS in 1h 15m 00s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-rally-ml2/8379917/ : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":101},{"id":"de7cca179835a221780e9a483b965cd2986df56f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-16 00:59:21.000000000","message":"Patch Set 101:\n\nrecheck","accounts_in_message":[],"_revision_number":101},{"id":"8b26f695ed4b4450460ec242cacf3380df1cddb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-16 03:14:18.000000000","message":"Patch Set 101: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/101/check/openstack-tox-pep8/db249c0/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/101/check/openstack-tox-py27/4a212df/ : POST_FAILURE in 13m 08s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/101/check/openstack-tox-py35/eede14c/ : POST_FAILURE in 12m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/101/check/build-openstack-sphinx-docs/bff8841/html/ : SUCCESS in 7m 54s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/101/check/build-openstack-releasenotes/d46a98b/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/87/199387/101/check/openstack-tox-cover/a114f26/cover/ : SUCCESS in 13m 32s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2/84df8c2/ : SUCCESS in 51m 46s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-multinode-ml2/a48a61b/ : SUCCESS in 57m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full/d895421/ : SUCCESS in 1h 51m 04s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full-legacy/b11f0aa/ : SUCCESS in 1h 56m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-centos-7/04b0972/ : SUCCESS in 1h 10m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/101/check/networking-midonet-tempest-aio-ml2-full-centos-7/d5a4293/ : FAILURE in 13m 35s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-grenade-ml2/914a217/ : SUCCESS in 1h 16m 49s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/101/check/networking-midonet-rally-ml2/56ac681/ : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":101},{"id":"8f8c9c49ae4d74fa7019b2acca12429f10560858","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-16 05:01:33.000000000","message":"Patch Set 102: Commit message was updated.","accounts_in_message":[],"_revision_number":102},{"id":"af02ec5dab8ee0799e9535d4930bd43d93d5b752","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-16 08:13:31.000000000","message":"Patch Set 102: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/102/check/openstack-tox-pep8/cc2e696/ : SUCCESS in 9m 45s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/102/check/openstack-tox-py27/4ab90ba/ : SUCCESS in 14m 02s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/102/check/openstack-tox-py35/173cd53/ : SUCCESS in 13m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/102/check/build-openstack-sphinx-docs/4d7da2d/html/ : SUCCESS in 6m 58s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/102/check/build-openstack-releasenotes/36c2e95/html/ : SUCCESS in 5m 50s\n- openstack-tox-cover http://logs.openstack.org/87/199387/102/check/openstack-tox-cover/e571398/cover/ : SUCCESS in 13m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2/87d627b/ : FAILURE in 56m 22s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-multinode-ml2/c721ef8/ : FAILURE in 1h 14m 51s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2-full/e12c4e3/ : FAILURE in 1h 47m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2-full-legacy/5de34b0/ : FAILURE in 1h 30m 03s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2-centos-7/abde0a6/ : FAILURE in 1h 06m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2-full-centos-7/22995dd/ : FAILURE in 2h 03m 16s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/102/check/networking-midonet-grenade-ml2/c46eaf7/ : FAILURE in 1h 31m 37s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/102/check/networking-midonet-rally-ml2/4acb1f5/ : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":102},{"id":"6a90ef9ad86f6e1c78ef4b3cb049e236a264c8e9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-16 10:34:33.000000000","message":"Patch Set 102:\n\nrecheck","accounts_in_message":[],"_revision_number":102},{"id":"3808a79579b0ff72b261e0b693d3f4ae5d4e5345","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-16 12:47:23.000000000","message":"Patch Set 102:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/102/check/openstack-tox-pep8/e8454c1/ : SUCCESS in 8m 23s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/102/check/openstack-tox-py27/fe86f55/ : SUCCESS in 11m 43s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/102/check/openstack-tox-py35/4bff832/ : SUCCESS in 12m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/102/check/build-openstack-sphinx-docs/5e02eaf/html/ : SUCCESS in 10m 13s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/102/check/build-openstack-releasenotes/50d7cac/html/ : SUCCESS in 5m 17s\n- openstack-tox-cover http://logs.openstack.org/87/199387/102/check/openstack-tox-cover/36aa6af/cover/ : SUCCESS in 12m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2/fcc3ddb/ : FAILURE in 55m 09s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-multinode-ml2/c8d7f33/ : FAILURE in 1h 09m 44s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2-full/e5ff885/ : FAILURE in 1h 39m 41s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2-full-legacy/53c55e7/ : FAILURE in 1h 38m 49s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2-centos-7/39f67d3/ : FAILURE in 1h 02m 09s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/102/check/networking-midonet-tempest-aio-ml2-full-centos-7/532104c/ : FAILURE in 1h 46m 34s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/102/check/networking-midonet-grenade-ml2/aed19fd/ : FAILURE in 1h 12m 41s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/102/check/networking-midonet-rally-ml2/6526d68/ : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":102},{"id":"7ba1f6371a384ff7c08c574d7fe6b25f1a217852","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-26 10:00:30.000000000","message":"Patch Set 103: Commit message was updated.","accounts_in_message":[],"_revision_number":103},{"id":"8e1b7f7b37ed5d3d71944664d91d8cb791679eda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-26 12:20:37.000000000","message":"Patch Set 103: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/103/check/openstack-tox-pep8/355d895/ : SUCCESS in 8m 41s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/103/check/openstack-tox-py27/37fba89/ : SUCCESS in 13m 09s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/103/check/openstack-tox-py35/37fef57/ : SUCCESS in 33m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/103/check/build-openstack-sphinx-docs/37912e7/html/ : SUCCESS in 8m 51s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/103/check/build-openstack-releasenotes/4dd0a1e/html/ : SUCCESS in 5m 24s\n- openstack-tox-cover http://logs.openstack.org/87/199387/103/check/openstack-tox-cover/4a1b624/cover/ : SUCCESS in 12m 32s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2/dddd8d7/ : SUCCESS in 1h 02m 36s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-multinode-ml2/5834adc/ : SUCCESS in 1h 00m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2-full/8c690cf/ : SUCCESS in 1h 23m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2-full-legacy/b1ce552/ : SUCCESS in 1h 39m 16s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2-centos-7/b6773c2/ : FAILURE in 13m 40s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2-full-centos-7/392c5d1/ : FAILURE in 2h 05m 44s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/103/check/networking-midonet-grenade-ml2/e8e1f84/ : SUCCESS in 1h 11m 36s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/103/check/networking-midonet-rally-ml2/8fba080/ : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":103},{"id":"adf50593673ab516df1b73e155c71ee2ff4f7a75","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-28 11:55:47.000000000","message":"Patch Set 103:\n\nrecheck","accounts_in_message":[],"_revision_number":103},{"id":"8f74b4dfb69a87c4a639e4b6416584c0bdd99946","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-28 19:21:18.000000000","message":"Patch Set 103:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/103/check/openstack-tox-pep8/b98e363/ : SUCCESS in 8m 33s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/103/check/openstack-tox-py27/dc37128/ : SUCCESS in 11m 38s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/103/check/openstack-tox-py35/903c9ca/ : SUCCESS in 16m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/103/check/build-openstack-sphinx-docs/296ca1d/html/ : SUCCESS in 8m 17s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/103/check/build-openstack-releasenotes/69c2856/html/ : SUCCESS in 5m 11s\n- openstack-tox-cover http://logs.openstack.org/87/199387/103/check/openstack-tox-cover/ae65fa3/cover/ : SUCCESS in 13m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2/7608ba3/ : SUCCESS in 56m 52s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-multinode-ml2/da82336/ : SUCCESS in 1h 02m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2-full/e567e83/ : SUCCESS in 1h 42m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2-full-legacy/f5b6c92/ : SUCCESS in 1h 39m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2-centos-7/c4c31c2/ : SUCCESS in 57m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/103/check/networking-midonet-tempest-aio-ml2-full-centos-7/bf83773/ : FAILURE in 15m 45s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/103/check/networking-midonet-grenade-ml2/88051ef/ : SUCCESS in 1h 12m 58s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/103/check/networking-midonet-rally-ml2/92e364d/ : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":103},{"id":"a49a11c854c466b84873af218652d47705f3faf1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-01-30 15:08:20.000000000","message":"Patch Set 104: Commit message was updated.","accounts_in_message":[],"_revision_number":104},{"id":"0ab6262c84ff9d3807b664bf20b21d7401e335ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 17:41:07.000000000","message":"Patch Set 104: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/104/check/openstack-tox-pep8/5080b5e/ : SUCCESS in 10m 41s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/104/check/openstack-tox-py27/bc34d3f/ : SUCCESS in 14m 24s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/104/check/openstack-tox-py35/5a136c2/ : SUCCESS in 12m 23s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/104/check/build-openstack-sphinx-docs/2f78c79/html/ : SUCCESS in 9m 58s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/104/check/build-openstack-releasenotes/5485b64/html/ : SUCCESS in 8m 50s\n- openstack-tox-cover http://logs.openstack.org/87/199387/104/check/openstack-tox-cover/23df520/cover/ : SUCCESS in 14m 59s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2/44a7a6f/ : FAILURE in 58m 54s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-multinode-ml2/5bf5012/ : FAILURE in 1h 51m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full/e083f51/ : FAILURE in 1h 43m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-legacy/46047bf/ : FAILURE in 2h 07m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-centos-7/d468ecd/ : SUCCESS in 1h 05m 18s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-centos-7/5765d5e/ : SUCCESS in 1h 37m 57s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-grenade-ml2/f303361/ : SUCCESS in 1h 29m 20s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-rally-ml2/f4d03ce/ : SUCCESS in 1h 03m 56s","accounts_in_message":[],"_revision_number":104},{"id":"978078384ec621faaccc0b7faf1fd604aeec9ad7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-02-01 03:02:50.000000000","message":"Patch Set 104:\n\nrecheck","accounts_in_message":[],"_revision_number":104},{"id":"aa5a5718d46a86891d3f6e3294d9d260aa694efa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-01 05:44:03.000000000","message":"Patch Set 104: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/104/check/openstack-tox-pep8/e759e88/ : SUCCESS in 8m 01s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/104/check/openstack-tox-py27/44fb87c/ : SUCCESS in 11m 59s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/104/check/openstack-tox-py35/432abf2/ : SUCCESS in 11m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/104/check/build-openstack-sphinx-docs/aad6b1d/html/ : SUCCESS in 7m 15s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/104/check/build-openstack-releasenotes/6c7e1f2/html/ : SUCCESS in 5m 37s\n- openstack-tox-cover http://logs.openstack.org/87/199387/104/check/openstack-tox-cover/3e7d757/cover/ : SUCCESS in 12m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2/bc62547/ : SUCCESS in 48m 32s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-multinode-ml2/977bb12/ : SUCCESS in 1h 14m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full/459fa28/ : FAILURE in 1h 26m 58s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-legacy/9299265/ : FAILURE in 1h 35m 37s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-centos-7/1d8569e/ : FAILURE in 9m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-centos-7/cfc3e4d/ : FAILURE in 13m 23s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-grenade-ml2/0b7791c/ : SUCCESS in 1h 14m 13s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-rally-ml2/8b3413e/ : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":104},{"id":"f6c0f6bbb068a533d57f39c9998ee69814fe6b17","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-02-02 14:26:41.000000000","message":"Patch Set 104:\n\nrecheck","accounts_in_message":[],"_revision_number":104},{"id":"fe62076faa49545aa5e013b3a0f29ee6c6e2f7d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-02 16:02:46.000000000","message":"Patch Set 104:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/104/check/openstack-tox-pep8/78c49da/ : SUCCESS in 7m 56s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/104/check/openstack-tox-py27/4f7ef9a/ : SUCCESS in 10m 57s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/104/check/openstack-tox-py35/0c4b619/ : SUCCESS in 11m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/104/check/build-openstack-sphinx-docs/1ad60c3/html/ : SUCCESS in 7m 12s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/104/check/build-openstack-releasenotes/d7d9037/html/ : SUCCESS in 4m 55s\n- openstack-tox-cover http://logs.openstack.org/87/199387/104/check/openstack-tox-cover/65643e0/cover/ : SUCCESS in 13m 12s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2/38e8289/ : SUCCESS in 52m 18s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-multinode-ml2/e7bcd24/ : SUCCESS in 1h 07m 48s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full/6641f57/ : FAILURE in 1h 29m 22s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-legacy/f0db5ef/ : FAILURE in 1h 27m 09s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-centos-7/731a3e1/ : SUCCESS in 1h 04m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-centos-7/1ec8903/ : FAILURE in 8m 55s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-grenade-ml2/bc595d5/ : SUCCESS in 1h 20m 57s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-rally-ml2/26f0d2d/ : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":104},{"id":"9702abeb150771504b3474246301941203df13ce","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-02-06 12:04:07.000000000","message":"Patch Set 104:\n\nrecheck","accounts_in_message":[],"_revision_number":104},{"id":"2d7564a27ddfe0794ca6f1f27287d6854f0b0df2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-07 00:13:22.000000000","message":"Patch Set 104:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/104/check/openstack-tox-pep8/9ff858f/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/104/check/openstack-tox-py27/5e5b9f6/ : SUCCESS in 10m 58s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/104/check/openstack-tox-py35/dc3cff0/ : SUCCESS in 12m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/104/check/build-openstack-sphinx-docs/5b6e0fc/html/ : SUCCESS in 8m 04s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/104/check/build-openstack-releasenotes/f50b5e9/html/ : SUCCESS in 5m 04s\n- openstack-tox-cover http://logs.openstack.org/87/199387/104/check/openstack-tox-cover/71ce91e/cover/ : SUCCESS in 20m 37s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2/226feb0/ : SUCCESS in 45m 40s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-multinode-ml2/58ac0d2/ : SUCCESS in 1h 17m 50s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full/d50906c/ : FAILURE in 1h 31m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-legacy/6c98461/ : FAILURE in 1h 52m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-centos-7/361fea3/ : SUCCESS in 1h 16m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-centos-7/c6d8c58/ : FAILURE in 2h 04m 04s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-grenade-ml2/64a8a89/ : SUCCESS in 1h 15m 37s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-rally-ml2/e676fb9/ : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":104},{"id":"2b0c19d53f413adc798874645e98c3cb3e2a6c0f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-02-08 05:44:37.000000000","message":"Patch Set 104:\n\nrecheck","accounts_in_message":[],"_revision_number":104},{"id":"70a6bd0a6ba3c248bb9fe8e01e6dfbf721d8a0c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-08 07:22:53.000000000","message":"Patch Set 104:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/104/check/openstack-tox-pep8/18bc066/ : SUCCESS in 6m 19s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/104/check/openstack-tox-py27/4839aa5/ : SUCCESS in 10m 55s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/104/check/openstack-tox-py35/2486076/ : SUCCESS in 10m 53s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/104/check/build-openstack-sphinx-docs/cbce2ca/html/ : SUCCESS in 5m 42s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/104/check/build-openstack-releasenotes/23ac3f0/html/ : SUCCESS in 4m 46s\n- openstack-tox-cover http://logs.openstack.org/87/199387/104/check/openstack-tox-cover/e29c2e0/cover/ : SUCCESS in 11m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2/b41b52c/ : SUCCESS in 47m 43s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-multinode-ml2/57cd12f/ : SUCCESS in 1h 03m 23s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full/ebdff42/ : SUCCESS in 1h 34m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-legacy/c765069/ : FAILURE in 1h 27m 39s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-centos-7/713c952/ : SUCCESS in 51m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/104/check/networking-midonet-tempest-aio-ml2-full-centos-7/7179cc8/ : FAILURE in 1h 36m 28s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-grenade-ml2/4b71499/ : SUCCESS in 1h 12m 11s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/104/check/networking-midonet-rally-ml2/035c612/ : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":104},{"id":"3a80a73d741851c8ffb619ea99118be054ced6c4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-02-08 22:34:33.000000000","message":"Patch Set 105: Commit message was updated.","accounts_in_message":[],"_revision_number":105},{"id":"e0289a8fbe466bc6c79c2a31b842d7a1dd0ad01d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-08 23:23:48.000000000","message":"Patch Set 105: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":105},{"id":"027aadc9493ca258365737b6772e58d3af8e094d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-02-21 07:25:26.000000000","message":"Patch Set 106: Commit message was updated.","accounts_in_message":[],"_revision_number":106},{"id":"7354583112209a29df780baafa61639591ec8d59","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-02-21 07:25:41.000000000","message":"Patch Set 106: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 5a3bfe17525b8c388e68b235d5606edab790c612","accounts_in_message":[],"_revision_number":106},{"id":"74d214a5761dc804a33eedc0fb52382f2738b119","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-21 09:17:55.000000000","message":"Patch Set 106: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/106/check/openstack-tox-pep8/ac06859/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/106/check/openstack-tox-py27/0765988/ : SUCCESS in 10m 08s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/106/check/openstack-tox-py35/54aaaba/ : SUCCESS in 12m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/106/check/build-openstack-sphinx-docs/793bd5f/html/ : SUCCESS in 4m 44s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/106/check/build-openstack-releasenotes/6403064/html/ : SUCCESS in 3m 51s\n- openstack-tox-cover http://logs.openstack.org/87/199387/106/check/openstack-tox-cover/7262506/cover/ : SUCCESS in 11m 58s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2/3b6b86d/ : FAILURE in 1h 04m 08s\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full/9b493a8/ : FAILURE in 1h 47m 03s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full-legacy/6a13cd3/ : FAILURE in 1h 28m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-centos-7/cd60a6c/ : FAILURE in 1h 14m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full-centos-7/6964e60/ : FAILURE in 1h 37m 51s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-grenade-ml2/9ed748e/ : FAILURE in 37m 38s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-rally-ml2/51f0e86/ : SUCCESS in 1h 05m 00s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-multinode-ml2/fd1f857/ : FAILURE in 1h 18m 30s (non-voting)","accounts_in_message":[],"_revision_number":106},{"id":"e9c3ed310661ebc9141841c9c8ce792e61aed872","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-01 09:54:58.000000000","message":"Patch Set 106:\n\nrecheck","accounts_in_message":[],"_revision_number":106},{"id":"5741d24010f577f737bf757f42cb2cf0556de06e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-01 11:36:03.000000000","message":"Patch Set 106:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/106/check/openstack-tox-pep8/9658aff/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/106/check/openstack-tox-py27/0af51dd/ : SUCCESS in 11m 16s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/106/check/openstack-tox-py35/0f6c378/ : SUCCESS in 11m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/106/check/build-openstack-sphinx-docs/22f3dee/html/ : SUCCESS in 5m 43s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/106/check/build-openstack-releasenotes/c6b9e96/html/ : SUCCESS in 4m 22s\n- openstack-tox-cover http://logs.openstack.org/87/199387/106/check/openstack-tox-cover/7f31fef/cover/ : SUCCESS in 12m 56s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2/530ff97/ : FAILURE in 1h 05m 50s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-multinode-ml2/4a7eaf7/ : FAILURE in 1h 17m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full/6f7312d/ : FAILURE in 1h 35m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full-legacy/f83eb29/ : FAILURE in 1h 39m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-centos-7/c80b1dc/ : FAILURE in 1h 04m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full-centos-7/550d939/ : FAILURE in 1h 35m 48s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-grenade-ml2/f2b1fb8/ : FAILURE in 33m 50s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-rally-ml2/3c06d8e/ : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":106},{"id":"4f2f64e4cc8cc5f5b8bcb946fbde93fcd69d2b28","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-06 04:40:21.000000000","message":"Patch Set 106:\n\nrecheck","accounts_in_message":[],"_revision_number":106},{"id":"d7acb3656fdd8dacd4cd037dab820cac64c44ba2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-06 06:50:26.000000000","message":"Patch Set 106:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/106/check/openstack-tox-pep8/41b9885/ : SUCCESS in 7m 24s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/106/check/openstack-tox-py27/0709dbe/ : SUCCESS in 10m 38s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/106/check/openstack-tox-py35/95be744/ : SUCCESS in 13m 49s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/106/check/build-openstack-sphinx-docs/232d1ac/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/106/check/build-openstack-releasenotes/c7170ce/html/ : SUCCESS in 4m 46s\n- openstack-tox-cover http://logs.openstack.org/87/199387/106/check/openstack-tox-cover/ef67458/cover/ : SUCCESS in 12m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2/9123648/ : FAILURE in 58m 07s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-multinode-ml2/290d4ce/ : FAILURE in 1h 29m 31s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full/e620e20/ : FAILURE in 1h 36m 11s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full-legacy/ac68877/ : FAILURE in 1h 42m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-centos-7/7b44df5/ : FAILURE in 1h 21m 59s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full-centos-7/8d8745a/ : FAILURE in 2h 04m 02s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-grenade-ml2/2922ce2/ : FAILURE in 34m 05s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-rally-ml2/7ee49f1/ : FAILURE in 21m 16s","accounts_in_message":[],"_revision_number":106},{"id":"efa75f8ff03b9f831868544062454556e961ebae","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-16 15:19:47.000000000","message":"Patch Set 106:\n\nrecheck","accounts_in_message":[],"_revision_number":106},{"id":"7b2529e1dfa52a5b5db7471fd736210c3987bc4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-16 17:44:45.000000000","message":"Patch Set 106:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/106/check/openstack-tox-pep8/f5c4d4d/ : FAILURE in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/106/check/openstack-tox-py27/aab8326/ : FAILURE in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/106/check/openstack-tox-py35/913fafa/ : FAILURE in 3m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/106/check/build-openstack-sphinx-docs/d03626f/ : RETRY_LIMIT in 4m 44s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/106/check/build-openstack-releasenotes/119ce82/html/ : SUCCESS in 5m 04s\n- openstack-tox-cover http://logs.openstack.org/87/199387/106/check/openstack-tox-cover/e89aa03/ : FAILURE in 3m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2/dc55e8a/ : FAILURE in 55m 30s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-multinode-ml2/6185b34/ : FAILURE in 1h 12m 04s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full/7ee0076/ : FAILURE in 1h 26m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full-legacy/5c92236/ : FAILURE in 2h 02m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-centos-7/a8b112a/ : FAILURE in 1h 11m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/106/check/networking-midonet-tempest-aio-ml2-full-centos-7/e246873/ : FAILURE in 1h 46m 26s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-grenade-ml2/ce2f336/ : FAILURE in 33m 56s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/106/check/networking-midonet-rally-ml2/0e7df32/ : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":106},{"id":"a3f95787b5ed0bc3eedbae8f834e9ba3097f52bc","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-19 05:13:31.000000000","message":"Patch Set 107: Commit message was updated.","accounts_in_message":[],"_revision_number":107},{"id":"0a99d34d5dc77475b28e66859dc0e0e5d1547ac9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-19 06:45:42.000000000","message":"Patch Set 107: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/107/check/openstack-tox-pep8/f0dfbbe/ : SUCCESS in 7m 32s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/107/check/openstack-tox-py27/5b7d017/ : SUCCESS in 10m 37s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/107/check/openstack-tox-py35/123c863/ : SUCCESS in 17m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/107/check/build-openstack-sphinx-docs/4fc7a50/html/ : SUCCESS in 7m 22s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/107/check/build-openstack-releasenotes/df7a961/html/ : SUCCESS in 5m 57s\n- openstack-tox-cover http://logs.openstack.org/87/199387/107/check/openstack-tox-cover/001f439/cover/ : SUCCESS in 12m 08s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/107/check/networking-midonet-tempest-aio-ml2/607c4d5/ : FAILURE in 1h 16m 22s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/107/check/networking-midonet-tempest-multinode-ml2/fc091ac/ : SUCCESS in 1h 10m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/107/check/networking-midonet-tempest-aio-ml2-full/62669a6/ : SUCCESS in 1h 29m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/107/check/networking-midonet-tempest-aio-ml2-full-legacy/5bf5739/ : SUCCESS in 1h 24m 41s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/107/check/networking-midonet-tempest-aio-ml2-centos-7/c94eb3d/ : SUCCESS in 1h 10m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/107/check/networking-midonet-tempest-aio-ml2-full-centos-7/956dd24/ : SUCCESS in 1h 28m 10s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/107/check/networking-midonet-grenade-ml2/d0a99b5/ : SUCCESS in 1h 13m 30s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/107/check/networking-midonet-rally-ml2/1ad55ac/ : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":107},{"id":"84b4661949dba39c080733638f2b22d2cb66fd8e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-22 03:08:07.000000000","message":"Patch Set 107:\n\nrecheck","accounts_in_message":[],"_revision_number":107},{"id":"47101616ff829ca5a5a26f4550130bc00038f915","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-22 03:09:08.000000000","message":"Patch Set 108: Commit message was updated.","accounts_in_message":[],"_revision_number":108},{"id":"fc78cb289d1266fa4e408c4efbac7f05bb9c1d89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-22 05:31:14.000000000","message":"Patch Set 108: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/108/check/openstack-tox-pep8/f804a06/ : SUCCESS in 8m 53s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/108/check/openstack-tox-py27/d8bfdea/ : FAILURE in 14m 20s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/108/check/openstack-tox-py35/c060a6f/ : FAILURE in 14m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/108/check/build-openstack-sphinx-docs/950a8ea/html/ : SUCCESS in 6m 18s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/108/check/build-openstack-releasenotes/b61926e/html/ : SUCCESS in 4m 53s\n- openstack-tox-cover http://logs.openstack.org/87/199387/108/check/openstack-tox-cover/510c6a7/ : FAILURE in 13m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/108/check/networking-midonet-tempest-aio-ml2/e46b0b6/ : SUCCESS in 1h 04m 37s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/108/check/networking-midonet-tempest-multinode-ml2/5ada0d5/ : SUCCESS in 1h 01m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/108/check/networking-midonet-tempest-aio-ml2-full/ef470b6/ : POST_FAILURE in 2h 19m 13s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/108/check/networking-midonet-tempest-aio-ml2-full-legacy/b30c10b/ : FAILURE in 1h 43m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/108/check/networking-midonet-tempest-aio-ml2-centos-7/a44d962/ : SUCCESS in 1h 12m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/108/check/networking-midonet-tempest-aio-ml2-full-centos-7/6c199bb/ : FAILURE in 1h 52m 20s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/108/check/networking-midonet-grenade-ml2/d06f590/ : FAILURE in 1h 22m 13s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/108/check/networking-midonet-rally-ml2/d1d6586/ : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":108},{"id":"31e28b05ecd18ac5d7a6ed77d6191d1b889f3e9c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-23 08:45:25.000000000","message":"Patch Set 109: Commit message was updated.","accounts_in_message":[],"_revision_number":109},{"id":"25d4d371621fc8ba47fef0019949f299bbaadd8f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-23 09:16:05.000000000","message":"Patch Set 110: Commit message was updated.","accounts_in_message":[],"_revision_number":110},{"id":"edc2f76866f490932b2a9ab58b237532bf2b1cda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-23 11:14:02.000000000","message":"Patch Set 110: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/110/check/openstack-tox-pep8/203c02b/ : SUCCESS in 11m 11s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/110/check/openstack-tox-py27/3962328/ : SUCCESS in 12m 34s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/110/check/openstack-tox-py35/7a1fbe8/ : SUCCESS in 12m 51s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/110/check/build-openstack-sphinx-docs/7835bd8/html/ : SUCCESS in 6m 31s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/110/check/build-openstack-releasenotes/ac83571/html/ : SUCCESS in 5m 42s\n- openstack-tox-cover http://logs.openstack.org/87/199387/110/check/openstack-tox-cover/7517894/cover/ : SUCCESS in 15m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2/d1ed8c2/ : SUCCESS in 50m 25s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-multinode-ml2/30de9a2/ : SUCCESS in 59m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2-full/7f10ae8/ : SUCCESS in 1h 27m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2-full-legacy/b20cf4e/ : SUCCESS in 1h 51m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2-centos-7/1b9cbb8/ : SUCCESS in 1h 03m 37s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2-full-centos-7/91dd869/ : SUCCESS in 1h 49m 36s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/110/check/networking-midonet-grenade-ml2/dca3eaf/ : FAILURE in 1h 21m 18s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/110/check/networking-midonet-rally-ml2/f15d94b/ : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":110},{"id":"24b23b575790b6197ad82065d56e2550679d4441","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-23 12:24:14.000000000","message":"Patch Set 110:\n\nrecheck test_server_sequence_suspend_resume","accounts_in_message":[],"_revision_number":110},{"id":"acd1a547182c1910102b2b7b40ec826e7b6a5a18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-23 14:24:40.000000000","message":"Patch Set 110: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/110/check/openstack-tox-pep8/79033b9/ : SUCCESS in 9m 14s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/110/check/openstack-tox-py27/45e0455/ : SUCCESS in 16m 13s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/110/check/openstack-tox-py35/54579bd/ : SUCCESS in 14m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/110/check/build-openstack-sphinx-docs/b5b1874/html/ : SUCCESS in 6m 47s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/110/check/build-openstack-releasenotes/870bc77/html/ : SUCCESS in 5m 16s\n- openstack-tox-cover http://logs.openstack.org/87/199387/110/check/openstack-tox-cover/2bdb126/cover/ : SUCCESS in 15m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2/cf047d8/ : SUCCESS in 1h 02m 32s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-multinode-ml2/0b3594f/ : SUCCESS in 1h 04m 58s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2-full/cf472ad/ : SUCCESS in 1h 41m 48s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2-full-legacy/e3c2f11/ : SUCCESS in 1h 50m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2-centos-7/1a8dbb8/ : FAILURE in 56m 56s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/110/check/networking-midonet-tempest-aio-ml2-full-centos-7/18fcece/ : SUCCESS in 1h 53m 46s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/110/check/networking-midonet-grenade-ml2/309e12e/ : SUCCESS in 1h 07m 59s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/110/check/networking-midonet-rally-ml2/4ec4a15/ : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":110},{"id":"8106bbb8696502d5f68cb3124a8a432e85234bd5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-28 01:02:04.000000000","message":"Patch Set 111: Commit message was updated.","accounts_in_message":[],"_revision_number":111},{"id":"06578ac74a645027e169c3d9f8234ef52af907a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 02:47:29.000000000","message":"Patch Set 111: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/111/check/openstack-tox-pep8/ddb1068/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/111/check/openstack-tox-py27/0fba1c9/ : SUCCESS in 12m 26s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/111/check/openstack-tox-py35/ba1282e/ : SUCCESS in 13m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/111/check/build-openstack-sphinx-docs/6e829bf/html/ : SUCCESS in 6m 20s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/111/check/build-openstack-releasenotes/98c6079/html/ : SUCCESS in 4m 38s\n- openstack-tox-cover http://logs.openstack.org/87/199387/111/check/openstack-tox-cover/1093278/cover/ : SUCCESS in 14m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/111/check/networking-midonet-tempest-aio-ml2/e79d191/ : SUCCESS in 49m 51s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/111/check/networking-midonet-tempest-multinode-ml2/720e5f9/ : FAILURE in 37m 38s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/111/check/networking-midonet-tempest-aio-ml2-full/552f6c7/ : SUCCESS in 1h 26m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/111/check/networking-midonet-tempest-aio-ml2-full-legacy/da9bffd/ : SUCCESS in 1h 37m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/111/check/networking-midonet-tempest-aio-ml2-centos-7/993b581/ : SUCCESS in 54m 03s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/111/check/networking-midonet-tempest-aio-ml2-full-centos-7/567e1ae/ : SUCCESS in 1h 43m 28s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/111/check/networking-midonet-grenade-ml2/6f5d460/ : SUCCESS in 1h 10m 34s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/111/check/networking-midonet-rally-ml2/4918de9/ : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":111},{"id":"e20c4be93a3a446bcc9186324f9f0c1215d6ee35","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-03-28 12:28:09.000000000","message":"Patch Set 112: Commit message was updated.","accounts_in_message":[],"_revision_number":112},{"id":"fab0609dfb95aec53b59677fea181f1415933d7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-28 14:11:57.000000000","message":"Patch Set 112: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/112/check/openstack-tox-pep8/abc7d81/ : SUCCESS in 8m 09s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/112/check/openstack-tox-py27/8a1d91e/ : SUCCESS in 13m 43s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/112/check/openstack-tox-py35/98d7c49/ : SUCCESS in 13m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/112/check/build-openstack-sphinx-docs/894edb6/html/ : SUCCESS in 6m 12s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/112/check/build-openstack-releasenotes/27f0db5/html/ : SUCCESS in 4m 43s\n- openstack-tox-cover http://logs.openstack.org/87/199387/112/check/openstack-tox-cover/77219be/cover/ : SUCCESS in 13m 18s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2/7c4390b/ : SUCCESS in 49m 19s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-multinode-ml2/ad12cc0/ : FAILURE in 31m 49s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full/7d2c1c0/ : SUCCESS in 1h 37m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full-legacy/986b0fa/ : SUCCESS in 1h 32m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-centos-7/dce1af7/ : SUCCESS in 1h 01m 47s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full-centos-7/347dd32/ : SUCCESS in 1h 39m 21s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-grenade-ml2/446789d/ : SUCCESS in 1h 09m 44s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-rally-ml2/502849a/ : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":112},{"id":"7927eb37ececdcc19ac382923944aedfe5b41340","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-04-10 15:25:12.000000000","message":"Patch Set 112:\n\nrecheck","accounts_in_message":[],"_revision_number":112},{"id":"ced1ad386853795a3c6529f36646b06c9b15b678","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-10 17:48:25.000000000","message":"Patch Set 112: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/112/check/openstack-tox-pep8/577595f/ : FAILURE in 8m 32s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/112/check/openstack-tox-py27/8af7bcd/ : FAILURE in 12m 23s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/112/check/openstack-tox-py35/a7bec3a/ : FAILURE in 14m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/112/check/build-openstack-sphinx-docs/b14e193/html/ : SUCCESS in 5m 53s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/112/check/build-openstack-releasenotes/dfa2096/html/ : SUCCESS in 5m 17s\n- openstack-tox-cover http://logs.openstack.org/87/199387/112/check/openstack-tox-cover/4d1166c/ : FAILURE in 15m 18s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2/f6e7909/ : SUCCESS in 55m 46s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-multinode-ml2/9e68e7e/ : SUCCESS in 58m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full/1865f17/ : SUCCESS in 1h 21m 24s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full-legacy/f8634c1/ : SUCCESS in 2h 00m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-centos-7/ddc8526/ : SUCCESS in 1h 24m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full-centos-7/a71e5bd/ : FAILURE in 2h 05m 31s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-grenade-ml2/a1f119e/ : SUCCESS in 1h 05m 04s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-rally-ml2/97e97c5/ : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":112},{"id":"2b23f88b04a18957b99a87c702d65b44a36c5d50","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-04-24 14:23:01.000000000","message":"Patch Set 112:\n\nrecheck","accounts_in_message":[],"_revision_number":112},{"id":"2914b5e4bd9fbf46cafdfc3bd265a4e3fc977b96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-24 17:14:43.000000000","message":"Patch Set 112:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/112/check/openstack-tox-pep8/b2854b9/ : SUCCESS in 8m 29s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/112/check/openstack-tox-py27/2fccb4b/ : SUCCESS in 12m 16s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/112/check/openstack-tox-py35/c231c0c/ : SUCCESS in 12m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/112/check/build-openstack-sphinx-docs/8fb9133/ : RETRY_LIMIT in 6m 23s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/112/check/build-openstack-releasenotes/f4a7278/html/ : SUCCESS in 4m 38s\n- openstack-tox-cover http://logs.openstack.org/87/199387/112/check/openstack-tox-cover/5c49be1/cover/ : SUCCESS in 12m 52s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2/b303cc8/ : SUCCESS in 1h 04m 50s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-multinode-ml2/abb52ea/ : SUCCESS in 1h 17m 41s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full/aa50340/ : SUCCESS in 1h 44m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full-legacy/4e156fb/ : FAILURE in 2h 03m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-centos-7/4ed8df4/ : SUCCESS in 56m 53s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/112/check/networking-midonet-tempest-aio-ml2-full-centos-7/9c9cf3a/ : FAILURE in 2h 05m 30s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-grenade-ml2/3435ebf/ : FAILURE in 30m 27s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/112/check/networking-midonet-rally-ml2/f1fc2e1/ : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":112},{"id":"05974fb4f37f8e52f9fd119b8ab2e300e43f67c6","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-05-03 13:09:49.000000000","message":"Patch Set 112:\n\nI see the following error here and in a recent patch I have on this queue::\n\n  ERROR: You must give at least one requirement to install (see \"pip help install\")\n\n\nSeems similar to an issue I fixed here FYI: https://review.openstack.org/#/c/561615/4/tools/tox_install.sh","accounts_in_message":[],"_revision_number":112},{"id":"c4779e364ce1b5a844a3e637ffe5da13ae3fd3e5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-05-28 10:55:57.000000000","message":"Patch Set 113: Commit message was updated.","accounts_in_message":[],"_revision_number":113},{"id":"3f3438f236da07686ade0dd83da6a15573f74362","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-28 12:49:02.000000000","message":"Patch Set 113: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/113/check/openstack-tox-pep8/a544870/ : SUCCESS in 7m 40s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/113/check/openstack-tox-py27/60e32ea/ : SUCCESS in 13m 06s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/113/check/openstack-tox-py35/33134eb/ : SUCCESS in 11m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/113/check/build-openstack-sphinx-docs/d97a433/html/ : SUCCESS in 5m 31s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/113/check/build-openstack-releasenotes/71b1dad/html/ : SUCCESS in 4m 59s\n- openstack-tox-cover http://logs.openstack.org/87/199387/113/check/openstack-tox-cover/561ca3b/cover/ : SUCCESS in 15m 38s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2/7a60fc4/ : SUCCESS in 1h 08m 05s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-multinode-ml2/739bfac/ : SUCCESS in 1h 19m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2-full/f29c185/ : SUCCESS in 1h 35m 26s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2-full-legacy/e557e4d/ : FAILURE in 1h 37m 15s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2-centos-7/060b024/ : SUCCESS in 55m 59s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2-full-centos-7/39cff50/ : SUCCESS in 1h 51m 31s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/113/check/networking-midonet-grenade-ml2/d775faf/ : SUCCESS in 1h 22m 58s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/113/check/networking-midonet-rally-ml2/dcb7dd6/ : SUCCESS in 1h 04m 07s","accounts_in_message":[],"_revision_number":113},{"id":"7b0e0034632e6c6b748c0e379137bfca77cb4930","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-01 12:46:38.000000000","message":"Patch Set 113:\n\nrecheck","accounts_in_message":[],"_revision_number":113},{"id":"11f765333327c237fe86222cd11a3c24ac4956a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-01 14:46:11.000000000","message":"Patch Set 113:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/113/check/openstack-tox-pep8/701d635/ : SUCCESS in 8m 21s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/113/check/openstack-tox-py27/3b8bec9/ : SUCCESS in 11m 23s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/113/check/openstack-tox-py35/21520ce/ : SUCCESS in 11m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/113/check/build-openstack-sphinx-docs/b5a287e/html/ : SUCCESS in 6m 53s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/113/check/build-openstack-releasenotes/f9e0f4b/html/ : SUCCESS in 4m 59s\n- openstack-tox-cover http://logs.openstack.org/87/199387/113/check/openstack-tox-cover/d82dfb3/cover/ : SUCCESS in 13m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2/b3e63b8/ : SUCCESS in 1h 00m 13s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-multinode-ml2/27d9627/ : SUCCESS in 1h 08m 39s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2-full/efb2d8e/ : SUCCESS in 1h 53m 31s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2-full-legacy/7e1de2f/ : SUCCESS in 1h 39m 23s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2-centos-7/7b145e5/ : SUCCESS in 1h 08m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/113/check/networking-midonet-tempest-aio-ml2-full-centos-7/a71d165/ : SUCCESS in 1h 44m 04s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/113/check/networking-midonet-grenade-ml2/69d9d79/ : SUCCESS in 1h 14m 15s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/113/check/networking-midonet-rally-ml2/a1878ff/ : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":113},{"id":"c8f7f61316ce0d2143c26f8a431cee55f16cdf6d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-08 16:21:28.000000000","message":"Patch Set 114: Commit message was updated.","accounts_in_message":[],"_revision_number":114},{"id":"02e7b43fed4b1f1d1e6f32049ee61617d8101c13","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-09 08:35:21.000000000","message":"Patch Set 114:\n\nrecheck","accounts_in_message":[],"_revision_number":114},{"id":"5d341df5253c5fbf5743532b6babb80f41bb3f73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-09 10:46:10.000000000","message":"Patch Set 114: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/114/check/openstack-tox-pep8/a999ea4/ : SUCCESS in 8m 02s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/114/check/openstack-tox-py27/262cc0e/ : SUCCESS in 11m 56s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/114/check/openstack-tox-py35/e44f6be/ : SUCCESS in 12m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/114/check/build-openstack-sphinx-docs/06777f3/html/ : SUCCESS in 5m 54s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/114/check/build-openstack-releasenotes/3a352bf/html/ : SUCCESS in 5m 39s\n- openstack-tox-cover http://logs.openstack.org/87/199387/114/check/openstack-tox-cover/3c89f83/cover/ : SUCCESS in 15m 59s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2/fe6d439/ : SUCCESS in 1h 21m 24s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-multinode-ml2/76f9334/ : SUCCESS in 1h 02m 16s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full/3a2850a/ : FAILURE in 1h 51m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full-legacy/3ab3b80/ : FAILURE in 2h 03m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-centos-7/a098ea8/ : SUCCESS in 57m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full-centos-7/c22d799/ : FAILURE in 2h 04m 00s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-grenade-ml2/1cfad26/ : SUCCESS in 1h 05m 04s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-rally-ml2/a928320/ : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":114},{"id":"54f99a1b5f7d8b8020dc44549d0fbe51398b0654","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-10 09:39:20.000000000","message":"Patch Set 114:\n\nrecheck","accounts_in_message":[],"_revision_number":114},{"id":"50ca801ca0e0cf735d78951d62220da934ba18fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-10 11:44:23.000000000","message":"Patch Set 114:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/114/check/openstack-tox-pep8/db839d8/ : SUCCESS in 8m 24s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/114/check/openstack-tox-py27/b8a4c7d/ : SUCCESS in 11m 06s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/114/check/openstack-tox-py35/ffcc7bd/ : SUCCESS in 12m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/114/check/build-openstack-sphinx-docs/8eadc29/html/ : SUCCESS in 6m 01s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/114/check/build-openstack-releasenotes/5519ea4/html/ : SUCCESS in 4m 45s\n- openstack-tox-cover http://logs.openstack.org/87/199387/114/check/openstack-tox-cover/057d312/cover/ : SUCCESS in 13m 04s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2/9f881ca/ : SUCCESS in 59m 07s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-multinode-ml2/140cbb7/ : SUCCESS in 1h 06m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full/1fc0b8b/ : SUCCESS in 1h 31m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full-legacy/bcb8004/ : SUCCESS in 1h 55m 15s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-centos-7/65bce0d/ : SUCCESS in 57m 11s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full-centos-7/0567969/ : FAILURE in 2h 03m 18s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-grenade-ml2/dcb8262/ : SUCCESS in 1h 20m 22s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-rally-ml2/cd84968/ : SUCCESS in 1h 10m 47s","accounts_in_message":[],"_revision_number":114},{"id":"d790f0b0497b25c40cb42adc24686cdc06a56f17","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-19 06:14:06.000000000","message":"Patch Set 114:\n\nrecheck","accounts_in_message":[],"_revision_number":114},{"id":"5ee0d936f155ac4e00dbc03511f5f57705b9001c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-19 08:21:25.000000000","message":"Patch Set 114: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/114/check/openstack-tox-pep8/a6972f0/ : SUCCESS in 9m 19s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/114/check/openstack-tox-py27/4190a2d/ : SUCCESS in 13m 13s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/114/check/openstack-tox-py35/630408e/ : SUCCESS in 15m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/114/check/build-openstack-sphinx-docs/4deb4cb/ : FAILURE in 6m 23s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/114/check/build-openstack-releasenotes/58800ea/html/ : SUCCESS in 5m 25s\n- openstack-tox-cover http://logs.openstack.org/87/199387/114/check/openstack-tox-cover/927dd6a/cover/ : SUCCESS in 15m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2/574b0eb/ : SUCCESS in 1h 03m 35s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-multinode-ml2/05e5a12/ : SUCCESS in 1h 10m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full/d3f24a4/ : FAILURE in 1h 48m 08s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full-legacy/1e41410/ : SUCCESS in 1h 54m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-centos-7/8b06050/ : SUCCESS in 1h 10m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/114/check/networking-midonet-tempest-aio-ml2-full-centos-7/2b93f01/ : SUCCESS in 1h 53m 47s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-grenade-ml2/2c7ce4a/ : SUCCESS in 1h 27m 10s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/114/check/networking-midonet-rally-ml2/eea76f0/ : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":114},{"id":"6ecb120bc0ecb3468115f15552ab64a7fb51b129","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-21 08:20:52.000000000","message":"Patch Set 115: Commit message was updated.","accounts_in_message":[],"_revision_number":115},{"id":"e2e33bac005ff8b635036b73235b13d22b6aba0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-21 10:17:33.000000000","message":"Patch Set 115: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/115/check/openstack-tox-pep8/5638373/ : SUCCESS in 8m 52s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/115/check/openstack-tox-py27/5a5d2b3/ : SUCCESS in 13m 08s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/115/check/openstack-tox-py35/bffd268/ : SUCCESS in 12m 42s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/115/check/build-openstack-sphinx-docs/7c6e73e/ : FAILURE in 6m 20s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/115/check/build-openstack-releasenotes/b3950d6/html/ : SUCCESS in 6m 42s\n- openstack-tox-cover http://logs.openstack.org/87/199387/115/check/openstack-tox-cover/4a03016/cover/ : SUCCESS in 14m 22s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2/77c831c/ : SUCCESS in 54m 29s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-multinode-ml2/db22d67/ : SUCCESS in 1h 02m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full/df23d82/ : FAILURE in 1h 31m 12s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full-legacy/6de8bef/ : SUCCESS in 1h 49m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-centos-7/88dc3a6/ : SUCCESS in 1h 01m 53s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full-centos-7/60a0e38/ : SUCCESS in 1h 41m 50s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-grenade-ml2/3b4a210/ : SUCCESS in 1h 26m 00s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-rally-ml2/3e0b4aa/ : FAILURE in 25m 00s","accounts_in_message":[],"_revision_number":115},{"id":"856af709c6c7ae639a10db7d4906ec6f80fff74a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-25 07:59:16.000000000","message":"Patch Set 115:\n\nrecheck","accounts_in_message":[],"_revision_number":115},{"id":"006295b6d3108f86a3421107b824b1b41c8b38c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-25 10:06:34.000000000","message":"Patch Set 115:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/115/check/openstack-tox-pep8/0639cba/ : SUCCESS in 9m 38s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/115/check/openstack-tox-py27/d0902ee/ : SUCCESS in 12m 18s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/115/check/openstack-tox-py35/ef0575e/ : SUCCESS in 13m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/115/check/build-openstack-sphinx-docs/d203179/ : FAILURE in 5m 39s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/115/check/build-openstack-releasenotes/df30bfb/html/ : SUCCESS in 5m 33s\n- openstack-tox-cover http://logs.openstack.org/87/199387/115/check/openstack-tox-cover/eb4b25a/cover/ : SUCCESS in 14m 03s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2/0df4173/ : SUCCESS in 1h 02m 24s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-multinode-ml2/df7d700/ : SUCCESS in 1h 16m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full/ec487a6/ : FAILURE in 2h 04m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full-legacy/ed23dff/ : SUCCESS in 1h 39m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-centos-7/5bd48b7/ : SUCCESS in 1h 10m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full-centos-7/168b900/ : SUCCESS in 1h 43m 31s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-grenade-ml2/fd3b8ec/ : SUCCESS in 1h 10m 02s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-rally-ml2/46a503a/ : FAILURE in 26m 47s","accounts_in_message":[],"_revision_number":115},{"id":"2004d081024202874c84add059f2b497d122bcf5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-25 16:32:19.000000000","message":"Patch Set 115:\n\nrecheck","accounts_in_message":[],"_revision_number":115},{"id":"e556366ed839764fcd211b99a6a4df142acb3772","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-25 16:36:40.000000000","message":"Patch Set 115:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":115},{"id":"6cc04fe6a9b32b68ff81219f78c40f9e27cada5f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-25 16:43:30.000000000","message":"Patch Set 115:\n\nrecheck","accounts_in_message":[],"_revision_number":115},{"id":"9d34ea95d159781b5c610bde1a125de9c77a1dd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-25 18:54:03.000000000","message":"Patch Set 115:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/115/check/openstack-tox-pep8/0582086/ : SUCCESS in 9m 49s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/115/check/openstack-tox-py27/be28ccc/ : SUCCESS in 14m 37s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/115/check/openstack-tox-py35/ba1808b/ : SUCCESS in 13m 38s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/115/check/build-openstack-sphinx-docs/a65b9d0/ : FAILURE in 7m 16s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/115/check/build-openstack-releasenotes/566fbed/html/ : SUCCESS in 8m 11s\n- openstack-tox-cover http://logs.openstack.org/87/199387/115/check/openstack-tox-cover/3750bbf/cover/ : SUCCESS in 21m 56s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2/e78c88d/ : SUCCESS in 1h 00m 55s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-multinode-ml2/a499c48/ : SUCCESS in 1h 13m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full/9f3293d/ : FAILURE in 1h 22m 39s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full-legacy/e66a8e2/ : SUCCESS in 1h 50m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-centos-7/edbebff/ : SUCCESS in 1h 05m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full-centos-7/6457666/ : SUCCESS in 1h 49m 06s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-grenade-ml2/4681dad/ : SUCCESS in 1h 19m 13s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-rally-ml2/0f00baa/ : FAILURE in 28m 08s","accounts_in_message":[],"_revision_number":115},{"id":"4f1ba88a7ab16d5cc067b50505d61323f034161e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-06-26 16:30:29.000000000","message":"Patch Set 115:\n\nrecheck","accounts_in_message":[],"_revision_number":115},{"id":"78a767073ce2e47c3df8acc137fd1617082cd063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-26 19:28:42.000000000","message":"Patch Set 115:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/115/check/openstack-tox-pep8/baed5d9/ : SUCCESS in 9m 46s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/115/check/openstack-tox-py27/59ce4b4/ : SUCCESS in 13m 18s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/115/check/openstack-tox-py35/3db9336/ : SUCCESS in 15m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/115/check/build-openstack-sphinx-docs/189ea6a/ : FAILURE in 7m 20s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/115/check/build-openstack-releasenotes/be52cc0/html/ : SUCCESS in 7m 08s\n- openstack-tox-cover http://logs.openstack.org/87/199387/115/check/openstack-tox-cover/82a95fb/cover/ : SUCCESS in 14m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2/881eb18/ : SUCCESS in 1h 11m 49s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-multinode-ml2/399e331/ : FAILURE in 1h 14m 26s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full/d43d871/ : SUCCESS in 1h 58m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full-legacy/2807d9b/ : SUCCESS in 1h 39m 53s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-centos-7/6dd67d0/ : FAILURE in 1h 03m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/115/check/networking-midonet-tempest-aio-ml2-full-centos-7/82f31ef/ : SUCCESS in 1h 45m 30s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-grenade-ml2/2800ca3/ : SUCCESS in 1h 36m 48s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/115/check/networking-midonet-rally-ml2/e05febd/ : FAILURE in 32m 59s","accounts_in_message":[],"_revision_number":115},{"id":"2c0587ba41c79c056089647e7844de7d819d17a9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-02 07:12:40.000000000","message":"Patch Set 116: Commit message was updated.","accounts_in_message":[],"_revision_number":116},{"id":"3e0fee2377033fed185a26ecd186ac8d6df08112","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-02 09:32:33.000000000","message":"Patch Set 116: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/116/check/openstack-tox-pep8/ebcee75/ : SUCCESS in 8m 54s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/116/check/openstack-tox-py27/a172f60/ : SUCCESS in 13m 36s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/116/check/openstack-tox-py35/8e9b764/ : SUCCESS in 13m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/116/check/build-openstack-sphinx-docs/80af44e/ : FAILURE in 5m 59s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/116/check/build-openstack-releasenotes/2dac3ef/html/ : SUCCESS in 5m 34s\n- openstack-tox-cover http://logs.openstack.org/87/199387/116/check/openstack-tox-cover/b02355f/ : FAILURE in 9m 07s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2/02b36c9/ : FAILURE in 58m 12s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-multinode-ml2/55071c7/ : FAILURE in 2h 07m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2-full/e25aed6/ : SUCCESS in 1h 25m 58s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2-full-legacy/f910fb7/ : SUCCESS in 2h 02m 59s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2-centos-7/48056d6/ : SUCCESS in 1h 15m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2-full-centos-7/b9a3015/ : SUCCESS in 1h 44m 23s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/116/check/networking-midonet-grenade-ml2/7ae55a2/ : SUCCESS in 1h 12m 26s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/116/check/networking-midonet-rally-ml2/a8d654d/ : FAILURE in 31m 37s","accounts_in_message":[],"_revision_number":116},{"id":"f108b5c8fc00a5265a85b674e9049897fd3593e1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-03 00:52:45.000000000","message":"Patch Set 116:\n\nrecheck","accounts_in_message":[],"_revision_number":116},{"id":"4663a2ffe1edc45731bc7fc884722ffff6b845c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-03 02:52:02.000000000","message":"Patch Set 116:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/116/check/openstack-tox-pep8/3bfbda0/ : SUCCESS in 8m 52s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/116/check/openstack-tox-py27/d0b22bf/ : SUCCESS in 15m 19s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/116/check/openstack-tox-py35/dea9a92/ : SUCCESS in 15m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/116/check/build-openstack-sphinx-docs/a509328/ : FAILURE in 5m 38s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/116/check/build-openstack-releasenotes/5c30a60/html/ : SUCCESS in 5m 37s\n- openstack-tox-cover http://logs.openstack.org/87/199387/116/check/openstack-tox-cover/02105ca/ : FAILURE in 7m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2/35e7ea0/ : SUCCESS in 1h 05m 02s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-multinode-ml2/432958a/ : SUCCESS in 1h 22m 03s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2-full/09ea9b8/ : SUCCESS in 1h 52m 28s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2-full-legacy/fca1f95/ : SUCCESS in 1h 45m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2-centos-7/0d9d4fb/ : SUCCESS in 1h 20m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/116/check/networking-midonet-tempest-aio-ml2-full-centos-7/e1430d0/ : SUCCESS in 1h 53m 01s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/116/check/networking-midonet-grenade-ml2/7c241a8/ : SUCCESS in 1h 32m 35s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/116/check/networking-midonet-rally-ml2/bb23b4a/ : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":116},{"id":"781cb28234e51883a8b0f178aa059f00c55cfdfb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-10 09:58:05.000000000","message":"Patch Set 116:\n\nrecheck","accounts_in_message":[],"_revision_number":116},{"id":"deb587ebf699141e8388b44a092688f290af68e9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-10 10:21:43.000000000","message":"Patch Set 117: Commit message was updated.","accounts_in_message":[],"_revision_number":117},{"id":"2253ba4c899811962f371352ed2caa89e6c88a70","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-10 10:36:21.000000000","message":"Patch Set 118: Commit message was updated.","accounts_in_message":[],"_revision_number":118},{"id":"83f4e548da643fb591279b010889fe8ccd70d5e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 12:41:03.000000000","message":"Patch Set 118: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/118/check/openstack-tox-pep8/ab89fb5/ : SUCCESS in 8m 40s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/118/check/openstack-tox-py27/d676a73/ : SUCCESS in 12m 44s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/118/check/openstack-tox-py35/a640712/ : SUCCESS in 13m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/118/check/build-openstack-sphinx-docs/ab94f22/html/ : SUCCESS in 8m 57s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/118/check/build-openstack-releasenotes/d766b6a/html/ : SUCCESS in 6m 47s\n- openstack-tox-cover http://logs.openstack.org/87/199387/118/check/openstack-tox-cover/97eec14/ : FAILURE in 10m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2/57dd1b6/ : SUCCESS in 1h 01m 08s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-multinode-ml2/6a7919f/ : SUCCESS in 1h 07m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full/05e3dfa/ : FAILURE in 1h 58m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full-legacy/ccf8dfd/ : SUCCESS in 1h 45m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-centos-7/0651feb/ : SUCCESS in 52m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full-centos-7/4534c67/ : FAILURE in 30m 12s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-grenade-ml2/1cb5ee6/ : SUCCESS in 1h 47m 32s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-rally-ml2/e6c3a40/ : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":118},{"id":"452c7fce022c125f5dbdb1d705acab83457bceb1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-17 06:59:11.000000000","message":"Patch Set 118:\n\nrecheck","accounts_in_message":[],"_revision_number":118},{"id":"5fd66febab82f970b712094f6577b9dd54f8cad0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-17 08:59:46.000000000","message":"Patch Set 118:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/118/check/openstack-tox-pep8/ef28003/ : SUCCESS in 9m 02s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/118/check/openstack-tox-py27/089c9ac/ : SUCCESS in 13m 59s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/118/check/openstack-tox-py35/8e8ed41/ : SUCCESS in 12m 15s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/118/check/build-openstack-sphinx-docs/f5df6fe/html/ : SUCCESS in 7m 40s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/118/check/build-openstack-releasenotes/cad721d/html/ : SUCCESS in 6m 18s\n- openstack-tox-cover http://logs.openstack.org/87/199387/118/check/openstack-tox-cover/5875036/cover/ : SUCCESS in 13m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2/6d33bd4/ : SUCCESS in 52m 26s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-multinode-ml2/b95fdde/ : SUCCESS in 1h 29m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full/1c0bc63/ : SUCCESS in 1h 54m 49s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full-legacy/06fefdf/ : SUCCESS in 1h 52m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-centos-7/af5439c/ : SUCCESS in 59m 13s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full-centos-7/aeba367/ : SUCCESS in 1h 42m 15s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-grenade-ml2/8e91787/ : SUCCESS in 1h 30m 31s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-rally-ml2/4ecf2a4/ : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":118},{"id":"d60e05908520ed851770bc4e4685680c453f0bd0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-27 06:43:34.000000000","message":"Patch Set 118:\n\nrecheck","accounts_in_message":[],"_revision_number":118},{"id":"6ebe53ea3e9233a81b2403f2a2ef9a2615bc5bf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-27 08:25:43.000000000","message":"Patch Set 118: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/118/check/openstack-tox-pep8/f754318/ : SUCCESS in 10m 31s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/118/check/openstack-tox-py27/d96ffbb/ : FAILURE in 14m 50s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/118/check/openstack-tox-py35/9e3afff/ : FAILURE in 14m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/118/check/build-openstack-sphinx-docs/ef77806/html/ : SUCCESS in 9m 14s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/118/check/build-openstack-releasenotes/8f880a1/html/ : SUCCESS in 6m 34s\n- openstack-tox-cover http://logs.openstack.org/87/199387/118/check/openstack-tox-cover/572f594/ : FAILURE in 20m 41s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2/52eeb53/ : SUCCESS in 1h 22m 49s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-multinode-ml2/629ede9/ : FAILURE in 1h 21m 56s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full/9ddd823/ : FAILURE in 29m 53s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full-legacy/7fd0f70/ : FAILURE in 28m 16s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-centos-7/e01a474/ : SUCCESS in 1h 12m 50s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full-centos-7/69c5331/ : FAILURE in 50m 25s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-grenade-ml2/3c46814/ : FAILURE in 1h 24m 18s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-rally-ml2/de42500/ : FAILURE in 26m 29s","accounts_in_message":[],"_revision_number":118},{"id":"278c91cd1820bf8aaffc14200e6dee2f28712536","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-28 02:26:15.000000000","message":"Patch Set 118:\n\nrecheck after https://review.openstack.org/#/c/577549/ got merged","accounts_in_message":[],"_revision_number":118},{"id":"cc5761a8d2088c03c06d4627a5f4e0d88f1cf03a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-28 04:35:19.000000000","message":"Patch Set 118: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/118/check/openstack-tox-pep8/b784c07/ : SUCCESS in 9m 53s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/118/check/openstack-tox-py27/5827f34/ : SUCCESS in 14m 18s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/118/check/openstack-tox-py35/1771b56/ : SUCCESS in 12m 00s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/118/check/build-openstack-sphinx-docs/6caefa4/html/ : SUCCESS in 6m 40s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/118/check/build-openstack-releasenotes/263aed0/html/ : SUCCESS in 5m 34s\n- openstack-tox-cover http://logs.openstack.org/87/199387/118/check/openstack-tox-cover/b679f18/cover/ : SUCCESS in 12m 22s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2/c1a74bc/ : SUCCESS in 1h 08m 24s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-multinode-ml2/ee65a26/ : SUCCESS in 1h 29m 22s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full/0fa42af/ : FAILURE in 2h 03m 16s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full-legacy/b412491/ : FAILURE in 1h 54m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-centos-7/9783d15/ : SUCCESS in 1h 13m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/118/check/networking-midonet-tempest-aio-ml2-full-centos-7/37b93f8/ : SUCCESS in 1h 56m 49s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-grenade-ml2/c65c10e/ : SUCCESS in 1h 21m 49s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/118/check/networking-midonet-rally-ml2/e1867a4/ : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":118},{"id":"23f36595212f5cc701c23c95c2a194c1c6f64c89","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-30 06:13:12.000000000","message":"Patch Set 120: Commit message was updated.","accounts_in_message":[],"_revision_number":120},{"id":"bfbb7d9a2cf46687ad439ec16636902ff04953cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-30 08:05:26.000000000","message":"Patch Set 120: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/120/check/openstack-tox-pep8/39f1211/ : SUCCESS in 9m 25s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/120/check/openstack-tox-py27/6ea4a9d/ : SUCCESS in 13m 08s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/120/check/openstack-tox-py35/f0b4672/ : SUCCESS in 13m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/120/check/build-openstack-sphinx-docs/bea8d5c/html/ : SUCCESS in 6m 03s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/120/check/build-openstack-releasenotes/c12e908/html/ : SUCCESS in 5m 16s\n- openstack-tox-cover http://logs.openstack.org/87/199387/120/check/openstack-tox-cover/efc9cca/cover/ : SUCCESS in 15m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2/7e2721d/ : SUCCESS in 1h 06m 48s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-multinode-ml2/6f1d45e/ : SUCCESS in 1h 30m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2-full/20b8694/ : FAILURE in 1h 46m 15s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2-full-legacy/27e512c/ : FAILURE in 1h 43m 51s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2-centos-7/c145d5f/ : SUCCESS in 1h 02m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2-full-centos-7/dd872e6/ : FAILURE in 1h 43m 23s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/120/check/networking-midonet-grenade-ml2/04ff509/ : FAILURE in 1h 28m 49s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/120/check/networking-midonet-rally-ml2/5275d40/ : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":120},{"id":"6d34d45e9358fa3d34196fde01e93c2424059f6d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-07-30 09:08:40.000000000","message":"Patch Set 120:\n\nrecheck","accounts_in_message":[],"_revision_number":120},{"id":"988af145d50f273a51c773d6c97215a70fe7e9e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-30 11:20:13.000000000","message":"Patch Set 120:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/120/check/openstack-tox-pep8/f767efd/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/120/check/openstack-tox-py27/061ff9d/ : SUCCESS in 12m 16s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/120/check/openstack-tox-py35/7fbf4b0/ : SUCCESS in 13m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/120/check/build-openstack-sphinx-docs/0183c1c/html/ : SUCCESS in 5m 54s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/120/check/build-openstack-releasenotes/de066e8/html/ : SUCCESS in 5m 43s\n- openstack-tox-cover http://logs.openstack.org/87/199387/120/check/openstack-tox-cover/870ffed/cover/ : SUCCESS in 13m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2/8fa5a46/ : SUCCESS in 52m 44s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-multinode-ml2/290b13d/ : SUCCESS in 1h 02m 53s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2-full/7eb51e9/ : FAILURE in 2h 03m 44s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2-full-legacy/44dbfd3/ : FAILURE in 1h 59m 50s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2-centos-7/18a2dcc/ : SUCCESS in 1h 02m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/120/check/networking-midonet-tempest-aio-ml2-full-centos-7/90e1802/ : FAILURE in 2h 03m 14s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/120/check/networking-midonet-grenade-ml2/5f5bfc1/ : FAILURE in 1h 15m 02s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/120/check/networking-midonet-rally-ml2/97923d8/ : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":120},{"id":"ad1feb9b84bc1ff676808ec9184f4dcfce0a4127","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-08-27 00:37:48.000000000","message":"Patch Set 121: Commit message was updated.","accounts_in_message":[],"_revision_number":121},{"id":"b4d6ef9419b29a1d7f189411f96a8d6ad3142433","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-27 02:29:31.000000000","message":"Patch Set 121: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/121/check/openstack-tox-pep8/194fe0b/ : SUCCESS in 8m 19s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/121/check/openstack-tox-py27/9567b48/ : FAILURE in 7m 57s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/121/check/openstack-tox-py35/ce26249/ : FAILURE in 7m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/121/check/build-openstack-sphinx-docs/249f9fc/html/ : SUCCESS in 5m 49s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/121/check/build-openstack-releasenotes/4eff456/html/ : SUCCESS in 4m 47s\n- openstack-tox-cover http://logs.openstack.org/87/199387/121/check/openstack-tox-cover/fdbda88/ : FAILURE in 7m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2/c741cb6/ : FAILURE in 1h 11m 56s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-multinode-ml2/0b5946c/ : FAILURE in 1h 17m 09s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full/9396d5f/ : FAILURE in 1h 49m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full-legacy/35a2950/ : FAILURE in 1h 49m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-centos-7/af965e3/ : FAILURE in 53m 48s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full-centos-7/5215873/ : FAILURE in 20m 52s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-grenade-ml2/a762834/ : FAILURE in 1h 17m 06s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-rally-ml2/cbab006/ : SUCCESS in 1h 10m 53s","accounts_in_message":[],"_revision_number":121},{"id":"85a57f8c563f5cc9bee8169ab6ded34d6b550f12","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-08-28 09:27:38.000000000","message":"Patch Set 121:\n\nrecheck","accounts_in_message":[],"_revision_number":121},{"id":"db8a2b0ee478cebd72272eaa7b2abae4f5e6e71b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-28 11:59:21.000000000","message":"Patch Set 121:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/121/check/openstack-tox-pep8/77ef3b0/ : SUCCESS in 11m 01s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/121/check/openstack-tox-py27/adfc913/ : SUCCESS in 15m 22s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/121/check/openstack-tox-py35/429834b/ : SUCCESS in 14m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/121/check/build-openstack-sphinx-docs/4b93dcd/html/ : SUCCESS in 6m 57s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/121/check/build-openstack-releasenotes/9465a6d/html/ : SUCCESS in 7m 25s\n- openstack-tox-cover http://logs.openstack.org/87/199387/121/check/openstack-tox-cover/bfe18a5/cover/ : SUCCESS in 15m 40s (non-voting)\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2/3c3566e/ : FAILURE in 55m 13s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-multinode-ml2/309a2be/ : FAILURE in 1h 27m 13s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full/e5e3527/ : FAILURE in 1h 45m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full-legacy/9cb61a4/ : FAILURE in 2h 05m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-centos-7/18411f9/ : FAILURE in 1h 23m 13s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full-centos-7/cbe908b/ : FAILURE in 28m 47s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-grenade-ml2/af71188/ : FAILURE in 1h 17m 27s\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-rally-ml2/8ec96ce/ : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":121},{"id":"9f49b945b86138c48da02b52071f8acf26a95951","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-09-07 03:28:35.000000000","message":"Patch Set 121:\n\nrecheck","accounts_in_message":[],"_revision_number":121},{"id":"744712ae5228b6f298d19482dfb24ed71313d34a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-07 05:52:32.000000000","message":"Patch Set 121: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/121/check/openstack-tox-pep8/d3c180a/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/121/check/openstack-tox-py27/e304583/ : SUCCESS in 12m 08s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/121/check/openstack-tox-py35/7aa94b7/ : SUCCESS in 14m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/121/check/build-openstack-sphinx-docs/834206f/html/ : SUCCESS in 8m 07s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/121/check/build-openstack-releasenotes/f85ff54/html/ : SUCCESS in 6m 53s\n- openstack-tox-cover http://logs.openstack.org/87/199387/121/check/openstack-tox-cover/45dd319/cover/ : SUCCESS in 14m 43s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/87/199387/121/check/openstack-tox-py36/db6fd76/ : SUCCESS in 15m 09s\n- openstack-tox-docs http://logs.openstack.org/87/199387/121/check/openstack-tox-docs/852d084/html/ : SUCCESS in 7m 38s\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2/a16d086/ : SUCCESS in 1h 09m 44s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-multinode-ml2/7b65594/ : SUCCESS in 1h 20m 28s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full/54ab192/ : SUCCESS in 1h 47m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full-legacy/72d9ecf/ : FAILURE in 2h 05m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-centos-7/db1c431/ : FAILURE in 2h 08m 58s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full-centos-7/0999e1e/ : FAILURE in 28m 53s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-grenade-ml2/b94df1d/ : FAILURE in 1h 33m 38s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-rally-ml2/afcf262/ : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":121},{"id":"126ff313f3479b53249629bb993595c1675fde92","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-09-13 06:07:18.000000000","message":"Patch Set 121:\n\nrecheck","accounts_in_message":[],"_revision_number":121},{"id":"7c739758cd89dac6ee9739410016bcf931530a15","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-09-13 06:35:00.000000000","message":"Patch Set 121: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 7b650baafd691e523ffb46006e0d74b0653c6a74","accounts_in_message":[],"_revision_number":121},{"id":"27b4508c1f14001a6bd718ea3799e275620101cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-13 08:12:38.000000000","message":"Patch Set 121:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/121/check/openstack-tox-pep8/ef14e3b/ : SUCCESS in 9m 10s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/121/check/openstack-tox-py27/2fdfe0c/ : SUCCESS in 20m 40s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/121/check/openstack-tox-py35/4cd2ed6/ : SUCCESS in 14m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/121/check/build-openstack-sphinx-docs/6762062/html/ : SUCCESS in 7m 07s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/121/check/build-openstack-releasenotes/4298dae/html/ : SUCCESS in 6m 52s\n- openstack-tox-cover http://logs.openstack.org/87/199387/121/check/openstack-tox-cover/0af8202/cover/ : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/87/199387/121/check/openstack-tox-py36/53a4c50/ : SUCCESS in 12m 27s\n- openstack-tox-docs http://logs.openstack.org/87/199387/121/check/openstack-tox-docs/673a8cd/html/ : SUCCESS in 6m 54s\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2/ce0e386/ : SUCCESS in 1h 00m 47s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-multinode-ml2/597dfdd/ : SUCCESS in 1h 28m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full/6a68dce/ : FAILURE in 1h 36m 10s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full-legacy/26419eb/ : SUCCESS in 1h 53m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-centos-7/fc98ecf/ : SUCCESS in 58m 04s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/121/check/networking-midonet-tempest-aio-ml2-full-centos-7/c4c6950/ : FAILURE in 23m 40s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-grenade-ml2/f2d9a4d/ : FAILURE in 1h 56m 18s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/121/check/networking-midonet-rally-ml2/dbfd5de/ : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":121},{"id":"2420cc837a1a5acc413f7f82bee9874b8e306033","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-10-10 03:32:55.000000000","message":"Uploaded patch set 122.","accounts_in_message":[],"_revision_number":122},{"id":"f5652d012263615af35b3506cf71cd1780959018","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-10 05:41:05.000000000","message":"Patch Set 122: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/122/check/openstack-tox-pep8/f08b9f4/ : SUCCESS in 9m 28s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/122/check/openstack-tox-py27/9b1caa8/ : SUCCESS in 12m 49s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/122/check/openstack-tox-py35/e97bdb8/ : SUCCESS in 15m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/122/check/build-openstack-sphinx-docs/79a55ab/html/ : SUCCESS in 5m 40s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/122/check/build-openstack-releasenotes/643de89/html/ : SUCCESS in 6m 23s\n- test-release-openstack-python3 http://logs.openstack.org/87/199387/122/check/test-release-openstack-python3/a80c213/ : SUCCESS in 4m 39s\n- openstack-tox-cover http://logs.openstack.org/87/199387/122/check/openstack-tox-cover/3a852ad/cover/ : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/87/199387/122/check/openstack-tox-py36/9d73ce0/ : SUCCESS in 12m 45s\n- openstack-tox-docs http://logs.openstack.org/87/199387/122/check/openstack-tox-docs/9523331/html/ : SUCCESS in 6m 47s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/122/check/networking-midonet-tempest-multinode-ml2/38fcd39/ : SUCCESS in 1h 24m 25s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/122/check/networking-midonet-tempest-multinode-ml2-full/3214122/ : SUCCESS in 1h 59m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/122/check/networking-midonet-tempest-aio-ml2-full-legacy/8b521fa/ : SUCCESS in 1h 39m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/122/check/networking-midonet-tempest-aio-ml2-centos-7/aca243d/ : SUCCESS in 1h 13m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/122/check/networking-midonet-tempest-aio-ml2-full-centos-7/e67ed0d/ : FAILURE in 28m 47s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/122/check/networking-midonet-grenade-ml2/6a28673/ : FAILURE in 34m 18s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/122/check/networking-midonet-rally-ml2/e99409e/ : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":122},{"id":"95ee29bce4808cfeed44efdc1aedd02c97245ae1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-10-11 05:45:55.000000000","message":"Uploaded patch set 123.","accounts_in_message":[],"_revision_number":123},{"id":"5287539213c03be02ecd6e0359f0609f571463a2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-10-11 05:46:56.000000000","message":"Patch Set 124: Commit message was updated.","accounts_in_message":[],"_revision_number":124},{"id":"233a569488e766c17a7cd0369f077de759a0b012","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-11 08:10:45.000000000","message":"Patch Set 124: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/124/check/openstack-tox-pep8/9264b1f/ : SUCCESS in 9m 29s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/124/check/openstack-tox-py27/b36dd33/ : SUCCESS in 14m 29s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/124/check/openstack-tox-py35/7f27741/ : SUCCESS in 13m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/87/199387/124/check/build-openstack-sphinx-docs/6ec7d39/html/ : SUCCESS in 5m 22s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/124/check/build-openstack-releasenotes/58afd6f/html/ : SUCCESS in 6m 07s\n- test-release-openstack-python3 http://logs.openstack.org/87/199387/124/check/test-release-openstack-python3/c73542e/ : SUCCESS in 3m 18s\n- openstack-tox-cover http://logs.openstack.org/87/199387/124/check/openstack-tox-cover/4a3b040/cover/ : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py36 http://logs.openstack.org/87/199387/124/check/openstack-tox-py36/404eaad/ : SUCCESS in 14m 13s\n- openstack-tox-docs http://logs.openstack.org/87/199387/124/check/openstack-tox-docs/7214217/html/ : SUCCESS in 5m 55s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-multinode-ml2/c5f6fa7/ : FAILURE in 1h 32m 24s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-multinode-ml2-full/194f0a3/ : FAILURE in 2h 07m 03s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-full-legacy/fbd1072/ : FAILURE in 2h 05m 07s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-centos-7/32a504d/ : SUCCESS in 1h 10m 13s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-full-centos-7/249acc5/ : FAILURE in 19m 47s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-grenade-ml2/d3575fe/ : FAILURE in 39m 41s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-rally-ml2/8551427/ : FAILURE in 1h 17m 05s","accounts_in_message":[],"_revision_number":124},{"id":"03bece78c58118f4265e8ac1b54728414a6f8aa9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-10-22 13:18:13.000000000","message":"Patch Set 124:\n\nrecheck","accounts_in_message":[],"_revision_number":124},{"id":"9742a20aea5baa3d8555cb2e51a754c4b12990e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-22 15:31:36.000000000","message":"Patch Set 124: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack-python3 http://logs.openstack.org/87/199387/124/check/test-release-openstack-python3/3a83711/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/124/check/openstack-tox-pep8/b97ae0b/ : SUCCESS in 8m 57s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/124/check/openstack-tox-py27/511d28f/ : SUCCESS in 14m 52s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/124/check/openstack-tox-py35/6e689ce/ : SUCCESS in 13m 34s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/124/check/openstack-tox-py36/74d3857/ : SUCCESS in 12m 54s\n- openstack-tox-docs http://logs.openstack.org/87/199387/124/check/openstack-tox-docs/a30979f/html/ : SUCCESS in 5m 46s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/124/check/build-openstack-releasenotes/121d9e4/html/ : SUCCESS in 5m 22s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-multinode-ml2/7d6e5e4/ : SUCCESS in 1h 28m 19s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-multinode-ml2-full/432a63a/ : FAILURE in 2h 04m 50s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-full-legacy/039da28/ : FAILURE in 1h 55m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-centos-7/983d656/ : FAILURE in 1h 32m 22s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-full-centos-7/ef4fe1a/ : FAILURE in 21m 28s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-grenade-ml2/549b03e/ : FAILURE in 37m 18s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-rally-ml2/8652ccb/ : SUCCESS in 1h 22m 32s\n- openstack-tox-cover http://logs.openstack.org/87/199387/124/check/openstack-tox-cover/e67d1d0/cover/ : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":124},{"id":"68ab9b45454b784bb20fd4568319dc3c499f4cdb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-10-23 06:44:27.000000000","message":"Patch Set 124:\n\nrecheck","accounts_in_message":[],"_revision_number":124},{"id":"ac1a68a856eda1544703d4a3a024c044f88fc668","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-23 09:09:16.000000000","message":"Patch Set 124: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack-python3 http://logs.openstack.org/87/199387/124/check/test-release-openstack-python3/8b2c7ce/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/124/check/openstack-tox-pep8/07971ad/ : SUCCESS in 9m 57s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/124/check/openstack-tox-py27/7e20f42/ : SUCCESS in 12m 22s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/124/check/openstack-tox-py35/cc9badb/ : SUCCESS in 13m 31s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/124/check/openstack-tox-py36/65998cc/ : SUCCESS in 14m 12s\n- openstack-tox-docs http://logs.openstack.org/87/199387/124/check/openstack-tox-docs/6685c8a/html/ : SUCCESS in 6m 54s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/124/check/build-openstack-releasenotes/6f31921/html/ : SUCCESS in 6m 01s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-multinode-ml2/6153a17/ : FAILURE in 1h 16m 41s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-multinode-ml2-full/451d025/ : FAILURE in 2h 06m 18s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-full-legacy/c5252c8/ : FAILURE in 2h 07m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-centos-7/45993e1/ : FAILURE in 1h 08m 38s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/124/check/networking-midonet-tempest-aio-ml2-full-centos-7/7d5878c/ : FAILURE in 19m 38s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-grenade-ml2/bb85d79/ : FAILURE in 44m 18s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/124/check/networking-midonet-rally-ml2/ff0a9a2/ : SUCCESS in 1h 27m 01s\n- openstack-tox-cover http://logs.openstack.org/87/199387/124/check/openstack-tox-cover/2049c64/cover/ : SUCCESS in 15m 01s","accounts_in_message":[],"_revision_number":124},{"id":"b827db38ca1e90ff6f5d1a01b4aed16b2929f5a8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-10-24 03:58:31.000000000","message":"Patch Set 124:\n\nrecheck","accounts_in_message":[],"_revision_number":124},{"id":"b3620e4fe72a4bc99c196d65a5425411ef39750a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 04:15:02.000000000","message":"Patch Set 124:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":124},{"id":"f184aee27088518f93bde5474cda4dc38abda15d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-05 05:14:44.000000000","message":"Patch Set 125: Commit message was updated.","accounts_in_message":[],"_revision_number":125},{"id":"83f385c423ac044637c683cad6cae69053da6760","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 07:16:09.000000000","message":"Patch Set 125: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/125/check/test-release-openstack/58aa90b/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/125/check/openstack-tox-pep8/59a30f5/ : SUCCESS in 9m 03s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/125/check/openstack-tox-py27/4a583e7/ : FAILURE in 15m 26s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/125/check/openstack-tox-py35/227dd6b/ : FAILURE in 11m 18s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/125/check/openstack-tox-py36/1e7cc86/ : FAILURE in 11m 49s\n- openstack-tox-docs http://logs.openstack.org/87/199387/125/check/openstack-tox-docs/f003d45/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/125/check/build-openstack-releasenotes/accceed/html/ : SUCCESS in 5m 09s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-multinode-ml2/75fa0fd/ : SUCCESS in 1h 15m 15s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-multinode-ml2-full/23d11ce/ : FAILURE in 1h 40m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-full-legacy/d6e5045/ : FAILURE in 1h 59m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-centos-7/e3131ce/ : SUCCESS in 1h 19m 37s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-full-centos-7/133fa3e/ : FAILURE in 29m 42s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-grenade-ml2/8a5e890/ : FAILURE in 29m 21s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-rally-ml2/90899e4/ : SUCCESS in 1h 05m 38s\n- openstack-tox-cover http://logs.openstack.org/87/199387/125/check/openstack-tox-cover/cfd03d5/ : FAILURE in 13m 22s","accounts_in_message":[],"_revision_number":125},{"id":"ca14aad24371dffada0a0098a722790cd6a6eb39","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-05 13:52:05.000000000","message":"Patch Set 125:\n\nrecheck after https://review.openstack.org/#/c/597563/","accounts_in_message":[],"_revision_number":125},{"id":"4df0f06e8ce6caa50092cec5db02a5819fc9336e","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-11-05 15:41:39.000000000","message":"Patch Set 125:\n\nrecheck https://review.openstack.org/#/c/597563/ has landed","accounts_in_message":[],"_revision_number":125},{"id":"3cb1e71656874979683f490ec6e66d02f81fac67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 16:01:26.000000000","message":"Patch Set 125:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/125/check/test-release-openstack/87b283e/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/125/check/openstack-tox-pep8/d97ec99/ : SUCCESS in 9m 15s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/125/check/openstack-tox-py27/403b980/ : POST_FAILURE in 12m 32s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/125/check/openstack-tox-py35/aee407e/ : SUCCESS in 12m 30s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/125/check/openstack-tox-py36/0fda69e/ : SUCCESS in 12m 42s\n- openstack-tox-docs http://logs.openstack.org/87/199387/125/check/openstack-tox-docs/ea86e04/html/ : SUCCESS in 5m 09s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/125/check/build-openstack-releasenotes/9ceb9f2/html/ : SUCCESS in 5m 56s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-multinode-ml2/3a78beb/ : SUCCESS in 1h 05m 56s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-multinode-ml2-full/fbdfc18/ : FAILURE in 2h 02m 15s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-full-legacy/f3c0d34/ : FAILURE in 2h 00m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-centos-7/cd4e4b1/ : SUCCESS in 1h 16m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-full-centos-7/000ac77/ : FAILURE in 22m 55s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-grenade-ml2/980c1f2/ : FAILURE in 31m 04s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-rally-ml2/e9e1173/ : SUCCESS in 1h 19m 25s\n- openstack-tox-cover http://logs.openstack.org/87/199387/125/check/openstack-tox-cover/48a0583/cover/ : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":125},{"id":"2fb9aed138c147143ce31114d97b0b7e5186dbf8","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-07 12:20:47.000000000","message":"Patch Set 125:\n\nrecheck","accounts_in_message":[],"_revision_number":125},{"id":"6a35b611421041284241a51ae96d1afec40a8db1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-07 14:24:07.000000000","message":"Patch Set 125: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/125/check/test-release-openstack/7e3bc4c/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/125/check/openstack-tox-pep8/f29c6e7/ : SUCCESS in 8m 42s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/125/check/openstack-tox-py27/4ec97c4/ : SUCCESS in 12m 52s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/125/check/openstack-tox-py35/9e5b0f0/ : SUCCESS in 12m 40s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/125/check/openstack-tox-py36/df2d6cc/ : SUCCESS in 24m 12s\n- openstack-tox-docs http://logs.openstack.org/87/199387/125/check/openstack-tox-docs/a370773/html/ : SUCCESS in 4m 55s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/125/check/build-openstack-releasenotes/37cbc3b/html/ : SUCCESS in 5m 34s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-multinode-ml2/20c1e23/ : SUCCESS in 1h 09m 40s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-multinode-ml2-full/4e934d4/ : FAILURE in 2h 01m 08s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-full-legacy/bef89bc/ : FAILURE in 1h 53m 10s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-centos-7/a0abcee/ : SUCCESS in 1h 16m 51s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/125/check/networking-midonet-tempest-aio-ml2-full-centos-7/7020d19/ : FAILURE in 24m 46s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-grenade-ml2/be6433a/ : FAILURE in 1h 11m 01s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/125/check/networking-midonet-rally-ml2/5ce3996/ : SUCCESS in 1h 07m 06s\n- openstack-tox-cover http://logs.openstack.org/87/199387/125/check/openstack-tox-cover/23325d3/cover/ : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":125},{"id":"8229f209a047f397a86da21f361c8cb04f9d5f23","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-09 03:36:18.000000000","message":"Patch Set 126: Commit message was updated.","accounts_in_message":[],"_revision_number":126},{"id":"0e1d8d1f5af4303221c9fcf2af607be735cd6a50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 05:37:12.000000000","message":"Patch Set 126: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/126/check/test-release-openstack/6047d90/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/126/check/openstack-tox-pep8/0fd5ed0/ : SUCCESS in 8m 59s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/126/check/openstack-tox-py27/57fe3fb/ : SUCCESS in 12m 25s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/126/check/openstack-tox-py35/5fe1f0d/ : SUCCESS in 12m 23s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/126/check/openstack-tox-py36/d226511/ : SUCCESS in 11m 59s\n- openstack-tox-docs http://logs.openstack.org/87/199387/126/check/openstack-tox-docs/46d43be/html/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/126/check/build-openstack-releasenotes/4a3aa70/html/ : SUCCESS in 5m 48s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/126/check/networking-midonet-tempest-multinode-ml2/42e1177/ : SUCCESS in 1h 26m 10s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/126/check/networking-midonet-tempest-multinode-ml2-full/db908a5/ : FAILURE in 1h 52m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/126/check/networking-midonet-tempest-aio-ml2-full-legacy/5f47fc4/ : FAILURE in 1h 58m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/126/check/networking-midonet-tempest-aio-ml2-centos-7/80ee62d/ : SUCCESS in 1h 09m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/126/check/networking-midonet-tempest-aio-ml2-full-centos-7/3c618e2/ : FAILURE in 1h 43m 53s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/126/check/networking-midonet-grenade-ml2/fab961c/ : FAILURE in 43m 09s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/126/check/networking-midonet-rally-ml2/b175cd0/ : SUCCESS in 1h 16m 09s\n- openstack-tox-cover http://logs.openstack.org/87/199387/126/check/openstack-tox-cover/455c41e/cover/ : SUCCESS in 20m 19s","accounts_in_message":[],"_revision_number":126},{"id":"0635aa70041b4b6a0f3673bafe5858e40e7af92b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-09 07:05:15.000000000","message":"Patch Set 127: Commit message was updated.","accounts_in_message":[],"_revision_number":127},{"id":"0cd7ef57d2b7f21feb98e80c0bd628f6e6bf7b98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 08:56:56.000000000","message":"Patch Set 127: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/efe56db/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/5028823/ : SUCCESS in 9m 20s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/6e422f2/ : SUCCESS in 13m 00s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/a1c348a/ : SUCCESS in 26m 19s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/030d0fa/ : SUCCESS in 12m 36s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/a64dbfa/html/ : SUCCESS in 6m 07s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/5884f4a/html/ : SUCCESS in 5m 19s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/80d48f3/ : SUCCESS in 1h 14m 40s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/67fbe16/ : SUCCESS in 1h 49m 19s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/acbf076/ : SUCCESS in 1h 48m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/340038d/ : SUCCESS in 1h 13m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/a3d4205/ : FAILURE in 16m 55s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/3b39868/ : FAILURE in 35m 55s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/620eef7/ : SUCCESS in 1h 17m 10s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/fab30fd/cover/ : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":127},{"id":"a3f0c5653b437b481c6a0093e6082e30087ff230","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-09 10:47:05.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"11f797ad2c1da056891c4bae03e8f444f100d234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-09 12:58:54.000000000","message":"Patch Set 127:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/430c6cf/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/1c9e708/ : SUCCESS in 9m 00s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/57489d2/ : SUCCESS in 13m 12s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/87d700b/ : SUCCESS in 13m 00s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/1761ab5/ : SUCCESS in 18m 45s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/747c6da/html/ : SUCCESS in 5m 25s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/5493ad4/html/ : SUCCESS in 5m 46s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/ee1bfc0/ : SUCCESS in 1h 08m 34s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/2a7202f/ : FAILURE in 2h 04m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/a11ce4c/ : SUCCESS in 2h 02m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/d238f15/ : SUCCESS in 1h 13m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/4cf23e2/ : FAILURE in 19m 51s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/de9fc62/ : FAILURE in 33m 00s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/17fb79b/ : SUCCESS in 1h 19m 04s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/e054285/cover/ : SUCCESS in 20m 23s","accounts_in_message":[],"_revision_number":127},{"id":"8ec6ec2471540f247c032dc5cdb3378645f981b2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-12 17:26:15.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"42c118281e881b80e6f198bb2cc994a29473f760","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-12 19:39:02.000000000","message":"Patch Set 127:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/72ca995/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/22c11b9/ : SUCCESS in 9m 58s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/ac5d289/ : SUCCESS in 11m 49s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/7be89c4/ : SUCCESS in 11m 52s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/1689f1a/ : SUCCESS in 13m 47s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/7ad455e/html/ : SUCCESS in 6m 55s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/04db525/html/ : SUCCESS in 6m 04s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/11463d9/ : SUCCESS in 1h 12m 08s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/bbb4aea/ : FAILURE in 2h 10m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/68d7e6d/ : FAILURE in 2h 04m 08s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/9b1e613/ : SUCCESS in 1h 09m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/d387dd7/ : FAILURE in 30m 28s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/5f4b0ba/ : FAILURE in 51m 18s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/df4c8ed/ : SUCCESS in 1h 18m 49s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/7d0d6a3/cover/ : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":127},{"id":"06febc2af971454931be102748c587f86a4cf46a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-13 07:48:42.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"5f43463327e9a9242510426f7006c44aa7960c37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-13 09:59:18.000000000","message":"Patch Set 127:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/0995985/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/47b4749/ : SUCCESS in 9m 13s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/56b0aa2/ : SUCCESS in 13m 18s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/6979f44/ : SUCCESS in 16m 21s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/cb85908/ : SUCCESS in 11m 55s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/7eaf508/html/ : SUCCESS in 5m 32s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/f2d9710/html/ : SUCCESS in 5m 27s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/ad34491/ : SUCCESS in 1h 33m 02s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/f5fd451/ : FAILURE in 2h 08m 12s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/202636c/ : SUCCESS in 1h 50m 31s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/1e3c925/ : SUCCESS in 1h 00m 13s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/129e046/ : FAILURE in 19m 58s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/b14979e/ : FAILURE in 33m 55s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/eacda14/ : SUCCESS in 1h 15m 32s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/4d33352/cover/ : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":127},{"id":"532d2e9cd764cd696927b90066ff0f1fc651b5ca","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2018-11-20 13:53:17.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"c3787a9b8a6be0a5bb1f0f07ee173c831f1336a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-20 16:05:38.000000000","message":"Patch Set 127: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/3564086/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/2023882/ : SUCCESS in 9m 43s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/e22b2f1/ : FAILURE in 25m 33s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/98ef46f/ : FAILURE in 12m 54s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/03119b4/ : FAILURE in 26m 33s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/937fb10/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/dbf1018/html/ : SUCCESS in 5m 55s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/63e4499/ : FAILURE in 1h 48m 38s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/c621e20/ : FAILURE in 1h 55m 18s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/a245a51/ : SUCCESS in 1h 38m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/c47b5ed/ : SUCCESS in 1h 10m 12s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/1564b06/ : FAILURE in 2h 09m 07s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/ba94fe6/ : FAILURE in 34m 32s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/9620dd8/ : SUCCESS in 1h 07m 56s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/d58e931/ : FAILURE in 16m 14s","accounts_in_message":[],"_revision_number":127},{"id":"e9bf0a9b53c2dc709dcc49609cb5e52ba3c5a7d6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-01-22 23:17:41.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"db754673d807487a8a4db5853bd732504db383c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-23 01:29:06.000000000","message":"Patch Set 127: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/9ea8f7c/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/9cb0750/ : SUCCESS in 8m 34s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/b56fc56/ : SUCCESS in 11m 24s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/45977c5/ : SUCCESS in 14m 00s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/e2ae110/ : SUCCESS in 11m 56s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/567d54a/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/053d401/html/ : SUCCESS in 5m 25s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/c17e014/ : SUCCESS in 1h 22m 53s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/d5fc326/ : FAILURE in 2h 04m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/5f3d0ab/ : FAILURE in 2h 04m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/3c01c0c/ : SUCCESS in 1h 21m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/ff51795/ : FAILURE in 2h 05m 24s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/f183ee1/ : FAILURE in 41m 36s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/3459666/ : SUCCESS in 1h 28m 32s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/b506599/cover/ : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":127},{"id":"c160133f44e5a88f75a2f32a6fe55183ab65cf8c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-01-31 15:18:24.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"951aa24f0431c32d3d071280e3a078069da48753","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 17:35:37.000000000","message":"Patch Set 127: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/faff45c/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/efe3043/ : SUCCESS in 10m 44s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/44d256f/ : SUCCESS in 19m 17s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/a5b288c/ : SUCCESS in 18m 05s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/c935be7/ : SUCCESS in 14m 34s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/0857361/html/ : SUCCESS in 5m 55s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/ee4399b/html/ : SUCCESS in 7m 41s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/d261f2a/ : FAILURE in 30m 45s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/0c8a926/ : FAILURE in 2h 06m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/1414350/ : FAILURE in 2h 06m 03s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/265f0ca/ : FAILURE in 27m 11s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/8755018/ : FAILURE in 2h 06m 22s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/34a780b/ : FAILURE in 37m 55s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/2dee715/ : FAILURE in 25m 04s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/0819fac/cover/ : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":127},{"id":"18abd943ae1ec0b23afd7c90b7608f6064198c17","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-02-04 06:16:34.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"56ef5ec17443be5deb67400fc4ba438250bc79d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 06:56:42.000000000","message":"Patch Set 127:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/6fe6ed6/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/6ca4c35/ : SUCCESS in 9m 45s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/894dfff/ : SUCCESS in 14m 11s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/7b7beb0/ : SUCCESS in 17m 18s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/daa330b/ : SUCCESS in 12m 51s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/5e4332e/html/ : SUCCESS in 5m 40s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/9dcf5a4/html/ : SUCCESS in 6m 04s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/15a84e2/ : FAILURE in 22m 43s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/e256e04/ : FAILURE in 27m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/c3d1359/ : FAILURE in 29m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/4283481/ : FAILURE in 25m 04s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/58c1aec/ : FAILURE in 33m 03s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/de35f0f/ : FAILURE in 35m 50s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/e2ccee8/ : FAILURE in 21m 14s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/437c986/cover/ : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":127},{"id":"0412f026a13bf6c9045d0e196d5eab68fd875860","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-02-26 12:18:15.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"e0ed7e38346f541720b37e3ddaf9e06f93f9cec3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 14:02:30.000000000","message":"Patch Set 127:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/58d4b57/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/912f1d3/ : SUCCESS in 10m 00s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/90850de/ : SUCCESS in 14m 54s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/8091e05/ : SUCCESS in 13m 06s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/7068f7e/ : SUCCESS in 13m 38s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/2f4ffc9/html/ : SUCCESS in 5m 48s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/44f246c/html/ : SUCCESS in 5m 56s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/84487ea/ : FAILURE in 34m 31s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/3a5f2c2/ : FAILURE in 30m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/3d92624/ : FAILURE in 28m 28s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/860a2c6/ : FAILURE in 11m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/3b54510/ : FAILURE in 42m 43s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/be9dacf/ : FAILURE in 38m 31s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/c61ebc6/ : FAILURE in 1h 36m 05s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/9e3bd3e/cover/ : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":127},{"id":"dd05d198d95294b43cb8f2298a2efd2a55a17de6","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-27 16:21:34.000000000","message":"Patch Set 127:\n\nThis looks like the issue: http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/84487ea/logs/devstacklog.txt.gz#_2019-02-26_13_01_01_812\n\n\nBut not clear to me how to resolve.","accounts_in_message":[],"_revision_number":127},{"id":"bf698d4e940464ce13ae0c2665b1ffc82ad88219","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-02-28 00:53:05.000000000","message":"Patch Set 127:\n\nboden,\n\ni filed a bug https://bugs.launchpad.net/networking-midonet/+bug/1817973","accounts_in_message":[],"_revision_number":127},{"id":"c3d77fbb44d1f93acc3766cfdaa4e9f89ddcf1a6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-03-02 08:56:50.000000000","message":"Patch Set 127:\n\nrecheck","accounts_in_message":[],"_revision_number":127},{"id":"dabe3eead919f079e5b21b34bb2dee5dce7a8cf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-02 10:47:01.000000000","message":"Patch Set 127:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/127/check/test-release-openstack/9f45c2e/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/127/check/openstack-tox-pep8/340f69c/ : FAILURE in 7m 14s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/127/check/openstack-tox-py27/c01863e/ : SUCCESS in 10m 24s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/127/check/openstack-tox-py35/17d818a/ : SUCCESS in 17m 52s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/127/check/openstack-tox-py36/3db3d3e/ : SUCCESS in 10m 14s\n- openstack-tox-docs http://logs.openstack.org/87/199387/127/check/openstack-tox-docs/52721fa/html/ : SUCCESS in 4m 59s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/127/check/build-openstack-releasenotes/f743fd3/html/ : SUCCESS in 5m 08s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2/0119e8a/ : SUCCESS in 1h 28m 19s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-multinode-ml2-full/24b658f/ : SUCCESS in 1h 35m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-legacy/3430c11/ : SUCCESS in 1h 18m 09s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-centos-7/e407dfa/ : SUCCESS in 1h 31m 47s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/127/check/networking-midonet-tempest-aio-ml2-full-centos-7/54ebbe1/ : FAILURE in 27m 24s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-grenade-ml2/c7c7aae/ : FAILURE in 32m 15s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/127/check/networking-midonet-rally-ml2/7eb9501/ : SUCCESS in 1h 46m 38s\n- openstack-tox-cover http://logs.openstack.org/87/199387/127/check/openstack-tox-cover/44d8b35/cover/ : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":127},{"id":"8beada94638193827a5b899740de7b3eb465fda6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-03-05 11:07:35.000000000","message":"Patch Set 128: Commit message was updated.","accounts_in_message":[],"_revision_number":128},{"id":"a0ed54bd0e47efb4964d02cba8cfb1e64472ebf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 13:11:19.000000000","message":"Patch Set 128: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/128/check/test-release-openstack/6a0520a/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/128/check/openstack-tox-pep8/1d07196/ : SUCCESS in 9m 01s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/128/check/openstack-tox-py27/3e8534c/ : SUCCESS in 12m 35s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/128/check/openstack-tox-py35/1649071/ : SUCCESS in 12m 42s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/128/check/openstack-tox-py36/44dc55f/ : SUCCESS in 12m 50s\n- openstack-tox-docs http://logs.openstack.org/87/199387/128/check/openstack-tox-docs/127c556/html/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/128/check/build-openstack-releasenotes/33565a7/html/ : SUCCESS in 6m 15s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/128/check/networking-midonet-tempest-multinode-ml2/ae357d6/ : SUCCESS in 1h 49m 14s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/128/check/networking-midonet-tempest-multinode-ml2-full/a2fea99/ : SUCCESS in 1h 45m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/128/check/networking-midonet-tempest-aio-ml2-full-legacy/e1c6528/ : SUCCESS in 1h 36m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/128/check/networking-midonet-tempest-aio-ml2-centos-7/75593ab/ : SUCCESS in 1h 23m 56s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/128/check/networking-midonet-tempest-aio-ml2-full-centos-7/c405c92/ : SUCCESS in 1h 57m 22s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/128/check/networking-midonet-grenade-ml2/4c98dcd/ : FAILURE in 38m 22s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/128/check/networking-midonet-rally-ml2/87fae6f/ : FAILURE in 1h 34m 26s\n- openstack-tox-cover http://logs.openstack.org/87/199387/128/check/openstack-tox-cover/4764fa1/cover/ : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":128},{"id":"1944487fb470b436939701f3ff6c866a0c95ab48","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-03-13 02:48:57.000000000","message":"Patch Set 129: Commit message was updated.","accounts_in_message":[],"_revision_number":129},{"id":"2c4fa7156fced30abf6fbe06613d44bf1ce9667b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 03:36:45.000000000","message":"Patch Set 129: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/129/check/test-release-openstack/ba69827/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/129/check/openstack-tox-pep8/3386b7f/ : SUCCESS in 9m 22s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/129/check/openstack-tox-py27/6a46047/ : FAILURE in 9m 00s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/129/check/openstack-tox-py35/98528a5/ : FAILURE in 9m 31s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/129/check/openstack-tox-py36/dae7888/ : FAILURE in 8m 55s\n- openstack-tox-docs http://logs.openstack.org/87/199387/129/check/openstack-tox-docs/5d5ee29/html/ : SUCCESS in 6m 24s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/129/check/build-openstack-releasenotes/9f9d099/html/ : SUCCESS in 6m 06s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/129/check/networking-midonet-tempest-multinode-ml2/43672b7/ : FAILURE in 24m 15s\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/129/check/networking-midonet-tempest-multinode-ml2-full/5218779/ : FAILURE in 31m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/129/check/networking-midonet-tempest-aio-ml2-full-legacy/8ec487e/ : FAILURE in 23m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/129/check/networking-midonet-tempest-aio-ml2-centos-7/4368121/ : FAILURE in 34m 23s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/129/check/networking-midonet-tempest-aio-ml2-full-centos-7/da8e628/ : FAILURE in 34m 20s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/129/check/networking-midonet-grenade-ml2/fc57f1c/ : FAILURE in 38m 48s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/129/check/networking-midonet-rally-ml2/9284fa8/ : FAILURE in 27m 00s\n- openstack-tox-cover http://logs.openstack.org/87/199387/129/check/openstack-tox-cover/96854f8/ : FAILURE in 10m 10s","accounts_in_message":[],"_revision_number":129},{"id":"7c79e492030e0e7a6643ccb2781668922d0cd36f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-04-08 02:14:35.000000000","message":"Patch Set 130: Commit message was updated.","accounts_in_message":[],"_revision_number":130},{"id":"e1e2a082269c33c70d75ce552e8fbefad2e919e0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-04-08 02:15:04.000000000","message":"Patch Set 130: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 49b2d38b64a173ad6be42280a26be06e15b9a391","accounts_in_message":[],"_revision_number":130},{"id":"46aec62f522ad221567c76c7381173f2a91dcef2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 04:06:59.000000000","message":"Patch Set 130: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/130/check/test-release-openstack/79ac00c/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/130/check/openstack-tox-pep8/5d4946c/ : SUCCESS in 9m 19s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/130/check/openstack-tox-py27/d55f7c5/ : FAILURE in 13m 35s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/130/check/openstack-tox-py35/9ef6ea2/ : FAILURE in 15m 22s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/130/check/openstack-tox-py36/94498cd/ : FAILURE in 15m 28s\n- openstack-tox-docs http://logs.openstack.org/87/199387/130/check/openstack-tox-docs/cf92128/html/ : SUCCESS in 6m 10s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/130/check/build-openstack-releasenotes/38eceac/html/ : SUCCESS in 7m 16s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/130/check/networking-midonet-tempest-multinode-ml2/c94df68/ : FAILURE in 9m 11s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/130/check/networking-midonet-tempest-multinode-ml2-full/d28987f/ : FAILURE in 15m 11s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/130/check/networking-midonet-tempest-aio-ml2-full-legacy/01a764e/ : FAILURE in 15m 23s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/130/check/networking-midonet-tempest-aio-ml2-centos-7/9fc9671/ : SUCCESS in 1h 47m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/130/check/networking-midonet-tempest-aio-ml2-full-centos-7/2d7a507/ : FAILURE in 36m 18s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/130/check/networking-midonet-grenade-ml2/715b096/ : FAILURE in 15m 25s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/130/check/networking-midonet-rally-ml2/ae620b3/ : FAILURE in 12m 52s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/130/check/openstack-tox-cover/b05d405/ : FAILURE in 13m 45s","accounts_in_message":[],"_revision_number":130},{"id":"1d4598de3f57397885332d6b26c19ae3a6ba6ae4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-04-11 10:42:05.000000000","message":"Patch Set 131: Commit message was updated.","accounts_in_message":[],"_revision_number":131},{"id":"be2ff663b54d235bffd22566fd4c0f3b32e93d00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 12:37:26.000000000","message":"Patch Set 131: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/131/check/test-release-openstack/af2fdf4/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/131/check/openstack-tox-pep8/c950440/ : SUCCESS in 8m 55s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/131/check/openstack-tox-py27/7b0a47f/ : FAILURE in 11m 25s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/131/check/openstack-tox-py35/809084d/ : FAILURE in 13m 00s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/131/check/openstack-tox-py36/6bc861d/ : FAILURE in 11m 50s\n- openstack-tox-docs http://logs.openstack.org/87/199387/131/check/openstack-tox-docs/2fdd6ab/html/ : SUCCESS in 5m 19s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/131/check/build-openstack-releasenotes/a285eb6/html/ : SUCCESS in 5m 50s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/131/check/networking-midonet-tempest-multinode-ml2/5433606/ : FAILURE in 13m 30s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/131/check/networking-midonet-tempest-multinode-ml2-full/3ab3277/ : FAILURE in 13m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/131/check/networking-midonet-tempest-aio-ml2-full-legacy/31e5443/ : FAILURE in 11m 33s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/131/check/networking-midonet-tempest-aio-ml2-centos-7/f38a961/ : SUCCESS in 1h 06m 37s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/131/check/networking-midonet-tempest-aio-ml2-full-centos-7/5ea1329/ : SUCCESS in 1h 51m 54s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/131/check/networking-midonet-grenade-ml2/c82ec21/ : FAILURE in 11m 43s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/131/check/networking-midonet-rally-ml2/ce187b2/ : FAILURE in 11m 05s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/131/check/openstack-tox-cover/bf7f5d4/ : FAILURE in 13m 35s","accounts_in_message":[],"_revision_number":131},{"id":"bfdd7ce95e33a355eded0dfc92a28eade0b6ea88","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-04-16 07:03:29.000000000","message":"Patch Set 132: Commit message was updated.","accounts_in_message":[],"_revision_number":132},{"id":"33a738340f9b2a849a59e655a96967de1a615d25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 09:17:40.000000000","message":"Patch Set 132: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/132/check/test-release-openstack/ba9337d/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/132/check/openstack-tox-pep8/782d35e/ : SUCCESS in 8m 32s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/132/check/openstack-tox-py27/8324ad9/ : SUCCESS in 14m 11s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/132/check/openstack-tox-py35/e696e94/ : SUCCESS in 13m 44s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/132/check/openstack-tox-py36/9da00a1/ : SUCCESS in 12m 19s\n- openstack-tox-docs http://logs.openstack.org/87/199387/132/check/openstack-tox-docs/782b8bc/html/ : SUCCESS in 5m 26s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/132/check/build-openstack-releasenotes/88474b2/html/ : SUCCESS in 6m 27s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-multinode-ml2/f573db8/ : FAILURE in 19m 11s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-multinode-ml2-full/e072d35/ : FAILURE in 12m 38s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-aio-ml2-full-legacy/88933ef/ : FAILURE in 14m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-aio-ml2-centos-7/490cb27/ : SUCCESS in 1h 44m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-aio-ml2-full-centos-7/cdad79a/ : FAILURE in 2h 06m 08s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/132/check/networking-midonet-grenade-ml2/c3f06d5/ : FAILURE in 15m 15s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/132/check/networking-midonet-rally-ml2/2fd5d67/ : FAILURE in 13m 05s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/132/check/openstack-tox-cover/621f372/cover/ : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":132},{"id":"9fda96aaea2a3d1a282877f89fc7def65f8de68f","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-18 15:59:38.000000000","message":"Patch Set 132:\n\nIt does seem that the master branch gate is broken to me.\nSee: https://review.openstack.org/#/c/653728/","accounts_in_message":[],"_revision_number":132},{"id":"7bbc0073e62c1777b29926ecc4cbb6b46345e33c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-04-18 22:53:27.000000000","message":"Patch Set 132:\n\nboden,\n\nyes, it\u0027s why this change has Depends-On.","accounts_in_message":[],"_revision_number":132},{"id":"7ae121b40850f70af879dd83507341302ee26b5b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-04-20 07:16:11.000000000","message":"Patch Set 132:\n\nrecheck after repo renames","accounts_in_message":[],"_revision_number":132},{"id":"d738184a1757f719ecfb65d8481e760609848621","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:53:15.000000000","message":"Patch Set 132: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/132/check/test-release-openstack/f4eaf5a/ : SUCCESS in 6m 12s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/132/check/openstack-tox-pep8/0bc9a75/ : SUCCESS in 9m 27s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/132/check/openstack-tox-py27/9540ea5/ : FAILURE in 23m 42s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/132/check/openstack-tox-py35/b5feae7/ : FAILURE in 14m 21s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/132/check/openstack-tox-py36/b637ef1/ : FAILURE in 24m 34s\n- openstack-tox-docs http://logs.openstack.org/87/199387/132/check/openstack-tox-docs/25449ac/html/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/132/check/build-openstack-releasenotes/cb1f935/html/ : SUCCESS in 6m 26s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-multinode-ml2/7d2810c/ : FAILURE in 5m 19s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-multinode-ml2-full/f389d0d/ : FAILURE in 12m 18s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-aio-ml2-full-legacy/dc9f9c2/ : FAILURE in 10m 16s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-aio-ml2-centos-7/793b184/ : FAILURE in 12m 08s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/132/check/networking-midonet-tempest-aio-ml2-full-centos-7/641b80b/ : FAILURE in 12m 01s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/132/check/networking-midonet-grenade-ml2/5fd744d/ : FAILURE in 9m 56s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/132/check/networking-midonet-rally-ml2/4bc9738/ : FAILURE in 9m 15s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/132/check/openstack-tox-cover/8d32403/ : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":132},{"id":"f6f2058d53399ba989141a92a17d3f955a5712a4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-04-22 01:43:30.000000000","message":"Patch Set 133: Commit message was updated.","accounts_in_message":[],"_revision_number":133},{"id":"3c92b99be57b71ca1fe79a01a443828e65e54445","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 03:38:43.000000000","message":"Patch Set 133: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/133/check/test-release-openstack/597f4e0/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/133/check/openstack-tox-pep8/39a0ded/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/133/check/openstack-tox-py27/ad480e0/ : SUCCESS in 11m 20s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/133/check/openstack-tox-py35/347fd3c/ : SUCCESS in 11m 43s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/133/check/openstack-tox-py36/1594b94/ : SUCCESS in 12m 05s\n- openstack-tox-docs http://logs.openstack.org/87/199387/133/check/openstack-tox-docs/3d9825b/html/ : SUCCESS in 4m 33s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/133/check/build-openstack-releasenotes/42ffc66/html/ : SUCCESS in 5m 23s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/133/check/networking-midonet-tempest-multinode-ml2/7244938/ : FAILURE in 15m 17s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/133/check/networking-midonet-tempest-multinode-ml2-full/21531d9/ : FAILURE in 15m 10s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/133/check/networking-midonet-tempest-aio-ml2-full-legacy/755745c/ : FAILURE in 14m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/133/check/networking-midonet-tempest-aio-ml2-centos-7/77bb618/ : SUCCESS in 1h 24m 08s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/133/check/networking-midonet-tempest-aio-ml2-full-centos-7/fb750f8/ : SUCCESS in 1h 49m 02s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/133/check/networking-midonet-grenade-ml2/fa3289d/ : FAILURE in 13m 47s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/133/check/networking-midonet-rally-ml2/00bb3b5/ : FAILURE in 13m 01s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/133/check/openstack-tox-cover/4a9e4c7/cover/ : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":133},{"id":"84dcde09c5706236987ce0291a2e683d9071ed7b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-04-29 14:02:54.000000000","message":"Patch Set 134: Commit message was updated.","accounts_in_message":[],"_revision_number":134},{"id":"9b507c8000996995f71ba087e13bfe9aa759abf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 16:07:14.000000000","message":"Patch Set 134: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/134/check/test-release-openstack/d5f459e/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/134/check/openstack-tox-pep8/8934785/ : SUCCESS in 9m 08s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/134/check/openstack-tox-py27/508d3ce/ : SUCCESS in 14m 00s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/134/check/openstack-tox-py35/2eb0ee5/ : SUCCESS in 13m 17s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/134/check/openstack-tox-py36/e5e5df5/ : SUCCESS in 11m 23s\n- openstack-tox-docs http://logs.openstack.org/87/199387/134/check/openstack-tox-docs/0438d4e/html/ : SUCCESS in 5m 00s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/134/check/build-openstack-releasenotes/2801cd6/html/ : SUCCESS in 5m 40s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-multinode-ml2/98a13ef/ : FAILURE in 16m 21s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-multinode-ml2-full/259964f/ : FAILURE in 17m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-aio-ml2-full-legacy/0ca476c/ : FAILURE in 15m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-aio-ml2-centos-7/9ab0af3/ : SUCCESS in 1h 22m 10s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-aio-ml2-full-centos-7/4a16e14/ : SUCCESS in 1h 51m 34s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/134/check/networking-midonet-grenade-ml2/9d40cac/ : FAILURE in 14m 19s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/134/check/networking-midonet-rally-ml2/b651e1b/ : FAILURE in 12m 11s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/134/check/openstack-tox-cover/ebdc62a/cover/ : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":134},{"id":"476cb7b4cde5cc710d31b22603d9ccae282bfb6d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-05-10 13:46:23.000000000","message":"Patch Set 134:\n\nrecheck","accounts_in_message":[],"_revision_number":134},{"id":"b2b0deedc107b25471de33f351b6df2108bc7242","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:01:08.000000000","message":"Patch Set 134: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/134/check/test-release-openstack/81717d9/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/134/check/openstack-tox-pep8/25787f7/ : FAILURE in 8m 42s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/134/check/openstack-tox-py27/4a17357/ : SUCCESS in 12m 28s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/134/check/openstack-tox-py35/1f40362/ : SUCCESS in 12m 16s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/134/check/openstack-tox-py36/c8fc938/ : SUCCESS in 11m 06s\n- openstack-tox-docs http://logs.openstack.org/87/199387/134/check/openstack-tox-docs/ed87be7/html/ : SUCCESS in 6m 20s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/134/check/build-openstack-releasenotes/bdb86eb/html/ : SUCCESS in 5m 31s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-multinode-ml2/acf93f5/ : FAILURE in 15m 38s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-multinode-ml2-full/aa0359b/ : FAILURE in 17m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-aio-ml2-full-legacy/578d387/ : FAILURE in 12m 35s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-aio-ml2-centos-7/a86e55a/ : FAILURE in 2h 06m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/134/check/networking-midonet-tempest-aio-ml2-full-centos-7/624ead0/ : SUCCESS in 1h 40m 19s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/134/check/networking-midonet-grenade-ml2/eeb7fee/ : FAILURE in 15m 28s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/134/check/networking-midonet-rally-ml2/65f9cb7/ : FAILURE in 19m 37s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/134/check/openstack-tox-cover/41ec122/cover/ : SUCCESS in 30m 12s","accounts_in_message":[],"_revision_number":134},{"id":"70efdd466b6b89a7616b715677015f4e9b76693c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-05-15 13:54:33.000000000","message":"Patch Set 135: Commit message was updated.","accounts_in_message":[],"_revision_number":135},{"id":"95698a4b2f6302de49c7a92f2d35c8e421bea1c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 15:32:51.000000000","message":"Patch Set 135: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/135/check/test-release-openstack/b5eeefe/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/135/check/openstack-tox-pep8/8a5977e/ : FAILURE in 13m 58s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/135/check/openstack-tox-py27/f5a5860/ : SUCCESS in 14m 52s\n- openstack-tox-py35 http://logs.openstack.org/87/199387/135/check/openstack-tox-py35/c0ab79e/ : SUCCESS in 14m 20s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/135/check/openstack-tox-py36/cb6701b/ : SUCCESS in 13m 13s\n- openstack-tox-docs http://logs.openstack.org/87/199387/135/check/openstack-tox-docs/4f6c4b8/html/ : SUCCESS in 4m 50s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/135/check/build-openstack-releasenotes/1ce310a/html/ : SUCCESS in 5m 45s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/135/check/networking-midonet-tempest-multinode-ml2/cb5b2be/ : FAILURE in 15m 53s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/135/check/networking-midonet-tempest-multinode-ml2-full/69f7c0a/ : FAILURE in 20m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/135/check/networking-midonet-tempest-aio-ml2-full-legacy/cad61e6/ : FAILURE in 14m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/135/check/networking-midonet-tempest-aio-ml2-centos-7/3aa091e/ : SUCCESS in 1h 31m 41s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/135/check/networking-midonet-tempest-aio-ml2-full-centos-7/216b1ba/ : FAILURE in 41m 04s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/135/check/networking-midonet-grenade-ml2/97d0672/ : FAILURE in 13m 54s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/135/check/networking-midonet-rally-ml2/2064e8a/ : FAILURE in 13m 47s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/87/199387/135/check/openstack-tox-cover/aafb1da/cover/ : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":135},{"id":"cb11e5372723ef0b9b54ae00074d4a0a817ceb4e","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-06-06 07:46:47.000000000","message":"Patch Set 136: Commit message was updated.","accounts_in_message":[],"_revision_number":136},{"id":"d3929cc130d41d59b119ebda7f17ce8155068f66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 09:52:35.000000000","message":"Patch Set 136: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/136/check/test-release-openstack/8318413/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/136/check/openstack-tox-pep8/d78497e/ : SUCCESS in 7m 37s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/136/check/openstack-tox-py27/5ff37d4/ : SUCCESS in 12m 43s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/136/check/openstack-tox-py36/7335017/ : SUCCESS in 11m 04s\n- openstack-tox-py37 http://logs.openstack.org/87/199387/136/check/openstack-tox-py37/7cf6d74/ : SUCCESS in 12m 14s\n- openstack-tox-docs http://logs.openstack.org/87/199387/136/check/openstack-tox-docs/8ee8baa/html/ : SUCCESS in 4m 52s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/136/check/build-openstack-releasenotes/2718668/html/ : SUCCESS in 5m 51s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-multinode-ml2/b41a1f7/ : FAILURE in 15m 23s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-multinode-ml2-full/d55e114/ : FAILURE in 16m 51s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-aio-ml2-full-legacy/0403136/ : FAILURE in 16m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-aio-ml2-centos-7/8f6a0da/ : SUCCESS in 1h 46m 38s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-aio-ml2-full-centos-7/d7eb5d9/ : SUCCESS in 1h 54m 21s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/136/check/networking-midonet-grenade-ml2/299bb79/ : FAILURE in 16m 29s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/136/check/networking-midonet-rally-ml2/10a1e5c/ : FAILURE in 15m 30s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/87/199387/136/check/openstack-tox-py35/2d8ebb1/ : SUCCESS in 11m 03s\n- openstack-tox-cover http://logs.openstack.org/87/199387/136/check/openstack-tox-cover/4d2f86d/cover/ : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":136},{"id":"114227f6e69a563ed74d299e484db0e4da27fc86","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-07-01 07:24:05.000000000","message":"Patch Set 136:\n\nrecheck","accounts_in_message":[],"_revision_number":136},{"id":"44e9286c26ddb5abeec0ce4e04d36b880cb3c51f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 09:19:31.000000000","message":"Patch Set 136: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/87/199387/136/check/test-release-openstack/d5dc2ae/ : SUCCESS in 2m 05s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/136/check/openstack-tox-pep8/65749e5/ : RETRY_LIMIT in 2m 26s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/136/check/openstack-tox-py27/6d650ac/ : RETRY_LIMIT in 2m 41s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/136/check/openstack-tox-py36/965b7c7/ : RETRY_LIMIT in 3m 16s\n- openstack-tox-py37 http://logs.openstack.org/87/199387/136/check/openstack-tox-py37/4ceb3e2/ : RETRY_LIMIT in 2m 16s\n- openstack-tox-docs http://logs.openstack.org/87/199387/136/check/openstack-tox-docs/4a49e81/html/ : SUCCESS in 3m 46s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/136/check/build-openstack-releasenotes/88b09a7/html/ : SUCCESS in 4m 19s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-multinode-ml2/bcb9e83/ : FAILURE in 19m 03s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-multinode-ml2-full/65d4379/ : FAILURE in 16m 11s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-aio-ml2-full-legacy/0214669/ : FAILURE in 13m 48s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-aio-ml2-centos-7/2bb0a3d/ : SUCCESS in 1h 38m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/136/check/networking-midonet-tempest-aio-ml2-full-centos-7/082d886/ : SUCCESS in 1h 42m 25s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/136/check/networking-midonet-grenade-ml2/fed7294/ : FAILURE in 11m 53s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/136/check/networking-midonet-rally-ml2/1c1fbe8/ : FAILURE in 11m 11s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/87/199387/136/check/openstack-tox-py35/42c3996/ : RETRY_LIMIT in 2m 22s\n- openstack-tox-cover http://logs.openstack.org/87/199387/136/check/openstack-tox-cover/2ce7a69/ : RETRY_LIMIT in 2m 26s","accounts_in_message":[],"_revision_number":136},{"id":"60c9e1e3a7b06ea39739feb719a0e9f03eb14859","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-07-12 04:23:41.000000000","message":"Uploaded patch set 137.","accounts_in_message":[],"_revision_number":137},{"id":"ae8efb752f8f2de1e7db8042a1b78d0e2a45bda1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 06:12:31.000000000","message":"Patch Set 137: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/87/199387/137/check/test-release-openstack/e81f0bd/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/87/199387/137/check/openstack-tox-pep8/7392e1c/ : SUCCESS in 7m 08s\n- openstack-tox-py27 http://logs.openstack.org/87/199387/137/check/openstack-tox-py27/bc40c89/ : SUCCESS in 11m 14s\n- openstack-tox-py36 http://logs.openstack.org/87/199387/137/check/openstack-tox-py36/fbc5c84/ : SUCCESS in 10m 40s\n- openstack-tox-py37 http://logs.openstack.org/87/199387/137/check/openstack-tox-py37/ede2f35/ : SUCCESS in 12m 46s\n- openstack-tox-docs http://logs.openstack.org/87/199387/137/check/openstack-tox-docs/26cd8e5/html/ : SUCCESS in 3m 51s\n- build-openstack-releasenotes http://logs.openstack.org/87/199387/137/check/build-openstack-releasenotes/34634f8/html/ : SUCCESS in 4m 49s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/87/199387/137/check/networking-midonet-tempest-multinode-ml2/cf00b9c/ : FAILURE in 15m 14s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/87/199387/137/check/networking-midonet-tempest-multinode-ml2-full/1599d19/ : FAILURE in 17m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-full-legacy/6fedd3a/ : FAILURE in 13m 05s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-centos-7/2b25b7f/ : SUCCESS in 1h 43m 32s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-full-centos-7/b006449/ : SUCCESS in 1h 25m 02s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/87/199387/137/check/networking-midonet-grenade-ml2/de61aaa/ : FAILURE in 11m 03s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/87/199387/137/check/networking-midonet-rally-ml2/8fe180c/ : FAILURE in 10m 32s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/87/199387/137/check/openstack-tox-py35/6b767a6/ : SUCCESS in 9m 56s\n- openstack-tox-cover http://logs.openstack.org/87/199387/137/check/openstack-tox-cover/0997641/cover/ : SUCCESS in 12m 07s","accounts_in_message":[],"_revision_number":137},{"id":"abbe1a08e6c5e19917e73587c316271c61220d69","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-01 09:31:00.000000000","message":"Patch Set 137:\n\nrecheck","accounts_in_message":[],"_revision_number":137},{"id":"5290b60d432419b98fd48c1c2914a9609969e884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 10:01:41.000000000","message":"Patch Set 137:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/87/199387/137/check/test-release-openstack/9cb27f5/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://logs.opendev.org/87/199387/137/check/openstack-tox-pep8/ae987b4/ : SUCCESS in 9m 02s\n- openstack-tox-py27 https://logs.opendev.org/87/199387/137/check/openstack-tox-py27/b832d4f/ : SUCCESS in 13m 50s\n- openstack-tox-py36 https://logs.opendev.org/87/199387/137/check/openstack-tox-py36/3192170/ : SUCCESS in 13m 35s\n- openstack-tox-py37 https://logs.opendev.org/87/199387/137/check/openstack-tox-py37/33bbb37/ : SUCCESS in 11m 43s\n- openstack-tox-docs https://logs.opendev.org/87/199387/137/check/openstack-tox-docs/ad69af4/html/ : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://logs.opendev.org/87/199387/137/check/build-openstack-releasenotes/5efbd47/html/ : SUCCESS in 6m 08s\n- networking-midonet-tempest-multinode-ml2 https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-multinode-ml2/ef2a1d5/ : FAILURE in 16m 42s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-multinode-ml2-full/6b5d7e0/ : FAILURE in 14m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-full-legacy/b1b8415/ : FAILURE in 13m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-centos-7/9554da3/ : FAILURE in 15m 32s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-full-centos-7/cbe8dd1/ : FAILURE in 17m 24s (non-voting)\n- networking-midonet-grenade-ml2 https://logs.opendev.org/87/199387/137/check/networking-midonet-grenade-ml2/c44ca2a/ : FAILURE in 14m 39s (non-voting)\n- networking-midonet-rally-ml2 https://logs.opendev.org/87/199387/137/check/networking-midonet-rally-ml2/103a1f6/ : FAILURE in 21m 56s (non-voting)\n- openstack-tox-py35 https://logs.opendev.org/87/199387/137/check/openstack-tox-py35/e1144f8/ : SUCCESS in 10m 35s\n- openstack-tox-cover https://logs.opendev.org/87/199387/137/check/openstack-tox-cover/56d129f/cover/ : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":137},{"id":"f89777b94c36befec19bf683037d017039bb54cb","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-06 02:29:13.000000000","message":"Patch Set 137:\n\nrecheck","accounts_in_message":[],"_revision_number":137},{"id":"85701239d47c72e78b8392a2105b5b19a58c1f08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 04:02:48.000000000","message":"Patch Set 137: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack finger://ze02.openstack.org/040e224142a64759a0e9d994e981ff9e : POST_FAILURE in 3m 51s\n- openstack-tox-pep8 finger://ze01.openstack.org/eb749604e37f4576af2ab4d795a0aa8b : POST_FAILURE in 7m 36s\n- openstack-tox-py27 finger://ze02.openstack.org/8b0b093739fa426890554d930e5409b9 : POST_FAILURE in 10m 40s\n- openstack-tox-py36 finger://ze02.openstack.org/7f9e1833012e463f81c71dd7a90ae048 : POST_FAILURE in 11m 26s\n- openstack-tox-py37 finger://ze02.openstack.org/bb1ded4633ec4bdc97e8f497b1a118e1 : POST_FAILURE in 13m 20s\n- openstack-tox-docs finger://ze02.openstack.org/e76d4e6918724d419d3190cfb481a58e : POST_FAILURE in 4m 24s\n- build-openstack-releasenotes finger://ze01.openstack.org/e262535f0bc849d498ba4668cf5f16e0 : POST_FAILURE in 4m 37s\n- networking-midonet-tempest-multinode-ml2 finger://ze03.openstack.org/57e6c58a7e5d49f3828622beefdaf992 : POST_FAILURE in 13m 28s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full finger://ze05.openstack.org/2cee31d2df584529874d1039ecd3cc61 : POST_FAILURE in 15m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy finger://ze03.openstack.org/3371e369f36d45c2be700d3b787952f2 : POST_FAILURE in 12m 16s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 finger://ze02.openstack.org/99037ddcd60848aeabad895256bfca20 : POST_FAILURE in 1h 31m 50s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 finger://ze05.openstack.org/9680d8f9d80f444588d1a3b9cdd5101e : POST_FAILURE in 1h 32m 43s (non-voting)\n- networking-midonet-grenade-ml2 finger://ze03.openstack.org/75ab6f7b1cf4499ba1415cb466bd02cd : POST_FAILURE in 14m 30s (non-voting)\n- networking-midonet-rally-ml2 finger://ze02.openstack.org/febadfc05eb243c488ef40aefbe24305 : POST_FAILURE in 13m 59s (non-voting)\n- openstack-tox-cover finger://ze05.openstack.org/82a4bf4d96e24d7f84a6fa5dbc7eb4d6 : POST_FAILURE in 14m 37s","accounts_in_message":[],"_revision_number":137},{"id":"39794f7438fd7c3dc04834f821c82192138a7d6c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-06 11:04:04.000000000","message":"Patch Set 137:\n\nrecheck","accounts_in_message":[],"_revision_number":137},{"id":"12e2e19cadca6783bda3333f11f2a17eff48338a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:47:28.000000000","message":"Patch Set 137: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/87/199387/137/check/test-release-openstack/a46f79c/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://logs.opendev.org/87/199387/137/check/openstack-tox-pep8/8f811b6/ : SUCCESS in 8m 23s\n- openstack-tox-py27 https://logs.opendev.org/87/199387/137/check/openstack-tox-py27/4b164a1/ : SUCCESS in 12m 50s\n- openstack-tox-py36 https://logs.opendev.org/87/199387/137/check/openstack-tox-py36/9542f0d/ : SUCCESS in 12m 12s\n- openstack-tox-py37 https://logs.opendev.org/87/199387/137/check/openstack-tox-py37/7bd1399/ : SUCCESS in 14m 26s\n- openstack-tox-docs https://logs.opendev.org/87/199387/137/check/openstack-tox-docs/09dda3d/html/ : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://logs.opendev.org/87/199387/137/check/build-openstack-releasenotes/92664fd/html/ : SUCCESS in 5m 21s\n- networking-midonet-tempest-multinode-ml2 https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-multinode-ml2/e7cb973/ : FAILURE in 17m 23s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-multinode-ml2-full/acdb5e7/ : FAILURE in 18m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-full-legacy/05fa677/ : FAILURE in 14m 56s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-centos-7/997c3be/ : FAILURE in 1h 39m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://logs.opendev.org/87/199387/137/check/networking-midonet-tempest-aio-ml2-full-centos-7/1002c6d/ : FAILURE in 1h 32m 13s (non-voting)\n- networking-midonet-grenade-ml2 https://logs.opendev.org/87/199387/137/check/networking-midonet-grenade-ml2/2776af5/ : FAILURE in 13m 55s (non-voting)\n- networking-midonet-rally-ml2 https://logs.opendev.org/87/199387/137/check/networking-midonet-rally-ml2/516ab27/ : FAILURE in 15m 52s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/87/199387/137/check/openstack-tox-cover/e37bba5/cover/ : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":137},{"id":"c3310d99a1aa354eb750979dbfc7f5be748e6983","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-07 07:24:53.000000000","message":"Patch Set 138: Commit message was updated.","accounts_in_message":[],"_revision_number":138},{"id":"00fa498e546c79003fea652b221230c20658b259","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 08:00:08.000000000","message":"Patch Set 138: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":138},{"id":"7beef8ff76fde58cad858d50179c02a139980513","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-07 08:00:37.000000000","message":"Patch Set 138:\n\nrecheck","accounts_in_message":[],"_revision_number":138},{"id":"888d276b7cfea43a6c400f853cd1afaee608fd75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 10:04:18.000000000","message":"Patch Set 138:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/87/199387/138/check/test-release-openstack/8c93ca2/ : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://logs.opendev.org/87/199387/138/check/openstack-tox-pep8/97586e5/ : SUCCESS in 9m 18s\n- openstack-tox-py27 https://logs.opendev.org/87/199387/138/check/openstack-tox-py27/6e76d5a/ : FAILURE in 12m 00s\n- openstack-tox-py36 https://logs.opendev.org/87/199387/138/check/openstack-tox-py36/a02b9b2/ : FAILURE in 13m 43s\n- openstack-tox-py37 https://logs.opendev.org/87/199387/138/check/openstack-tox-py37/9c30f42/ : FAILURE in 16m 47s\n- openstack-tox-docs https://logs.opendev.org/87/199387/138/check/openstack-tox-docs/a8420e4/html/ : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://logs.opendev.org/87/199387/138/check/build-openstack-releasenotes/6b836bd/html/ : SUCCESS in 4m 39s\n- networking-midonet-tempest-multinode-ml2 https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-multinode-ml2/da19ee5/ : FAILURE in 14m 22s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-multinode-ml2-full/32b4383/ : FAILURE in 14m 10s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-aio-ml2-full-legacy/b0f08ef/ : FAILURE in 15m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-aio-ml2-centos-7/3b58730/ : FAILURE in 1h 29m 01s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-aio-ml2-full-centos-7/167ad9b/ : FAILURE in 1h 56m 16s (non-voting)\n- networking-midonet-grenade-ml2 https://logs.opendev.org/87/199387/138/check/networking-midonet-grenade-ml2/26a1ac2/ : FAILURE in 17m 05s (non-voting)\n- networking-midonet-rally-ml2 https://logs.opendev.org/87/199387/138/check/networking-midonet-rally-ml2/58ce20c/ : FAILURE in 13m 26s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/87/199387/138/check/openstack-tox-cover/39dbd15/ : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":138},{"id":"9d8d1a0dc5135aba886430850c5ce991e25fdbff","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-07 10:44:27.000000000","message":"Patch Set 138:\n\nrecheck","accounts_in_message":[],"_revision_number":138},{"id":"7b9be3f3da6b6ea8745d7aa6ac5c6554e2c702af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 12:34:26.000000000","message":"Patch Set 138: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/87/199387/138/check/test-release-openstack/8feb8cd/ : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://logs.opendev.org/87/199387/138/check/openstack-tox-pep8/e8cadc2/ : SUCCESS in 9m 56s\n- openstack-tox-py27 https://logs.opendev.org/87/199387/138/check/openstack-tox-py27/48704e0/ : SUCCESS in 13m 43s\n- openstack-tox-py36 https://logs.opendev.org/87/199387/138/check/openstack-tox-py36/4811b46/ : SUCCESS in 12m 52s\n- openstack-tox-py37 https://logs.opendev.org/87/199387/138/check/openstack-tox-py37/81fe132/ : SUCCESS in 12m 36s\n- openstack-tox-docs https://logs.opendev.org/87/199387/138/check/openstack-tox-docs/0defa9b/html/ : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://logs.opendev.org/87/199387/138/check/build-openstack-releasenotes/6f5e0e9/html/ : SUCCESS in 5m 45s\n- networking-midonet-tempest-multinode-ml2 https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-multinode-ml2/7332fb5/ : FAILURE in 18m 20s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-multinode-ml2-full/801be69/ : FAILURE in 19m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-aio-ml2-full-legacy/1d72352/ : FAILURE in 18m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-aio-ml2-centos-7/accb4fd/ : FAILURE in 1h 41m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://logs.opendev.org/87/199387/138/check/networking-midonet-tempest-aio-ml2-full-centos-7/bfdaa3c/ : FAILURE in 1h 31m 54s (non-voting)\n- networking-midonet-grenade-ml2 https://logs.opendev.org/87/199387/138/check/networking-midonet-grenade-ml2/0eb666b/ : FAILURE in 19m 50s (non-voting)\n- networking-midonet-rally-ml2 https://logs.opendev.org/87/199387/138/check/networking-midonet-rally-ml2/338b385/ : FAILURE in 11m 34s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/87/199387/138/check/openstack-tox-cover/27ae46a/cover/ : SUCCESS in 12m 57s","accounts_in_message":[],"_revision_number":138},{"id":"333d4b39a405c2562ebeef5bb58614ffeed09e54","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-07 13:44:11.000000000","message":"Patch Set 138:\n\nrecheck","accounts_in_message":[],"_revision_number":138},{"id":"525700d63939d76c39b932deb30f040030f1ddab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:39:41.000000000","message":"Patch Set 138: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":138},{"id":"c925a838e73f1e7fbcf4dec0a72ced23921e20a3","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-08 08:23:14.000000000","message":"Patch Set 139: Commit message was updated.","accounts_in_message":[],"_revision_number":139},{"id":"2fa0c2800f382591c5555587e9d716c5799ce449","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 10:41:20.000000000","message":"Patch Set 139: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/87/199387/139/check/test-release-openstack/0518ebb/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://logs.opendev.org/87/199387/139/check/openstack-tox-pep8/fe94f06/ : SUCCESS in 8m 18s\n- openstack-tox-py27 https://logs.opendev.org/87/199387/139/check/openstack-tox-py27/f72fa35/ : SUCCESS in 11m 48s\n- openstack-tox-py36 https://logs.opendev.org/87/199387/139/check/openstack-tox-py36/96d52c0/ : SUCCESS in 11m 48s\n- openstack-tox-py37 https://logs.opendev.org/87/199387/139/check/openstack-tox-py37/8427664/ : SUCCESS in 12m 02s\n- openstack-tox-docs https://logs.opendev.org/87/199387/139/check/openstack-tox-docs/62a3f4b/html/ : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://logs.opendev.org/87/199387/139/check/build-openstack-releasenotes/fcff9f3/html/ : SUCCESS in 5m 53s\n- networking-midonet-tempest-multinode-ml2 https://logs.opendev.org/87/199387/139/check/networking-midonet-tempest-multinode-ml2/fd95487/ : FAILURE in 13m 10s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://logs.opendev.org/87/199387/139/check/networking-midonet-tempest-multinode-ml2-full/07eb40b/ : FAILURE in 18m 08s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://logs.opendev.org/87/199387/139/check/networking-midonet-tempest-aio-ml2-full-legacy/7a49400/ : FAILURE in 12m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/87/199387/139/check/networking-midonet-tempest-aio-ml2-centos-7/5032ebf/ : FAILURE in 1h 33m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://logs.opendev.org/87/199387/139/check/networking-midonet-tempest-aio-ml2-full-centos-7/98a5aef/ : FAILURE in 1h 49m 30s (non-voting)\n- networking-midonet-grenade-ml2 https://logs.opendev.org/87/199387/139/check/networking-midonet-grenade-ml2/a10c3ab/ : FAILURE in 17m 52s (non-voting)\n- networking-midonet-rally-ml2 https://logs.opendev.org/87/199387/139/check/networking-midonet-rally-ml2/09a1ff7/ : FAILURE in 14m 25s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/87/199387/139/check/openstack-tox-cover/f180341/cover/ : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":139},{"id":"a804d710f664cbb1042ab20c6eb5c19d1667a6e2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-14 09:19:46.000000000","message":"Patch Set 139:\n\nrecheck","accounts_in_message":[],"_revision_number":139},{"id":"2b07bd6510fcd42385aadc0df3e9e0b43fcebb12","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-14 11:05:39.000000000","message":"Patch Set 140: Commit message was updated.","accounts_in_message":[],"_revision_number":140},{"id":"b96da9cfa4a495e9a9892567bacdd986fa4ec077","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 13:03:58.000000000","message":"Patch Set 140: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/32fcbc7a827b442cb963ae35c5c7a44f : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517a457811b749f19f1461d1c0a68c20 : SUCCESS in 9m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d5444556df44251ae14c89f2bed0041 : SUCCESS in 14m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f4f32655faa495c81aef56420c39410 : SUCCESS in 15m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d42043b06a7427bb3a5ab0cca1e95f5 : SUCCESS in 14m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe92d61fc96402e87472a51397d6a67 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed7af6bf9bac4b0987f107e00a765285 : SUCCESS in 6m 06s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/10e90b07304140649e79ca21a1f8db92 : FAILURE in 1h 30m 17s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/2b1e7fccd67a4b01b0c947225eba34b8 : FAILURE in 1h 56m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56a85671ab4d4d58b065c0c144ec6211 : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":140},{"id":"61995c98c70b455fafb6f31e7e68d4e765fdbd5c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-14 15:29:37.000000000","message":"Patch Set 141: Commit message was updated.","accounts_in_message":[],"_revision_number":141},{"id":"7c43302be8f8ea1fc6b289084c63fbf8d819be64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 17:24:08.000000000","message":"Patch Set 141: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0b0206660d4b45d6b0d23f0d743030cc : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29bbb04fab9e4cce88308b3ce73fd836 : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de084286d8d44084b4436b2f57f0b52c : SUCCESS in 11m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aca97abac3a43f0bbc55681dde7e752 : SUCCESS in 13m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb6562a133ec4bdcb1396666dca81fa9 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34cd1bad4cd146f5b2df7e9c8380114c : POST_FAILURE in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ddde979e808460ca2fb72886539d9a1 : SUCCESS in 5m 27s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/2204aa64bbb043998d51ef337caaf361 : SUCCESS in 1h 32m 28s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/888b4f268c9f44fcb168dcdc9f7d977b : SUCCESS in 1h 46m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/822f635c41334328a0547dafa0584805 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":141},{"id":"e8da844b414b4ce4bce79296de6cc1cf0adbe72b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-09-30 23:42:40.000000000","message":"Patch Set 141:\n\nrecheck","accounts_in_message":[],"_revision_number":141},{"id":"f2d7f72939348f3a5b50921d29d2ecd196533722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 23:43:10.000000000","message":"Patch Set 141:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":141},{"id":"cec5cfdf47291ec2fe1df2abf6d8e5d69df9a30d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-10-01 23:41:23.000000000","message":"Patch Set 140:\n\nrecheck","accounts_in_message":[],"_revision_number":140},{"id":"d36efe120cfbdf682b60f7e2025b5825143c8200","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-08 05:13:45.000000000","message":"Patch Set 142: Commit message was updated.","accounts_in_message":[],"_revision_number":142},{"id":"38d5d8576033ed0ab104a46e86f08a7647fd5865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 07:20:54.000000000","message":"Patch Set 142: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/97f3d8b7f8114af7adb2be6a30face53 : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3596d6b98b9e4b078b3a488102f98c0c : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61921e741b8348feb6e48d72fd03059d : SUCCESS in 11m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bd623becc5a459e8eb50d55807f8e25 : SUCCESS in 10m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30d3e5773da5484f848bc5ea02f228ee : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ed56dc371c480ba01468b058cb8079 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/589fc45d32024614a15f8faf9e81f33c : SUCCESS in 5m 02s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/82384239cd3d46189eca2776a88b662d : FAILURE in 1h 28m 12s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/4227cf6017464dc8b847201a79f9be67 : FAILURE in 2h 01m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4784b8696b04c19adae756aa5e12f7b : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":142},{"id":"7aa507f25068aa84ab880bbe0090f488e659e3a4","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-15 13:49:58.000000000","message":"Patch Set 142:\n\nrecheck","accounts_in_message":[],"_revision_number":142},{"id":"2ed783991c2f9fbf1c6ca8e9612cd0fed61fad08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 15:51:06.000000000","message":"Patch Set 142:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e01be82e13b945e090c691f8125ed073 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c668619ef048449bb7f52f9a37e5a6e2 : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1cebb5cb601497f8b99d5eec8d31a46 : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42e7eebd2b8e4f39b8812f16316364df : SUCCESS in 14m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc512266cf5c4cdeb73c0f29b056d81b : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b9b20b1d574db48bc4b40f59d02519 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f665fc6621874ad9b37b4cca04e7c988 : SUCCESS in 5m 19s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/134c24d794b948f9bb4b8cb47811af65 : FAILURE in 1h 38m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/327bcb2083514250b40216877a868aa9 : FAILURE in 1h 54m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b66c1385687492385dab27b33947eb2 : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":142},{"id":"eb35d21177087cce8e339968144a9fc6d9c13d66","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-23 02:32:47.000000000","message":"Patch Set 142:\n\nrecheck","accounts_in_message":[],"_revision_number":142},{"id":"5816d15b7682ce5fdc50970688d89d1f721d6268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 03:54:41.000000000","message":"Patch Set 142:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d02709c44fd46fca928c69ad5c3c014 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/233a0babdc614c7487657970345a0842 : SUCCESS in 7m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce1a72a4e4de4a9787ade545bf9f0f7b : SUCCESS in 11m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d32be4644eb546fb9e52474bed14f9fe : SUCCESS in 13m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/649f32fa81744e088f2f446a572bbe9a : SUCCESS in 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df15641bd4a409b89c2f20ce291381f : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/590332c70874461cb9bc98a2dda885b8 : SUCCESS in 5m 07s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/bf8119f0fb394837b443158b73ff4e75 : FAILURE in 1h 19m 34s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/876188322cbe4c239d5fb5ddbb96785e : FAILURE in 15m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a07723aa0964393aa30b1a4f61fcf11 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":142},{"id":"d28f54ebbead81c1528c24dc895dbf0d929b9c4e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:32:08.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment and currently blocked by a core reviewer with a -2. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and contacting the reviewer with the -2 on this review to ensure you address their concerns.","accounts_in_message":[],"_revision_number":142},{"id":"7dd322906721052514a3a934eb9802e505879fd9","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-06 09:46:15.000000000","message":"Restored","accounts_in_message":[],"_revision_number":142},{"id":"1d5123945f534967cdfc9d85aac7cd5c00c489cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 10:11:39.000000000","message":"Patch Set 142:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7665d8249c004ceca02c474e5fc71ad7 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c5888dedfc4c7285d404821d1282a5 : SUCCESS in 10m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d51774d77df349ecad76b344f5bd6394 : SUCCESS in 14m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8345d9acd3cc4cd1990a90b27048649a : SUCCESS in 14m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/048371075d1f422c8e5c152cb4502445 : SUCCESS in 17m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff9e2ad94744866ab79a96f3973832f : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d7b2cd5d5624296815772fa40c2e020 : SUCCESS in 8m 08s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/155b466d3f044d879cde6ed41b32e04b : FAILURE in 18m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/f09ebb0b370a4676b23b121b76e28eeb : FAILURE in 17m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38f28ab9fb7946b093a7d3bba3540904 : SUCCESS in 17m 03s","accounts_in_message":[],"_revision_number":142},{"id":"ecf93a1ab5bd44b78bd0f13a758ba27706240561","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-10 03:27:31.000000000","message":"Patch Set 142:\n\nrecheck after https://review.opendev.org/#/c/681201/","accounts_in_message":[],"_revision_number":142},{"id":"998c5a5947e30eb763718729a82394dda03eaa38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 05:04:57.000000000","message":"Patch Set 142:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/75559b1fe667479586ff71cd63acb9ae : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a62e29282c2447d3b2269e875c98644b : SUCCESS in 10m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3dc5cba6ccd149ae98f1312c411f9cfa : SUCCESS in 11m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/899093a51a714115b91f512b76e5301e : SUCCESS in 11m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e6563ca65c641b4aa9b204355d1a613 : SUCCESS in 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc57b7b6e91947979120fbb06dcc5ea0 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/714889cc198947a4918d0128e155f704 : SUCCESS in 4m 34s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/7c1636d13dc74790b178651b383c3b9a : SUCCESS in 1h 22m 23s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/64c42035675a431da4ee702dc4b599b1 : FAILURE in 1h 35m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/767f5bc2a3ee4435a5cbf4ac69f4003e : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":142},{"id":"9bf4da606e39ad0154102bd32bdae2af632c6f79","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-12 06:02:25.000000000","message":"Patch Set 142:\n\nrecheck","accounts_in_message":[],"_revision_number":142},{"id":"f75ef9e31f1f9af1467cbb96ee60fe06e3cb2821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 08:00:44.000000000","message":"Patch Set 142:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7bcf7acfa53d44bc9eb42d7dbaf15e62 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa7378236b64e1da22fbcf6beae8d5d : SUCCESS in 8m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21869dbadcdf49f6b30fda41ba57dbc2 : SUCCESS in 19m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/547e27a25cc24ee09d47512d0490201c : SUCCESS in 13m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de6a122eefdf4b14ace61c0c378b26a2 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c2784dd9ad4c6b92bd338e1f88407d : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c0989343be44884985daab5e674b954 : SUCCESS in 6m 26s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/123977fdfeae43318bc69a894f6d5a52 : SUCCESS in 1h 42m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/8a67c81730934e85ae8d7b9be916a7cc : FAILURE in 1h 26m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a873cd1f96e41679ce771feaf38fc65 : SUCCESS in 23m 57s","accounts_in_message":[],"_revision_number":142},{"id":"70d6eee88aa7bdc08cb44d089984c93c87b5d3fc","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-16 02:13:34.000000000","message":"Patch Set 142:\n\nrecheck","accounts_in_message":[],"_revision_number":142},{"id":"5181f9adb65daedfac12c29e4bd5e0de0445c96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 02:43:46.000000000","message":"Patch Set 142:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/df9b8c74d92946e68ff3a767e7d2aade : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/defcf2cb0dd5429e9edf95783b60afd2 : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1089d8ae86e444b9915fba0e3701d459 : SUCCESS in 10m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/925b2ae9c6894455962e7015f1912f8b : SUCCESS in 10m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/036c4100429f427dbe92365fbfaa60ed : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed1a24638f4a476ebc95108b7a6b68e8 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f63b3a11c28d45999f1b93aa74a7725f : SUCCESS in 5m 25s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/4cd61e90467544069b47909776eaff86 : FAILURE in 24m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/dba199062b0f43f9a70e5aa5ab8fc665 : FAILURE in 22m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb80b65e859744cd9a2e84a8611c17d2 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":142},{"id":"38cd3d4a1a9e2eba5681e2b898622ea691973ab3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:33:31.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment and currently blocked by a core reviewer with a -2. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and contacting the reviewer with the -2 on this review to ensure you address their concerns.","accounts_in_message":[],"_revision_number":142}],"current_revision_number":142,"current_revision":"9d21c8e231e0f186e2be721a811a92e957a5e2ff","revisions":{"38c61ca544f5d5ad26f6d23875c95357703d308b":{"kind":"REWORK","_number":1,"created":"2015-07-08 03:21:22.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/1"}}},"commit":{"parents":[{"commit":"048dcecdfb4e99dd674d13cf1a64029bfb8ddae2","subject":"Allow users to run \u0027tox -epy34\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/048dcecdfb4e99dd674d13cf1a64029bfb8ddae2"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/38c61ca544f5d5ad26f6d23875c95357703d308b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/38c61ca544f5d5ad26f6d23875c95357703d308b"}]},"branch":"refs/heads/master"},"e5ae6a0dc51753401e7384ab2350f4aa94782eb2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2015-07-09 08:20:50.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/2"}}},"commit":{"parents":[{"commit":"eb9da0f5d1e906b23a8830b25ecc965a3e9949eb","subject":"Fix python3 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/eb9da0f5d1e906b23a8830b25ecc965a3e9949eb"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-09 08:20:14.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e5ae6a0dc51753401e7384ab2350f4aa94782eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e5ae6a0dc51753401e7384ab2350f4aa94782eb2"}]},"branch":"refs/heads/master"},"a337ceb16f9e6500afa36bd722193ace2e26031e":{"kind":"NO_CHANGE","_number":3,"created":"2015-07-09 08:22:40.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/3"}}},"commit":{"parents":[{"commit":"69a5e174bdf798839459d32536da907a56033782","subject":"Fix python3 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/69a5e174bdf798839459d32536da907a56033782"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-09 08:22:10.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a337ceb16f9e6500afa36bd722193ace2e26031e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a337ceb16f9e6500afa36bd722193ace2e26031e"}]},"branch":"refs/heads/master"},"3f818c16a3d5060cc905ba09729c891041a62346":{"kind":"REWORK","_number":4,"created":"2015-08-27 06:05:35.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/4"}}},"commit":{"parents":[{"commit":"9a961ddd1ce99da85b5a4792240f4947a728927d","subject":"Merge \"Define VIF_TYPE_MIDONET in networking-midonet repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9a961ddd1ce99da85b5a4792240f4947a728927d"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-08-27 06:05:11.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3f818c16a3d5060cc905ba09729c891041a62346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3f818c16a3d5060cc905ba09729c891041a62346"}]},"branch":"refs/heads/master"},"17a4f28af0a119765be1dd2a1f3935daac2911e3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2015-10-28 06:30:11.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/5"}}},"commit":{"parents":[{"commit":"3943328ffa6753a88b82ac163b3c1023eee4a884","subject":"devstack: Set midolman log level to DEBUG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3943328ffa6753a88b82ac163b3c1023eee4a884"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-10-28 06:30:11.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/17a4f28af0a119765be1dd2a1f3935daac2911e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/17a4f28af0a119765be1dd2a1f3935daac2911e3"}]},"branch":"refs/heads/master"},"ebc0e2fa3884f381179450d768582eebe7700052":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2015-11-06 08:59:10.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/6"}}},"commit":{"parents":[{"commit":"3943328ffa6753a88b82ac163b3c1023eee4a884","subject":"devstack: Set midolman log level to DEBUG","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3943328ffa6753a88b82ac163b3c1023eee4a884"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-11-06 08:59:10.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ic9c58a92f66fe8b1bd7788d7f3daf361545dc8be\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ebc0e2fa3884f381179450d768582eebe7700052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ebc0e2fa3884f381179450d768582eebe7700052"}]},"branch":"refs/heads/master"},"78b3106e38d856109df4089a9e25c649ce845543":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2015-11-06 08:59:14.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/7"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-11-06 08:59:14.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ic9c58a92f66fe8b1bd7788d7f3daf361545dc8be\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/78b3106e38d856109df4089a9e25c649ce845543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/78b3106e38d856109df4089a9e25c649ce845543"}]},"branch":"refs/heads/master"},"8ec19377adac5259794ff2161fef2dd458a6f1b5":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2015-11-30 09:15:53.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/8"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-11-30 09:15:53.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8ec19377adac5259794ff2161fef2dd458a6f1b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8ec19377adac5259794ff2161fef2dd458a6f1b5"}]},"branch":"refs/heads/master"},"3871e8ceeefb1a43022f0a74dfa0b6793fb6a83e":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2015-12-15 10:41:18.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/9"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-12-15 10:41:18.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I7595f61af05d4667dc561be60dda3f3c83957c2c\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3871e8ceeefb1a43022f0a74dfa0b6793fb6a83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3871e8ceeefb1a43022f0a74dfa0b6793fb6a83e"}]},"branch":"refs/heads/master"},"4eea2049db6092c880dbd8bdc1976b6fafdaf338":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2015-12-15 14:10:09.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/10"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-12-15 14:10:09.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4eea2049db6092c880dbd8bdc1976b6fafdaf338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4eea2049db6092c880dbd8bdc1976b6fafdaf338"}]},"branch":"refs/heads/master"},"2b8bfd55962ad316b0856ce7aea03f7335cde826":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2016-01-18 14:02:31.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/11"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-01-18 14:02:17.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I54baa82d15b68fe8609a8560a83ca5999d122402\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2b8bfd55962ad316b0856ce7aea03f7335cde826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2b8bfd55962ad316b0856ce7aea03f7335cde826"}]},"branch":"refs/heads/master"},"3d887caa8e515dcdcc1123a7ec135a1326faa77a":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2016-01-20 04:03:54.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/12"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-01-20 04:03:38.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: If3dbf0db572b4ea92ebcd8d4769807351ea84ff7\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3d887caa8e515dcdcc1123a7ec135a1326faa77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3d887caa8e515dcdcc1123a7ec135a1326faa77a"}]},"branch":"refs/heads/master"},"72ef5b71899d093aea896acfb033215aa66b20d1":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2016-01-25 12:58:18.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/13"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-01-25 12:58:07.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I72f4ae947cb3f58938c75691a4d7da8d6203466f\nDepends-On: If3dbf0db572b4ea92ebcd8d4769807351ea84ff7\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/72ef5b71899d093aea896acfb033215aa66b20d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/72ef5b71899d093aea896acfb033215aa66b20d1"}]},"branch":"refs/heads/master"},"12ce3b0a4f41b813ceba0e5b3974424d3c306f16":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2016-01-25 21:44:59.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/14"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-01-25 21:44:53.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/12ce3b0a4f41b813ceba0e5b3974424d3c306f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/12ce3b0a4f41b813ceba0e5b3974424d3c306f16"}]},"branch":"refs/heads/master"},"de653484220511ace4107e3c1c3ecdd22e5e84ff":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2016-01-26 09:06:49.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/15"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-01-26 09:06:45.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I7242867e41abc625eb1085983118c09a28249b85\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/de653484220511ace4107e3c1c3ecdd22e5e84ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/de653484220511ace4107e3c1c3ecdd22e5e84ff"}]},"branch":"refs/heads/master"},"b6b2fb610a700cc9d879eef1b854071bcadc7376":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2016-01-27 12:07:29.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/16"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-01-27 12:07:25.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I11ca340a26b90f4424a40226e8e8679895163ea3\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b6b2fb610a700cc9d879eef1b854071bcadc7376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b6b2fb610a700cc9d879eef1b854071bcadc7376"}]},"branch":"refs/heads/master"},"92dd1ddd5dc3aeaa6e20349efef0b69dde4f5b10":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2016-01-28 09:55:20.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/17"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-01-28 09:55:16.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I4826d04665875df3b07f3d692a8bdb070c5f3b91\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/92dd1ddd5dc3aeaa6e20349efef0b69dde4f5b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/92dd1ddd5dc3aeaa6e20349efef0b69dde4f5b10"}]},"branch":"refs/heads/master"},"30ef0a09fc5bf55f9cabdb6e7d0dadbc33e9829d":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2016-02-02 05:26:41.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/18"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-02-02 05:26:05.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/30ef0a09fc5bf55f9cabdb6e7d0dadbc33e9829d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/30ef0a09fc5bf55f9cabdb6e7d0dadbc33e9829d"}]},"branch":"refs/heads/master"},"6bd3130564b6c3da4e3ad23344500b7717a75842":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2016-03-01 13:21:05.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/19"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-03-01 13:21:00.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I74eed67e34185e63d0e97c8098cd189d4c54a61a\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6bd3130564b6c3da4e3ad23344500b7717a75842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6bd3130564b6c3da4e3ad23344500b7717a75842"}]},"branch":"refs/heads/master"},"813de919501a940989a99e32d0d288fe2d3a20af":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2016-03-03 03:32:08.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/20"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-03-03 03:31:48.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I8888d42bc986fd71665526bd99ae2053ef5473e8\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/813de919501a940989a99e32d0d288fe2d3a20af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/813de919501a940989a99e32d0d288fe2d3a20af"}]},"branch":"refs/heads/master"},"ef3a53360cdf1b423886901f1ca4f9bf28a3c074":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2016-03-07 03:49:20.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/21"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-03-07 03:49:16.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I8888d42bc986fd71665526bd99ae2053ef5473e8\nDepends-On: I9a660922b1aa6826ec19696ee81abb1e9f11d6a4\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ef3a53360cdf1b423886901f1ca4f9bf28a3c074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ef3a53360cdf1b423886901f1ca4f9bf28a3c074"}]},"branch":"refs/heads/master"},"51ce182afdaf0904247bf2327f1d3f93c975574f":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2016-03-07 07:58:46.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/22"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-03-07 07:58:37.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I8888d42bc986fd71665526bd99ae2053ef5473e8\nDepends-On: I9a660922b1aa6826ec19696ee81abb1e9f11d6a4\nDepends-On: Iedd880aa2879fe2e6a3dfafbc9d23d8e8ca0fcff\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/51ce182afdaf0904247bf2327f1d3f93c975574f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/51ce182afdaf0904247bf2327f1d3f93c975574f"}]},"branch":"refs/heads/master"},"2568b5a85879abae0f9b5be0ee1ef0823f58d8e2":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2016-03-13 23:56:58.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/23"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-03-13 23:56:54.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2568b5a85879abae0f9b5be0ee1ef0823f58d8e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2568b5a85879abae0f9b5be0ee1ef0823f58d8e2"}]},"branch":"refs/heads/master"},"8738cb928f70dc1b3e5ce0d1af6c0528fbd79339":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2016-04-14 09:52:32.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/24"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-04-14 09:52:28.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I0544f1d47ae53d572adda872847a56fa0b202d2e\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8738cb928f70dc1b3e5ce0d1af6c0528fbd79339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8738cb928f70dc1b3e5ce0d1af6c0528fbd79339"}]},"branch":"refs/heads/master"},"f7154511e690ed5e3b3939a314953f8be46c2afe":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2016-04-14 16:55:34.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/25"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-04-14 16:55:28.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f7154511e690ed5e3b3939a314953f8be46c2afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f7154511e690ed5e3b3939a314953f8be46c2afe"}]},"branch":"refs/heads/master"},"035cdac38b50f836b9c1634bd253ddf6739dfbed":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2016-04-20 05:24:55.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/26"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-04-20 05:24:46.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Iaa33face9de8fbf1f28a90fa54e3b955bae3abb2\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/035cdac38b50f836b9c1634bd253ddf6739dfbed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/035cdac38b50f836b9c1634bd253ddf6739dfbed"}]},"branch":"refs/heads/master"},"3ea6c1a50e995df3634754fa7123b677f8d301a9":{"kind":"NO_CODE_CHANGE","_number":27,"created":"2016-04-22 07:41:03.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/27"}}},"commit":{"parents":[{"commit":"6ed85a7e31a93f4d99b56a3bab99056df92a4bb3","subject":"Merge \"Use stevedore entry point instead of module path for interface driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6ed85a7e31a93f4d99b56a3bab99056df92a4bb3"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-04-22 07:40:55.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3ea6c1a50e995df3634754fa7123b677f8d301a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3ea6c1a50e995df3634754fa7123b677f8d301a9"}]},"branch":"refs/heads/master"},"8fe2092502626c9a35625d7f0ef1a2d72de7a1e6":{"kind":"REWORK","_number":28,"created":"2016-05-16 05:15:55.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/28"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-05-16 05:15:49.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8fe2092502626c9a35625d7f0ef1a2d72de7a1e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8fe2092502626c9a35625d7f0ef1a2d72de7a1e6"}]},"branch":"refs/heads/master"},"cc7506ea65a0ee20eaa31ce06f4a89580abab0ed":{"kind":"NO_CODE_CHANGE","_number":29,"created":"2016-06-22 14:23:35.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/29"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-06-22 14:23:29.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I1cef944115c1911d4b7e60fb4dff389b1911ddbb\nDepends-On: I1737a47cb5662ae7f9ebaeb789a64445f12fd147\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/cc7506ea65a0ee20eaa31ce06f4a89580abab0ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/cc7506ea65a0ee20eaa31ce06f4a89580abab0ed"}]},"branch":"refs/heads/master"},"e7283ca982ec9dba3e5463c7a28a201a0d6df7a6":{"kind":"NO_CODE_CHANGE","_number":30,"created":"2016-06-23 06:37:57.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/30"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-06-23 06:37:48.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I1cef944115c1911d4b7e60fb4dff389b1911ddbb\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e7283ca982ec9dba3e5463c7a28a201a0d6df7a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e7283ca982ec9dba3e5463c7a28a201a0d6df7a6"}]},"branch":"refs/heads/master"},"ac797ad254f5c7cb3057b300e5c17ab526c31c5f":{"kind":"NO_CODE_CHANGE","_number":31,"created":"2016-06-27 04:10:51.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/31"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-06-27 04:10:45.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie8f3ec1b23c14f36992886510c3114cf956769d4\nDepends-On: I1cef944115c1911d4b7e60fb4dff389b1911ddbb\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ac797ad254f5c7cb3057b300e5c17ab526c31c5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ac797ad254f5c7cb3057b300e5c17ab526c31c5f"}]},"branch":"refs/heads/master"},"8b93ad37d70269c20568ddb6d72b73385c57a8e5":{"kind":"NO_CODE_CHANGE","_number":32,"created":"2016-06-27 05:13:55.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/32"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-06-27 05:13:52.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I1cef944115c1911d4b7e60fb4dff389b1911ddbb\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8b93ad37d70269c20568ddb6d72b73385c57a8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8b93ad37d70269c20568ddb6d72b73385c57a8e5"}]},"branch":"refs/heads/master"},"c9ec9aeaad168caeadeaca1c9b11dc74a69b62fe":{"kind":"NO_CODE_CHANGE","_number":33,"created":"2016-06-27 11:53:54.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/33"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-06-27 11:53:50.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie8f3ec1b23c14f36992886510c3114cf956769d4\nDepends-On: I1cef944115c1911d4b7e60fb4dff389b1911ddbb\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c9ec9aeaad168caeadeaca1c9b11dc74a69b62fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c9ec9aeaad168caeadeaca1c9b11dc74a69b62fe"}]},"branch":"refs/heads/master"},"6facd8f440752d8234b97b399ed75c69ba439ddb":{"kind":"NO_CODE_CHANGE","_number":34,"created":"2016-06-27 13:28:06.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/34"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-06-27 13:28:01.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie8f3ec1b23c14f36992886510c3114cf956769d4\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6facd8f440752d8234b97b399ed75c69ba439ddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6facd8f440752d8234b97b399ed75c69ba439ddb"}]},"branch":"refs/heads/master"},"9e5504657636946c562fe80c11782a122db85d0c":{"kind":"NO_CODE_CHANGE","_number":35,"created":"2016-07-27 06:17:06.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/35"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-07-27 06:16:54.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9e5504657636946c562fe80c11782a122db85d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9e5504657636946c562fe80c11782a122db85d0c"}]},"branch":"refs/heads/master"},"6c12f41ee62b20587494a190ef03d13ce6d21f83":{"kind":"NO_CODE_CHANGE","_number":36,"created":"2016-08-09 11:19:52.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/36"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-08-09 11:04:54.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I9924600c57648f7eccaa5abb6979419d9547a2ff\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6c12f41ee62b20587494a190ef03d13ce6d21f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6c12f41ee62b20587494a190ef03d13ce6d21f83"}]},"branch":"refs/heads/master"},"2771f4a1218f4c015814cfb1ab40b1dbc5f84c88":{"kind":"NO_CODE_CHANGE","_number":37,"created":"2016-08-09 22:17:02.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/37"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-08-09 22:16:57.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I3f05d3ce402669160d56417c977cbf28d366c4c2\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2771f4a1218f4c015814cfb1ab40b1dbc5f84c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2771f4a1218f4c015814cfb1ab40b1dbc5f84c88"}]},"branch":"refs/heads/master"},"18e59ba14e63937728a6d9806364c85e8dc3f9de":{"kind":"NO_CODE_CHANGE","_number":38,"created":"2016-08-10 04:33:38.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/38"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-08-10 04:33:31.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I70afcea717b7ca8434be0746a7809b94929b6803\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/18e59ba14e63937728a6d9806364c85e8dc3f9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/18e59ba14e63937728a6d9806364c85e8dc3f9de"}]},"branch":"refs/heads/master"},"d8da487a643bb9a1dab2b9b0f569c77791452042":{"kind":"NO_CODE_CHANGE","_number":39,"created":"2016-08-11 12:24:39.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/39"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-08-11 12:24:31.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d8da487a643bb9a1dab2b9b0f569c77791452042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d8da487a643bb9a1dab2b9b0f569c77791452042"}]},"branch":"refs/heads/master"},"3587638f4c0baf6005b36720c173f9e4c39ab68a":{"kind":"NO_CODE_CHANGE","_number":40,"created":"2016-08-26 06:02:54.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/40"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-08-26 06:02:49.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie8f3ec1b23c14f36992886510c3114cf956769d4\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3587638f4c0baf6005b36720c173f9e4c39ab68a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3587638f4c0baf6005b36720c173f9e4c39ab68a"}]},"branch":"refs/heads/master"},"3d03f791d416123c43a45ad317ac33f7ec1105a3":{"kind":"NO_CODE_CHANGE","_number":41,"created":"2016-09-09 08:13:21.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/41"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-09-09 08:13:13.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3d03f791d416123c43a45ad317ac33f7ec1105a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3d03f791d416123c43a45ad317ac33f7ec1105a3"}]},"branch":"refs/heads/master"},"fb85ed3b4078080714f3e02dec6d6bd083d9f418":{"kind":"NO_CODE_CHANGE","_number":42,"created":"2016-09-30 09:39:10.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/42"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-09-30 09:38:47.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Iebd5a44d2195276e92111cb8964b90fc2955e2de\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fb85ed3b4078080714f3e02dec6d6bd083d9f418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fb85ed3b4078080714f3e02dec6d6bd083d9f418"}]},"branch":"refs/heads/master"},"61fd0a4e688a4d030fe2630d5a5121e24f617834":{"kind":"NO_CODE_CHANGE","_number":43,"created":"2016-10-03 06:11:32.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/43"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-10-03 06:11:25.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/61fd0a4e688a4d030fe2630d5a5121e24f617834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/61fd0a4e688a4d030fe2630d5a5121e24f617834"}]},"branch":"refs/heads/master"},"4b1e57f2fb8a913ee2f606872ee75064eab2d62d":{"kind":"NO_CODE_CHANGE","_number":44,"created":"2016-10-27 08:06:07.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/44"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-10-27 08:06:01.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I5022f42392c0a076aa4980082640392cad1688a0\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4b1e57f2fb8a913ee2f606872ee75064eab2d62d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4b1e57f2fb8a913ee2f606872ee75064eab2d62d"}]},"branch":"refs/heads/master"},"35e551a7c36d58314dc77485fe1e0fe9833e6bbf":{"kind":"NO_CODE_CHANGE","_number":45,"created":"2016-11-01 12:16:34.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/45"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-11-01 12:16:29.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/35e551a7c36d58314dc77485fe1e0fe9833e6bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/35e551a7c36d58314dc77485fe1e0fe9833e6bbf"}]},"branch":"refs/heads/master"},"3fe8caaa4fc5a7b3b28241fa221b60b902afa322":{"kind":"NO_CODE_CHANGE","_number":46,"created":"2016-12-14 00:10:06.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/46"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2016-12-14 00:09:57.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie64aa161f1a2bb17098d88db20195b7e84bb0eae\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3fe8caaa4fc5a7b3b28241fa221b60b902afa322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3fe8caaa4fc5a7b3b28241fa221b60b902afa322"}]},"branch":"refs/heads/master"},"e2b285aa97dadbf3896d770b77cb6c6fb5707dfb":{"kind":"NO_CODE_CHANGE","_number":47,"created":"2017-01-13 08:10:25.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/47"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-01-13 08:10:17.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e2b285aa97dadbf3896d770b77cb6c6fb5707dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e2b285aa97dadbf3896d770b77cb6c6fb5707dfb"}]},"branch":"refs/heads/master"},"ad3f8757aeccfb008815ecfe5939035355a56443":{"kind":"NO_CODE_CHANGE","_number":48,"created":"2017-01-13 09:29:39.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/48"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-01-13 09:29:35.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I2f166a43f0b980ad22617f8a3f7b4cc7f4786c48\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ad3f8757aeccfb008815ecfe5939035355a56443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ad3f8757aeccfb008815ecfe5939035355a56443"}]},"branch":"refs/heads/master"},"c2a534bf0a4040c8bbbe9a24730393868637f2dc":{"kind":"NO_CODE_CHANGE","_number":49,"created":"2017-01-17 03:14:16.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/49"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-01-17 03:13:50.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I946b23b5a9446d7263c64cf29bf3fd295ed6cf00\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c2a534bf0a4040c8bbbe9a24730393868637f2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c2a534bf0a4040c8bbbe9a24730393868637f2dc"}]},"branch":"refs/heads/master"},"f6bf90761c48175b239097cf3460fd538cdaf092":{"kind":"NO_CODE_CHANGE","_number":50,"created":"2017-01-23 22:30:55.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/50"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-01-23 22:30:50.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ibc480d8b8f09c978f874e1e74bd52e3d615a57f6\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f6bf90761c48175b239097cf3460fd538cdaf092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f6bf90761c48175b239097cf3460fd538cdaf092"}]},"branch":"refs/heads/master"},"155cbbe225c809b50690f088da61f5b80844d9cd":{"kind":"NO_CODE_CHANGE","_number":51,"created":"2017-01-24 06:58:19.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/51"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-01-24 06:58:13.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ibc480d8b8f09c978f874e1e74bd52e3d615a57f6\nDepends-On: I38626690320c8c7844fe653baba27ff9315b1f17\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/155cbbe225c809b50690f088da61f5b80844d9cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/155cbbe225c809b50690f088da61f5b80844d9cd"}]},"branch":"refs/heads/master"},"c5f6959fedc70c4c632b717119714fff83891ff4":{"kind":"NO_CODE_CHANGE","_number":52,"created":"2017-01-25 03:36:47.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/52"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-01-25 03:36:38.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I620f55a2e3e1031a196888c2192ac21ffcf886f9\nDepends-On: I38626690320c8c7844fe653baba27ff9315b1f17\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c5f6959fedc70c4c632b717119714fff83891ff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c5f6959fedc70c4c632b717119714fff83891ff4"}]},"branch":"refs/heads/master"},"018f8f552ba0b50584e853a21049a3c3888f33e5":{"kind":"NO_CODE_CHANGE","_number":53,"created":"2017-02-14 09:22:06.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/53"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-02-14 09:22:01.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I6a3d52b606e43a9d00ec1a7c7db86ee32c1511ca\nDepends-On: I4a1d8f5b2f100f86dd1ee7dca3983881fb86e143\nDepends-On: I620f55a2e3e1031a196888c2192ac21ffcf886f9\nDepends-On: I38626690320c8c7844fe653baba27ff9315b1f17\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/018f8f552ba0b50584e853a21049a3c3888f33e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/018f8f552ba0b50584e853a21049a3c3888f33e5"}]},"branch":"refs/heads/master"},"2e61de8e4d8453f5cf45411d259b9008e573260f":{"kind":"NO_CODE_CHANGE","_number":54,"created":"2017-02-16 02:49:07.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/54"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-02-16 02:49:03.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ib7d1c2081654dda82b5e015c71d20f378f3308f7\nDepends-On: I6a3d52b606e43a9d00ec1a7c7db86ee32c1511ca\nDepends-On: I4a1d8f5b2f100f86dd1ee7dca3983881fb86e143\nDepends-On: I620f55a2e3e1031a196888c2192ac21ffcf886f9\nDepends-On: I38626690320c8c7844fe653baba27ff9315b1f17\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2e61de8e4d8453f5cf45411d259b9008e573260f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2e61de8e4d8453f5cf45411d259b9008e573260f"}]},"branch":"refs/heads/master"},"62cb9aaf35642b2eecc8b277acbb42a5f817b5e4":{"kind":"NO_CODE_CHANGE","_number":55,"created":"2017-02-16 05:18:30.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/55"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-02-16 05:18:16.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie29acfbe323b60205ade9d660f7497f5bf4a35ca\nDepends-On: I6a3d52b606e43a9d00ec1a7c7db86ee32c1511ca\nDepends-On: I4a1d8f5b2f100f86dd1ee7dca3983881fb86e143\nDepends-On: I620f55a2e3e1031a196888c2192ac21ffcf886f9\nDepends-On: I38626690320c8c7844fe653baba27ff9315b1f17\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/62cb9aaf35642b2eecc8b277acbb42a5f817b5e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/62cb9aaf35642b2eecc8b277acbb42a5f817b5e4"}]},"branch":"refs/heads/master"},"e214ccccaef0800cfd0208804ee55b5a675486e8":{"kind":"NO_CODE_CHANGE","_number":56,"created":"2017-02-20 19:04:32.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/56"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-02-20 19:04:27.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie29acfbe323b60205ade9d660f7497f5bf4a35ca\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e214ccccaef0800cfd0208804ee55b5a675486e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e214ccccaef0800cfd0208804ee55b5a675486e8"}]},"branch":"refs/heads/master"},"4d6a4a38219d5db6c37b4e81cb1040772ef73227":{"kind":"NO_CODE_CHANGE","_number":57,"created":"2017-02-20 21:31:29.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/57"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-02-20 21:31:19.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie29acfbe323b60205ade9d660f7497f5bf4a35ca\nDepends-On: I425c65faf8ceb1261107c489e42ac92a7f978dca\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4d6a4a38219d5db6c37b4e81cb1040772ef73227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4d6a4a38219d5db6c37b4e81cb1040772ef73227"}]},"branch":"refs/heads/master"},"31d4748664c4430240c05c84cb2119270545121e":{"kind":"NO_CODE_CHANGE","_number":58,"created":"2017-02-22 16:06:53.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/58"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-02-22 16:06:47.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/31d4748664c4430240c05c84cb2119270545121e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/31d4748664c4430240c05c84cb2119270545121e"}]},"branch":"refs/heads/master"},"0383322941dac7cbe36bca07da28d6afc540a7b3":{"kind":"NO_CODE_CHANGE","_number":59,"created":"2017-04-04 05:55:31.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/59"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-04 05:55:24.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I17647e8a671fe7e5c6786647184dcd412938a003\nDepends-On: Iceaf178be2885b0299d6d281983e9f7f8ae6def0\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0383322941dac7cbe36bca07da28d6afc540a7b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0383322941dac7cbe36bca07da28d6afc540a7b3"}]},"branch":"refs/heads/master"},"c9ec1f0f33805d842e21cffe6a5bb36b801db896":{"kind":"NO_CODE_CHANGE","_number":60,"created":"2017-04-04 09:36:13.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/60"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-04 09:36:08.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I17647e8a671fe7e5c6786647184dcd412938a003\nDepends-On: Iceaf178be2885b0299d6d281983e9f7f8ae6def0\nDepends-On: I61bf6483267f1b6c7b2e014b860667c6a3f768b7\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c9ec1f0f33805d842e21cffe6a5bb36b801db896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c9ec1f0f33805d842e21cffe6a5bb36b801db896"}]},"branch":"refs/heads/master"},"79546dbc9671db7141f5b12636c012e2ad0a329c":{"kind":"NO_CODE_CHANGE","_number":61,"created":"2017-04-05 06:33:01.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/61"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-05 04:08:26.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I17647e8a671fe7e5c6786647184dcd412938a003\nDepends-On: Iceaf178be2885b0299d6d281983e9f7f8ae6def0\nDepends-On: I61bf6483267f1b6c7b2e014b860667c6a3f768b7\nDepends-On: Idabc7b6b4174575c89c8010886c2b440de6a8e05\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/79546dbc9671db7141f5b12636c012e2ad0a329c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/79546dbc9671db7141f5b12636c012e2ad0a329c"}]},"branch":"refs/heads/master"},"bf800601d3014395ee21e263acd6ef11cf6e2de9":{"kind":"NO_CODE_CHANGE","_number":62,"created":"2017-04-06 08:49:31.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/62"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-06 08:49:25.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Iee81f2ac700b70daebaaaa3476d612cf8f38c488\nDepends-On: I17647e8a671fe7e5c6786647184dcd412938a003\nDepends-On: Iceaf178be2885b0299d6d281983e9f7f8ae6def0\nDepends-On: I61bf6483267f1b6c7b2e014b860667c6a3f768b7\nDepends-On: Idabc7b6b4174575c89c8010886c2b440de6a8e05\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bf800601d3014395ee21e263acd6ef11cf6e2de9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bf800601d3014395ee21e263acd6ef11cf6e2de9"}]},"branch":"refs/heads/master"},"d180306664df4607390b0a925a0cab8ed914d614":{"kind":"NO_CODE_CHANGE","_number":63,"created":"2017-04-07 12:59:23.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/63"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-07 12:59:03.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Iee81f2ac700b70daebaaaa3476d612cf8f38c488\nDepends-On: I17647e8a671fe7e5c6786647184dcd412938a003\nDepends-On: I61bf6483267f1b6c7b2e014b860667c6a3f768b7\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d180306664df4607390b0a925a0cab8ed914d614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d180306664df4607390b0a925a0cab8ed914d614"}]},"branch":"refs/heads/master"},"839636858593fce116741784778e292cd58000f5":{"kind":"NO_CODE_CHANGE","_number":64,"created":"2017-04-10 08:42:52.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/64"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-10 08:42:48.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I3ca282437fcfde9aae9089bac4066db223f8774d\nDepends-On: Iee81f2ac700b70daebaaaa3476d612cf8f38c488\nDepends-On: I17647e8a671fe7e5c6786647184dcd412938a003\nDepends-On: I61bf6483267f1b6c7b2e014b860667c6a3f768b7\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/839636858593fce116741784778e292cd58000f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/839636858593fce116741784778e292cd58000f5"}]},"branch":"refs/heads/master"},"7d491e74da2eda111bb1d277fe13223722eab04e":{"kind":"NO_CODE_CHANGE","_number":65,"created":"2017-04-12 03:48:19.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/65"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-12 03:48:10.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Iee81f2ac700b70daebaaaa3476d612cf8f38c488\nDepends-On: I17647e8a671fe7e5c6786647184dcd412938a003\nDepends-On: I61bf6483267f1b6c7b2e014b860667c6a3f768b7\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7d491e74da2eda111bb1d277fe13223722eab04e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7d491e74da2eda111bb1d277fe13223722eab04e"}]},"branch":"refs/heads/master"},"29655ff3235c23866e75a26224fb10298060696c":{"kind":"NO_CODE_CHANGE","_number":66,"created":"2017-04-18 10:21:43.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/66"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-18 10:21:38.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I590ce615699c5e26abd47b84eedb27a9a6fd1455\nDepends-On: I0095e160481f1d4572e38ad1d3bbc8e183039b84\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/29655ff3235c23866e75a26224fb10298060696c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/29655ff3235c23866e75a26224fb10298060696c"}]},"branch":"refs/heads/master"},"0421aef3abfc03e6119650380811a6980bd8b630":{"kind":"NO_CODE_CHANGE","_number":67,"created":"2017-04-20 03:45:54.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/67"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-20 03:45:47.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I590ce615699c5e26abd47b84eedb27a9a6fd1455\nDepends-On: I0095e160481f1d4572e38ad1d3bbc8e183039b84\nDepends-On: Idfb9817e8e055eba437547e1aa69f34bcc6c8c89\nDepends-On: I2ae8a8f1f34e5105d002905230a43a8b04093de5\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0421aef3abfc03e6119650380811a6980bd8b630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0421aef3abfc03e6119650380811a6980bd8b630"}]},"branch":"refs/heads/master"},"39dc64bc0d2a19471d2172ed6a9a208360745c0c":{"kind":"NO_CODE_CHANGE","_number":68,"created":"2017-04-20 04:35:09.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/68"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-20 04:35:04.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I590ce615699c5e26abd47b84eedb27a9a6fd1455\nDepends-On: I0095e160481f1d4572e38ad1d3bbc8e183039b84\nDepends-On: Idfb9817e8e055eba437547e1aa69f34bcc6c8c89\nDepends-On: I2ae8a8f1f34e5105d002905230a43a8b04093de5\nDepends-On: I8460ca4b0156c9ab28d3e5b638a033931ca78e39\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/39dc64bc0d2a19471d2172ed6a9a208360745c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/39dc64bc0d2a19471d2172ed6a9a208360745c0c"}]},"branch":"refs/heads/master"},"2cfbe564cd04986e9e1aac3dcd2dbbd96b94786e":{"kind":"NO_CODE_CHANGE","_number":69,"created":"2017-04-25 04:31:50.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/69"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-25 04:31:40.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I0095e160481f1d4572e38ad1d3bbc8e183039b84\nDepends-On: Idfb9817e8e055eba437547e1aa69f34bcc6c8c89\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2cfbe564cd04986e9e1aac3dcd2dbbd96b94786e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2cfbe564cd04986e9e1aac3dcd2dbbd96b94786e"}]},"branch":"refs/heads/master"},"363b9a7f995d37537793f0d4066f1114008e2b99":{"kind":"NO_CODE_CHANGE","_number":70,"created":"2017-04-26 02:30:57.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/70"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-26 02:30:52.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Idfb9817e8e055eba437547e1aa69f34bcc6c8c89\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/363b9a7f995d37537793f0d4066f1114008e2b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/363b9a7f995d37537793f0d4066f1114008e2b99"}]},"branch":"refs/heads/master"},"d34ed081bb348111ba79952428cea0317b5d4ad3":{"kind":"NO_CODE_CHANGE","_number":71,"created":"2017-04-27 05:55:01.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/71"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-04-27 05:54:56.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Idfb9817e8e055eba437547e1aa69f34bcc6c8c89\nDepends-On: I8d996df4674dc40ef8965522bd6d744329b2fb3b\nDepends-On: I80696fa1227d1d36ae67d8f2b70e4206c4cdcd70\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d34ed081bb348111ba79952428cea0317b5d4ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d34ed081bb348111ba79952428cea0317b5d4ad3"}]},"branch":"refs/heads/master"},"e56625abcca45f786582612c8631722b882e495c":{"kind":"NO_CODE_CHANGE","_number":72,"created":"2017-05-01 04:15:47.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/72"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-05-01 04:15:38.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e56625abcca45f786582612c8631722b882e495c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e56625abcca45f786582612c8631722b882e495c"}]},"branch":"refs/heads/master"},"3187d17f1aff8cdfe2dc364cf2256ee467480e00":{"kind":"NO_CODE_CHANGE","_number":73,"created":"2017-05-08 07:53:16.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/73"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-05-08 07:53:11.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I36357e2ef967db3a73c2341903cd18f5109a006b\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3187d17f1aff8cdfe2dc364cf2256ee467480e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3187d17f1aff8cdfe2dc364cf2256ee467480e00"}]},"branch":"refs/heads/master"},"6be5afdeb0eca3df8a02e42b1db2f0fc92e85c24":{"kind":"NO_CODE_CHANGE","_number":74,"created":"2017-05-30 09:19:10.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/74"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-05-30 09:18:52.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6be5afdeb0eca3df8a02e42b1db2f0fc92e85c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6be5afdeb0eca3df8a02e42b1db2f0fc92e85c24"}]},"branch":"refs/heads/master"},"caeff4d526809c400f0af86687a0a39a8e563abf":{"kind":"NO_CODE_CHANGE","_number":75,"created":"2017-06-05 10:03:15.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/75"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-05 09:54:39.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I0629153f973daeb3bef3a6968360cbc2d427f9ad\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/caeff4d526809c400f0af86687a0a39a8e563abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/caeff4d526809c400f0af86687a0a39a8e563abf"}]},"branch":"refs/heads/master"},"a58531b4e565584f0b608dbd2d1333aa4ee3726f":{"kind":"NO_CODE_CHANGE","_number":76,"created":"2017-06-07 10:13:16.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/76"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-07 10:13:11.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I677721aeada6fd74e8201c7f308771c887b2a1e9\nDepends-On: I0629153f973daeb3bef3a6968360cbc2d427f9ad\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a58531b4e565584f0b608dbd2d1333aa4ee3726f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a58531b4e565584f0b608dbd2d1333aa4ee3726f"}]},"branch":"refs/heads/master"},"8884edf9dc080686235e47de382ebf72ac8d3285":{"kind":"NO_CODE_CHANGE","_number":77,"created":"2017-06-15 13:48:06.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/77"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-15 13:47:57.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I4452d61ee6e9d21add3d37ec39301efcbd3bd66d\nDepends-On: Ib5a8ffe63eaec15bc29bfdd133db7169507bab82\nDepends-On: I677721aeada6fd74e8201c7f308771c887b2a1e9\nDepends-On: I0629153f973daeb3bef3a6968360cbc2d427f9ad\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8884edf9dc080686235e47de382ebf72ac8d3285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8884edf9dc080686235e47de382ebf72ac8d3285"}]},"branch":"refs/heads/master"},"7c22412150b37cea597b4433070c460e7a6f6f99":{"kind":"NO_CODE_CHANGE","_number":78,"created":"2017-06-19 04:25:53.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/78"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-19 04:25:45.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I10a17d2910409781ba8317eba96fadb829801346\nDepends-On: I4452d61ee6e9d21add3d37ec39301efcbd3bd66d\nDepends-On: Ib5a8ffe63eaec15bc29bfdd133db7169507bab82\nDepends-On: I677721aeada6fd74e8201c7f308771c887b2a1e9\nDepends-On: I0629153f973daeb3bef3a6968360cbc2d427f9ad\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7c22412150b37cea597b4433070c460e7a6f6f99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7c22412150b37cea597b4433070c460e7a6f6f99"}]},"branch":"refs/heads/master"},"fe0f87f95cce93e82b656f5024b884f237b70bde":{"kind":"NO_CODE_CHANGE","_number":79,"created":"2017-06-19 06:58:09.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/79"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-19 06:58:04.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ic94d6e4944002b46125bac128e13a610cbc3a999\nDepends-On: I4452d61ee6e9d21add3d37ec39301efcbd3bd66d\nDepends-On: Ib5a8ffe63eaec15bc29bfdd133db7169507bab82\nDepends-On: I677721aeada6fd74e8201c7f308771c887b2a1e9\nDepends-On: I0629153f973daeb3bef3a6968360cbc2d427f9ad\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fe0f87f95cce93e82b656f5024b884f237b70bde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fe0f87f95cce93e82b656f5024b884f237b70bde"}]},"branch":"refs/heads/master"},"80a4dac11d48d6a336364797d4294b1171b657a6":{"kind":"NO_CODE_CHANGE","_number":80,"created":"2017-06-19 15:44:56.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/80"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-19 15:44:45.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ifef2561ef4ff2a44068fc008475b216fdabe7095\nDepends-On: Ic94d6e4944002b46125bac128e13a610cbc3a999\nDepends-On: I4452d61ee6e9d21add3d37ec39301efcbd3bd66d\nDepends-On: Ib5a8ffe63eaec15bc29bfdd133db7169507bab82\nDepends-On: I677721aeada6fd74e8201c7f308771c887b2a1e9\nDepends-On: I0629153f973daeb3bef3a6968360cbc2d427f9ad\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/80a4dac11d48d6a336364797d4294b1171b657a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/80a4dac11d48d6a336364797d4294b1171b657a6"}]},"branch":"refs/heads/master"},"f9c8fee7cb28788d521d5f753f7381bb842af791":{"kind":"NO_CODE_CHANGE","_number":81,"created":"2017-06-20 11:24:39.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/81"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-20 11:24:22.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ifef2561ef4ff2a44068fc008475b216fdabe7095\nDepends-On: Ic94d6e4944002b46125bac128e13a610cbc3a999\nDepends-On: I4452d61ee6e9d21add3d37ec39301efcbd3bd66d\nDepends-On: Ib5a8ffe63eaec15bc29bfdd133db7169507bab82\nDepends-On: I677721aeada6fd74e8201c7f308771c887b2a1e9\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f9c8fee7cb28788d521d5f753f7381bb842af791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f9c8fee7cb28788d521d5f753f7381bb842af791"}]},"branch":"refs/heads/master"},"f48251d526e8f605721a6ece460afe30549c6873":{"kind":"NO_CODE_CHANGE","_number":82,"created":"2017-06-25 10:57:41.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/82"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-25 10:57:36.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I467672cc0ba1e579cddfff6020658c82e80fa5af\nDepends-On: Ifef2561ef4ff2a44068fc008475b216fdabe7095\nDepends-On: Ic94d6e4944002b46125bac128e13a610cbc3a999\nDepends-On: I4452d61ee6e9d21add3d37ec39301efcbd3bd66d\nDepends-On: Ib5a8ffe63eaec15bc29bfdd133db7169507bab82\nDepends-On: I677721aeada6fd74e8201c7f308771c887b2a1e9\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f48251d526e8f605721a6ece460afe30549c6873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f48251d526e8f605721a6ece460afe30549c6873"}]},"branch":"refs/heads/master"},"a8edb0666f47f67cc0eb1ad20ad4238fc0736e0e":{"kind":"NO_CODE_CHANGE","_number":83,"created":"2017-06-26 04:55:36.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/83"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-26 04:55:32.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I677721aeada6fd74e8201c7f308771c887b2a1e9\nDepends-On: Iaa7121ced07f7877292e4ff15926bf02b5e7aaa1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a8edb0666f47f67cc0eb1ad20ad4238fc0736e0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a8edb0666f47f67cc0eb1ad20ad4238fc0736e0e"}]},"branch":"refs/heads/master"},"4f6c40e61dfc660522d37b2f715e3095159d38f7":{"kind":"NO_CODE_CHANGE","_number":84,"created":"2017-06-29 07:37:35.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/84"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-06-29 07:37:28.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4f6c40e61dfc660522d37b2f715e3095159d38f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4f6c40e61dfc660522d37b2f715e3095159d38f7"}]},"branch":"refs/heads/master"},"7acb4990389a0cc9ff1bbd2a6dd2088b0b7aa896":{"kind":"NO_CODE_CHANGE","_number":85,"created":"2017-07-05 12:44:27.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/85"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-07-05 12:44:21.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I1cbbb201f494502a5fe3167eb66a1dbb389f6427\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7acb4990389a0cc9ff1bbd2a6dd2088b0b7aa896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7acb4990389a0cc9ff1bbd2a6dd2088b0b7aa896"}]},"branch":"refs/heads/master"},"c6c28aaeac767d4b9dfcc113dc9f16ab5a4472b9":{"kind":"NO_CODE_CHANGE","_number":86,"created":"2017-07-10 05:16:25.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/86"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-07-10 05:16:17.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I4f1f399ac8e1c9acec9c7bce2f57c053c98e3031\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c6c28aaeac767d4b9dfcc113dc9f16ab5a4472b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c6c28aaeac767d4b9dfcc113dc9f16ab5a4472b9"}]},"branch":"refs/heads/master"},"e51581fd06b6db17a721a9d5c2a32edf6d2f6bee":{"kind":"NO_CODE_CHANGE","_number":87,"created":"2017-08-07 00:52:44.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/87"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-08-07 00:52:35.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: If64c20f10f963f372bdc7c9c6c50361425d75ab1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e51581fd06b6db17a721a9d5c2a32edf6d2f6bee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e51581fd06b6db17a721a9d5c2a32edf6d2f6bee"}]},"branch":"refs/heads/master"},"9bdd5695c98850e92e255e7cd5cb890c620884a9":{"kind":"NO_CODE_CHANGE","_number":88,"created":"2017-08-07 23:51:56.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/88"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-08-07 23:51:51.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9bdd5695c98850e92e255e7cd5cb890c620884a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9bdd5695c98850e92e255e7cd5cb890c620884a9"}]},"branch":"refs/heads/master"},"d8ea43160d751ea7c1dc9bc9eb851b0972dfd586":{"kind":"NO_CODE_CHANGE","_number":89,"created":"2017-10-17 05:59:33.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/89"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-10-17 05:59:23.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I31ac7679e72749dbb89e925f0bab1daa5677c079\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d8ea43160d751ea7c1dc9bc9eb851b0972dfd586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d8ea43160d751ea7c1dc9bc9eb851b0972dfd586"}]},"branch":"refs/heads/master"},"f9a3e0d0c10a014e794479409779fa7dd141a31a":{"kind":"NO_CODE_CHANGE","_number":90,"created":"2017-10-17 09:43:58.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/90"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-10-17 09:43:53.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ib2227e332c52d2f7225c41f57108c38641348183\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f9a3e0d0c10a014e794479409779fa7dd141a31a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f9a3e0d0c10a014e794479409779fa7dd141a31a"}]},"branch":"refs/heads/master"},"66435ff4f2613710648e89098111478bd62976cf":{"kind":"NO_CODE_CHANGE","_number":91,"created":"2017-10-23 07:00:45.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/91"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-10-23 07:00:34.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/66435ff4f2613710648e89098111478bd62976cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/66435ff4f2613710648e89098111478bd62976cf"}]},"branch":"refs/heads/master"},"dd3b025f1f6d62d11896fa862d7e95c7472b2951":{"kind":"NO_CODE_CHANGE","_number":92,"created":"2017-10-25 05:31:36.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/92"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-10-25 05:31:26.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie8fe87bc4b0b36dcf8b7c042149fbe8658e385b1\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/dd3b025f1f6d62d11896fa862d7e95c7472b2951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/dd3b025f1f6d62d11896fa862d7e95c7472b2951"}]},"branch":"refs/heads/master"},"54315369800077b85484273e48493390b1444497":{"kind":"NO_CODE_CHANGE","_number":93,"created":"2017-10-31 02:09:38.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/93"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-10-31 02:09:33.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/54315369800077b85484273e48493390b1444497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/54315369800077b85484273e48493390b1444497"}]},"branch":"refs/heads/master"},"95bddf445ecb6e66b47363ca8cfd6d2faa24ddae":{"kind":"NO_CODE_CHANGE","_number":94,"created":"2017-11-01 06:22:21.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/94"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-11-01 06:22:14.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I06ae0c42d983e88e1f38c501d5c85a7956f195ad\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/95bddf445ecb6e66b47363ca8cfd6d2faa24ddae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/95bddf445ecb6e66b47363ca8cfd6d2faa24ddae"}]},"branch":"refs/heads/master"},"94e530b04d31345dc759c28dbfca4efdb7c90ea6":{"kind":"NO_CODE_CHANGE","_number":95,"created":"2017-11-06 04:06:39.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/95"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-11-06 04:06:30.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/94e530b04d31345dc759c28dbfca4efdb7c90ea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/94e530b04d31345dc759c28dbfca4efdb7c90ea6"}]},"branch":"refs/heads/master"},"55e1d4457c01ff533fa27de3cb323cc00e025b36":{"kind":"NO_CODE_CHANGE","_number":96,"created":"2017-11-06 05:20:48.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/96"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-11-06 05:20:41.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Id33dc37118f57a0dbbfbe5ee71d1c7a48eefeb3e\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/55e1d4457c01ff533fa27de3cb323cc00e025b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/55e1d4457c01ff533fa27de3cb323cc00e025b36"}]},"branch":"refs/heads/master"},"8273309c1c1da42fc89d489270987ceb5b72714f":{"kind":"NO_CODE_CHANGE","_number":97,"created":"2017-11-10 09:59:47.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/97"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-11-10 09:59:42.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8273309c1c1da42fc89d489270987ceb5b72714f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8273309c1c1da42fc89d489270987ceb5b72714f"}]},"branch":"refs/heads/master"},"424ecfde4547e0dd2789300414511f2a5cf6d0a5":{"kind":"NO_CODE_CHANGE","_number":98,"created":"2017-11-11 13:23:44.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/98"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-11-11 13:23:39.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ie5c773165948d6db55f1b7ba3b7e312a5bb1318b\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/424ecfde4547e0dd2789300414511f2a5cf6d0a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/424ecfde4547e0dd2789300414511f2a5cf6d0a5"}]},"branch":"refs/heads/master"},"83323b337b5f5c1d7325bccd8843ac051917e9ab":{"kind":"NO_CODE_CHANGE","_number":99,"created":"2017-11-14 10:39:41.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/99"}}},"commit":{"parents":[{"commit":"a45b4a99d4b86365e89bad49d21e41f023cb08b5","subject":"Merge \"devstackgaterc: Enable tests from Neutron tempest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a45b4a99d4b86365e89bad49d21e41f023cb08b5"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-11-14 10:39:36.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/83323b337b5f5c1d7325bccd8843ac051917e9ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/83323b337b5f5c1d7325bccd8843ac051917e9ab"}]},"branch":"refs/heads/master"},"2b3f9071aa0050da6c1f37d260ecb38435911c04":{"kind":"REWORK","_number":100,"created":"2017-11-29 10:06:35.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/100"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2017-11-29 10:06:26.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2b3f9071aa0050da6c1f37d260ecb38435911c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2b3f9071aa0050da6c1f37d260ecb38435911c04"}]},"branch":"refs/heads/master"},"0827a703748b4258bf31cbc21b96f333eb365d2c":{"kind":"NO_CODE_CHANGE","_number":101,"created":"2018-01-11 14:18:48.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/101","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/101","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/101 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/101 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/101 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/101"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-01-11 14:18:42.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Id67037252aafbacbe43f6af786f30a687321952d\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0827a703748b4258bf31cbc21b96f333eb365d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0827a703748b4258bf31cbc21b96f333eb365d2c"}]},"branch":"refs/heads/master"},"738418be2d1dde57f5538bf11f6a14d0dfab6b8f":{"kind":"NO_CODE_CHANGE","_number":102,"created":"2018-01-16 05:01:33.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/102","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/102","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/102 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/102 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/102 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/102"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-01-16 05:01:27.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Id67037252aafbacbe43f6af786f30a687321952d\nDepends-On: I04fe57630d902f9aae3bb3405619d89c837a8564\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/738418be2d1dde57f5538bf11f6a14d0dfab6b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/738418be2d1dde57f5538bf11f6a14d0dfab6b8f"}]},"branch":"refs/heads/master"},"589934f49e8595dc5eb60d9f3060f15a8a47ae1a":{"kind":"NO_CODE_CHANGE","_number":103,"created":"2018-01-26 10:00:30.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/103","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/103","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/103 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/103 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/103 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/103"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-01-26 10:00:25.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/589934f49e8595dc5eb60d9f3060f15a8a47ae1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/589934f49e8595dc5eb60d9f3060f15a8a47ae1a"}]},"branch":"refs/heads/master"},"9f9e42a5c3f9cbf0b1cb7cd5c5e52ebf7bea2ee6":{"kind":"NO_CODE_CHANGE","_number":104,"created":"2018-01-30 15:08:20.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/104","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/104","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/104 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/104 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/104 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/104"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-01-30 15:08:15.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ic96f31dce81ecf4db2daa2e9cba649eaff8f81d6\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9f9e42a5c3f9cbf0b1cb7cd5c5e52ebf7bea2ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9f9e42a5c3f9cbf0b1cb7cd5c5e52ebf7bea2ee6"}]},"branch":"refs/heads/master"},"ce7f12b7380fdd14723c3396e7fcc7b8168ea815":{"kind":"NO_CODE_CHANGE","_number":105,"created":"2018-02-08 22:34:33.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/105","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/105","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/105 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/105 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/105 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/105"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-02-08 22:34:05.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I15ec238c18a464f977f7d1079605b82965052311\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ce7f12b7380fdd14723c3396e7fcc7b8168ea815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ce7f12b7380fdd14723c3396e7fcc7b8168ea815"}]},"branch":"refs/heads/master"},"28bbdea2b208511972008504a3fb655d266a1ad1":{"kind":"NO_CODE_CHANGE","_number":106,"created":"2018-02-21 07:25:26.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/106","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/106","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/106 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/106 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/106 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/106"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-02-21 07:25:15.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/28bbdea2b208511972008504a3fb655d266a1ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/28bbdea2b208511972008504a3fb655d266a1ad1"}]},"branch":"refs/heads/master"},"571cfee7a703b7919638c621b8ac396342638c34":{"kind":"NO_CODE_CHANGE","_number":107,"created":"2018-03-19 05:13:31.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/107","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/107","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/107 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/107 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/107 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/107"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-03-19 05:13:24.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/#/c/553035/\nDepends-On: https://review.openstack.org/#/c/553817/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/571cfee7a703b7919638c621b8ac396342638c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/571cfee7a703b7919638c621b8ac396342638c34"}]},"branch":"refs/heads/master"},"2070535e907221d0f159146016467dcdcbbd3962":{"kind":"NO_CODE_CHANGE","_number":108,"created":"2018-03-22 03:09:08.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/108","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/108","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/108 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/108 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/108 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/108"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-03-22 03:09:02.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2070535e907221d0f159146016467dcdcbbd3962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2070535e907221d0f159146016467dcdcbbd3962"}]},"branch":"refs/heads/master"},"b2f200765456d8d3b72199915b4cca03fd374349":{"kind":"NO_CODE_CHANGE","_number":109,"created":"2018-03-23 08:45:25.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/109","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/109","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/109 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/109 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/109 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/109"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-03-23 08:45:19.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/555699\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b2f200765456d8d3b72199915b4cca03fd374349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b2f200765456d8d3b72199915b4cca03fd374349"}]},"branch":"refs/heads/master"},"37ec3452f3bde1cf2c3882e8141b81ce1283d0d1":{"kind":"NO_CODE_CHANGE","_number":110,"created":"2018-03-23 09:16:05.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/110","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/110","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/110 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/110 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/110 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/110"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-03-23 09:16:00.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/555307\nDepends-On: https://review.openstack.org/555699\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/37ec3452f3bde1cf2c3882e8141b81ce1283d0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/37ec3452f3bde1cf2c3882e8141b81ce1283d0d1"}]},"branch":"refs/heads/master"},"ac3bb8ee639b55b70592c7b1b23db52031170a2d":{"kind":"NO_CODE_CHANGE","_number":111,"created":"2018-03-28 01:02:04.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/111","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/111","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/111 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/111 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/111 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/111"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-03-28 01:01:55.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ac3bb8ee639b55b70592c7b1b23db52031170a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ac3bb8ee639b55b70592c7b1b23db52031170a2d"}]},"branch":"refs/heads/master"},"ef55da40625e9d966aee1a34b7258ba2e241cd10":{"kind":"NO_CODE_CHANGE","_number":112,"created":"2018-03-28 12:28:09.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/112","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/112","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/112 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/112 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/112 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/112"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-03-28 12:27:49.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/#/c/557127/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ef55da40625e9d966aee1a34b7258ba2e241cd10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ef55da40625e9d966aee1a34b7258ba2e241cd10"}]},"branch":"refs/heads/master"},"beb207fd43b7464dbb02b27fd3c079766197f718":{"kind":"NO_CODE_CHANGE","_number":113,"created":"2018-05-28 10:55:57.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/113","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/113","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/113 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/113 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/113 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/113"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-05-28 10:55:53.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/beb207fd43b7464dbb02b27fd3c079766197f718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/beb207fd43b7464dbb02b27fd3c079766197f718"}]},"branch":"refs/heads/master"},"08f59fe5952acf719fd35953c0f0ac6ff6b9321f":{"kind":"NO_CODE_CHANGE","_number":114,"created":"2018-06-08 16:21:28.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/114","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/114","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/114 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/114 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/114 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/114"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-06-08 16:21:23.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/#/c/414251\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/08f59fe5952acf719fd35953c0f0ac6ff6b9321f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/08f59fe5952acf719fd35953c0f0ac6ff6b9321f"}]},"branch":"refs/heads/master"},"7ce8e3a0ec0171983dd75dfb0e82f7447566cb5e":{"kind":"NO_CODE_CHANGE","_number":115,"created":"2018-06-21 08:20:52.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/115","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/115","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/115 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/115 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/115 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/115"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-06-21 08:20:46.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/#/c/414251\nDepends-On: https://review.openstack.org/#/c/577028\nDepends-On: https://review.openstack.org/#/c/577029\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7ce8e3a0ec0171983dd75dfb0e82f7447566cb5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7ce8e3a0ec0171983dd75dfb0e82f7447566cb5e"}]},"branch":"refs/heads/master"},"9cc68799312ec9dc2d37ee651197835cd6b259c9":{"kind":"NO_CODE_CHANGE","_number":116,"created":"2018-07-02 07:12:40.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/116","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/116","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/116 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/116 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/116 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/116"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-07-02 07:12:31.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9cc68799312ec9dc2d37ee651197835cd6b259c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9cc68799312ec9dc2d37ee651197835cd6b259c9"}]},"branch":"refs/heads/master"},"75c88f487aa594f27fa5d2f45d30eaaa15d98afa":{"kind":"NO_CODE_CHANGE","_number":117,"created":"2018-07-10 10:21:43.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/117","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/117","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/117 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/117 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/117 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/117"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-07-10 10:21:36.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Iebcadda5186f2f577c993276090aba2e99d13780\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/75c88f487aa594f27fa5d2f45d30eaaa15d98afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/75c88f487aa594f27fa5d2f45d30eaaa15d98afa"}]},"branch":"refs/heads/master"},"8ab1f152918db6ceb2502f504bdf45ec047cd932":{"kind":"NO_CODE_CHANGE","_number":118,"created":"2018-07-10 10:36:21.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/118","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/118","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/118 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/118 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/118 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/118"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-07-10 10:36:06.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8ab1f152918db6ceb2502f504bdf45ec047cd932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8ab1f152918db6ceb2502f504bdf45ec047cd932"}]},"branch":"refs/heads/master"},"02950aaeeac9604791dd2fbc481969a8a69b0701":{"kind":"NO_CODE_CHANGE","_number":120,"created":"2018-07-30 06:13:12.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/120","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/120","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/120 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/120 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/120 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/120"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-07-30 06:13:07.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/#/c/586911/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/02950aaeeac9604791dd2fbc481969a8a69b0701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/02950aaeeac9604791dd2fbc481969a8a69b0701"}]},"branch":"refs/heads/master"},"46218e6b32e9ae10fac136f89d1338f7c47598e2":{"kind":"NO_CODE_CHANGE","_number":121,"created":"2018-08-27 00:37:48.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/121","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/121","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/121 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/121 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/121 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/121"}}},"commit":{"parents":[{"commit":"b9a9c1334fe9cd3d65d822792583d88a6dc32121","subject":"Merge \"devstack: Bump the heap size of midonet-cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b9a9c1334fe9cd3d65d822792583d88a6dc32121"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-08-27 00:37:42.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/46218e6b32e9ae10fac136f89d1338f7c47598e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/46218e6b32e9ae10fac136f89d1338f7c47598e2"}]},"branch":"refs/heads/master"},"be1f026bd7d8fed2746dd3d8a7a29d34b89757b4":{"kind":"REWORK","_number":122,"created":"2018-10-10 03:32:55.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/122","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/122","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/122 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/122 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/122 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/122"}}},"commit":{"parents":[{"commit":"4d2e0008e3382c6891fe0d7df581b73314b93a5f","subject":"Merge \"Make ml2-full job multinode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4d2e0008e3382c6891fe0d7df581b73314b93a5f"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-10-10 03:32:46.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/be1f026bd7d8fed2746dd3d8a7a29d34b89757b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/be1f026bd7d8fed2746dd3d8a7a29d34b89757b4"}]},"branch":"refs/heads/master"},"3c122dab2256b3f7188b5915757f4bfedc9ddf9a":{"kind":"REWORK","_number":123,"created":"2018-10-11 05:45:55.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/123","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/123","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/123 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/123 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/123 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/123"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-10-11 05:45:45.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3c122dab2256b3f7188b5915757f4bfedc9ddf9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3c122dab2256b3f7188b5915757f4bfedc9ddf9a"}]},"branch":"refs/heads/master"},"cfa2e0585c3d2aa4730a0aaff1f38fdccd68fc5c":{"kind":"NO_CODE_CHANGE","_number":124,"created":"2018-10-11 05:46:56.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/124","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/124","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/124 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/124 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/124 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/124"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-10-11 05:46:51.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I04fe57630d902f9aae3bb3405619d89c837a8564\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/cfa2e0585c3d2aa4730a0aaff1f38fdccd68fc5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/cfa2e0585c3d2aa4730a0aaff1f38fdccd68fc5c"}]},"branch":"refs/heads/master"},"be4876e97156b38d0e7a11eaecbb4e1ac49d2cfe":{"kind":"NO_CODE_CHANGE","_number":125,"created":"2018-11-05 05:14:44.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/125","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/125","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/125 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/125 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/125 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/125"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-11-05 05:14:38.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/be4876e97156b38d0e7a11eaecbb4e1ac49d2cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/be4876e97156b38d0e7a11eaecbb4e1ac49d2cfe"}]},"branch":"refs/heads/master"},"f589a6058c07a6c49e993961362bb862dce6d81b":{"kind":"NO_CODE_CHANGE","_number":126,"created":"2018-11-09 03:36:18.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/126","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/126","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/126 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/126 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/126 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/126"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-11-09 03:36:09.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: Ib865c894f24fc1eccce7c459fb485937487215db\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f589a6058c07a6c49e993961362bb862dce6d81b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f589a6058c07a6c49e993961362bb862dce6d81b"}]},"branch":"refs/heads/master"},"f600d7f066c2fd61f5a6f9e241f272bacc58a3f2":{"kind":"NO_CODE_CHANGE","_number":127,"created":"2018-11-09 07:05:15.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/127","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/127","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/127 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/127 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/127 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/127"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2018-11-09 07:05:10.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I21a068b73e366ea5c815ea14249cd3b3c6725315\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f600d7f066c2fd61f5a6f9e241f272bacc58a3f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f600d7f066c2fd61f5a6f9e241f272bacc58a3f2"}]},"branch":"refs/heads/master"},"689f39b751dd28e7d123bd9490e0dacb994abc4f":{"kind":"NO_CODE_CHANGE","_number":128,"created":"2019-03-05 11:07:35.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/128","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/128","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/128 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/128 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/128 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/128"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-03-05 11:07:29.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: I40a6d48ab80ad73501de67317712937e02bccee3\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/689f39b751dd28e7d123bd9490e0dacb994abc4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/689f39b751dd28e7d123bd9490e0dacb994abc4f"}]},"branch":"refs/heads/master"},"85f99527f4e68861ec1f027a7fd33f0e72e66e9f":{"kind":"NO_CODE_CHANGE","_number":129,"created":"2019-03-13 02:48:57.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/129","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/129","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/129 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/129 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/129 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/129"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-03-13 02:48:52.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/85f99527f4e68861ec1f027a7fd33f0e72e66e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/85f99527f4e68861ec1f027a7fd33f0e72e66e9f"}]},"branch":"refs/heads/master"},"8c8497a5db832c104524e793a27e69e31f79c591":{"kind":"NO_CODE_CHANGE","_number":130,"created":"2019-04-08 02:14:35.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/130","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/130","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/130 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/130 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/130 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/130"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-04-08 02:14:30.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/#/c/650599/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8c8497a5db832c104524e793a27e69e31f79c591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8c8497a5db832c104524e793a27e69e31f79c591"}]},"branch":"refs/heads/master"},"e001e34e95db466eb36b2e8637d335fa3bb6e6b8":{"kind":"NO_CODE_CHANGE","_number":131,"created":"2019-04-11 10:42:05.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/131","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/131","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/131 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/131 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/131 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/131"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-04-11 10:41:59.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e001e34e95db466eb36b2e8637d335fa3bb6e6b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e001e34e95db466eb36b2e8637d335fa3bb6e6b8"}]},"branch":"refs/heads/master"},"6be9e3f3c1480674eb7aca1215e0315a06da6651":{"kind":"NO_CODE_CHANGE","_number":132,"created":"2019-04-16 07:03:29.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/132","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/132","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/132 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/132 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/132 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/132"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-04-16 07:03:25.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.openstack.org/#/c/651781/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6be9e3f3c1480674eb7aca1215e0315a06da6651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6be9e3f3c1480674eb7aca1215e0315a06da6651"}]},"branch":"refs/heads/master"},"9589059aadbe2554297fe5a44798fa07263e7746":{"kind":"NO_CODE_CHANGE","_number":133,"created":"2019-04-22 01:43:30.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/133","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/133","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/133 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/133 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/133 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/133"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-04-22 01:43:25.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.opendev.org/#/c/651781/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9589059aadbe2554297fe5a44798fa07263e7746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9589059aadbe2554297fe5a44798fa07263e7746"}]},"branch":"refs/heads/master"},"a9bdcc014a8b921fd92f9a92520a9e62f7a2c992":{"kind":"NO_CODE_CHANGE","_number":134,"created":"2019-04-29 14:02:54.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/134","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/134","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/134 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/134 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/134 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/134"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-04-29 14:02:49.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a9bdcc014a8b921fd92f9a92520a9e62f7a2c992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a9bdcc014a8b921fd92f9a92520a9e62f7a2c992"}]},"branch":"refs/heads/master"},"4f2105ddf516014511cfa9a08e4dcff6b7d4d6ec":{"kind":"NO_CODE_CHANGE","_number":135,"created":"2019-05-15 13:54:33.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/135","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/135","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/135 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/135 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/135 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/135"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-05-15 13:54:27.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.opendev.org/#/c/659294/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4f2105ddf516014511cfa9a08e4dcff6b7d4d6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4f2105ddf516014511cfa9a08e4dcff6b7d4d6ec"}]},"branch":"refs/heads/master"},"31a73265fde310b5fbfcb0dd074190edfba1cc5d":{"kind":"NO_CODE_CHANGE","_number":136,"created":"2019-06-06 07:46:47.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/136","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/136","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/136 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/136 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/136 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/136"}}},"commit":{"parents":[{"commit":"8e5f538811a21e4ef07e7da7b0b15aa98da4e77e","subject":"Merge \"Remove extra publish-openstack-python-branch-tarball job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8e5f538811a21e4ef07e7da7b0b15aa98da4e77e"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-06-06 07:46:43.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/31a73265fde310b5fbfcb0dd074190edfba1cc5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/31a73265fde310b5fbfcb0dd074190edfba1cc5d"}]},"branch":"refs/heads/master"},"ab1a22f282b73a5b7a80247c61f5357293446cd0":{"kind":"REWORK","_number":137,"created":"2019-07-12 04:23:41.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/137","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/137","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/137 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/137 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/137 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/137"}}},"commit":{"parents":[{"commit":"fa99271e8fcc45664388c3c8b35770fdd14a0635","subject":"Merge \"Remove ocata jobs from branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fa99271e8fcc45664388c3c8b35770fdd14a0635"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-07-12 04:23:29.000000000","tz":540},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ab1a22f282b73a5b7a80247c61f5357293446cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/ab1a22f282b73a5b7a80247c61f5357293446cd0"}]},"branch":"refs/heads/master"},"4de49ad57bcece1b3dcb31ea57c735d87aaadac7":{"kind":"NO_CODE_CHANGE","_number":138,"created":"2019-08-07 07:24:53.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/138","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/138","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/138 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/138 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/138 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/138"}}},"commit":{"parents":[{"commit":"fa99271e8fcc45664388c3c8b35770fdd14a0635","subject":"Merge \"Remove ocata jobs from branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fa99271e8fcc45664388c3c8b35770fdd14a0635"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-08-07 07:24:47.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.opendev.org/#/c/675044/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4de49ad57bcece1b3dcb31ea57c735d87aaadac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4de49ad57bcece1b3dcb31ea57c735d87aaadac7"}]},"branch":"refs/heads/master"},"f6857f2deacdbe73aa3a923e8814e505f98aec53":{"kind":"NO_CODE_CHANGE","_number":139,"created":"2019-08-08 08:23:14.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/139","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/139","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/139 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/139 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/139 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/139"}}},"commit":{"parents":[{"commit":"fa99271e8fcc45664388c3c8b35770fdd14a0635","subject":"Merge \"Remove ocata jobs from branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fa99271e8fcc45664388c3c8b35770fdd14a0635"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-08-08 08:23:08.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f6857f2deacdbe73aa3a923e8814e505f98aec53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f6857f2deacdbe73aa3a923e8814e505f98aec53"}]},"branch":"refs/heads/master"},"45836cd260a3397e85ee13df97f98980a8dd9ede":{"kind":"NO_CODE_CHANGE","_number":140,"created":"2019-08-14 11:05:39.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/140","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/140","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/140 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/140 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/140 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/140"}}},"commit":{"parents":[{"commit":"fa99271e8fcc45664388c3c8b35770fdd14a0635","subject":"Merge \"Remove ocata jobs from branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fa99271e8fcc45664388c3c8b35770fdd14a0635"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-08-14 11:05:29.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.opendev.org/#/c/676388\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/45836cd260a3397e85ee13df97f98980a8dd9ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/45836cd260a3397e85ee13df97f98980a8dd9ede"}]},"branch":"refs/heads/master"},"bc536ca766f44c67fb01eabe1181e286786e9f54":{"kind":"NO_CODE_CHANGE","_number":141,"created":"2019-08-14 15:29:37.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/141","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/141","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/141 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/141 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/141 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/141"}}},"commit":{"parents":[{"commit":"fa99271e8fcc45664388c3c8b35770fdd14a0635","subject":"Merge \"Remove ocata jobs from branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fa99271e8fcc45664388c3c8b35770fdd14a0635"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-08-14 15:29:28.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\nDepends-On: https://review.opendev.org/#/c/675021/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bc536ca766f44c67fb01eabe1181e286786e9f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bc536ca766f44c67fb01eabe1181e286786e9f54"}]},"branch":"refs/heads/master"},"9d21c8e231e0f186e2be721a811a92e957a5e2ff":{"kind":"NO_CODE_CHANGE","_number":142,"created":"2019-10-08 05:13:45.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/87/199387/142","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/87/199387/142","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/142 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/142 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/142 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/87/199387/142"}}},"commit":{"parents":[{"commit":"fa99271e8fcc45664388c3c8b35770fdd14a0635","subject":"Merge \"Remove ocata jobs from branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fa99271e8fcc45664388c3c8b35770fdd14a0635"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2015-07-08 03:21:01.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-10-08 05:13:31.000000000","tz":0},"subject":"TEST FOR CI DO NOT REVIEW","message":"TEST FOR CI DO NOT REVIEW\n\n// Depends-On: https://review.opendev.org/#/c/675021/\nChange-Id: Ifd462e0fac8f6b3237766a29f928152c9669133b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9d21c8e231e0f186e2be721a811a92e957a5e2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9d21c8e231e0f186e2be721a811a92e957a5e2ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
