)]}'
{"id":"openstack%2Fnetworking-midonet~660594","triplet_id":"openstack%2Fnetworking-midonet~master~I26b8bd324daca07dfd0e8874bc26883c1d4f080c","project":"openstack/networking-midonet","branch":"master","topic":"opendev","hashtags":[],"change_id":"I26b8bd324daca07dfd0e8874bc26883c1d4f080c","subject":"Update the rest of github URLs","status":"MERGED","created":"2019-05-22 08:03:00.000000000","updated":"2019-05-28 07:41:39.000000000","submitted":"2019-05-28 07:41:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660594-1559029299452-b5105cdb","meta_rev_id":"c726a0ebaa7e0df2a4f15968fb9a82d05fb1924b","_number":660594,"virtual_id_number":660594,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":2,"date":"2019-05-28 07:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"all":[{"value":2,"date":"2019-05-28 07:25:12.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-05-22 08:43:39.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"all":[{"value":1,"date":"2019-05-28 07:25:12.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 08:43:39.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2019-05-28 07:25:12.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-05-28 07:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf003a5444ab1b04730ab63b01533707b0b0e0b5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-05-22 08:03:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd9ade438914ee89755f3d63378ba71a03ee71bb","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2019-05-22 08:43:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a2a3c59b65123dd993bcb8b0f15b211f758fdd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 09:37:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/94/660594/1/check/openstack-tox-pep8/2f82d64/ : SUCCESS in 8m 00s\n- openstack-tox-py27 http://logs.openstack.org/94/660594/1/check/openstack-tox-py27/775b11c/ : SUCCESS in 11m 14s\n- openstack-tox-py35 http://logs.openstack.org/94/660594/1/check/openstack-tox-py35/9bf291c/ : SUCCESS in 12m 30s\n- openstack-tox-py36 http://logs.openstack.org/94/660594/1/check/openstack-tox-py36/4c55d18/ : SUCCESS in 12m 35s\n- openstack-tox-docs http://logs.openstack.org/94/660594/1/check/openstack-tox-docs/b57f3a8/html/ : SUCCESS in 5m 11s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/94/660594/1/check/networking-midonet-tempest-multinode-ml2/7f07fb6/ : FAILURE in 16m 49s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/94/660594/1/check/networking-midonet-tempest-multinode-ml2-full/1cbc228/ : FAILURE in 16m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/94/660594/1/check/networking-midonet-tempest-aio-ml2-full-legacy/02ad8cb/ : FAILURE in 9m 51s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/94/660594/1/check/networking-midonet-tempest-aio-ml2-centos-7/4d07ed8/ : SUCCESS in 1h 27m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/94/660594/1/check/networking-midonet-tempest-aio-ml2-full-centos-7/87bf92e/ : FAILURE in 40m 31s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/94/660594/1/check/networking-midonet-grenade-ml2/96f0ebf/ : FAILURE in 15m 40s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/94/660594/1/check/networking-midonet-rally-ml2/f874089/ : FAILURE in 12m 30s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/94/660594/1/check/openstack-tox-cover/15c0b43/cover/ : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b174a96679ebd5ed73399b54c244e8b61486f4d3","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-05-28 07:25:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8ddd499e28f6626aa8a5827740b41bdb5565f48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:25:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2d33eec2ce6bd0a2457c903c18c6cfa98308679","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:41:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/94/660594/1/gate/openstack-tox-pep8/e17ab63/ : SUCCESS in 7m 30s\n- openstack-tox-py27 http://logs.openstack.org/94/660594/1/gate/openstack-tox-py27/6a65bda/ : SUCCESS in 11m 21s\n- openstack-tox-py35 http://logs.openstack.org/94/660594/1/gate/openstack-tox-py35/2f382f0/ : SUCCESS in 11m 37s\n- openstack-tox-py36 http://logs.openstack.org/94/660594/1/gate/openstack-tox-py36/4969bd5/ : SUCCESS in 10m 47s\n- openstack-tox-docs http://logs.openstack.org/94/660594/1/gate/openstack-tox-docs/1bc2036/html/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b617c8a5b2d4a0d2e80db02587194d1f3e2cf7c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:41:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd592c7da02d588da7387d4cb4fdeaf4ba619269","revisions":{"cd592c7da02d588da7387d4cb4fdeaf4ba619269":{"kind":"REWORK","_number":1,"created":"2019-05-22 08:03:00.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/94/660594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/94/660594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/94/660594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/94/660594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/94/660594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/94/660594/1"}}},"commit":{"parents":[{"commit":"66c93dc3dac846f8d4b4ec6e26140d458ae59812","subject":"Use opendev repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/66c93dc3dac846f8d4b4ec6e26140d458ae59812"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-05-22 07:51:14.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-05-22 08:01:28.000000000","tz":540},"subject":"Update the rest of github URLs","message":"Update the rest of github URLs\n\nRemove or migrate to opendev.org.\nI left ones under specs/ as they are historical.\n\nChange-Id: I26b8bd324daca07dfd0e8874bc26883c1d4f080c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/cd592c7da02d588da7387d4cb4fdeaf4ba619269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/cd592c7da02d588da7387d4cb4fdeaf4ba619269"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
