)]}'
{"id":"openstack%2Fnetworking-midonet~664614","triplet_id":"openstack%2Fnetworking-midonet~master~I9461ca708f05d037fb771b89ec7a95c403c5d1f6","project":"openstack/networking-midonet","branch":"master","topic":"bug/1819260","hashtags":[],"change_id":"I9461ca708f05d037fb771b89ec7a95c403c5d1f6","subject":"Join on explcit relationship paths","status":"MERGED","created":"2019-06-11 15:04:57.000000000","updated":"2019-06-12 20:36:51.000000000","submitted":"2019-06-12 20:36:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664614-1560371811206-22ee9161","meta_rev_id":"fea5d1ce53d52d5e8146ef97cd56b0d4a60fec39","_number":664614,"virtual_id_number":664614,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2019-06-12 13:22:56.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-06-12 20:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":2,"date":"2019-06-12 13:40:11.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2019-06-12 20:08:02.000000000","_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-06-12 20:23:10.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":1,"date":"2019-06-12 20:08:02.000000000","_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 13:40:11.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-06-12 20:08:02.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-12 20:23:10.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-06-12 20:36:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"216c46540b0702bded39410b30bdd6d8092957b0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-11 15:04:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce0c67f29ef67ac8252652fe3a24a411b88e4270","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-11 15:05:31.000000000","message":"Patch Set 1:\n\nThis is an attempt to change the co-check neutron job:\n\nhttp://logs.openstack.org/54/663754/1/check/networking-midonet-tempest-aio-ml2-centos-7/5ed3eda/logs/screen-neutron-api.txt.gz?level\u003dWARNING","accounts_in_message":[],"_revision_number":1},{"id":"7c5e1d6e80a6f21ebe433945c9dab347a61c502e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-11 15:07:26.000000000","message":"Patch Set 1:\n\ns/change/fix the neutron co-gating job","accounts_in_message":[],"_revision_number":1},{"id":"953aadb2fcd4f4ae5b0bb6c28da065932b457068","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-11 16:12:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7452e080453061e1a4b7c95968fc0f2dd93218e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-11 16:14:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2bdc1e1ad9e766c7b95344951a96fbace55c3f06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 18:14:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/664614/3/check/openstack-tox-pep8/4ce67b7/ : SUCCESS in 10m 57s\n- openstack-tox-py27 http://logs.openstack.org/14/664614/3/check/openstack-tox-py27/b1625f8/ : SUCCESS in 13m 48s\n- openstack-tox-py36 http://logs.openstack.org/14/664614/3/check/openstack-tox-py36/5fa4bfc/ : SUCCESS in 18m 33s\n- openstack-tox-py37 http://logs.openstack.org/14/664614/3/check/openstack-tox-py37/43c9c8b/ : SUCCESS in 11m 59s\n- openstack-tox-docs http://logs.openstack.org/14/664614/3/check/openstack-tox-docs/64dc5bb/html/ : SUCCESS in 8m 38s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/14/664614/3/check/networking-midonet-tempest-multinode-ml2/e76300f/ : FAILURE in 15m 16s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/14/664614/3/check/networking-midonet-tempest-multinode-ml2-full/969c4b6/ : FAILURE in 18m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/14/664614/3/check/networking-midonet-tempest-aio-ml2-full-legacy/3101070/ : FAILURE in 13m 44s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/14/664614/3/check/networking-midonet-tempest-aio-ml2-centos-7/d0e9a01/ : SUCCESS in 1h 39m 41s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/14/664614/3/check/networking-midonet-tempest-aio-ml2-full-centos-7/099b9a1/ : SUCCESS in 1h 45m 14s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/14/664614/3/check/networking-midonet-grenade-ml2/06cdc62/ : FAILURE in 9m 33s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/14/664614/3/check/networking-midonet-rally-ml2/3b90b4e/ : FAILURE in 10m 26s (non-voting)\n- openstack-tox-py35 http://logs.openstack.org/14/664614/3/check/openstack-tox-py35/c0ca9f0/ : SUCCESS in 11m 34s\n- openstack-tox-cover http://logs.openstack.org/14/664614/3/check/openstack-tox-cover/ec15942/cover/ : SUCCESS in 12m 03s","accounts_in_message":[],"_revision_number":3},{"id":"80827bdc8035fafa463b252a509c1aa596939ff1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-11 18:20:48.000000000","message":"Patch Set 3:\n\nCreated https://review.opendev.org/#/c/664651/ to verify this fixes neutron co-gating job.","accounts_in_message":[],"_revision_number":3},{"id":"691c22497c9386fb382afaa452004a8ac5ad9160","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-12 13:22:56.000000000","message":"Patch Set 3:\n\nThis has at least gotten the neutron co-gating job to pass","accounts_in_message":[],"_revision_number":3},{"id":"2198be406b1ad87d47cb8fb868840f679d15bdb7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-06-12 13:40:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c24194a0d652246b9a99f1d05bc560ed32ec2752","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-06-12 13:40:15.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"21994eedd179d24e3c691f78f74f14ea8d8571dd","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-12 20:08:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5a0b461ec5386ca5a69db2bc5e712b9d65b9a1de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:08:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f3a67ece9975f0f7ae768033a77b69ca16163e54","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-06-12 20:23:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dbe8e130d77d32f015d3b7b940cefd1a45902328","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:36:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/664614/3/gate/openstack-tox-pep8/84f9781/ : SUCCESS in 8m 05s\n- openstack-tox-py27 http://logs.openstack.org/14/664614/3/gate/openstack-tox-py27/dd899cd/ : SUCCESS in 14m 25s\n- openstack-tox-py36 http://logs.openstack.org/14/664614/3/gate/openstack-tox-py36/a00f58f/ : SUCCESS in 12m 34s\n- openstack-tox-py37 http://logs.openstack.org/14/664614/3/gate/openstack-tox-py37/3fc1e42/ : SUCCESS in 13m 00s\n- openstack-tox-docs http://logs.openstack.org/14/664614/3/gate/openstack-tox-docs/6d527b1/html/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/14/664614/3/gate/openstack-tox-py35/4e46d02/ : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":3},{"id":"a2cd6489566201604a830d9604acd2698ee27816","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:36:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cad4706e294057e503aca5a306af61802703cbe9","revisions":{"03e6427167bcab691bafceec046c12e9b613dbf0":{"kind":"REWORK","_number":1,"created":"2019-06-11 15:04:57.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/14/664614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/14/664614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/1"}}},"commit":{"parents":[{"commit":"df7c0c6070b0e302dc40273f47bb3c5da198dc9e","subject":"Merge \"Remove CINDER_PERIODIC_INTERVAL setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/df7c0c6070b0e302dc40273f47bb3c5da198dc9e"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-06-11 15:02:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-06-11 15:02:25.000000000","tz":-240},"subject":"Join on explcit relationship paths","message":"Join on explcit relationship paths\n\nThe join() in get_router_for_floatingip() is joining from entity\nto entity without an explicit ON clause which creates an ambiguous\nsituation.  SQLAlchemy 1.3 guards against this now, so use the\nreal relationship-bound path so that the ORM does not need to guess.\n\nCopied from https://review.opendev.org/#/c/642117/\n\nRelated-bug: #1819260\n\nChange-Id: I9461ca708f05d037fb771b89ec7a95c403c5d1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/03e6427167bcab691bafceec046c12e9b613dbf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/03e6427167bcab691bafceec046c12e9b613dbf0"}]},"branch":"refs/heads/master"},"6590d46bb3530bd5d5f8aacb0f950b1e03af6ea2":{"kind":"REWORK","_number":2,"created":"2019-06-11 16:12:51.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/14/664614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/14/664614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/2"}}},"commit":{"parents":[{"commit":"df7c0c6070b0e302dc40273f47bb3c5da198dc9e","subject":"Merge \"Remove CINDER_PERIODIC_INTERVAL setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/df7c0c6070b0e302dc40273f47bb3c5da198dc9e"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-06-11 15:02:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-06-11 16:12:38.000000000","tz":-240},"subject":"Join on explcit relationship paths","message":"Join on explcit relationship paths\n\nThe join() in get_router_for_floatingip() is joining from entity\nto entity without an explicit ON clause which creates an ambiguous\nsituation.  SQLAlchemy 1.3 guards against this now, so use the\nreal relationship-bound path so that the ORM does not need to guess.\n\nCopied from https://review.opendev.org/#/c/642117/\n\nRelated-bug: #1819260\nChange-Id: I9461ca708f05d037fb771b89ec7a95c403c5d1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6590d46bb3530bd5d5f8aacb0f950b1e03af6ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6590d46bb3530bd5d5f8aacb0f950b1e03af6ea2"}]},"branch":"refs/heads/master"},"cad4706e294057e503aca5a306af61802703cbe9":{"kind":"REWORK","_number":3,"created":"2019-06-11 16:14:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/14/664614/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/14/664614/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/14/664614/3"}}},"commit":{"parents":[{"commit":"df7c0c6070b0e302dc40273f47bb3c5da198dc9e","subject":"Merge \"Remove CINDER_PERIODIC_INTERVAL setting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/df7c0c6070b0e302dc40273f47bb3c5da198dc9e"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-06-11 15:02:25.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-06-11 16:14:28.000000000","tz":-240},"subject":"Join on explcit relationship paths","message":"Join on explcit relationship paths\n\nThe join() in get_router_for_floatingip() is joining from entity\nto entity without an explicit ON clause which creates an ambiguous\nsituation.  SQLAlchemy 1.3 guards against this now, so use the\nreal relationship-bound path so that the ORM does not need to guess.\n\nCopied from https://review.opendev.org/#/c/642117/\n\nRelated-bug: #1819260\nChange-Id: I9461ca708f05d037fb771b89ec7a95c403c5d1f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/cad4706e294057e503aca5a306af61802703cbe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/cad4706e294057e503aca5a306af61802703cbe9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
