)]}'
{"id":"openstack%2Fnetworking-midonet~669059","triplet_id":"openstack%2Fnetworking-midonet~stable%2Fstein~I57958f8388505aee5654f20c0c9c60f0e3ceea15","project":"openstack/networking-midonet","branch":"stable/stein","topic":"stein/backport/667356","hashtags":[],"change_id":"I57958f8388505aee5654f20c0c9c60f0e3ceea15","subject":"Remove ocata jobs from branch","status":"MERGED","created":"2019-07-04 06:17:55.000000000","updated":"2019-10-01 13:56:18.000000000","submitted":"2019-10-01 13:50:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669059-1569937821672-fe29816b","meta_rev_id":"e81d1b5fafa6c4f05bad814aa999f58c272c7981","_number":669059,"virtual_id_number":669059,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 13:50:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":1,"date":"2019-10-01 04:45:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2019-10-01 09:22:56.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-10-01 12:16:15.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2019-10-01 12:16:15.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 04:45:45.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-10-01 09:22:56.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-01 12:16:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-01 13:50:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d54eb639fb3b2df93809d124c87fd6e52be7a128","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-04 06:17:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b86cecc92345e271a8cc8cc3cc4968d6d4331340","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-04 06:19:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8c561497334de790b580e0ca70abb839b39006ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 07:56:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/59/669059/2/check/openstack-tox-pep8/bc58670/ : RETRY_LIMIT in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/59/669059/2/check/openstack-tox-py27/851f7bc/ : RETRY_LIMIT in 2m 37s\n- openstack-tox-py35 http://logs.openstack.org/59/669059/2/check/openstack-tox-py35/ccecf7a/ : RETRY_LIMIT in 2m 35s\n- openstack-tox-py36 http://logs.openstack.org/59/669059/2/check/openstack-tox-py36/9336f03/ : RETRY_LIMIT in 2m 40s\n- openstack-tox-docs http://logs.openstack.org/59/669059/2/check/openstack-tox-docs/7a30af7/html/ : SUCCESS in 4m 07s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/59/669059/2/check/networking-midonet-tempest-multinode-ml2/9c7f9df/ : SUCCESS in 1h 22m 38s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full http://logs.openstack.org/59/669059/2/check/networking-midonet-tempest-multinode-ml2-full/eacb9b8/ : FAILURE in 53m 26s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/59/669059/2/check/networking-midonet-tempest-aio-ml2-full-legacy/2b8124b/ : FAILURE in 14m 40s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/59/669059/2/check/networking-midonet-tempest-aio-ml2-centos-7/3d2309c/ : SUCCESS in 1h 25m 58s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/59/669059/2/check/networking-midonet-tempest-aio-ml2-full-centos-7/d82d429/ : FAILURE in 49m 58s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/59/669059/2/check/networking-midonet-grenade-ml2/a73b34f/ : FAILURE in 16m 41s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/59/669059/2/check/networking-midonet-rally-ml2/d3ba1d7/ : FAILURE in 16m 20s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/59/669059/2/check/openstack-tox-cover/6d4afa3/ : RETRY_LIMIT in 2m 24s","accounts_in_message":[],"_revision_number":2},{"id":"2c0c1378979017548357bf7475ced21901be1fcf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-01 04:45:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"293a789654c9ddd312956ba08dbce13969bab275","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-01 04:45:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8f6517649654ac3826e4666a411cdce9db8ca464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:39:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31204db035794c5e9470a635c4a53190 : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de8283773d4447e38f21bf0f2fcf408d : SUCCESS in 11m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/932030ab7ea54edca3ce34c1ba02cb04 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a6a2accf84a447ca6c0dfceca3a9874 : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf466dd225b439e87dfaf50c801c4c8 : SUCCESS in 5m 13s\n- networking-midonet-tempest-multinode-ml2 https://zuul.opendev.org/t/openstack/build/1c9022b4a9dc415e86a205eb4dd5bc18 : SUCCESS in 1h 32m 25s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://zuul.opendev.org/t/openstack/build/1d5bf32615dc4a828bd881fbbea8a762 : FAILURE in 35m 32s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://zuul.opendev.org/t/openstack/build/482e70b4ee2648d6b43fd0c02e9b04d5 : FAILURE in 11m 31s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/ad444cb721cb4834852978ce8a850307 : SUCCESS in 1h 21m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/aeb750cbb142401da82a8790d6a78e47 : FAILURE in 34m 44s (non-voting)\n- networking-midonet-grenade-ml2 https://zuul.opendev.org/t/openstack/build/2d45ceba8d26414a88e4dbc3d737bbad : FAILURE in 16m 17s (non-voting)\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/7ee50977d6c64bc0be98977fb270b550 : FAILURE in 11m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/425350eb67fa4a78ac161796602df695 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":3},{"id":"dae178ae32b3726f912519428c11e66cdffe105f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-01 09:22:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"15c2759314c01c24a49def46b558dba53301d0ff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-01 12:16:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ae962e0901073e2cad367eca7432be5feca4400d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:16:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d233e496ff03a939fc9bb509ac02d699197aef7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:50:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e933d3181c554cb9af3f81cba5232975 : SUCCESS in 8m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8da2187065934edeb03ffc81088db90b : SUCCESS in 13m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d70900c547d544f39bd5be4c0b0d255f : SUCCESS in 11m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ed451f1bdab40b49a23283d41ae52f8 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c0429eaa38462ba99c90a665eb8da2 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"aebd71f6c826f70b0107f0286eff4da090b0d3e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:50:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dd0fd3a02a7e249048b7a156e449243ce124de97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:56:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c63ee62d31849d88decdb4192baf6ea : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5165d0ad661837f89195f61c54783b2bcc050897","revisions":{"d46950da08afa2c7bcd4a465e0d3473d4dff5b68":{"kind":"REWORK","_number":1,"created":"2019-07-04 06:17:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/59/669059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/59/669059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/1"}}},"commit":{"parents":[{"commit":"5b4437e94e107aefb2ec0a1601fe7ea7895eca53","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5b4437e94e107aefb2ec0a1601fe7ea7895eca53"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-25 12:45:53.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-07-04 06:17:55.000000000","tz":0},"subject":"Remove ocata jobs from branch","message":"Remove ocata jobs from branch\n\nThis repo configures two ocata jobs to run as periodic jobs, these\nneed to be only in the stable/ocata and not in this branch - remove\nthem.\n\nThe stable/ocata branch has these jobs configured, so the jobs continue\nto run - but the Zuul setup is simplified.\n\nChange-Id: I57958f8388505aee5654f20c0c9c60f0e3ceea15\n(cherry picked from commit 431ad460f85136fcad773818f3b12f3a17917b95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d46950da08afa2c7bcd4a465e0d3473d4dff5b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/d46950da08afa2c7bcd4a465e0d3473d4dff5b68"}]},"branch":"refs/heads/stable/stein"},"df9985ba4ae3237fe842e630afc51ba8470b1e7e":{"kind":"NO_CHANGE","_number":2,"created":"2019-07-04 06:19:32.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/59/669059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/59/669059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/2"}}},"commit":{"parents":[{"commit":"5b4437e94e107aefb2ec0a1601fe7ea7895eca53","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5b4437e94e107aefb2ec0a1601fe7ea7895eca53"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-25 12:45:53.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-04 06:18:49.000000000","tz":120},"subject":"Remove ocata jobs from branch","message":"Remove ocata jobs from branch\n\nThis repo configures two ocata jobs to run as periodic jobs, these\nneed to be only in the stable/ocata and not in this branch - remove\nthem.\n\nThe stable/ocata branch has these jobs configured, so the jobs continue\nto run - but the Zuul setup is simplified.\n\nChange-Id: I57958f8388505aee5654f20c0c9c60f0e3ceea15\n(cherry picked from commit 431ad460f85136fcad773818f3b12f3a17917b95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/df9985ba4ae3237fe842e630afc51ba8470b1e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/df9985ba4ae3237fe842e630afc51ba8470b1e7e"}]},"branch":"refs/heads/stable/stein"},"5165d0ad661837f89195f61c54783b2bcc050897":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-01 04:45:14.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/59/669059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/59/669059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/59/669059/3"}}},"commit":{"parents":[{"commit":"0ec1fe15dc6009a840014c19a94852a053f0ca4c","subject":"Update test_update_provider_net_unsupported test case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0ec1fe15dc6009a840014c19a94852a053f0ca4c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-25 12:45:53.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-10-01 04:45:14.000000000","tz":0},"subject":"Remove ocata jobs from branch","message":"Remove ocata jobs from branch\n\nThis repo configures two ocata jobs to run as periodic jobs, these\nneed to be only in the stable/ocata and not in this branch - remove\nthem.\n\nThe stable/ocata branch has these jobs configured, so the jobs continue\nto run - but the Zuul setup is simplified.\n\nChange-Id: I57958f8388505aee5654f20c0c9c60f0e3ceea15\n(cherry picked from commit 431ad460f85136fcad773818f3b12f3a17917b95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5165d0ad661837f89195f61c54783b2bcc050897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5165d0ad661837f89195f61c54783b2bcc050897"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
