)]}'
{"id":"openstack%2Fnetworking-midonet~669063","triplet_id":"openstack%2Fnetworking-midonet~stable%2Frocky~I57958f8388505aee5654f20c0c9c60f0e3ceea15","project":"openstack/networking-midonet","branch":"stable/rocky","topic":"rocky/backport/667356","hashtags":[],"change_id":"I57958f8388505aee5654f20c0c9c60f0e3ceea15","subject":"Remove ocata jobs from branch","status":"ABANDONED","created":"2019-07-04 06:21:33.000000000","updated":"2021-04-14 13:56:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"342cddb1796415b1582785c1106907092b21173e","_number":669063,"virtual_id_number":669063,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-02 10:10:09.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"date":"2021-04-14 13:56:06.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2019-07-04 08:32:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-09-30 01:45:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 08:32:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 15:35:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-30 01:45:19.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"}],"messages":[{"id":"d98f20a65a0e822415b6521078f02969b013892f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-04 06:21:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30f950ef729f80b7b79901d2890106fa97159201","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 08:32:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/63/669063/1/check/openstack-tox-pep8/5a16b3e/ : RETRY_LIMIT in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/63/669063/1/check/openstack-tox-py27/f75094f/ : RETRY_LIMIT in 2m 55s\n- openstack-tox-py35 http://logs.openstack.org/63/669063/1/check/openstack-tox-py35/f490297/ : RETRY_LIMIT in 3m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/63/669063/1/check/build-openstack-sphinx-docs/ce06534/html/ : SUCCESS in 4m 57s\n- networking-midonet-tempest-aio-ml2 http://logs.openstack.org/63/669063/1/check/networking-midonet-tempest-aio-ml2/07469f6/ : FAILURE in 1h 19m 35s\n- networking-midonet-tempest-multinode-ml2 http://logs.openstack.org/63/669063/1/check/networking-midonet-tempest-multinode-ml2/26ce7c8/ : FAILURE in 1h 42m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-full http://logs.openstack.org/63/669063/1/check/networking-midonet-tempest-aio-ml2-full/61e9e0a/ : FAILURE in 30m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy http://logs.openstack.org/63/669063/1/check/networking-midonet-tempest-aio-ml2-full-legacy/d36500b/ : FAILURE in 31m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 http://logs.openstack.org/63/669063/1/check/networking-midonet-tempest-aio-ml2-centos-7/3e862ff/ : FAILURE in 1h 13m 36s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 http://logs.openstack.org/63/669063/1/check/networking-midonet-tempest-aio-ml2-full-centos-7/06c20ee/ : FAILURE in 36m 54s (non-voting)\n- networking-midonet-grenade-ml2 http://logs.openstack.org/63/669063/1/check/networking-midonet-grenade-ml2/b621fa3/ : FAILURE in 1h 25m 11s (non-voting)\n- networking-midonet-rally-ml2 http://logs.openstack.org/63/669063/1/check/networking-midonet-rally-ml2/5d4b167/ : FAILURE in 23m 10s\n- openstack-tox-cover http://logs.openstack.org/63/669063/1/check/openstack-tox-cover/197cc9d/ : RETRY_LIMIT in 2m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6752438e2d1388b0198be6daf4646819c6539287","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-12-02 10:10:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b359f6876f6224c011e9eefb9457c99b04aac169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 11:31:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a7f75f4d984befb28ce05becb9ae94 : FAILURE in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1dc687045ee3416fa9f5244d75ada716 : RETRY_LIMIT in 2m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1cc5a791cca941f9a271efc8c76ae933 : RETRY_LIMIT in 2m 23s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/049dc1008e71436e886c594901f31f4b : SUCCESS in 4m 52s\n- networking-midonet-tempest-aio-ml2 https://zuul.opendev.org/t/openstack/build/444cec8bf73e4fe2aa09316be4b7a874 : FAILURE in 1h 04m 27s\n- networking-midonet-tempest-multinode-ml2 https://zuul.opendev.org/t/openstack/build/70f6737cf20d4bdf86effb1273bb7042 : FAILURE in 1h 12m 27s (non-voting)\n- networking-midonet-tempest-aio-ml2-full https://zuul.opendev.org/t/openstack/build/a93f0c767a4e4c79b2049113a52a5841 : FAILURE in 31m 04s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://zuul.opendev.org/t/openstack/build/b7e71969460443199d4f8e62452d1d52 : FAILURE in 29m 49s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/6bd2f9528ffc4acf84e5159799980471 : FAILURE in 12m 45s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/b466936406694f08bdbd70f981486846 : FAILURE in 17m 46s (non-voting)\n- networking-midonet-grenade-ml2 https://zuul.opendev.org/t/openstack/build/545551bfa8034f9ebe36ab03f8bdd6f2 : FAILURE in 1h 19m 24s (non-voting)\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/d00bd377fd4f42d58e82fc532d2d1b5b : FAILURE in 29m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/019e60a9481e4aeb9ee4c6329d18e1bb : RETRY_LIMIT in 2m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab27f5e2673437a76931242459c4ca01bfa6a9c4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:35:20.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"96fa5c1041ef7dcdeecfe480f72fc60d04da7624","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-25 20:03:52.000000000","message":"Restored\n\nThis is needed - the breakage is unrelated","accounts_in_message":[],"_revision_number":1},{"id":"a28919ecfdae6519e25dc35b287bcd911406fc98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 20:04:15.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/networking-midonet on branch stable/rocky.  The error\nwas:\n\n  Unknown projects: openstack/networking-l2gw\n\nThe error appears in the following job stanza:\n\n  job:\n      name: networking-midonet-tempest-multinode-ml2\n      parent: legacy-dsvm-base-multinode\n      run: playbooks/tempest-multinode-ml2/run.yaml\n      post-run: playbooks/tempest-multinode-ml2/post.yaml\n      timeout: 7800\n      required-projects:\n        - openstack/devstack-gate\n        - openstack/networking-l2gw\n        - openstack/networking-midonet\n        - openstack/neutron-dynamic-routing\n        - openstack/neutron-tempest-plugin\n        - x/tap-as-a-service\n        - openstack/tempest\n      nodeset: legacy-ubuntu-xenial-2-node\n\n  in \"openstack/networking-midonet/.zuul.yaml@stable/rocky\", line 258, column 3","accounts_in_message":[],"_revision_number":1},{"id":"d1bd59bc843f93cf8437866f40e8d6191f8814e5","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2020-09-30 01:45:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"342cddb1796415b1582785c1106907092b21173e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:56:09.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2942ad97d426d3cb5beaf4aca72927dd73e92e9","revisions":{"a2942ad97d426d3cb5beaf4aca72927dd73e92e9":{"kind":"REWORK","_number":1,"created":"2019-07-04 06:21:33.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/63/669063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/63/669063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/669063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/669063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/669063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/63/669063/1"}}},"commit":{"parents":[{"commit":"dca16fa9d7760472ab14e62dc618ce059c42a646","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/dca16fa9d7760472ab14e62dc618ce059c42a646"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-06-25 12:45:53.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-04 06:21:11.000000000","tz":120},"subject":"Remove ocata jobs from branch","message":"Remove ocata jobs from branch\n\nThis repo configures two ocata jobs to run as periodic jobs, these\nneed to be only in the stable/ocata and not in this branch - remove\nthem.\n\nThe stable/ocata branch has these jobs configured, so the jobs continue\nto run - but the Zuul setup is simplified.\n\nChange-Id: I57958f8388505aee5654f20c0c9c60f0e3ceea15\n(cherry picked from commit 431ad460f85136fcad773818f3b12f3a17917b95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a2942ad97d426d3cb5beaf4aca72927dd73e92e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a2942ad97d426d3cb5beaf4aca72927dd73e92e9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
