)]}'
{"id":"openstack%2Fnetworking-midonet~674397","triplet_id":"openstack%2Fnetworking-midonet~master~Ie83a4dcbfc57759972714889fd53f7da4c6da741","project":"openstack/networking-midonet","branch":"master","topic":"disable-centos-7-jobs","hashtags":[],"change_id":"Ie83a4dcbfc57759972714889fd53f7da4c6da741","subject":"Stop running the centos-7 integration jobs.","status":"ABANDONED","created":"2019-08-02 18:02:15.000000000","updated":"2022-11-22 14:52:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"08c6b50b367738061bae398a426d5706b61f6862","_number":674397,"virtual_id_number":674397,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":1,"date":"2019-08-02 18:27:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"all":[{"value":-1,"date":"2019-08-05 07:11:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2019-08-02 18:52:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 18:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 18:52:12.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-08-05 07:11:47.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"}],"messages":[{"id":"ead881335db76f3ed8154265ebf086de74b4f077","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-08-02 18:02:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aefda9019a59d5ea9d847192211f0b007a5dd512","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 18:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/97/674397/1/check/openstack-tox-pep8/a6fdf1b/ : SUCCESS in 10m 47s\n- openstack-tox-py27 https://logs.opendev.org/97/674397/1/check/openstack-tox-py27/f332a7f/ : SUCCESS in 13m 59s\n- openstack-tox-py36 https://logs.opendev.org/97/674397/1/check/openstack-tox-py36/d2db200/ : SUCCESS in 12m 14s\n- openstack-tox-py37 https://logs.opendev.org/97/674397/1/check/openstack-tox-py37/48a630d/ : SUCCESS in 13m 02s\n- openstack-tox-docs https://logs.opendev.org/97/674397/1/check/openstack-tox-docs/858ec6a/html/ : SUCCESS in 3m 19s\n- networking-midonet-tempest-multinode-ml2 https://logs.opendev.org/97/674397/1/check/networking-midonet-tempest-multinode-ml2/5bbdcb3/ : FAILURE in 19m 10s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://logs.opendev.org/97/674397/1/check/networking-midonet-tempest-multinode-ml2-full/2758988/ : FAILURE in 16m 54s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://logs.opendev.org/97/674397/1/check/networking-midonet-tempest-aio-ml2-full-legacy/0c0c2c0/ : FAILURE in 11m 46s (non-voting)\n- networking-midonet-grenade-ml2 https://logs.opendev.org/97/674397/1/check/networking-midonet-grenade-ml2/628969c/ : FAILURE in 14m 37s (non-voting)\n- networking-midonet-rally-ml2 https://logs.opendev.org/97/674397/1/check/networking-midonet-rally-ml2/4a76124/ : FAILURE in 15m 16s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/97/674397/1/check/openstack-tox-cover/564fd2f/cover/ : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bd0d53f16c71ec9dd5145a1b3927b2d5185f502d","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-08-02 18:52:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33f2401f4ec29c8f32ad25f1d455aed121c6552c","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-05 07:11:47.000000000","message":"Patch Set 1: Code-Review-1\n\nthis seems overkill as we can use http as a workaround.\ni notified someone to renew the cert.","accounts_in_message":[],"_revision_number":1},{"id":"08c6b50b367738061bae398a426d5706b61f6862","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 14:52:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03bc785623e264a2fed4ab31bcf2e03788722015","revisions":{"03bc785623e264a2fed4ab31bcf2e03788722015":{"kind":"REWORK","_number":1,"created":"2019-08-02 18:02:15.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/97/674397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/97/674397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/674397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/674397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/674397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/97/674397/1"}}},"commit":{"parents":[{"commit":"c405a30f636effd04c6ce50760b2e3efecb3eece","subject":"Remove py35 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c405a30f636effd04c6ce50760b2e3efecb3eece"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-08-02 18:00:50.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2019-08-02 18:02:13.000000000","tz":-420},"subject":"Stop running the centos-7 integration jobs.","message":"Stop running the centos-7 integration jobs.\n\nThese jobs depend on packages from builds.midonet.org but that repo\u0027s\nssl cert has expired and the jobs fail.\n\nThe error message:\n  https://builds.midonet.org/devstack/unstable/el7/repodata/repomd.xml: [Errno 14] curl#60 - \"Peer\u0027s Certificate has expired.\"\n\nThese jobs can be readded when they are modified to pull from a\ndifferent repo or the upstream repo fixes its cert.\n\nChange-Id: Ie83a4dcbfc57759972714889fd53f7da4c6da741\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/03bc785623e264a2fed4ab31bcf2e03788722015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/03bc785623e264a2fed4ab31bcf2e03788722015"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
