)]}'
{"id":"openstack%2Fnetworking-midonet~674538","triplet_id":"openstack%2Fnetworking-midonet~master~I61be97e518548f8c41ef4ce66a97e598637e266b","project":"openstack/networking-midonet","branch":"master","topic":"http","hashtags":[],"change_id":"I61be97e518548f8c41ef4ce66a97e598637e266b","subject":"Use http instead of https for builds.midonet.org","status":"MERGED","created":"2019-08-05 07:10:17.000000000","updated":"2019-08-15 12:31:46.000000000","submitted":"2019-08-05 15:00:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674538-1565017207757-04ad8034","meta_rev_id":"ff91961c3f37004f3eabc0a672531693d8332efa","_number":674538,"virtual_id_number":674538,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-08-05 15:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2019-08-05 13:54:17.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-08-05 14:42:54.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-08-05 14:42:54.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 13:54:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-08-05 14:42:54.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-08-05 15:00:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b39dbb29e6c175379d7e27925a37cc1de3cf5a39","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-05 07:10:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7a04eb1133633ad0ee7f5dddd909fc08ccd008f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-05 07:25:28.000000000","message":"Patch Set 1: Code-Review+2\n\nThx a lot for that :) I hope it will help.","accounts_in_message":[],"_revision_number":1},{"id":"5a64bf0e70ded35e977b4eff4355eb8c9652c9b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 07:32:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/38/674538/1/check/openstack-tox-pep8/2d0d23c/ : SUCCESS in 8m 41s\n- openstack-tox-py27 https://logs.opendev.org/38/674538/1/check/openstack-tox-py27/2b2fee4/ : SUCCESS in 10m 50s\n- openstack-tox-py36 https://logs.opendev.org/38/674538/1/check/openstack-tox-py36/8289e3f/ : SUCCESS in 15m 35s\n- openstack-tox-py37 https://logs.opendev.org/38/674538/1/check/openstack-tox-py37/85c307f/ : SUCCESS in 11m 05s\n- openstack-tox-docs https://logs.opendev.org/38/674538/1/check/openstack-tox-docs/42e25e0/html/ : SUCCESS in 4m 12s\n- networking-midonet-tempest-multinode-ml2 https://logs.opendev.org/38/674538/1/check/networking-midonet-tempest-multinode-ml2/851f60f/ : FAILURE in 16m 28s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://logs.opendev.org/38/674538/1/check/networking-midonet-tempest-multinode-ml2-full/4627354/ : FAILURE in 15m 59s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://logs.opendev.org/38/674538/1/check/networking-midonet-tempest-aio-ml2-full-legacy/78395c7/ : FAILURE in 14m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/38/674538/1/check/networking-midonet-tempest-aio-ml2-centos-7/254b213/ : FAILURE in 16m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://logs.opendev.org/38/674538/1/check/networking-midonet-tempest-aio-ml2-full-centos-7/8234866/ : FAILURE in 18m 13s (non-voting)\n- networking-midonet-grenade-ml2 https://logs.opendev.org/38/674538/1/check/networking-midonet-grenade-ml2/dd303b8/ : FAILURE in 15m 10s (non-voting)\n- networking-midonet-rally-ml2 https://logs.opendev.org/38/674538/1/check/networking-midonet-rally-ml2/5edb7a6/ : FAILURE in 13m 07s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/38/674538/1/check/openstack-tox-cover/48de9a1/cover/ : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a7e51759fe8d0bcdb655d43ec50566be41d74e3b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-05 07:48:40.000000000","message":"Patch Set 1: Code-Review-1\n\nIt seems that You also need to change url to repo key in devstack/midonet-pkg/configure_repo_centos.sh","accounts_in_message":[],"_revision_number":1},{"id":"2844c9d5429cef0cd7c9236028f286daa0b1db6d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-05 11:48:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb9210f97e91d458b461cd39ad55290e050d50fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:48:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/38/674538/2/check/openstack-tox-pep8/05858c6/ : SUCCESS in 8m 43s\n- openstack-tox-py27 https://logs.opendev.org/38/674538/2/check/openstack-tox-py27/3ef1b7b/ : SUCCESS in 12m 50s\n- openstack-tox-py36 https://logs.opendev.org/38/674538/2/check/openstack-tox-py36/79b09e9/ : SUCCESS in 20m 44s\n- openstack-tox-py37 https://logs.opendev.org/38/674538/2/check/openstack-tox-py37/8e09c7b/ : SUCCESS in 16m 02s\n- openstack-tox-docs https://logs.opendev.org/38/674538/2/check/openstack-tox-docs/ce311e8/html/ : SUCCESS in 3m 24s\n- networking-midonet-tempest-multinode-ml2 https://logs.opendev.org/38/674538/2/check/networking-midonet-tempest-multinode-ml2/3dc567b/ : FAILURE in 10m 48s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://logs.opendev.org/38/674538/2/check/networking-midonet-tempest-multinode-ml2-full/89cf940/ : FAILURE in 11m 24s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://logs.opendev.org/38/674538/2/check/networking-midonet-tempest-aio-ml2-full-legacy/11c9ed1/ : FAILURE in 11m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://logs.opendev.org/38/674538/2/check/networking-midonet-tempest-aio-ml2-centos-7/cd364cf/ : FAILURE in 1h 57m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://logs.opendev.org/38/674538/2/check/networking-midonet-tempest-aio-ml2-full-centos-7/303f6f8/ : FAILURE in 1h 41m 26s (non-voting)\n- networking-midonet-grenade-ml2 https://logs.opendev.org/38/674538/2/check/networking-midonet-grenade-ml2/ba9b9e3/ : FAILURE in 14m 46s (non-voting)\n- networking-midonet-rally-ml2 https://logs.opendev.org/38/674538/2/check/networking-midonet-rally-ml2/0b6b027/ : FAILURE in 13m 18s (non-voting)\n- openstack-tox-cover https://logs.opendev.org/38/674538/2/check/openstack-tox-cover/a058cb9/cover/ : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":2},{"id":"01210a6f8696e798a6137bdf5556783cacf7a217","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-05 13:54:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1acffec8333ca7491c8a38d9b420f8ed87b969f7","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-08-05 14:42:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3860edb8ec2598cfd28112a3e256111e4bd78c9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:43:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ed6d3bdf799fa0ae858a42874dab736e2cfc352","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:00:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/38/674538/2/gate/openstack-tox-pep8/984607c/ : SUCCESS in 8m 07s\n- openstack-tox-py27 https://logs.opendev.org/38/674538/2/gate/openstack-tox-py27/dc640f1/ : SUCCESS in 14m 23s\n- openstack-tox-py36 https://logs.opendev.org/38/674538/2/gate/openstack-tox-py36/466aca2/ : SUCCESS in 14m 49s\n- openstack-tox-py37 https://logs.opendev.org/38/674538/2/gate/openstack-tox-py37/bbde2a6/ : SUCCESS in 15m 23s\n- openstack-tox-docs https://logs.opendev.org/38/674538/2/gate/openstack-tox-docs/75e453d/html/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f4f6d56a25a0f3607faf00366b91768cfc992138","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:00:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0c53fca6fe1b54afc8fc2329bddbdb0d93f1eed1","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-08-15 12:31:46.000000000","message":"Created a revert of this change as I19be0b171d435198e0be0a05f33a4b375780b172","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02e25cc65601add1d96b7150ed70403c3de4243b","revisions":{"289fdc7bc1507345d16f93046608c5777d729659":{"kind":"REWORK","_number":1,"created":"2019-08-05 07:10:17.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/38/674538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/38/674538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/38/674538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/38/674538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/38/674538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/38/674538/1"}}},"commit":{"parents":[{"commit":"c405a30f636effd04c6ce50760b2e3efecb3eece","subject":"Remove py35 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c405a30f636effd04c6ce50760b2e3efecb3eece"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-08-05 07:09:21.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-08-05 07:09:21.000000000","tz":540},"subject":"Use http instead of https for centos job","message":"Use http instead of https for centos job\n\nAs a temporary workaround for cert expiration.\n\nChange-Id: I61be97e518548f8c41ef4ce66a97e598637e266b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/289fdc7bc1507345d16f93046608c5777d729659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/289fdc7bc1507345d16f93046608c5777d729659"}]},"branch":"refs/heads/master"},"02e25cc65601add1d96b7150ed70403c3de4243b":{"kind":"REWORK","_number":2,"created":"2019-08-05 11:48:21.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/38/674538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/38/674538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/38/674538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/38/674538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/38/674538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/38/674538/2"}}},"commit":{"parents":[{"commit":"c405a30f636effd04c6ce50760b2e3efecb3eece","subject":"Remove py35 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c405a30f636effd04c6ce50760b2e3efecb3eece"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-08-05 07:09:21.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-08-05 11:47:19.000000000","tz":540},"subject":"Use http instead of https for builds.midonet.org","message":"Use http instead of https for builds.midonet.org\n\nAs a temporary workaround for cert expiration.\n\nChange-Id: I61be97e518548f8c41ef4ce66a97e598637e266b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/02e25cc65601add1d96b7150ed70403c3de4243b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/02e25cc65601add1d96b7150ed70403c3de4243b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
