)]}'
{"id":"openstack%2Fnetworking-midonet~678880","triplet_id":"openstack%2Fnetworking-midonet~master~I42907fecf49b4549252b6a6d3ac9b4ff8c556c92","project":"openstack/networking-midonet","branch":"master","hashtags":[],"change_id":"I42907fecf49b4549252b6a6d3ac9b4ff8c556c92","subject":"PDF documentation build","status":"MERGED","created":"2019-08-27 15:20:15.000000000","updated":"2019-10-01 04:00:43.000000000","submitted":"2019-10-01 03:59:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678880-1569902373075-b153f933","meta_rev_id":"de0c7881140218798f2d7c41c1dbe3be982ec9f9","_number":678880,"virtual_id_number":678880,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 03:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":2,"date":"2019-09-30 21:59:09.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":1,"date":"2019-08-27 15:40:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-10-01 03:41:23.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":1,"date":"2019-10-01 03:41:23.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 15:40:31.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-09-30 21:59:09.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-10-01 03:41:23.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2019-10-01 03:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78ce8507053b53ee924628bc0e75a5a5362d529a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-27 15:20:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"045e57e779b290428f8b2dd8620e4180b15b8d2d","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-08-27 15:40:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aa3d8ffbe0cb722377ad21d944ef4764d9a2f600","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 17:04:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06db0516f574403b876754f09a3cd499 : SUCCESS in 8m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7de924d9de264311a988dacbf0973671 : SUCCESS in 11m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/214504259ea646808f54d3bbe8a1fda5 : SUCCESS in 12m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46d86b645efd46dd812df69d3a0c7e4d : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad504c191d4344f49902b1723d833285 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74b4da1195cf4c3b8ddaa20ba5e52ea3 : SUCCESS in 5m 25s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/def650eba36d4a54a72c34f2cbd2aac2 : FAILURE in 1h 33m 47s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/92493ba64ccb4706979385197e07521a : FAILURE in 43m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5fb216de0f14db68c5c30f01cbd29b6 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9ef775eb7810192f8b6cbffefd992df324018229","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-09-30 21:59:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fff5fa383f83fd28c2940ecb437f6c601bdeb60","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2019-10-01 03:41:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c341f247d19c9a5bfd7f392747923bc905bd6bf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 03:41:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84d59603188cac7cc9c20085d4efd2ca5caeed1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 03:59:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2821861cf8b4745a2921f69b472dabe : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b32482f6c914478698cebe2d5dbe94ca : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe33674d523b44efa47e05d06207be87 : SUCCESS in 9m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2eda717e18f4e2ab18a1848938bff65 : SUCCESS in 11m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bdd07d38b294249b94420b8782eb7d4 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b5c5b341a5e449dad449ed56ba63f8b : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5c5454a84e9dc2852c3f47eb4f99e4c242bb277e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 03:59:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"de0c7881140218798f2d7c41c1dbe3be982ec9f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 04:00:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca08f5fd5d84fe0932a1de81142dcf3 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a621acc416fd4653b0301e99534a1f34 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c33dde5cacdad72f534a471bd9a2ad53fd87ae96","revisions":{"c33dde5cacdad72f534a471bd9a2ad53fd87ae96":{"kind":"REWORK","_number":1,"created":"2019-08-27 15:20:15.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/80/678880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/80/678880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/80/678880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/80/678880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/80/678880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/80/678880/1"}}},"commit":{"parents":[{"commit":"5af46b5a7605b626896c68fe74a61963e6470974","subject":"Merge \"Revert \"Use http instead of https for builds.midonet.org\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5af46b5a7605b626896c68fe74a61963e6470974"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-27 15:18:54.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-08-27 15:20:12.000000000","tz":540},"subject":"PDF documentation build","message":"PDF documentation build\n\nNote that toctree_only option [1] is set to True in latex_documents\nin doc/source/conf.py to get a better document structure in\nthe generated PDF file.\n\nChange-Id: I42907fecf49b4549252b6a6d3ac9b4ff8c556c92\nStory: 2006099\nTask: 35123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c33dde5cacdad72f534a471bd9a2ad53fd87ae96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c33dde5cacdad72f534a471bd9a2ad53fd87ae96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
