)]}'
{"id":"openstack%2Fnetworking-midonet~684870","triplet_id":"openstack%2Fnetworking-midonet~master~I812031838a6fc84bbb9896b280f9806313659ad5","project":"openstack/networking-midonet","branch":"master","topic":"constraints","hashtags":[],"change_id":"I812031838a6fc84bbb9896b280f9806313659ad5","subject":"Update the constraints url","status":"ABANDONED","created":"2019-09-26 02:12:34.000000000","updated":"2019-12-04 08:43:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4f3a4f5f249b6aba0edbf3c4d52718691212370f","_number":684870,"virtual_id_number":684870,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-04 08:43:11.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-26 03:29:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 03:29:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 08:43:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"0fbece7c424f3f3c9cda443dc6c82c9c7e49c042","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 02:12:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"828dc46aacf1c093e0644ea70f55e0e80874040f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:29:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca30496d4bd47709db9237c4bd94da9 : FAILURE in 7m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af17b59140284364a2a129d5887cc774 : FAILURE in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/754400d6aeca4070b3268706f6d49824 : FAILURE in 10m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69d0cdc718e440f08feddb318dcbfb93 : FAILURE in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479c4b00ca574fbe8d20f26117f9e7bb : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8079048ac348475e9da4593989884726 : SUCCESS in 4m 40s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/3325cc0024594710978ab0f4da8c7d27 : FAILURE in 1h 10m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/4b05c41f045d48bbb1d6179c81bfc8b6 : FAILURE in 1h 09m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/272c77bad46a4d6887ca08c504d7e329 : FAILURE in 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6263a650b21dd1741ec92e08e3543b3f3e351ea3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:43:11.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a0a32613e099be55a9628c1233c900d69122d95","revisions":{"0a0a32613e099be55a9628c1233c900d69122d95":{"kind":"REWORK","_number":1,"created":"2019-09-26 02:12:34.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/70/684870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/70/684870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/70/684870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/70/684870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/70/684870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/70/684870/1"}}},"commit":{"parents":[{"commit":"5af46b5a7605b626896c68fe74a61963e6470974","subject":"Merge \"Revert \"Use http instead of https for builds.midonet.org\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5af46b5a7605b626896c68fe74a61963e6470974"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 02:12:37.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 02:13:15.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I812031838a6fc84bbb9896b280f9806313659ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0a0a32613e099be55a9628c1233c900d69122d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0a0a32613e099be55a9628c1233c900d69122d95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
