)]}'
{"id":"openstack%2Fnetworking-midonet~685385","triplet_id":"openstack%2Fnetworking-midonet~master~I36e2c5efe54bfefe1dcf046f87e399e747a7deea","project":"openstack/networking-midonet","branch":"master","topic":"_update_fip_assoc","hashtags":[],"change_id":"I36e2c5efe54bfefe1dcf046f87e399e747a7deea","subject":"Fix arguments in call to _update_fip_assoc()","status":"MERGED","created":"2019-09-27 15:41:33.000000000","updated":"2019-09-30 05:57:58.000000000","submitted":"2019-09-30 05:44:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685385-1569822272048-fa124ec6","meta_rev_id":"21691f66faf46e0cd83ed6535f21d7a230e54ac5","_number":685385,"virtual_id_number":685385,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-30 04:43:36.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-30 05:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2019-09-30 05:28:44.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-09-30 02:10:03.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2019-09-29 22:27:59.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":1,"date":"2019-09-30 05:28:44.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 17:00:53.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-09-29 22:27:59.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-09-30 02:10:03.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2019-09-30 04:43:36.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-30 05:28:44.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-30 05:44:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"540e2c5d0722da16bcdfbd8c5d0934cab6d6b272","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-27 15:41:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e969f0f1272c7e04cd87f1e9433aeb431df4735","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-09-27 17:00:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"311244dbbb47c083bf8cfb64ac233b0e997fb129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:41:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/891ce9cfe6e34627b56247f01e784865 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb154de5a3fc408a9817b31cb57fa007 : FAILURE in 12m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f184ac9c1c494408b8a4497b59d51995 : FAILURE in 15m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0873e6a4090c4dfca3f329e76b3f597c : FAILURE in 10m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/687d912b942041a7bcc904a9f0b0f5c8 : SUCCESS in 4m 15s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/da9737b293014129ab7c3d7595a86231 : FAILURE in 1h 44m 30s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/1502850ff39e4fd9926c606d58cc7498 : FAILURE in 1h 53m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3874d0627bab466f85a2f8c0d1e5c960 : FAILURE in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"19c57155caef29213e0293b8264784f255b4717e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-09-27 18:50:34.000000000","message":"Patch Set 1:\n\nThere is another create port failure here, which seems to have been around for a while too, I don\u0027t know the obvious way to fix that, sorry.","accounts_in_message":[],"_revision_number":1},{"id":"912ee38d394e671cfe720e0716e00aba974cba24","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-09-29 22:27:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b575d7763526525d8e27686762d96896e477b884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 00:25:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3612ea1b555e4d11bf2fdd5f2e1f4657 : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1da7452fc0e74e4aadd4c2b78c312e7f : SUCCESS in 21m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74a409aaf6ac48ee9eb7e15db70611dc : SUCCESS in 12m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae4f92fef0994ab5a0ac9f574ac9563a : SUCCESS in 24m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae92c73094944fd85945bbe4f64a053 : SUCCESS in 4m 15s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/58ae8d860a5741d3a880ba66614b5345 : FAILURE in 1h 52m 18s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/8a32f278627146b9b5044d3cc6c90f7c : FAILURE in 1h 39m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0925d009fc334bac869c0dfea61106cb : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":2},{"id":"fdf3ca0a323188c111c163cd74fd5f959c51047c","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2019-09-30 02:10:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"640e04bc6b81479d41bfe1e7161e8cd5f41d532f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-30 04:43:36.000000000","message":"Patch Set 2:\n\nthis needs backporting to stable/train","accounts_in_message":[],"_revision_number":2},{"id":"100e1f01d24f08365c15b18a7da3f34c65a37c77","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-30 05:28:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7ac31035a1b6578f7c4ff55ab375ddc429e220cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 05:28:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a5a40ba42054b6361fc11e9cbb243d35e6f4feb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 05:44:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff625e1b771e4f9793dc30eefbeca31f : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0a4bd2bc6464b9dbab21a357c69e507 : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf667de8a7344b3699630562432afdb7 : SUCCESS in 12m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7247322f0d441359b8806e511002f24 : SUCCESS in 13m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03ad0db275e475fa5dd7d3166dd766b : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d1b66b44310df58438fc391f3b5d0a5f5c961382","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 05:44:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"40f77cf0105a8c2cf617f4df0a6b15227c81c541","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 05:45:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fbf29ddd8194ea0befb4ee2a8cc1c0f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2},{"id":"04143fa87ee2b6becba43387938b8089df0d9359","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-30 05:57:58.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/685649 is the backport","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3017c0f0050e912495cd5ceb155f0bed49016fa1","revisions":{"07e972d781c857d9abbe11e2b050323947fc1767":{"kind":"REWORK","_number":1,"created":"2019-09-27 15:41:33.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/85/685385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/85/685385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/85/685385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/85/685385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/85/685385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/85/685385/1"}}},"commit":{"parents":[{"commit":"5af46b5a7605b626896c68fe74a61963e6470974","subject":"Merge \"Revert \"Use http instead of https for builds.midonet.org\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5af46b5a7605b626896c68fe74a61963e6470974"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-09-27 15:38:49.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-09-27 15:38:49.000000000","tz":-240},"subject":"Fix arguments in call to _update_fip_assoc()","message":"Fix arguments in call to _update_fip_assoc()\n\nThe external_port argument was removed from _update_fip_assoc()\nin https://review.opendev.org/#/c/680976/ and is no longer\nneeded.\n\nChange-Id: I36e2c5efe54bfefe1dcf046f87e399e747a7deea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/07e972d781c857d9abbe11e2b050323947fc1767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/07e972d781c857d9abbe11e2b050323947fc1767"}]},"branch":"refs/heads/master"},"3017c0f0050e912495cd5ceb155f0bed49016fa1":{"kind":"REWORK","_number":2,"created":"2019-09-29 22:27:59.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/85/685385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/85/685385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/85/685385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/85/685385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/85/685385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/85/685385/2"}}},"commit":{"parents":[{"commit":"5af46b5a7605b626896c68fe74a61963e6470974","subject":"Merge \"Revert \"Use http instead of https for builds.midonet.org\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5af46b5a7605b626896c68fe74a61963e6470974"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-09-27 15:38:49.000000000","tz":-240},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-29 22:27:40.000000000","tz":540},"subject":"Fix arguments in call to _update_fip_assoc()","message":"Fix arguments in call to _update_fip_assoc()\n\nThe external_port argument was removed from _update_fip_assoc()\nin https://review.opendev.org/#/c/680976/ and is no longer\nneeded.\n\nThis commit includes the following unrelated change to\npass the gate.\n----------------------------------------\nSpecify CONNECTIVITY_L2\n\nNecessary after the recent Neutron change. [1]\n\n[1] https://review.opendev.org/#/c/678027/\n\nbp boot-vm-with-unaddressed-port\nRelated-Bug: #1821058\nChange-Id: Ia2282f2af5f19f1f7fc82614d2fde5974f7316a2\n----------------------------------------\n\nCo-authored-by: YAMAMOTO Takashi \u003cyamamoto@midokura.com\u003e\nChange-Id: I36e2c5efe54bfefe1dcf046f87e399e747a7deea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3017c0f0050e912495cd5ceb155f0bed49016fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3017c0f0050e912495cd5ceb155f0bed49016fa1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
