)]}'
{"id":"openstack%2Fnetworking-midonet~685790","triplet_id":"openstack%2Fnetworking-midonet~stable%2Fstein~Ia109c6eba2cff529949567b0af821985a7f9efdf","project":"openstack/networking-midonet","branch":"stable/stein","topic":"bindep-fallback","hashtags":[],"change_id":"Ia109c6eba2cff529949567b0af821985a7f9efdf","subject":"Fix versions of neutron and friends","status":"ABANDONED","created":"2019-09-30 20:59:49.000000000","updated":"2019-10-01 01:00:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"742ac7860b725e4573cdf8f4d62edc75d973a471","_number":685790,"virtual_id_number":685790,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-01 00:35:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 00:35:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bed430b3532225241de105b0ac5f0f15ef637144","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-09-30 20:59:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f16b57d8a92354fa704bf7f6b94c3252645793e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 21:25:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/70a14618b65849ddb0687bb1112979c8 : RETRY_LIMIT in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa1aadb90bc487581a1c717b86b09d6 : FAILURE in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0009c5edd9f941d78e2fa4032416e332 : RETRY_LIMIT in 3m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a677f3b7e46549b49aedd6b78af80e02 : RETRY_LIMIT in 2m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc90af643d704c9d86e18d19a379e422 : RETRY_LIMIT in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06364976a20b4258ba50ba3264e14f34 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48321fbfc0fa4d98b017c6a4acf4d749 : SUCCESS in 5m 07s\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/8233b27656d9474eb4688600f7b873c3 : FAILURE in 12m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f580c9e820d42ae8ce02a20c9c84b3a : RETRY_LIMIT in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"07106c2320f3a44112b2cce033edb1e31b5891ba","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-09-30 22:02:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8d078ac63de034ed2f1590ec88bea13c4246a079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 22:29:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/64d259c008fc4de080eaf3685caeaacc : FAILURE in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5020735b6964cc9bdf8c7dc5dc761b4 : FAILURE in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70a636daf4f941709df91c01847f93ae : FAILURE in 12m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e342bfab74884579aea5a62a8f8e5d65 : FAILURE in 13m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/212181e893224af6b7f805ce41db028e : FAILURE in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e5be068eff442df9efa0ce58149d0ca : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9ac61a4ad7e44ef8fcf16c1b2cd3f60 : SUCCESS in 5m 08s\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/f1e60716abea4d28a7c085525ce51a9e : FAILURE in 21m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71be285cba6c41a68f0be3c6108dc7f8 : FAILURE in 11m 54s","accounts_in_message":[],"_revision_number":2},{"id":"72815ce86ac02608b5d245b4083613d0bd710a0a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-09-30 22:51:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d0bebafc37d11ce4c25f3f2213df65ceb42e7a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 23:35:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/60e2a4f075274e4799f27d1b3f99a0b2 : FAILURE in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1143fd76034c404e8b1231881dbeb079 : FAILURE in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dddc700b71ad47e29b215ac4c537725e : FAILURE in 12m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ad8c14efbcbb4122b4b0e3234b7bd6b6 : FAILURE in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58f99250e4d94053968fe40550d9fc01 : FAILURE in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f42023b3aa4898a8285420ffca54f5 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6a6a5d410b44802a87e5904f9a0fb88 : SUCCESS in 5m 23s\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/51bac84cf286448ba7780401aeca541a : FAILURE in 13m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8d03565834c4906ad5b153ec28cef3b : FAILURE in 10m 28s","accounts_in_message":[],"_revision_number":3},{"id":"a9250148642bc12721ebf569d9a2109ea6d2ada7","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-01 00:12:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"47781775463cc006aa14d158212e4831f47727a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 00:35:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7fe778829218491fae8c51045a42d38f : FAILURE in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56af152b37a4c31889d31fe21057dab : SUCCESS in 7m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/740327890079427c9a54f238c67dec26 : FAILURE in 11m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c1835a6a68e7411fa5b1afe7e1e90227 : FAILURE in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c198abfabc744181af6347665888c4a1 : FAILURE in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869a7911cfb0499a8e1dd4e50bb15b58 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55ddbef539c040949b5e7db7c6096075 : SUCCESS in 5m 07s\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/c06c7da4e1ba4acdac5fad54ff570e38 : FAILURE in 16m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87b9f095136c4b5cbc031a7c6d8b2b9d : FAILURE in 12m 28s","accounts_in_message":[],"_revision_number":4},{"id":"e6798df5151e854e829b90d911b7f1b9c37a3e1d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-01 01:00:32.000000000","message":"Abandoned\n\nmerged into https://review.opendev.org/#/c/668824/","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"51b9c0de504c875064128972c3a4b4a57ddbf6c7","revisions":{"5765607b329ed732d4dd347b555b394a6eca0a8d":{"kind":"REWORK","_number":1,"created":"2019-09-30 20:59:49.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/90/685790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/90/685790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/1"}}},"commit":{"parents":[{"commit":"5b4437e94e107aefb2ec0a1601fe7ea7895eca53","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5b4437e94e107aefb2ec0a1601fe7ea7895eca53"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 20:59:06.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 20:59:06.000000000","tz":540},"subject":"Cap versions of neutron and friends","message":"Cap versions of neutron and friends\n\nChange-Id: Ia109c6eba2cff529949567b0af821985a7f9efdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5765607b329ed732d4dd347b555b394a6eca0a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5765607b329ed732d4dd347b555b394a6eca0a8d"}]},"branch":"refs/heads/stable/stein"},"e3266a0351bbfc170adb14898c87848a91773cf8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-30 22:02:28.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/90/685790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/90/685790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/2"}}},"commit":{"parents":[{"commit":"4c1ce091ca662c65b45c78b7cc807654c59e8854","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c1ce091ca662c65b45c78b7cc807654c59e8854"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 20:59:06.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 22:02:14.000000000","tz":540},"subject":"Cap versions of neutron and friends","message":"Cap versions of neutron and friends\n\nChange-Id: Ia109c6eba2cff529949567b0af821985a7f9efdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e3266a0351bbfc170adb14898c87848a91773cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e3266a0351bbfc170adb14898c87848a91773cf8"}]},"branch":"refs/heads/stable/stein"},"28190b567daf72e2047a03c7e5d82bd68aa0d855":{"kind":"REWORK","_number":3,"created":"2019-09-30 22:51:17.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/90/685790/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/90/685790/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/3"}}},"commit":{"parents":[{"commit":"4c1ce091ca662c65b45c78b7cc807654c59e8854","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c1ce091ca662c65b45c78b7cc807654c59e8854"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 20:59:06.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 22:51:01.000000000","tz":540},"subject":"Cap versions of neutron and friends","message":"Cap versions of neutron and friends\n\nChange-Id: Ia109c6eba2cff529949567b0af821985a7f9efdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/28190b567daf72e2047a03c7e5d82bd68aa0d855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/28190b567daf72e2047a03c7e5d82bd68aa0d855"}]},"branch":"refs/heads/stable/stein"},"51b9c0de504c875064128972c3a4b4a57ddbf6c7":{"kind":"REWORK","_number":4,"created":"2019-10-01 00:12:53.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/90/685790/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/90/685790/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/90/685790/4"}}},"commit":{"parents":[{"commit":"4c1ce091ca662c65b45c78b7cc807654c59e8854","subject":"Add bindep.txt to fix unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c1ce091ca662c65b45c78b7cc807654c59e8854"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 20:59:06.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-10-01 00:12:27.000000000","tz":540},"subject":"Fix versions of neutron and friends","message":"Fix versions of neutron and friends\n\nChange-Id: Ia109c6eba2cff529949567b0af821985a7f9efdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/51b9c0de504c875064128972c3a4b4a57ddbf6c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/51b9c0de504c875064128972c3a4b4a57ddbf6c7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
