)]}'
{"id":"openstack%2Fnetworking-midonet~685889","triplet_id":"openstack%2Fnetworking-midonet~master~Idbea72514b8869a9a194987ddde36abe369b5496","project":"openstack/networking-midonet","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Idbea72514b8869a9a194987ddde36abe369b5496","subject":"Correct the PDF filename","status":"MERGED","created":"2019-10-01 08:02:58.000000000","updated":"2019-10-01 09:00:20.000000000","submitted":"2019-10-01 08:58:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685889-1569920293208-a241494d","meta_rev_id":"9c9c8befc520d2f5710c0c52d5f7049ec38abd4c","_number":685889,"virtual_id_number":685889,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-01 08:24:34.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 08:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-10-01 08:43:01.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-10-01 08:19:18.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-10-01 08:43:01.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 08:19:18.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2019-10-01 08:43:01.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-10-01 08:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6844c8f8f39af0375104d37ac5f3b63b5c56acc9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-01 08:02:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"294f2866e2116194776791030b4c43904a8beec1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-01 08:07:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c1386f43dbcc4da290322a9b9a55b8c7da88768e","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2019-10-01 08:19:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ce7f914888c662a334ad5883e7a5bfd6f057c34f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:19:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc6db3b6eef2467cb444ffd5fa1b6952 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e959ea90de19484098256b013927e002 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":2},{"id":"7f8b2a1c363a350a848ee97f3c15e80b5b03ed4e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-01 08:24:34.000000000","message":"Patch Set 2:\n\nThe PDF file is found expectedly https://zuul.opendev.org/t/openstack/build/e959ea90de19484098256b013927e002","accounts_in_message":[],"_revision_number":2},{"id":"c96b6b9c4ca4d43cc2d7fb71516bb7bba9d79bd9","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-10-01 08:43:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"392b8a001baebe6917eedc5c69673ec4a9b5636a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:43:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c2072a7962c70203fa44779d4c3fadae69ca332","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:58:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272910031ed14057bbe06a9b2445b7bc : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f7584e21f1a425fbab60ef397140266 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4efc5045214548ad7f84552a61102af7e2452953","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:58:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7a5af4468ef2b9dc7c8786536b00363adb17e153","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 09:00:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e97a2469f56497d9482d134cffaf995 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5d3de26ea56c2eaf340a1340536bf687a8dfa631","revisions":{"7de186d9d4aeb24788c47f5fcc9ddd27d4c82086":{"kind":"REWORK","_number":1,"created":"2019-10-01 08:02:58.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/89/685889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/89/685889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/89/685889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/89/685889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/89/685889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/89/685889/1"}}},"commit":{"parents":[{"commit":"b52018e83d2b4cd6f086141231487ab04f901539","subject":"Merge \"PDF documentation build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b52018e83d2b4cd6f086141231487ab04f901539"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-01 08:00:52.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-01 08:00:52.000000000","tz":540},"subject":"Correct the PDF filename","message":"Correct the PDF filename\n\nThe zuul job assumes doc-\u003crepo\u003e.pdf as the generated PDF file.\nIt was changed after the original PDF build patch was pushed\nand I forgot to update it. This commit fixes it.\n\nChange-Id: Idbea72514b8869a9a194987ddde36abe369b5496\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7de186d9d4aeb24788c47f5fcc9ddd27d4c82086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7de186d9d4aeb24788c47f5fcc9ddd27d4c82086"}]},"branch":"refs/heads/master"},"5d3de26ea56c2eaf340a1340536bf687a8dfa631":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-01 08:07:01.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/89/685889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/89/685889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/89/685889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/89/685889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/89/685889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/89/685889/2"}}},"commit":{"parents":[{"commit":"b52018e83d2b4cd6f086141231487ab04f901539","subject":"Merge \"PDF documentation build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b52018e83d2b4cd6f086141231487ab04f901539"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-01 08:00:52.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-01 08:06:58.000000000","tz":0},"subject":"Correct the PDF filename","message":"Correct the PDF filename\n\nThe zuul job assumes doc-\u003crepo\u003e.pdf as the generated PDF file.\nIt was changed after the original PDF build patch was pushed\nand I forgot to update it. This commit fixes it.\n\nChange-Id: Idbea72514b8869a9a194987ddde36abe369b5496\nStory: 2006099\nTask: 35123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5d3de26ea56c2eaf340a1340536bf687a8dfa631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5d3de26ea56c2eaf340a1340536bf687a8dfa631"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
