)]}'
{"id":"openstack%2Fnetworking-midonet~686127","triplet_id":"openstack%2Fnetworking-midonet~stable%2Fstein~Ie96d1cae98e0d071f51ab98c4c6516df5ed11806","project":"openstack/networking-midonet","branch":"stable/stein","topic":"bug/1819893","hashtags":[],"change_id":"Ie96d1cae98e0d071f51ab98c4c6516df5ed11806","subject":"Switch to tap-as-a-service-tempest-plugin","status":"ABANDONED","created":"2019-10-02 10:28:44.000000000","updated":"2021-04-14 13:51:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bcdcc615d33c3711736cbe9a2c61f09833f515b0","_number":686127,"virtual_id_number":686127,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2020-11-17 23:57:27.000000000","_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2020-11-17 14:30:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2020-11-17 14:09:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-10-17 11:01:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-11-17 14:09:47.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"date":"2021-04-14 13:51:35.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-10 04:35:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-17 11:01:13.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-17 14:09:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-11-17 14:30:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 23:57:27.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"}],"messages":[{"id":"8340bfc446725cecc78ebb67a5b3296146d4dd2b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-02 10:28:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de4be0c6762ebe001ba27ff88d8a76cc6e815d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 12:40:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78856d6fe5bb41988f5c71b5c14072cc : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e722e59a332f44b98030ab381409056a : SUCCESS in 12m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5b64af20833740d1a475728d0b110c1a : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f97821a40964408ebaed8077ef4cc241 : SUCCESS in 16m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93626b15b62a428c908128217f2cccd7 : SUCCESS in 5m 27s\n- networking-midonet-tempest-multinode-ml2 https://zuul.opendev.org/t/openstack/build/b06ce7669ed940b49c7e8b97b90c3db0 : SUCCESS in 1h 52m 39s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://zuul.opendev.org/t/openstack/build/19ba5d0f5d7d48afa11d3b5e260c1f02 : FAILURE in 1h 51m 21s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://zuul.opendev.org/t/openstack/build/116e7c0d84844729a185c349d45fcecf : FAILURE in 18m 31s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/1a4c8bcb382f4f1998924e40c2033832 : SUCCESS in 1h 36m 51s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/aeb2382609654423a966f8e69e4e8d10 : FAILURE in 2h 02m 36s (non-voting)\n- networking-midonet-grenade-ml2 https://zuul.opendev.org/t/openstack/build/7634b868fc2f4ad2b5d2a9dcf3ceca04 : FAILURE in 14m 49s (non-voting)\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/c77dc05ad5154ef487bb6f08e0f24076 : FAILURE in 18m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/012f7d9065f849a480f8fd6acc58dcec : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9af332a53e3328c89012b441933c8ef17d824b36","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-17 11:01:13.000000000","message":"Patch Set 1: Code-Review+2\n\nConflicts are on openstack/x prefix change","accounts_in_message":[],"_revision_number":1},{"id":"8a864d195126def75316b15751c0bd765377bcd8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-17 14:09:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d719a35b4da0e329af5fba601c721a77d1d0459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 14:10:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcf4f8115ece35a8e980549552015be02d847c4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 14:30:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d7996dd8334f36a9ba889481a2a31f : FAILURE in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/586c413618a840d0b52dde24a79a4efe : SUCCESS in 11m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6b6fed3e8f48463c9b40dd4802089f34 : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b32dac04d9804af682b315977fcc13d8 : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3604f87a94d04e75a6958cf046099d69 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5b438aaa266e720ce12b817897b6721f776001db","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-17 23:57:27.000000000","message":"Patch Set 1:\n\nWill need 18663cd4d37bda46589ad105aca8b05f058a8cfd backported I think","accounts_in_message":[],"_revision_number":1},{"id":"bcdcc615d33c3711736cbe9a2c61f09833f515b0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:51:36.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"141b966c6a9304c12a9d7d828e013aa95b34f4f7","revisions":{"141b966c6a9304c12a9d7d828e013aa95b34f4f7":{"kind":"REWORK","_number":1,"created":"2019-10-02 10:28:44.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/27/686127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/27/686127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/27/686127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/27/686127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/27/686127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/27/686127/1"}}},"commit":{"parents":[{"commit":"5165d0ad661837f89195f61c54783b2bcc050897","subject":"Remove ocata jobs from branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5165d0ad661837f89195f61c54783b2bcc050897"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-04-02 12:58:26.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-10-02 10:28:17.000000000","tz":540},"subject":"Switch to tap-as-a-service-tempest-plugin","message":"Switch to tap-as-a-service-tempest-plugin\n\nCloses-Bug: #1819893\nChange-Id: Ie96d1cae98e0d071f51ab98c4c6516df5ed11806\n(cherry picked from commit 4d3f7569c458051996e532b292fb3fba987a0a34)\nConflicts:\n    .zuul.yaml\n    playbooks/tempest-aio-ml2-full-centos-7/run.yaml\n    playbooks/tempest-aio-ml2-full-legacy/run.yaml\n    playbooks/tempest-multinode-ml2-full/run.yaml\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/141b966c6a9304c12a9d7d828e013aa95b34f4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/141b966c6a9304c12a9d7d828e013aa95b34f4f7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
