)]}'
{"id":"openstack%2Fnetworking-midonet~686165","triplet_id":"openstack%2Fnetworking-midonet~stable%2Fstein~Ibb4d6fe6d810b967307242a3c68dd038ab73958c","project":"openstack/networking-midonet","branch":"stable/stein","topic":"bug/1819893","hashtags":[],"change_id":"Ibb4d6fe6d810b967307242a3c68dd038ab73958c","subject":"Switch to use neutron-tempest-plugin for neutron-dynamic-routing tests","status":"ABANDONED","created":"2019-10-02 13:35:38.000000000","updated":"2021-04-14 13:51:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c6726fd3deff36b61eac47aaebcdb73f78971808","_number":686165,"virtual_id_number":686165,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"date":"2019-10-15 04:57:28.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"date":"2020-11-17 23:57:23.000000000","_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-17 20:59:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2019-10-17 11:12:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2019-10-16 08:04:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-11-16 10:32:55.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-16 08:04:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-11-16 10:32:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-11-17 20:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 23:57:23.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"}],"messages":[{"id":"f6c15be5c31c7d0a12f341d831cf34531df426de","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-02 13:35:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03047093ef90744a481664145bb7257a17f27c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 15:27:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15395f4129de4b5babb2d94afae1b307 : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0b08490076d44e3924f875c80b38956 : SUCCESS in 11m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f67533e629e04576a111794dea026ade : SUCCESS in 12m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9352bdff16ca4b1e9db837862780139c : SUCCESS in 14m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfad096eb24e4f59be0571169a9c9fa0 : SUCCESS in 4m 56s\n- networking-midonet-tempest-multinode-ml2 https://zuul.opendev.org/t/openstack/build/c9b438acb3fc4adc97979c2f74dca566 : SUCCESS in 1h 45m 23s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://zuul.opendev.org/t/openstack/build/aea57e4c69904f2db31cf7ac6e5e0884 : FAILURE in 30m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://zuul.opendev.org/t/openstack/build/12d67d760f82428fbc95e483e4c35453 : FAILURE in 10m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/c8b4d9610c4143568b236fdc07536686 : SUCCESS in 1h 32m 46s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/4644c78516d249728faf6a869b91e5d3 : FAILURE in 41m 01s (non-voting)\n- networking-midonet-grenade-ml2 https://zuul.opendev.org/t/openstack/build/34888fc824ac48ec8548537f4226a093 : FAILURE in 15m 55s (non-voting)\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/9c7e2c1875de44cf9f29c5e13329204a : FAILURE in 13m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5c7369daf9647beb274976f5f9c403d : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6a5e2d475508e67bdc28334ad83582479f2b307d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-03 01:10:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"713dbdf589bff1d9699bf5c3f0ef594852c4fca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 02:52:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da13f1694b34656b6b7dae0cccd922c : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15a07f51f8c54476a266e392c6c3b497 : SUCCESS in 13m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd9a7006538f4f38a15b0647b5ad9e9b : SUCCESS in 20m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc5efe76743043a1a03cd397401be9ce : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a40ca7e5d8d34867b399763d291f1f4e : SUCCESS in 4m 34s\n- networking-midonet-tempest-multinode-ml2 https://zuul.opendev.org/t/openstack/build/ac56eb507f74469987e17f8d73231920 : SUCCESS in 1h 34m 59s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://zuul.opendev.org/t/openstack/build/4cd31e95effc439ca14fb10533591bc2 : SUCCESS in 1h 32m 49s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://zuul.opendev.org/t/openstack/build/15da6dbf897f4fe3a5003147483ee7b9 : FAILURE in 14m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/819a9a4980f5423796d279b86d4f2e3e : SUCCESS in 1h 23m 20s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/8dc17a6b962e4e37a52b0b73b8064b37 : SUCCESS in 1h 35m 11s (non-voting)\n- networking-midonet-grenade-ml2 https://zuul.opendev.org/t/openstack/build/095cc930bbc04d1499688e0b3b945aa3 : FAILURE in 14m 10s (non-voting)\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/eadf11d7ed4b4e6a98916c3439502a29 : FAILURE in 15m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64b38444b0dd456d86a1f3ba85019e47 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":2},{"id":"508be68a68ad5c3cff1c4e9d06559fee838e5a51","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-10 04:34:42.000000000","message":"Patch Set 2:\n\nlegacy job failure looks unrelated. for some reasons it seems running on bionic.","accounts_in_message":[],"_revision_number":2},{"id":"fb5461fe813fb58ca79727405f78d4d41066a888","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-11 10:56:20.000000000","message":"Patch Set 2: Code-Review-1\n\nin stable branches those tests are still in neutron-dynamic-routing repo. We don\u0027t run this job defined in neutron-tempest-plugin for stable branches of neutron-dynamic-routing. So do we really need such change in stable branches for networking-midonet?","accounts_in_message":[],"_revision_number":2},{"id":"ca734ca5039758f02ae705f907598139ff61d455","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-11 13:58:14.000000000","message":"Patch Set 2:\n\nSlawek,\n\ntempest and its plugin should be branchless.\nit\u0027s one of the reasons we introduced neutron-tempest-plugin in the first place.\n\ni don\u0027t see any reason to prefer the copy in neutron-dynamic-routing.\nhaving tempest-plugin in the same repo has been a problem for us. eg. it has been one of blockers of lower-constraints job on this repo.\n\nIMO, neutron-dynamic-routing should switch to use neutron-tempest-plugin as well. I don\u0027t insist though.","accounts_in_message":[],"_revision_number":2},{"id":"fa22fcc14656eb4fcf1783c771ffc07750139917","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-14 08:10:30.000000000","message":"Patch Set 2:\n\n\u003e Slawek,\n \u003e \n \u003e tempest and its plugin should be branchless.\n \u003e it\u0027s one of the reasons we introduced neutron-tempest-plugin in the\n \u003e first place.\n \u003e \n \u003e i don\u0027t see any reason to prefer the copy in neutron-dynamic-routing.\n \u003e having tempest-plugin in the same repo has been a problem for us.\n \u003e eg. it has been one of blockers of lower-constraints job on this\n \u003e repo.\n \u003e \n \u003e IMO, neutron-dynamic-routing should switch to use neutron-tempest-plugin\n \u003e as well. I don\u0027t insist though.\n\nI know that neutron-tempest-plugin is branchless. But from the other hand with this migration of tests from e.g. neutron-dynamic-routing to the neutron-tempest-plugin repo we can be only sure that those tests works for ussuri (current master) and newer as we are not running those tests against e.g. stable/train or stein. For those branches neutron-dynamic-routing tempest plugin is still branched and placed in neutron-dynamic-routing repo.\nAnd we can\u0027t remove those tests from stable branches easily as this could e.g. cause problems with packaging of those stable branches for distros like rdo.\n\nBecause of that I think that it would be better to run those tests only on branches on which we are running them also in neutron-tempest-plugin queues (so ussuri and newer).\n\nIs my understanding wrong here?","accounts_in_message":[],"_revision_number":2},{"id":"6b46ed7c48cd9115102298b9a9c2b31d92784b5a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-15 04:57:28.000000000","message":"Patch Set 2:\n\nSlawek,\n\ni guess you are missing two points:\n\n- this patch actually skips all tests migrated from neutron-dynamic-routing.\n\n- for networking-midonet, actually there\u0027s no choice because the test we care (namely neutron_tempest_plugin.neutron_dynamic_routing.scenario.test_simple_bgp.Bgp) is not available in neutron-dynamic-routing repo. it has been migrated from networking-midonet repo, not neutron-dynamic-routing.\n\nwrt supporting older branches, i\u0027m fine as far as there\u0027s a reasonable way to skip them. (as this patch does)","accounts_in_message":[],"_revision_number":2},{"id":"a9c9a793f04ec883e809f85826f3734c0b9d53b3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-16 08:04:33.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e Slawek,\n \u003e \n \u003e i guess you are missing two points:\n \u003e \n \u003e - this patch actually skips all tests migrated from\n \u003e neutron-dynamic-routing.\n \u003e \n \u003e - for networking-midonet, actually there\u0027s no choice because the\n \u003e test we care (namely neutron_tempest_plugin.neutron_dynamic_routing.scenario.test_simple_bgp.Bgp)\n \u003e is not available in neutron-dynamic-routing repo. it has been\n \u003e migrated from networking-midonet repo, not neutron-dynamic-routing.\n \u003e \n \u003e wrt supporting older branches, i\u0027m fine as far as there\u0027s a\n \u003e reasonable way to skip them. (as this patch does)\n\nOk, I see now. Thx.","accounts_in_message":[],"_revision_number":2},{"id":"83f8d1e0ad2247e739857ed6207ececb24eb3601","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-17 11:12:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nOK so this is to get an almost \"networking-midonet test\" in, lgtm","accounts_in_message":[],"_revision_number":2},{"id":"1e33db4f958e7d3681bed966636bc01ed29f8698","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-18 13:35:37.000000000","message":"Patch Set 2: -Workflow\n\nLet\u0027s get train https://review.opendev.org/#/c/686163 in first","accounts_in_message":[],"_revision_number":2},{"id":"024091d6ca45fa20f5cdba200c71bb7ad9cadd11","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-16 10:32:55.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"76fe58c89200dbd9aa760b494d1790a9a36b718c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 14:10:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"716e8d76aba4e94a268b4a553ebcfa0ded256430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 14:30:38.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"42d9bea6739cee3e9d7a4dd0942393f66e2c897d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-11-17 20:23:32.000000000","message":"Patch Set 2:\n\nrecheck test pep8 result","accounts_in_message":[],"_revision_number":2},{"id":"3c74167d3b3e517565cbca08b0117c026baeab3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 20:59:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec96799576748aba66540551d20c0a4 : FAILURE in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e09db4d34b348c6b3afb40bb4656dbd : SUCCESS in 11m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2412e14107874db5977a1883c6074e18 : SUCCESS in 13m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/419020756bf34be7bd546ef80656af17 : SUCCESS in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9fa23ca71f7459f91c49a01688a11de : SUCCESS in 5m 26s\n- networking-midonet-tempest-multinode-ml2 https://zuul.opendev.org/t/openstack/build/65b94f8cdf5b4abba416f6c4dee3bd77 : FAILURE in 25m 30s (non-voting)\n- networking-midonet-tempest-multinode-ml2-full https://zuul.opendev.org/t/openstack/build/b13fb35a9cf3488f8c76a09e10edca2d : FAILURE in 16m 25s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://zuul.opendev.org/t/openstack/build/faff4eef644c42b487e074f948b17103 : FAILURE in 10m 38s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/a4c5f59e01214ed386b9b184b33d5123 : FAILURE in 21m 07s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/7401444ffe59483d8b8378999a011327 : FAILURE in 24m 45s (non-voting)\n- networking-midonet-grenade-ml2 https://zuul.opendev.org/t/openstack/build/26824af805204b4886844766cbd30e51 : FAILURE in 19m 37s (non-voting)\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/055c15dc262b4078b07e797a588cfec7 : FAILURE in 13m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e00ce724afb046818a3d6e8545c81d73 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e583794466850c6cdc4be49a7fef1da8306d40fc","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-17 23:57:23.000000000","message":"Patch Set 2:\n\nWill need 18663cd4d37bda46589ad105aca8b05f058a8cfd backported I think","accounts_in_message":[],"_revision_number":2},{"id":"c6726fd3deff36b61eac47aaebcdb73f78971808","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:51:43.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"516a160e9c8b8a06ee03177cc5fa6bff3c3be849","revisions":{"920ca73ac34845f6ba11ae3024ff157fe469c012":{"kind":"REWORK","_number":1,"created":"2019-10-02 13:35:38.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/65/686165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/65/686165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/65/686165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/65/686165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/65/686165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/65/686165/1"}}},"commit":{"parents":[{"commit":"5165d0ad661837f89195f61c54783b2bcc050897","subject":"Remove ocata jobs from branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/5165d0ad661837f89195f61c54783b2bcc050897"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 21:27:33.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-10-02 13:35:06.000000000","tz":540},"subject":"Switch to use neutron-tempest-plugin for neutron-dynamic-routing tests","message":"Switch to use neutron-tempest-plugin for neutron-dynamic-routing tests\n\nDepends-On: https://review.opendev.org/#/c/652099/\nChange-Id: Ibb4d6fe6d810b967307242a3c68dd038ab73958c\n(cherry picked from commit 5ad54cdd3542e02110c2e60616fcde76dee6eee6)\nConflicts:\n    devstack/ci/gate_hook.sh\n    devstack/ci/hooks/tempest_plugins_advsvc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/920ca73ac34845f6ba11ae3024ff157fe469c012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/920ca73ac34845f6ba11ae3024ff157fe469c012"}]},"branch":"refs/heads/stable/stein"},"516a160e9c8b8a06ee03177cc5fa6bff3c3be849":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-03 01:10:06.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/65/686165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/65/686165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/65/686165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/65/686165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/65/686165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/65/686165/2"}}},"commit":{"parents":[{"commit":"141b966c6a9304c12a9d7d828e013aa95b34f4f7","subject":"Switch to tap-as-a-service-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/141b966c6a9304c12a9d7d828e013aa95b34f4f7"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-09-30 21:27:33.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-10-03 01:09:49.000000000","tz":540},"subject":"Switch to use neutron-tempest-plugin for neutron-dynamic-routing tests","message":"Switch to use neutron-tempest-plugin for neutron-dynamic-routing tests\n\nDepends-On: https://review.opendev.org/#/c/652099/\nChange-Id: Ibb4d6fe6d810b967307242a3c68dd038ab73958c\n(cherry picked from commit 5ad54cdd3542e02110c2e60616fcde76dee6eee6)\nConflicts:\n    devstack/ci/gate_hook.sh\n    devstack/ci/hooks/tempest_plugins_advsvc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/516a160e9c8b8a06ee03177cc5fa6bff3c3be849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/516a160e9c8b8a06ee03177cc5fa6bff3c3be849"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
