)]}'
{"id":"openstack%2Fnetworking-midonet~687208","triplet_id":"openstack%2Fnetworking-midonet~stable%2Frocky~Ie96d1cae98e0d071f51ab98c4c6516df5ed11806","project":"openstack/networking-midonet","branch":"stable/rocky","topic":"bindep-fallback","hashtags":[],"change_id":"Ie96d1cae98e0d071f51ab98c4c6516df5ed11806","subject":"Switch to tap-as-a-service-tempest-plugin","status":"ABANDONED","created":"2019-10-08 03:34:56.000000000","updated":"2020-09-25 15:33:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d0b0da3009fd5a9687c311b430012f3ff40d13b5","_number":687208,"virtual_id_number":687208,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-25 15:33:51.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-08 05:13:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 05:13:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 15:33:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"60f73c06a178b772f6571ab912392a7f5b16ff54","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-10-08 03:34:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1dc993225407610f3f86fd053ae5cec63cd78a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 05:13:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067a108817764024a9292f80b8ca3c1a : SUCCESS in 8m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89fe82a503754406af622902b0ff30a1 : FAILURE in 16m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e3e6b3829ff4b0a9a0243ec46e466c9 : FAILURE in 15m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/745e89d090c04aaeb1cac631306f812e : SUCCESS in 4m 43s\n- networking-midonet-tempest-aio-ml2 https://zuul.opendev.org/t/openstack/build/96b03fcc6acc4f29a33f574bab8e54bc : SUCCESS in 59m 23s\n- networking-midonet-tempest-multinode-ml2 https://zuul.opendev.org/t/openstack/build/939e92ca54e6420ab1df6ff84c7a9fea : SUCCESS in 1h 17m 02s (non-voting)\n- networking-midonet-tempest-aio-ml2-full https://zuul.opendev.org/t/openstack/build/443a7ae2e9ed4b0ea1fcc19eabfe734d : FAILURE in 27m 40s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://zuul.opendev.org/t/openstack/build/77fb69c9bcd64beb82df02c60afd4c31 : FAILURE in 33m 19s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/562448cdc14a4727bca414053cd980e1 : SUCCESS in 1h 27m 00s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/2e1e88bcba834f29b453fab99292b7da : FAILURE in 46m 40s (non-voting)\n- networking-midonet-grenade-ml2 https://zuul.opendev.org/t/openstack/build/6271856d68b44d28b3244ab6090bbfdf : SUCCESS in 1h 29m 52s (non-voting)\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/f75c2b165b3c4a2ca6e8e8b7452f526e : FAILURE in 30m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae4353af76e3446bb48c91941239c6dc : FAILURE in 13m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4a2d9b616fce35746d0a361fd69bfbdb3c65062","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-04 08:39:30.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"ba1cfc16e549e872b34c2f1cfe7d670e46c054c6","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-05 01:21:21.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"4423be9c7922c11c7442762e5f78ea7578838095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 03:34:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac49826130a465d983ecd439f552fd4 : FAILURE in 8m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95f664578ee141ccac7a6b75b5a4d226 : SUCCESS in 14m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e4ebacfeb1446a7a6dda995bef5827c : SUCCESS in 14m 38s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/19d913e327a64c4a946049430ca01d4e : SUCCESS in 5m 33s\n- networking-midonet-tempest-aio-ml2 https://zuul.opendev.org/t/openstack/build/5c0d46a0889642058770ef31808bc02f : SUCCESS in 1h 25m 50s\n- networking-midonet-tempest-multinode-ml2 https://zuul.opendev.org/t/openstack/build/2fb42a5461874aba8247f4378a419c4b : SUCCESS in 1h 26m 42s (non-voting)\n- networking-midonet-tempest-aio-ml2-full https://zuul.opendev.org/t/openstack/build/6f10774bd45f472d848ebee6bc95354f : FAILURE in 2h 00m 19s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-legacy https://zuul.opendev.org/t/openstack/build/d668b52234f24f82aca471ae1f50df7b : FAILURE in 2h 04m 06s (non-voting)\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/299e0a442e574ca68d51f136906bd32e : FAILURE in 10m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/ca31d5dc9ed94ca584b3ccbcddbe0fd6 : FAILURE in 12m 40s (non-voting)\n- networking-midonet-grenade-ml2 https://zuul.opendev.org/t/openstack/build/6923397f2843445e93e18fad718da147 : SUCCESS in 1h 19m 16s (non-voting)\n- networking-midonet-rally-ml2 https://zuul.opendev.org/t/openstack/build/ff09b915cfa2477a96c4ca481d97f54d : FAILURE in 30m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7da0142c1f344aaa3f54d0acec2fef1 : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b0219fa0e86a45536a3f962f8d11e3e4bdc971a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:33:51.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f7c2f462685e54d6856673cb468ea0ca700dd2d","revisions":{"9f7c2f462685e54d6856673cb468ea0ca700dd2d":{"kind":"REWORK","_number":1,"created":"2019-10-08 03:34:56.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/08/687208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/08/687208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/08/687208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/08/687208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/08/687208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/08/687208/1"}}},"commit":{"parents":[{"commit":"fe0c20b33155c809e31c85cfb92c9818032315f6","subject":"Remove tempest plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/fe0c20b33155c809e31c85cfb92c9818032315f6"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-04-02 12:58:26.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-10-08 03:34:06.000000000","tz":540},"subject":"Switch to tap-as-a-service-tempest-plugin","message":"Switch to tap-as-a-service-tempest-plugin\n\nCloses-Bug: #1819893\nChange-Id: Ie96d1cae98e0d071f51ab98c4c6516df5ed11806\n(cherry picked from commit 4d3f7569c458051996e532b292fb3fba987a0a34)\nConflicts:\n    .zuul.yaml\n    playbooks/tempest-aio-ml2-full-centos-7/run.yaml\n    playbooks/tempest-aio-ml2-full-legacy/run.yaml\n    playbooks/tempest-multinode-ml2-full/run.yaml\n(cherry picked from commit 141b966c6a9304c12a9d7d828e013aa95b34f4f7)\nConflicts:\n    devstack/ci/gate_hook.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9f7c2f462685e54d6856673cb468ea0ca700dd2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9f7c2f462685e54d6856673cb468ea0ca700dd2d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
