)]}'
{"id":"openstack%2Fnetworking-midonet~693330","triplet_id":"openstack%2Fnetworking-midonet~master~I6307c90b70cbd9c4bac1d26f36a5becad538b293","project":"openstack/networking-midonet","branch":"master","topic":"policy","hashtags":[],"change_id":"I6307c90b70cbd9c4bac1d26f36a5becad538b293","subject":"Remove policy.json as it doesn\u0027t seem to be used","status":"MERGED","created":"2019-11-07 10:12:27.000000000","updated":"2019-11-13 02:38:42.000000000","submitted":"2019-11-13 02:37:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693330-1573612655626-aa86e2ef","meta_rev_id":"c276ce92050c40f9a7019423235f1085a0c4db8c","_number":693330,"virtual_id_number":693330,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 02:37:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2019-11-12 14:10:35.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-11-08 02:20:48.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-11-13 02:12:51.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 14:10:35.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-11-13 02:12:51.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2019-11-13 02:37:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bdc4688386c5a50da81a32f6e0dfeec38c22a2d","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-11-07 10:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a81e7c08056eca77ae2a37d1b5dfeb3a6f6b2eac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 10:36:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e6cf8f44724f3eba1a91e1565152df : SUCCESS in 9m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b70fd7208b604f91b84bc9bc9dd5785f : SUCCESS in 19m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5868b47189704e94bc69d6aa0fc3dd8f : SUCCESS in 16m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f93c2c56098e41d18951b53e3f9448ea : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba410ef82e34d9996c1e79a30bb5b7b : SUCCESS in 7m 58s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/fa07710bd8c043409ccba82c3692e345 : FAILURE in 11m 53s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/edb41a44360f4cfb9667f3b825504b40 : FAILURE in 18m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bf989fb44744c77bd9d39314983213e : SUCCESS in 11m 50s","accounts_in_message":[],"_revision_number":1},{"id":"58d6e9876118bf298f48acdc3e29b9b792578094","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2019-11-08 02:20:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d689db9de36c16c0cedf9e0711c6289e97e74180","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-11-12 14:10:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"457388068f794a474ac131e13dd3b37f10dfbceb","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2019-11-13 02:12:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86c9b6007ee28d564c46d29f6d0e5d9451fec4ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 02:13:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ac283485ee32f3bb796024e0986550fb54f2a6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 02:37:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e14503466c24809a80405bde34fb52c : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51c5a439be864996a1b95481d49d550a : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a960b63d030429ebb66f0ce38769f98 : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7b042ea55554fcbb6cc4f4076eab9bf : SUCCESS in 16m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e5e1d0f6e24f9aa1dbcafbf455ff35 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cae2f660392bde7d9e1404874eb087524d1f7b60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 02:37:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"703489e82ca8d767904155b0ebb6aac6d8b7c9fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 02:38:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf5e8bc04c34abca9db063714316003 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5277ee401427bfc38ade804ca39d7d0d854774d","revisions":{"b5277ee401427bfc38ade804ca39d7d0d854774d":{"kind":"REWORK","_number":1,"created":"2019-11-07 10:12:27.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/30/693330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/30/693330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/30/693330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/30/693330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/30/693330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/30/693330/1"}}},"commit":{"parents":[{"commit":"0b39a9904f1ce1d760caeb007ea027583fbaa227","subject":"Merge \"Bump the openstackdocstheme extension to 1.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/0b39a9904f1ce1d760caeb007ea027583fbaa227"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-11-07 09:16:39.000000000","tz":480},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-11-07 09:16:59.000000000","tz":480},"subject":"Remove policy.json as it doesn\u0027t seem to be used","message":"Remove policy.json as it doesn\u0027t seem to be used\n\nChange-Id: I6307c90b70cbd9c4bac1d26f36a5becad538b293\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b5277ee401427bfc38ade804ca39d7d0d854774d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b5277ee401427bfc38ade804ca39d7d0d854774d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
