)]}'
{"id":"openstack%2Fnetworking-midonet~697648","triplet_id":"openstack%2Fnetworking-midonet~master~I71c6a20f32650bc58953d237da9aa80bef4d1f60","project":"openstack/networking-midonet","branch":"master","topic":"py","hashtags":[],"change_id":"I71c6a20f32650bc58953d237da9aa80bef4d1f60","subject":"Specify USE_PYTHON3\u003dFalse explicitly","status":"MERGED","created":"2019-12-06 11:39:34.000000000","updated":"2019-12-09 08:49:47.000000000","submitted":"2019-12-09 08:48:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697648-1575881319254-8628b750","meta_rev_id":"0569366649a9b5dfecd968460248ff0e71923ad8","_number":697648,"virtual_id_number":697648,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-08 13:08:13.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 08:48:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2019-12-09 08:23:15.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":2,"date":"2019-12-09 03:39:48.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":1,"date":"2019-12-09 08:23:08.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 03:39:48.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2019-12-09 08:23:08.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-12-09 08:48:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1513ab7df27d6dec378c57725090671a198f7299","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-06 11:39:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d3c4ff9dcd4c66571fc4d7b43b3d37e08aed38f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-06 11:39:45.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2d522e42627718b55b5f5f1d13a783dc2f239a15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 13:44:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3de58cab32ef48999bb3e4f48a967a2b : SUCCESS in 8m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab8994e7870940fcbb13ba133584e00f : SUCCESS in 12m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb417e5748634786a2a362b51769438f : SUCCESS in 14m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e0740e2e07444beb7726a88e42c8571 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0593967e5bc04f5f84c525cbef6ce8cc : SUCCESS in 6m 15s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/56d0a91a7e3740bc9202721dcfed4c2d : FAILURE in 2h 03m 29s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/b7082900696c4404b6a9a06f92b42ede : FAILURE in 1h 56m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/886d5ff3f8b6463caef86cf831a861e3 : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":1},{"id":"152ab0e02e5c6e3be875bbf93d4aa492a2c98814","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-07 03:52:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec72ca8ef38583199fe3998f3140f77710eb5ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 04:36:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9033ee502f2d4ceca0146b033e7a2850 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e9f73c747e246ef8b3c2d32ab76c51a : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63d0b11b614d49489f26774ad199be0b : SUCCESS in 10m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39e5d42156ca4f24aae54de41f44aeb1 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08204f97abf47e296dcd42c56fd6ed6 : SUCCESS in 5m 21s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/773954bc2ec2417d9ffb0614297ac9e2 : FAILURE in 21m 55s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/75aa1e4e74ee4c208b6679fb33a56239 : FAILURE in 21m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be9318b0c6bf4accaa98aa1b0cccc473 : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":2},{"id":"27936169426adf63e62cbf65aaeb53850ae8b70c","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-12-08 10:52:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e6c69152383d3839a2abdb021161687320599fdd","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-08 13:08:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14453e3da76ae78e6a9fb77b07d153b34d3b4cdd","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2019-12-09 03:39:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7fe00ed451303fa355bdafc787e41909f11b318f","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-12-09 08:23:08.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3a8905eaa47e039cdacaea973a1d17fb81517e43","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-12-09 08:23:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d74fa275352b3279bd2eb320da3faeec1cc21fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:23:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"23c5047e588f532266c5b34d5a07f280c91e2d5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:48:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e00005bf4cc67857640e84e0bf224f35d8e257a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:48:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26430e0c77d0499582de53629fe9bfb4 : SUCCESS in 8m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a7a71cddd884d3f92dee605ff463820 : SUCCESS in 16m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd7bbb0dc30744cca229e5dc57e836d6 : SUCCESS in 16m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8af55ddb5fc84d15bab3126110f3c327 : SUCCESS in 13m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4feafbc3fb43bd948f328185774e83 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1f64a8e00179109913d997a1b0ba0d8833b44199","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:49:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15deb055eb3c431a97b07d4f0d498deb : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52b7d13449383b83326537824f25f323e38075b2","revisions":{"19680254bf0b5902d3a83d4c1794826058c3d01b":{"kind":"REWORK","_number":1,"created":"2019-12-06 11:39:34.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/48/697648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/48/697648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/48/697648/1"}}},"commit":{"parents":[{"commit":"1096e13d961665c6b7ba81c762f6d0208a2029fa","subject":"doc: Add policy reference","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1096e13d961665c6b7ba81c762f6d0208a2029fa"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-06 11:38:55.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-06 11:38:55.000000000","tz":540},"subject":"test USE_PYTHON3\u003dFalse","message":"test USE_PYTHON3\u003dFalse\n\nChange-Id: I71c6a20f32650bc58953d237da9aa80bef4d1f60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/19680254bf0b5902d3a83d4c1794826058c3d01b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/19680254bf0b5902d3a83d4c1794826058c3d01b"}]},"branch":"refs/heads/master"},"52b7d13449383b83326537824f25f323e38075b2":{"kind":"REWORK","_number":2,"created":"2019-12-07 03:52:51.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/48/697648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/48/697648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/48/697648/2"}}},"commit":{"parents":[{"commit":"1096e13d961665c6b7ba81c762f6d0208a2029fa","subject":"doc: Add policy reference","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1096e13d961665c6b7ba81c762f6d0208a2029fa"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-06 11:38:55.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 03:43:15.000000000","tz":540},"subject":"Specify USE_PYTHON3\u003dFalse explicitly","message":"Specify USE_PYTHON3\u003dFalse explicitly\n\nCurrently our devstack jobs are with python2 because\nthey all are on centos-7, where python3 is not ready.\nThey recently started failing with \"python3.5 not found\" for\nsome reasons. (I haven\u0027t investigated why.)\nSpecify USE_PYTHON3\u003dFalse as a workaround.\n\nCloses-Bug: #1855512\nChange-Id: I71c6a20f32650bc58953d237da9aa80bef4d1f60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/52b7d13449383b83326537824f25f323e38075b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/52b7d13449383b83326537824f25f323e38075b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
