)]}'
{"id":"openstack%2Fnetworking-midonet~697746","triplet_id":"openstack%2Fnetworking-midonet~master~I2bd88fdea22884296ce418148cbb6a867c93233c","project":"openstack/networking-midonet","branch":"master","topic":"py","hashtags":[],"change_id":"I2bd88fdea22884296ce418148cbb6a867c93233c","subject":"Skip test_resize_server_revert_deleted_flavor for now","status":"MERGED","created":"2019-12-07 03:52:51.000000000","updated":"2019-12-09 08:50:38.000000000","submitted":"2019-12-09 08:49:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697746-1575881366873-0f63802f","meta_rev_id":"3a0b4e316a4b98c2efa9cb35448cd4c66a03bb71","_number":697746,"virtual_id_number":697746,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 08:49:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"recommended":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"all":[{"value":1,"date":"2019-12-08 15:08:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":2,"date":"2019-12-08 10:53:06.000000000","_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-09 03:40:31.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"all":[{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-09 03:40:31.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-08 10:53:06.000000000","updated_by":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"reviewer":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"state":"REVIEWER"},{"updated":"2019-12-08 15:08:02.000000000","updated_by":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"reviewer":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"state":"REVIEWER"},{"updated":"2019-12-09 03:40:31.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2019-12-09 08:49:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af8a34a52d5c89768d6f2854ed3455e722816942","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-07 03:52:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"691b542af9e0322a11af4c00bc1cc4974ba20975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 04:20:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f38bc92cf19d4ab9ab57f670b34cbb3c : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e001872c6c4948e9900f88a9291cbd81 : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e0faaa044f84186b85b8e9ec31321c7 : SUCCESS in 12m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc57f946c1eb496aae65d6fd3f286df8 : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb2b50fd6be4c1aa8ea04ddf6ebb2a9 : SUCCESS in 5m 12s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/f12ad203d31a47568c5c486e6b7d7611 : FAILURE in 23m 41s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/74edbdf496c046cf8830223346299cb3 : FAILURE in 22m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3e8bc607d68466b8c701d517dc2870a : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cb8c40ed231225d45fb028b5cc5a5b401ecfa11f","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-07 10:11:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc6330cefe57442f4fc6ba5e33096dce0b8a0f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 10:43:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b76a27f7faf4ffb82361ec66416067c : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2687f5006994a34a73aa998244184cf : SUCCESS in 16m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b158a450115f484eae455f5a9349f75f : SUCCESS in 15m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/010b16d6901f423b91baf1ce6a945d50 : SUCCESS in 15m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de76d94bd2f48ebbc4d4aedf3c1343e : SUCCESS in 5m 29s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/5c7f8e70e8ab454ea5bcda03d23ad7e9 : FAILURE in 26m 43s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/596ce3eaa4164691a89d713071d4f1de : FAILURE in 25m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7268f580de0c492f9e49c190e9d74c27 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5beba0ac63c3a02af4f1e7e7ae57dcb6a4612065","author":{"_account_id":19307,"name":"Antonio Ojea","email":"antonio.ojea.garcia@gmail.com","username":"itsuugo"},"date":"2019-12-08 10:53:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62079c94c276a4a1e740feab10dc05cd4be2d267","author":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"date":"2019-12-08 15:08:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7fc9d9c44b9b65584f46421501a72d84880f1dbe","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2019-12-09 03:40:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"10af9a2a184f428153bde6438f3b1ee4cfa8204d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:23:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"46498b5d141d07c466f6e120716705abe5a1388f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:49:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"52c2c118aad0c8c0fe73270054ba28abd4069f06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:49:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed228f1f141841de9bf640e4cff83b88 : SUCCESS in 7m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c4b875436f64655b6806496748bb285 : SUCCESS in 16m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f375b6951e6f456abed0c65f0e1232d0 : SUCCESS in 13m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9676b79df46347d79def25f58887514e : SUCCESS in 15m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/283b07944728490bb7004633d1ac86f0 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":2},{"id":"1653da935c714482053a68a7e46c9300a2bdd1a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 08:50:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f68bb100b8524200afc0403002347554 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c3e8b36a7b0e8c1ca81bc2c24f1bae27395dbcd6","revisions":{"bb31fe4349de2f83b2188ceb30e36206680b8cd9":{"kind":"REWORK","_number":1,"created":"2019-12-07 03:52:51.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/46/697746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/697746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/697746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/697746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/697746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/697746/1"}}},"commit":{"parents":[{"commit":"52b7d13449383b83326537824f25f323e38075b2","subject":"Specify USE_PYTHON3\u003dFalse explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/52b7d13449383b83326537824f25f323e38075b2"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 03:50:05.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 03:50:51.000000000","tz":540},"subject":"Skip test_resize_server_revert_deleted_flavor for now","message":"Skip test_resize_server_revert_deleted_flavor for now\n\nIt has been broken since a nova change. [1]\nNova folks don\u0027t want to revert it.\nI don\u0027t have time to fix it for the meantime either.\nSkip the test case for now to avoid hiding other breakage.\n\n[1] https://review.opendev.org/#/c/667177/\n\nRelated-Bug: #1839169\nChange-Id: I2bd88fdea22884296ce418148cbb6a867c93233c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bb31fe4349de2f83b2188ceb30e36206680b8cd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bb31fe4349de2f83b2188ceb30e36206680b8cd9"}]},"branch":"refs/heads/master"},"c3e8b36a7b0e8c1ca81bc2c24f1bae27395dbcd6":{"kind":"REWORK","_number":2,"created":"2019-12-07 10:11:16.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/46/697746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/697746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/697746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/697746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/697746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/697746/2"}}},"commit":{"parents":[{"commit":"52b7d13449383b83326537824f25f323e38075b2","subject":"Specify USE_PYTHON3\u003dFalse explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/52b7d13449383b83326537824f25f323e38075b2"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 03:50:05.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 10:10:57.000000000","tz":540},"subject":"Skip test_resize_server_revert_deleted_flavor for now","message":"Skip test_resize_server_revert_deleted_flavor for now\n\nIt has been broken since a nova change. [1]\nNova folks don\u0027t want to revert it.\nI don\u0027t have time to fix it for the meantime either.\nSkip the test case for now to avoid hiding other breakage.\n\n[1] https://review.opendev.org/#/c/667177/\n\nRelated-Bug: #1839169\nChange-Id: I2bd88fdea22884296ce418148cbb6a867c93233c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c3e8b36a7b0e8c1ca81bc2c24f1bae27395dbcd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c3e8b36a7b0e8c1ca81bc2c24f1bae27395dbcd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
