)]}'
{"id":"openstack%2Fnetworking-midonet~697748","triplet_id":"openstack%2Fnetworking-midonet~master~I2c8489ff1068e6471ae13272f555ce52237c0c6e","project":"openstack/networking-midonet","branch":"master","topic":"py","hashtags":[],"change_id":"I2c8489ff1068e6471ae13272f555ce52237c0c6e","subject":"DO NOT MERGE: test a revert","status":"ABANDONED","created":"2019-12-07 05:16:25.000000000","updated":"2022-11-22 14:52:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d6994475e2f1c6fea80aff263eefc318aaf13bf","_number":697748,"virtual_id_number":697748,"owner":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-09 07:23:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 07:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6595d5ef8b7356ebd2687ec3a0447a3c25367cb2","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-07 05:16:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24beb873e09b6229700bf7145c66afb84fda9d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 07:06:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd9df2d655146f182ecb88b006cb77c : SUCCESS in 6m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2da30d062de14fb2b6c4110ef6ff39da : SUCCESS in 10m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0793de3977284e44ace6182bd8507f8f : SUCCESS in 11m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d282b7ab2ca4285a1e86325da377875 : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678fe989fd5f4d4aa42c78e4b5fa400b : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a769865930646a4a88a9d3daf4d86d5 : SUCCESS in 5m 22s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/cab6525a238f4bbab0cd9d6fb565a441 : FAILURE in 1h 43m 51s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/df0888b6e81b4ee49028818a98a2b1c2 : FAILURE in 1h 34m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ace989e3f1c40338c20552f3f8e03f2 : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e6bdb1ea9542a7018e2e224e4dfc86d4d4517f1b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-07 10:11:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5713716e2309f921b52250d32f5046c49c526562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 10:24:26.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"82508e5e243ae806e5fe9388aa2315bf0bbc51dd","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-07 11:22:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c729c3e5e2d4770f1527387a7a4074ba07bd6e77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 13:14:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a7b8f7a60bf4585b428228d5b009b65 : SUCCESS in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff36fafad0de497a92e2562477df479d : SUCCESS in 14m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9acf5f1915d4de690361162791cd4c2 : SUCCESS in 14m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d652d0a162eb4646a8bba73afda1b2e5 : SUCCESS in 14m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa5a54da334442a90ee6d18d05e7c89 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d41ebd73a29b433d884e518c956b5179 : SUCCESS in 4m 40s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/d071d78ba28d41528470dee118e280a8 : SUCCESS in 1h 41m 57s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/1afd8dffeb2e4b19946bdc21f8b552bc : FAILURE in 1h 26m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01d66c92e41c4b6493fcc3dbd34fe661 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e8f4f966af1915d013a9f8321c43d57e89faec4a","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2019-12-09 05:25:52.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5aeaf6496ef399a0208fa7ba5ad4e4d54a203ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 07:23:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/081ea24051c4407993fab858871e712f : SUCCESS in 6m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af3f4ffa0be6417d9734791b4c61f282 : SUCCESS in 11m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/095fe4d6bedd496fbbbb316eeb49f30b : SUCCESS in 10m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b248e2a1118d4ff7ad01fb2fa8d0b828 : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484393743d164ee69bfb25198cfada62 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c753ca903477429c8d888290403afc22 : SUCCESS in 4m 38s\n- networking-midonet-tempest-aio-ml2-centos-7 https://zuul.opendev.org/t/openstack/build/179fd59a8dbb465c93014bab38390243 : SUCCESS in 1h 37m 14s (non-voting)\n- networking-midonet-tempest-aio-ml2-full-centos-7 https://zuul.opendev.org/t/openstack/build/b91d8f752efe432dadbf189c64ab0cd1 : SUCCESS in 1h 52m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afd23c003c6f49ca864a32d84da01fb3 : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5d6994475e2f1c6fea80aff263eefc318aaf13bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 14:52:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"148ab72bf77cd4a118716a3664b9c8ebf636a036","revisions":{"eef738d8728a215c99010fface8b6a73c0bae6f1":{"kind":"REWORK","_number":1,"created":"2019-12-07 05:16:25.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/48/697748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/48/697748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/1"}}},"commit":{"parents":[{"commit":"bb31fe4349de2f83b2188ceb30e36206680b8cd9","subject":"Skip test_resize_server_revert_deleted_flavor for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bb31fe4349de2f83b2188ceb30e36206680b8cd9"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 05:15:43.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 05:15:43.000000000","tz":540},"subject":"DO NOT MERGE: test a revert","message":"DO NOT MERGE: test a revert\n\nDepends-On: https://review.opendev.org/#/c/697747/\nChange-Id: I2c8489ff1068e6471ae13272f555ce52237c0c6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/eef738d8728a215c99010fface8b6a73c0bae6f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/eef738d8728a215c99010fface8b6a73c0bae6f1"}]},"branch":"refs/heads/master"},"7699e6e516914ae1ded7ffdf5587d37b78967c06":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-07 10:11:16.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/48/697748/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/48/697748/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/2"}}},"commit":{"parents":[{"commit":"c3e8b36a7b0e8c1ca81bc2c24f1bae27395dbcd6","subject":"Skip test_resize_server_revert_deleted_flavor for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c3e8b36a7b0e8c1ca81bc2c24f1bae27395dbcd6"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 05:15:43.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 10:10:57.000000000","tz":540},"subject":"DO NOT MERGE: test a revert","message":"DO NOT MERGE: test a revert\n\nDepends-On: https://review.opendev.org/#/c/697747/\nChange-Id: I2c8489ff1068e6471ae13272f555ce52237c0c6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7699e6e516914ae1ded7ffdf5587d37b78967c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7699e6e516914ae1ded7ffdf5587d37b78967c06"}]},"branch":"refs/heads/master"},"148ab72bf77cd4a118716a3664b9c8ebf636a036":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-09 05:25:52.000000000","uploader":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"ref":"refs/changes/48/697748/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/48/697748/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/48/697748/3"}}},"commit":{"parents":[{"commit":"c3e8b36a7b0e8c1ca81bc2c24f1bae27395dbcd6","subject":"Skip test_resize_server_revert_deleted_flavor for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/c3e8b36a7b0e8c1ca81bc2c24f1bae27395dbcd6"}]}],"author":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-07 05:15:43.000000000","tz":540},"committer":{"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","date":"2019-12-09 05:25:48.000000000","tz":0},"subject":"DO NOT MERGE: test a revert","message":"DO NOT MERGE: test a revert\n\nDepends-On: https://review.opendev.org/#/c/697948/\n//Depends-On: https://review.opendev.org/#/c/697747/\nChange-Id: I2c8489ff1068e6471ae13272f555ce52237c0c6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/148ab72bf77cd4a118716a3664b9c8ebf636a036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/148ab72bf77cd4a118716a3664b9c8ebf636a036"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
