)]}'
{"id":"openstack%2Fnetworking-midonet~720788","triplet_id":"openstack%2Fnetworking-midonet~master~Ibab90934e138f10e36e055c038e404f89d01f41e","project":"openstack/networking-midonet","branch":"master","topic":"bug/1871568","hashtags":[],"change_id":"Ibab90934e138f10e36e055c038e404f89d01f41e","subject":"Make gate great again","status":"MERGED","created":"2020-04-17 17:41:10.000000000","updated":"2020-05-08 14:15:20.000000000","submitted":"2020-05-08 14:13:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720788-1588947235686-7b8e95d9","meta_rev_id":"8e1b9fa1b5704774bb8081df7510dc2e6a6b4333","_number":720788,"virtual_id_number":720788,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 14:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":2,"date":"2020-05-08 13:52:47.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-05-08 13:08:29.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"all":[{"value":1,"date":"2020-05-08 13:52:47.000000000","_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 11:12:34.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2020-05-06 12:32:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-05-08 13:08:29.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-05-08 13:52:47.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2020-05-08 14:13:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54023e65b2a20382aaee4fe34659b1e4455c63c9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-17 17:41:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0e15a7c25ab74d6f9a18facf6756dd9df701ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 18:09:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda7ac56f9744821b9a24a200ba10ba0 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6706032607e43e390ce2041287e31ab : FAILURE in 12m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59a4d65be204424ca987e0a168f889ed : FAILURE in 13m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66ea037f566d476d9648fc4fd153bd02 : FAILURE in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/176a6dc3a7874c6985e27ac12c725172 : SUCCESS in 12m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd35ea9caa5e4d24afb9b47728a4e1b4 : FAILURE in 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9861180429f97da6947b57c0e85e58f901e6346e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-18 09:32:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"60b3b6a1ef186a659b6c7f39405be1b623d4eb7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 09:50:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1cbf2d7ccd8413c9e710117e87ae5b1 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9acfa09585674c5282e3c7387d21776c : FAILURE in 11m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a20d6c544a24ad0a0e3cf0cc4eb6d18 : FAILURE in 12m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/378f9e7a555a417d89147aac06910a5f : FAILURE in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128536c62a0f47fc9b615d5b750e0cc3 : SUCCESS in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46b1a2e4efae453aab40250731289f37 : FAILURE in 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c22f92c52b96a9b920082ccc3a0be905036bb992","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2020-05-03 11:22:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"741444b752f4197bf9f5b60cfa0610b68226e319","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2020-05-06 11:12:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fcd93f7b87d622ad1c9a0dc9fdab428f5eb79d37","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-06 12:32:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c55de23f25e55817f98179634b3248e38a3eb331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:52:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4815ba1f7723488682fc2b3f9ed7b823 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4a410f1b4dc46b7bc129a604d8ee115 : FAILURE in 13m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c058238eaa14554aa9b7681493d4c7a : FAILURE in 17m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1013353317c45678fa3873b7951a377 : FAILURE in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9755c582852344bb950b41255a4cc6d9 : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffe56df0296543a7adb7ed5df512ec34 : FAILURE in 14m 06s","accounts_in_message":[],"_revision_number":3},{"id":"fc333aca3b10e69aa44073a8e84514820a7cf2f6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-06 20:13:18.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"7b52126177bddfab43ddc003937617d6e7e7cf03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 20:40:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd29039a8c24f8a8bf757d3eae7089b : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f651790000114e2aaac6ab32718ded1c : FAILURE in 17m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b02fbbfa59894607ac8720e38fc18974 : FAILURE in 16m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eca11d73217a4f3caf8c4eb52ee4d5e5 : FAILURE in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213c1130955a4afaa1a565da5a665fbb : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ed3344908df4c0993407e02fd3e8992 : FAILURE in 14m 11s","accounts_in_message":[],"_revision_number":4},{"id":"12f37a7b442f12e5f3b95ab33e2e4401cab36e62","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-08 09:40:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"48fd8b696d400ff9f18ac926d4c8c886bc81bda8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 09:58:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf59f88002546c992c1278e7694c5d0 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7c8ad4040384bcba87c15dffcadb389 : SUCCESS in 13m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4d819350093412cbfe46223e9069c7f : SUCCESS in 16m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7a179f23b744ce684a6d0d7ed07da81 : FAILURE in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2269547c83214190a0e72b8aaee708d2 : SUCCESS in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8e66f64ab92475590309fb8e52c2ba1 : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":5},{"id":"fe156aeda239f6e3edd58208f7aff798d285bb99","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-05-08 12:41:44.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 6e79ba7478254e9b327526b1ef2f1214df4c0151","accounts_in_message":[],"_revision_number":5},{"id":"e697a588a576222681b6f27d3b9cca62f058ba49","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-08 13:08:29.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bae4a0626661ca0b39dd128da0045275b0e1d05b","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2020-05-08 13:52:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nthank you","accounts_in_message":[],"_revision_number":5},{"id":"5b5a8104c4f4b3cf104a8f9cfc5f1edb7fd9f726","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 13:52:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"36d081179ad4efda69152934fed7671c04f1dc2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 14:13:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5f3aae13b34f592b93bbe478c8a14ec571571991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 14:13:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81abe1fdada4413484e4c4b07999abc5 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cde77cfdd59844d4a711a71acabbe9fc : SUCCESS in 10m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7433f74d2016471eba9b3707644ce46b : SUCCESS in 14m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd280adc44cf442b89004b4d43b0959b : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":5},{"id":"023fa9d64c1a0d0c4bb4dbe92d210e2177bf4e58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 14:15:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e219be300ff411ebe87c8056c11cc0e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"86463fe85f106e95688bb505e7f14569dbbcea1b","revisions":{"521f0a893fca088b6924381a91f95ce37df3e8c0":{"kind":"REWORK","_number":1,"created":"2020-04-17 17:41:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/720788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/88/720788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/1"}}},"commit":{"parents":[{"commit":"42f1cfad564948d0b8dd0514c776e57197f8bc58","subject":"Merge \"Update for removal of agent tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/42f1cfad564948d0b8dd0514c776e57197f8bc58"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 17:38:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 17:41:02.000000000","tz":0},"subject":"Fix MidonetInterfaceDriver.plug_new signature","message":"Fix MidonetInterfaceDriver.plug_new signature\n\nSince [1], the LinuxInterfaceDriver.plug_new signature includes a new\nkwarg.\n\n[1]https://review.opendev.org/#/c/707406\n\nChange-Id: Ibab90934e138f10e36e055c038e404f89d01f41e\nCloses-Bug: #TBD\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/521f0a893fca088b6924381a91f95ce37df3e8c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/521f0a893fca088b6924381a91f95ce37df3e8c0"}]},"branch":"refs/heads/master"},"e6d2ae300be03a8937f03c6cbc2a55da3ba5c7b7":{"kind":"REWORK","_number":2,"created":"2020-04-18 09:32:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/720788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/88/720788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/2"}}},"commit":{"parents":[{"commit":"42f1cfad564948d0b8dd0514c776e57197f8bc58","subject":"Merge \"Update for removal of agent tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/42f1cfad564948d0b8dd0514c776e57197f8bc58"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 17:38:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-18 09:29:38.000000000","tz":0},"subject":"Fix MidonetInterfaceDriver.plug_new signature","message":"Fix MidonetInterfaceDriver.plug_new signature\n\nSince [1], the \"LinuxInterfaceDriver.plug_new\" signature includes a\nnew kwarg, \"link_up\". Because \"MidonetInterfaceDriver\" inherits from\n\"LinuxInterfaceDriver\" and overrides \"plug_new\" function, the local\nimplementation should include this new parameter.\n\n[1]https://review.opendev.org/#/c/707406\n\nChange-Id: Ibab90934e138f10e36e055c038e404f89d01f41e\nCloses-Bug: #1871568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e6d2ae300be03a8937f03c6cbc2a55da3ba5c7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e6d2ae300be03a8937f03c6cbc2a55da3ba5c7b7"}]},"branch":"refs/heads/master"},"b280adcfbca0ce9c95397ccd9dce5444a915e490":{"kind":"REWORK","_number":3,"created":"2020-05-06 12:32:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/88/720788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/88/720788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/3"}}},"commit":{"parents":[{"commit":"42f1cfad564948d0b8dd0514c776e57197f8bc58","subject":"Merge \"Update for removal of agent tempest tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/42f1cfad564948d0b8dd0514c776e57197f8bc58"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 17:38:39.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-06 12:32:00.000000000","tz":120},"subject":"Make gate great again","message":"Make gate great again\n\nThis patch combines two fixes into one patch to fix UT failures:\n\n* Fix MidonetInterfaceDriver.plug_new signature\n\n  Since [1], the \"LinuxInterfaceDriver.plug_new\" signature includes a\n  new kwarg, \"link_up\". Because \"MidonetInterfaceDriver\" inherits from\n  \"LinuxInterfaceDriver\" and overrides \"plug_new\" function, the local\n  implementation should include this new parameter.\n\n* Fix Mysql db migration schema\n  Field \"midonet_tasks.data\" was defined as sa.Text() in db model but in\n  migration script for mysql engine it is defined as Text type with\n  lenght long enough to create LONGTEXT field in the DB. That caused\n  problem with types of field in the db model and migration script.\n  So this patch fixes that by using sa.Text() with LONGTEXT variant for\n  mysql engine in db model.\n\n[1]https://review.opendev.org/#/c/707406\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: Ibab90934e138f10e36e055c038e404f89d01f41e\nCloses-Bug: #1871568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b280adcfbca0ce9c95397ccd9dce5444a915e490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/b280adcfbca0ce9c95397ccd9dce5444a915e490"}]},"branch":"refs/heads/master"},"74e8b8a14a1f9c78417b664d53726b50103e0a71":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-06 20:13:18.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/88/720788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/88/720788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/4"}}},"commit":{"parents":[{"commit":"98c9eaeb52990cf484500b5f0985d35e173800b5","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/98c9eaeb52990cf484500b5f0985d35e173800b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 17:38:39.000000000","tz":0},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-05-06 12:48:24.000000000","tz":120},"subject":"Make gate great again","message":"Make gate great again\n\nThis patch combines two fixes into one patch to fix UT failures:\n\n* Fix MidonetInterfaceDriver.plug_new signature\n\n  Since [1], the \"LinuxInterfaceDriver.plug_new\" signature includes a\n  new kwarg, \"link_up\". Because \"MidonetInterfaceDriver\" inherits from\n  \"LinuxInterfaceDriver\" and overrides \"plug_new\" function, the local\n  implementation should include this new parameter.\n\n* Fix Mysql db migration schema\n  Field \"midonet_tasks.data\" was defined as sa.Text() in db model but in\n  migration script for mysql engine it is defined as Text type with\n  lenght long enough to create LONGTEXT field in the DB. That caused\n  problem with types of field in the db model and migration script.\n  So this patch fixes that by using sa.Text() with LONGTEXT variant for\n  mysql engine in db model.\n\n[1]https://review.opendev.org/#/c/707406\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: Ibab90934e138f10e36e055c038e404f89d01f41e\nCloses-Bug: #1871568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/74e8b8a14a1f9c78417b664d53726b50103e0a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/74e8b8a14a1f9c78417b664d53726b50103e0a71"}]},"branch":"refs/heads/master"},"86463fe85f106e95688bb505e7f14569dbbcea1b":{"kind":"REWORK","_number":5,"created":"2020-05-08 09:40:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/720788/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/88/720788/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/88/720788/5"}}},"commit":{"parents":[{"commit":"98c9eaeb52990cf484500b5f0985d35e173800b5","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/98c9eaeb52990cf484500b5f0985d35e173800b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 17:38:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-08 09:39:56.000000000","tz":0},"subject":"Make gate great again","message":"Make gate great again\n\nThis patch combines two fixes into one patch to fix UT failures:\n\n* Fix MidonetInterfaceDriver.plug_new signature\n\n  Since [1], the \"LinuxInterfaceDriver.plug_new\" signature includes a\n  new kwarg, \"link_up\". Because \"MidonetInterfaceDriver\" inherits from\n  \"LinuxInterfaceDriver\" and overrides \"plug_new\" function, the local\n  implementation should include this new parameter.\n\n* Fix Mysql db migration schema\n  Field \"midonet_tasks.data\" was defined as sa.Text() in db model but in\n  migration script for mysql engine it is defined as Text type with\n  lenght long enough to create LONGTEXT field in the DB. That caused\n  problem with types of field in the db model and migration script.\n  So this patch fixes that by using sa.Text() with LONGTEXT variant for\n  mysql engine in db model.\n\n[1]https://review.opendev.org/#/c/707406\n\nCo-Authored-By: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n\nChange-Id: Ibab90934e138f10e36e055c038e404f89d01f41e\nCloses-Bug: #1871568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/86463fe85f106e95688bb505e7f14569dbbcea1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/86463fe85f106e95688bb505e7f14569dbbcea1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
