)]}'
{"id":"openstack%2Fnetworking-midonet~738046","triplet_id":"openstack%2Fnetworking-midonet~master~I9231b2b362a1f2316307908b7e9ad57a709700f6","project":"openstack/networking-midonet","branch":"master","topic":"complete-retirement-networking-l2gw","hashtags":[],"change_id":"I9231b2b362a1f2316307908b7e9ad57a709700f6","subject":"Use the networking-l2gw project new location","status":"MERGED","created":"2020-06-25 14:43:06.000000000","updated":"2020-09-12 04:38:54.000000000","submitted":"2020-09-09 21:52:55.000000000","submitter":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738046-1599688375748-87da3b05","meta_rev_id":"e26404b135e2069d9d8a762096309edc53da6995","_number":738046,"virtual_id_number":738046,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-09-09 21:52:42.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-09 16:35:01.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-09-09 21:52:42.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-09-09 21:52:42.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 07:55:43.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-09-09 16:01:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-09 16:35:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-09 21:52:42.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"20312da57e4e31276379225d568650e83e5c889d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 14:43:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb3649cf35adddef4e97528b162b767a8fed2cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:43:41.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"e7269e259d7dda11cc772bb902af0788f0d9520f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 15:17:27.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease remove the depends-on, let\u0027s merge these *first*. Otherwise, we have a broken config until this merges.","accounts_in_message":[],"_revision_number":1},{"id":"a16b878399764779720269557b0d6df7c318c7b4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 15:45:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"650bc06a541acd61b7aeb0eca5e02eeea116ce07","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 15:45:30.000000000","message":"Patch Set 2: Code-Review+1\n\nRemoved depends-on","accounts_in_message":[],"_revision_number":2},{"id":"3bc82f894e8c66a0822fb3504ed16c60352bd6da","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 15:47:00.000000000","message":"Patch Set 2:\n\nImportant: This needs backporting to all branches that use networking-l2gw","accounts_in_message":[],"_revision_number":2},{"id":"76eae590393f57dd0ecbf98112684394d3527b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 16:08:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cef366e1b4245a69d50f2669b8fa1c4 : SUCCESS in 10m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d1a14530fa94a3eb5ae2e32720e4ff0 : SUCCESS in 15m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d0aae10d3fa4a7d97a4eb4727d64523 : SUCCESS in 14m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3348f38a459494689b36fa828552190 : SUCCESS in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec302df42b1340d8a370fd87f488e585 : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":2},{"id":"79dc1928964b8602756b807fad1ac6ecd2cfa94f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 17:45:37.000000000","message":"Patch Set 2: Workflow-1\n\nwe have code dependency also, discussing with neutron team.","accounts_in_message":[],"_revision_number":2},{"id":"375c9009fe8fdd7249a308f270ae67ca35477796","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-27 22:22:16.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7b3d058cdfb9b2bccae502698dc5f90204c68882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 22:36:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83906350dd54839ba05e4fff21f2d7d : FAILURE in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb848d81445949da9e4b84f22c123879 : FAILURE in 12m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cb99f9392b1498283f18de3a7fd417e : FAILURE in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95a3297aa7e43c7917352063cf79cbf : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16a50a75e7b346f5becb60d9d63749a8 : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":3},{"id":"92c81b3e05fbac953b9a6b1ab5da02efa86a6c38","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-27 22:50:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2fc8e4de8c23a808a70fc7290157dd5274506ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 23:09:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8a0a3454b5a4a2da6ff5cc3683358e6 : FAILURE in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd29f09150df417fa2c8a54676a84ba7 : FAILURE in 13m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c3d75742eef4349a66e195928911b8d : FAILURE in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c6413373b44f5e88192b77d64c91d3 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2595addb5d444782b8e27653f21f369f : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/062d2f193b7d4df6bb58e4e16367966b : FAILURE in 13m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c296c62f1b005a67e4fa7781ea69e6edec1a5827","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-02 07:55:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3f6ed24024176eae4de077fd40e1384bc7299b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 08:16:34.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecaace3bf53043c48abe291993fb06fe : FAILURE in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45a669b1919c4348a733701b7654c952 : FAILURE in 13m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27053894e9484dfdb24d1fc17bc63362 : FAILURE in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7fe75133fe24de7bc129ad3087d75c6 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31939e0a678d4a32af0b177b6be70a63 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59c0f093a8c040b5ae840c2ae35573c1 : FAILURE in 13m 34s","accounts_in_message":[],"_revision_number":4},{"id":"98938fd2faa3204b36af39f9006f1cd17f43964c","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-03 05:54:09.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6cc9d11895b70d417f634adc237de4ef3370ab4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 06:09:54.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f447cac8aed340d0b8fd714135978281 : FAILURE in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb7ff155c8b74fa9a1c43c5ed740cad7 : FAILURE in 14m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dcc99cead804823b50eabfb1beebbfd : FAILURE in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49138d780b6c4df68fc75f3bb2020461 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05ca59ee40de47e8ab0d673a39b0dea1 : SUCCESS in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e553c8d20012463f9531a418b6278243 : FAILURE in 13m 06s","accounts_in_message":[],"_revision_number":5},{"id":"d352baa94975cc72aeb4b8c8c5ae058dde3c967e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-09 16:01:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"72d346fc66d1301f48f3d32b1079874401b10d3f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-09 16:02:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cfc3feffeff66c704c3b048b9ba464935d5354fa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-09 16:02:56.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"7fb8ccff1c62ef7284779a8dbff82f5818974a6c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-09 16:03:45.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"bc638237f33838614f69c251a1f390e5185a99c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 16:35:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf02122ca5c9435f9d20a88cb77c776a : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb2c3f0220c4759820df9b9cbb23c7e : FAILURE in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/befeb7fae29e45a7ac7341a436d83cbc : FAILURE in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/941970c40dff49c7ae919ac690e53300 : FAILURE in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523fb80c63844c60944fac5649c877f2 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c1123229ef1471f96863f7b49bd324e : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29ed4987182f429f9d14f8576f1cde1f : FAILURE in 10m 44s","accounts_in_message":[],"_revision_number":7},{"id":"81db3c9a272506b8114a582b0a7ce7bbb916866d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-09 21:52:42.000000000","message":"Patch Set 7: Verified+2 Code-Review+2 Workflow+1\n\nWill be submitting this manually in order to address zuul configuration errors resulting from this project rename.","accounts_in_message":[],"_revision_number":7},{"id":"cd89b0bccad763dd5e3607b3b81c932dd43d072f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-09 21:52:55.000000000","message":"Change has been successfully merged by Clark Boylan","accounts_in_message":[],"_revision_number":7},{"id":"47f3555aeba8757dd1fecdf6aafae966f86a8699","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 21:58:04.000000000","message":"Patch Set 7:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f939ea9d5794c2e922dabff54ebf024 : FAILURE in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65adbac37a064658b147cc22cdb550e2 : FAILURE in 53s","accounts_in_message":[],"_revision_number":7},{"id":"eaa3adf8b0ae0d23b17562e6536553cb0579aab6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-09 22:14:37.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 002f38c9b784f3625fe0a0ed01985b1fe4e7b2ef","accounts_in_message":[],"_revision_number":7},{"id":"a6aab29a6840fc8ca38bef83ee51962ddcfafa93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 23:09:26.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0ab8d4e8184345bd86d2d4624a19d9d0 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3b9fc7aa4c46f084b1e707721460fa : FAILURE in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0177851298224317bdcdc4fc0f653534 : FAILURE in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3afe090c83734fa688f5038c857ecce8 : RETRY_LIMIT in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17991d3bc5e04308a7d55f3cb32690b0 : SUCCESS in 16m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbf7e1d9564549339ae2293f0bed7dd8 : SUCCESS in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6db85120bdad4026934b3abea493c173 : RETRY_LIMIT in 4m 57s","accounts_in_message":[],"_revision_number":7},{"id":"49911533d1d8c15984f122e045169320660879d0","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2020-09-12 04:38:54.000000000","message":"Patch Set 7:\n\nlgtm except that this needs to be merged with other changes to pass the gate","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6428376d4ecdf911208f455231ecd2bf484e86b7","revisions":{"f31607af90c967298c91146679e4bf5aef705f35":{"kind":"REWORK","_number":1,"created":"2020-06-25 14:43:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/738046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/738046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/1"}}},"commit":{"parents":[{"commit":"4c97b0249c0bf43dd3cde14ebf39627a624f7313","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c97b0249c0bf43dd3cde14ebf39627a624f7313"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:42:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:43:05.000000000","tz":-300},"subject":"Remove the networking-l2gw from zuul jobs dependency","message":"Remove the networking-l2gw from zuul jobs dependency\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nfinishing the retirement process and no longer will be\nmaitained under openstack[1].\n\nDepends-On: https://review.opendev.org/#/c/738035/\nDepends-On: https://review.opendev.org/#/c/738038/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I9231b2b362a1f2316307908b7e9ad57a709700f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f31607af90c967298c91146679e4bf5aef705f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f31607af90c967298c91146679e4bf5aef705f35"}]},"branch":"refs/heads/master"},"e4a9c943404a283012aa51ebc1aa401bc30c5346":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-25 15:45:19.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/46/738046/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/738046/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/2"}}},"commit":{"parents":[{"commit":"4c97b0249c0bf43dd3cde14ebf39627a624f7313","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c97b0249c0bf43dd3cde14ebf39627a624f7313"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:42:03.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-06-25 15:45:07.000000000","tz":0},"subject":"Remove the networking-l2gw from zuul jobs dependency","message":"Remove the networking-l2gw from zuul jobs dependency\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nfinishing the retirement process and no longer will be\nmaitained under openstack[1].\n\nSee also:\nhttps://review.opendev.org/#/c/738035/\nhttps://review.opendev.org/#/c/738038/\n\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I9231b2b362a1f2316307908b7e9ad57a709700f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e4a9c943404a283012aa51ebc1aa401bc30c5346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/e4a9c943404a283012aa51ebc1aa401bc30c5346"}]},"branch":"refs/heads/master"},"8b073c82c880670788c34f8e7e6f55f1e712d946":{"kind":"REWORK","_number":3,"created":"2020-08-27 22:22:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/738046/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/738046/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/3"}}},"commit":{"parents":[{"commit":"4c97b0249c0bf43dd3cde14ebf39627a624f7313","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c97b0249c0bf43dd3cde14ebf39627a624f7313"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:42:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-27 22:22:13.000000000","tz":0},"subject":"Use the networking-l2gw project new location","message":"Use the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I9231b2b362a1f2316307908b7e9ad57a709700f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8b073c82c880670788c34f8e7e6f55f1e712d946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8b073c82c880670788c34f8e7e6f55f1e712d946"}]},"branch":"refs/heads/master"},"1e86e58b88c9a936768403f32e19d2db27c371db":{"kind":"REWORK","_number":4,"created":"2020-08-27 22:50:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/738046/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/738046/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/4"}}},"commit":{"parents":[{"commit":"4c97b0249c0bf43dd3cde14ebf39627a624f7313","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c97b0249c0bf43dd3cde14ebf39627a624f7313"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:42:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-27 22:50:20.000000000","tz":-300},"subject":"Use the networking-l2gw project new location","message":"Use the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I9231b2b362a1f2316307908b7e9ad57a709700f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1e86e58b88c9a936768403f32e19d2db27c371db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1e86e58b88c9a936768403f32e19d2db27c371db"}]},"branch":"refs/heads/master"},"637a29e334fa8acf7d9c39b241a780f15213336c":{"kind":"NO_CHANGE","_number":5,"created":"2020-09-03 05:54:09.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/46/738046/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/738046/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/5"}}},"commit":{"parents":[{"commit":"4c97b0249c0bf43dd3cde14ebf39627a624f7313","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c97b0249c0bf43dd3cde14ebf39627a624f7313"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:42:03.000000000","tz":-300},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-01 22:21:29.000000000","tz":600},"subject":"Use the networking-l2gw project new location","message":"Use the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I9231b2b362a1f2316307908b7e9ad57a709700f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/637a29e334fa8acf7d9c39b241a780f15213336c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/637a29e334fa8acf7d9c39b241a780f15213336c"}]},"branch":"refs/heads/master"},"8121cbc430ca547ffc4834232a02eab072ef9609":{"kind":"REWORK","_number":6,"created":"2020-09-09 16:02:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/738046/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/738046/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/6"}}},"commit":{"parents":[{"commit":"4c97b0249c0bf43dd3cde14ebf39627a624f7313","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c97b0249c0bf43dd3cde14ebf39627a624f7313"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:42:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-09 16:02:55.000000000","tz":0},"subject":"Use the networking-l2gw project new location","message":"Use the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I9231b2b362a1f2316307908b7e9ad57a709700f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8121cbc430ca547ffc4834232a02eab072ef9609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/8121cbc430ca547ffc4834232a02eab072ef9609"}]},"branch":"refs/heads/master"},"6428376d4ecdf911208f455231ecd2bf484e86b7":{"kind":"REWORK","_number":7,"created":"2020-09-09 16:03:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/46/738046/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/46/738046/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/46/738046/7"}}},"commit":{"parents":[{"commit":"4c97b0249c0bf43dd3cde14ebf39627a624f7313","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/4c97b0249c0bf43dd3cde14ebf39627a624f7313"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:42:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-09 16:03:44.000000000","tz":0},"subject":"Use the networking-l2gw project new location","message":"Use the networking-l2gw project new location\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nimported to new location now and supposed to be retired from\nopenstack/namespace[1].\n\n-https://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I9231b2b362a1f2316307908b7e9ad57a709700f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6428376d4ecdf911208f455231ecd2bf484e86b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6428376d4ecdf911208f455231ecd2bf484e86b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
