)]}'
{"id":"openstack%2Fnetworking-midonet~751963","triplet_id":"openstack%2Fnetworking-midonet~master~I2f2b930a8edf8137b5c3bb84dec6f5c58538261f","project":"openstack/networking-midonet","branch":"master","hashtags":[],"change_id":"I2f2b930a8edf8137b5c3bb84dec6f5c58538261f","subject":"[DNM] - Test tempest jobs with focal","status":"ABANDONED","created":"2020-09-15 03:15:19.000000000","updated":"2021-04-14 13:57:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"64c840805a2b1165d65ce8d8affb2b71675def20","_number":751963,"virtual_id_number":751963,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-18 05:45:40.000000000","_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-18 04:33:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 04:33:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 13:57:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"6c9b2173f86f11de979fc9214692b17800eac748","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-15 03:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6eb93c2c71fb8afe821ec61318031428e4df2bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 03:31:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d4296512f1949dba389bd1a697acd9d : FAILURE in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6d56b3bb97d44a99e3b3f55a5ceef44 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f5a8f93dce245e890da5f4613632252 : FAILURE in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c1245097d446798826a5ff131b3f20 : FAILURE in 14m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f8a4514e2e442e18fb023323db3a2ee : FAILURE in 4m 53s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/64d9b5d229e04fb6831f33e4a683dfd0 : FAILURE in 10m 19s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/0785e26dadaa44c6a7df0e664421b058 : FAILURE in 11m 29s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/c77de189d86d4734801fd096594ed79c : FAILURE in 11m 13s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/3a3107b9b6164e579133e0fc057cebdc : FAILURE in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df3e1a430bb3f257abd608b975f24f2a7d50aa2d","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-18 04:10:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82e916c8ed226ea7a229829b8c43ce07acd498b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 04:33:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64cd7c429d5c4594b9a32decef037a78 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecdde5328e7440edb1d10922b2667d49 : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e645150229c44bf49cfce94e4de37c3b : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6997754f3b54134ad4b2086433bb8a7 : SUCCESS in 13m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fecbd2ff897248478ac68ab44404e7ad : SUCCESS in 11m 42s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/3396452042e44396a021a975575ebd82 : FAILURE in 10m 09s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/18d89dddf8e34cf1bb35ba5a36abac33 : FAILURE in 17m 03s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/fd7a50f11a3a49158e574b42f6f01673 : FAILURE in 14m 04s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/c0c600026df845ed94a1eb8ddafe1919 : FAILURE in 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5cd5b4d638fea6fd890e97cce1f341d9f257e0a4","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-18 05:45:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad704ff2f4ed864eb7db81ad973ae347a5e0501b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 06:27:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6001f5907f2049f6b2693db2048678b1 : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4bacb8e2f6a455792d597305b2133c6 : SUCCESS in 11m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b12aa380ff84f03ab9939e30253abcc : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/828a852c889e440b97dc225644d4a4e5 : SUCCESS in 13m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4a0b5e28bef4fb6a6f86bd553fc98fe : SUCCESS in 11m 16s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/d0f123e5884f476993461ccd797a09e7 : FAILURE in 35m 16s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/5e6d8c9e30a64d1986ba11f0b94456b8 : FAILURE in 19m 30s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/3b23aa4c971145e78057d37d3f920195 : FAILURE in 28m 57s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/ebec812b0a174bfca8856a2745f23c97 : FAILURE in 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64c840805a2b1165d65ce8d8affb2b71675def20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:57:21.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"26075909e591c904eecd821f80c38570fdc5666f","revisions":{"37609561a6dc1431b62c35cb8180651acfc544d1":{"kind":"REWORK","_number":1,"created":"2020-09-15 03:15:19.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/63/751963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/63/751963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/751963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/751963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/751963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/63/751963/1"}}},"commit":{"parents":[{"commit":"af9b8a7444c72d69c1accdce70bcedba530f0ff8","subject":"Refactored zuul jobs and converted to zuul v3 syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/af9b8a7444c72d69c1accdce70bcedba530f0ff8"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-15 03:14:26.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-15 03:14:26.000000000","tz":600},"subject":"[DNM] - Test tempest jobs with focal","message":"[DNM] - Test tempest jobs with focal\n\nChange-Id: I2f2b930a8edf8137b5c3bb84dec6f5c58538261f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/37609561a6dc1431b62c35cb8180651acfc544d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/37609561a6dc1431b62c35cb8180651acfc544d1"}]},"branch":"refs/heads/master"},"26075909e591c904eecd821f80c38570fdc5666f":{"kind":"REWORK","_number":2,"created":"2020-09-18 04:10:32.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/63/751963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/63/751963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/751963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/751963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/63/751963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/63/751963/2"}}},"commit":{"parents":[{"commit":"af9b8a7444c72d69c1accdce70bcedba530f0ff8","subject":"Refactored zuul jobs and converted to zuul v3 syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/af9b8a7444c72d69c1accdce70bcedba530f0ff8"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-15 03:14:26.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-18 04:10:23.000000000","tz":600},"subject":"[DNM] - Test tempest jobs with focal","message":"[DNM] - Test tempest jobs with focal\n\nChange-Id: I2f2b930a8edf8137b5c3bb84dec6f5c58538261f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/26075909e591c904eecd821f80c38570fdc5666f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/26075909e591c904eecd821f80c38570fdc5666f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
