)]}'
{"id":"openstack%2Fnetworking-midonet~753875","triplet_id":"openstack%2Fnetworking-midonet~master~I7a52bc55f3cd468e0818e4bb28de8e8a92f03714","project":"openstack/networking-midonet","branch":"master","hashtags":[],"change_id":"I7a52bc55f3cd468e0818e4bb28de8e8a92f03714","subject":"Fix up devstack plugin, libvirt bin renamed in ubuntu","status":"MERGED","created":"2020-09-24 01:35:29.000000000","updated":"2020-09-26 21:09:38.000000000","submitted":"2020-09-26 21:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753875-1601154503667-5dd22bf2","meta_rev_id":"8a93f5ce5499a69132f868bf1bd3ce6c9231ebea","_number":753875,"virtual_id_number":753875,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-26 21:08:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2020-09-26 20:57:05.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-09-24 03:25:51.000000000","_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2020-09-26 20:57:05.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 03:25:51.000000000","updated_by":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"reviewer":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"state":"REVIEWER"},{"updated":"2020-09-26 20:57:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-26 21:08:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c45e3644503b6ded14d877921859d24a4f06094e","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-24 01:35:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4d520183a0542fe2180cc2f2859c1e821b442d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:24:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020bb626ead04fdcbe5690a271282a45 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/363cae4a786140f39b6a32b5dcf5417f : SUCCESS in 10m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd259ad60cdb4fa687536dd3929614b1 : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcdb7947d9245439e27b9decf4fa71c : SUCCESS in 7m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0784d7ea4b8471ea4de37878ed563e5 : SUCCESS in 12m 16s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/1aab2e1e35524e6d8199e8236f6706ce : SUCCESS in 34m 57s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/ef0addfffbed48e1879495efe810244b : FAILURE in 10m 28s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/2298e6c054ad4319a914ca57051d5080 : SUCCESS in 47m 47s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/f938f3afddc341a99fa22c355df20fad : FAILURE in 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"832c7d17fca21bd5cbd1c77c11c20e819ff85929","author":{"_account_id":156,"name":"Ryu Ishimoto","email":"ryu@midokura.com","username":"ryu-midokura"},"date":"2020-09-24 03:25:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6fac53414a4d56b1f6b37a2aac2139d7d9135ff","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-26 20:57:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9bc071943e95a91dcf1549615269bf5bdc17e0d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 20:57:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbbb9c815a5649ad9c2aa5d78eb8d6df866c696d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 21:08:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"255b3bb04b527580e0a7eb09eb5029a31980a723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 21:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65733a2e6bf64769a2ffbf3d132c4049 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21bbf2b0c011480ca01c99ca79266dc9 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1325cf92944d918f7d41a01ef0e24c : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1761e501332c47e28f2184da4f604d24 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8a93f5ce5499a69132f868bf1bd3ce6c9231ebea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 21:09:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d2ebc00d6f4a088e179f6dc3650f13 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af999193cb0e20ed00cdc6849fb81b58101cd208","revisions":{"af999193cb0e20ed00cdc6849fb81b58101cd208":{"kind":"REWORK","_number":1,"created":"2020-09-24 01:35:29.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/75/753875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/75/753875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/75/753875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/75/753875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/75/753875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/75/753875/1"}}},"commit":{"parents":[{"commit":"af9b8a7444c72d69c1accdce70bcedba530f0ff8","subject":"Refactored zuul jobs and converted to zuul v3 syntax","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/af9b8a7444c72d69c1accdce70bcedba530f0ff8"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-24 01:34:54.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-24 01:35:13.000000000","tz":600},"subject":"Fix up devstack plugin, libvirt bin renamed in ubuntu","message":"Fix up devstack plugin, libvirt bin renamed in ubuntu\n\nChange-Id: I7a52bc55f3cd468e0818e4bb28de8e8a92f03714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/af999193cb0e20ed00cdc6849fb81b58101cd208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/af999193cb0e20ed00cdc6849fb81b58101cd208"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
