)]}'
{"id":"openstack%2Fnetworking-midonet~754922","triplet_id":"openstack%2Fnetworking-midonet~master~I77c9222ce9c49a83e3284136db490b53c54b92de","project":"openstack/networking-midonet","branch":"master","hashtags":[],"change_id":"I77c9222ce9c49a83e3284136db490b53c54b92de","subject":"Run devstack jobs on bionic","status":"ABANDONED","created":"2020-09-29 07:47:14.000000000","updated":"2021-04-14 13:49:02.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e9043146d672d7ac45270140ad52a6f9e74f698","_number":754922,"virtual_id_number":754922,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-27 07:37:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 08:52:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-30 01:54:03.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2020-10-01 14:25:25.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"aa6aa8c4060078260899e3299a8274a6296af8a6","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-29 07:47:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99bd151affd2fbdadc304e92be2cdce52919d1ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 08:52:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ca555f8d4843a5ac79a8031d520c74 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1040de04c73b4fd7803be5840064084a : SUCCESS in 12m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06174895db8640dea9019df230df7d34 : SUCCESS in 13m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a6828b41064aa1b53c6af6bd32338c : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d79f5210acc8465088292d21b2081d07 : SUCCESS in 11m 29s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/6f465b15290541359aa0c82ff1591c1b : SUCCESS in 42m 11s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/5e2ae3c550e0489582338e4662735f66 : FAILURE in 17m 42s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/a0a8dcf9966e4a07883dda17f51a1e8f : SUCCESS in 55m 10s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/b2c4e51f9a004e7e90892d94919cbfa5 : FAILURE in 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b1c1617c177510e2e05503528782c1173f26179","author":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"date":"2020-09-30 01:54:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a4f6b209572dd3acc3eb6f1be4e2da32f8b18c2","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-30 03:55:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c03b2576dffeb3e1ed55974f97e67068a6b65c55","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-01 14:25:25.000000000","message":"Patch Set 1: Code-Review+2\n\nIt\u0027s still better than Xenial :)","accounts_in_message":[],"_revision_number":1},{"id":"df90c1e2543d5c886ef837d2a03c35774d52efa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2021-01-27 06:02:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e80cf5d0423ea56c5fbb88dd71013656bdb16d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 07:37:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c47337a22a774ba2b1b28f778595b4d0 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24f32126fbb145fa9318693c6ba9a054 : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68dc6bb37f8641f4a91d6e99646591de : FAILURE in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62f45685c64943a6b1c9defe9dae1179 : FAILURE in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73051739666d44f5b3515305cb62734f : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35047259dd214d8c90049084e0bb90b0 : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bf068772bbb4e1a952d84de5aeb14f2 : FAILURE in 8m 05s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/8c7b6b898f3c4588a3e4d9328bb1c6f1 : FAILURE in 1h 08m 27s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/06c6081bc68c430fb33d02f1b670883d : FAILURE in 17m 58s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/c2f815006fd64724a1b32a5b69d7cac1 : SUCCESS in 1h 12m 55s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/ce59b13a8c0f4033b7baa4940a135ccc : FAILURE in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e9043146d672d7ac45270140ad52a6f9e74f698","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:49:02.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"33544927c0c3467b563beb99776b224268b1828a","revisions":{"36dc4ae599674ff9f3ef388d4ae622e6fa45472c":{"kind":"REWORK","_number":1,"created":"2020-09-29 07:47:14.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/22/754922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/22/754922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/22/754922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/22/754922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/22/754922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/22/754922/1"}}},"commit":{"parents":[{"commit":"1dcb2d2947d46189959deea7bc9ca8be8e762917","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1dcb2d2947d46189959deea7bc9ca8be8e762917"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-29 07:44:50.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-29 07:44:50.000000000","tz":600},"subject":"Run devstack jobs on bionic","message":"Run devstack jobs on bionic\n\nCurrently the zookeeper that is packaged with Ubuntu Focal does\nnot run on jdk8 (even though no dependencies allow it)\n\nUntil a proper fix is found continue to use bionic.\n\nChange-Id: I77c9222ce9c49a83e3284136db490b53c54b92de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/36dc4ae599674ff9f3ef388d4ae622e6fa45472c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/36dc4ae599674ff9f3ef388d4ae622e6fa45472c"}]},"branch":"refs/heads/master"},"33544927c0c3467b563beb99776b224268b1828a":{"kind":"REWORK","_number":2,"created":"2021-01-27 06:02:46.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/22/754922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/22/754922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/22/754922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/22/754922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/22/754922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/22/754922/2"}}},"commit":{"parents":[{"commit":"1dcb2d2947d46189959deea7bc9ca8be8e762917","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1dcb2d2947d46189959deea7bc9ca8be8e762917"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-29 07:44:50.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2021-01-27 06:02:21.000000000","tz":660},"subject":"Run devstack jobs on bionic","message":"Run devstack jobs on bionic\n\nCurrently the zookeeper that is packaged with Ubuntu Focal does\nnot run on jdk8 (even though no dependencies allow it)\n\nUntil a proper fix is found continue to use bionic.\n\n(Also fix some dependency issues and sync flake settings with neutron project)\n\nChange-Id: I77c9222ce9c49a83e3284136db490b53c54b92de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/33544927c0c3467b563beb99776b224268b1828a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/33544927c0c3467b563beb99776b224268b1828a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
