)]}'
{"id":"openstack%2Fnetworking-midonet~757326","triplet_id":"openstack%2Fnetworking-midonet~master~I62902c76b016861c008a9f8e705645c4c7c3730f","project":"openstack/networking-midonet","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I62902c76b016861c008a9f8e705645c4c7c3730f","subject":"Imported Translations from Zanata","status":"ABANDONED","created":"2020-10-11 07:19:47.000000000","updated":"2021-04-14 13:49:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1543157072d67cf9d16c432364cb9637df44fa6a","_number":757326,"virtual_id_number":757326,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-27 04:55:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2021-04-14 13:49:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-11 08:09:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-11 18:32:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-12-01 21:41:55.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"CC"}],"messages":[{"id":"f64f772ce91b82045a00619e5a7a0426e7191d43","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-11 07:19:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49c0247da5bb02b903519f4b20d70d53970cd372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 08:09:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8af471a3c5479485418290c625e934 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6151d369c34ba5a45630e02da7a08e : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28518b96bbe44d7981db48f07719fbfd : SUCCESS in 4m 17s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/74d2f03734684864aeb852450361e158 : FAILURE in 13m 56s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/c1403363e1b246a5a79880b10bd3bc82 : FAILURE in 15m 59s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/933315ba3b58422c9e9b334554d70abd : SUCCESS in 44m 29s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/ca91d5468e14415b8692e18008e7f810 : FAILURE in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84b34bbbcc34272766a474985a2b412cab7bd904","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-10-11 18:32:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1bbb5eadebaa0a3f5d8a2370ed8a3c225c9a18f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 19:15:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548083268fde4e02baabef0e485b833e : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1d664fd8e14de1a2861d65ab8dd36d : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a96458b9149546379fdcd0d216f97810 : SUCCESS in 5m 28s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/5b59391afa484973a3cda8334e5c3cac : FAILURE in 9m 38s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/c1b612669db441a08c4071f6f89af16e : FAILURE in 10m 56s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/724ba0e461d6478fb68d658844f942e9 : SUCCESS in 41m 36s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/01ed18c256b1413c8043814ba94fd9f7 : FAILURE in 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57def7f0e6bb89acefd1297baf565cb57941e03a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-12-01 08:47:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3da3ec9a6d42be00215f94569c4390f81f9c6227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 09:24:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b85cfdd191401882da45d473f2bdd7 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0f96fea0744ca3930542070eb89702 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f951b994a89e427f97b7fa62f9fa703d : RETRY_LIMIT in 2m 39s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/d2ab986d8b834449b158994fc6adf2f6 : FAILURE in 10m 54s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/d1dd9bd230b54a0ab806c82fbb452888 : FAILURE in 12m 31s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/c709d166ab68482a82d60af2e9d410d4 : FAILURE in 32m 08s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/5ff16e5c991940c0ac7d60df1e352cae : FAILURE in 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f028c3836f98bc67278696f194ec6c4935b73eeb","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-12-01 21:41:55.000000000","message":"Patch Set 1:\n\nThis will need https://review.opendev.org/c/openstack/networking-midonet/+/754922 which is still not merged","accounts_in_message":[],"_revision_number":1},{"id":"bb7e29056d2c3cd81e2cff7f6ceecd9039f655ba","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-25 22:08:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"33841dd5e8cb9751760eebbc43a794b5e5718fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 23:19:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f969176409c41caad573deea547efab : FAILURE in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd7d7f573c343d99a5831476e1294c8 : FAILURE in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d80c14eb80542cca59c02cbdfc366e0 : RETRY_LIMIT in 3m 16s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/34462532a67548c8832194c7b83d40d0 : FAILURE in 14m 19s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/5b3c2d88e8fe44f28efb4def3e9ea47a : FAILURE in 13m 18s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/513e2ddaad4e47f2ae1457063bba9497 : FAILURE in 1h 07m 09s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/d87bca588394472eb68b090447a7737c : FAILURE in 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec810186e2f0736ec7857eb66aeb809f6e364a91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2021-01-27 04:03:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bbc96278e7007c154b3100bfc50c5f7ea5ad9b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 04:55:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07b8d4fe02144d6a5b2d05fe2e50738 : FAILURE in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f603e48c7a5d450786d73f178bb6bd87 : FAILURE in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e47a08586cb46c9b9504e1be1f54cec : RETRY_LIMIT in 2m 19s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/1caef13cdbfe44f9acca045e639ec148 : SUCCESS in 26m 02s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/dda6471933be43cab208d6653b575276 : FAILURE in 13m 26s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/0c1f6575a1e8437a8ce18633200d8e06 : SUCCESS in 39m 52s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/f886a1a3b5ab49c6b1d5f63f7516ab72 : FAILURE in 48m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6ddf052a7cc176849ab67e178b6dc237eede9ed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-14 13:49:18.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c5b90b487f6cd9351fa46a5c149630e81380933","revisions":{"1332c7221fab543e1118607bf15d918b6977ba9f":{"kind":"REWORK","_number":1,"created":"2020-10-11 07:19:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/757326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/26/757326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/26/757326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/26/757326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/26/757326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/26/757326/1"}}},"commit":{"parents":[{"commit":"1dcb2d2947d46189959deea7bc9ca8be8e762917","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1dcb2d2947d46189959deea7bc9ca8be8e762917"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-11 07:19:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-11 07:19:43.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I62902c76b016861c008a9f8e705645c4c7c3730f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1332c7221fab543e1118607bf15d918b6977ba9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1332c7221fab543e1118607bf15d918b6977ba9f"}]},"branch":"refs/heads/master"},"2c5b90b487f6cd9351fa46a5c149630e81380933":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-27 04:03:49.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/26/757326/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/26/757326/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/26/757326/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/26/757326/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/26/757326/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/26/757326/2"}}},"commit":{"parents":[{"commit":"7c6a9d3f8ab28c846283a7f35db7dd854295b3ee","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7c6a9d3f8ab28c846283a7f35db7dd854295b3ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-11 07:19:43.000000000","tz":0},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2021-01-27 04:03:49.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I62902c76b016861c008a9f8e705645c4c7c3730f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2c5b90b487f6cd9351fa46a5c149630e81380933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2c5b90b487f6cd9351fa46a5c149630e81380933"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
