)]}'
{"id":"openstack%2Fnetworking-midonet~770797","triplet_id":"openstack%2Fnetworking-midonet~master~Ib237ad06ba15c265caf25bfdf91e89938438be4e","project":"openstack/networking-midonet","branch":"master","topic":"bp/enginefacade-switch","hashtags":[],"change_id":"Ib237ad06ba15c265caf25bfdf91e89938438be4e","subject":"Switch to the new engine facade in networking-midonet","status":"ABANDONED","created":"2021-01-14 12:13:44.000000000","updated":"2021-04-06 07:15:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e8c1bb5fe8551036f68932e2389bb869237589c","_number":770797,"virtual_id_number":770797,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-27 05:06:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 13:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 08:25:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-03-29 11:53:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"10118b4bf714d0c7c8293d5858e5f12b84cf45db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-14 12:13:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef60add3db4102477c1fa78ae4ee0c3e9313d848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:22:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98623b1fd4d74cd1b7c8e1a08061d856 : FAILURE in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f0a18e517a04d0386b9f57a5f0cab54 : FAILURE in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da919e4dec1e499a8663d2e4a5dead95 : FAILURE in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5cf4409e1e42458e81eca5e07d5d16 : FAILURE in 4m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94c55ddffc1b4db5ac1639822066d237 : FAILURE in 4m 07s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/42182b1d683a4ecabe7de76eb398a072 : FAILURE in 14m 16s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/a9b81477c6f84cb2bad64ff74c05b32a : FAILURE in 10m 54s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/03d886444d204c6eae50f05ae22aa065 : SUCCESS in 1h 01m 37s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/5d7426ffd2ce44388d67016b5ea19120 : FAILURE in 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecd8bf7ae779dbb822df441e2d17ecf2a6c1f170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-01-18 15:16:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24c81ba54062120f684e3e0412b010fd78e383e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:19:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ade22f7a1345faa6ee384a360c75b3 : FAILURE in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ec43704d7e244e68b5259caf9bd1105 : FAILURE in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16047cfb43534cac8067946da61da1cc : FAILURE in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335b2e7858654ba19aeb58e83bc9882e : FAILURE in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/add14fab01894b06977535eb2a2fdc32 : FAILURE in 5m 35s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/5703aea7c9ce4b978286d8801f3cd793 : FAILURE in 9m 12s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/8d95a942bbbd4c6faf8518aac43b6181 : FAILURE in 15m 20s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/4e2beab158f64e6bac9283b796b7ca21 : FAILURE in 52m 18s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/445fe67620f94fbb93a659db866ca5c9 : FAILURE in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b3b99dd0c0a4bc6c536489ac94fb66889ccd83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2021-01-27 04:04:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1f93d0640189ebda2fb2b07818aeeb894f155b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 05:06:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264958501c3140e6844d4b31df1815fb : FAILURE in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25197fc45ed84fc29b572acbe98170d8 : FAILURE in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43a36e0c2cac44ab9fabe14284eb0613 : FAILURE in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1faaabbb47f94f9b93013b494349cd3d : FAILURE in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a129867f8b46b79ab27450253f271d : FAILURE in 13m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fab5d3d85694176a317172e263abd0a : FAILURE in 3m 39s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/d2302e577bdd43169ac5bc0fd5e5ed39 : SUCCESS in 26m 58s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/f3391e6882ae466e94962ddbb6b63e06 : FAILURE in 11m 46s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/784fbc0d34fa44c9ab354ff4187d8409 : SUCCESS in 59m 01s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/634bc728661243a98db034206d0e91e2 : FAILURE in 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"71ee911a5eaa3cf13be21deae4db1e5e883e714c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-29 08:25:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"578898bc7756a09246046387ea43ffd11a4b6549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 09:14:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f077885bfd1c472099819b811140fcf8 : FAILURE in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b10c606e3aaf4b34965c9f93294f5b5d : FAILURE in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/717186c1d763420a89bf15d872f90d31 : FAILURE in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9fd2080cb4b4e319b14d1f783c4d6b3 : FAILURE in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f059a74aae004f92a225129ad5245db0 : FAILURE in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fd4ad34623b4ecf84616a5f27898f44 : FAILURE in 5m 09s\n- networking-midonet-ml2-tempest-aio https://zuul.opendev.org/t/openstack/build/f902109edc4841e4a07ba56076056a8e : SUCCESS in 35m 22s\n- networking-midonet-ml2-tempest-aio-centos8 https://zuul.opendev.org/t/openstack/build/9f5564190e974e84b10f9d1209ae50b8 : FAILURE in 12m 40s (non-voting)\n- networking-midonet-ml2-tempest-multinode https://zuul.opendev.org/t/openstack/build/6675f925271b4de28af94ce32dc2481a : SUCCESS in 45m 00s\n- networking-midonet-ml2-grenade https://zuul.opendev.org/t/openstack/build/c97b4dc202be49edb7a03412b3edd206 : FAILURE in 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e8c1bb5fe8551036f68932e2389bb869237589c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-06 07:15:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3e30e72862e686b194ca49cda0700816b20a6c65","revisions":{"f734299bec80419772ced34c504fd09c3ce8ab44":{"kind":"REWORK","_number":1,"created":"2021-01-14 12:13:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/770797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/97/770797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/1"}}},"commit":{"parents":[{"commit":"1dcb2d2947d46189959deea7bc9ca8be8e762917","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1dcb2d2947d46189959deea7bc9ca8be8e762917"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-14 12:12:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-14 12:13:40.000000000","tz":60},"subject":"Switch to the new engine facade in networking-midonet","message":"Switch to the new engine facade in networking-midonet\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib237ad06ba15c265caf25bfdf91e89938438be4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f734299bec80419772ced34c504fd09c3ce8ab44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/f734299bec80419772ced34c504fd09c3ce8ab44"}]},"branch":"refs/heads/master"},"56867997b9b00caa7a39eb7e5d15bc34bfee026c":{"kind":"REWORK","_number":2,"created":"2021-01-18 15:16:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/97/770797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/97/770797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/2"}}},"commit":{"parents":[{"commit":"1dcb2d2947d46189959deea7bc9ca8be8e762917","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/1dcb2d2947d46189959deea7bc9ca8be8e762917"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-14 12:12:39.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-18 15:16:34.000000000","tz":60},"subject":"Switch to the new engine facade in networking-midonet","message":"Switch to the new engine facade in networking-midonet\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib237ad06ba15c265caf25bfdf91e89938438be4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/56867997b9b00caa7a39eb7e5d15bc34bfee026c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/56867997b9b00caa7a39eb7e5d15bc34bfee026c"}]},"branch":"refs/heads/master"},"3e30e72862e686b194ca49cda0700816b20a6c65":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-27 04:04:19.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/97/770797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/97/770797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/97/770797/3"}}},"commit":{"parents":[{"commit":"2c5b90b487f6cd9351fa46a5c149630e81380933","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/2c5b90b487f6cd9351fa46a5c149630e81380933"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-01-14 12:12:39.000000000","tz":60},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2021-01-27 04:04:19.000000000","tz":0},"subject":"Switch to the new engine facade in networking-midonet","message":"Switch to the new engine facade in networking-midonet\n\nPartially-Implements blueprint: enginefacade-switch\n\nChange-Id: Ib237ad06ba15c265caf25bfdf91e89938438be4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3e30e72862e686b194ca49cda0700816b20a6c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/3e30e72862e686b194ca49cda0700816b20a6c65"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
