)]}'
{"id":"openstack%2Fnetworking-midonet~796498","triplet_id":"openstack%2Fnetworking-midonet~stable%2Fstein~Ice02fbdad1ab8d81ffc20668ff292ff18805a2e2","project":"openstack/networking-midonet","branch":"stable/stein","topic":"clean-non-voting-jobs-stable/stein","hashtags":[],"change_id":"Ice02fbdad1ab8d81ffc20668ff292ff18805a2e2","subject":"[EM releases] Move non-voting jobs to the experimental queue","status":"ABANDONED","created":"2021-06-15 14:52:48.000000000","updated":"2021-08-15 08:30:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8cbb916e58594e3a31cd1ebff2becfc324f21906","_number":796498,"virtual_id_number":796498,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-15 15:16:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2021-06-16 07:44:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 15:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 07:44:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"edb9e885422ce309c8fa6af98230cd855adb7d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-06-15 14:52:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca85c918af0c5eae5511b67f736da4eb81ef3bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:16:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbbb21e9ce234a8b9a29041406a16241 : FAILURE in 10m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ed7830c21de491d9d99e23aae47947d : FAILURE in 9m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2e474bb7d06646598205349f83b4d32d : SUCCESS in 21m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aef1c7cb9d9495ebbbcafa2213542f2 : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2373edd18dea4d20a21b0c9eff2ee763 : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7618b412cf2401d846d8eabcb79c188 : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":1},{"id":"eb979bc0ba56dbc2bb23c56b532325dc66a818e0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-06-16 07:44:40.000000000","message":"Patch Set 1: Code-Review+2\n\nSimilar issue to other repos on unfixed:\nAttributeError: module \u0027isort\u0027 has no attribute \u0027SortImports\u0027\n\nBut this cleans a lot of the jobs indeed","accounts_in_message":[],"_revision_number":1},{"id":"8cbb916e58594e3a31cd1ebff2becfc324f21906","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-15 08:30:28.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd6ffa4095be2949cc4a8636c0122fffd49c38d0","revisions":{"bd6ffa4095be2949cc4a8636c0122fffd49c38d0":{"kind":"REWORK","_number":1,"created":"2021-06-15 14:52:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/98/796498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/98/796498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/98/796498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/98/796498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/98/796498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/98/796498/1"}}},"commit":{"parents":[{"commit":"a8f1b922e50594a0dba9e03bf365b487f7cb097e","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/a8f1b922e50594a0dba9e03bf365b487f7cb097e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 14:52:34.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-06-15 14:52:34.000000000","tz":120},"subject":"[EM releases] Move non-voting jobs to the experimental queue","message":"[EM releases] Move non-voting jobs to the experimental queue\n\nDuring the CI meeting we agreed that non-voting jobs in the branches\nwhich are in Extened Maintenance (EM) phase should be moved to from\nthe check to the experimental queue.\n\nThis patch is doing exactly that.\n\nChange-Id: Ice02fbdad1ab8d81ffc20668ff292ff18805a2e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bd6ffa4095be2949cc4a8636c0122fffd49c38d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/bd6ffa4095be2949cc4a8636c0122fffd49c38d0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
