)]}'
{"id":"openstack%2Fnetworking-midonet~864866","triplet_id":"openstack%2Fnetworking-midonet~stable%2Fussuri~I6fd40b1ec59a840272261fca039c8fb49fc5e478","project":"openstack/networking-midonet","branch":"stable/ussuri","topic":"fix-zuul-errors","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-11-23 13:44:28.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-11-23 13:33:27.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I6fd40b1ec59a840272261fca039c8fb49fc5e478","subject":"Replace \"x/tap-as-a-service\" with \"openstack/tap-as-a-service\"","status":"MERGED","created":"2022-11-17 10:27:54.000000000","updated":"2022-11-23 13:44:28.000000000","submitted":"2022-11-23 13:44:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864866-fix-zuul-errors","meta_rev_id":"163c4b9b6b66925aca3edb0fcad2763fcda2b7d8","_number":864866,"virtual_id_number":864866,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 13:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-11-23 13:33:27.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-11-23 12:04:57.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-11-23 13:33:27.000000000","_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-17 10:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 12:04:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-23 12:08:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"62a9f144280cbae04eedf73fc1c28ddf1a9a5fb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-17 10:27:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9834265f9c89329aaae3ed4c27cf0bcb699d7af9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 10:42:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70cef11b8332420ba095bb72bf11d27a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ce5bdfe923d4087b2161c23620ec3d0 : RETRY_LIMIT in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b822cb9b64ef481ba12b048addded732 : FAILURE in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c96ead9512a4c4b813fb812200f2bca : FAILURE in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70bb8eaec7e24faf8a010595160b8597 : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d0878e130144870ab5f21268a281809 : FAILURE in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e83ed0358a5b4843bdd90a98c08ee1cf : FAILURE in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da16f3037e4441e09209093aff37dcbe : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d24bd1022497a44cde5ad30f10e5b5adb5c4ef23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 15:14:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed91ad101fd724080de0d67e473a1b88f3eb97b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 15:31:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ac78ea6aecc4f5bbefe341db7267ace\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc125faf72ed4cae95aeca0aae164fb7 : RETRY_LIMIT in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90100c753557420c8f72447880a85fdc : FAILURE in 2m 50s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/574965955bba4183a7f4f002e0bb85c2 : FAILURE in 3m 23s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/316cd26e84cc4d43bb5d365563ee12f8 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4009d22f732f4ad2860da879ef6d1de1 : FAILURE in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/198fb79e5fd04542a1592ccd5eea8a22 : FAILURE in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d21723092ec4eec83fea1a364df2224 : FAILURE in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9bee31362fcfcfcc5e987a63961b5c900ac0ea29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-22 15:38:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2fe67ee47f12f3b06c0b6c881b67886185bb7a0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 15:50:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7cf19f17e1403cb84629b01596b3d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a2d60ce360d4a6c9a94d48eab174235 : FAILURE in 3m 38s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92d194d4ce614844a11f019cfb6dc7ea : FAILURE in 3m 26s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa1c4a559fec412499b7f17c2bae81f6 : FAILURE in 3m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27e096aa39e64d01b60128d68d2a10b9 : FAILURE in 2m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"239e7f10528826e24881db1bd4cf041fe4475c3b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-23 12:04:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c4eb1b9d34ce65cbbf6e083c89529ff0ebcd06fc","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-23 13:33:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f825821f03a053f7f099dfebeac25d21eb813bd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 13:34:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d546d98ae5023b275f24a8411f6cd4269fc511a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 13:44:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a8dcfefb7f143f0b9ef09695cf17149\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f10bd3a9cdc4f8bbb55492ade030559 : FAILURE in 4m 20s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3928937809e47bf94518ea44885a951 : FAILURE in 4m 33s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdbfcf5a2cc74d069d035bf4eed467c9 : FAILURE in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"163c4b9b6b66925aca3edb0fcad2763fcda2b7d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 13:44:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bebd5e1ee8415e1808b10c960f5722a207e79f7","revisions":{"6a6a713b9a614bd6e52b6f9e6c2cdc5a9ed0cab4":{"kind":"REWORK","_number":1,"created":"2022-11-17 10:27:54.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/864866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/66/864866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/1"}}},"commit":{"parents":[{"commit":"002f38c9b784f3625fe0a0ed01985b1fe4e7b2ef","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/002f38c9b784f3625fe0a0ed01985b1fe4e7b2ef"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 17:30:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 17:30:49.000000000","tz":60},"subject":"Replace \"x/tap-as-a-service\" with \"openstack/tap-as-a-service\"","message":"Replace \"x/tap-as-a-service\" with \"openstack/tap-as-a-service\"\n\nThe project \"tap-as-a-service\" was restored and now is being\nactively developed again.\n\nChange-Id: I6fd40b1ec59a840272261fca039c8fb49fc5e478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6a6a713b9a614bd6e52b6f9e6c2cdc5a9ed0cab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/6a6a713b9a614bd6e52b6f9e6c2cdc5a9ed0cab4"}]},"branch":"refs/heads/stable/ussuri"},"9d3392dec4ad281425af8adffba9ea63175cc78e":{"kind":"REWORK","_number":2,"created":"2022-11-22 15:14:18.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/864866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/66/864866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/2"}}},"commit":{"parents":[{"commit":"002f38c9b784f3625fe0a0ed01985b1fe4e7b2ef","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/002f38c9b784f3625fe0a0ed01985b1fe4e7b2ef"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 17:30:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 15:44:26.000000000","tz":60},"subject":"Replace \"x/tap-as-a-service\" with \"openstack/tap-as-a-service\"","message":"Replace \"x/tap-as-a-service\" with \"openstack/tap-as-a-service\"\n\nThe project \"tap-as-a-service\" was restored and now is being\nactively developed again.\n\nThis patch is also marking all check and gate jobs as non-voting\nuntil they are fixed.\n\nChange-Id: I6fd40b1ec59a840272261fca039c8fb49fc5e478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9d3392dec4ad281425af8adffba9ea63175cc78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/9d3392dec4ad281425af8adffba9ea63175cc78e"}]},"branch":"refs/heads/stable/ussuri"},"7bebd5e1ee8415e1808b10c960f5722a207e79f7":{"kind":"REWORK","_number":3,"created":"2022-11-22 15:38:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/864866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-midonet","ref":"refs/changes/66/864866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-midonet refs/changes/66/864866/3"}}},"commit":{"parents":[{"commit":"002f38c9b784f3625fe0a0ed01985b1fe4e7b2ef","subject":"Use the networking-l2gw project new location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/002f38c9b784f3625fe0a0ed01985b1fe4e7b2ef"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-13 17:30:49.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-11-15 16:07:48.000000000","tz":60},"subject":"Replace \"x/tap-as-a-service\" with \"openstack/tap-as-a-service\"","message":"Replace \"x/tap-as-a-service\" with \"openstack/tap-as-a-service\"\n\nThe project \"tap-as-a-service\" was restored and now is being\nactively developed again.\n\nThis patch is also marking all check and gate jobs as non-voting\nuntil they are fixed. Some templates (check-requirements,\npublish-openstack-docs-pti and release-notes-jobs-python3) are\nremoved too.\n\nChange-Id: I6fd40b1ec59a840272261fca039c8fb49fc5e478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7bebd5e1ee8415e1808b10c960f5722a207e79f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-midonet/commit/7bebd5e1ee8415e1808b10c960f5722a207e79f7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}]}],"submit_requirements":[]}
