)]}'
{"id":"openstack%2Fnetworking-odl~652489","triplet_id":"openstack%2Fnetworking-odl~master~Ibda74052c2975d95f343f9e1f6c1a341aa1bbeb5","project":"openstack/networking-odl","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"Ibda74052c2975d95f343f9e1f6c1a341aa1bbeb5","subject":"Add Python 3 Train unit tests","status":"ABANDONED","created":"2019-04-15 02:11:41.000000000","updated":"2019-07-18 11:30:13.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1f0786eaa7ceb662f1fde76ced0eec063b82ed6","_number":652489,"virtual_id_number":652489,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"date":"2019-07-18 11:30:06.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":-1,"date":"2019-07-18 04:26:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":-1,"date":"2019-07-18 10:57:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-16 08:38:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-04-16 16:43:38.000000000","updated_by":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"reviewer":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"state":"REVIEWER"},{"updated":"2019-06-12 20:13:23.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-07-17 20:14:22.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2019-07-18 04:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-18 10:57:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"78c60af989a12d4291b59e2522b4297abbfc25d8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 02:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"734fe0e0f353f1672d9968d00fd81fcc7c45442c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 04:54:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/89/652489/1/check/test-release-openstack/60056f2/ : SUCCESS in 2m 00s\n- openstack-tox-pep8 http://logs.openstack.org/89/652489/1/check/openstack-tox-pep8/886b31d/ : SUCCESS in 6m 41s\n- openstack-tox-py27 http://logs.openstack.org/89/652489/1/check/openstack-tox-py27/012751b/ : SUCCESS in 6m 19s\n- openstack-tox-py36 http://logs.openstack.org/89/652489/1/check/openstack-tox-py36/935a173/ : SUCCESS in 6m 10s\n- build-openstack-releasenotes http://logs.openstack.org/89/652489/1/check/build-openstack-releasenotes/823c6b8/html/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/89/652489/1/check/openstack-tox-docs/ecb2e58/html/ : SUCCESS in 6m 12s\n- openstack-tox-cover http://logs.openstack.org/89/652489/1/check/openstack-tox-cover/b22ebf3/cover/ : SUCCESS in 6m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/652489/1/check/openstack-tox-lower-constraints/31c0b07/ : SUCCESS in 8m 05s\n- networking-odl-grenade http://logs.openstack.org/89/652489/1/check/networking-odl-grenade/f35c837/ : FAILURE in 21m 32s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/89/652489/1/check/networking-odl-tempest-oxygen/3d66243/ : POST_FAILURE in 21m 59s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/89/652489/1/check/networking-odl-tempest-fluorine/a21fafe/ : TIMED_OUT in 2h 37m 09s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/89/652489/1/check/networking-odl-tempest-oxygen-multinode/7543875/ : FAILURE in 46m 06s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/89/652489/1/check/networking-odl-tempest-fluorine-multinode/c9d2462/ : FAILURE in 48m 39s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/89/652489/1/check/networking-odl-rally-oxygen/29c6972/ : SUCCESS in 58m 25s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/89/652489/1/check/networking-odl-rally-fluorine/be48d9f/ : SUCCESS in 1h 05m 51s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/89/652489/1/check/networking-odl-functional-oxygen/4bd0b64/ : SUCCESS in 16m 27s\n- networking-odl-functional-fluorine http://logs.openstack.org/89/652489/1/check/networking-odl-functional-fluorine/5cd0607/ : SUCCESS in 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"95a89c241ce1795af8482828f9c93d3c6a211c85","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-04-15 14:06:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nWhere there is only py35 maybe it would be better to change to py36 instead of just removing py35.","accounts_in_message":[],"_revision_number":1},{"id":"679bf6cc1553b67afeb79a4cd7ebc75fe652cd8b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 16:26:21.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e61949b4d1355d197d850110b785617ef6fdf955","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-15 16:28:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1004d94fd48c5cb795776b0f29fe888b7c6d3e8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 19:32:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/89/652489/2/check/test-release-openstack/46d2bcc/ : SUCCESS in 2m 52s\n- openstack-tox-pep8 http://logs.openstack.org/89/652489/2/check/openstack-tox-pep8/3742cf8/ : SUCCESS in 7m 46s\n- openstack-tox-py27 http://logs.openstack.org/89/652489/2/check/openstack-tox-py27/e953d26/ : SUCCESS in 7m 04s\n- openstack-tox-py36 http://logs.openstack.org/89/652489/2/check/openstack-tox-py36/f1f2b7b/ : SUCCESS in 7m 44s\n- build-openstack-releasenotes http://logs.openstack.org/89/652489/2/check/build-openstack-releasenotes/66477d7/html/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/89/652489/2/check/openstack-tox-docs/e4dc620/html/ : SUCCESS in 6m 27s\n- openstack-tox-cover http://logs.openstack.org/89/652489/2/check/openstack-tox-cover/6040dc2/cover/ : SUCCESS in 8m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/652489/2/check/openstack-tox-lower-constraints/c683ef8/ : SUCCESS in 8m 39s\n- networking-odl-grenade http://logs.openstack.org/89/652489/2/check/networking-odl-grenade/61768dd/ : FAILURE in 17m 32s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/89/652489/2/check/networking-odl-tempest-oxygen/b590b48/ : POST_FAILURE in 23m 29s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/89/652489/2/check/networking-odl-tempest-fluorine/8e50c43/ : TIMED_OUT in 2h 37m 32s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/89/652489/2/check/networking-odl-tempest-oxygen-multinode/5e2fef2/ : FAILURE in 54m 38s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/89/652489/2/check/networking-odl-tempest-fluorine-multinode/621acf2/ : FAILURE in 49m 52s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/89/652489/2/check/networking-odl-rally-oxygen/b055653/ : SUCCESS in 1h 04m 29s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/89/652489/2/check/networking-odl-rally-fluorine/b4448d6/ : SUCCESS in 1h 02m 47s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/89/652489/2/check/networking-odl-functional-oxygen/1a7cdf9/ : SUCCESS in 17m 49s\n- networking-odl-functional-fluorine http://logs.openstack.org/89/652489/2/check/networking-odl-functional-fluorine/b9f4e5f/ : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"6c7ba7db3c34482d4b4f212658aab47d8adf8457","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-04-16 08:38:19.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks! Looks good!","accounts_in_message":[],"_revision_number":2},{"id":"e6ee883f5cb212c4c6a9585066ec4c7d8abaa1a3","author":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"date":"2019-04-16 16:43:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2ca2b4d84cd481ec7bacd4685557afd4fc51f911","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-10 10:18:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b89d5e064ce47ddd22d5a3c31865d68d14b2f834","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-12 20:13:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"27457767653c9216c5e02b35f2a3aa27cfbaa600","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:13:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fae38875983573f44762971a68133b8bb7fa5937","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 21:12:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/89/652489/2/gate/test-release-openstack/4ca9923/ : SUCCESS in 1m 56s\n- openstack-tox-pep8 http://logs.openstack.org/89/652489/2/gate/openstack-tox-pep8/703a923/ : FAILURE in 7m 51s\n- openstack-tox-py27 http://logs.openstack.org/89/652489/2/gate/openstack-tox-py27/34aa1d0/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/89/652489/2/gate/openstack-tox-py36/0128ffa/ : SUCCESS in 6m 16s\n- build-openstack-releasenotes http://logs.openstack.org/89/652489/2/gate/build-openstack-releasenotes/5408966/html/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/89/652489/2/gate/openstack-tox-docs/79f0671/html/ : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/652489/2/gate/openstack-tox-lower-constraints/5f7ed75/ : FAILURE in 8m 36s\n- networking-odl-functional-oxygen http://logs.openstack.org/89/652489/2/gate/networking-odl-functional-oxygen/4f9ac47/ : SUCCESS in 18m 19s\n- networking-odl-functional-fluorine http://logs.openstack.org/89/652489/2/gate/networking-odl-functional-fluorine/8e34dae/ : SUCCESS in 17m 57s","accounts_in_message":[],"_revision_number":2},{"id":"73cd0db0da3d53df4dbef8ea13754f3da3ecff16","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-06-13 11:27:27.000000000","message":"Patch Set 2:\n\nI think this must go in first: https://review.opendev.org/658352 to fix bandit related issues.","accounts_in_message":[],"_revision_number":2},{"id":"542bd8ac14b61332e007f723953e1985662935d2","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-17 20:14:22.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nCan we abandon this in favor of https://review.opendev.org/#/c/669420/ ?","accounts_in_message":[],"_revision_number":2},{"id":"8ae5649611a21caf79cf4c20e6f9d49c86d6bd4e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-18 02:52:06.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Can we abandon this in favor of https://review.opendev.org/#/c/669420/\n \u003e ?\n\nI will add the py37 changes here. Because this is removing functional 35 jobs also.","accounts_in_message":[],"_revision_number":2},{"id":"e0b81fe635b289085bbc3b990f8d951fabad821b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-18 02:54:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"de53c3a7147f32ba613912f70ccd153a7a847b58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 02:54:21.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"c042d5f4649c9eb80370bb311a83cf62cf2b7bcc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-18 02:56:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9c9d866487a019bc039bd6a07142a264c1b987ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 04:26:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/89/652489/4/check/test-release-openstack/c38b42d/ : FAILURE in 3m 06s\n- openstack-tox-pep8 http://logs.openstack.org/89/652489/4/check/openstack-tox-pep8/e10c3c6/ : SUCCESS in 9m 04s\n- openstack-tox-py27 http://logs.openstack.org/89/652489/4/check/openstack-tox-py27/6f7f3dc/ : SUCCESS in 9m 52s\n- openstack-tox-py36 http://logs.openstack.org/89/652489/4/check/openstack-tox-py36/a93ca99/ : SUCCESS in 9m 54s\n- openstack-tox-py37 http://logs.openstack.org/89/652489/4/check/openstack-tox-py37/338eb49/ : FAILURE in 5m 24s\n- build-openstack-releasenotes http://logs.openstack.org/89/652489/4/check/build-openstack-releasenotes/b683177/html/ : SUCCESS in 4m 54s\n- openstack-tox-docs http://logs.openstack.org/89/652489/4/check/openstack-tox-docs/64b1caf/html/ : SUCCESS in 6m 48s\n- openstack-tox-cover http://logs.openstack.org/89/652489/4/check/openstack-tox-cover/83f81a6/cover/ : SUCCESS in 12m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/652489/4/check/openstack-tox-lower-constraints/5ba88a7/ : SUCCESS in 12m 59s\n- networking-odl-grenade http://logs.openstack.org/89/652489/4/check/networking-odl-grenade/7f3abed/ : FAILURE in 17m 08s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/89/652489/4/check/networking-odl-tempest-oxygen/ef41e41/ : RETRY_LIMIT in 10m 12s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/89/652489/4/check/networking-odl-tempest-fluorine/217f57f/ : RETRY_LIMIT in 7m 19s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/89/652489/4/check/networking-odl-tempest-oxygen-multinode/c8eb4ca/ : FAILURE in 50m 16s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/89/652489/4/check/networking-odl-tempest-fluorine-multinode/1ae2014/ : FAILURE in 50m 09s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/89/652489/4/check/networking-odl-rally-oxygen/eb2fe5b/ : SUCCESS in 1h 11m 43s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/89/652489/4/check/networking-odl-rally-fluorine/27784e0/ : SUCCESS in 1h 14m 13s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/89/652489/4/check/networking-odl-functional-oxygen/550d8a8/ : SUCCESS in 24m 14s\n- networking-odl-functional-fluorine http://logs.openstack.org/89/652489/4/check/networking-odl-functional-fluorine/2473f2f/ : SUCCESS in 21m 02s","accounts_in_message":[],"_revision_number":4},{"id":"d0af3adb41283a3f97774bd0ca7e669c06eb9d04","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-18 10:57:54.000000000","message":"Patch Set 4: Code-Review-1\n\nThanks gmann for uploading this patch.\nQuestion: There is another patch which as I see is basically the same, see [0].\n\nNote: Anyway I have a fix for failing tests, I rebased [0] on top of it, see [1].\n\n[0]: https://review.opendev.org/669420\n[1]: https://review.opendev.org/659767","accounts_in_message":[],"_revision_number":4},{"id":"5d2a4809bae1b937cb5ff13219b7003679b9cb91","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-18 11:30:06.000000000","message":"Patch Set 4:\n\n\u003e Thanks gmann for uploading this patch.\n \u003e Question: There is another patch which as I see is basically the\n \u003e same, see [0].\n \u003e \n \u003e Note: Anyway I have a fix for failing tests, I rebased [0] on top\n \u003e of it, see [1].\n \u003e \n \u003e [0]: https://review.opendev.org/669420\n \u003e [1]: https://review.opendev.org/659767\n\nSure, thanks for fix. I think other patch need more updates to remove py35 tox also. We mentioned in goal doc to use existing patch to add Train PTI but I am ok to abandon this.","accounts_in_message":[],"_revision_number":4},{"id":"ab6304a9a23e6bfc86dcf6aa2e828e9a68290d0c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-18 11:30:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6efac0baf3b088b4a85389abc10897eab0cd53ed","revisions":{"64524e7f66aee00cddf82a1e5c1c37a6fe1e8e42":{"kind":"REWORK","_number":1,"created":"2019-04-15 02:11:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/89/652489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/89/652489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/1"}}},"commit":{"parents":[{"commit":"5752d506bd42097aa68e8b21766606b0945b35b1","subject":"Merge \"Fix HOST_IP fetching method in local.conf.example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5752d506bd42097aa68e8b21766606b0945b35b1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 02:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 02:11:38.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Ibda74052c2975d95f343f9e1f6c1a341aa1bbeb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/64524e7f66aee00cddf82a1e5c1c37a6fe1e8e42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/64524e7f66aee00cddf82a1e5c1c37a6fe1e8e42"}]},"branch":"refs/heads/master"},"d4a95b5e0923954a46c1c3c3fc15041400f2f02e":{"kind":"REWORK","_number":2,"created":"2019-04-15 16:28:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/89/652489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/89/652489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/2"}}},"commit":{"parents":[{"commit":"5752d506bd42097aa68e8b21766606b0945b35b1","subject":"Merge \"Fix HOST_IP fetching method in local.conf.example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5752d506bd42097aa68e8b21766606b0945b35b1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 02:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 16:28:07.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: Ibda74052c2975d95f343f9e1f6c1a341aa1bbeb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d4a95b5e0923954a46c1c3c3fc15041400f2f02e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d4a95b5e0923954a46c1c3c3fc15041400f2f02e"}]},"branch":"refs/heads/master"},"a57f098f114ec95fe54ff8d37d28f894cdee801e":{"kind":"REWORK","_number":3,"created":"2019-07-18 02:54:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/89/652489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/89/652489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/3"}}},"commit":{"parents":[{"commit":"5752d506bd42097aa68e8b21766606b0945b35b1","subject":"Merge \"Fix HOST_IP fetching method in local.conf.example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5752d506bd42097aa68e8b21766606b0945b35b1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 02:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-07-18 02:54:04.000000000","tz":0},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nAs per below ML thread, we are good to drop the py35\ntesting also:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\nStory: #2005924\nTask: #34225\n\nChange-Id: Ibda74052c2975d95f343f9e1f6c1a341aa1bbeb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/a57f098f114ec95fe54ff8d37d28f894cdee801e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/a57f098f114ec95fe54ff8d37d28f894cdee801e"}]},"branch":"refs/heads/master"},"6efac0baf3b088b4a85389abc10897eab0cd53ed":{"kind":"REWORK","_number":4,"created":"2019-07-18 02:56:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/89/652489/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/89/652489/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/89/652489/4"}}},"commit":{"parents":[{"commit":"d9c32a9390532e33e9f1a4fca2c68e54e4a4d6a8","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d9c32a9390532e33e9f1a4fca2c68e54e4a4d6a8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-15 02:11:18.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-07-18 02:56:36.000000000","tz":0},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nAs per below ML thread, we are good to drop the py35\ntesting also:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\nStory: #2005924\nTask: #34225\n\nChange-Id: Ibda74052c2975d95f343f9e1f6c1a341aa1bbeb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6efac0baf3b088b4a85389abc10897eab0cd53ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/6efac0baf3b088b4a85389abc10897eab0cd53ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
