)]}'
{"id":"openstack%2Fnetworking-odl~655063","triplet_id":"openstack%2Fnetworking-odl~master~Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa","project":"openstack/networking-odl","branch":"master","topic":"opendev","hashtags":[],"change_id":"Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:39:55.000000000","updated":"2019-07-05 13:47:03.000000000","submitted":"2019-07-05 13:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655063-1562334423419-29425bd4","meta_rev_id":"d3b77e411b55933ea70908ea97574ed901b315b9","_number":655063,"virtual_id_number":655063,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2019-07-05 13:47:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2019-07-05 06:38:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"date":"2019-07-05 06:38:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-07-05 13:08:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 15:35:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-04-24 16:20:45.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-05-11 04:33:15.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-06-09 16:38:30.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-05 06:38:27.000000000","updated_by":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"reviewer":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"state":"REVIEWER"},{"updated":"2019-07-05 13:08:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-07-05 13:47:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"023369e87b8acb14df4083ddfa707067570a3eb6","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4217177b1ec8b4d06bdd68c0121806a359831b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 14:40:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/63/655063/1/check/test-release-openstack/6e2b6ff/ : SUCCESS in 2m 18s\n- openstack-tox-pep8 http://logs.openstack.org/63/655063/1/check/openstack-tox-pep8/a977d72/ : SUCCESS in 7m 03s\n- openstack-tox-py27 http://logs.openstack.org/63/655063/1/check/openstack-tox-py27/a569f5b/ : SUCCESS in 7m 07s\n- openstack-tox-py35 http://logs.openstack.org/63/655063/1/check/openstack-tox-py35/a846889/ : SUCCESS in 7m 14s\n- openstack-tox-py36 http://logs.openstack.org/63/655063/1/check/openstack-tox-py36/f2bfc2e/ : SUCCESS in 6m 04s\n- build-openstack-releasenotes http://logs.openstack.org/63/655063/1/check/build-openstack-releasenotes/cd98ab4/html/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/63/655063/1/check/openstack-tox-docs/b5f6e9b/html/ : SUCCESS in 5m 38s\n- openstack-tox-cover http://logs.openstack.org/63/655063/1/check/openstack-tox-cover/091da35/cover/ : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/655063/1/check/openstack-tox-lower-constraints/c1883d9/ : SUCCESS in 8m 54s\n- networking-odl-grenade http://logs.openstack.org/63/655063/1/check/networking-odl-grenade/85acfb8/ : POST_FAILURE in 52m 15s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/63/655063/1/check/networking-odl-tempest-oxygen/06d1785/ : POST_FAILURE in 22m 20s (non-voting)\n- networking-odl-tempest-fluorine finger://ze04.openstack.org/be34a3dfde1848d9ba8b4a31c0514c1d : RETRY_LIMIT in 52m 50s (non-voting)\n- networking-odl-tempest-oxygen-multinode finger://ze03.openstack.org/0e17abe5345046828eff9b89f144198c : RETRY_LIMIT in 38m 28s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/63/655063/1/check/networking-odl-tempest-fluorine-multinode/c8ba52e/ : RETRY_LIMIT in 48m 57s (non-voting)\n- networking-odl-rally-oxygen finger://ze05.openstack.org/1dc19fd547c1410792c9f67cca07e146 : RETRY_LIMIT in 13m 35s (non-voting)\n- networking-odl-rally-fluorine finger://ze07.openstack.org/9e0370f8bdd14b83830cae24ffe988f3 : RETRY_LIMIT in 40m 15s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/63/655063/1/check/networking-odl-functional-oxygen/1dcd53f/ : FAILURE in 16m 51s\n- networking-odl-functional-fluorine http://logs.openstack.org/63/655063/1/check/networking-odl-functional-fluorine/b749f81/ : POST_FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"780c822fd7140dc8e681025f180dcebc56bb4441","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-04-24 15:35:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(16 comments)\n\nOverall looks good, but some occurrences are still there.\n\nFurthermore, functional jobs fail on upper-constraint file URL (see comment inline)","accounts_in_message":[],"_revision_number":1},{"id":"4a043fe53c1e01667efcfe0f24b6e25fe63fdb4a","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-24 16:20:45.000000000","message":"Patch Set 1: Code-Review-1\n\nneed to fix the more","accounts_in_message":[],"_revision_number":1},{"id":"c3c5ba5197aa83dede111ae106df26292ee85b79","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-24 16:33:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af6f56c5085856abf559508c27a55643f2dc84b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 19:13:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/63/655063/2/check/test-release-openstack/2827a05/ : SUCCESS in 2m 32s\n- openstack-tox-pep8 http://logs.openstack.org/63/655063/2/check/openstack-tox-pep8/e190cd0/ : SUCCESS in 8m 12s\n- openstack-tox-py27 http://logs.openstack.org/63/655063/2/check/openstack-tox-py27/ce0e8e1/ : SUCCESS in 7m 02s\n- openstack-tox-py35 http://logs.openstack.org/63/655063/2/check/openstack-tox-py35/529b61c/ : SUCCESS in 9m 01s\n- openstack-tox-py36 http://logs.openstack.org/63/655063/2/check/openstack-tox-py36/109eadf/ : FAILURE in 6m 33s\n- build-openstack-releasenotes http://logs.openstack.org/63/655063/2/check/build-openstack-releasenotes/5c45a87/html/ : SUCCESS in 3m 01s\n- openstack-tox-docs http://logs.openstack.org/63/655063/2/check/openstack-tox-docs/75227ea/html/ : SUCCESS in 6m 23s\n- openstack-tox-cover http://logs.openstack.org/63/655063/2/check/openstack-tox-cover/2216457/cover/ : SUCCESS in 7m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/655063/2/check/openstack-tox-lower-constraints/6df3acc/ : SUCCESS in 7m 49s\n- networking-odl-grenade http://logs.openstack.org/63/655063/2/check/networking-odl-grenade/85d8ad7/ : FAILURE in 16m 38s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/63/655063/2/check/networking-odl-tempest-oxygen/62aaf82/ : POST_FAILURE in 23m 13s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/63/655063/2/check/networking-odl-tempest-fluorine/81253f6/ : TIMED_OUT in 2h 35m 50s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/63/655063/2/check/networking-odl-tempest-oxygen-multinode/63f0b41/ : FAILURE in 42m 33s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/63/655063/2/check/networking-odl-tempest-fluorine-multinode/e046282/ : FAILURE in 50m 45s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/63/655063/2/check/networking-odl-rally-oxygen/ae6177d/ : SUCCESS in 1h 13m 26s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/63/655063/2/check/networking-odl-rally-fluorine/9874011/ : SUCCESS in 1h 10m 59s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/63/655063/2/check/networking-odl-functional-oxygen/3e0ac0c/ : SUCCESS in 20m 33s\n- networking-odl-functional-fluorine http://logs.openstack.org/63/655063/2/check/networking-odl-functional-fluorine/9d4ec36/ : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e7c23a93b53929ccae3ecdae24134f428de6fd5e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-10 10:30:18.000000000","message":"Patch Set 2: Code-Review-1\n\nsmall nits, could you please include these as well:\nhttps://opendev.org/openstack/networking-odl/src/branch/master/devstack/devstackgaterc#L19\n\nhttps://opendev.org/openstack/networking-odl/src/branch/master/bindep.txt#L2","accounts_in_message":[],"_revision_number":2},{"id":"826b2a2769414fac65518023b43a7bf7232ca81b","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-11 04:32:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7fac8534d0e8de6b229ad6db647eae17c4049f34","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-11 04:33:15.000000000","message":"Patch Set 3:\n\n\u003e small nits, could you please include these as well:\n \u003e https://opendev.org/openstack/networking-odl/src/branch/master/devstack/devstackgaterc#L19\n \u003e \n \u003e https://opendev.org/openstack/networking-odl/src/branch/master/bindep.txt#L2\n\ndone, thanks for review","accounts_in_message":[],"_revision_number":3},{"id":"7d21fb944036dd12aa25d44e10533adba9302cfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 05:40:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/63/655063/3/check/test-release-openstack/c5dcb0b/ : SUCCESS in 1m 58s\n- openstack-tox-pep8 http://logs.openstack.org/63/655063/3/check/openstack-tox-pep8/065418c/ : FAILURE in 5m 56s\n- openstack-tox-py27 http://logs.openstack.org/63/655063/3/check/openstack-tox-py27/837cff5/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/63/655063/3/check/openstack-tox-py35/a543a16/ : SUCCESS in 7m 05s\n- openstack-tox-py36 http://logs.openstack.org/63/655063/3/check/openstack-tox-py36/2fd2f94/ : SUCCESS in 6m 25s\n- build-openstack-releasenotes http://logs.openstack.org/63/655063/3/check/build-openstack-releasenotes/78315cf/html/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/63/655063/3/check/openstack-tox-docs/a907641/html/ : SUCCESS in 4m 45s\n- openstack-tox-cover http://logs.openstack.org/63/655063/3/check/openstack-tox-cover/cd5910a/cover/ : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/655063/3/check/openstack-tox-lower-constraints/d08fda5/ : SUCCESS in 6m 37s\n- networking-odl-grenade http://logs.openstack.org/63/655063/3/check/networking-odl-grenade/f0d8d8c/ : FAILURE in 14m 09s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/63/655063/3/check/networking-odl-tempest-oxygen/034cc72/ : POST_FAILURE in 15m 40s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/63/655063/3/check/networking-odl-tempest-fluorine/6a6173a/ : FAILURE in 29m 01s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/63/655063/3/check/networking-odl-tempest-oxygen-multinode/3ad2b48/ : FAILURE in 46m 39s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/63/655063/3/check/networking-odl-tempest-fluorine-multinode/b104a22/ : FAILURE in 48m 13s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/63/655063/3/check/networking-odl-rally-oxygen/5dc540a/ : SUCCESS in 1h 06m 58s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/63/655063/3/check/networking-odl-rally-fluorine/4286993/ : SUCCESS in 1h 06m 56s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/63/655063/3/check/networking-odl-functional-oxygen/7b0dcb2/ : SUCCESS in 18m 51s\n- networking-odl-functional-fluorine http://logs.openstack.org/63/655063/3/check/networking-odl-functional-fluorine/2b648f3/ : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d02071d40aea963137df2b166e449ba3b26aa149","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-13 08:47:07.000000000","message":"Patch Set 3: Code-Review+1\n\nDue to bandit bug please rebase to this patch:\nhttps://review.opendev.org/658352\n\nHope that this one will go in before new bandit will be released :-)","accounts_in_message":[],"_revision_number":3},{"id":"9d8c733585d75fae0a8ea545d04e7de942ca1768","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-05-16 13:19:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c8fa1b625ed41be9ae58fcb15e7db64fb3aedea5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 14:39:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/63/655063/4/check/test-release-openstack/e854b1b/ : SUCCESS in 2m 00s\n- openstack-tox-pep8 http://logs.openstack.org/63/655063/4/check/openstack-tox-pep8/320809a/ : SUCCESS in 8m 35s\n- openstack-tox-py27 http://logs.openstack.org/63/655063/4/check/openstack-tox-py27/6b16b51/ : SUCCESS in 6m 27s\n- openstack-tox-py35 http://logs.openstack.org/63/655063/4/check/openstack-tox-py35/3023e4d/ : SUCCESS in 7m 05s\n- openstack-tox-py36 http://logs.openstack.org/63/655063/4/check/openstack-tox-py36/15f3dd5/ : SUCCESS in 5m 58s\n- build-openstack-releasenotes http://logs.openstack.org/63/655063/4/check/build-openstack-releasenotes/1926ee7/html/ : SUCCESS in 2m 59s\n- openstack-tox-docs http://logs.openstack.org/63/655063/4/check/openstack-tox-docs/990f0ff/html/ : SUCCESS in 5m 15s\n- openstack-tox-cover http://logs.openstack.org/63/655063/4/check/openstack-tox-cover/f908cb2/cover/ : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/655063/4/check/openstack-tox-lower-constraints/0ebe6e3/ : FAILURE in 8m 21s\n- networking-odl-grenade http://logs.openstack.org/63/655063/4/check/networking-odl-grenade/6a9379e/ : FAILURE in 19m 00s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/63/655063/4/check/networking-odl-tempest-oxygen/db37961/ : POST_FAILURE in 19m 58s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/63/655063/4/check/networking-odl-tempest-fluorine/eb0a0c7/ : FAILURE in 34m 52s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/63/655063/4/check/networking-odl-tempest-oxygen-multinode/df44b04/ : FAILURE in 45m 53s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/63/655063/4/check/networking-odl-tempest-fluorine-multinode/f8ac60d/ : FAILURE in 57m 41s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/63/655063/4/check/networking-odl-rally-oxygen/e95aa5a/ : SUCCESS in 1h 14m 31s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/63/655063/4/check/networking-odl-rally-fluorine/b5a5a58/ : SUCCESS in 1h 17m 05s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/63/655063/4/check/networking-odl-functional-oxygen/c44e17e/ : FAILURE in 20m 35s\n- networking-odl-functional-fluorine http://logs.openstack.org/63/655063/4/check/networking-odl-functional-fluorine/416436d/ : FAILURE in 23m 12s","accounts_in_message":[],"_revision_number":4},{"id":"597377724fad43fe0a219f25080712b5e159cefc","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-09 16:38:30.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"30ee6d54b312518ea465b4bc9bdcc474085c3d07","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-05 06:20:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a37b00ff87c90684ef488c4c9c11412463482f62","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-05 06:23:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aef81bc391bb6900977e1e3c390f906c159998cb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-05 06:35:17.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nThanks for taking care of this. Only one small regarding neutron_lbaas in tox.ini is why I give -1.\nI think this can be backported to stable branches as well, of course the neutron_lbaas there is still alive :-)","accounts_in_message":[],"_revision_number":6},{"id":"17dfaadb4915fc2d026971f2a8aed98367ce35aa","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-05 06:37:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"45e46944c116061d3ef4220273cbb211b29ee4d0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-05 06:38:18.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"624ce0039007763a44bdc6d3cb8f7d5fca566911","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-05 07:47:28.000000000","message":"Topic set to opendev","accounts_in_message":[],"_revision_number":7},{"id":"a32cb858617d68bd8d569f1d7ff0a0da48646461","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 08:28:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/63/655063/7/check/test-release-openstack/c9833a9/ : SUCCESS in 2m 22s\n- openstack-tox-pep8 http://logs.openstack.org/63/655063/7/check/openstack-tox-pep8/69ac61f/ : SUCCESS in 7m 46s\n- openstack-tox-py27 http://logs.openstack.org/63/655063/7/check/openstack-tox-py27/736eb3c/ : SUCCESS in 6m 35s\n- openstack-tox-py35 http://logs.openstack.org/63/655063/7/check/openstack-tox-py35/980d7d5/ : SUCCESS in 6m 57s\n- openstack-tox-py36 http://logs.openstack.org/63/655063/7/check/openstack-tox-py36/5bfac69/ : SUCCESS in 6m 43s\n- build-openstack-releasenotes http://logs.openstack.org/63/655063/7/check/build-openstack-releasenotes/26e8974/html/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/63/655063/7/check/openstack-tox-docs/f12d49d/html/ : SUCCESS in 4m 37s\n- openstack-tox-cover http://logs.openstack.org/63/655063/7/check/openstack-tox-cover/5c35ccf/cover/ : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/655063/7/check/openstack-tox-lower-constraints/3bd1d9b/ : SUCCESS in 6m 56s\n- networking-odl-grenade http://logs.openstack.org/63/655063/7/check/networking-odl-grenade/cbf62b4/ : FAILURE in 21m 08s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/63/655063/7/check/networking-odl-tempest-oxygen/edeb8a9/ : RETRY_LIMIT in 10m 08s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/63/655063/7/check/networking-odl-tempest-fluorine/7582d6c/ : RETRY_LIMIT in 10m 26s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/63/655063/7/check/networking-odl-tempest-oxygen-multinode/2f196d1/ : FAILURE in 1h 12m 50s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/63/655063/7/check/networking-odl-tempest-fluorine-multinode/0ca5dd6/ : FAILURE in 57m 02s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/63/655063/7/check/networking-odl-rally-oxygen/9311568/ : SUCCESS in 1h 15m 48s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/63/655063/7/check/networking-odl-rally-fluorine/26aaac9/ : SUCCESS in 1h 33m 32s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/63/655063/7/check/networking-odl-functional-oxygen/7b05658/ : SUCCESS in 22m 07s\n- networking-odl-functional-fluorine http://logs.openstack.org/63/655063/7/check/networking-odl-functional-fluorine/6a35c0d/ : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":7},{"id":"61eeafc4099f8006e6ffd91db854065de00b938c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-05 13:08:04.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6d12a6be1e8b11cf0a21c4a969791ba1476cac74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 13:08:16.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"23d0a1e063b1e254b10559a17127466c0e5e8d90","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-05 13:11:04.000000000","message":"Patch Set 7:\n\nAs this patch is only about url changes, I added wf+1","accounts_in_message":[],"_revision_number":7},{"id":"a951434927c51cce5ffe5c006b87e600eae6bee8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 13:47:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d3b77e411b55933ea70908ea97574ed901b315b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 13:47:03.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/63/655063/7/gate/test-release-openstack/5986c90/ : SUCCESS in 2m 27s\n- openstack-tox-pep8 http://logs.openstack.org/63/655063/7/gate/openstack-tox-pep8/1899e9e/ : SUCCESS in 7m 58s\n- openstack-tox-py27 http://logs.openstack.org/63/655063/7/gate/openstack-tox-py27/b8cdac0/ : SUCCESS in 6m 59s\n- openstack-tox-py35 http://logs.openstack.org/63/655063/7/gate/openstack-tox-py35/e90059a/ : SUCCESS in 6m 33s\n- openstack-tox-py36 http://logs.openstack.org/63/655063/7/gate/openstack-tox-py36/b7d8710/ : SUCCESS in 6m 07s\n- build-openstack-releasenotes http://logs.openstack.org/63/655063/7/gate/build-openstack-releasenotes/c250041/html/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/63/655063/7/gate/openstack-tox-docs/41fc836/html/ : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/63/655063/7/gate/openstack-tox-lower-constraints/e275329/ : SUCCESS in 8m 50s\n- networking-odl-functional-oxygen http://logs.openstack.org/63/655063/7/gate/networking-odl-functional-oxygen/925a6aa/ : SUCCESS in 20m 39s\n- networking-odl-functional-fluorine http://logs.openstack.org/63/655063/7/gate/networking-odl-functional-fluorine/f20707c/ : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d9c32a9390532e33e9f1a4fca2c68e54e4a4d6a8","revisions":{"c6c63555d519566713bbfab1751d6c415650453e":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:39:55.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/63/655063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/655063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/1"}}},"commit":{"parents":[{"commit":"73be89f774611ec37f1857dac4942e7a4af40676","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/73be89f774611ec37f1857dac4942e7a4af40676"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:36.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:53.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/c6c63555d519566713bbfab1751d6c415650453e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/c6c63555d519566713bbfab1751d6c415650453e"}]},"branch":"refs/heads/master"},"252e77301ad7ebc6956d1aae17faf24519f6c496":{"kind":"REWORK","_number":2,"created":"2019-04-24 16:33:24.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/63/655063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/655063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/2"}}},"commit":{"parents":[{"commit":"73be89f774611ec37f1857dac4942e7a4af40676","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/73be89f774611ec37f1857dac4942e7a4af40676"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:36.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-24 16:33:09.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/252e77301ad7ebc6956d1aae17faf24519f6c496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/252e77301ad7ebc6956d1aae17faf24519f6c496"}]},"branch":"refs/heads/master"},"29616a7a84980825137258f9c5a09abaffbf39ee":{"kind":"REWORK","_number":3,"created":"2019-05-11 04:32:01.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/63/655063/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/655063/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/3"}}},"commit":{"parents":[{"commit":"b1e38fdc14ad018304b8b1dcfdfc98bb67412062","subject":"stop using common db mixin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/b1e38fdc14ad018304b8b1dcfdfc98bb67412062"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:36.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-11 04:31:53.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/29616a7a84980825137258f9c5a09abaffbf39ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/29616a7a84980825137258f9c5a09abaffbf39ee"}]},"branch":"refs/heads/master"},"5f44231aa0f490ff9430d930b5d240dbe95788f3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-16 13:19:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/63/655063/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/655063/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/4"}}},"commit":{"parents":[{"commit":"fb042884671a849dba8b2a0bed2b45d32b71ddc9","subject":"Blacklist bandit 1.6.0 due to directory exclusion bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/fb042884671a849dba8b2a0bed2b45d32b71ddc9"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:36.000000000","tz":480},"committer":{"name":"Lajos Katona","email":"lajos.katona@ericsson.com","date":"2019-05-16 13:19:27.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5f44231aa0f490ff9430d930b5d240dbe95788f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/5f44231aa0f490ff9430d930b5d240dbe95788f3"}]},"branch":"refs/heads/master"},"ce42b1d3dd811855af5e7b302969935d93ca2308":{"kind":"REWORK","_number":5,"created":"2019-07-05 06:20:51.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/63/655063/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/655063/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/5"}}},"commit":{"parents":[{"commit":"a66aa842da4dc73804f79b9eb5ace6387e3a0d06","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/a66aa842da4dc73804f79b9eb5ace6387e3a0d06"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:36.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-07-05 06:20:42.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/ce42b1d3dd811855af5e7b302969935d93ca2308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/ce42b1d3dd811855af5e7b302969935d93ca2308"}]},"branch":"refs/heads/master"},"170ecc238d10a738a5cdbc8d387539002276bb2d":{"kind":"REWORK","_number":6,"created":"2019-07-05 06:23:37.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/63/655063/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/655063/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/6"}}},"commit":{"parents":[{"commit":"a66aa842da4dc73804f79b9eb5ace6387e3a0d06","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/a66aa842da4dc73804f79b9eb5ace6387e3a0d06"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:36.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-07-05 06:23:21.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/170ecc238d10a738a5cdbc8d387539002276bb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/170ecc238d10a738a5cdbc8d387539002276bb2d"}]},"branch":"refs/heads/master"},"d9c32a9390532e33e9f1a4fca2c68e54e4a4d6a8":{"kind":"REWORK","_number":7,"created":"2019-07-05 06:37:04.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/63/655063/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/63/655063/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/63/655063/7"}}},"commit":{"parents":[{"commit":"a66aa842da4dc73804f79b9eb5ace6387e3a0d06","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/a66aa842da4dc73804f79b9eb5ace6387e3a0d06"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:36.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-07-05 06:36:50.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ia40dbbd46d69efba5804f530a4027ff3cc9ffdaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d9c32a9390532e33e9f1a4fca2c68e54e4a4d6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/d9c32a9390532e33e9f1a4fca2c68e54e4a4d6a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
