)]}'
{"id":"openstack%2Fnetworking-odl~668660","triplet_id":"openstack%2Fnetworking-odl~master~I7dfc3adf52131211dd3b9e180c82370d39779541","project":"openstack/networking-odl","branch":"master","topic":"python3-first","hashtags":[],"change_id":"I7dfc3adf52131211dd3b9e180c82370d39779541","subject":"Convert filter_metadata_diff result to list","status":"MERGED","created":"2019-07-02 11:48:14.000000000","updated":"2019-08-19 21:51:41.000000000","submitted":"2019-08-19 21:51:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668660-1566251501101-28760403","meta_rev_id":"fb1ba824ece5b0e5b05ca3527daec92de8c08108","_number":668660,"virtual_id_number":668660,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-07-04 11:12:00.000000000","_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-08-19 21:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-08-13 09:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2019-08-19 21:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"date":"2019-07-11 11:47:52.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"date":"2019-07-12 06:11:56.000000000","_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"value":0,"date":"2019-07-11 11:47:56.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2019-08-19 21:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-11 11:49:16.000000000","_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 11:47:52.000000000","updated_by":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"reviewer":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"state":"REVIEWER"},{"updated":"2019-07-11 11:47:56.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-07-11 11:49:16.000000000","updated_by":{"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"},"reviewer":{"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"},"state":"REVIEWER"},{"updated":"2019-07-12 06:11:56.000000000","updated_by":{"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},"reviewer":{"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},"state":"REVIEWER"},{"updated":"2019-08-13 09:23:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-08-19 21:25:28.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-08-19 21:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"305d684879cf1a3714c72af41f6e4b4e7e040071","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-02 11:48:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44c5c99b8c07bcbbe84e40f85254dca5e952eb83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 13:21:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/60/668660/1/check/openstack-tox-pep8/a0c89b2/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/60/668660/1/check/openstack-tox-py27/2f5d09e/ : SUCCESS in 6m 10s\n- openstack-tox-py35 http://logs.openstack.org/60/668660/1/check/openstack-tox-py35/6b3bc57/ : SUCCESS in 7m 57s\n- openstack-tox-py36 http://logs.openstack.org/60/668660/1/check/openstack-tox-py36/3d76b61/ : SUCCESS in 6m 21s\n- openstack-tox-docs http://logs.openstack.org/60/668660/1/check/openstack-tox-docs/1504aa7/html/ : SUCCESS in 4m 45s\n- openstack-tox-cover http://logs.openstack.org/60/668660/1/check/openstack-tox-cover/40e8195/ : FAILURE in 6m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/668660/1/check/openstack-tox-lower-constraints/63ce144/ : SUCCESS in 7m 49s\n- networking-odl-grenade http://logs.openstack.org/60/668660/1/check/networking-odl-grenade/3c4c40a/ : FAILURE in 22m 28s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-oxygen/d79f198/ : RETRY_LIMIT in 8m 42s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-fluorine/0c82eef/ : RETRY_LIMIT in 11m 47s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-oxygen-multinode/618176f/ : FAILURE in 51m 03s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-fluorine-multinode/261e0e7/ : FAILURE in 48m 21s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-rally-oxygen/dd4dd17/ : SUCCESS in 1h 20m 30s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-rally-fluorine/c6dfeda/ : SUCCESS in 1h 08m 59s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-functional-oxygen/fffdfc9/ : SUCCESS in 17m 49s\n- networking-odl-functional-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-functional-fluorine/1907452/ : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a41b6263147d3b41a529afc6a9597a629b098393","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-03 11:09:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c0d80db1a34143d20a5f1a01e01bc8e08741fc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 12:38:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/60/668660/1/check/openstack-tox-pep8/babf580/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/60/668660/1/check/openstack-tox-py27/f86e8ca/ : SUCCESS in 6m 22s\n- openstack-tox-py35 http://logs.openstack.org/60/668660/1/check/openstack-tox-py35/22c87b3/ : SUCCESS in 9m 23s\n- openstack-tox-py36 http://logs.openstack.org/60/668660/1/check/openstack-tox-py36/0f73591/ : SUCCESS in 7m 20s\n- openstack-tox-docs http://logs.openstack.org/60/668660/1/check/openstack-tox-docs/2d710d2/html/ : SUCCESS in 5m 13s\n- openstack-tox-cover http://logs.openstack.org/60/668660/1/check/openstack-tox-cover/c0a85d2/ : FAILURE in 6m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/668660/1/check/openstack-tox-lower-constraints/60d387b/ : SUCCESS in 12m 21s\n- networking-odl-grenade http://logs.openstack.org/60/668660/1/check/networking-odl-grenade/30716b6/ : FAILURE in 25m 51s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-oxygen/3cc0bcc/ : RETRY_LIMIT in 10m 27s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-fluorine/b2b7854/ : RETRY_LIMIT in 10m 41s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-oxygen-multinode/b3c4e6f/ : FAILURE in 53m 26s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-fluorine-multinode/bed36b4/ : FAILURE in 47m 46s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-rally-oxygen/9518072/ : SUCCESS in 1h 21m 19s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-rally-fluorine/9d4b24f/ : SUCCESS in 1h 15m 40s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-functional-oxygen/e9caafe/ : SUCCESS in 23m 36s\n- networking-odl-functional-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-functional-fluorine/c00a248/ : SUCCESS in 26m 23s","accounts_in_message":[],"_revision_number":1},{"id":"740ffd5f3fd22c9d9a3999c6814546a4952d1741","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-07-04 11:12:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e49b398d0b6d347357010afe33e6cfd21e660ff3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:36:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/60/668660/1/check/openstack-tox-pep8/3182b43/ : SUCCESS in 6m 39s\n- openstack-tox-py27 http://logs.openstack.org/60/668660/1/check/openstack-tox-py27/47561e9/ : SUCCESS in 6m 43s\n- openstack-tox-py35 http://logs.openstack.org/60/668660/1/check/openstack-tox-py35/c2260a8/ : SUCCESS in 7m 01s\n- openstack-tox-py36 http://logs.openstack.org/60/668660/1/check/openstack-tox-py36/d7799f4/ : SUCCESS in 6m 52s\n- openstack-tox-docs http://logs.openstack.org/60/668660/1/check/openstack-tox-docs/89f9d27/html/ : SUCCESS in 5m 04s\n- openstack-tox-cover http://logs.openstack.org/60/668660/1/check/openstack-tox-cover/13194a7/cover/ : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/60/668660/1/check/openstack-tox-lower-constraints/b1d4674/ : SUCCESS in 7m 01s\n- networking-odl-grenade http://logs.openstack.org/60/668660/1/check/networking-odl-grenade/3292cfa/ : FAILURE in 26m 35s (non-voting)\n- networking-odl-tempest-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-oxygen/552e33d/ : RETRY_LIMIT in 10m 12s (non-voting)\n- networking-odl-tempest-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-fluorine/d72cd5c/ : RETRY_LIMIT in 10m 13s (non-voting)\n- networking-odl-tempest-oxygen-multinode http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-oxygen-multinode/7d1382d/ : FAILURE in 51m 45s (non-voting)\n- networking-odl-tempest-fluorine-multinode http://logs.openstack.org/60/668660/1/check/networking-odl-tempest-fluorine-multinode/d7483b1/ : FAILURE in 49m 11s (non-voting)\n- networking-odl-rally-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-rally-oxygen/7d8ccc5/ : SUCCESS in 1h 16m 25s (non-voting)\n- networking-odl-rally-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-rally-fluorine/185f876/ : SUCCESS in 1h 12m 25s (non-voting)\n- networking-odl-functional-oxygen http://logs.openstack.org/60/668660/1/check/networking-odl-functional-oxygen/08b9073/ : SUCCESS in 20m 56s\n- networking-odl-functional-fluorine http://logs.openstack.org/60/668660/1/check/networking-odl-functional-fluorine/ae6436e/ : SUCCESS in 19m 50s","accounts_in_message":[],"_revision_number":1},{"id":"437e2caca08a5449f711a373efe01a1c5ee4057e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-13 09:23:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"569375447a894ee12af8e6d20f9f0ceb6d2b8e75","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-19 21:25:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"799673947082ec85aebcc25e65f60d78d7c145db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 21:25:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f57451e4a2943071303acf937408c471f33dfaba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 21:51:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012b89a6aa734d1b9ba6a0bb58b52d67 : SUCCESS in 8m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e88f1040023433592dea0ee4b2b7406 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c8c2a113dac423785363576b5224d65 : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8119a5c11a4b4780a769957c695d8306 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/301c503eff8e43df82bd2ced7fb32986 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3146c13a5f124230953afe24e5db9399 : SUCCESS in 9m 06s\n- networking-odl-functional-oxygen https://zuul.opendev.org/t/openstack/build/620884fe2b7f46e3b99d6034ece5f361 : SUCCESS in 19m 59s\n- networking-odl-functional-fluorine https://zuul.opendev.org/t/openstack/build/5226458a1ec94eedb563a84866d1fc5b : SUCCESS in 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"411cf21ef973a7effa82ad9ca14643d26496046a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 21:51:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c03dc0dd6ec9165a06bf1592d62d5f68dfe1b799","revisions":{"c03dc0dd6ec9165a06bf1592d62d5f68dfe1b799":{"kind":"REWORK","_number":1,"created":"2019-07-02 11:48:14.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/60/668660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/60/668660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/60/668660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/60/668660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/60/668660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/60/668660/1"}}},"commit":{"parents":[{"commit":"2db2cab66c27019ab8e11cad951a172d40eecd1b","subject":"Blacklist bandit, bump neutron-lib and retire neutron-lbaas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/2db2cab66c27019ab8e11cad951a172d40eecd1b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-07-02 11:37:40.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-07-02 11:47:08.000000000","tz":120},"subject":"Convert filter_metadata_diff result to list","message":"Convert filter_metadata_diff result to list\n\nfilter_metadata_diff method gives a way to whitelist/blacklist\ndifferences between migration scripts and db models, for the\ntest_models_sync test.\nThe test fails on python3 because filter builtin on python3 returns\niterable instead of a list.\nFix this problem by returning always a list.\n\nChange-Id: I7dfc3adf52131211dd3b9e180c82370d39779541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/c03dc0dd6ec9165a06bf1592d62d5f68dfe1b799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/c03dc0dd6ec9165a06bf1592d62d5f68dfe1b799"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
