)]}'
{"id":"openstack%2Fnetworking-odl~683152","triplet_id":"openstack%2Fnetworking-odl~master~I8585f8429acb7359e1f372a5e4f378ca5ce6b540","project":"openstack/networking-odl","branch":"master","topic":"use-latest-odl_and_python3","hashtags":[],"change_id":"I8585f8429acb7359e1f372a5e4f378ca5ce6b540","subject":"Change function.func_doc to function.__doc__","status":"MERGED","created":"2019-09-19 13:18:44.000000000","updated":"2019-11-20 14:39:39.000000000","submitted":"2019-11-20 14:38:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683152-1574260681005-64a7fc44","meta_rev_id":"023c047d41fd82bfe9a7deee2db3caf0b7416682","_number":683152,"virtual_id_number":683152,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-20 14:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"value":1,"date":"2019-11-18 14:15:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-11-18 14:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-11-19 16:07:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 11:41:56.000000000","updated_by":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"reviewer":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"state":"REVIEWER"},{"updated":"2019-09-24 13:06:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-09-24 13:21:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-09-24 14:00:18.000000000","updated_by":{"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},"reviewer":{"_account_id":9361,"name":"Vivekanandan Narasimhan","email":"n.vivekanandan@ericsson.com","username":"viveknarasimhan"},"state":"REVIEWER"},{"updated":"2019-09-24 14:00:24.000000000","updated_by":{"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"},"reviewer":{"_account_id":25550,"name":"Achuth","email":"achuth.maniyedath@gmail.com","username":"achuthm"},"state":"REVIEWER"},{"updated":"2019-11-20 14:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"397ce97976f883980365a03606416f3c6676b945","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-09-19 13:18:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c40e2b9b2dd3bcbba3390bee822f11034221b677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 17:24:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331326fd2a484a6ba2af231c69427816 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48f65fa6358447a499c6ac6ccdc93516 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dd9ab2ab6314496bea2cc939cdef2e0 : SUCCESS in 8m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/393cc4d3a33f44e082adb07ec54c26ee : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e8d5b739dc41a19770e8a35b2f1f08 : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd91da70f8db4eb78b70047de3eaa623 : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/014b9d0ba3e448ae9cd7a4b8d6f6225b : SUCCESS in 10m 14s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/47540c6518bb4915bf76e4e6578757f3 : FAILURE in 26m 02s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/11a8c1651c2d4b358d172e94aee5802f : RETRY_LIMIT in 10m 49s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/4c94a0098bc1448eb414fc3475ce0f18 : RETRY_LIMIT in 11m 03s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/4394a0cb86e04607987ae39226b9f992 : FAILURE in 50m 00s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/63839b7967194d9aaf4998860b3d20df : FAILURE in 52m 45s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/e3293bdf10be49d7867547af45f50e3e : SUCCESS in 1h 51m 43s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/4d44cdba057b4b7082cde64ceb98f9d5 : SUCCESS in 1h 32m 19s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/37680b2a2cc34e8283ff0a64608663a9 : SUCCESS in 20m 12s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/80c2bb4c6a0d4c578ce68e86ede5c485 : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0600cd925e57508b383380448f8004728da129fe","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-09-24 13:06:28.000000000","message":"Patch Set 1: Code-Review+1\n\nIn python2 both .func_doc and .__doc__ work, but in python3 only the latter. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"efd92496bae3c7943a9a0c8a47b0262e1b83d595","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-24 13:21:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8220164ee5c10b62014428810394517a99815c2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-13 11:21:21.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f9a0fdcb7d1e1979b763806f557cec5ae42ca5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 11:21:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfe78a0c69e4780e618e197b99cf0a3e3f31b697","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 11:52:17.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"12741bd055f64fcfc951ec58d14f450aa275ae4c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-18 14:15:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1f79c6d3e0167b4c87e2b01e48710d1019b9fe38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 16:15:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe26f07c9eb4cc7b149611a0562a991 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d810075ec6c498b876884765d4b6de2 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1c0ddb44247403f9e7d1cf94d66deb8 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eef1fcd82b814257b334d4bbd9d8a9c2 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb796c4fcfd743fc88fb0cccc7ba474c : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49b8ce30a0c84073a00315de16f09487 : SUCCESS in 7m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49a48033a127462d86f8a86d1396c2fe : SUCCESS in 8m 07s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/5ac240f979a94264ab8f285d7ad6ca41 : FAILURE in 19m 45s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/f43c7d43a7c4485fa0fdba66c21fb2a6 : RETRY_LIMIT in 6m 10s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/ef3c84fa79e345eb9fef409ccf875f76 : RETRY_LIMIT in 9m 44s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/23eed288c3f34336b7d32c4e1b7ba22a : FAILURE in 53m 37s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/1a554541928d4f67ba06b0547366f946 : FAILURE in 25m 38s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/62dde7cd92294cc08c9efd15dc65c785 : SUCCESS in 1h 34m 45s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/00b5b78fb3dd4687bfe7d16a3df90a19 : SUCCESS in 1h 51m 35s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/2a60cac3ea1247a8afaba4b7b8f38d97 : SUCCESS in 23m 00s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/a124c1b8a06142d4badb4c23b6119640 : FAILURE in 18m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4e3278c1093c44ba3882921bde9323b9d695ddf3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-19 08:13:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"77d003611b94e89b37795075811e3d1e6368cde7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 09:06:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf4df544f15249e28d664e753c119510 : SUCCESS in 6m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/541be7634c494431aa75af4bccfefaf1 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70d0fda34ccf48769e5062c7bb28ae7d : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a60dbea5ffe48a28558471ece9cbae4 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b43f2ba4e70497589931fe42ea81bbc : SUCCESS in 7m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a99fb6045f048c7925dfe084c684512 : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a4b5a994dbf47ff910a9dc981625d28 : SUCCESS in 8m 15s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/f9ec7b429360417db5a47c33c8443f71 : FAILURE in 23m 38s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/ee96ed984b984816838204e415a29dd5 : RETRY_LIMIT in 6m 44s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/a4b60e2e087d4f95af08e8757199a830 : RETRY_LIMIT in 7m 34s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/488396f23cb24db0acdc0f8964ab8389 : FAILURE in 24m 39s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/891401aa185e46799690e59873f8f1d6 : FAILURE in 20m 35s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/60fc0c8107914497bdf4d62c0c0ff0e6 : FAILURE in 17m 32s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/70bd86d7757142a0a36e1a36b5436675 : FAILURE in 15m 34s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/7584a2bcd8d44988af6a757b5aa49206 : SUCCESS in 17m 52s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/3341f4f58c144749a28c4b0fbeaa7e56 : SUCCESS in 25m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2ea9e117c678b2be757703ac114d4095142a216a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-19 16:07:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b80a76a8cb96376dd69936ad8586dd05319d0dc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 16:07:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ee1fe6ed20207797e62c98179d98f5639589119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 16:33:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9a06acd15ce49e3b83711af6576304f : SUCCESS in 7m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8417df84b88b4b81b6567e3edef9da2b : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8625b6aaf12842c89249845c8323b8f9 : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6728f29ef992476198c8f735c626c1d1 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44e104563c74890af8fb238d6f1f930 : SUCCESS in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37f60972295d4c69902f563664625f72 : SUCCESS in 10m 06s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/b15c87e9b186401d9fdbe3395f033234 : FAILURE in 23m 18s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/18c390e1170248ca99774d4e45056194 : SUCCESS in 23m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9c327f6961de60088594890cd3eb6804311c0d28","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-20 12:18:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cda2bde5219e624195de283415bc2c5c07b187c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 14:11:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1625a4e38823430bb0e89e8dd0890e1c : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d7192dfdf0e4f42bef49b94db0a9753 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b369e2e08824c328ca9b4ec8e9ea421 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcfe949fc3ea480187169c66b640bff2 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74cc5806d924640baa902f7aeebe2e2 : SUCCESS in 7m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cea276c37a946cdad93341eb23f1be9 : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d6de773e5f344a78f514ca99ae88160 : SUCCESS in 9m 56s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/b6d5d619e2ee471cbd20fbb6ab913dcb : FAILURE in 21m 37s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/67372411785747fa8f4f44a747521dff : RETRY_LIMIT in 6m 07s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/eb47fbb0c17e4a64b6e1c0412c648c95 : RETRY_LIMIT in 10m 21s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/c5b1517e7cc04754b4c8bcbd9a3c13e9 : FAILURE in 54m 00s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/7b7f6e8195944a5eaa5d2e519ec29a5b : FAILURE in 23m 10s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/3bdedf5d5a6c43999c07da67d8c756e5 : SUCCESS in 1h 42m 58s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/9944acfa6820497fb587e9cb0b204018 : SUCCESS in 1h 40m 09s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/17b967f4cbb740089587e1547bebfbf6 : SUCCESS in 22m 28s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/0aa53411d25245f39abea99b81d72a9a : SUCCESS in 22m 15s","accounts_in_message":[],"_revision_number":2},{"id":"399b76cad1b2875342e80aa082117f8e419ae9eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 14:11:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4b1347980ed5a2e534114ea4e68e69b9b23215d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 14:38:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31a354045a38459bbe14a450454602c6 : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/857bc2eb86fb45d6a02da52af3d875a3 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd0fb5d8477a4b0cabe8e2c15dceab48 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f5fb1e44cfc40a0aa7f5b36c4dc94b5 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc8f5143a24c4974a1fd891648b4005e : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0d3caab5ce642a2b6b7aa71c786f42a : SUCCESS in 8m 39s\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/010bce75631c493f9c467f60105651a7 : SUCCESS in 19m 39s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/594bc9fcd149480885477ecbd61d34c6 : SUCCESS in 20m 01s","accounts_in_message":[],"_revision_number":2},{"id":"582d5d10fa33332c46508c0df3af587ef86af423","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 14:38:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"36704bc7ce987f1c2bccec6cd44e7db3a889bfa1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 14:39:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f26dfcfcbc1e40bab8257e8fafd8d662 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"924141c66354d4b165cc02402fa5a6b092a97c4e","revisions":{"26ca2489f70727f900a3effcfca77bd095c2745f":{"kind":"REWORK","_number":1,"created":"2019-09-19 13:18:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/683152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/52/683152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/52/683152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/52/683152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/52/683152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/52/683152/1"}}},"commit":{"parents":[{"commit":"f8336170a0c5c51c80c5706f17df1965b182260a","subject":"Bump pylint version to one that supports python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/f8336170a0c5c51c80c5706f17df1965b182260a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-09-19 13:08:50.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-09-19 13:08:50.000000000","tz":120},"subject":"Change function.func_doc to function.__doc__","message":"Change function.func_doc to function.__doc__\n\nIn python3 func_doc was renamed to __doc__, see [0].\n\n[0]: https://docs.python.org/3.0/whatsnew/3.0.html#operators-and-special-methods\nCloses-Bug: #1844660\n\nChange-Id: I8585f8429acb7359e1f372a5e4f378ca5ce6b540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/26ca2489f70727f900a3effcfca77bd095c2745f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/26ca2489f70727f900a3effcfca77bd095c2745f"}]},"branch":"refs/heads/master"},"924141c66354d4b165cc02402fa5a6b092a97c4e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-18 14:15:55.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/683152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/52/683152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/52/683152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/52/683152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/52/683152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/52/683152/2"}}},"commit":{"parents":[{"commit":"432dd3d4ea0ddfbf24da240a2b09fdd05ec4847f","subject":"Merge \"Bump pylint version to one that supports python3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/432dd3d4ea0ddfbf24da240a2b09fdd05ec4847f"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2019-09-19 13:08:50.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2019-11-18 14:15:55.000000000","tz":0},"subject":"Change function.func_doc to function.__doc__","message":"Change function.func_doc to function.__doc__\n\nIn python3 func_doc was renamed to __doc__, see [0].\n\n[0]: https://docs.python.org/3.0/whatsnew/3.0.html#operators-and-special-methods\nCloses-Bug: #1844660\n\nChange-Id: I8585f8429acb7359e1f372a5e4f378ca5ce6b540\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/924141c66354d4b165cc02402fa5a6b092a97c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/924141c66354d4b165cc02402fa5a6b092a97c4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
