)]}'
{"id":"openstack%2Fnetworking-odl~699449","triplet_id":"openstack%2Fnetworking-odl~master~I560aa5db3053a2b19819d64fc514f4eff8928603","project":"openstack/networking-odl","branch":"master","hashtags":[],"change_id":"I560aa5db3053a2b19819d64fc514f4eff8928603","subject":"DNM: test CI","status":"ABANDONED","created":"2019-12-17 16:23:19.000000000","updated":"2020-07-08 14:08:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eb103e270c2787971ed219bf37d2346a1d640017","_number":699449,"virtual_id_number":699449,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-17 18:07:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 18:07:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26f068a18b057f90da2c3689f49107b0b4340bb5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-17 16:23:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9bf3564e876bc52de6c0df7f3b3d3917983f19a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-17 16:25:43.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit deed05df5edc8ed10def9175a187be9b358b6360","accounts_in_message":[],"_revision_number":1},{"id":"c5fb3718c399d286aa72892dd9255504a272dce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 18:07:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfb702502cd24b999518d67c1b50ba4c : SUCCESS in 9m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9ba6a074dd94fafb8d94c9c067930fd : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ab86f1a7705466f8e472e3a12130617 : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c613e9e7ef146ef80c4d0b09ad0034b : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78dcb00d61dd4026abe4f3355f9edd77 : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f166d23e9e65433d8a5069b6fd4d4837 : SUCCESS in 9m 47s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/ec8ca69f42f246bba916ba9086e560d6 : FAILURE in 28m 19s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/b0ebc98c93cb4acbb9495c62db300275 : RETRY_LIMIT in 8m 05s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/841837c1b3094c82af5af7710cff5e7b : RETRY_LIMIT in 7m 46s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/baa0fd9e6fb44bba99c5c501ce82c2e1 : FAILURE in 17m 21s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/32d9a24b89224dd2aa45456dc7c1f407 : FAILURE in 17m 51s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/5caf282dfb3d4a0c8f79e960c79b7e82 : SUCCESS in 1h 29m 37s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/bc6f7c93e31a4e0fba5997aabb374edc : SUCCESS in 1h 36m 32s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/992c68bd691f45b993c692f028065ea2 : SUCCESS in 23m 38s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/d3f268f783bd4ecca3434104c9219003 : SUCCESS in 24m 13s","accounts_in_message":[],"_revision_number":1},{"id":"eb103e270c2787971ed219bf37d2346a1d640017","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-08 14:08:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30d0de73898d23e8cf265e95869f2693d5f6386c","revisions":{"30d0de73898d23e8cf265e95869f2693d5f6386c":{"kind":"REWORK","_number":1,"created":"2019-12-17 16:23:19.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/49/699449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/49/699449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/49/699449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/49/699449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/49/699449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/49/699449/1"}}},"commit":{"parents":[{"commit":"e765aa81808eec272cd2bec17c79fe9473d25a62","subject":"Merge \"Remove networking-odl python2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/e765aa81808eec272cd2bec17c79fe9473d25a62"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-17 16:21:51.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-17 16:21:51.000000000","tz":60},"subject":"DNM: test CI","message":"DNM: test CI\n\nChange-Id: I560aa5db3053a2b19819d64fc514f4eff8928603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/30d0de73898d23e8cf265e95869f2693d5f6386c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/30d0de73898d23e8cf265e95869f2693d5f6386c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
