)]}'
{"id":"openstack%2Fnetworking-odl~699451","triplet_id":"openstack%2Fnetworking-odl~stable%2Ftrain~I560aa5db3053a2b19819d64fc514f4eff8928603","project":"openstack/networking-odl","branch":"stable/train","hashtags":[],"change_id":"I560aa5db3053a2b19819d64fc514f4eff8928603","subject":"DNM: test CI","status":"ABANDONED","created":"2019-12-17 16:25:43.000000000","updated":"2020-09-25 15:32:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"061dd5c28ca5dd7320255910fba36d92fb0b4964","_number":699451,"virtual_id_number":699451,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-10 13:31:15.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"date":"2020-09-25 15:32:56.000000000","_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-13 12:56:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 12:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 15:32:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"38946f7be6e9b259b30acf17d02b47f1d76f227b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-17 16:25:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f877e9187934eb2fb670cef0e60d9dec9b0c5b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 18:29:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c7259f1cfd42b997567c9a10e6210d : FAILURE in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba7ca47f916d41c0bb8dc5bfacac44c1 : FAILURE in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/532e9ca15e2c4b2a9c045db0ab6d39eb : FAILURE in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/873533e8efd84cd6889877ed025e5518 : FAILURE in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee35726493b421eb6a4abb835ffcf2d : SUCCESS in 8m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a3232a6fad142c896d51aefa7c1a242 : FAILURE in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cbf3ceccd654c56b9fcf3837692c352 : SUCCESS in 10m 24s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/d816111ec0c745da923905c57ed39376 : FAILURE in 18m 15s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/1febb911e724447789d59ff92818c50c : RETRY_LIMIT in 13m 53s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/0ca72d2797ba469caba6a05db7e240bc : RETRY_LIMIT in 9m 02s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/3e6475277d3b47bd887c1abd0f5ae424 : FAILURE in 1h 00m 56s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/76b7da3023c44193b41818581e79d4f7 : FAILURE in 28m 46s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/100e5767b12041f09e458c014aafc050 : SUCCESS in 1h 45m 25s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/345cbaead15d4165a5442685567b7d7f : SUCCESS in 1h 26m 25s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/7e3e0bf000ce4c2384957a2618b65427 : FAILURE in 16m 04s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/fe597051e8b74d34aea4cbe22e4a10db : FAILURE in 24m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7789af6b36dafaa7e96acad250f8042628f4e423","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-17 20:19:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0c4c60b00619d23523f2f33069305321bc779598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 21:10:23.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b23639310ceaa79207a9773cc40cca8c977d5562","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-07 13:11:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"26ef6214e129c054ae710c997a18585d49eed520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 14:37:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e70f4dc7494abc8717906a0fdf993f : SUCCESS in 8m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed8ebd704ce94cc2833059df702cfa15 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa837057832f46ac921f0958ea18b3b2 : SUCCESS in 9m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2b22f23c24e42ac9c12359ab52d6983 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0b2f7db53141bea5aed046abd9b1c9 : SUCCESS in 7m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0934264e4203418e9cef2670e7ad18cf : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/860849df3b77458e9e82ad0deac9027f : SUCCESS in 7m 34s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/e5f63bb7e0ba4bd3bceff72248c331a6 : FAILURE in 29m 42s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/2028f0fa5249458292a52e46ba060222 : RETRY_LIMIT in 7m 23s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/f6e6436dbcaf40a5a33a3a8f40638e05 : RETRY_LIMIT in 11m 14s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/f5e84e52d23148baa0d4f3b497adf2f3 : FAILURE in 53m 12s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/c957c5f737ee4f7d810540c3c1c7ee14 : FAILURE in 40m 01s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/a7f7216a036744ed8cde29ae84b9e378 : SUCCESS in 1h 22m 44s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/e36b4bc2b81142cd9581a63d4209ee38 : SUCCESS in 1h 24m 55s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/c1629319cf664e539962c67b2567322e : SUCCESS in 18m 05s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/556a2dcc7f314c56a9ee30720808b5d6 : SUCCESS in 17m 57s","accounts_in_message":[],"_revision_number":2},{"id":"753b4b019b8de6c0ae0b58f44cdc7281bbdc8d72","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-13 11:22:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"64b009d5599e805cc9f2fafefb9c1dc3055c9dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 12:56:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da534d2ad51c46f0832bc5c4c5484cd5 : SUCCESS in 8m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dfde493b2f5841ac814530b40b4f9b8b : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b0ff3b6e7cb4063a045ec4251338f85 : FAILURE in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4652dccfcfa74e668bb3844d07bee212 : FAILURE in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9467ce649c1548338d0b6a798e6dce7e : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a2087af28be41f9aa65c235bdf204ac : SUCCESS in 7m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fc0be9b0269401cb13acf28f7036074 : SUCCESS in 8m 58s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/abd7322a8f1c4e24884da18a8603b603 : FAILURE in 20m 06s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/db6999ce43e8413689a9666097ef5c52 : RETRY_LIMIT in 1m 08s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/4bee1b7502974317a6619544e14498af : RETRY_LIMIT in 6m 33s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/679d487e3c464306ab877d1b64887405 : FAILURE in 55m 20s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/115312cead7e4291a0ad0fdc32892081 : FAILURE in 23m 55s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/995f1fb04b684bdc838871bd845b066f : FAILURE in 18m 26s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/f7224b97b8964a55aed23d0db1b469ac : SUCCESS in 1h 28m 27s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/fdfdd8b53f37461a9ae05e2ba1ca5985 : RETRY_LIMIT in 12m 15s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/4ce5c846cb84402f80ef1052a1f672f6 : SUCCESS in 23m 09s","accounts_in_message":[],"_revision_number":2},{"id":"fb48fe609830b28058c8184ab5d0f20ab4bd34af","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-03 10:30:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2337594f1ca1e1dd062b9ec1148796d05b1be359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 11:55:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fcdf0f601b44d40a09d9d335cc376ff : FAILURE in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d78a30cb0944d469e5573b9f9319c95 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ac216c090d4efea8465afe73293aa4 : FAILURE in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/311adc69ee43490cb60996910e6fd92e : FAILURE in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b268551d22cc499faf10269a7a4a84fc : SUCCESS in 9m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a419a4cd93ca4911a58051a069513de7 : FAILURE in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/450bd2a42a3b44dfa9a9397ce1c339f6 : FAILURE in 9m 18s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/08658147de4c4f379bb8d201d789b557 : FAILURE in 14m 38s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/e733fa87c629441fa0f244c21d2f74ce : RETRY_LIMIT in 5m 40s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/247f67daf4b141238e385463c4be128b : RETRY_LIMIT in 12m 21s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/05f424d245af4bf286298660fa929756 : FAILURE in 27m 51s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/0d7e3503849646e49f92ba765308f1a5 : FAILURE in 21m 10s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/456a0e72d17e4c20a3301f0d6092a03f : FAILURE in 16m 36s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/3afea2a9ab664f53978d4cff2894ecf1 : FAILURE in 23m 14s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/92bacd691ddd47cf8ea69de48c18c98e : RETRY_LIMIT in 10m 44s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/bb8d001b58954550ae22a7e15d7162f9 : RETRY_LIMIT in 14m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ef82d6b1651c71133b9e12406cfebca2f0fca986","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-04 10:18:22.000000000","message":"Patch Set 2:\n\nrecheck - pip 20.2.1 released, let\u0027s see whether it helps with the \"reinstalling packages in phase 2 without using constraints\"","accounts_in_message":[],"_revision_number":2},{"id":"b4b1b8c4b5e1567556589363ac762243804e6e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 11:10:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e1b0b2a1089441cadc3fcbd3b5e4df1 : FAILURE in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12d53d9c1564410f86c4d4f9c407635c : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f095aef44f44f749527a565a7ec7a95 : FAILURE in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27688079af4a4413bb90ecc81882fa92 : FAILURE in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97dd3299ea01466bb552e5e7d9930238 : SUCCESS in 8m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b383734c706840b891b513b3bb8bed0b : FAILURE in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6052a2f421314c2fa489dd148310f63f : FAILURE in 8m 41s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/aa1236310fc54398836c4f2a5a1658ed : FAILURE in 23m 50s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/538ad16510a240b2bf42db1b5bc43c56 : RETRY_LIMIT in 6m 28s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/2b208f3e314d4aa09c9a34eaf019cc08 : RETRY_LIMIT in 7m 02s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/fc23a46d9b9b4bf9b32d2d4fe91d98b1 : FAILURE in 28m 50s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/45121e1be6cf44068e23dcf7873dbc61 : FAILURE in 23m 33s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/684769063d7c472fbd1a0775e6ab491f : FAILURE in 22m 38s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/3f5a2c0a92534e5188ab95ce1d89c82d : FAILURE in 22m 32s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/fd61450fded745d082e8bff019bc0ec5 : RETRY_LIMIT in 10m 47s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/dc6c292f955c49a980713412e39faa7d : RETRY_LIMIT in 10m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4065c44fb1c2e41e9298b9575e23a1cbf7503de7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-10 13:31:15.000000000","message":"Patch Set 2:\n\nrecheck - virtualenv was released with the new pip one week ago, so let\u0027s see again.","accounts_in_message":[],"_revision_number":2},{"id":"89d9bc5397955d28fd06a7e809343b89214306fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 14:56:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713f705845ac4359a2ada29d16e2cefa : FAILURE in 11m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37322b16d8e94dea8a1de2cc36564356 : SUCCESS in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43dc006fa1b442eaa4f4f8eb33f17188 : SUCCESS in 10m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/724305a3ffe54583989f36d959e526c8 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a188b8292c4101852384cc4ea22bd8 : SUCCESS in 13m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/356e2c5ef9d7439cac682dd2310c2cc3 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/badf9b1bfd00424e8d605a87c487a00b : SUCCESS in 11m 00s\n- networking-odl-grenade https://zuul.opendev.org/t/openstack/build/3126336687724edc8647955a7bc5c0dc : FAILURE in 26m 15s (non-voting)\n- networking-odl-tempest-oxygen https://zuul.opendev.org/t/openstack/build/24313a2008564777aaa2623747fd2ba3 : RETRY_LIMIT in 16m 48s (non-voting)\n- networking-odl-tempest-fluorine https://zuul.opendev.org/t/openstack/build/841f1114e5a941c0b7c2fbe552db925e : RETRY_LIMIT in 14m 00s (non-voting)\n- networking-odl-tempest-oxygen-multinode https://zuul.opendev.org/t/openstack/build/7f4998e14fed4785916c1fae7aed974a : FAILURE in 28m 02s (non-voting)\n- networking-odl-tempest-fluorine-multinode https://zuul.opendev.org/t/openstack/build/5ec2ae38f0144b1eb9433abae759de34 : FAILURE in 28m 25s (non-voting)\n- networking-odl-rally-neon https://zuul.opendev.org/t/openstack/build/a9b1eb1b69eb40199d8d3c745a7adda3 : FAILURE in 22m 45s (non-voting)\n- networking-odl-rally-sodium https://zuul.opendev.org/t/openstack/build/93976eaf4df746f187e7cc9be06cc976 : FAILURE in 17m 47s (non-voting)\n- networking-odl-functional-neon https://zuul.opendev.org/t/openstack/build/516fe74cfd8d4eb1babc8fd70d146ffb : RETRY_LIMIT in 13m 37s\n- networking-odl-functional-sodium https://zuul.opendev.org/t/openstack/build/87165f605a29429ba0d62256a089537f : RETRY_LIMIT in 16m 44s","accounts_in_message":[],"_revision_number":2},{"id":"061dd5c28ca5dd7320255910fba36d92fb0b4964","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-25 15:32:56.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c47d970adabb719cbbc20890e00b4676a186f4b8","revisions":{"deed05df5edc8ed10def9175a187be9b358b6360":{"kind":"REWORK","_number":1,"created":"2019-12-17 16:25:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/51/699451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/51/699451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/51/699451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/51/699451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/51/699451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/51/699451/1"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-17 16:21:51.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-17 16:25:43.000000000","tz":0},"subject":"DNM: test CI","message":"DNM: test CI\n\nChange-Id: I560aa5db3053a2b19819d64fc514f4eff8928603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/deed05df5edc8ed10def9175a187be9b358b6360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/deed05df5edc8ed10def9175a187be9b358b6360"}]},"branch":"refs/heads/stable/train"},"c47d970adabb719cbbc20890e00b4676a186f4b8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-12-17 20:19:23.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/51/699451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-odl","ref":"refs/changes/51/699451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/51/699451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/51/699451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-odl refs/changes/51/699451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-odl refs/changes/51/699451/2"}}},"commit":{"parents":[{"commit":"8efa1b38fe355887576782b756b9aec9f647a574","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/8efa1b38fe355887576782b756b9aec9f647a574"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-17 16:21:51.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-12-17 20:18:23.000000000","tz":60},"subject":"DNM: test CI","message":"DNM: test CI\n\nDepends-On: https://review.opendev.org/#/c/699456/\nChange-Id: I560aa5db3053a2b19819d64fc514f4eff8928603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/c47d970adabb719cbbc20890e00b4676a186f4b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-odl/commit/c47d970adabb719cbbc20890e00b4676a186f4b8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
